)]}'
{"id":"openstack%2Fglance~874078","triplet_id":"openstack%2Fglance~master~Id27f12aecce5a981dedf0bc3513a2f19a1e2a24d","project":"openstack/glance","branch":"master","topic":"antelope-m3-configrefresh","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-02-21 13:28:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id27f12aecce5a981dedf0bc3513a2f19a1e2a24d","subject":"Refresh Glance example configs for antelope milestone 3","status":"MERGED","created":"2023-02-16 10:28:27.000000000","updated":"2023-02-21 13:29:57.000000000","submitted":"2023-02-21 13:28:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874078-antelope-m3-configrefresh","meta_rev_id":"70917456b80da67f5c31ef43f2194c3b1b26d3ca","_number":874078,"virtual_id_number":874078,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 13:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-02-21 04:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-02-20 13:48:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-02-20 13:48:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-16 10:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-16 13:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-16 13:59:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-02-20 13:48:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-02-21 04:53:04.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"61146a8caf2f79790a82db6634ebe454eaa5bb65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 10:28:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"842a8c5fecd1b4a37c8ac75e00ddf044a15ad640","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 10:30:18.000000000","message":"Topic secure-rbac removed","accounts_in_message":[],"_revision_number":1},{"id":"4cb1451c7661d4b04f7bb72c4e4c24b7bd6aa39e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 10:30:36.000000000","message":"Topic set to antelope-m3-configrefresh","accounts_in_message":[],"_revision_number":1},{"id":"bde2667d32a876ef84ecf701dce3737ce12d7a44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 10:42:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c788e7bea5d4827a087d5654c4e00cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c3092d850074e4b80b52c070f0bd047 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/241f1bf03c194421a17232d5ea2fe7bc : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44991980cf34fe5af3ce9b7551f845119afc4f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 11:39:09.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a810ef4175da9dcd8361b1128762dd23a7faf4bf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 11:46:45.000000000","message":"Topic secure-rbac removed","accounts_in_message":[],"_revision_number":2},{"id":"9049a01d1ae8087a29257b180719cc4807993bfc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 11:46:56.000000000","message":"Topic set to antelope-m3-configrefresh","accounts_in_message":[],"_revision_number":2},{"id":"dbac5e9a11bdf44383b5f7e2d9f7fe37fd7177cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 11:48:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c10704dc576346f990d2219c69ccc97b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/636dca156617461489a93f0f680d2eb3 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bbf779ad0b3429d900597a008025172 : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ef65267b66efa38a2741bcb0fe3e92c613ed9feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 13:56:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/43a2f8917cf44522b59295e882d10b6f\n\n- grenade https://zuul.opendev.org/t/openstack/build/15fb9f562d894d84ac03467bceefc6e6 : SUCCESS in 49m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/63c3f3b08fad4346b11e9d907571ed09 : SUCCESS in 52m 42s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf474d1a0efc475fa43876ef6a0414aa : FAILURE in 1h 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58a50ab93e024464a5b289bf2b552732 : SUCCESS in 1h 11m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153785d5672d46eda0becbcfdaea7f07 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d84ffff5df6742d891591804a2965ee1 : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b143ce80e48e4f48b2a27d522a9de5a0 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b95788b416847ab9f950f1ac0f5f9a9 : SUCCESS in 7m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0c786b9105604363ab11cf5a4522997a : SUCCESS in 11m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/397a66626abc445c981fca7f94179b12 : SUCCESS in 8m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/519cd0c995c343928a45885bb74083e4 : SUCCESS in 7m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/050ca3e833db4414978745a985ed6ea4 : SUCCESS in 1h 01m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f1d78ab7be34624a86e22a360d6974a : SUCCESS in 1h 25m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b5d379f0fd74942a93563d71dfcc5a3 : SUCCESS in 1h 28m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/31fd7df563194e0bb4e643aa78cbf0b1 : SUCCESS in 1h 47m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a0ca1801d6cf41aca916898e94f2fb95 : SUCCESS in 1h 59m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9fd5c8e9c774d7bac2aa3b26c5f00f3 : SUCCESS in 57m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5cc7bae614eb47b5a5f9d3a6b602c47e : FAILURE in 2h 04m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/76ad2f819600466fb6a53fd76a1a9c46 : SUCCESS in 37m 53s","accounts_in_message":[],"_revision_number":2},{"id":"86448bee8b7ae822731cfcd6e9519f2e522dc29d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-16 13:59:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4e6ada5d1283a1ef3ead77db024cf31ffc7af570","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 14:17:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb9fa9b09963468fb1a0cfe2137688a4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13ffccd6a88e421a900636a9005b17f6 : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c57a8b9ff2e742f0ac44df25a946ed6c : SUCCESS in 16m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fc57f8a73b4cc008a7083bd7dc1630dacae6ce90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 16:06:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c134ea378e544ce593c4e5f7d206f8fb\n\n- grenade https://zuul.opendev.org/t/openstack/build/4483f1b971c64d8dbec50bb7dce7d9db : SUCCESS in 56m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bc5063e25aa34039b6235e597a5cfd51 : FAILURE in 1h 11m 22s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1612bdc6f8745baaea8de49757b578e : SUCCESS in 1h 50m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9da07dd645ae4ab496975dd913cddf3f : SUCCESS in 59m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/991d62f323b145f9b8b2f922f8c3163e : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec05f05399204902807d85437845a43e : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16e9f3e14371484e8fd6dae1e5533a3f : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad68c042309e41ecb66897abb39a8c6c : SUCCESS in 8m 23s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/645eab9e76174de69173899467f07a4d : SUCCESS in 15m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7edcfed28bd4666a3dad4578f3ba9f4 : SUCCESS in 9m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4cf736dd2e524a48bd0613153eab06be : SUCCESS in 9m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/62f8bbd03ada4c369ce325e67449fe61 : FAILURE in 1h 14m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1d083d6506e648b3b4ec79cd4ee247c4 : SUCCESS in 1h 30m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0afb1ed5576a449b99b242da8100154d : SUCCESS in 1h 56m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/24a35b01507043e8b1f893a3e2552b7a : SUCCESS in 1h 29m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/12e0aeeaa2f04978a8ab8e4a6cefebd5 : FAILURE in 1h 13m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d47dfaf437ec4b8d8e80de8cdcf73454 : SUCCESS in 45m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5056e3b37dd34277b7723415e9772895 : FAILURE in 1h 20m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/afa13bd0e33f4a43a4dd8825c5985f78 : SUCCESS in 55m 31s","accounts_in_message":[],"_revision_number":2},{"id":"5db75f4223bc1835fd56f1f8f5358161eb9089bd","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-16 18:08:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fe729a66140ce47e9d27b4ba833977cc6453881f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 18:14:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f831a46360741608077d2c52609a16a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ec1520863b943e3bf2af67e81b18396 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62f0077c81ef4c25a7b2c4493d35232f : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ce23fc42e160ea0736abc8cc4605375a5c6cd51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-16 21:39:35.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/325823ff0f5d4f55af8ad9b49e5b01ad\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c1e5119799944ebb5f0ab9258a15d24 : SUCCESS in 52m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b00f824534dd49058f9093a0fdab3c60 : SUCCESS in 52m 35s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97924947f3cf4fa5ab55c9bf84a190c4 : FAILURE in 28m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e20db74f2be34da8a3953581e2230ebc : SUCCESS in 1h 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41cf314fe8d24fb1990633bbbba2c298 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4264e9bf57db47739b288a7819621d64 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab4e67ab803d4334a208391ed063c4a3 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c9318c70a2d4a4db2d78dbed9d304bd : SUCCESS in 6m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1603c6e3ff9149a9b69af333f396e993 : SUCCESS in 12m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e43047bc55f843449c175ecea458252f : SUCCESS in 9m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/00a4fd0f144f4c089cdceeed2da8a2d6 : SUCCESS in 9m 47s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4166aa1c871649609bf31ad2cde0326f : SUCCESS in 1h 14m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a7bb81a91ecb447c9e90e78e62c7032d : FAILURE in 38m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/323773116bb845ffa8a3e7b0a135bef0 : SUCCESS in 1h 30m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ae65bb3021144cd843c20258724a46d : SUCCESS in 1h 33m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d2961f84158e43679f50e671fd08f994 : SUCCESS in 1h 30m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8cafd94343b48b185d5951ea9810985 : SUCCESS in 52m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64e55ff2a83a47d2afb9ae4a91f019e8 : SUCCESS in 1h 34m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/61a4514555bd4831bfcdb2c2f7661148 : SUCCESS in 44m 51s","accounts_in_message":[],"_revision_number":2},{"id":"f9d5958f678315964f92558efa25b1c9c966321f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-17 04:39:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"753f45c2db58138ac9ca8335479b0e24aadf3d55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 04:58:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/539b0be343784ca48e5cdf0febc6be2c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b979e38ec30b426f957d942420141001 : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b612b1ff85c343e99c65d77d5cc50414 : SUCCESS in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"01032d63eeaa52cc2d0890719d1161b3d6047ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 06:45:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fed2c904dd44f909baffd391d0e67ae\n\n- grenade https://zuul.opendev.org/t/openstack/build/8d469cd7430c4b9e9804a0b3786e9e08 : SUCCESS in 56m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d139b9310861419eac5d244c9d51742d : SUCCESS in 55m 50s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46b98918c47f43eeb4fd8dcfade659db : SUCCESS in 1h 06m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/886f18e9374a402499d19aa6549383ba : SUCCESS in 1h 00m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83f00419dbcc4c10ab73d6ae0e347388 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24c92cc8910049f29e34645dd913765b : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2964b691472f44ae8c005cff46571d2c : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b03b6f02ff4158ba216377ef04b1f8 : SUCCESS in 7m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/874dd71ae9444164a0ea28bf7f119efb : SUCCESS in 12m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7d5a8a1e4dd1480d97c04f75b528bb72 : SUCCESS in 12m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3a51a57c0fd249fa9cb91cf2c702ef01 : SUCCESS in 12m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c1a40d5d640b4898a70438a1fdc6fd62 : SUCCESS in 2h 03m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d71661fd3df54ee8b7d007226f8999d7 : SUCCESS in 1h 38m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6aba24a5d72340a4af58b6335a92bb23 : SUCCESS in 1h 29m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9cc5e877ccca48cc8feb3b3e36f600db : SUCCESS in 1h 35m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/05efc397fc884eccb2247d60a93dea7c : FAILURE in 1h 33m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1993c267a91449faefb5c7680d59f46 : SUCCESS in 49m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c87dfab9682403ba81279a690f8cf9a : SUCCESS in 1h 57m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a5bb649d3ae146f99b6432e7adff4d0b : SUCCESS in 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"aecf089939b031bb6d60bb55ce5fd13eff30b9bc","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-17 07:13:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9fba2db3724d07794880d911fc0eea10b73e0ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 07:33:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3be4e387bf9b4bfcb7ac0a3535b67eda\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a65511664ae54619b174a0fa1d97361d : SUCCESS in 17m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b8c95d322f24629ad19c0d966575b31 : SUCCESS in 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1bdb882fd9c80b8044fe145521e18d54eac0ad52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 09:11:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abd1bff691af4ddc93bab43a98e0e266\n\n- grenade https://zuul.opendev.org/t/openstack/build/27957e6a52c64ed988dde08e9bc047f9 : SUCCESS in 1h 10m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/da6af0af9336493e946e67c9475fcfef : SUCCESS in 1h 05m 08s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4b565fb2f8e461db557961556815aed : SUCCESS in 1h 40m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a8d4873e38d4248b31941b283fe88b0 : SUCCESS in 1h 03m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d17034d887141d6979ca31e5527bc1d : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1421a9055232433fb7ce0158920f0f28 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c125b20773cd45248871b3b6f2cfccea : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9f0cb10e3b4740b92425f269ff41ad : SUCCESS in 5m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ebca658708f842249b64db543b641d44 : SUCCESS in 13m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ad015e40b995431b8a1a791bb61821bd : POST_FAILURE in 7m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/41eec6f5bd8a43499e119ff557938949 : SUCCESS in 8m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1b0a49b90a949e49d053833300a048b : SUCCESS in 1h 05m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ed51ad88816d4fb5be3936049e3829f4 : SUCCESS in 1h 25m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7eccd931da5d48aa9cfb9717a8d1c3ce : SUCCESS in 1h 16m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a7a85e972d864b76b87acec62653159f : SUCCESS in 1h 14m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7bc1193ce0f415ba20df162bc226287 : SUCCESS in 1h 32m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79bd9800a74640579a7159e4986891da : SUCCESS in 50m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0a877ed6465f47889f97ed3bb097ec0c : SUCCESS in 1h 11m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bcc46ed18afb4124bea95e3dba23999d : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":2},{"id":"2d19564e6b27981d9abac9725f9a72e5f2e20bd6","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-17 12:20:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0090a078f9be0f197b9e4bb774b2574566a9412","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 12:26:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b8b578b7f94fb481f22ea8f5d577c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3aa86fc6ddc24e438b6baf5b2e6704da : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3df2e9a4f5d94ee0bb85cdabf15ea923 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a23b1f7ff3b2f27da77fd1805814626d2cfd0310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-17 14:28:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f870051f648043e2a4c28ee9eddae8b2\n\n- grenade https://zuul.opendev.org/t/openstack/build/c462e6f72cbe4030905fac533adf2935 : SUCCESS in 56m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/30ce15a395b94d8eb2e9a2063a1704b6 : SUCCESS in 49m 22s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8160274c2da94ad4ac1eb570dd235d2f : FAILURE in 1h 14m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f37e045671764d288697b8e8525099f6 : SUCCESS in 51m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fa3cf5f99e44a2fb5830802089ed425 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1fa19ae056774f3f97d3592cabe2dfdf : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8541398a24144c0852b95ae4919a3cc : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb5a90beb3e64f1c9f843200630eb7be : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cb6f2850facd42a68eb75ebe9894fce8 : SUCCESS in 16m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef9c36138ea24beab112c5a1203e6ce1 : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3c249ae8fe854e8aa6c049b45a4f4fa1 : SUCCESS in 11m 17s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e4559f91739748c8aa4b03521919ef79 : SUCCESS in 1h 16m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a8ed2c79a9544da7aee58b23bb5f84ee : SUCCESS in 1h 35m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a81c0a81a80d465e98c7e003fd7d3225 : SUCCESS in 2h 01m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/39ad9b8b2df24f1c9f6c9f1ee8441dad : SUCCESS in 1h 30m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/397cdd859fde4b90b2b9be5b86780afa : FAILURE in 1h 17m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80a599e80cce48a99dc0afd711814d44 : SUCCESS in 50m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2319e00c978a48e38f445a30c92a03fa : SUCCESS in 1h 21m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1baf132b3ba544be87aea198b0536a7a : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3bb47ab46fdf2228453bc98c6188777170135b5e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-20 05:40:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"20cd14d4c731cd5984e3b24671ad2f9d63b28023","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 05:58:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae01ed542f3640799c2a2132edb75a61\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/990de7a786754962b4887450e0bef0c0 : SUCCESS in 13m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30387c3632124bedbe74e0f107f7e161 : SUCCESS in 14m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"982a96c3337710183577d956f1d97049ab74c682","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 07:31:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c5a7b5f9ce942439b8eaeebc43638e9\n\n- grenade https://zuul.opendev.org/t/openstack/build/83f471f1719540cf965c87a5cf8c63ee : SUCCESS in 56m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f933c92ce2a24705947dc99553a678fc : SUCCESS in 54m 22s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/448a10d06413472690fbacd6ac15624b : SUCCESS in 1h 45m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f79888ba9f845a1acbf1597d2fef0de : SUCCESS in 48m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f930876606466fb7675eb8e29c330c : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d907f8513b743038c083ee0c942ad18 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b76b93ae13e4b3bbf5ba1d6f37e367b : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27cca022b53b434aa7c1df5bfa754bee : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/863aac428c2d462c981b739b1da7c78e : SUCCESS in 13m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2c845cd6bc0497ca8636fab2319f523 : SUCCESS in 8m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bdce309c61a04cc090e0b8f5270e0e17 : SUCCESS in 9m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eeb73710b3174088871f349f01541bdc : SUCCESS in 1h 16m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/49997665db0d4013856eb710b0b06876 : SUCCESS in 1h 38m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/61a6f0db7dba4266998337179397f0ce : SUCCESS in 1h 14m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/97a3908a5a914eb1b176a44dc297646a : SUCCESS in 1h 16m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f17f2eef784c4e719a820404e3fbbef0 : SUCCESS in 1h 40m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bba2c6e8d354a8cae03636e633079dd : SUCCESS in 43m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c0583b317e04b7b87d9679a1d7859c9 : SUCCESS in 1h 10m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a070f9806184965aa1999bfc8f43c4f : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":2},{"id":"a2442ff38928e77eeeb598ecb324aa6d5448c4e7","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-20 13:48:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a2a395f029e4945d809ffcbc01933320c6583eb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 13:49:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b28fe279ac1a778f622000d0f37fc2a360ea5f73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-20 15:25:32.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/873372 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"e65bf18c5de6ea5c72ff5a968c3731b7950f9044","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-21 04:53:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"991be48547fe304687fcb8e795897a4ff635647d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 05:04:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2432fe168fc4b9f9807c936b81050ac\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a73220ea02d4503b96387d8d5716f1b : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0d4ce5f0749541a9a6b2ab2cc36de1fc : SUCCESS in 9m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b9a7a54380d36b71f5254d32b57dbd76c4d19dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 07:04:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94f24fca78864bd79ec22b7fda0f4318\n\n- grenade https://zuul.opendev.org/t/openstack/build/239ce1bc70b340b4bc46451584c1bb70 : SUCCESS in 49m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3ee036b97da242fa9578dffd465c72d5 : SUCCESS in 51m 07s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a6a62bdf487040999f9f28b4b62e563b : SUCCESS in 1h 16m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/83adce2928ed457cbdc673c975f360e0 : SUCCESS in 54m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97f6b1b4af4246da8366f1ad22e978b0 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/329a11513f4f43c09fe4c17d8325fab5 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f968512dcc24a33a6108faeb565a291 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c04ef43253a54dde807b7b5de6969160 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0a4b81cfeff9498bab8ff3e530dded35 : SUCCESS in 13m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/472fc88a737e4c4088c17f712ecd3394 : SUCCESS in 7m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5f588f7b75d241d89bc80532bca79435 : SUCCESS in 8m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca6fa7d70dc746e996742c1505187326 : TIMED_OUT in 2h 05m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/10900ffa23724cab85d3f5bd4e319ee4 : SUCCESS in 1h 32m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0870ff8ae2704e2f9bce7a3c86cf6855 : SUCCESS in 1h 22m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60d201fad87f443b945ff6fdec6555dd : SUCCESS in 1h 21m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/647340236972479787b6b4381c395973 : SUCCESS in 1h 40m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c81562bbda34a08ae4613cccde22bb0 : SUCCESS in 36m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f8022775be8c4919a363d2ad6f963171 : SUCCESS in 1h 24m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7690a899c3ef479d9a535d1f83586441 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":2},{"id":"c3ede7298c98033738ceaa974453a9f9302b9471","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 07:05:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af031172e2fe2366851f496921ee76f7b197c80d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 08:39:24.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab4ad6d814e0427dbf47793cc3695abc\n\n- grenade https://zuul.opendev.org/t/openstack/build/d630ee61fc9d48ac9cda1dc7485b4b85 : SUCCESS in 1h 07m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a447025fe1a74513b0ec503e52505c4e : FAILURE in 1h 33m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/590fac5203c74ca3894874354bae42a8 : SUCCESS in 56m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e5ffbf5310450cb71f2f208f048f7a : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cfcac36e420473b8ebaa209a2ce84b5 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20c7133a74314456bb0ada8db99b41b8 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41e0421e49c541c5bf2f6a535e7ebe75 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d285c3e849994d08ad83f0cae6d4114c : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/355063a6811249bf8eb7c907624df824 : SUCCESS in 8m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c165421bbc9a447dac16d85274798142 : SUCCESS in 1h 08m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b90fe047657c4672a94149e8f4db3a29 : FAILURE in 1h 02m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8296c32c84ad4725b1efd35b4f1d4eaa : SUCCESS in 1h 13m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6d17a33031074af1a5f19584128e93ac : SUCCESS in 48m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed63598306d04e889ab42b5d92c75025 : SUCCESS in 40m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9133b090a8104b2cb21ce5bffc3a0f6a : SUCCESS in 57m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c61bb165bd0b48bf9c57cf0bacf1c681 : SUCCESS in 21m 44s","accounts_in_message":[],"_revision_number":2},{"id":"221926e5193d3873424f8829baeb86024fa0c09f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-21 08:42:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6f38cffb9395c2872b3c68506eeb62b20e9f215f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 08:54:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04222dba8c794fcb9791611165bb18eb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2d30cf8a03c4130ad134cb6e00f2b5a : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f40ce4eaeac84e6eba61eef0e1ffce5e : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa7510f1e9ef213340419c904378553a464bc5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 10:41:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e2cec16e0b04f59bacc8238e230a419\n\n- grenade https://zuul.opendev.org/t/openstack/build/dcfcd42a9935481795c52ff3dfc2347a : SUCCESS in 43m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/868da7c50272488e98d5e18329232527 : SUCCESS in 1h 04m 28s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c3bcb2781b94c9cbdbe6990d1de9758 : SUCCESS in 1h 22m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9263583a581d4ccfb7e9aa8df4e28ff3 : SUCCESS in 55m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8ba67883e34404bc0b64df5f60dd5b : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cffd7f4e3f94edea046ad445f63d429 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7189d5bf09a48beaec52732487524a8 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7d09a80e0864b2d9eb3d8f50462b7b4 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/45ba6f43917c4e3c89fcd22e40a9f221 : SUCCESS in 14m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/43c45bd6d085444dba728a4e664f24d9 : SUCCESS in 7m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dc3c66f9c2bb4e10a8311d9d0662eff2 : SUCCESS in 8m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0ebe3f7852234d7fb9f73a9a1a72d4d0 : SUCCESS in 1h 14m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/94fb8c143e6d43f7b91333afab11dd4f : SUCCESS in 1h 21m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee7673fc9e5c4902b37491d9a77fa673 : SUCCESS in 1h 22m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1dd5140d4ef04f22a3f64f3d8137be54 : SUCCESS in 1h 32m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/78ca90cdb93b44bda0504b635603e00f : SUCCESS in 1h 39m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/653a5339b0134a13b7b2e4d5f4ff6f69 : SUCCESS in 58m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ab5690876d604cdaac82a2027f43d389 : SUCCESS in 1h 08m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/88c797ff65d446cf9760ab63bd38bb12 : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5505b05199f5e60d54c3c24da1706f96cc23c955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 10:42:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"56a487b6ba8964e0f78d1d9b51c9649110f1ba0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 13:28:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/996855d89b3844fd88656898c1000c65\n\n- grenade https://zuul.opendev.org/t/openstack/build/03ad9f820aa94d3ab65a124c862c89ee : SUCCESS in 49m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1d39df418a794919a00a52b9a23701f7 : SUCCESS in 1h 19m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f21f4d33f1142f4b3b8598aef95885e : SUCCESS in 1h 14m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90bd2a4a8cde489abe646737727f0b3c : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c05e392f9f594681af6652e0163088b1 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30e3d402cf284dc0a969cb6ff17b2268 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4431a215b45840a69857ee69c752f204 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/30dad5e447874d4b863143060299356b : SUCCESS in 11m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5d960354f7346c9b1b71e539ecb5505 : SUCCESS in 8m 17s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7a75620a603d4530a81935b977ddd81e : SUCCESS in 1h 29m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/69fa43f54776433fa200e24c6c22508c : SUCCESS in 1h 28m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6a1408afb0ee4ec79d091e099313357d : SUCCESS in 1h 29m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8dc99eb970a14029b099ac0d7591a207 : SUCCESS in 1h 00m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc1b9be4ddc84217bb070af790a2286e : SUCCESS in 53m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ceab77f6486402497f1d76ca972da4c : SUCCESS in 1h 15m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/575e7a424f904b0fa6ad4cf8b6c2844b : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":2},{"id":"00581f31958f99f0db18925f1fc2bd703306857c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 13:28:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"70917456b80da67f5c31ef43f2194c3b1b26d3ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 13:29:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4ffd37eb63147faa203f2ebee110358\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aca805a76f84cc2879d89e91c118c7e : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d54dddaa03fc8c76b25846a379e7662abbd3fb8","revisions":{"44985466660ce9b8bfbd446e67be1744dea37ae5":{"kind":"REWORK","_number":1,"created":"2023-02-16 10:28:27.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/78/874078/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/874078/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/874078/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/874078/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/874078/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/874078/1"}}},"commit":{"parents":[{"commit":"5d7bec57924886b37a7959b870595f6684b2336a","subject":"Remove deprecated ``enforce_secure_rbac`` option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d7bec57924886b37a7959b870595f6684b2336a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-16 09:37:25.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-16 09:37:25.000000000","tz":0},"subject":"Refresh Glance example configs for antelope milestone 3","message":"Refresh Glance example configs for antelope milestone 3\n\nChange-Id: Id27f12aecce5a981dedf0bc3513a2f19a1e2a24d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44985466660ce9b8bfbd446e67be1744dea37ae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44985466660ce9b8bfbd446e67be1744dea37ae5"}]},"branch":"refs/heads/master"},"1d54dddaa03fc8c76b25846a379e7662abbd3fb8":{"kind":"NO_CHANGE","_number":2,"created":"2023-02-16 11:39:09.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/78/874078/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/78/874078/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/874078/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/874078/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/78/874078/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/78/874078/2"}}},"commit":{"parents":[{"commit":"b20cc91e6f5c0a84965b0498c75b2ad40f361361","subject":"Remove deprecated ``enforce_secure_rbac`` option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b20cc91e6f5c0a84965b0498c75b2ad40f361361"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-16 09:37:25.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-16 11:15:59.000000000","tz":0},"subject":"Refresh Glance example configs for antelope milestone 3","message":"Refresh Glance example configs for antelope milestone 3\n\nChange-Id: Id27f12aecce5a981dedf0bc3513a2f19a1e2a24d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d54dddaa03fc8c76b25846a379e7662abbd3fb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d54dddaa03fc8c76b25846a379e7662abbd3fb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
