)]}'
{"id":"openstack%2Fglance~874698","triplet_id":"openstack%2Fglance~master~I3f67671c4426ff7f0080ccc5bcfb69e7f386c9fc","project":"openstack/glance","branch":"master","topic":"version","attention_set":{},"removed_from_attention_set":{"7130":{"account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"last_update":"2023-03-04 11:56:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f67671c4426ff7f0080ccc5bcfb69e7f386c9fc","subject":"Enclose all APIv2 versions in single quotes","status":"MERGED","created":"2023-02-21 20:46:11.000000000","updated":"2023-03-04 11:57:53.000000000","submitted":"2023-03-04 11:56:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874698-version","meta_rev_id":"c3414d51725302fdbb4a8d3cb6fe7b5959c341ac","_number":874698,"virtual_id_number":874698,"owner":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-04 11:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-02-22 18:01:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-02-23 13:09:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-02-23 13:09:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 20:52:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-21 21:02:16.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-02-21 22:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-22 04:25:35.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-02-28 10:43:08.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"3d2c0e856f1a6a077a13db89cd11d60b8267ef59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-02-21 20:46:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33b744ad3539751f217d1dd8d66b7852938ed4a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 20:52:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c7d071711694c35ace2722b217900cd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51de06b25ee14d258ae7dd2b4cd785bd : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b50ae93514d84623909344ba4a6d1e0f : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3914e2bca5ca4492f7f609e88d8a5934368f4092","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-21 21:02:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e9a33baea79c2e2ba9278245c7619af9f3acb55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 22:39:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09cfae3e05d845cb8c0ee88c10fd0768\n\n- grenade https://zuul.opendev.org/t/openstack/build/c64b9b9b5cdd4994a78ceff9a87a3d39 : SUCCESS in 47m 02s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8604d0a93b3f4cd2bb65c0dce031b0fb : SUCCESS in 47m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc6725bfd2644c5e8396610f37496d90 : SUCCESS in 1h 13m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba42681489d5455fa7441868b5490456 : SUCCESS in 1h 03m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c39473a493414d69b4198f4970d60706 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3b1f887b10441469298eb84b33c9349 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2def59697dcb4303bfa27e5adec16830 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e5902dfb5a4398ad4cb4b5d895870f : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/39ab9cd031a942979b76a749832fe341 : SUCCESS in 13m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abe678f1abbe47679fd49b3d19b7ed91 : SUCCESS in 10m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/156804f4150a46218cc0e30147582840 : SUCCESS in 15m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d12bfecb3adc4454a925a8a43d4f712f : SUCCESS in 1h 38m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/952ac84e26734fd7b77a95f834684374 : SUCCESS in 1h 34m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2f9f23b03082498eb39a59eaf8d39c47 : SUCCESS in 1h 34m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d892e6f70ea4412a8d5aa02e59d82446 : SUCCESS in 1h 21m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2670e8385f274f049eedff01fade1b92 : SUCCESS in 1h 45m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f732fda9161a4dabbf232fcc438a10c3 : SUCCESS in 50m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/421f7ac09a6e469ebc8fb03f79609e07 : SUCCESS in 1h 26m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6dd6253133454e1890c6436c730c9041 : SUCCESS in 43m 07s","accounts_in_message":[],"_revision_number":1},{"id":"fb8c832798a75024754c490dfaae43872d0e2ea8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-22 04:25:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a87a5fc451d61c21b29786bdb41663f716f1f7df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 04:26:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3611a341b7c7a8dfb23086af19d06f7ba43e2775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 05:54:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70170d108d5649c8ae6c449ac9a2c3f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/23571421d0df4f84972609b5fd056543 : SUCCESS in 41m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57eb4274ee2b418bb970c390073a5655 : SUCCESS in 1h 03m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d1f18848aa94275baabeafca69d3aa3 : SUCCESS in 53m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12ff136a8d144e6dbdad0869684a5f0d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bca3dc1874541babb37a46cf68f136c : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a8831e94f5041f5b2eff48d7c7a21a4 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4543bdc48f1b4178819d3d4d803be10b : SUCCESS in 6m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/041a64c5111e435cbcd54215c6eb5dbf : SUCCESS in 7m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/470af5e758334246ab6ccc6c410b7d72 : SUCCESS in 8m 45s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/17f153c289504f1ba26469293f69ab35 : SUCCESS in 1h 19m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a208afc04a58447eacd99fc946dd8d43 : SUCCESS in 1h 21m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/340a770da3454094a91cdc540c80c6a6 : FAILURE in 1h 12m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/41b9b04a5d9c4915950632aecd4e7c69 : SUCCESS in 54m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28761e91a49b46609ac0b87d25473cd7 : SUCCESS in 53m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/16b85b1c3fd74b3bb540870171a02851 : SUCCESS in 1h 08m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/92d1623cdc564497bece169024f362e0 : SUCCESS in 31m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c4c8394d4afb2b86aa90349fac289dcbff69f92b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-22 06:03:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7179d3605434020d09fae03357134f3caf4ade8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 06:08:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb685d29b53542c9a0b9dd2362249f9d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f7d2a1ae39d4bf6bac88f722897a09c : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4558bb6bae3444a5bf75bb338298e8d2 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c947458d2bf1aa983cb104446187ef78d462ca93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 07:51:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5c1360aa32948db9cecda07423b24b9\n\n- grenade https://zuul.opendev.org/t/openstack/build/f9003a45ad6d40799af7fe681b136b03 : SUCCESS in 47m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bb345abda6a94b6d8491e10226729d68 : SUCCESS in 47m 02s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1faf4053450b4b51a3e54324c76b9749 : FAILURE in 1h 22m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94c6a6be648544539f95c773e163e2ec : SUCCESS in 47m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59f5047d65c4e3d88a271ee4fc4f66a : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be7afbeedf5d4c73813a08953de84f3e : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e454453974d4f0fa214a06fe1ba149c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef8ead7c57ec4060ba416e5d5b4586c2 : SUCCESS in 5m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/857e341a6e22451294140ea1432e1f0c : SUCCESS in 12m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6df3e0cff9104226ab0d02add76c0139 : SUCCESS in 7m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e3a5aebfe1224a768718379a2c7abd1d : SUCCESS in 8m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/00fd0aa41eac4b75a454a3332714da41 : FAILURE in 57m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/95dd2d06f39744e0938e1155e2927e7b : SUCCESS in 1h 25m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dc8a564e79ff4887972a778107bbb7d0 : SUCCESS in 1h 19m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/afa9b6413fa04be0a204911071acd751 : SUCCESS in 1h 46m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d10ab1bec19f4fe5b5958fb860c1ff0f : FAILURE in 1h 17m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/115fced3f5644f1aad619dee86b0b3a2 : SUCCESS in 1h 00m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8de592dd4bc4fd5a11713b42a247b68 : SUCCESS in 1h 20m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eeb0af2efb1646eba53fe518f5cd7fd4 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d4664426c5ad772aff8a35090cf2524ccb956292","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-02-22 17:40:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"13e95b3ef9b8dc5b4d81b5e0de3cafc7d0a63cee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-02-22 17:46:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8e48af7aecf5f6f88f0741171fc341c0b2205bbf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-22 18:01:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"966e082022f3a45b71cbeb2dac66c13ef0c28554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 18:01:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de5da045d4f448fd9cd9b1aac2fe7a45\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1475d739a384d619b8130bc473d9f43 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c3b0f4d6f9484beb9648046c5bf15ae2 : SUCCESS in 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b6c5ead6136314f980b225eff18e08653590d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 21:30:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e0f5991a3e84d5bb4a07f2b26a06296\n\n- grenade https://zuul.opendev.org/t/openstack/build/29887217f4f4486e9a630b506b8be0ce : SUCCESS in 56m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f0883e0c6d60458db78dbbf2f56d5373 : SUCCESS in 1h 05m 37s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4049adad04d84198b7a8109023368a30 : SUCCESS in 1h 24m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fba9a27d499847419c49b0c29adc05ea : SUCCESS in 45m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf908e861d8f43c9a1d2ca257188db6e : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c1bcc18130348a5a260649d1addedcf : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89b700284c9a4556aaae99553245d443 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29b7af6346b841ada334202239354327 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/828a5f1b2f61466cb8f761ddbe26eed0 : SUCCESS in 14m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1357ec1c4b1047088ebd3a440285099d : SUCCESS in 8m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/87d154534a8740d9b4ae8fe2547772e9 : SUCCESS in 9m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1364d997916d41ffa34f2deb912a87c8 : SUCCESS in 1h 17m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7ad64fbe39d0470e87f0f2cad28715ac : SUCCESS in 1h 20m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b4be121846f7403297f93a008eab17f0 : SUCCESS in 1h 26m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fdb15e59f4744d618f088dc7b04761cd : FAILURE in 1h 44m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/300a54d48e1a430ab28aa9d05cb3bdf3 : FAILURE in 1h 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/026929d225bc49c28ea9fee14b31e16d : SUCCESS in 52m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/89a569b05f604f62ae0aa5218de17dae : SUCCESS in 1h 13m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f55685e7d86443e09795a5fa96556c4a : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":2},{"id":"d0bc862ae23b45f442c2c7a06c849cedede157f1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-02-23 13:09:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2605e3ebe5aa6e8e3c97959b2628c2cc2c777a71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 13:16:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16cfc9a3ebd94c95a0709facc60c2b95\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9103fbd5f514e9584a0cd69cf3acf44 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2beae7122b247079ea576ea197b9fe1 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"354644273960271329619c234008b137c848f85f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 15:15:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33763828e42e4f13aa3207bd6a287b05\n\n- grenade https://zuul.opendev.org/t/openstack/build/dea4fc37241145248fec767ab7bd959b : SUCCESS in 58m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/033d21675db14ad4bd4c2f7b7601e703 : SUCCESS in 59m 46s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a73d6e8ab15c493cb7a658521a1e2892 : SUCCESS in 1h 28m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a9078d1538243a7bae92e4ab83e776b : SUCCESS in 48m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf519c594544e94bdc2bdaa5c6ba7e5 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/989fa9211fc84c279c4b42dfdf1f5884 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8875c107e184b52a137a9546df66f9f : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51f9083804594a61a4b9a2a626f3f79c : SUCCESS in 6m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4a149db81a7e43c2821366880d4c44c5 : SUCCESS in 13m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0f2fe5be4784729b1d1d9f265561e7b : SUCCESS in 8m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a8fb8d495de7424e9ad8543c2cac99ec : SUCCESS in 8m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/79883d1682cd410ba75902e802d49a0d : FAILURE in 1h 45m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/84005e2462684db68d85a854283d2e31 : SUCCESS in 1h 19m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e7b0ab2599f4da8b64357bc4597dfc4 : SUCCESS in 1h 16m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6af9e9f9251c4ca1b562e4b5e61e48b0 : SUCCESS in 1h 22m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd1de00b9a074805baed87038385ca25 : SUCCESS in 1h 38m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00766cfcba5e4b17b761364b438ce7bd : SUCCESS in 50m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3c95e5bde3042b8b70df61daae0d84b : FAILURE in 1h 47m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/29a0702f9f93440d830ac0997df7e34c : SUCCESS in 41m 40s","accounts_in_message":[],"_revision_number":2},{"id":"4fd1c6db4a7eb87d340a5ec8ada6215573dec6a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-02-28 04:55:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ef436059559dd0c084ad66ea5078e2c8e093d241","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 05:09:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a317675d36fa41ff8c356181ab11d082\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9023c63770b24f77ba0fda6c578d30e8 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd95f8b0ec5947d99a382a44dd1a56b4 : SUCCESS in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6175e808ebce7d812940f6b517e7a711a59daedf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 06:42:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb36f541d5e94572bc93ca509bcd0426\n\n- grenade https://zuul.opendev.org/t/openstack/build/08bd294753994539bee0620d9ed81e0d : SUCCESS in 1h 00m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8693781a57824dadbe81e4bc6f77432d : SUCCESS in 1h 18m 10s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff2c35c5bd5944ba89057f0319874fa1 : SUCCESS in 1h 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/005689e8e22f49d7b505e857bacca0fb : SUCCESS in 1h 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9abfad6b6ccd49629a64c354dd2717e5 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c94aaa1a435d497e9713aaee82f299c4 : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10f33e9d87284dd18478071d7845b692 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54941c7baa2f426daef8fd44d92bb134 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c273ceebe24041649d9b33b0dbf6edeb : SUCCESS in 13m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e106d24f253548ca94d73cd79f52a4fd : SUCCESS in 8m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/505ced8f9f824da2931d41c40110fb35 : SUCCESS in 7m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/610dc36dfcfa460b8cad6300186e0a11 : SUCCESS in 1h 19m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43d7912133bc47c280c5916edbe13ec8 : FAILURE in 1h 01m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f2f30e0cf33f4c9abfbfc59eba1a7bd3 : SUCCESS in 1h 19m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/159b71cef7604b08af7325267166db76 : SUCCESS in 1h 26m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/534b668e29cf4bacb4fba88edb0a095a : SUCCESS in 1h 23m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a4afa5689d441f48e2547d59b47c0ac : SUCCESS in 53m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/597fdce830564704a677a64e6b99ee84 : SUCCESS in 1h 15m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cde41e85a46746e8bfe4b844d1a64d55 : FAILURE in 49m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ec6d700e33a4e6c55c9ea57ab60277eb133a100b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-28 10:43:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b495c207fdae84626abadf9a3afa4deec7587826","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 10:49:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f5a6c9adf4f497f9405f4493c2dadef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee9ee56f8ca04fef8f37de7e86b9ee64 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc82fe1c7cf2412eb85d8f85aa50a0b6 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2fd41d496faad3481cf4fed4af54180639b2e337","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 13:19:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ab948d811104538a268e006e38cfbe7\n\n- grenade https://zuul.opendev.org/t/openstack/build/945368011e1a4588a0b23537d88f0885 : SUCCESS in 1h 06m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/81877deea75e48d49152da10d3a8ee20 : SUCCESS in 46m 38s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e3bdbdcfcbf448a940ad98119786969 : SUCCESS in 1h 26m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/472ce0a495ff4af3aca7aec992edd822 : SUCCESS in 1h 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d39961570444134b2a47af2c27a83fa : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f81423a1a743cfa0c064bb6014e6c7 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02f05429ee464f528dcbfcdf0edca9dd : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caede42d0e74e62baee1b6ced386094 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4957331d8d8a4ecf868542fd1605c581 : SUCCESS in 13m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/118f24ec6a074667a6007721726ba252 : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/717862c5fc6b4234b9d3140dea14e533 : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d9688f72a59e41cba9cc97238f39108c : SUCCESS in 1h 11m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1ef042ccfe49444ba91cde34e3cad21d : SUCCESS in 1h 26m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e83d179c26264719915c6d558a5ed537 : SUCCESS in 1h 22m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c9a070cc7543404ab002a625cc351819 : SUCCESS in 1h 33m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f92f06258e5b41a4855184d1e5af4342 : SUCCESS in 1h 54m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/849b4dacaba3496a8996f2e726de834f : SUCCESS in 46m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1837039ea369420087906c9156d17336 : SUCCESS in 1h 54m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0f972051170a4f4cbc2e669669cf0b17 : FAILURE in 33m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b01ca6bd15c27f0f0d4143906e6422b047c1c0db","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-03 04:54:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90b4e2a9141069410ac281e6e1d29f7cc271470a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:59:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7894a9ffe7f1494582f9180789754434\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d586c30f4ba14a69bb078fbffa388b58 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2071b050e82e452fb040ed858bf00eed : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d01cd10824a2c49c051b3f2da1259e9714a58f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 06:38:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/678135aeb0144b509492e460c2f68779\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f8e04e754e7409da7a5fd9d8e753031 : SUCCESS in 1h 34m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0cc4341f12c142e5bb7253aadf417695 : SUCCESS in 55m 09s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3f518329ea7492f9087df17b3093e10 : SUCCESS in 1h 05m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e3ec39b62674c8f970c8490f23b5257 : SUCCESS in 1h 01m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbfdde4ccf4f4225881f0620319b1a59 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1961e9969c45ee864350d9a3e0d007 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebc58af7d23b456ebf370fb531bf74f9 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c815c5e184764561b1bb1e207283c51d : SUCCESS in 5m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ffe1e561ce6a4599b978ce0f404738df : SUCCESS in 12m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6305d7ab773148cb8f217d7c7abd1169 : SUCCESS in 7m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4473f3a99dc54c80990f1737c1026608 : SUCCESS in 8m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fdf0dda1492344dabe646846081c48c1 : FAILURE in 1h 17m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f361a134fa364864922acc0b1164fbb2 : SUCCESS in 1h 15m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d7efe528e14540669106a51ee923772c : SUCCESS in 1h 32m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2515fd2f1b4142ec8d983b2c149d2c76 : SUCCESS in 1h 20m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fcb0d38e28b2401fbc2303ddabdc45a1 : SUCCESS in 1h 35m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6baa9f7e5f86484cb8b5f97d9096c864 : SUCCESS in 53m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e45ecadecd714a919aed6b657c64f1c0 : SUCCESS in 1h 21m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b02e41218614dff88053d65db275821 : FAILURE in 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"28e8c5617ee18161441e6bfe38caa587c2ad0128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 06:38:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6d5bd07662cf3ddeb9651dd8959e857e98481e52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 08:15:27.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3474b259a9b548edb5043ece5c95631e\n\n- grenade https://zuul.opendev.org/t/openstack/build/cc127a30731d4f38877a487ac133d087 : SUCCESS in 1h 05m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c68ce46a1514480983936c8e35982c4 : SUCCESS in 1h 29m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e55bc12e7756412e848967ff79f4e3c2 : SUCCESS in 1h 10m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d1083ecc053476caeb96084f0500abf : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf47b9e35e814235a593c6364e72d5e1 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cb493b4b5e2418183db43f662ac71c4 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1749024fa7462bbfc216e250c9b554 : SUCCESS in 6m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5116a08e3e9841a4be6a0905651189c1 : SUCCESS in 7m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/341638d806fa4b91b480932f2ee98c1b : SUCCESS in 8m 19s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5aec21f6e0eb4698aa917568f06ab133 : FAILURE in 1h 02m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1adc1ebf66624648a50a78694ee72c71 : SUCCESS in 1h 10m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6f3b4fdab56145648053c578ee3f969f : SUCCESS in 1h 31m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9da59dcb4f74426ab328ef3845da606b : SUCCESS in 49m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca0b6ad6c4af474daf7ea33a2821576d : SUCCESS in 51m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a0bf9b2acd140b8835d5dc78207009d : SUCCESS in 1h 11m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cb15b50592bd459f94ce5e492338b173 : FAILURE in 35m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"976f87d794c93ce5a9bf77701ff69ef3626d7b82","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-03-03 08:26:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"872aec107f4986a09c20fe81bbef84cfcd57277c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 08:32:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b31062eb55e843b3afe4171e1a866097\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49e697d572cc493ebd921391c78276e3 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1e7ee20e30841229ef483995c217d06 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e7393ffc7d7413a0c9e84405e5043c386ff4538","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 10:51:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb9e78c54d224f949ae39c71575fc9bf\n\n- grenade https://zuul.opendev.org/t/openstack/build/fc3ac5a331114f219884e7c943add4fd : SUCCESS in 1h 08m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/db6c55d3bfa64809b59bc1ff56541afc : SUCCESS in 48m 28s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aaacb0b393594cef92cf7fd8a1f362b5 : SUCCESS in 1h 14m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14dd011b2fc94f2fb81e5d5f32eb00fa : SUCCESS in 46m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d353b0c7e3ae485f9401e04ce79b0888 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d436de13a1c74e00bf9156f5db9b34bd : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23890fe86d304923baa383e51a63d4cc : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e58ba592f04e3fb94079f394d7a03b : SUCCESS in 6m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/50f3ec97f93e49be9b0187f55a8d6347 : SUCCESS in 12m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c1ea176946154445b746fd268da005d3 : SUCCESS in 9m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/21733a81ae5e48709b4a0c0a039248d9 : SUCCESS in 12m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/54643e365d2647028984247f9d47fe87 : SUCCESS in 1h 17m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/05373ac4da0f48c19867dfbcda6614b0 : SUCCESS in 1h 27m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/734323b6f9914dc2b482cf9d161e0c49 : SUCCESS in 1h 24m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8659d1a1f2234a22a8a962c85bef5bd9 : SUCCESS in 1h 24m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cd1dcb059ff948acabd8b96f80b1a946 : SUCCESS in 1h 43m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/287b0cf65dfb4a11aa41b36e562950bb : SUCCESS in 52m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/568bf7c1f7e748449ce28c12850ad414 : SUCCESS in 1h 09m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/75af544db9414a96aaa194c7b7da5762 : FAILURE in 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4ea704158cc892cd669ce612d1cf07710ed8a31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 10:51:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1952039760f1a679d24e695f466d16565727430","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:02:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/527bcc24c2ae49deb6c422c10933a7df\n\n- grenade https://zuul.opendev.org/t/openstack/build/8112624c4dad4f379136cb9dbd21a594 : SUCCESS in 52m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9299bf9e4b3546afa4615d07f70ed0a1 : SUCCESS in 1h 36m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/992dbfdd3ef14add8fb25cad7bd2df2d : SUCCESS in 1h 02m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5902c113694f45a6ab9569d48d507d06 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7574db7ea902482987d33dc9eb27ec3f : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f494548c91dc42508b419b0a24890a94 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe17e38b639f47308c63359a028b9a19 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42a28396036e4bcfb9caca4fdffb5548 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40a1734b201b479280d922077e1f48aa : SUCCESS in 8m 41s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f081663492d4b0583cc3cfb347550a0 : SUCCESS in 1h 39m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d2b15800667947ebb003f245cacb7358 : TIMED_OUT in 2h 04m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e5750c7634884f52b1153c71d69ba822 : SUCCESS in 1h 33m 09s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f6b4cc0d9ad2453e8a9b47eb51481bc4 : SUCCESS in 1h 00m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/601d408f9d904bdda678f1cf747a5e95 : SUCCESS in 54m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63b3e108a6ce4083aa0841fc5f64995c : SUCCESS in 1h 21m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c92bc6093ece48b88368febc2b5b7ffd : FAILURE in 32m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6435fc42d53690c1eecc21474d5250ec3d55707","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-03-03 13:05:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd08931b19d6a2a98e741e000edf859b863cd447","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:25:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80548c0a44384529bd6fb727debf7953\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdc51ed3ea2a4456aa0ffd60ee35e675 : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/df01443bb8ac4523b7bdbf532b9eb7b1 : SUCCESS in 15m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fef145dff2c01c1b5822c7ae64d3c50c8f76f40b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 15:03:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd4524c8209545b7b4d85e164d7fa24a\n\n- grenade https://zuul.opendev.org/t/openstack/build/52f8949c275c4144989da6c93eec2e5f : SUCCESS in 54m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/14f5a7dfa784415999cd875c1250cb35 : SUCCESS in 58m 18s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/482b618c64d44fb198e44113ca46fdcf : SUCCESS in 1h 24m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/320d7bce797a4d2ea5b116563dbd4b92 : SUCCESS in 53m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d3e6286f00549e0b75b0c289169c203 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/030cbb1d279546c68151f5c2163d5550 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d79148380174d9099cc3ff5e15b61af : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d89b3b2f67f4fd18bcc54b76350e8a5 : SUCCESS in 7m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8f337bd5d90141d2b92d947bcab9041c : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e7d038aebbf4b02bf2aee00a5be08e6 : SUCCESS in 10m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a777c46617a94f5fb89f78e709d6bd19 : SUCCESS in 8m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e983340bb1404548820dc086f1062a07 : SUCCESS in 1h 08m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e43dbdef35ab4cfdb9ac2348bec8e123 : SUCCESS in 1h 29m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e00a48ad1b86451090fc4d4eac0e76de : FAILURE in 1h 51m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e30a50e46638451db5ee2f3068126f9e : SUCCESS in 1h 20m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d80960915cb48aa95538dff949efd40 : FAILURE in 25m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a34f38b761544bc9850fb75019998a5 : SUCCESS in 42m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/82461a41a2564762ad01530300f6d99e : SUCCESS in 1h 12m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1adf3d8843ee44d9ba393228c84fb777 : FAILURE in 30m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"61460cd3699b54218cf0a0fa6c51766d447a10e7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-03 15:13:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3ee60e2b84b8fc5ddac07974ba7f8c20b37e1b2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 15:19:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5994290874040a2bc98d2a95b11e0db\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0fa0980b2604d28842fd7ff2eed2a20 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f490e6fd660d4305a4a9c3add7836d38 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f150e9cd23ad2f7910ac3218b768954f2ccd78ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 17:25:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ce0e6579ce144c3825fdd2c0f204525\n\n- grenade https://zuul.opendev.org/t/openstack/build/035aad47845c4a05bd3b5c53d406b1c2 : SUCCESS in 1h 40m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f72182633f2498f84f2d08452a0575a : SUCCESS in 1h 09m 08s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dce0e0c09fd14874b48a13b0c9732cb6 : TIMED_OUT in 2h 04m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33bffefb82084d33bffa809a350c9567 : SUCCESS in 1h 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7adf620d0af422886bd5bdae242e0dc : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc43a47aa07142afaab17596e40ab454 : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ed13d46a15540bab1bf9506a2bf2e3d : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d1359d3b5549dfa687a5fe15b0f003 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d6aa7273682944e2b4b0f86508435479 : SUCCESS in 11m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f94da5d5ca0450994aad33ded5608e7 : SUCCESS in 10m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f8c931d980144b77837c7a90753e828c : SUCCESS in 10m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/739252c803374619a7fc43465994fa75 : SUCCESS in 1h 08m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2f9e24291c974220aa19397f24b8f79b : FAILURE in 1h 18m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/67fc33a3cc44467c8c3cb97345ea274d : SUCCESS in 1h 41m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/acb56c9473e243a4a22817fd91a2dfbd : SUCCESS in 1h 02m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/712f049bf5dd47ca9d6b809d86561a4f : FAILURE in 1h 40m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffd4b31fc1ac47b28d6d5c7af9bac4bf : SUCCESS in 1h 00m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65f8b961028c4bf8aca9a7d969ffda08 : SUCCESS in 1h 18m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ce88719e14e444ddaefaa3d0f7389696 : FAILURE in 28m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c4d651b1a243cf3b82041783cb40c3ba4a8fe5e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-03-04 05:19:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"23bbc7598bdefb75d5c610d67097cf25448a7898","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 05:25:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21c040fad4424afd9a14a490d3440e24\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ba69cf4945848068d9f0c565d9554f2 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d70817297b8b40929c60682a6f94f98f : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10e5c5dbb08fb426f5ef14acdf817af5a07a1419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 07:12:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40dd51069b9647078ec8faeb288cd6cb\n\n- grenade https://zuul.opendev.org/t/openstack/build/c3f1a227c2e94ea89495849e007ec47a : SUCCESS in 50m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2af00d5a235c4d3abb5b3bf36d281071 : SUCCESS in 49m 26s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c8d87adb21d34b75832991bd174f7cb4 : SUCCESS in 1h 12m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/856772c47b0640939c6984609b312fd8 : SUCCESS in 50m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b7fafc6f8b842e6988ab832e96aefa7 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5c065ca6ad342b09d45676bacfbdeaf : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28c713dc9a104fcdb8162203af2aa645 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0932944f1484204be96caa9494f69b3 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c42d7fb6f149435f84f10237789a551f : SUCCESS in 12m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0849aaacaf5d4afabee0466e7fcaf6f8 : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7a37f2e87fd945f981ada41a9a03ca48 : SUCCESS in 8m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a546357c02194eccae47133f3774afdd : SUCCESS in 1h 21m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6929f6a702de489596195e6eb0103448 : SUCCESS in 1h 11m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3bacc5eb650a42b4b1c26609cfb499b6 : SUCCESS in 1h 33m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1c3b7aa7b0854aab8190a5ca1fdcd9c6 : SUCCESS in 1h 26m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1d36be1f8634e4ca204f35d644b751d : FAILURE in 1h 37m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41ba2610265d43648919150b92b126f7 : SUCCESS in 1h 02m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c8cb7f9bc164d728739ed2f2f679652 : SUCCESS in 58m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52ea24c9ea6040b6b5528481fb65fc6b : FAILURE in 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"431a456437b960403964695b8d7deeb12c7eba40","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-03-04 08:37:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8478a8dcb5f6e02aad6bf14a8ae71130d811a33b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 08:43:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b4a9e9403f74d1fad606bda238becfb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b541b6333d494f789a6e9ea8dc7d5082 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/423415ec407741bd967730c56b755d7a : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c656ca58d66b3fa9c66497e48b5e339907984c9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 10:22:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e728729e0be948358a75a226ed97cd31\n\n- grenade https://zuul.opendev.org/t/openstack/build/ae0d4fd50ef34d3184a83c4073da2094 : SUCCESS in 1h 38m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c1ccbc0967f849a699f509a7108bd590 : SUCCESS in 40m 43s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36cbbc570c054bf596dd528ed1f21fba : SUCCESS in 1h 08m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f780e42b444472288eb1711cf21131f : SUCCESS in 1h 03m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24def8d23ab84b908d9a3ca979b73d49 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d81cab5662664870ae148d668a9642cf : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/641c0fc274e649fd95563abcc18bffb6 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657e120277934b928bf7b394243fedc4 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f14a74e46f0f4c9ab2b5801523d25ccf : SUCCESS in 14m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66b01d24f06f46e2b45dbce49a4bd942 : SUCCESS in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fe1c79174fa94e109bf1ea9d1d45d6d0 : SUCCESS in 8m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/101f583b5de24986840914d6eecd62e1 : SUCCESS in 1h 11m 47s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d4fb7f49794b418a90467bd826756203 : SUCCESS in 1h 24m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/36d1103292074f8c8b97aff44d7c233f : SUCCESS in 1h 01m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9be5764b321e47c79e82e735fe619b79 : SUCCESS in 1h 24m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d55b9ae9778548a586d967ac446f86ab : SUCCESS in 1h 31m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e73a1b00559248d3b802d5fd96024cd3 : SUCCESS in 39m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d0be329894d44608ba3b7e67e2b5d428 : SUCCESS in 1h 22m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c86f955d94c342bfa0092f58421b6976 : SUCCESS in 30m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"79cd1051e4353e4a0be8c9dee6eeaec3556fd6fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 10:22:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"22b2a89aaab7b7956ed9096753ac12baaba11bd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 11:56:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ece2ec2476a1446080dfd7a1a0383a86\n\n- grenade https://zuul.opendev.org/t/openstack/build/93080273e23544d0b6936e53a3e51b35 : SUCCESS in 59m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6acc5c34cfea427ab4172ba8c7e32df7 : SUCCESS in 1h 24m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ff3555d4cb6f4098be6bbd6bc6a9f0a4 : SUCCESS in 1h 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2402d6cb13344b3cb756f96515120800 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b40639e3bae649d3986ecf7f93750d13 : SUCCESS in 4m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2367f0322c542a5a4317e507e7b1b2d : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/616a019e30264c94ae8f083b3f0eb43c : SUCCESS in 6m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/37d55d68d5ff4fe5b25dcbe114b404ec : SUCCESS in 8m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8dc85776dc444e0c95ceb8184dfd6c92 : SUCCESS in 8m 19s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5d8f93df5af64ef3a1289d29e807ccb7 : SUCCESS in 1h 24m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3907c983fecc41829f32a8d74611cf39 : SUCCESS in 1h 03m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec465cdf243140ca8420a6c753aea7bf : SUCCESS in 1h 29m 01s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b5e5360dc11e447683e128e2aec6eeae : SUCCESS in 45m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d7635ed18754e36835d4117cad5de12 : SUCCESS in 56m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ed42adc53ae42ad9c7804adee616a9c : SUCCESS in 58m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c27261d693d34b15834cf7861780838d : SUCCESS in 34m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e1ccbf59310946b70559235d8d57808332eaf9d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 11:56:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c3414d51725302fdbb4a8d3cb6fe7b5959c341ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 11:57:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/caf1fc9b7af8471889df671eed810967\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7286a2f87f454a999d377d0533c71b1f : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1da3e48fa8c7d791016ccf6b4edaba16d87a1c67","revisions":{"b7363e071b7e5fa1280b7c0babec493c52bfd2dc":{"kind":"REWORK","_number":1,"created":"2023-02-21 20:46:11.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/98/874698/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/874698/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/874698/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/874698/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/874698/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/874698/1"}}},"commit":{"parents":[{"commit":"f0371614d5093463ede1b846227db6473f9ece82","subject":"Merge \"Release notes for Antelope Milestone 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0371614d5093463ede1b846227db6473f9ece82"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-02-21 20:44:02.000000000","tz":-300},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-02-21 20:46:10.000000000","tz":-300},"subject":"Enclose all APIv2 versions in single quotes","message":"Enclose all APIv2 versions in single quotes\n\nWhen version detection broke for v2.10, only 2.10 was changed to\nbe enclosed between single quotes but all of them should be the same\nformat, a string.  In this case, some of the versions are floats\nand converted to strings in build_version_object which is not pretty.\n\nChange-Id: I3f67671c4426ff7f0080ccc5bcfb69e7f386c9fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7363e071b7e5fa1280b7c0babec493c52bfd2dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7363e071b7e5fa1280b7c0babec493c52bfd2dc"}]},"branch":"refs/heads/master"},"1da3e48fa8c7d791016ccf6b4edaba16d87a1c67":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-02-22 17:46:18.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/98/874698/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/98/874698/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/874698/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/874698/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/98/874698/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/98/874698/2"}}},"commit":{"parents":[{"commit":"f0371614d5093463ede1b846227db6473f9ece82","subject":"Merge \"Release notes for Antelope Milestone 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0371614d5093463ede1b846227db6473f9ece82"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-02-21 20:44:02.000000000","tz":-300},"committer":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-02-22 17:46:08.000000000","tz":-300},"subject":"Enclose all APIv2 versions in single quotes","message":"Enclose all APIv2 versions in single quotes\n\nWhen version detection broke for v2.10, only 2.10 was changed to\nbe enclosed between single quotes but all of them should be the same\nformat, a string.  In this case, some of the versions are floats\nand converted to strings in build_version_object which is not pretty.\n\nChange-Id: I3f67671c4426ff7f0080ccc5bcfb69e7f386c9fc\nCloses-bug: #2008116\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1da3e48fa8c7d791016ccf6b4edaba16d87a1c67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1da3e48fa8c7d791016ccf6b4edaba16d87a1c67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
