)]}'
{"id":"openstack%2Fglance~875733","triplet_id":"openstack%2Fglance~master~I7d1c9c17b56a6c98f687ef79945215b1116192ab","project":"openstack/glance","branch":"master","topic":"test","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-03-09 06:07:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7d1c9c17b56a6c98f687ef79945215b1116192ab","subject":"[DNM] Test whether SRBAC job passes or not","status":"ABANDONED","created":"2023-02-28 14:23:48.000000000","updated":"2023-03-09 06:07:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"adc9d04cb48560de440632f51b734d23a82609a3","_number":875733,"virtual_id_number":875733,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-01 11:35:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-28 14:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-02-28 17:07:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b32ad16cd009d25d898cd0423f9eddbd1bde124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-28 14:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0a00fb4061eb88f4d56e7b0d1e2ec37d6da75a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 14:29:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c341e110482a489b950425f2a0c201ca\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/790444103e0747f4ae9c22416c33ebb8 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6aada078099b4f869d2ef3657f78d35f : SUCCESS in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a6722411a370329762f6c3fa143642a6bce91b04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-02-28 14:32:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dc7b50081be77c539f592cfc296435f967bf4e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:07:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/375b42a63e25467b93422485b4c91f4e\n\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/553b9b6989e34587a27c95a8933ec4c9 : FAILURE in 37m 35s","accounts_in_message":[],"_revision_number":2},{"id":"09ff3759b2d711a7586931fab1e80765091e9c61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-01 10:31:48.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"317b863576c1c7a43faeb4c5261371e80009d982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 11:35:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2eb745e4fd68403f9d6b7a10b6b39d7c\n\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e838f6f717894831859fbc21a355186e : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":3},{"id":"adc9d04cb48560de440632f51b734d23a82609a3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-03-09 06:07:30.000000000","message":"Abandoned\n\nNot required any more","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"928d2112ffd37e152b31fbb110e4369004930c27","revisions":{"6ca63e33bf1ef0451372ded2bd87658f97d2b126":{"kind":"REWORK","_number":1,"created":"2023-02-28 14:23:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/33/875733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/875733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/875733/1"}}},"commit":{"parents":[{"commit":"f0371614d5093463ede1b846227db6473f9ece82","subject":"Merge \"Release notes for Antelope Milestone 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0371614d5093463ede1b846227db6473f9ece82"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-02-28 13:58:47.000000000","tz":330},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-02-28 14:23:44.000000000","tz":330},"subject":"[DNM] Test whether SRBAC job passes or not","message":"[DNM] Test whether SRBAC job passes or not\n\nChange-Id: I7d1c9c17b56a6c98f687ef79945215b1116192ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ca63e33bf1ef0451372ded2bd87658f97d2b126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ca63e33bf1ef0451372ded2bd87658f97d2b126"}]},"branch":"refs/heads/master"},"8f598383343118f6de9b4c4d092dbb5afab7c5bb":{"kind":"REWORK","_number":2,"created":"2023-02-28 14:32:01.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/33/875733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/875733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/875733/2"}}},"commit":{"parents":[{"commit":"f0371614d5093463ede1b846227db6473f9ece82","subject":"Merge \"Release notes for Antelope Milestone 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0371614d5093463ede1b846227db6473f9ece82"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-02-28 13:58:47.000000000","tz":330},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-02-28 14:31:15.000000000","tz":330},"subject":"[DNM] Test whether SRBAC job passes or not","message":"[DNM] Test whether SRBAC job passes or not\n\nChange-Id: I7d1c9c17b56a6c98f687ef79945215b1116192ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f598383343118f6de9b4c4d092dbb5afab7c5bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8f598383343118f6de9b4c4d092dbb5afab7c5bb"}]},"branch":"refs/heads/master"},"928d2112ffd37e152b31fbb110e4369004930c27":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-03-01 10:31:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/33/875733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/875733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/875733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/875733/3"}}},"commit":{"parents":[{"commit":"f0371614d5093463ede1b846227db6473f9ece82","subject":"Merge \"Release notes for Antelope Milestone 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0371614d5093463ede1b846227db6473f9ece82"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-02-28 13:58:47.000000000","tz":330},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-03-01 10:31:48.000000000","tz":0},"subject":"[DNM] Test whether SRBAC job passes or not","message":"[DNM] Test whether SRBAC job passes or not\n\nDepends-On: https://review.opendev.org/c/openstack/glance-tempest-plugin/+/875863\nChange-Id: I7d1c9c17b56a6c98f687ef79945215b1116192ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928d2112ffd37e152b31fbb110e4369004930c27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928d2112ffd37e152b31fbb110e4369004930c27"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX and -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}},{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX and -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX and -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}}]}
