)]}'
{"id":"openstack%2Fglance~876164","triplet_id":"openstack%2Fglance~master~Ibbfd8904b3a87b870ed8cde26245176c5b8e74aa","project":"openstack/glance","branch":"master","topic":"srbac-tempest-non-voting","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-03-03 04:51:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibbfd8904b3a87b870ed8cde26245176c5b8e74aa","subject":"Make glance-secure-rbac-protection-functional job non-voting","status":"MERGED","created":"2023-03-02 14:52:19.000000000","updated":"2023-03-03 18:38:18.000000000","submitted":"2023-03-03 04:51:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876164-srbac-tempest-non-voting","meta_rev_id":"0fc441e1a8a1544d7e78af3c405f0de9f0a570a5","_number":876164,"virtual_id_number":876164,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-03 04:51:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":2,"date":"2023-03-02 15:18:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-03-02 15:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"date":"2023-03-02 18:16:14.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2023-03-02 15:18:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2023-03-02 14:53:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 15:07:21.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-03-02 15:08:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-02 15:18:55.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2023-03-02 18:12:13.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-03-02 20:08:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"174900ed05a06af6eae306a5171b5115c1d9caaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-02 14:52:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84c697d270e04b71aba9f1faa375e7076588c62e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-03-02 14:53:43.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"64a2302c5ede249dfe060ed2b8a6828309779a60","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-03-02 15:07:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c042cfc464d4bc75739adefac0ba5bcc082fa3ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:08:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae4ce825ddc04ae79e99ced166333a4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b663ccd8d684d0d922f1114c1d33a3e : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28896d49fa214878b2a2fba476d5d288 : SUCCESS in 12m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2333e57a79e9771b4c6f9bd93065bb76a77ffc5","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-03-02 15:18:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b9c5b0f163f3180601ed790e18c31baf42c1600","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-02 18:12:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce5fb3c87eb3f4b2dc71b8b87654c946f48d6cbc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-02 18:16:14.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b33aaa77a70d538391fa1f84f21314c09728a240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 20:08:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b630aa57cf2640089ff57f0b113f1ecd\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1a0b8033ea674c0d9097f69771735cfa : SUCCESS in 58m 56s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f786526d4f1a45e98731d54b1bb56545 : SUCCESS in 1h 15m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81054091fae049848fac329a589c3788 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69840f9590ce45c091b1522f518812d0 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45277c96c6994d54a1b68866b5afecdf : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91ba153cf4047f2be4dbf1ed71bdbf8 : SUCCESS in 5m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a59c359475b644dda722b97745dd857f : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/af64c70272e543eb804f3b11873890b7 : FAILURE in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0e0ba4d15a584eacb108a122edfb2eb7 : SUCCESS in 8m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6cd762bafea343bf924648347344720f : FAILURE in 37m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66d4744b6da3171f13cc04fc20e9695e38b31743","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2023-03-03 00:01:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d5f82cc1e73b08afe035c5aadec2a323827df90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:07:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ec04771aab94741b0d56f3eecdc6d0d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b57c419830640329d11309e89da518d : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9b70ac578abc44ee8707718f0ce5f972 : SUCCESS in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"215cecd83ee0e9e378b5450e0c04b3cbebf9dc56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:57:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bddcb51838894046982174ee31b5bfa8\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/116bce1b0808423dad5db7dc407b0038 : SUCCESS in 54m 35s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e59365538a64488e838dd4bee6e94db5 : SUCCESS in 45m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e281f077df3e40cbbb3b8155df0ad9ad : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24c6bbda80a648d9a5fba1069cf6a2b6 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0643fd0ce9dd490a8f882975dd5a2db7 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c758d77221664830a05f297c927c7187 : SUCCESS in 6m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2d3f8a190f0e438c9279cac8b91a7361 : SUCCESS in 12m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c936546d5590401d90252d693627926a : SUCCESS in 11m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f9d67d9fba745739971ebb0dcb122ad : SUCCESS in 10m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0def1bcc8e134a08b5d851ed583545f4 : FAILURE in 28m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ee13dd64f8efa1ecaf880181ff95cc73354f81c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:57:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c8eaf280e94e93ee81ba27498cf0c6500b556cdc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:51:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a3af89e716a4f0787ece06f6adbb23c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59e6afb230e4434bacf5889a2148518a : SUCCESS in 1h 00m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cfe129209c44141bd538bf684e1a832 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e182a94e1d074702bb9574f3c4ab7260 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb78c1fe05354cfa9dd58aec984034f7 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85e8b7bcc9ed45dbb8f5a4ca713710a5 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26880a86f7c8480b9a9f8934b4bfbae1 : SUCCESS in 7m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6418dab69f7341fbbb7741b0b856ab13 : SUCCESS in 8m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e646772ecb74068a557a82f1435c7a0 : FAILURE in 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c213b95e2a502c24adfb1df4ee3af9cf94c03bce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:51:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e90ff891d4dbd2cb93f1d8742e2ef544759d7b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 04:54:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905c22cfd12f4888920c9efde3752745\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09dadfad4b848d1b20cc587a747fa0b : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1},{"id":"0fc441e1a8a1544d7e78af3c405f0de9f0a570a5","tag":"autogenerated:gerrit:revert","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-03 18:38:18.000000000","message":"Created a revert of this change as I94bc43b3e88e284fdc1158404c3ee3ee2b0125c0","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ae2ad39795a3b4c8e0d4f7169ead740b29ee033","revisions":{"1ae2ad39795a3b4c8e0d4f7169ead740b29ee033":{"kind":"REWORK","_number":1,"created":"2023-03-02 14:52:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/64/876164/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/64/876164/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/876164/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/876164/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/876164/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/64/876164/1"}}},"commit":{"parents":[{"commit":"f0371614d5093463ede1b846227db6473f9ece82","subject":"Merge \"Release notes for Antelope Milestone 3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0371614d5093463ede1b846227db6473f9ece82"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-03-02 14:47:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-03-02 14:47:38.000000000","tz":0},"subject":"Make glance-secure-rbac-protection-functional job non-voting","message":"Make glance-secure-rbac-protection-functional job non-voting\n\nTempest change [1] has broken glance srbac tempest tests, so\nin order to unblock the gate we need to turn the job non-voting and\nonce we fix those tests we will re-enable the same.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/871018\n\nChange-Id: Ibbfd8904b3a87b870ed8cde26245176c5b8e74aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ae2ad39795a3b4c8e0d4f7169ead740b29ee033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ae2ad39795a3b4c8e0d4f7169ead740b29ee033"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[]}
