)]}'
{"id":"openstack%2Fglance~876461","triplet_id":"openstack%2Fglance~stable%2Fzed~I3f67671c4426ff7f0080ccc5bcfb69e7f386c9fc","project":"openstack/glance","branch":"stable/zed","topic":"version-stable/zed","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-05-04 14:12:02.000000000","reason":"Change was abandoned"},"7130":{"account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"last_update":"2023-03-08 15:03:23.000000000","reason":"\u003cGERRIT_ACCOUNT_7130\u003e replied on the change","reason_account":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"}}},"hashtags":[],"change_id":"I3f67671c4426ff7f0080ccc5bcfb69e7f386c9fc","subject":"Enclose all APIv2 versions in single quotes","status":"ABANDONED","created":"2023-03-06 13:31:21.000000000","updated":"2023-05-04 14:12:02.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":874698,"cherry_pick_of_patch_set":2,"meta_rev_id":"143f150acf213d6903c7803075c1b44d7de67872","_number":876461,"virtual_id_number":876461,"owner":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-03-06 15:43:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":-1,"date":"2023-03-06 16:54:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 15:43:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-06 16:54:24.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"b307cc25f8388b131c91d177331dbc8dac41ad01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-03-06 13:31:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8423307e382bae39828ff3730343e0e30b1c0e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 15:43:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7d18a8cfc9a47f09ab49fe43047ad37\n\n- grenade https://zuul.opendev.org/t/openstack/build/1769ca7611144ec4a15be1879f1965b8 : SUCCESS in 58m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/55cf0e3654744fbdab5a5bf9718e9d12 : SUCCESS in 1h 22m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10071bb1b21b4b648b82e047df5a0a1f : SUCCESS in 56m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91235e2745384eaea60eee164d19e187 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c903466900a7452dbe035728a6b0fc2e : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59388b28a6ee4fbe80bf58208e7027a5 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d71f52a89f2c40d0b336a7783523ae46 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0b355e9c5434db2bc3140c12347cfbf : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f09efb2c0cc34d70bd2c74641350d840 : SUCCESS in 13m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9d479adba2a491d863cf542d659b1a4 : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/94bbccaceb5b4601b424409e620bfbf6 : SUCCESS in 8m 46s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2528911ca2044f3195ae4ad032b092de : SUCCESS in 1h 27m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8ed5327a2a564e67a3983a44483a6a9e : SUCCESS in 1h 02m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d28689f053204aafae2ad54efa941025 : SUCCESS in 1h 06m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a68e63dc8ad444be93767d4cd4b4436d : SUCCESS in 1h 03m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/273db70539584c7b84af32b1a028a806 : SUCCESS in 57m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0754747463a04729b9fa2be6531a1880 : SUCCESS in 1h 09m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9066321423be40aeab420259f0b422d3 : SUCCESS in 25m 40s","accounts_in_message":[],"_revision_number":1},{"id":"36c1da05fa5425ee90d7b45753945fbb7bd83712","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-03-06 16:54:24.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d2f642d829d6634c8ea15cd9a5fc30637c1b4a1","author":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"date":"2023-03-08 15:03:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"143f150acf213d6903c7803075c1b44d7de67872","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-05-04 14:12:02.000000000","message":"Abandoned\n\nSee comment inline.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9638722ca03683124d4a92c4a541ec347d17ac9e","revisions":{"9638722ca03683124d4a92c4a541ec347d17ac9e":{"kind":"REWORK","_number":1,"created":"2023-03-06 13:31:21.000000000","uploader":{"_account_id":7130,"name":"David Hill","email":"davidchill@hotmail.com","username":"dhill"},"ref":"refs/changes/61/876461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/876461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/876461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/876461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/876461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/876461/1"}}},"commit":{"parents":[{"commit":"b1e5292248fefb7cd1ff4292c9b94d336f4cf73b","subject":"Merge \"Enforce image safety during image_conversion\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1e5292248fefb7cd1ff4292c9b94d336f4cf73b"}]}],"author":{"name":"David Hill","email":"dhill@redhat.com","date":"2023-02-21 20:44:02.000000000","tz":-300},"committer":{"name":"David Hill","email":"davidchill@hotmail.com","date":"2023-03-06 13:31:21.000000000","tz":0},"subject":"Enclose all APIv2 versions in single quotes","message":"Enclose all APIv2 versions in single quotes\n\nWhen version detection broke for v2.10, only 2.10 was changed to\nbe enclosed between single quotes but all of them should be the same\nformat, a string.  In this case, some of the versions are floats\nand converted to strings in build_version_object which is not pretty.\n\nChange-Id: I3f67671c4426ff7f0080ccc5bcfb69e7f386c9fc\nCloses-bug: #2008116\n(cherry picked from commit 1da3e48fa8c7d791016ccf6b4edaba16d87a1c67)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9638722ca03683124d4a92c4a541ec347d17ac9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9638722ca03683124d4a92c4a541ec347d17ac9e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
