)]}'
{"id":"openstack%2Fglance~878633","triplet_id":"openstack%2Fglance~master~I5d398277b01a114cfb74af6458f8bee30a812cc5","project":"openstack/glance","branch":"master","topic":"bug-2012898","attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-07-10 15:07:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-16 14:08:00.000000000","reason":"\u003cGERRIT_ACCOUNT_33607\u003e replied on the change","reason_account":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"}},"29037":{"account":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"last_update":"2026-07-10 15:07:38.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"removed_from_attention_set":{"33607":{"account":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"last_update":"2026-07-16 14:08:00.000000000","reason":"\u003cGERRIT_ACCOUNT_33607\u003e replied on the change","reason_account":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"}}},"hashtags":[],"change_id":"I5d398277b01a114cfb74af6458f8bee30a812cc5","subject":"image_cache: do not reraise GlanceException on checksum mismatch","status":"NEW","created":"2023-03-27 08:57:17.000000000","updated":"2026-07-16 14:08:00.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7d5c88f01148bd0c59d9bb0b333d73d7a79ae504","_number":878633,"virtual_id_number":878633,"owner":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-10 12:12:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"all":[{"value":1,"date":"2026-07-16 14:08:00.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-27 09:10:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-03-27 11:22:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-26 14:14:19.000000000","updated_by":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"reviewer":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"state":"REVIEWER"},{"updated":"2026-06-30 17:33:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2026-06-30 17:33:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"e757a3bba64bc7897abb9e3350bc248678d958da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"date":"2023-03-27 08:57:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee3fc93b8ebd31e5abf54dda4e68667dd5b8b70c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 09:10:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4233c9054d1544a5be202fb1c6b7e1e2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42e55654b133466c91737bb72b8d993a : FAILURE in 5m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12f1920b9cfc4b59a3d8416bf708dcb8 : FAILURE in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7428b2f353c7af1dbac2c72a036a17f44730a64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"date":"2023-03-27 09:38:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fb57e6c991015a7979afb391d2dd2ab6453317ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 09:51:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b6f3fdbe03247cbbaf3d1793712e7c4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d21af9b8561740049030fa3511becc10 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f6cf143264d4ac99756bd48b32fb4bb : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"43f29102b2d50c4ff30a5273b04f9a6fe1336bdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 11:22:42.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1152c02e00145cab8be8ac654649f7f\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b9039c49d8949778658eef933b51f89 : SUCCESS in 50m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5c1d732fce242b790a973204534e9b5 : SUCCESS in 1h 37m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc9dbe4ef90d434f9d0e3f400a374bb0 : SUCCESS in 56m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebff6a436e1f4dffa9c73e0c064243d6 : FAILURE in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e0d81c7bcbf47a4a38f0769b8f6ca88 : SUCCESS in 8m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a9f3d0d4f12438991d5dbd72ca7414c : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9558fd1e93f74874b2eea328312b6a35 : SUCCESS in 7m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ef472e19c92f4f0a9436568504b927b5 : TIMED_OUT in 31m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d2905bc7b66846ccade2bf05d8e07500 : TIMED_OUT in 30m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/935d0c3574b84c70955a6ec01c5957a8 : TIMED_OUT in 31m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/821dd6319b7c404bb492da6c43aabcba : SUCCESS in 1h 10m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96a67ccd67684f7980b9eb95be7bef69 : SUCCESS in 1h 30m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/26967242a38b42c59c42a38679f85e72 : SUCCESS in 1h 19m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/144e0a1d30804c36b8eb4d5ec73c3fdb : SUCCESS in 1h 41m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/01b090d163ff45e484ca1516243fc98f : FAILURE in 1h 17m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca2bcd1469584a21ae7d62f1db52bd53 : SUCCESS in 46m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ee24d842cce407da96553b9ac178fea : SUCCESS in 1h 18m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e271269e137e4dc4b9245536b483e5f1 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f8e10ba7b3ff4bda55d769b96025e1a28c853f7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"date":"2023-03-27 12:05:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee55a065328b1ecd943e38dfb3975badb2bc180a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 12:11:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/548dc4f9c17b4e028489e3e109b32cea\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49f238bb8f4d4f55ac96274e93af7d2a : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7d71034f06aa473797cfb82e7f8dd9f5 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6bbd6cdfb65320968dcd2803b0b6976e175d9aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-27 13:48:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/97171fd114d94bf6ac941220a14009e8\n\n- grenade https://zuul.opendev.org/t/openstack/build/11682a64ae9440528c249b6c0d59104a : SUCCESS in 50m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65829cd2860e49429503dc583788acc6 : SUCCESS in 1h 35m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a44be640da704bd2964052e93042ca4e : SUCCESS in 57m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f15cd5445214c25b647bd16ef9c93be : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2618ec2a06fd4da6b1fa3d907ed0f84d : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40f2d9f36d7e4345a252b10fd5f0e471 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929dc95579ed47c6a13f0bfc79af19f7 : SUCCESS in 5m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a9a17e029025432c8a4eb508ff7133c4 : TIMED_OUT in 30m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2df69019c574423eaf19230bdfb3b0f0 : TIMED_OUT in 30m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/18d6450155c24706a20099e18598c893 : TIMED_OUT in 30m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/20884d53dd004b4fafc5b102f3803c3f : SUCCESS in 1h 07m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a58d45d85c7e43fda3a15631391e5267 : SUCCESS in 1h 14m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6bd0f281355a445696430ea2904acce8 : SUCCESS in 1h 12m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5cc8cfc7413d42afb35a5ef242fdc85b : SUCCESS in 1h 12m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27a4ec8eeb6a4537b77ea93d7dd885a4 : SUCCESS in 1h 37m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29d5dfbe566343d89cbd4a662d8b5b40 : SUCCESS in 40m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e79e364df184d11b8ee81e70aef371a : SUCCESS in 1h 13m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/65b43d169e6b4aae90b6f7a6f445f83d : SUCCESS in 25m 54s","accounts_in_message":[],"_revision_number":3},{"id":"5eed405d5ddd8239ee3cfc14a6979537e3d896ac","author":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"date":"2023-03-28 07:50:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2fc46317d4804d83b846bc655fb49c03c13cffde","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 08:03:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fce39a1bd6545909d6043c7880ba385\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/61d082ac42c14217a28f073cbce23dc1 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c00df5a60b114c5cb91d8d6292a20d51 : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d0164e2f3c28cd4e4b0072634d8d533571b75834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-28 09:38:20.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b522f46abad4402afe220d3ad7bed2f\n\n- grenade https://zuul.opendev.org/t/openstack/build/70914fee3577461eac5fc1327319ffcf : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53e5deceacee4bd691bd5f1d667a47ae : SUCCESS in 1h 29m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/092352e3b1c3421c8562c60bce3c6f01 : SUCCESS in 58m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/122a24b7682545b8b2c45e133f72e346 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/022fdadb26834cbc88f7ee260672b5ed : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8644f3df16a40b583a648a903305334 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79eb9f0c83364161a8a60b15c2237cbd : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b131d9163d00438a8cb85e02f7699ca3 : TIMED_OUT in 30m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/093c418367af4f2a927a64f0d3a6c74a : TIMED_OUT in 30m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f045bdc19ee64ad7bb8fde7d699d8fcd : TIMED_OUT in 30m 46s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/08f0a4e9a5a1499997d6dd3f74914a4a : SUCCESS in 1h 11m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/547251198b1048ae9a0fbe03fc161c82 : SUCCESS in 1h 22m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5950362e26bc4c09a7a0c3088e9e7e56 : SUCCESS in 1h 21m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0a010a2bf53472eb0f1d06ef625373f : SUCCESS in 1h 35m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a370f2add64e40c98309f426a9656f15 : SUCCESS in 1h 34m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b556cd970bec49ca9be7e3ffd8d043da : SUCCESS in 55m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b54c16ed0dfd46b1a581e421246108f9 : SUCCESS in 1h 16m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6d7b2378b666448183875c0b4a1fb1ea : SUCCESS in 39m 55s","accounts_in_message":[],"_revision_number":3},{"id":"a094a9bf974ea84972c395c2ffcbceb41aa53fe2","author":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"date":"2023-04-14 08:15:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f7fc7db733a221e9a617c0000b37597ea264b86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 08:21:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ed0f3cf7ae4d7e9da063d86bfe58b9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa42c18e371e4e6c990611e87f3d36f7 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d96efa5606434f1892cd203552ea2d17 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4aaca199306cc4552f6af3eae11195070914fb5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 10:23:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a422f009f9204c02a78610b2da1f6028\n\n- grenade https://zuul.opendev.org/t/openstack/build/1d72aee2852f4aa3a0160c1dce1d8e53 : SUCCESS in 56m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9a8bfa2308414b6eb09d6e584144ce43 : SUCCESS in 1h 20m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47ed8e0e7c3b42e69d5a7ad1f7ebb1d1 : SUCCESS in 1h 05m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6289eefcd104b4cba765ac01e289944 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fa883bf51a749b0bbab8c5397aa6096 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08367953b5c14533a27c19b1c345e943 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136472f496c84546b709aea9dc711f9d : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e4789d8186d84d5d9f72985d756ff034 : SUCCESS in 13m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94012bbbf63048cbb77533dff112ca27 : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ae42c9febfd1465ebb2a8f9550cf3a83 : SUCCESS in 8m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/911f7b3ebf314914a90420ad5f560370 : SUCCESS in 1h 35m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f5f645cfe4541b7ad17ecd2a288e9ae : TIMED_OUT in 2h 04m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ea888c7afc4545ac8ee4ca9d20b7234b : SUCCESS in 1h 30m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2e784d6fa9df4c5d8cd28e67f2fe97c3 : SUCCESS in 1h 33m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6bb9679fcc5d4edcaa8b6e740aa32340 : SUCCESS in 1h 39m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d50ce4a5b3b4f5a833fff1c8d66d593 : SUCCESS in 52m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e881834409b34e0982000504a93ab15d : TIMED_OUT in 2h 05m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1540b53fac774ac99c1ece3616449fa1 : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":3},{"id":"c3d51dc7a02fc4e509cc426a2580bff00cc2e59b","author":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"date":"2023-10-27 13:41:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c22b8cda74a17037d3f45536b6e4a2673e978bd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 13:56:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/631dffb499de413386c86f71ac14b5c9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27f1018d0e1a4d8eb11ce37cd39ea7a0 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/434a695126924bd2aaf27ff34052f42e : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f0705384dc94a36ba794e3587127ff5 : SUCCESS in 11m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ff132dfd1bb930db24cf2b138e60f6c3923ceca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 15:45:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/129b867118fe4f31b99f7bc751963b68\n\n- grenade https://zuul.opendev.org/t/openstack/build/d736efc995ac4a45a5b3827cd90ddb92 : SUCCESS in 1h 00m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/29fce0a0a98c42d5a503d64b9956a1f0 : SUCCESS in 54m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c85ac19b496f46519d91113ce54e5ce0 : SUCCESS in 1h 19m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbf0f23572154719894f4c213b9ccefc : SUCCESS in 58m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e188e9089e749308da5aa8210944a0b : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91d692df53be4d628fb4ed1e4cfa6dba : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ad4afb90190f48a8858ed6fdc5e5148a : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db5f007c2db475a99e3371a31fc98f8 : SUCCESS in 6m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/59bf1049aa584611953eb2320ce4a5d7 : SUCCESS in 13m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/631dc8cf805240cea028617d6bff7732 : SUCCESS in 9m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1fc2ac53b7c445488327d9b9b64d9627 : SUCCESS in 9m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/42aa3d2de9d849d9ae3b06bb1ff11f3f : SUCCESS in 9m 01s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/fdfbbfd8824b4f3890509e99e0749991 : SUCCESS in 11m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/253b936714724b04a9d5e2188434f74d : SUCCESS in 2h 01m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/29a69301d4da4bbda5689fd93ab10ac5 : SUCCESS in 1h 47m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b691da954174453e8bb05945da1d1033 : SUCCESS in 1h 42m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f1fab5c572a24902b719535a8ba3dea9 : SUCCESS in 1h 41m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/807e8d7b0f9749f3b95fc1c0ed0506df : FAILURE in 1h 46m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a9631ad1a2c466f96b2a889534350e1 : SUCCESS in 47m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6835f1c49ca64ba292b393f835de6dc8 : SUCCESS in 1h 41m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1959204815484d36a54095e8366343ba : SUCCESS in 28m 13s","accounts_in_message":[],"_revision_number":3},{"id":"35b4a9b028b3d9563842b6db26058bdf7b141229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2025-09-26 14:14:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1e06a0d8bab91395c7ef1478ce12bf189055a3d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 14:29:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e5a706f46b24b74897d3d0cc31e9e7d\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c8995f8fe2cb4a9e8a154eaa75410f55 : SUCCESS in 8m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/29247c3457df4d90839aee220903fa5e : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/11000c42397e4cc992f099563afe60e4 : SUCCESS in 13m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"38b02bb9bfa16140c596dbad79ffd83a802ee99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 16:26:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f69b90a64b34f46ad6b471c4b9e900c\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbe2456360ad476cbdc2ba4a7765b333 : SUCCESS in 56m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/37253b42136b422cb63461ff737d4bb6 : SUCCESS in 1h 05m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f4ea051a04846fea15438571b62adcd : SUCCESS in 1h 13m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b24a9251d0ae4e979a4c49c02fc2337f : SUCCESS in 47m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/957e43a283994839aba36936124ff44f : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03aad64188f741febf3404219d0d0d10 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49092625b8c64a02a456287d288b334d : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/de6c5cfad5064b1a8c7b29fbf96d67a1 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e60ef0c24932481ba4b9b4b442ce470f : SUCCESS in 5m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a3aeb1c8de445aa8d8f713179ac936e : SUCCESS in 5m 33s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a213e773e0da46d19c77ce095b09a573 : SUCCESS in 11m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1096de59db9c471aa32ab81a6215ed6a : SUCCESS in 1h 40m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3592273c07f9414b96085d647b8d1a37 : SUCCESS in 2h 09m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/990e89687bd24b34a3a69d618493145c : SUCCESS in 2h 01m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cb56a62e295545bf8b4911f3e82bdc64 : SUCCESS in 2h 00m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b3dfce0a59842199f75d48c4cf1489c : SUCCESS in 58m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/93f87673f80646abb70ca70dbe2f77a4 : SUCCESS in 57m 56s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/754e8c36017a4ba29e13479ec10111fb : FAILURE in 35m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d6cb27a894ca49ae902e303fe6ab7275 : SUCCESS in 1h 45m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1c58c63f5c084df99817e6ef4bc04dd8 : SUCCESS in 24m 43s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/197b18b785584a43b4baebb5cf686912 : SUCCESS in 1h 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f13aebbf6329340866b1498e2198e2ad54ba37cc","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-30 17:33:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dcdd0176305bf85122fe7505e9f62507fc70d578","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-01 06:24:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e4f3d9829993e2125bb3e9db71e3539d33ae5d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2026-07-10 09:22:40.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"c38d84d9bb811386beee819f339e7d1b62a0da55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 12:12:22.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb346314f805475e8047c90e24cc1ef9\n\n- grenade https://zuul.opendev.org/t/openstack/build/1432ed138b1141a1b10b3eac0c4d20f8 : SUCCESS in 1h 07m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e13ee0f7660f469a8f8968db204fce50 : SUCCESS in 1h 10m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/54f83cb4958543e1a89cabeb55b48ad3 : SUCCESS in 1h 24m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5601732618e64d038ff0a795c622c037 : SUCCESS in 1h 04m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52566a1bbc9416b980e230b44ebfa9d : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4306855859c24013bcf9cdfb25da0b1b : SUCCESS in 7m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a760d28d5e4d489fbbd8d94878d0a9b8 : SUCCESS in 5m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a709ab23c7af446cabd17d2cedd33000 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b811d0a27cec4e65a8ef3f85161f0318 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/986b1f13c03646e2ba81c0da1c969e62 : SUCCESS in 8m 14s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/b82df78db5b74726be5ad3268bbf250e : SUCCESS in 6m 28s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/60c7ef9a548349e58b5d7b17a7a5a3ed : SUCCESS in 6m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fcc1637820214afb82f36f44bdb16bc3 : SUCCESS in 2h 00m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5a83451a4f944916aa61a7e02c6d3ea5 : SUCCESS in 1h 58m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8033a22905a34f83ae5753f47c06b948 : SUCCESS in 1h 11m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5efa0cf72ab642bfbe0d7f6f882bd372 : SUCCESS in 2h 39m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc986df334f74df795655d8e3b93d37f : SUCCESS in 1h 00m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1fbb605506e44e592396b0d34881675 : SUCCESS in 1h 56m 17s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f3b0c85ade944aafab36b0c6b586f5ef : SUCCESS in 2h 07m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef82130f23b747b992476c8c3b1d93b4 : SUCCESS in 34m 49s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/992e635311ee449c9b7a6f80f0c468fa : SUCCESS in 57m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d479d585ba4a9cb12e615bf3d1129934bcad1d6c","author":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"date":"2026-07-10 12:23:39.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b85eae19d921a8a24b343e07ce26e6a374c65150","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 13:43:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b568208ee32147769896bcba7e48b373\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/851dca207a484a5ea3eaad3a7fe8d6f1 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b13c83cef45f4d7db04f2b6eb2ae1546 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/9bda130843494d359c8ef49b8db6a4d4 : SUCCESS in 14m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a66995e3eb75a48743edba2c5661e96c48194680","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-10 15:07:38.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"7d5c88f01148bd0c59d9bb0b333d73d7a79ae504","author":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"date":"2026-07-16 14:08:00.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"582432b8232506a0aef5e2b563ccdf628f8efc84","revisions":{"d4f06afa2d5e0aa2a2501d621755b55e254e9963":{"kind":"REWORK","_number":1,"created":"2023-03-27 08:57:17.000000000","uploader":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"ref":"refs/changes/33/878633/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/878633/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/878633/1"}}},"commit":{"parents":[{"commit":"b023562e1d31d3e60c195d701e6a9ff52e2fc956","subject":"Merge \"Add a check on startup for staging directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b023562e1d31d3e60c195d701e6a9ff52e2fc956"}]}],"author":{"name":"Pierre Libeau","email":"pierre.libeau@corp.ovh.com","date":"2023-03-27 08:48:28.000000000","tz":120},"committer":{"name":"Pierre Libeau","email":"pierre.libeau@corp.ovh.com","date":"2023-03-27 08:57:15.000000000","tz":120},"subject":"Fix glance download with wrong checksum","message":"Fix glance download with wrong checksum\n\nThe glance download is broken with glance cache enabled and wrong\nchecksum because the save and reraise the exeption in glance cache.\n\nCloses-Bug: #2012898\nChange-Id: I5d398277b01a114cfb74af6458f8bee30a812cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4f06afa2d5e0aa2a2501d621755b55e254e9963"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4f06afa2d5e0aa2a2501d621755b55e254e9963"}]},"branch":"refs/heads/master"},"e90ed3941fb260e7b16e986b0dd62e2d6d2fc079":{"kind":"REWORK","_number":2,"created":"2023-03-27 09:38:47.000000000","uploader":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"ref":"refs/changes/33/878633/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/878633/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/878633/2"}}},"commit":{"parents":[{"commit":"b023562e1d31d3e60c195d701e6a9ff52e2fc956","subject":"Merge \"Add a check on startup for staging directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b023562e1d31d3e60c195d701e6a9ff52e2fc956"}]}],"author":{"name":"Pierre Libeau","email":"pierre.libeau@corp.ovh.com","date":"2023-03-27 08:48:28.000000000","tz":120},"committer":{"name":"Pierre Libeau","email":"pierre.libeau@corp.ovh.com","date":"2023-03-27 09:38:38.000000000","tz":120},"subject":"Fix glance download with wrong checksum","message":"Fix glance download with wrong checksum\n\nThe glance download is broken with glance cache enabled and wrong\nchecksum because the save and reraise the exeption in glance cache.\n\nCloses-Bug: #2012898\nChange-Id: I5d398277b01a114cfb74af6458f8bee30a812cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e90ed3941fb260e7b16e986b0dd62e2d6d2fc079"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e90ed3941fb260e7b16e986b0dd62e2d6d2fc079"}]},"branch":"refs/heads/master"},"5ea5a4f95dcf23b381d2c7fbfcac535ef8644c52":{"kind":"REWORK","_number":3,"created":"2023-03-27 12:05:20.000000000","uploader":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"},"ref":"refs/changes/33/878633/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/878633/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/878633/3"}}},"commit":{"parents":[{"commit":"b023562e1d31d3e60c195d701e6a9ff52e2fc956","subject":"Merge \"Add a check on startup for staging directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b023562e1d31d3e60c195d701e6a9ff52e2fc956"}]}],"author":{"name":"Pierre Libeau","email":"pierre.libeau@corp.ovh.com","date":"2023-03-27 08:48:28.000000000","tz":120},"committer":{"name":"Pierre Libeau","email":"pierre.libeau@corp.ovh.com","date":"2023-03-27 12:05:12.000000000","tz":120},"subject":"Fix glance download with wrong checksum","message":"Fix glance download with wrong checksum\n\nThe glance download is broken with glance cache enabled and wrong\nchecksum because the save and reraise the exeption in glance cache.\n\nCloses-Bug: #2012898\nChange-Id: I5d398277b01a114cfb74af6458f8bee30a812cc5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ea5a4f95dcf23b381d2c7fbfcac535ef8644c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ea5a4f95dcf23b381d2c7fbfcac535ef8644c52"}]},"branch":"refs/heads/master"},"5704ffe41db9519ff012d142017e3e5fa654d19f":{"kind":"REWORK","_number":4,"created":"2025-09-26 14:14:19.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/33/878633/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/878633/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/878633/4"}}},"commit":{"parents":[{"commit":"bcfd2cac74618f9cf115963488e80cd6d576a5b2","subject":"Merge \"Remove the last reference to OsloDBDeprecationWarning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bcfd2cac74618f9cf115963488e80cd6d576a5b2"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2025-09-26 14:14:06.000000000","tz":120},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2025-09-26 14:14:06.000000000","tz":120},"subject":"Fix glance download with wrong checksum","message":"Fix glance download with wrong checksum\n\nThe glance download is broken with glance cache enabled and wrong\nchecksum because the save and reraise the exeption in glance cache.\n\nCloses-Bug: #2012898\nChange-Id: I5d398277b01a114cfb74af6458f8bee30a812cc5\nSigned-off-by: Pierre-Samuel Le Stang \u003cpierre-samuel.le-stang@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5704ffe41db9519ff012d142017e3e5fa654d19f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5704ffe41db9519ff012d142017e3e5fa654d19f"}]},"branch":"refs/heads/master"},"582432b8232506a0aef5e2b563ccdf628f8efc84":{"kind":"REWORK","_number":5,"created":"2026-07-10 09:22:40.000000000","uploader":{"_account_id":29037,"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@ovhcloud.com","username":"pslestang"},"ref":"refs/changes/33/878633/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/33/878633/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/33/878633/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/33/878633/5"}}},"commit":{"parents":[{"commit":"bcfd2cac74618f9cf115963488e80cd6d576a5b2","subject":"Merge \"Remove the last reference to OsloDBDeprecationWarning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bcfd2cac74618f9cf115963488e80cd6d576a5b2"}]}],"author":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2025-09-26 14:14:06.000000000","tz":120},"committer":{"name":"Pierre-Samuel Le Stang","email":"pierre-samuel.le-stang@corp.ovh.com","date":"2026-07-10 09:12:35.000000000","tz":120},"subject":"image_cache: do not reraise GlanceException on checksum mismatch","message":"image_cache: do not reraise GlanceException on checksum mismatch\n\nWhen image cache is enabled and checksum verification fails while\ncaching an image, cache_tee_iter() saved and reraised the exception,\nwhich aborted the download for the user even though all the image\ndata had already been sent successfully.\n\nLog the checksum failure instead of reraising it, so the download\ncompletes for the user. The image is not kept in the cache.\n\nCloses-Bug: #2012898\nChange-Id: I5d398277b01a114cfb74af6458f8bee30a812cc5\nSigned-off-by: Pierre-Samuel Le Stang \u003cpierre-samuel.le-stang@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/582432b8232506a0aef5e2b563ccdf628f8efc84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/582432b8232506a0aef5e2b563ccdf628f8efc84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33607,"name":"Pierre Libeau","email":"pierre.libeau@ovhcloud.com","username":"plibeau"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
