)]}'
{"id":"openstack%2Fglance~880053","triplet_id":"openstack%2Fglance~master~I34707ecf7eb88dd0080707addeea852526f5beb0","project":"openstack/glance","branch":"master","topic":"bug/2007354","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-05-04 16:28:57.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I34707ecf7eb88dd0080707addeea852526f5beb0","subject":"Remove duplicate value in compute-host-capabilities.json","status":"MERGED","created":"2023-04-11 13:52:17.000000000","updated":"2023-05-04 20:51:56.000000000","submitted":"2023-05-04 20:48:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"880053-bug/2007354","meta_rev_id":"e4f537114469e0477afed093564e4fdc127ae3b8","_number":880053,"virtual_id_number":880053,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-04 20:48:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-05-04 13:08:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-05-04 14:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-05-04 14:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-05-04 13:08:56.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-11 14:05:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-11 15:28:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-13 05:23:00.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-05-04 13:07:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2023-05-04 13:08:56.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"d89c63bae6d17e05523ec79eba2fda35c3448f53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-04-11 13:52:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fa8bb4950c4f0050064895f770402bac9d60178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 14:05:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16176247cd3546a7b7a92bfd68cf0371\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/babcdf524e4547c4910c0083f1da9a55 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/321a7ea411ed43709ce8d0f687ce9802 : SUCCESS in 10m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2694d7bfc75b9ad56a6d85801dc5b4b61d7359ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-11 15:28:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4acb8c011d7f4f45867d7f18f1e0d411\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc49d0d837284b0a8d982ad4c80a255a : SUCCESS in 1h 11m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ee035b854214b31b79f890b2428866d : SUCCESS in 1h 30m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0caaf4ec22f415088612928577d2738 : SUCCESS in 1h 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b4d2f0b31d14492af82450ff72cc0fc : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19a795bce2734a10977266251e92def0 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/445595c9c82741769c1740774b291cc8 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cc53adf8e3b4415b20fbb64ff0fd034 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c6958771787d4ad88623c8b2a579a093 : TIMED_OUT in 31m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42ae9c186f0741d3b97827ee9a2ab6a3 : TIMED_OUT in 30m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/42408bbd4339410db83d0e4c17558d23 : FAILURE in 30m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e099aabb45944042ba06759131a16788 : SUCCESS in 1h 07m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4873682f2c15498faca76a2a2a8b1595 : SUCCESS in 1h 11m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/882c357483ce45ac9d0c5cd6a127c639 : SUCCESS in 1h 11m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/253cb2f86d894f0089799915957c07c1 : SUCCESS in 1h 17m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/528f743d1b2b4b83953dae906a17878c : FAILURE in 1h 21m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d93f1825c7c84ef99e9ca78ae2a454fb : SUCCESS in 51m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d42481f162384ae2a3df5fab99c7953a : SUCCESS in 1h 18m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e3884ac7e2048db964482568cf6f6d3 : SUCCESS in 32m 59s","accounts_in_message":[],"_revision_number":1},{"id":"9398a8069ef99923c1c7ae2d8b8f2318bdc212c1","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-04-12 21:10:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e506fb34dca84d09aa7903ba68f5291de4018a0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 21:17:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5eb0b4d354a4c5ea55de1fbb4801e65\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fefa31982a3403d97be3253e04c512e : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f9d8c652b6d421192194ebe92e0af59 : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e017d72e0f7124dec5b08c65fc3f00c03ad222a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 22:55:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/337468aeea27456bbf2cc1c907ff5469\n\n- grenade https://zuul.opendev.org/t/openstack/build/c0dc75b142ba48a391aa4a86b2119763 : SUCCESS in 1h 18m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/888b1b70e88648919ef245372e8946b0 : SUCCESS in 1h 21m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4e805ca6aa74c47b0a17fbbd15acaab : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d01c02c71ff94921917e749d719fd16d : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03f78d83429d414cbd30631fd02d225d : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bb56fcda3d84907947ec474c201520d : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e45c4bb244e4bab8905e7ef43a62537 : SUCCESS in 5m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bb6f5780055f4465b1cba0f883c88b1f : SUCCESS in 12m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/708663516e424fba8529cc66cffe2bc6 : SUCCESS in 8m 03s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f7d2d8777f54668bab59b78a9591e93 : SUCCESS in 8m 58s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/920916c1b6434731af7aed7c6be9e487 : SUCCESS in 1h 43m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fa1a91740fd344768c1ba60a7a296555 : FAILURE in 58m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/afe4d691ff3244fc8e2e3672f49b723f : SUCCESS in 1h 23m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88ba722babe74c148b1100f30e05989f : SUCCESS in 1h 00m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/04bfe86300d847b49227ad7766ecef71 : SUCCESS in 1h 39m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f17958372e444cac9d4e8c976d61e747 : SUCCESS in 1h 04m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d193272403fc4262bca1091efbb953c7 : SUCCESS in 1h 01m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb528a4e11b949089f26345812743485 : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"b1e440a6a2717e56c666e4a9bb6751e544783e09","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-04-13 05:23:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0cea7ca5bdaa86843869a1c2451933f6ea695d6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 05:29:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/431554bd32304b36adb501253ab7a41d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/406ec4b34dbc4c66946a70d699df55c7 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/603b3475600045d48201a74e15b58894 : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"727f36d8fb2d0c7a1d56f041d04713c05a89c3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 07:26:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af6cd820a99b4d40924ca9b796050f30\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f69ec11196542d58e2eda5b4a015e80 : SUCCESS in 53m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/da54e4e1d179449ca3cb154ec0a01d1e : SUCCESS in 1h 27m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99b0db2a72b945a0a9d73e4b549c2952 : SUCCESS in 57m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40de87d3d864446ba36eab29e9743ad1 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38b3fcb2988044a393d6fc9f7e9a3f58 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5ec462a5d90432e9e5e62e8e61dfd04 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40ba1d2c3ca4243969ffde818d56314 : SUCCESS in 6m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2509b41039e34a91995e6980f0bacd0e : SUCCESS in 12m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4500a4cbdb740f5b7cc2fad4ef0d9a9 : TIMED_OUT in 30m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f09763d86dd7475a9eb0b6016477db1d : SUCCESS in 10m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0982100c26d4451a8cb25708ee953a70 : SUCCESS in 1h 47m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cb75f4c2a6bb4d69a494008f4b1e4587 : FAILURE in 1h 52m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f324d4475d4485bb4682382ace02bbd : SUCCESS in 1h 35m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ea5e24999c134796ab38d63b5b57db8c : SUCCESS in 1h 23m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/04bf0a9d8f814f46accdb8e471dd61fc : SUCCESS in 2h 02m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04ed87ed5719455ea9870d5ee9aa8a86 : SUCCESS in 54m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a2f5f76dcd7f43e8b1d6f5d9758f0107 : SUCCESS in 1h 35m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1a7ba8f842f340cd8cb3d2ea62bceec1 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e1787317342d5b597b3abc3faa8ad919e3395820","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-04-13 07:35:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cbce420563f763b44de69bfcb10f5658c868af0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 07:42:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/969dfef1ad42408c8f409731de33b82a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbdadc277ce24c0aac13050773c9882d : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec6c331e29794aa887f36da9be5dc129 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c5f842018eaa2ae4751034ca7dbc033c1bfc9879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 09:18:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3593a5a4d150465d94f42bcd6f325d2f\n\n- grenade https://zuul.opendev.org/t/openstack/build/11c9c610e806441191b497c21aa3ac1c : SUCCESS in 42m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a08ebc530d8405cad65fa206a327dc5 : SUCCESS in 1h 14m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/232b5783f6844346b1e693d9afc7365b : SUCCESS in 47m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1040da51e43347d3803a6870f35200ac : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/218914d4f3c84decbc32f1bc2e951729 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23840fb5d97743f9ba045d7e204b1851 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4345d61102394509a84f1657ac772be3 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9f7a87e434f34b7da98f76ef1f5d16e0 : SUCCESS in 12m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/831183ab772a4459995fd46ed029a6ae : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2e95288b158f4516be5bd9d255f1ceb7 : SUCCESS in 8m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/039c2c652c8943e2be2595a1ce0baf2c : SUCCESS in 1h 19m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/79d3812a8b434f8d91fcbb2e659fb6e5 : FAILURE in 1h 12m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18a84b4346654fbfbbfdd17bec695f47 : FAILURE in 1h 16m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c75c4ed197204b6aa43ab8f07a48c3f7 : SUCCESS in 1h 13m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f22a28b744234e2f840aa8d85e0c861f : SUCCESS in 1h 36m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f74c9919a6274d15ba5dcd3e74ee1461 : SUCCESS in 42m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/724f6aa885f4404aa2c70788318fefa8 : FAILURE in 1h 15m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c20ad8137044d629d15f1c584c173ff : SUCCESS in 27m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fa46cfbe577ec7941ec956412e0d4e200ebde529","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-04-13 14:12:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"70120780656ef9400a34f7c91cbc93a6b2af4be4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 14:29:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4abe98506d984498809fa81f00a82ecd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55f450f32a4543c1afecb74236390d52 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e70be36c578447588f9e17367f0a627c : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ce88491edd90cf6025ffbad96e2f1cc7dfa572c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 16:26:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b125e3e49ab24e1baebceb300856c6c4\n\n- grenade https://zuul.opendev.org/t/openstack/build/37b7f6752da840ecac45928eecb75800 : SUCCESS in 56m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/88517392931c4986a386ba9055b170b5 : SUCCESS in 1h 30m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7967c8201d964c008aaff28f9eeca7ad : SUCCESS in 54m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/482bfecc847a4b40af2599e38314686a : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5d74de7bb2c43e5b8ea80f7cb42c41f : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2921bce73ca94752a40330a7ddddb7ad : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb509de009f747b7a10219355274737e : SUCCESS in 6m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3a770121ee96431096b4ef76203428c7 : SUCCESS in 13m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f09180f52f84f35a4ac88449992dbac : SUCCESS in 10m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/14ae9873a75248c3b2cda7abf20b3f55 : SUCCESS in 8m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9c2d8be4b77e4e7ba7d3378528ca26ab : SUCCESS in 1h 22m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/00f456a7c23041e2aac7540d518c591c : SUCCESS in 1h 31m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/de8aee5650db4a74a53a8f58829c5d2d : SUCCESS in 1h 21m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4252c93360354ed089315fb73fb1cb95 : SUCCESS in 1h 23m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/21f2dcdf4b16459f9c1e6ec176b2bf97 : FAILURE in 1h 30m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50928fbd37424f30afee9f3bb634eab5 : SUCCESS in 37m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb332e596a7b470d80122e8b55a4f689 : SUCCESS in 55m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60d8a9a93efd4886b35ab64681da9c7f : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":1},{"id":"10d739a6cb4b6fcc095be479013301b90f9260f3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-04-17 13:06:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5243579d10979b86b9f531b8cc56d3b9d9fbe25f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:12:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbf451c75ee84fd0924c34be1384a2b6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66be215beed34c809ee9f1846245e322 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b921a15111bc4e61b228c71a9c18b931 : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b3cd932da8190e72dc7a56749e0d8bdaf17cf34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:03:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f301c0c12cdd4a78ada4e8500cb14125\n\n- grenade https://zuul.opendev.org/t/openstack/build/57c10a498ce049549ccdd5d9e5bec234 : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7b6ee2b86b54ec088514a1c51a2c855 : FAILURE in 1h 18m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38f0d0a1f2754c54b6361e17270fa3b5 : SUCCESS in 59m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f70dd2ac46a42a9beddc3bcef540738 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea2642943ed0455e802cbd6f55cdf0d9 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed6a3543f40048bba770688877384057 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39390495208a4010b34e32a522fbea60 : SUCCESS in 6m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a525af552bfa4fe2970de99b434f0310 : SUCCESS in 10m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5127419a894f4cd49019f8b78f5a0c67 : SUCCESS in 8m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f5cb2401a4a54761b5fc86e5357ae63b : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f3b555091f7844d495432173f10c0d08 : SUCCESS in 1h 23m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/567cea62f09e4edab569b364d139c485 : SUCCESS in 1h 33m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f619b32bafb43f29a32a47c96b51f93 : SUCCESS in 1h 35m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f905347586404e95bdd6d76f1fd605a7 : SUCCESS in 1h 48m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2add5d2c039146cdbbbcf51dfe9740ec : FAILURE in 1h 08m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1aabd0ddbca4452a9c9072523962bf5 : SUCCESS in 57m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ba8f8eef876b46c5a55d57d5c261b877 : SUCCESS in 1h 31m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6f6111eb32e4a2c9e2871b3367751ca : SUCCESS in 49m 25s","accounts_in_message":[],"_revision_number":1},{"id":"35d5091bcafdfc6d2c7caa8c89756ae75f0c479c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-05-03 14:43:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a8705dfe08d3150ab0e8e49e575a7c44b4a0e1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 14:49:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f659ca05fb48a0a5c082f71e9fe249\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c2276c03d0f47ef9d56b35c44df6718 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa07cb2a8af54c5cb9bcdf38f143b460 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5bda07e552105cd6f70f1e2b8b2b3b775b0e85a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-03 16:23:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/751ae2526ffc423b84b600dae705ad42\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d95a38b609a4f35b401dd2bfc1ab057 : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7710390190cb41cc9968d2641f2cf0b7 : SUCCESS in 1h 23m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fceed77e282b412ebfb5bf837808b023 : SUCCESS in 1h 07m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41f400cca9a242939faeeb3004ce7af2 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b53c058c7594a809dee3a8164859ae8 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdf6d9dfd17c47b08cf07bdd610635b5 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e52d33e7d30487489cc18e57695b61f : SUCCESS in 7m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c241b533049e4447a75c8da1e43f97ed : SUCCESS in 11m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4c6a5b8641bd44acb71f5178da00029d : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d4a4774ccfd4e93b65583f3584c93ed : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e42a09b8b30a4d3babf04278d030da46 : SUCCESS in 7m 27s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05931c8755d0406abff5be452f51de77 : FAILURE in 1h 29m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97959f530974406eabd0fc73016d71d6 : SUCCESS in 1h 21m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/10e13c965af84d249eb8e0bf790d1288 : FAILURE in 1h 19m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/34296afa76be4fbb80226875af61e120 : SUCCESS in 1h 31m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d866327089bd494f8e697eab1079efe6 : FAILURE in 1h 21m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfbe924e5d284f95987f62deaaad1381 : SUCCESS in 46m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ab63b0c5a394303ba0eeabced681fa1 : SUCCESS in 1h 34m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b1a8eddc2564381834c23a4954b9fed : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":1},{"id":"fae306ff4578466292b4c4581f2604911056f0e6","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-04 13:07:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2e5ec0ac9d20c0a4d72cc18987801cb3170de0d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-05-04 13:08:56.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"703e0ce9d0f435af41936cc68c0d7a4551314dcd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 13:19:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e39096733468488b95c9900234ebdf43\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dbaedfa693814f58b5b2eaf3ff764469 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48f6f9ba7f0443359ff563955363757a : SUCCESS in 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"830874b89a4a9e1c5cdc9679549dbf5d742a05db","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-05-04 14:20:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed304bf1521492bb54a003e7ecbba335cfc056d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 14:30:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/509e4c1147ee4e60be1c48c03ec7dbdf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f17c11fd5ca040faae64b44cd8918442 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2ef8b28de7054b0d886196cee872fd5c : NODE_FAILURE Node request 300-0021135578 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4fe6fbd20129583692cbf8865969da579d4fa23d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 15:57:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d7bbccda0704464853bfc16e762b9d0\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f93dfa51abd42618c8654c61181618e : SUCCESS in 1h 01m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e457b82c2cfd40aa8d9dd95c48366fde : SUCCESS in 1h 39m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb1f38bf6ae6453a86d89b986ac781e8 : SUCCESS in 1h 04m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03b9362396e94782a9de9186849928b2 : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b332b3566254459d84e2adb8ac55f15f : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3519c60be62e4f4eb7d4f4f247b43b8f : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e8f14055fc41faba46dfd9779307e8 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/56e4d94fb6694a21aed2b4cf9c2eabc1 : SUCCESS in 14m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/283dcaaab20a4338ab41e5a5015205bd : SUCCESS in 9m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/259b6720ca4144178b4d4f3aad635a40 : SUCCESS in 18m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7d94f9d6c1df4db4addc6f46f647417c : FAILURE in 9m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57f903572e9046bdb5f1a5131d5341db : FAILURE in 1h 44m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a1df09b1accd443db9f44cc96ac9e6a1 : SUCCESS in 1h 36m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ecb5edd64cc24a4b8b7914a4e886d86c : TIMED_OUT in 2h 04m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b4487ef171014ef4a10bf9c15947b134 : TIMED_OUT in 2h 04m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d9611df6095e4658af7f52ee7ee0e702 : SUCCESS in 1h 35m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05c24b5800254439888c88e38dc53196 : SUCCESS in 49m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d72d621d0ce848708990fbf450192661 : TIMED_OUT in 2h 03m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d27d77e77c6a4f82b48c8a749ac4185b : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":1},{"id":"65a2774b70e8843f4def1ddac191c94266a5dfda","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-05-04 16:28:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"32f610ea4185066d91401d568bb1d2ab4bd565b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 16:45:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd1e2287f784d439f88b0fc68f9337c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/81046c2302e440fcab730bf76dde4db6 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/997fd8bc4db04355a7df263dd626f3f1 : SUCCESS in 13m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65c8c3e2d85b6ba729c2eef9aa81a6615bd47445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 19:09:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39c7607b43894924997a27d7f1b2baa5\n\n- grenade https://zuul.opendev.org/t/openstack/build/3344cdebeb6d498ea1eb30d5251784bb : SUCCESS in 1h 06m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b294ab36b53043e6807c8695324dc268 : SUCCESS in 1h 16m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/536b749480584e2790d51d2b2182a0b0 : SUCCESS in 50m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808bab613c1a453db5ccdd9efcc19133 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ec1c371709c439680e84ff9b75ca09c : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0822b8145f34690a4b662c2857ef34e : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a18ceb737b3434dbd87323a100c7287 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5ca63f6f9029450a940a33f3634da1b7 : SUCCESS in 13m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/99670bc27a9547e3878305c945d0b921 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/438fe2054cb142e1a0ab470e6de5fae9 : SUCCESS in 9m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0085f14c92c94039beee2eb3f959f625 : SUCCESS in 8m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0768cf43846544daa6531f5e3db06467 : FAILURE in 1h 18m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/98207744635a465584085b6da7b1551c : SUCCESS in 1h 32m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/05106476f91743f7a84b60424851d899 : SUCCESS in 1h 35m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1525b523d707431981716401619c1700 : SUCCESS in 1h 12m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/10275d9788e840038a9c863951fe5e7c : SUCCESS in 1h 39m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8c714f7167e405583f6e4b991179dea : SUCCESS in 1h 04m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1df4ec42c0aa4a61bc2a794d0da79804 : SUCCESS in 1h 30m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dc9acb4fea0f45a69425728cc63beceb : SUCCESS in 30m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0357fbc0a8c1b1e97590c219ca9f7fd8b60deb9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 19:11:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ff8ef7fd00f12676e225927230958f8a00593c7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 20:48:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82b2191961f14427be23fc62ec002c53\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f4cca8bffba48c0a058245244ddbe7e : SUCCESS in 1h 01m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/75e39aa8f9d147fea8d600a2926367ad : SUCCESS in 1h 33m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60451fe30c4c42e09d790fc57d237cac : SUCCESS in 58m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2ae853b6aa4323ac6bffd59a57107c : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/603a015ed20a4ce794b7900b01a706ca : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b914636754b94f0cb70a9b7e7370ae97 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b6482b525624de18dec76788fc1bf42 : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95a1a70dc2954cfd84dd6d3064a7731b : SUCCESS in 9m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ba3fc66c84dc4a688277fcf2a700a964 : SUCCESS in 9m 58s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c14b2470d3d84aae8f478c342d5756f1 : SUCCESS in 1h 23m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8e8a39fe69c64f76a150eae8d18595a7 : SUCCESS in 1h 11m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a223c286683f4cceb53bf08c121c423a : SUCCESS in 1h 18m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3ee9f9b6e2d4175a0b1d7f52aa725d0 : SUCCESS in 36m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cc103aff4d1941f5a701e4e7b9f45fb9 : SUCCESS in 1h 04m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d828bb2cc5940b39b1a71dd205bfcdf : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":1},{"id":"118a0a6594152acaa01b2afad1b3fd5e2da880d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 20:48:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4f537114469e0477afed093564e4fdc127ae3b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 20:51:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d95fde628274b918571ac883512a5a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2edc614d88644919556f643f77c8fb7 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"babf8f19a5ea9298119714a99733df5ac290f92f","revisions":{"babf8f19a5ea9298119714a99733df5ac290f92f":{"kind":"REWORK","_number":1,"created":"2023-04-11 13:52:17.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/53/880053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/880053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/880053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/880053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/880053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/880053/1"}}},"commit":{"parents":[{"commit":"cb4ac8d37deeadaa43e74125fa8950cd331d4b3e","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb4ac8d37deeadaa43e74125fa8950cd331d4b3e"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-04-11 13:51:44.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-04-11 13:51:44.000000000","tz":120},"subject":"Remove duplicate value in compute-host-capabilities.json","message":"Remove duplicate value in compute-host-capabilities.json\n\nChange-Id: I34707ecf7eb88dd0080707addeea852526f5beb0\nCloses-Bug: #2007354\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/babf8f19a5ea9298119714a99733df5ac290f92f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/babf8f19a5ea9298119714a99733df5ac290f92f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
