)]}'
{"id":"openstack%2Fglance~880219","triplet_id":"openstack%2Fglance~master~I1e557e731cceb3cfbc4ebc3f0887adf372a91d1d","project":"openstack/glance","branch":"master","topic":"fix-gate-paste-deploy","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-04-12 20:52:23.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-04-12 20:52:23.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-04-12 18:55:45.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-04-12 20:52:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1e557e731cceb3cfbc4ebc3f0887adf372a91d1d","subject":"Fix functional test failures with PasteDeploy 3.x","status":"MERGED","created":"2023-04-12 18:42:41.000000000","updated":"2023-04-12 20:53:43.000000000","submitted":"2023-04-12 20:52:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880219-fix-gate-paste-deploy","meta_rev_id":"8238112bec483b2e6c5f1aeb36d7e370929e9136","_number":880219,"virtual_id_number":880219,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 20:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-04-12 18:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-04-12 18:55:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-04-12 18:55:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-04-12 18:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-12 18:44:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-04-12 18:44:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-04-12 18:44:38.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-04-12 18:49:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-12 19:41:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"193b88676deda3b8341babc3bd88eb86494d3fea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-04-12 18:42:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fc49cc335c1e23498712eb3e9daa9602600687ff","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-04-12 18:44:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"24202063a277ad970c458a0aeebdad76a5dd46ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-04-12 18:44:53.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a93638998c1c29ba9d6347ba57ee7190d68d854","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 18:49:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47f6192ca7e8458eac6bf915a1a9b2cb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6b52ac386a542e99c1e983c1a4f27d2 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cd44ffa437544e43bbfbd25ecf98569e : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07a6a251f9d5597091a95a0eeed7ab4b9f04ae65","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-04-12 18:55:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c5f6cdd23496b1c4805bf9d4492b4dbb684ae9d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:41:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d657772d7a5f45f7833865a020b8b869\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dde557db9aed4c44911ba2d339c08bd8 : SUCCESS in 4m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/183eda71c0aa41eebacabb29e681688c : SUCCESS in 57m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9516ff6f908648969691b556a3933351 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b9fc3be35a6410b917d90b1c12efea0 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f3b5d1c59464642b0149d6bde20288a : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cbbb5d384c84aebbee1685b27413f3b : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88c96b16e37b48c4bff1e52fc7be5f2a : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/403bcf7695dc424e893b7c36c766eda4 : SUCCESS in 14m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fcabf1855a80403a9985d862d015c543 : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dde3b3fa5dab4f3f88531a2a3d55ad99 : SUCCESS in 8m 31s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0cb92f29bec740dd92e26f498ffa6e6b : SUCCESS in 25m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8033d67b6fa47c6e9e68794ec1e841b9fc4c849c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 19:42:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12d75c5ff40d2d93780a17731c9641040283c7b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 20:52:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85a4fb974fd14e5ebd071180cc6c357d\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2a207c9ed88f4bd89500408411cfe35a : SUCCESS in 4m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57ca7e5a625343b693b9904556585d07 : SUCCESS in 1h 03m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d876f8aa5d764614b05aa3afad1dbb44 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/871a8b833ce845e5b60ff97a2343415d : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b3404c6688d4d78bc9019915a19c406 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447f446151bf4da0bf51df669b0a0192 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3aff6ece3e3f4196a3d945733063daaf : SUCCESS in 6m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98c984996839427fadf61484983a250b : SUCCESS in 9m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1939b7c3c2f44bc48ca93799c995cebd : SUCCESS in 7m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/54b73951868a4f0da610c0f97471f419 : SUCCESS in 43m 50s","accounts_in_message":[],"_revision_number":1},{"id":"bc89105afdb1d5b8e72cd371593254ecabd171ef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 20:52:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8238112bec483b2e6c5f1aeb36d7e370929e9136","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 20:53:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06d458f8bac44213be78efbef8d02b24\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/459e758bc28e43e48630743add3dd804 : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbe0dd558876472c86cfcffd9548fb0d : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f76db1de5717474fb88b7387b0b5b4bb : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41da684fe41e48eca934aca1f19fbb779be07fc4","revisions":{"41da684fe41e48eca934aca1f19fbb779be07fc4":{"kind":"REWORK","_number":1,"created":"2023-04-12 18:42:41.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/19/880219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/880219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/880219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/880219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/880219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/880219/1"}}},"commit":{"parents":[{"commit":"cb4ac8d37deeadaa43e74125fa8950cd331d4b3e","subject":"Merge \"Update master for stable/2023.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb4ac8d37deeadaa43e74125fa8950cd331d4b3e"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-04-12 18:31:05.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-04-12 18:37:47.000000000","tz":120},"subject":"Fix functional test failures with PasteDeploy 3.x","message":"Fix functional test failures with PasteDeploy 3.x\n\nPasteDeploy replaced pkg_resources with importlib_metadata[1] in 3.0.\nThis causes functional test failures because the line breaks in the\nexample configs from glance/tests/functional/__init__.py are read as a\n\"prefix\" by PasteDeploy. For instance,\n\npaste.filter_factory \u003d\n  glance.api.middleware.version_negotiation:VersionNegotiationFilter.factory\n\nis read as \"\\nglance.api.middleware.version_negotiation:VersionNegotiationFilter.factory\".\n\nThis might be a bug in PasteDeploy or (quite unlikely) in\nconfigparser.ConfigParser rather than in our code, so this issue should\nbe revisited at some point. For now, remove the cosmetic line breaks in\norder to fix our gates.\n\n[1] https://github.com/Pylons/pastedeploy/commit/ef58c20961072e4f60c26e94564180c304af7144\n\nChange-Id: I1e557e731cceb3cfbc4ebc3f0887adf372a91d1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41da684fe41e48eca934aca1f19fbb779be07fc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41da684fe41e48eca934aca1f19fbb779be07fc4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
