)]}'
{"id":"openstack%2Fglance~880685","triplet_id":"openstack%2Fglance~master~Ibc21b107878f5ba50137da4082a7cbc6342d2aa9","project":"openstack/glance","branch":"master","topic":"2023.2-functests","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-04-20 16:02:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc21b107878f5ba50137da4082a7cbc6342d2aa9","subject":"Update functional jobs for 2023.2","status":"MERGED","created":"2023-04-17 22:08:56.000000000","updated":"2023-04-20 16:06:18.000000000","submitted":"2023-04-20 16:02:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880685-2023.2-functests","meta_rev_id":"7627cb01873c6d8a3d2bfca18493a31191fe40ed","_number":880685,"virtual_id_number":880685,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 16:02:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2023-04-20 14:31:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2023-04-20 14:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2023-04-20 14:31:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-17 22:14:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 23:21:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 14:31:04.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-04-20 14:31:23.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"b8c271ae2f4a0193aa1256f87cc382a16acb76c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-04-17 22:08:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7542d18f4375ee4b513731aa68b6e55eb4736056","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 22:14:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3130800265584dbb8d0455134f8118fd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/021567b9c07d4c18a22f4426fcbb28e3 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/98806abb3f3845a2be2da39d1378bdc7 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c1c89cf2fe31e7f94105c5050459c7582b600716","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 23:21:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f45eca9a23454fdd8c9b563bfce3077e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d5003505405e41b5b6e8f2a5117961c0 : SUCCESS in 4m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1faa3b2222249feb72047d390bfff56 : SUCCESS in 1h 06m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352ad8f5d0f24faaa8b83698c0ee9fe7 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87d963de311143898f1643129f437f8c : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3599f54b7b224991aec3f3a68da331e7 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7175d410cbb141df928ca678dd8a090b : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a6ab21ac5ad41df8b02f0af9a441a00 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0727eada8af440b39712a70254d06e08 : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3478f13a497f494ba0b4ccb193abc80c : SUCCESS in 9m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/61d582cd94414508975f1a4b42bf55fc : SUCCESS in 7m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5b201cc3c6b94c63a5294fa74dc2340c : SUCCESS in 8m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c91d68b2b00456e871c6d5ff0f78395 : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e29b9572f7183075e5e3cdc7eb7cff063eab3b5a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-04-20 14:31:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e125edf6f26a0ec4f9ef17172d76e02282d4f15b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-04-20 14:31:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3c2f07483cbf6102d543378690f17481e8093fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 14:32:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cfbde6a906a37290bb3dd8a14f1ab4d8fe373c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:02:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ef6e79ea9334e7fbc1fb27501466322\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8bb331e80f6a43bd9300601d3fd0eca8 : SUCCESS in 5m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b2ce56099c94bd3a34d0ab5f0773e8a : SUCCESS in 52m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b39cfdf326314d05ba06890b42699452 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb42e639427a4a90a07add4c1a0df15f : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/961a3c629ba84b799d09f90ab0bcab88 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ae1a42c18246c598ac17653c954b11 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9e0f20af8f24632a230f36ded469211 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fdae545aa5046a18c7016a7601df05e : SUCCESS in 9m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6cebade3c36a45cdb9b943e9abc9957b : SUCCESS in 9m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/640160cbd0f341a095251f5e16eca772 : SUCCESS in 28m 59s","accounts_in_message":[],"_revision_number":1},{"id":"305eb5bb815141eb04c09784d6e25483f944d2ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:02:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7627cb01873c6d8a3d2bfca18493a31191fe40ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 16:06:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07c573f82ff241a3aaebd5039abd69d3\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/914306f5333a453a9d027506d8ea54c6 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9923e2417699410bba45fc6fee467403 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71c95de3da1e473f8d180c08120aaee9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"084c8a32f5abddc6ad6ebfeb804cefc504928b45","revisions":{"084c8a32f5abddc6ad6ebfeb804cefc504928b45":{"kind":"REWORK","_number":1,"created":"2023-04-17 22:08:56.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/85/880685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/85/880685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/880685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/880685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/85/880685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/85/880685/1"}}},"commit":{"parents":[{"commit":"0b822f1fab40c84e790f753318964d4a5748a8b6","subject":"Merge \"Fix functional test failures with PasteDeploy 3.x\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b822f1fab40c84e790f753318964d4a5748a8b6"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-04-17 22:00:24.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-04-17 22:00:24.000000000","tz":-240},"subject":"Update functional jobs for 2023.2","message":"Update functional jobs for 2023.2\n\nBased on the python runtimes for 2023.2 [0], we should be running\nfunctional jobs on python 3.9 and python 3.10.  This patch adds\nfunctional testenvs for python 3.10 and 3.11 (the latter for local\ntesting) to tox.ini so the appropriate zuul jobs can be defined.\n\nThe functional-py38 testenv is not removed from tox.ini, as it may\nbe useful locally.\n\nFunctional py39 and py310 jobs are added to the check and the gate.\n\nThe current openstack-functional-py38-fips job is not removed; a\npy39 job has not yet been defined to replace it [1].\n\n[0] https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n[1] https://opendev.org/openstack/openstack-zuul-jobs/src/commit/b3cad4f7a36929c13d9c74f2de24789741aa5d85/zuul.d/jobs.yaml\n\nChange-Id: Ibc21b107878f5ba50137da4082a7cbc6342d2aa9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/084c8a32f5abddc6ad6ebfeb804cefc504928b45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/084c8a32f5abddc6ad6ebfeb804cefc504928b45"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
