)]}'
{"id":"openstack%2Fglance~881756","triplet_id":"openstack%2Fglance~master~I5c0651cb5dfe00afab9ee61d614227055988b930","project":"openstack/glance","branch":"master","topic":"bp/store-location-strategy","hashtags":[],"change_id":"I5c0651cb5dfe00afab9ee61d614227055988b930","subject":"PoC: Add new location strategy based on store identifier","status":"ABANDONED","created":"2023-04-27 18:00:31.000000000","updated":"2024-03-01 06:00:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"23fd66e1801cbba2fc927658c201a053eaef3932","_number":881756,"virtual_id_number":881756,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-27 19:39:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 18:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-27 19:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-22 18:27:52.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"2db763ba65f976c3b3d0d7961a03ecfd1bc2668b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-04-27 18:00:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4d45dcff74af08a27ac424bfd1f530f748ac2b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 18:11:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90b34e16593a4d08b7dea8547a01d9bc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5e126b22b054a318d49ec5bf00752c0 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a11710dde794d8fbaef67f41d5f55fe : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdbdee01a32adbbea176fe4dd40fa4778de3c071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 19:39:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddcb35bffe9d461a875848f4d7d410c7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f2a865daf7a342fa8f7516db6ddf7d4a : SUCCESS in 1m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/91ceecb66b354142ae5868ed84e29378 : SUCCESS in 51m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ec15b08035e499aacbfaa279df073f9 : SUCCESS in 1h 37m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23c81d9aa03f4b769f2153c578a7b7b4 : SUCCESS in 1h 00m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eccd22d503be4ff38303961402f9b281 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cda0b16e8d4457bac1d461121a9903f : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3040ec47075e46dcbb7b4bf94deded14 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0c50786ed8d4f06b46faef00fa6fd0f : SUCCESS in 6m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7d50f93967cd445bb6f210e443ce0df8 : SUCCESS in 13m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b00b8cb981724062b6c87f48e68fe2df : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52e3e210ef374b5fad96727b1d9beaf2 : SUCCESS in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b94cdf3fec7743148f83b8ddbfbbe800 : SUCCESS in 8m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca6cea392d854b8c93c37c6f1ccacd79 : FAILURE in 1h 17m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dbb5d5daee054c6e9c50547ead67acba : SUCCESS in 1h 19m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/19b1e08d0d284b51b36341b64c3477ec : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b2d87f3832ea4f42a84ab78838b85d2b : SUCCESS in 1h 04m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5eda4b120dab432493b10b780bbd53eb : SUCCESS in 1h 18m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7aab30b610714c3ab238e83d2809eb30 : SUCCESS in 42m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5409eac0717a4af5b2de9c2dccb8debc : SUCCESS in 1h 16m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ebbda5a57ddd4368b7f2e4669813c05e : SUCCESS in 38m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a01d11d275349c1278770f1095db081a1b70e5e2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2023-11-22 18:27:52.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- ext-glance-zuul : NOT_REGISTERED","accounts_in_message":[],"_revision_number":1},{"id":"23fd66e1801cbba2fc927658c201a053eaef3932","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-01 06:00:52.000000000","message":"Abandoned\n\nThis is not required now!","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b92b689797e58f7c995f18fc72d05e617992985","revisions":{"3b92b689797e58f7c995f18fc72d05e617992985":{"kind":"REWORK","_number":1,"created":"2023-04-27 18:00:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/56/881756/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/881756/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/881756/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/881756/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/881756/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/881756/1"}}},"commit":{"parents":[{"commit":"084c8a32f5abddc6ad6ebfeb804cefc504928b45","subject":"Update functional jobs for 2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/084c8a32f5abddc6ad6ebfeb804cefc504928b45"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-04-27 17:55:27.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-04-27 17:55:27.000000000","tz":0},"subject":"PoC: Add new location strategy based on store identifier","message":"PoC: Add new location strategy based on store identifier\n\nTest result - https://paste.opendev.org/show/bISLcM5wfwsDFldPftsN/\n\nChange-Id: I5c0651cb5dfe00afab9ee61d614227055988b930\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b92b689797e58f7c995f18fc72d05e617992985"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b92b689797e58f7c995f18fc72d05e617992985"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
