)]}'
{"id":"openstack%2Fglance~882302","triplet_id":"openstack%2Fglance~stable%2Fxena~Ie9a4f289aa1e7c2db1370804ec91e68444051965","project":"openstack/glance","branch":"stable/xena","topic":"glance-tempest-plugin-branchless","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-05-09 17:38:30.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie9a4f289aa1e7c2db1370804ec91e68444051965","subject":"DNM: testing gate","status":"ABANDONED","created":"2023-05-04 20:54:06.000000000","updated":"2023-05-09 17:38:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":882360,"cherry_pick_of_patch_set":1,"meta_rev_id":"49da224f97a6407a4ff6c33174941dc539a37d45","_number":882302,"virtual_id_number":882302,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-06 04:18:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 21:33:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3732b98a04e04df447b36a80f171cf3b2652c261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-04 20:54:06.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"2d0a366ce07b6a52b2354944c75c0b0fedfd098b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 21:33:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/14d6f30a4c084cca87b366b7890ed3e5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a135f6384944b38ffc2cf9b0f7a524 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2cc3547cad44dc6bd1b739e16f55528 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b19a30b3311f4201baf5898821ee7573 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/835db6829a78430e9f4dc4ac35c8a525 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/489852785f60469680c3d0c02cbca333 : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c535fd417ad040948f626a752dd4ecd4 : TIMED_OUT in 31m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f70c0fc8accb4d55bdab73c83a7f2bd6 : SUCCESS in 8m 38s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7f4881d3a14143e099ad20b40fa49216 : SUCCESS in 8m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ff3ba5b674f420b80576628b34440bc : FAILURE in 30m 43s","accounts_in_message":[],"_revision_number":1},{"id":"b91cf7772aecee1ed57bb0e75d722911623329a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-06 03:48:25.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b1b2f49c2abf3e56f79048649f3da53ac3a60b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-06 03:51:32.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"36afabb84ce1f0face8c8268a041363c9d42900d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-06 04:18:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8675543d920a4d1faeda2954835d049e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19c8aae78df45aea8e987a0306ebc5c : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7347af8e186e41db9f76b2382e4d6f76 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d464d7d06ff48c8ae1a1e07a5d66c36 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/790a9503e4d24e358d6c501ddd2b13cf : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4273bd0a30504ee5a103fc2bc9c0eee3 : SUCCESS in 6m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bb16f4eff9714ffabd4ba424851788fb : SUCCESS in 8m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b8a920696049464a92d847b8b9fdb18a : SUCCESS in 7m 49s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/4d026e4e1db74dd2a0e6a213d34ee4d1 : SUCCESS in 8m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d4f6715757aa4d449e79db3ddaa139d5 : SUCCESS in 24m 48s","accounts_in_message":[],"_revision_number":3},{"id":"dc71149db3f28d31df633980e7ca3cb4a05295ba","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-06 04:29:07.000000000","message":"Topic set to glance-tempest-plugin-branchless","accounts_in_message":[],"_revision_number":3},{"id":"49da224f97a6407a4ff6c33174941dc539a37d45","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-09 17:38:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7bbfdf010b0a3e14a53eac3e38d8f5192251dfb7","revisions":{"7841a6e3d3b162b3c28bbeb91424e931c6d90f22":{"kind":"REWORK","_number":1,"created":"2023-05-04 20:54:06.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/882302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/882302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/882302/1"}}},"commit":{"parents":[{"commit":"4fc2ea4a85bbca25fae84a7506bf78f08d6485bf","subject":"Merge \"Enforce image safety during image_conversion\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fc2ea4a85bbca25fae84a7506bf78f08d6485bf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-05-04 20:52:52.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-04 20:54:06.000000000","tz":0},"subject":"DNM: testing gate","message":"DNM: testing gate\n\nChange-Id: Ie9a4f289aa1e7c2db1370804ec91e68444051965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7841a6e3d3b162b3c28bbeb91424e931c6d90f22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7841a6e3d3b162b3c28bbeb91424e931c6d90f22"}]},"branch":"refs/heads/stable/xena"},"4a73a38cc64ceb0ebc2ff1eb367f654fc8e782c7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-06 03:48:25.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/882302/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/882302/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/882302/2"}}},"commit":{"parents":[{"commit":"4fc2ea4a85bbca25fae84a7506bf78f08d6485bf","subject":"Merge \"Enforce image safety during image_conversion\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fc2ea4a85bbca25fae84a7506bf78f08d6485bf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-05-04 20:52:52.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-06 03:48:25.000000000","tz":0},"subject":"DNM: testing gate","message":"DNM: testing gate\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/882366\n\nChange-Id: Ie9a4f289aa1e7c2db1370804ec91e68444051965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4a73a38cc64ceb0ebc2ff1eb367f654fc8e782c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4a73a38cc64ceb0ebc2ff1eb367f654fc8e782c7"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"},"7bbfdf010b0a3e14a53eac3e38d8f5192251dfb7":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-05-06 03:51:32.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/882302/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/02/882302/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/02/882302/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/02/882302/3"}}},"commit":{"parents":[{"commit":"4fc2ea4a85bbca25fae84a7506bf78f08d6485bf","subject":"Merge \"Enforce image safety during image_conversion\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fc2ea4a85bbca25fae84a7506bf78f08d6485bf"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-05-04 20:52:52.000000000","tz":-300},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2023-05-06 03:51:32.000000000","tz":0},"subject":"DNM: testing gate","message":"DNM: testing gate\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/882366\nDepends-On: https://review.opendev.org/c/openstack/glance-tempest-plugin/+/882362/\n\nChange-Id: Ie9a4f289aa1e7c2db1370804ec91e68444051965\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7bbfdf010b0a3e14a53eac3e38d8f5192251dfb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7bbfdf010b0a3e14a53eac3e38d8f5192251dfb7"}]},"branch":"refs/heads/stable/xena","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
