)]}'
{"id":"openstack%2Fglance~882329","triplet_id":"openstack%2Fglance~stable%2Fyoga~I0808dc0b1b34b527e38aa137c1dd25e1fc06409f","project":"openstack/glance","branch":"stable/yoga","topic":"dnm-debuggin-rbac-failures","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-06-28 14:23:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0808dc0b1b34b527e38aa137c1dd25e1fc06409f","subject":"Enabled new defaults and scope checks by default","status":"ABANDONED","created":"2023-05-04 15:21:14.000000000","updated":"2023-06-28 14:23:26.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21e19b017adca0222b632b73c4e083998096407f","_number":882329,"virtual_id_number":882329,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-04 19:07:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-04 15:21:14.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2023-05-04 19:07:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8b2426a69362f381ca93baaa2e329dee4a0dd3e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-05-04 15:21:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50bfc1270af5fcd64d7e62533461cbf3c24dd925","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 19:07:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b499e7175a04e6a8b768fe8524822bc\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/79774d4670b0406da00ab358e75ddc89 : SUCCESS in 2m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/19de27f0d89f48308661f2e619c166b1 : SUCCESS in 46m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70e67fb84a3e459abebf422bd976ce3a : SUCCESS in 1h 37m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59d63c8caf06420e909b905a903f0d76 : SUCCESS in 49m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745036a2b66742b18a47e768c51cf4af : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f17af9327eb4c258d3444cc565875e4 : FAILURE in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c740d3cc21a44fe1926003d7cf54a5a5 : FAILURE in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99c57603e228436a872a1e603249ea0b : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fef333140f014ec6ad64ae5235c3f54e : SUCCESS in 6m 45s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/51153c1548224134b6e2f148bd585521 : SUCCESS in 13m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a2ffafc3656147bbbd94bf4996d52010 : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/62396532587b447ea4590774a4ea4f2e : SUCCESS in 9m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/56e5cc2fb93e4d7e8b0704a624fe338b : SUCCESS in 1h 26m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ad156e1805b4457b9bca7d3f4635d72f : SUCCESS in 1h 33m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fd32a19f511c4d4994979c0c2befebc6 : SUCCESS in 1h 26m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/707d1b95fc364d5a862dbcf8f5f74405 : SUCCESS in 1h 27m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9813c2b7dda24f8989cbe22e39461308 : SUCCESS in 1h 26m 52s\n- glance-multistore-cinder-import-fips https://zuul.opendev.org/t/openstack/build/550ca1ffc2a042bfa41e72c91bfb0d89 : RETRY_LIMIT in 19m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cc90a56b89a45b2a13c5ea994995032 : SUCCESS in 52m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1d5e020420f74994bfd6a0a41f6ad148 : SUCCESS in 1h 19m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/132721ff189441e68be14b19948b4366 : FAILURE in 25m 41s","accounts_in_message":[],"_revision_number":1},{"id":"21e19b017adca0222b632b73c4e083998096407f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-28 14:23:26.000000000","message":"Abandoned\n\nYoga is going EOL.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"366c727d6b15b32c42f7363917c4e3a1a82fe867","revisions":{"366c727d6b15b32c42f7363917c4e3a1a82fe867":{"kind":"REWORK","_number":1,"created":"2023-05-04 15:21:14.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/29/882329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/882329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/882329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/882329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/882329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/882329/1"}}},"commit":{"parents":[{"commit":"641399fdd00966f0e6cedcaafffa787ac44dc7bd","subject":"Merge \"Enforce image safety during image_conversion\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/641399fdd00966f0e6cedcaafffa787ac44dc7bd"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-02-02 10:24:00.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-05-04 15:20:56.000000000","tz":120},"subject":"Enabled new defaults and scope checks by default","message":"Enabled new defaults and scope checks by default\n\nEnabling the enforce scope and new defaults by default in glance\n\nRelated blueprint secure-rbac\n\nChange-Id: I0808dc0b1b34b527e38aa137c1dd25e1fc06409f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/366c727d6b15b32c42f7363917c4e3a1a82fe867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/366c727d6b15b32c42f7363917c4e3a1a82fe867"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
