)]}'
{"id":"openstack%2Fglance~882500","triplet_id":"openstack%2Fglance~master~I4d21cbcd0b2b09b88dd73575d9f9e36bf037bded","project":"openstack/glance","branch":"master","topic":"bobcat-m1-config-refresh","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-05-11 15:00:01.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-05-10 01:20:06.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-05-11 12:45:10.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I4d21cbcd0b2b09b88dd73575d9f9e36bf037bded","subject":"Refresh Glance example configs for bobcat milestone 1","status":"MERGED","created":"2023-05-07 19:43:58.000000000","updated":"2023-05-11 15:01:18.000000000","submitted":"2023-05-11 15:00:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"882500-bobcat-m1-config-refresh","meta_rev_id":"837ee2c8f2ceb83c7908000fb02d4d18b3824bd8","_number":882500,"virtual_id_number":882500,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-11 15:00:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-05-10 01:20:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-05-11 12:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-05-11 12:45:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-07 19:59:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-07 23:05:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-08 06:28:24.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-05-08 06:28:24.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"0e6c0e9752622792acfa6f82743eeace1ce1c206","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-05-07 19:43:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ef1ded32b3f3e09f06237c8b3c908ded302e8c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-07 19:59:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2da41afa50a48eb9373b42a6074f397\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c08bfb2feae54ec0bd16181c84deff94 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4479b72d29e34421a81ea062640b2c3f : SUCCESS in 12m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"816cc7d07b8835d4a1894bb315354f8a641bb94c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-07 23:05:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/254a9dc9fd4b48a5a4a5a893a194a922\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6ac73853323415eb7e126c6c11810ac : SUCCESS in 45m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23436e6c2c7c42eabb0eb88d6aa58852 : SUCCESS in 1h 09m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34843198088348f2ad17da3fb8067ce4 : SUCCESS in 47m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9af0543d89a34724ae42455156e04ea8 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fac9eb79976148d383c79420bea0b52d : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/349cad2b695c4d78807a30a6afd60ff2 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51c6514dff745209c719d2952565d86 : SUCCESS in 5m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/908ed7f5b8fb4acf8482f638b9ddfd13 : SUCCESS in 10m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e31be7261fd4a3fa96ab0ed1af2fe0d : SUCCESS in 8m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4e6118888c4f4b61b08fc292f342a377 : SUCCESS in 7m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cc0f6e1e663e446899658c59ce8aa08e : SUCCESS in 7m 26s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c68bcd6666b84deeab5c46cb674d7bae : FAILURE in 1h 56m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7af5c639d3c4429aabc625d4e249ed45 : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f559e58818a942de887b738686bba718 : SUCCESS in 1h 31m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/26480225f90149fa8507fe154c6152ac : SUCCESS in 1h 00m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4a0131fdc6f742b6bdc239696a034778 : FAILURE in 1h 14m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d637f2d58f345c8bbdfacdef583280a : SUCCESS in 53m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4f659b33cf38425182d8ba7b3ba19d98 : SUCCESS in 1h 56m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bab49b0e969a4ff290f3b3f723c251bf : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":1},{"id":"714db0ac67c3a592865a2c083815da8b568efe37","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-05-08 06:28:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c7844673f7b5e58fe348cb63f367a9e4aa851dba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 06:44:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ce80cf0f46146648c5de29fc9356c7c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce57df31a4b04ea09667c326c6ec8b12 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0176e42e5491477fbc7a7877c0bebba5 : SUCCESS in 13m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f3ee94d8bb9e59c592472ca24d0df20c708d3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 08:47:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84ea2b302b32485eb9cd539d14570445\n\n- grenade https://zuul.opendev.org/t/openstack/build/c17f30b3170448ef978a99ebb5121773 : SUCCESS in 42m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/444984e0f06e4f41b49660467fc66b96 : SUCCESS in 1h 21m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d73edb145524e558997649a42fe9579 : SUCCESS in 42m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a42795eddec042c6999b99cfdf426152 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82b327f9dfdd4a3b87a6f1ca09df20d2 : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7753a7dfc8b04c4db9ad57923595575d : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0254d6f7e5404df1bf9217a449abfe1d : SUCCESS in 6m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7d7964c90bad4d59950bfd5ab72864e7 : SUCCESS in 14m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e41a0ed1e5bc4305bdd12d198fde51e2 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc8873a053cd48e9be59fa718a1c4d52 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c00d86264c70477d950525fcede08de7 : SUCCESS in 8m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d26b7885870645a59335bbe2baf114e0 : FAILURE in 2h 01m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7cc59f63e3b648109b385149f0f7d4b6 : SUCCESS in 1h 27m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b659ee31cecb40a99681ae4c08b9553a : SUCCESS in 1h 32m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6c443cba157c470096f90c7fd8001456 : SUCCESS in 1h 28m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0308e3b945d34d918061f9adf5fc9607 : SUCCESS in 2h 02m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e578ae9df05409895f96dbb30c297e4 : SUCCESS in 45m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38ee5e63def64b52a71733949cae1406 : TIMED_OUT in 2h 03m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a99acd8f05ac4963bcf23581a66083c1 : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a37a43aae4d6b928a47dcd20b6a7a54e3b5443a8","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-05-08 08:58:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"488aa8c20fd01005d8bafe477f220052bf469417","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 09:04:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb1d5090b75c45bf846b89b7494f04f8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba81b5f866e8440f877d5850b03462c5 : NODE_FAILURE Node request 300-0021161155 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ef80b73bda074ff58d257b0c978e13ff : NODE_FAILURE Node request 300-0021161156 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c217cc7e11a002f5a75d98423b56a8e7841d8a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-08 11:12:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85ca6a0c0d914975b91554d76cc84a65\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b86e52f94314421960c022defc8e26e : SUCCESS in 55m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94d7792270784f269a0ccf2e1aa25eca : SUCCESS in 1h 24m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/09066d7bb4074a9b8a0ac79b876c6144 : SUCCESS in 50m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d31706149b954bb0be8337e26676d0dd : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8569f13182924776892f09aa8773b2ed : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40ae9e0158be4f1cb623f0a7b67a6d67 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a1d3355ad5a4dfdabd70fbad4d43930 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/773ba761bbcb42bc83536392efa76d85 : SUCCESS in 14m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/426460bd23664e39af889289e74a3ca4 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/82c4cbce3be34f05939cd978f8797620 : SUCCESS in 7m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5da5cc9c518f4946b71facd7b643c735 : SUCCESS in 8m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c0ac2267dac24940b35daa4e2d937ce5 : FAILURE in 1h 46m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1773799d7aa24f2193b39c0001479676 : SUCCESS in 1h 28m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/03c2162d8bdf440caca13181f0533372 : SUCCESS in 1h 22m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/50cc2262b94c4525a1b3cad9a12212f3 : SUCCESS in 1h 18m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c006694daf3247aaac90b7897e520c9d : SUCCESS in 1h 39m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc835e68e0924347ab532049627a5ef8 : SUCCESS in 59m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e7194e12f3c0449db7a8f041ef045eff : SUCCESS in 1h 59m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc1b21d9c29349449c54ff72cc41f340 : SUCCESS in 32m 07s","accounts_in_message":[],"_revision_number":1},{"id":"e9d4987a0dd25cc2efcbb79d519ee2f37f3db28b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-05-10 01:20:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4a91f3406e985844c77ca827a344e1727417096","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-05-11 12:45:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44793567ab6f202ce5de3c2ac9de1865d4022372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 12:45:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7588082c7f186331fd60504d1b871ff58316f18a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 15:00:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8243666870484d099c5cdba798d0b5c3\n\n- grenade https://zuul.opendev.org/t/openstack/build/25d653f527794e17b902b60ea14d7508 : SUCCESS in 57m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a589b28bdc00408ba9ef0e5ef11f46c7 : SUCCESS in 1h 22m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6deb439f8d1e427eb02ab7dd3bd62ebb : SUCCESS in 1h 03m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35e9cdab43ae4fc58441802f51591068 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f551ccb0e9049d6bbc64259c273a72b : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2981592f0158409486d423a1c6b4b779 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8673e70db32645d58c16c9504379cb85 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4895b478e8584dbfa9d02e04f07fa5db : SUCCESS in 6m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fee73e32459746719ecad8da333fcdee : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d46b5611c48f42da8dba329ef3f55dae : SUCCESS in 8m 39s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f1bf9c8618ce4b4780fa122235e1306f : SUCCESS in 1h 22m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/57482a6ba154400d8ab171a3ac1989dc : SUCCESS in 1h 15m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/db2a00b752814c2fbe295a7115338801 : SUCCESS in 1h 23m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08fee0e8bb4343998ee49b8092ba564f : SUCCESS in 49m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d8bafbfafbe41ed9c1d46a6fd97d22d : SUCCESS in 2h 13m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ff4845f344d4e57be37f0f10368e088 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"13d6692b65844d96a5391ec7df000263d77eac0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 15:00:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"837ee2c8f2ceb83c7908000fb02d4d18b3824bd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-11 15:01:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e153051e6da54e5b9c4d08bc171e85ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaf299a027a84b948cf0aa0d395442d9 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02de7a7d2a21b4577bba5af10c001f546a970d13","revisions":{"02de7a7d2a21b4577bba5af10c001f546a970d13":{"kind":"REWORK","_number":1,"created":"2023-05-07 19:43:58.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/00/882500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/00/882500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/882500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/882500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/00/882500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/00/882500/1"}}},"commit":{"parents":[{"commit":"26d53c4adce210674a0cd00bfb9f4d555c288c1b","subject":"Merge \"Remove duplicate value in compute-host-capabilities.json\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26d53c4adce210674a0cd00bfb9f4d555c288c1b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-05-07 19:27:00.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-05-07 19:27:00.000000000","tz":0},"subject":"Refresh Glance example configs for bobcat milestone 1","message":"Refresh Glance example configs for bobcat milestone 1\n\nChange-Id: I4d21cbcd0b2b09b88dd73575d9f9e36bf037bded\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02de7a7d2a21b4577bba5af10c001f546a970d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02de7a7d2a21b4577bba5af10c001f546a970d13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
