)]}'
{"id":"openstack%2Fglance~883515","triplet_id":"openstack%2Fglance~master~I636e6db755daa770041e76657ba82c1c43aa82ea","project":"openstack/glance","branch":"master","topic":"test-rbd-revert","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-05-30 14:25:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I636e6db755daa770041e76657ba82c1c43aa82ea","subject":"DNM: Test glance-store revert revert","status":"ABANDONED","created":"2023-05-18 14:13:41.000000000","updated":"2023-05-30 14:25:22.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5cabe1be16c9e9ee8fe0d253c360ddc54a6445b5","_number":883515,"virtual_id_number":883515,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-19 20:35:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 14:26:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-18 16:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ac8ed08a7e8f3a8ee8e1c02e164a7ee52b1b541","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-18 14:13:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea449846b73188b1ec1ac6de818e977df1b7ab3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 14:26:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0346446874b4454b59a07edb7dfba99\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62ef0a042d5d435ea306b201d2acd3cb : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a5468074fff4f81985b58960a92a163 : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a124ee0f9c654081825e2d054289e302 : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b913b9a0465e011cfd694a1bb5e7c70007c0758f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 16:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279f0b95a5bc4d858d603aa9407a8b74\n\n- grenade https://zuul.opendev.org/t/openstack/build/0beb627392644f50be9a1723e0b8931f : SUCCESS in 1h 02m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4becc224d9734258927ff95e073aa8c1 : SUCCESS in 1h 19m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/579693dac9024e259def139bef5901dd : SUCCESS in 58m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a86afc8c4244ddbaed99f76ce47e85a : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/858525011ea44372b79453304147dfcf : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a70e4ec637447cfa3d5c4ab9b38b857 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca581bdf4106458483719a6ed61659ba : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83828e32f13141178a81ced86cf7c16b : SUCCESS in 8m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3e107762025a46569bbaff0e651dd0cb : SUCCESS in 13m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/75ace09388a74f9991bd3a22bf52aa11 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6319f4cbd68d4d4d9a3f7dc066f46ce5 : SUCCESS in 8m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/61d0ec427d354907801deeefca153004 : SUCCESS in 11m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/85af48cc5e434f1a9bbb42121e9c6cfd : SUCCESS in 2h 00m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/917bdf0fa189446fb2d01b878097f9e7 : SUCCESS in 1h 58m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/55c9a2decd154b389596143a2671bad6 : SUCCESS in 1h 37m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aadb706eca254dc7a6a2ddb2078a3dc9 : SUCCESS in 1h 55m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f45c197f64ed45feb3a664e6524a2f02 : SUCCESS in 1h 45m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47ca3f2f073d4ed5b886032e5052c993 : SUCCESS in 52m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/411106bef0704e7882b96d6c621541c2 : SUCCESS in 1h 57m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/074122beb9054957a19cda6ef15de1df : SUCCESS in 37m 37s","accounts_in_message":[],"_revision_number":1},{"id":"8fd50522de7af5c47c302a6aea2a2ac69bf5728a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-19 13:41:22.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"47d3fc387d8a7050e6334caff8e5f3fc94d96b47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 13:53:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c848ddcd9d0245888d4e248cedbc6966\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/80f988e63c1942cc85c846b47e432266 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/03d0060979e54e9f91ed125048211d7d : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/99e9aaaa4c6a41109f811c82dbe5953a : SUCCESS in 4m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e48f40fd27f054afe1a9c247b4843f627fb3458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 15:36:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14f08c1674154b37a5d76de6c9524fc3\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5841af90fdb4e5092ef048ac260ff19 : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a440eca7c6e49128435c24744e8ff74 : SUCCESS in 1h 27m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2b6fa6421774ec0b724c994135ffcc0 : SUCCESS in 1h 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f4e0c092ec4b29921dafd9c4603b16 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/072285a9be1847c7a114b2bcac422f5e : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01fca0ea0054483ca9cd8f00537461ee : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d542024c7494cf29cb2028e1793f980 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49939998b01d492b88a223b9e9a48e82 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1e40343aedbc4512bad0dec5dbe9f86d : SUCCESS in 13m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c5f170fde07456e865356cd4e44294b : SUCCESS in 9m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e5e606425f9e428499e0b6c1fb057667 : SUCCESS in 9m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/74965c0876ea43268e6e7190d5320286 : SUCCESS in 9m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/09e12a9c06f24bec8b1e37c70b5ae296 : SUCCESS in 1h 49m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/85ede245e90047c89a15a90b5efd0a5c : SUCCESS in 1h 38m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a9cb8ae16284c6baa967723b9e11817 : SUCCESS in 1h 25m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/664140bca7994ce4ac606f2d3d0f688e : SUCCESS in 1h 20m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/01983e0581c34784883778fcf0606837 : SUCCESS in 1h 47m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e66b9b620c34439864cb4a6755a78b6 : SUCCESS in 52m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e72b2a653513446286b806e1b2ac6c9b : SUCCESS in 1h 47m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de60e5493ffd47d5b904d806b538a2fa : SUCCESS in 40m 44s","accounts_in_message":[],"_revision_number":2},{"id":"88ee2b7325902a6d8b4974a7e54aa2378dec2361","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-19 16:37:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e8c3f5a950ccb394e338ce9340dc64092ec05347","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 16:49:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1846fe80396745daba74d49ce4ba6647\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc0512488971495983f0746db57c8919 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/397324bf072b47a79a02f38bec13f939 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4a3b3bb8e0b74478ae9ffe62f6c0e3dc : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5fa8132048d92405eefe6e00165be7b317ebfc8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-19 18:53:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"89a0a8ae617d19096f06467d447f6cf83a2e2644","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:54:26.000000000","message":"Patch Set 3: Verified-1\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027required_projects\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: glance-nova-ceph-multistore\n      parent: nova-ceph-multistore\n      required_projects:\n        - openstack/glance_store\n\n  in \"openstack/glance/.zuul.yaml@master\", line 202, column 3","accounts_in_message":[],"_revision_number":3},{"id":"9ac0b0a4c6d87add02093aa834c89700a9b32c72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 18:54:30.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  extra keys not allowed @ data[\u0027required_projects\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: glance-nova-ceph-multistore\n      parent: nova-ceph-multistore\n      required_projects:\n        - openstack/glance_store\n\n  in \"openstack/glance/.zuul.yaml@master\", line 202, column 3","accounts_in_message":[],"_revision_number":3},{"id":"7dd26d455bec0f2bb5bdd4075485309f6870d0a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-19 18:56:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"6a13648bc2ed16a3de1f8750be7fbdc9a8c03e87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 19:09:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7074510fe5aa4c668b416c123c7d2a55\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c8b55a7e5d44099a93ee12489924c5d : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75d28e9c401842f7b635b57a088fdc64 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d86348dfe34845b2bbf0005b3826f2c7 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6e3614fae44d2d9950d44a06509ad61fff8af800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:35:01.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5f5457cde794279aa0876fcd16985e2\n\n- grenade https://zuul.opendev.org/t/openstack/build/719a98e62f2a433f84f254af420227fa : SUCCESS in 50m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/017cab03fab64b3596395946e5e45618 : SUCCESS in 1h 08m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1568a60dc2f1414db54c821ee5ff1e10 : SUCCESS in 1h 09m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66b0b6e3cd1b478894f16709802ef1f4 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2d89634e7cd4e9a98453866874c1cd1 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7749c532a854d03ad52c647fac8d26c : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13253a2289bc45f5abe5e97edbe28351 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fd1c1a7eaa41f89a41777ba7c244fb : SUCCESS in 6m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/58c51e6690a74cf7a10f41f127faf30f : SUCCESS in 1h 29m 03s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1df00406d6e14746958cfccc18ce183e : FAILURE in 1h 29m 45s\n- glance-nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f7d9e89bf5b42919fed9219f7b89c72 : SUCCESS in 1h 36m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2a8f5426ae34488aaebd09e4a6763245 : SUCCESS in 30m 14s","accounts_in_message":[],"_revision_number":4},{"id":"01df0302f69b0833324d823f72973a4ef07ae329","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-19 20:37:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1958006379252268a97126081883601e62a428f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 20:50:44.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a33495333642f4ac7370c388f6478e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89197bddcf174cfba5d2e7ed2a7ae7c1 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9cf16c1a97f40eba46b7fb5a85556f8 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a90551c6fd4414c9e8b1d4f66e346cc : SUCCESS in 10m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"60ca8e03714905ced6fef7b48431327eedf89472","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 22:34:08.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c260359e21c54a02afb92346acc9b7d3\n\n- grenade https://zuul.opendev.org/t/openstack/build/d76491ae03d84201baecbbe10f920d8b : SUCCESS in 1h 02m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3bd3aec77bc44ccb5a311ed16f91b1a : SUCCESS in 1h 46m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b5a6d88ce8944889b69510c3606a37dc : SUCCESS in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3648d0793022437089f3b0bd21f70618 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e004b6081f14726901c9cbfdb07f3de : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/834d892412474694b26dadc8cb8d8da7 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edb9a280b86c48888da4a9b8b36c5ca1 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6364b8951de4c438816a0cf9e02e459 : SUCCESS in 5m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c1c79d52360745329f2b3be31af0f474 : SUCCESS in 1h 39m 21s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ee579f84b95f4d20bbe21753844cde03 : FAILURE in 1h 48m 41s\n- glance-nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6387adac3aaf4cf39a5a90c7e8dd6f26 : SUCCESS in 1h 43m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1a94fd68379443bab19addf7e194ad7f : SUCCESS in 40m 56s","accounts_in_message":[],"_revision_number":4},{"id":"5cabe1be16c9e9ee8fe0d253c360ddc54a6445b5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-30 14:25:22.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8910c6110051827c12c4eb7848a3d810ee1db260","revisions":{"76872ec1aa450273bc7133220a63057f289ff455":{"kind":"REWORK","_number":1,"created":"2023-05-18 14:13:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/883515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/883515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/883515/1"}}},"commit":{"parents":[{"commit":"428295969836a2a0bad0ae0593ea1a890e155898","subject":"Merge \"Deploy healthcheck middleware as app instead of filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/428295969836a2a0bad0ae0593ea1a890e155898"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-18 14:13:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-18 14:13:15.000000000","tz":-420},"subject":"DNM: Test glance-store revert","message":"DNM: Test glance-store revert\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/883532\nChange-Id: I636e6db755daa770041e76657ba82c1c43aa82ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76872ec1aa450273bc7133220a63057f289ff455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76872ec1aa450273bc7133220a63057f289ff455"}]},"branch":"refs/heads/master"},"f9ee45dfa1d649b1a952279a25ffb8ee04b35d9d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-05-19 13:41:22.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/883515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/883515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/883515/2"}}},"commit":{"parents":[{"commit":"428295969836a2a0bad0ae0593ea1a890e155898","subject":"Merge \"Deploy healthcheck middleware as app instead of filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/428295969836a2a0bad0ae0593ea1a890e155898"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-18 14:13:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-19 13:41:01.000000000","tz":-420},"subject":"DNM: Test glance-store revert revert","message":"DNM: Test glance-store revert revert\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/883670\nChange-Id: I636e6db755daa770041e76657ba82c1c43aa82ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9ee45dfa1d649b1a952279a25ffb8ee04b35d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9ee45dfa1d649b1a952279a25ffb8ee04b35d9d"}]},"branch":"refs/heads/master"},"ab26aa4d03b06ac14269003348baf2c3b311d250":{"kind":"REWORK","_number":3,"created":"2023-05-19 18:53:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/883515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/883515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/883515/3"}}},"commit":{"parents":[{"commit":"428295969836a2a0bad0ae0593ea1a890e155898","subject":"Merge \"Deploy healthcheck middleware as app instead of filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/428295969836a2a0bad0ae0593ea1a890e155898"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-18 14:13:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-19 18:27:43.000000000","tz":-420},"subject":"DNM: Test glance-store revert revert","message":"DNM: Test glance-store revert revert\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/883670\nChange-Id: I636e6db755daa770041e76657ba82c1c43aa82ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab26aa4d03b06ac14269003348baf2c3b311d250"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ab26aa4d03b06ac14269003348baf2c3b311d250"}]},"branch":"refs/heads/master"},"8910c6110051827c12c4eb7848a3d810ee1db260":{"kind":"REWORK","_number":4,"created":"2023-05-19 18:56:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/883515/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/883515/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/883515/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/883515/4"}}},"commit":{"parents":[{"commit":"428295969836a2a0bad0ae0593ea1a890e155898","subject":"Merge \"Deploy healthcheck middleware as app instead of filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/428295969836a2a0bad0ae0593ea1a890e155898"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-18 14:13:15.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-05-19 18:56:30.000000000","tz":-420},"subject":"DNM: Test glance-store revert revert","message":"DNM: Test glance-store revert revert\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/883670\nChange-Id: I636e6db755daa770041e76657ba82c1c43aa82ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8910c6110051827c12c4eb7848a3d810ee1db260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8910c6110051827c12c4eb7848a3d810ee1db260"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
