)]}'
{"id":"openstack%2Fglance~883619","triplet_id":"openstack%2Fglance~master~I14abd88944407da17ca81b77c079262565ff71c6","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-05-25 15:14:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I14abd88944407da17ca81b77c079262565ff71c6","subject":"Run integrated test if zuul.yaml file is changed","status":"MERGED","created":"2023-05-19 03:29:56.000000000","updated":"2023-05-25 15:17:02.000000000","submitted":"2023-05-25 15:14:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883619","meta_rev_id":"3ae0822f145ffd12e71307f93a1c328833a5ffe4","_number":883619,"virtual_id_number":883619,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 15:14:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-05-23 04:01:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-05-25 13:13:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-05-25 13:13:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-19 03:41:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-05-19 05:29:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-23 04:01:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-05-25 13:13:06.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"c4480712eab0cb827e9efd36a06824d9e4cbd26d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-19 03:29:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51774c640e9c774d8211a90c15f7189a5914c461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 03:41:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c39071b5e1244d9a40889eafc4d1923\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0cbec656ec442129c44fbc3d2a9062e : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78b2042ace5c4e228bb8e1f4dfbd8b9e : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c632be139314d34a12f5f4e577434af : SUCCESS in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8831f83acbab3c5770319ae361e6f00f77161f88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 05:29:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/decde791fac94bd08c99b6ddfc9860b0\n\n- grenade https://zuul.opendev.org/t/openstack/build/e520587672824a47a9bd898798a7991c : SUCCESS in 52m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f1c88e6270747b2a03c3348c4123174 : SUCCESS in 1h 39m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16ab768a08ff4614bed9f468ea81e8dd : SUCCESS in 57m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6e4912d54bc443cab1fad19f3d74b04 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a250fcbcaab84fb4b0841f91893c1722 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bd7ff75404c4906be8906a06f35bfbb : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec4933caa5ba4216835bc0b8b418efb1 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6dfbda6e5af4a25b614d94d0981ada8 : SUCCESS in 5m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7be2b7718865402dbfd4c99a752cbfc6 : SUCCESS in 14m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d8a9730c84024c32bf6870e0ba88cb79 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2363906cb12240da984eea7739f2afbd : SUCCESS in 7m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/65ec833677574802bee4db32ddf103dd : SUCCESS in 9m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7298879b95834b26ae86c50c679ba3bb : SUCCESS in 1h 32m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9c2c78ecb95453eb8278412775b0e06 : SUCCESS in 1h 11m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f8719f5fb5b42c9a64517a53ecc1186 : SUCCESS in 1h 35m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5f5920b2fbab4269851cfc3be35a9994 : SUCCESS in 1h 18m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/05492494d109435d8a7c6807834e8a9b : FAILURE in 1h 05m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3af33088a51e44149324cf4c311b91d3 : SUCCESS in 56m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/badc6afe38594b7986f56e8cb6e20e6b : SUCCESS in 1h 48m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2fcfd9abbce440559e6f2f78d5c0eeb1 : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"dd258d406735987560d74d08ce755de560ebeb16","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-20 01:51:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8736da90a05a5cb1989766c727664d277e09209e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 02:04:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b593f1f11054f62b5db789eb149afae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c05d4f32246147d29a9a34713f852f07 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c276956837014b39b4f7488376be15d6 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/460639f5242441e2b0d192d41b121f77 : SUCCESS in 11m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf6b6b03b32c2207ab7c3de23a99372141209aeb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-20 04:25:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65503b259b434adfb9fbaaeb30a93ca0\n\n- grenade https://zuul.opendev.org/t/openstack/build/41661aceb0a84449adfc0721a8ab2287 : SUCCESS in 1h 05m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e1de5d6f90c44edd8573a573c3e31a1c : SUCCESS in 1h 51m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da93924616044f92ac830504c042c051 : SUCCESS in 1h 14m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8c6cf0504a94eb9b00551d836739554 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/481db8ff46134daba38ab749e35c13bc : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb6bcc61b5824b29954212310ad902ee : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9f03c80b39848fc8d328a4dbbd91899 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fed0045499734f47932cffd71cfa5937 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6861fcc2724848c89d861f7af5984f11 : SUCCESS in 11m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c92fe7c874a348c688835a0a34734e01 : SUCCESS in 10m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aae71312fbcf4a2dae87b8d606ded655 : SUCCESS in 10m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c2b1cdb8bbd54633bd84f9b4bb13e665 : SUCCESS in 9m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/759ea93cca96472b90e37c6909d0ee28 : TIMED_OUT in 2h 02m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b63d983f8171423eaa016c6c1008a5f9 : SUCCESS in 1h 31m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb03861cccb7440aa353b4712b1fda16 : TIMED_OUT in 2h 04m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b20b5aa4c4bb4f0cb3206255b131bee6 : TIMED_OUT in 2h 04m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6daadd7c585149aa9755761254712861 : TIMED_OUT in 2h 04m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e97776b4b034d1095e1b24b234a33f4 : SUCCESS in 58m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b5d5c3778fc49f7b64a86ee62128470 : TIMED_OUT in 2h 32m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7923956bc569433cb33ad2ad2f9d5b7b : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":1},{"id":"77a75f72b380566f95376a2c66b37cf08c75b6db","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-22 17:31:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6a0acaaf4e147a8bede6d9afd42713b601faa57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 18:00:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d1fc9cfa1614ce5b058584386fb9318\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfec7bf34c074b088af6c77d63155d19 : NODE_FAILURE Node request 300-0021278876 failed in 0s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a6bbac4523c419388cc6cff86d21cce : NODE_FAILURE Node request 300-0021278877 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5fac381b0f394afdaf5ca775053b8499 : NODE_FAILURE Node request 300-0021278878 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d360c35a64db8e8cedb9fb6235c20a0554702349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 21:14:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff5a8aca4a949feb0d78f5cd9cbc4cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/e54f1f61e2944136969e3bd7eae02e10 : SUCCESS in 47m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8827d01d72124b79a35a77d73be4815b : SUCCESS in 1h 20m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fd06e8e318d4525ad603be0c7634d4e : SUCCESS in 59m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f22c57a595cf458492f3c7a0ff3a91e9 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c31793f9666041ed9d98f86e9eb431ee : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4988406b303405a829591898fab08f0 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/423dba1cef114db1a42497c4a9713203 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2537c861f4d14fbf9ea130684a5c8762 : SUCCESS in 5m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/25ddf7e8bc3d446595e1f89374b1643a : SUCCESS in 13m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f83986f4e4f4779889fa80f34762b2b : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0490675cf687457cb874f776ad8d22de : SUCCESS in 10m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/63861b06a4544cb2b22bb1dc749fb36a : SUCCESS in 9m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/51d65370d97848bbad726db6eb738da1 : SUCCESS in 1h 49m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/91e7dbe3f53646da85ca510546d96624 : SUCCESS in 1h 13m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b7d9716e3fa64698b5df6c146c266ed8 : SUCCESS in 1h 26m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d4e544ee21b04c0a93b4d9209ca15f73 : SUCCESS in 1h 17m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cf6b716f85504f488528b9f5375e5bf5 : SUCCESS in 1h 42m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38b243258ea04831bdb808a189a69f2a : SUCCESS in 1h 02m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/23d07920af0e4e4e8e653a8697b88aa8 : SUCCESS in 2h 10m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/70ae60628aff41b691a1a778dd9e14d5 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":1},{"id":"90ae84cce0fef84d9a80c44d78f9d7825806b151","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-05-23 04:01:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0e5ca1b1bcd12e11deec8a11e9429b09d48b48c9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-05-25 13:13:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f19fb5942be07298e8aa964e29c89f110119fdc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 13:13:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c9ee24223d76b0bdea8c3385a276ee3ed35b6d6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 15:14:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6000180c334ed49bd7fd03ff21e099\n\n- grenade https://zuul.opendev.org/t/openstack/build/68e517f0b3b94cdbac79474c781834e3 : SUCCESS in 56m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c0a76221744e41f4b534f49f1b549a89 : SUCCESS in 1h 19m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/670f10872eae4dbab50c7b2a9108b8a5 : SUCCESS in 1h 00m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea7996dbb58b4801a83eb6220ad2d217 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86d95b601a8b4ad68e8b2c2eb97b90d0 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e0719735cde4b1a92ec03b0078850ea : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8eea4e0d0984e05a8d74866173ef4e9 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04a65420e77547a6996936e87c1d337e : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3813320cf7c841f9bc021184340d0a99 : SUCCESS in 9m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a73ab013be548e6ab20143fbf853217 : SUCCESS in 8m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/145353b6c54a4162a6bfde048063203a : SUCCESS in 1h 28m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/506fb1eac3dd4df5a839fab600b3ecdf : SUCCESS in 1h 39m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/55988846f74749bd939acf3741005764 : SUCCESS in 1h 45m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/06288b11e2484e80880d708b409981f6 : SUCCESS in 1h 03m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1d3a2c6d4294da2994ca8eeadd00f31 : SUCCESS in 1h 55m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91c3176d3b7c431380001cd5ed29f61c : SUCCESS in 42m 04s","accounts_in_message":[],"_revision_number":1},{"id":"1af96f1af6337c4750351a8ab2fc6c5f56f52bfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 15:14:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3ae0822f145ffd12e71307f93a1c328833a5ffe4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 15:17:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa9b1332387545a5810ea5686b54b5f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc5b0fc578e4e04b605663ab8d8935e : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dca70d1ca8f6c5947bea0b5a623843b68cfb5d15","revisions":{"dca70d1ca8f6c5947bea0b5a623843b68cfb5d15":{"kind":"REWORK","_number":1,"created":"2023-05-19 03:29:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/883619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/19/883619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/883619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/883619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/19/883619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/19/883619/1"}}},"commit":{"parents":[{"commit":"428295969836a2a0bad0ae0593ea1a890e155898","subject":"Merge \"Deploy healthcheck middleware as app instead of filter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/428295969836a2a0bad0ae0593ea1a890e155898"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-05-19 03:27:30.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-05-19 03:27:30.000000000","tz":-300},"subject":"Run integrated test if zuul.yaml file is changed","message":"Run integrated test if zuul.yaml file is changed\n\nzuul.yaml file contain many jobs definition and base class.\nAny change in this file can impact other job definition. Also,\nmany times we want to test the full integrated gate if this\nfile is changed or if depends-on is changing some global setting,\nfor example: https://review.opendev.org/c/openstack/glance/+/883602/1\n\nChange-Id: I14abd88944407da17ca81b77c079262565ff71c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dca70d1ca8f6c5947bea0b5a623843b68cfb5d15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dca70d1ca8f6c5947bea0b5a623843b68cfb5d15"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
