)]}'
{"id":"openstack%2Fglance~886183","triplet_id":"openstack%2Fglance~master~I473684bb8f8a75c905b1afd95e6fd116a58a9d82","project":"openstack/glance","branch":"master","topic":"test-rbd-trash","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-06-16 17:14:10.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I473684bb8f8a75c905b1afd95e6fd116a58a9d82","subject":"DNM: Test glance-store RBD trash","status":"ABANDONED","created":"2023-06-15 14:58:23.000000000","updated":"2023-06-16 17:14:10.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9160bfa3f376144b59e8983b065947ce0c3e706b","_number":886183,"virtual_id_number":886183,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-15 21:00:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-15 15:14:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-15 18:07:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b3772b48900176d00da9219e0f339044357cc901","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-06-15 14:58:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ebfc21b8980a158391fa21b5217adf07195b92b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 15:14:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0c82b77bcc74302852c402e03724d85\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3001ed25d6043fd837e62435d9a3dec : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec7226e12f6449879b0ff68b8fa0b680 : SUCCESS in 13m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e321757aba2407bbc33de29a3a39cb1 : SUCCESS in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd5a4a37e3224f6aa9e245f8350decfdb6a42a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 18:07:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/edf6fda3ee20498fab36a15c832df03e\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff43bb3bd22d4120811a49e467cbc35c : TIMED_OUT in 3h 02m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c0f81aadfc9f4170ac6545819d8cda28 : FAILURE in 55m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed5ef96d981a4746b4fd52033bdd694c : POST_FAILURE in 39m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f0a4dbd50884936a617b5daff832126 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c010c9ecf3c6486888632799ccdbb0e6 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6913ca00991441eab661e1c1f0803f16 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/307d82b7c6e04706b33b8500066db490 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eab018568414caf90522992f4e40a66 : SUCCESS in 6m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/dc68f703fc724d2ab611ea534d640d55 : SUCCESS in 15m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a63d27baa0f49f5881cedb18bd1a196 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4899a8a1e42640d1b3a365c79d412f2e : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6aafce4cd2ee486684f361edca6ccf9b : SUCCESS in 9m 05s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/105e0a5d4ce941db83af2a83088148bf : TIMED_OUT in 2h 03m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d8e0c22ddcc4e5f91c8da370e8950f3 : SUCCESS in 1h 27m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c1d75e7ede4143dfb18a172ea56935d8 : SUCCESS in 1h 35m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4843686794ae4a3892b31e2ad0ec2e8e : SUCCESS in 1h 47m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cd548c15104346549c099d7e692a71e7 : TIMED_OUT in 2h 04m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/384ed567a63e4fbaa176aa4a161ceb1c : SUCCESS in 55m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92d477d7a0de480eb8e47dbe2cffc91f : TIMED_OUT in 2h 33m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5410a44c6ee34866b6dff2cfafbf7960 : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":1},{"id":"86b42c9c631ff5436359fdcfc6e965791a6d8eab","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-06-15 18:12:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d97179dd9f54d17d2b2dc7db48b89e8672a07b7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 18:24:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d9771c289448d5bf9462a871b34625\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/862c2401f4c74c45baf31d49c3c5c976 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/be850db8fae549208346dc967b23f21f : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a126b909c16a4ecf83be92a916ad6206 : SUCCESS in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff3a122142b9b2e55c52091b7c16bcb107cca4bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-06-15 18:57:52.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"5a974772506316807d1269867faba121a3ed9b5d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 19:10:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c099a63ecc52499fb1863c4473ce12bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa35403802644e739158d5f70676108c : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43e018b06a4948e08daeda63893fb62b : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b14aa0cf8fed4730b2d8f7983a020991 : SUCCESS in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"63ec95a9943dd31c60f4228e06db98db1df25a54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-15 21:00:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/88cbe2a113dd418091105e6436b2e6a2\n\n- grenade https://zuul.opendev.org/t/openstack/build/74911f5d883f4736a3c7cec2a81838f3 : SUCCESS in 53m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13b84b6a95fb4170bb5e68a8c74812d9 : SUCCESS in 1h 27m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18eb090b0f4740748b18b554da43947d : SUCCESS in 1h 02m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16f63c7632b34f24ad91a9468a0861e2 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3daa47c20e4c42e5af4abee0f15b52d6 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b16c50b466ba49e49e7c771d6d28f55f : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58e53b504c94424ca21e19f18e8d07cd : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8df44d96184eaebdfd746bdadf50f8 : SUCCESS in 7m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/317d6fdbae7d489bb68ea4bf72fe3e5f : SUCCESS in 11m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7ad2e1c8bfec4b2ba7c8f5ba4b81ceda : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bc040c1d0af5410a812152b05299f4b1 : SUCCESS in 10m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d0a2d918da9e451997c798c78bfd7fbb : SUCCESS in 9m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ba50f79555d945468714d286aab77ed0 : SUCCESS in 1h 43m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3fe171a9baee430ab81049c98a707894 : SUCCESS in 1h 39m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07807ea442e040679c7de5dfacd9ff6f : SUCCESS in 1h 28m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9e02aed1739145daa2de87efdfae1ee2 : SUCCESS in 1h 49m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0c82c29cb1fa42e5bbec88759113efe3 : SUCCESS in 1h 47m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e010abef9d2d440788a26821f50c967c : SUCCESS in 1h 07m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a43540f7156441be81d7a1130d5335c3 : FAILURE in 2h 02m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b3e5409b71b044b5be8e624cfc38c2c5 : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":2},{"id":"9160bfa3f376144b59e8983b065947ce0c3e706b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-06-16 17:14:10.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4d4641a54c4b344bcf629242984f6c5333e0989c","revisions":{"5794bf0f8548bd5d396b45e50f14aacc113c2509":{"kind":"REWORK","_number":1,"created":"2023-06-15 14:58:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/886183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/886183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/886183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/886183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/886183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/886183/1"}}},"commit":{"parents":[{"commit":"43b2116adb92e5fbe945b53edd115edadd73a016","subject":"Test glance RBAC old defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43b2116adb92e5fbe945b53edd115edadd73a016"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-06-15 14:57:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-06-15 14:57:55.000000000","tz":-420},"subject":"DNM: Test glance-store RBD trash","message":"DNM: Test glance-store RBD trash\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/884524\nChange-Id: I473684bb8f8a75c905b1afd95e6fd116a58a9d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5794bf0f8548bd5d396b45e50f14aacc113c2509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5794bf0f8548bd5d396b45e50f14aacc113c2509"}]},"branch":"refs/heads/master"},"4d4641a54c4b344bcf629242984f6c5333e0989c":{"kind":"REWORK","_number":2,"created":"2023-06-15 18:57:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/83/886183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/886183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/886183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/886183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/886183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/886183/2"}}},"commit":{"parents":[{"commit":"43b2116adb92e5fbe945b53edd115edadd73a016","subject":"Test glance RBAC old defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43b2116adb92e5fbe945b53edd115edadd73a016"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-06-15 14:57:55.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dms@danplanet.com","date":"2023-06-15 18:57:51.000000000","tz":0},"subject":"DNM: Test glance-store RBD trash","message":"DNM: Test glance-store RBD trash\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/884524\nChange-Id: I473684bb8f8a75c905b1afd95e6fd116a58a9d82\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d4641a54c4b344bcf629242984f6c5333e0989c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4d4641a54c4b344bcf629242984f6c5333e0989c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
