)]}'
{"id":"openstack%2Fglance~886754","triplet_id":"openstack%2Fglance~master~Ide428268f52527972165ab59ac6e3da0688968f4","project":"openstack/glance","branch":"master","topic":"remove-six-metaclass","attention_set":{},"removed_from_attention_set":{"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-07-10 08:20:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ide428268f52527972165ab59ac6e3da0688968f4","subject":"Remove the last occurrence of six.add_metaclass","status":"MERGED","created":"2023-06-22 13:50:25.000000000","updated":"2023-07-10 08:21:51.000000000","submitted":"2023-07-10 08:20:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886754-remove-six-metaclass","meta_rev_id":"e513c2d1a29dad8209caa3297dd7c100b8c89deb","_number":886754,"virtual_id_number":886754,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-10 08:20:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-06-28 14:12:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-06-28 14:06:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-29 11:26:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-06-28 14:12:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-22 14:04:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-22 18:01:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-23 14:36:54.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-06-28 14:06:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-06-28 14:12:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-06-29 11:26:49.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2023-07-05 10:08:41.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"ef071ea414202d825995e5ba0857f996daab416e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-22 13:50:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22cd0698c15761217cead7c74cdfba5a5d12237f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 14:04:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccba9998d02f4ec4b189d897c8aef33d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cceed292d5db4d279b086f65ceb67387 : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b11992212694cfaad4b68c50ae59de0 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0aaf609067754f3696d2cd99916fdb13 : SUCCESS in 4m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"96d342e465f5591d7b4af5868df20777bca37f6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 18:01:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4edc0df084344007a51e8489261c1457\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0c1a2ce77ef43a59c5128aeb271f69e : SUCCESS in 56m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/775697505775421c9751bd315ff47ae8 : SUCCESS in 1h 48m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec1babc5fd224c099a62e987b404f14f : SUCCESS in 1h 01m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4c805363c79494d9cedf6d4df0507b2 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/080b3ba98e224a329c151e19c4a1c6fb : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/313960c421214c69a0a638836b5bdcfc : SUCCESS in 7m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c1f609e169a40f798d41ff5781c2596 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fce30872f5e4e28bd409074981e7f42 : SUCCESS in 6m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/03311fb90b78471c94e0e5c592f02637 : SUCCESS in 14m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a2e402de8b842a7bb2f71d84b98a381 : SUCCESS in 9m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/62373ed6870f408ba7d49e756eddbf05 : SUCCESS in 11m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1850b5bac096425c8e17c151563465a7 : SUCCESS in 9m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8eef2918a47b4587a3e09d43bafec407 : SUCCESS in 1h 48m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/20b73ccb1c494e5abd282c438f7efbb3 : SUCCESS in 1h 44m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c2160a4e3dac4f8cb7163557eea59df8 : TIMED_OUT in 2h 03m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/898a8fd9c09c43dfb546811dd888dbed : SUCCESS in 1h 58m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/db00ac8a6b0447088c05d14a03f7478b : FAILURE in 1h 59m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/09c5daa4cc7f483cb4dbb664b01d2240 : SUCCESS in 55m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/33a3b41514bb4261829fe6453aec9597 : FAILURE in 1h 45m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3bb79dea35b04e49b269b8f3ba24c510 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":1},{"id":"687d3db0c4cff16c31a48910001727371f4e9091","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-06-23 14:36:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03297869adfb689c518570abeb481a035574b6d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 14:44:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a5a2281c4c745f5b1f176289479213f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c61aed6046d245e4becdd5034824f58f : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d9b81f754cd49ca954988233aef7e8d : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1de5410221484d0ebf9aef7a7a8d497c : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b671c4b9bf46b24309f126c295112cd06acdc72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-23 16:45:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe51a85a59324886ba62a904cf869ee2\n\n- grenade https://zuul.opendev.org/t/openstack/build/55bbfce8f928400aae1c4a03b4cee611 : SUCCESS in 48m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6dc997ad38154e55a1e4931fc64e112d : SUCCESS in 1h 56m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b9da9eab20f4750a21bacc66289586f : SUCCESS in 58m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ac8f50342b4923ac4a770c41dd80cf : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d306fced1654ba08e5bf93d16abc57d : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/31349ce4400d4b92bbcc3b6fe4d1eb7d : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0f958cfd51f4d7abe9a29fa699ea8aa : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e54ff8f9d3ba46ada5910faf83e678f4 : SUCCESS in 7m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1d6ec7a2d7c54bbb9de019b5ce5369ff : SUCCESS in 14m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/038c070cfe554daa87e67695193e1dc8 : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/65229e96df484509a85e698b0425ccf6 : SUCCESS in 10m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1ff241fe77a24563987bcb1d58ab99de : SUCCESS in 10m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f0941593ff534b49b25c8a76d89d2aa5 : SUCCESS in 2h 03m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1bef5173ad6d41c68084c797ccc3cd06 : SUCCESS in 1h 58m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/065f625e826449cca13bb92f8058e1d3 : SUCCESS in 1h 40m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/36ec7b1bf83543028f81c3410a2da84b : SUCCESS in 1h 42m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/64df3ac1241c4569a1f66fe77add1a85 : FAILURE in 1h 52m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/819b28069c1d4679b84abda0daf3fa2c : SUCCESS in 1h 07m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/184a2e70380e431cab5aa8c9ce9833ed : SUCCESS in 1h 23m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ddad86ce0efa4da4a65d3294c0c96a5e : SUCCESS in 46m 11s","accounts_in_message":[],"_revision_number":1},{"id":"153428458daa0e95da81047054e0b328760b5e79","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-26 18:23:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa699c6f97ca7ee8be66a16ca9a90a0327f49a43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 18:32:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed06dcff9a8840f580df704422802cfa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cf0e57f0b71454391288bacc35eea57 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b1ee471d2bd458aa6d6edbc3419186c : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/230f42d332364b5cb1d8ee2bcd807151 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"caabbe280ed493ff6ba8799075236a22898aba2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 20:35:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7031a75b0dee4a75a433b32c98dc5018\n\n- grenade https://zuul.opendev.org/t/openstack/build/6841e0ecccca4f7caef89478e855daac : SUCCESS in 40m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3617b3c4844f425388280fb199ba5750 : SUCCESS in 1h 33m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90b74f66739741fdbd927b28d28db687 : SUCCESS in 44m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03e145a9582e4ae29451a3312ae89cca : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58c5a3e9626446379a6d6cb3b08a4039 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd54a49742be42d384b3b6b9fe02a7f0 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2c4b34edc1a4244ac9be55e5ebbc4ac : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6050606e75a5459ab4719ca2a01ca6b7 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/038e2c1258bb4af48cd6194fc97a3db2 : SUCCESS in 15m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6de88a78b7ba42859e2fe970e9b10104 : SUCCESS in 11m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/def6ac5a2ee342619496efdcf64b6aff : SUCCESS in 10m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0a0e174359a34773b1ca8c575046f758 : SUCCESS in 11m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c9ecf7eccb5d4d83b9d30e3738595a4b : TIMED_OUT in 2h 03m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21aa4d9bec744fb590df5420ced573ff : SUCCESS in 1h 52m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0de0c42a6b70441b991c9af476c2004f : FAILURE in 1h 14m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/89f1762e196e416787e9b3e613488a0d : SUCCESS in 1h 21m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/19518683780b4d45976165bdb0d86314 : TIMED_OUT in 2h 04m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d5cb5d627a4249e9a0e344a92528f4bd : SUCCESS in 43m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56aeae983b0a454b8e1491c742798649 : FAILURE in 1h 58m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3cb9c790fb404bb49dac2e190a216632 : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d4be49a72e09cfe690934e9ed14b79a772ba991a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-28 01:00:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c0f33511d2c6b365d42178e4fa2a30b4af0f9f75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 01:06:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00eb463caf904d519060c25ad63c482d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3488ddf5016d4c6498631ffc8981325c : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54e39c9b8bc541fa9a975a0e5f5d61fe : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4e2ff7081f24aa3bb0e576246ff463a : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c2712b109251e2eb886ef7af685a81b6d5fa194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 03:02:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c15e215421d944719d87669584d8e7ec\n\n- grenade https://zuul.opendev.org/t/openstack/build/8e84eb3aee59425495582994ba5cf7ec : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c54139ff7064955a93d6ac22674b26b : SUCCESS in 1h 25m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5606f73f1649423883a4f500962afd41 : SUCCESS in 1h 03m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/500dcf821dca426c8ff3dd95f60f7e29 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f667c745a934139a7ab11e7a72ac23b : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a8951a04a9c4f94af69c1a435c5217c : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e0f6ca05e084c18928c85dba2fad99a : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f2d8fe6c2942d1a2ca3bbf7f63f297 : SUCCESS in 6m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/afa371e9537d477a9627c1fd87b1a5aa : SUCCESS in 12m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/189d13df00d44a7cb8fcd5992cdf3707 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3d319796875440e2934315b6c62b6299 : SUCCESS in 8m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4bacb13c07684c98bd1c1a0c9c8d0939 : SUCCESS in 9m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eddde1f521af484091028d0b51f41ebc : SUCCESS in 1h 53m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b09b2e12b08e4bef98cba786026ae030 : SUCCESS in 1h 55m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d339a1413d64187a4a5c2bc2905c363 : SUCCESS in 1h 41m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/71ce08d6d11b487b9bfbc0749568d3a6 : SUCCESS in 1h 28m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c34fe011b56e44548b3275d843a23f97 : SUCCESS in 1h 48m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/641a818ae9254ad6b53d9e219e92da35 : SUCCESS in 48m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1bfc9de007874392a8750da84782bb63 : SUCCESS in 1h 35m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5319db6cbc2c477ca5b4753fb5639700 : SUCCESS in 29m 06s","accounts_in_message":[],"_revision_number":1},{"id":"3ed24c783c5b7b7c696ba979f05f80c9c45f4db8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-06-28 14:06:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22b981de34ce650eb0fef331f4c054414a567630","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-06-28 14:12:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6cf6b3d6ceddb339b7e9bb37909502ad682c7e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 14:13:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1379380c8ee36d4e2be4f405d357b2b4b7921b4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 19:28:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54220de174bc4c18b3b1764b0726721f\n\n- grenade https://zuul.opendev.org/t/openstack/build/8a18c7fd91e3412fa6b4240e206a6962 : SUCCESS in 1h 07m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/885c10536d7040a39ebe087f2ba3d969 : SUCCESS in 1h 39m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/282085dd244e4e2bbc932f02e4fec2e2 : SUCCESS in 1h 14m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c427f4fc9be94cfc83bdad6650e0c02c : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74ed48ebae2447d0b94ab4a506987018 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96969bd4e69b4ffd927620f81b4b6ddc : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2583a9661b2947eaadee771c883f82cd : SUCCESS in 7m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3408ae6b9148dd9e7068a39fbbf70f : SUCCESS in 8m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f8ca68468f434ff2b259fad94d5f675d : SUCCESS in 9m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da7524edbbc7401585f79282e81a6740 : SUCCESS in 9m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/41e82982a00247419db29bbaae2ad207 : FAILURE in 2h 02m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/02f150b378304313ab056b73c445a848 : FAILURE in 1h 49m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/17874df3a1e146918ac9125ae13ab8eb : SUCCESS in 1h 54m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/859f0e5ebb434553935e855b4d590529 : SUCCESS in 1h 13m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/21a38bb12f1c4ca88f2d00d4192f5037 : SUCCESS in 1h 56m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1553a3de9ba0422f8f71c3d821aba7a6 : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":1},{"id":"987e58eaabc5c188c4c4c4213b645e682a5a787e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-06-28 21:29:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"93d2260f528aa310428a3fece8ec4b4aa39a5be4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 21:40:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e9bc4b747984de781a5760248ef265c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3dff52c5d02d430d87b9e2a49b64b05b : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ea8e5226ee345d883a461f1aa7fe81c : SUCCESS in 9m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dae6b69c099a41f78ca318303c308be8 : SUCCESS in 8m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea911adb52f5d32bc864811e1d184ec78585d0e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-28 23:51:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbdd6e99371b469cbf5ef66cf704d36d\n\n- grenade https://zuul.opendev.org/t/openstack/build/6d109aeb839148ca9013089022411e86 : SUCCESS in 59m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea94474aa5924ad4bbe428730d573cc4 : SUCCESS in 1h 45m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/327d6db6ab1e4fa3ba050e65c80bb61b : SUCCESS in 1h 09m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a3deb40efce40f5b76384caf66f430a : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16f9d5a5a25d434b92cf9c7403f696f4 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27c143aee78f42088bd265d981e2ee6b : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/767882f552b34aeba285257af14c958b : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7562f4c6b69c4f6e9dce59aed534eb6c : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8fa1a10da93a44bd8e7c00f56e562063 : SUCCESS in 11m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2f290d41e2f24af38c00f396014ee413 : SUCCESS in 8m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c4b5c19659f4061abed944ace7800a1 : SUCCESS in 7m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/12f2ef205f194224bd6503f4f84b43d5 : SUCCESS in 8m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/69ec523d3b6c424890c5e1604edf1be0 : SUCCESS in 1h 31m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9cc76c38ec1942ddb420e7cd2530da2e : SUCCESS in 1h 43m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/215d5899a4334d4291ce5fc0350e1d89 : SUCCESS in 1h 20m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1c1233f100274822adbc17af9e8ca1ac : SUCCESS in 1h 51m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5e85586183a046f481c5a4b5a480e597 : FAILURE in 2h 00m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9a768c5418d49138e046b0646520eda : SUCCESS in 1h 01m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6d7c998645bf48ab81f68ac63359bd42 : SUCCESS in 2h 15m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/69e3094d03e64461968a84c04964837f : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"4e75a701da86af64e17fc5004f2f4cef87fa8ce8","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-06-29 11:26:49.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bfcc11562982dbde3a52ab0e56494d00803c3ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 11:35:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941e2f709a3e4389b7a6a5aeb7027f7b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f3408366adc41afa85b0a82a804b7bb : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a08bd23fe9e4e33a74db7079f18991a : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59749bed350c403580dd9636fd0debe5 : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"828216e790f8a1b384d8723ab4517920b65b7398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 14:08:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/56a54505f6824f75a8f6dd257358c274\n\n- grenade https://zuul.opendev.org/t/openstack/build/9824b43a87344517a8f0464f03a15415 : SUCCESS in 53m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a64c179432642ba9d637df173477438 : SUCCESS in 1h 22m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcb248c09c8545259e54f82b906cf911 : SUCCESS in 1h 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16532531b93e497f846bb2bd44fd67bb : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27dc98844d25428bbbb7445d406679fa : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0169a9ec81414fd896f418b5a87be7f9 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce955a9f0bc9489486ddb1888fda1a1d : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47866feaaad645f7bdda99fd8ab8ca1f : SUCCESS in 6m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/917be6ed082e416b87fb4607cf9049f6 : SUCCESS in 12m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35f1a2b7da1b4963a9992aa4b7a07d97 : SUCCESS in 10m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/705c9edc86924b32893c6b8885dc34d6 : SUCCESS in 8m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/55d5afc4796c4395bc3f517589076bc3 : SUCCESS in 9m 58s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/22e8f5782f4d4ebb80532f38ed5f4ad6 : FAILURE in 1h 53m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a43b721081b8499f803a38e5eafaf4c2 : SUCCESS in 1h 43m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0131ff40c7d049c3aa333fc9397e3f8c : FAILURE in 1h 23m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3ae5b303c1d549b9ab768d41993c63b3 : FAILURE in 1h 24m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e6bdbec6357247c6ba071a6deb4e6ba8 : FAILURE in 1h 36m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2f2e94efa95444891cd78f974ed5911 : SUCCESS in 45m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/33c8a66f83cc47e3a67aa39158548098 : SUCCESS in 2h 05m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/85690c3fa92d452da93544db93dfbe14 : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":1},{"id":"8538ad462a6dc6c0d9c5b4fd4f2052dc2cefff31","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-03 18:39:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b1cd458258b07001c4520e00e3d66e86422993f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 18:46:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb9fdd1761214038a8a278152cdf722b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02cf59fd20ca4a5db94aefc0286c3726 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7eb5c180d004fc991d2687e1039311f : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/989e25f0a0f3485f80210f933d30bcc3 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"655dc1cfa143bc0a603a3b9a3165b0045ed710fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 20:45:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4073e5c56f4f4418bdc21a191ba2049a\n\n- grenade https://zuul.opendev.org/t/openstack/build/8d92d5348a8c4264bad78ff9cc012e5d : SUCCESS in 59m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b314c3c47a32464c8e78a1dbae416bef : SUCCESS in 1h 21m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/122de14c15494364b669ced4efaa5193 : SUCCESS in 51m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d34eb51111143cdb53be0580b03576d : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce49fdecacff450b8293eff501302479 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b0e7865052b496b85e85d1645e76804 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a6a7bb719d44740ba16f54eb07eaaac : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7ba81c34f27484eaa8e92d03ca8bd43 : SUCCESS in 6m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2263bf954d144cfaac5530f844df655c : SUCCESS in 14m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/911e5b5b91ff444c9e138d694682bea5 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/22b5111bbb5f4f1dab7fe49e045617ea : SUCCESS in 8m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa65f267e7ef4cc9985d162f1af01311 : SUCCESS in 8m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d9131df1b643417cae3a2e5a7d5e1a96 : SUCCESS in 2h 00m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/78631f5866de43778c3cd6062932cd7a : SUCCESS in 1h 25m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d6648cd5622457593beb47c77406629 : SUCCESS in 1h 07m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d4bb08514bf40d9846ee75a5a3b046d : SUCCESS in 1h 45m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/639bfc734a58459ba5f8d7bc35bb7f7c : SUCCESS in 1h 46m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8dfe5b68d6b457980c1ecc0340407f7 : SUCCESS in 36m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8960f0aa375a46bdb738f5e195a26401 : SUCCESS in 1h 24m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d90ab2966d5436e9e66e7a7873b8ce7 : SUCCESS in 38m 07s","accounts_in_message":[],"_revision_number":1},{"id":"98ab7faffeb4bddc378611af72b343bd9bfef03d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 20:45:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"19b449c0dc0fbd3e7b026f6bd146d3d85249e858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 23:24:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6dde1bc271d94308b448e3950c54a735\n\n- grenade https://zuul.opendev.org/t/openstack/build/9fe2783272b945a8a9e252fd288d8d55 : SUCCESS in 51m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/145f8ae21278422dbceb2e1f4d17e93d : SUCCESS in 1h 29m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf3ac04f84804fa5ab6f9cb5985cc19c : SUCCESS in 52m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df29289b8e2d4f73911cc869991fde8c : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19f4e90aca6b41148bff56fbdf7c94fa : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5cc1562bf8842e69c9976df24aeea90 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f108a13b0bab48b6adfb2a5ae122110c : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/753f5dcf77e4462bb4f5b19379e12190 : SUCCESS in 6m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9fd513356174ddbb5f6f0b7cfd60c88 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a77ed5a6c09d4daaaba4c81122cf2c89 : SUCCESS in 9m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0c7bde08c8fa4cd38970ee2a563ab16a : SUCCESS in 1h 49m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b85b62872d3c4c73898036802a0c29c3 : TIMED_OUT in 2h 13m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/36a214cf98354faa8764cf7ded299eaa : SUCCESS in 1h 50m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b491166d273b4ccba2b115791b8c9f24 : SUCCESS in 1h 04m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a876aae6bb244cd3919070dd1a55ca04 : TIMED_OUT in 2h 35m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ce9e74e2b37648eb8027c04a2ebb8d72 : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":1},{"id":"0d1bfff78e5ccf5b26b6446ef20117da63957452","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-04 13:51:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd2d80023d82e2f4b070c3b157d32e7da64362ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 13:59:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6644c3e3a3743ca8fb16c7ebcba3efc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7ee38d80d4648668e143c47c23581ff : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f741494c80a47ad853d8aac5b3f1bea : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c54e3c1a2f9f45248c12afb90a6b926d : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc81878055d809bbc8cc64989da7f419b9d5cdff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-04 15:42:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1db8f2ea062c42ef9475077cbfa114b8\n\n- grenade https://zuul.opendev.org/t/openstack/build/fecbbb54fac248fcb5384fe042273155 : SUCCESS in 44m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/027d92c93db541e783bca8afcf9bc2db : SUCCESS in 1h 34m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5088ac58d394a7cbe18757d58eb71f5 : SUCCESS in 1h 02m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d34fd39867144729b3884a6ed4f4f6c7 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/def648fdf69042f49a52df525ac397a5 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a6cf85db7644bdfa125e08e81e505a1 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e10aa5d4bc504e09a69790fcbbb96544 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8727aabcbf3e49c38d3e1f96ccff3a46 : SUCCESS in 5m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7725a00bbbae4e4f8ef8f62fdf5dfd15 : SUCCESS in 15m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2387f1d3413a4b75b514792fd91cdacf : FAILURE in 16m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4bb529b44bb4a4cadd9e07dea6c7c77 : TIMED_OUT in 30m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9d6f704f80b84d3f8941144d12a49ad8 : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/add6cc40e75c4fd6a8854c93686159c4 : FAILURE in 1h 48m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fc5251e357824ca19ca320a1f8518ecf : SUCCESS in 1h 05m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6e250e5a8d5e451b98bd5f8fcd23b474 : SUCCESS in 1h 41m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/18492b148ec64c41bf38bca317acb0b7 : SUCCESS in 1h 13m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/945cc04de4334eeebd1d322054551b9c : SUCCESS in 1h 46m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdb2568280204da593675bb26761f447 : SUCCESS in 50m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e493c54ce8624085b6d442b1407f4a53 : SUCCESS in 1h 49m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7cdca6298097443d801006e8a66d5ed1 : SUCCESS in 24m 41s","accounts_in_message":[],"_revision_number":1},{"id":"4aba09f75d3f258a2b6a1be2ea288860abc362bf","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-05 00:34:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"99e18b22b86be46287ab9cd40682864cc9819312","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 00:42:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebb309b275244d39a888a52d554628f4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5394100c73eb4389bd82a824befa94b1 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0be3b3045a5b4fb6b09fbbacad8c0038 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d88413dfe2ec4bb58ab47d39d5044c84 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5f463820ac94659d33a99b2a285af7cb9dd504ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 02:43:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd8334ec27d440f6913bd42ddc70e2a4\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c07860f858a44d0a63570a9fdcdb877 : SUCCESS in 1h 01m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9423626c0a3948c0b5e359c464834700 : SUCCESS in 1h 24m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/813da7936fbe44c6bc7b86874075f787 : SUCCESS in 52m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/813f24c22e9c4feeab75b2d2e6118978 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17e9c6aa071143e79370a47ad5d88a69 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e15d854735044f1580036e4ac1930e58 : POST_FAILURE in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b72d433bec8340b496a76dd70fc7f6be : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bb00d840a5c4467ad3776a1dbbb5325 : SUCCESS in 8m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/41fce010bb6443118fc633d2229160f2 : SUCCESS in 11m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d234732032694143955ba4f592988178 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f71d2530eb864544b2d8d853568a7538 : SUCCESS in 11m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7aa6476fa5eb473fa6f2ada4923f519f : SUCCESS in 8m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c9c2f636ae6402fab7bffb9896f4966 : SUCCESS in 1h 50m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fff738b4f26742e2990f417087ff5bad : SUCCESS in 1h 17m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a652bb99c0b7461db8bb261fc3672924 : SUCCESS in 1h 15m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/646651424946480287b636bcb22fb811 : SUCCESS in 1h 46m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0c1c05f849c543fda9e9a464cbf4e715 : SUCCESS in 1h 52m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf1fd1e0ffad4c48bcb021af411482b6 : SUCCESS in 52m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7e160f5ed6f7478092e87f19f2c0ec67 : SUCCESS in 2h 03m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ed2e52635744f93938c00c92da46747 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5dd9ee82721051c9e1e363cc884f28833aabbc81","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-07-05 05:02:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02a25a59e4338fb41cd4cccae277167013164dfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 05:18:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f967e3535484d2a8005a633a972f06f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f3293e4167344aa8e4716dbe107101e : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c7c8effbd584a41abae97f4df589d96 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3e95d4bb543d414a951de0b045953479 : SUCCESS in 15m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b07e32c5a913429a0a2181c38ca6573f24a4b5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 07:19:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d11c86ae41d24189be9c58f1ed7224a2\n\n- grenade https://zuul.opendev.org/t/openstack/build/ea2a0d848bae41bb815c8bd56cf3ae8b : POST_FAILURE in 57m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4dfd02ea1b6a405694cc8ecb5da91660 : POST_FAILURE in 1h 25m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db136d23a65c4296a5379056efc65045 : SUCCESS in 1h 00m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a306eb21d794ca8a2b0a7d043b2cf66 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c0e2d2ac1d24e308c44097c04ddebf5 : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6626f31e410f4a739ba2de4af73509ae : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5244002a56af4faaa403d6853719341a : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917cb4e686ef41f38b3a6349218a66f2 : POST_FAILURE in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/96daae155c604ca69f8ca71a0fcdfb77 : SUCCESS in 13m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/421308816ef342c6ab64333695c31e71 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/329e5e3e976e4b9c8e16fa60d8bd32df : SUCCESS in 8m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/92feb30d6cd04adf8eb724fdb79e32e9 : SUCCESS in 7m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d6f2878ac5f845ada264a184f106e825 : SUCCESS in 1h 22m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d1026933f0ba408fa2edfa4042e063f2 : POST_FAILURE in 1h 45m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/75512655fabb424c9838717a0c1cd068 : SUCCESS in 1h 36m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ac2abf607af64a3c958189481aa6f4ce : POST_FAILURE in 1h 34m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d43ff4cb4ba41998f303d21314b0149 : TIMED_OUT in 2h 10m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a349e6e4d44141e989f09a4271b249a8 : SUCCESS in 43m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e4335261b5d402e89726b3f6a7b0736 : POST_FAILURE in 1h 52m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/608b6a8114194663bfadf61d2e9a422e : POST_FAILURE in 22m 48s","accounts_in_message":[],"_revision_number":1},{"id":"22593727d2ab68cf3888ccf803e9e9663a21a6eb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-07-05 07:26:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d96afad26584cd10df7606d27f70b8ce9fafd65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 07:40:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79de2f163f244294ad3f98f97de28194\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8469bafb06d245da90db2040b3d708d7 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64effc766037411c86dbe844cbc5b930 : POST_FAILURE in 4m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d9d74736373c4947826e5c6e239184c9 : POST_FAILURE in 5m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8622e442cab56f2ba94864460a93e3ed571cb78f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 09:35:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e2513f09ea99483c928a3692442ffccb\n\n- grenade https://zuul.opendev.org/t/openstack/build/891031531ede46a785b2f75658e4a71e : SUCCESS in 52m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99fe900566b3415786a765bba0b65bc9 : SUCCESS in 1h 38m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a105fad7914541d3a46f99f9e8be116d : SUCCESS in 53m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ed8d88f3dc4f1a812c06ee6662dd4c : POST_FAILURE in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9bb3b157084e9b9a24db77aaa00735 : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d352e7c273345e88ac5f79c0a79c16f : POST_FAILURE in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c968289cfc5f498d8a842c9e6fa78831 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125ce5d883334f568d7c6630a3212b35 : SUCCESS in 8m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1c67d4a3c7e841808800c3f9ed1546ce : POST_FAILURE in 12m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3db677efef3b402a978974ab15a5121c : SUCCESS in 9m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/551960c963494b2d9e2809ee1cc2d804 : SUCCESS in 8m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6f26e9abd64c4c5591726584fca919bf : POST_FAILURE in 9m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/38d90881d908446da673d5648a03fb61 : TIMED_OUT in 2h 03m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/52330ff22ff14dcca8c2e6924100d1f3 : SUCCESS in 1h 30m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7639cab562f641d397ac3e7b4f6afeeb : SUCCESS in 1h 52m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1f0a9cd0978247178f4fc67f0f4e10eb : SUCCESS in 1h 41m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aac58e36257140618cb335261e51cedc : SUCCESS in 1h 54m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6a62f053a40473b84c98cbaff6ed27f : SUCCESS in 1h 02m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4b873477bfeb4e3db1f23067aacc4e57 : SUCCESS in 1h 49m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0801b8625aa43d48e6a624d58c22be6 : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":1},{"id":"509661c7cbea2d92ad311ef254d42e95b696a169","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-05 10:08:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d2c458b70baab544348f6587164cf54f34a1fb9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 10:16:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4411b37b235405e9e92857db5809620\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7489496d99b842b19092dd8cf94ee31d : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9dfb0c1ccf354ec0954ad49d2795a32f : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1043bb50fce14b43a17ad8a28cd70dbe : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adac4827be8d61aedffbb41abc3306fafd1560bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 12:42:08.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa08293644564dd0b3bba79d1e67f8fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/84f481a32a6042cd840d479337f7f87c : SUCCESS in 49m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/866b90f316a04f198e3b37c252c43b91 : SUCCESS in 1h 50m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0e606e8644b741339b32400c808556b1 : SUCCESS in 1h 07m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5d0ffb2534f4acb8601588452b04f40 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/411eb720c7b74036b70067f039c1d3a5 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/339feb7f55a04b44aa133bbc4307b2e0 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7530215cc37e43c489797369369149cb : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb71c883cc4b43a88136e960c115b3db : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4d65a75da9f44632896c1897e1f766b7 : SUCCESS in 16m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/73036f8173464c27b36f08528dc00f9c : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/434505ad81d840ee8fe993425f1155be : SUCCESS in 10m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/64dbfa0c74fd4abd9c0d3c15d9e961c9 : SUCCESS in 9m 47s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/184829f80a844fc0a1aa4b4b78606bb1 : TIMED_OUT in 2h 03m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd63c28ce06e4b5d88af26a5fe945369 : SUCCESS in 1h 35m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ea5339f766b84fc9908c60f645b11dff : SUCCESS in 1h 45m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bb9168e1fbb14fa6b5f58a1299d6f080 : SUCCESS in 1h 58m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fa5a540c75cf404aa9cba608c53f8d56 : FAILURE in 2h 03m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6720f6d5d16546d8b98f17b5b17697c3 : SUCCESS in 1h 12m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a6943d4965dd46578b7d3b5fc2510bab : SUCCESS in 2h 22m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/acf2c3c689e54e31822897e09f2d164e : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":1},{"id":"c427edce8e7f4b6f8ca2887b3399ba118ae75e23","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-06 14:01:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"25eec21e0e07165d9edb6d5bae8af7f47f431e6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 14:09:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d65b778d72d6465c80d2f7ce678d8f81\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/98916d16e4f24b91a3f07b1047665732 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1655a1e86cd549d19ed4a10f89450fbd : NODE_FAILURE Node request 300-0021628441 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1cca6fde17c84836a28b1ccfc113a5dd : NODE_FAILURE Node request 300-0021628442 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1872a8b74f3a03340b3d22ae7f8f757dbdf31aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 17:12:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a40890e2a95a4ba7a87b097b2ccc5cd0\n\n- grenade https://zuul.opendev.org/t/openstack/build/3aba314d8040440d9efc805acdb11635 : SUCCESS in 53m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1e5cf01f39041fbaac074bf1cbd5afc : FAILURE in 1h 52m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ecf0d71f7154d8ca5fbfb3f36a08bfe : SUCCESS in 48m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cbf0444e4744d9db9d6de9ddadf64ad : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f08e6386de3e457a99c6519977962c8b : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3935121d2b6d44d5b5ca1b22fd146b59 : SUCCESS in 6m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3cd2f39b85dc45e6a00749499da1156a : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdee659b90274a56992ab56545a4e32a : SUCCESS in 5m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5462ef555fe442bfa3acf531f71748ad : SUCCESS in 12m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40ec86979b8d45458b4a252b4866a6cb : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d9497425f7184facb0afe6d3f9e86467 : SUCCESS in 11m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa96cf5658ad42598f5e4b904283f036 : SUCCESS in 8m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a6b50db7993846a4969ce183f2cb731e : TIMED_OUT in 2h 03m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2006626e6e8248a9ac7319f17d45807c : SUCCESS in 1h 43m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/56454a89de3b4c1091759b467813b51e : SUCCESS in 1h 54m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cd7b5ec0d05c4417934fcb743ff8765c : SUCCESS in 1h 22m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5b0dfa1dfbd3476b84b4b72e12872bcb : TIMED_OUT in 2h 04m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58bf25245d87481b835e0af514e39d76 : SUCCESS in 1h 19m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/df40cb25f51549a4b89c182f87cfc0e6 : SUCCESS in 2h 28m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ccfea8d6e3af4cbd94556d6f89b02a23 : SUCCESS in 44m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4da082b2f8d3528452d1576ee4bd1948f0113ee2","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-07 14:15:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"961c0a3256fe5bcc08527ec692623e2b67680c80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 14:22:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27c6ad8978ab41a4a1dbd63a1f8986ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2e97267be0547a3a570f2b4ca0ef633 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de3a207ec3f947cd83e2f46be9b94338 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/87d18efeca4341e8b455111a39d0ede8 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f2958c527b009b819a06490939ef272f9a2d73ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 16:34:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/31449d5da4ad4ca6a78e7ef0dcad1220\n\n- grenade https://zuul.opendev.org/t/openstack/build/f772bab0e7fa47d4bfe99365ecefaee7 : SUCCESS in 1h 06m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68632ca3f6db43fbb9b302eca5bfcb1a : SUCCESS in 1h 33m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87f0394d25b54cdaaacd3f957f1c98fb : SUCCESS in 53m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7644a2397504446fac59386afdf9a291 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c76ba149529472c8ebd8f73193b6d28 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/424fa5791ec9438988bea34035caab61 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a734091afba94d628ada17edfab80b5b : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb5880db8514d78a5b12b99fef75077 : SUCCESS in 7m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3a01d7fe1b644b6bb696e5173001499c : SUCCESS in 12m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4262f551b4534b908ae9789ec73aeec0 : SUCCESS in 10m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0412d7be02bf407aa6881ac417f5dcc5 : SUCCESS in 9m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/86396410e67c4189bdc51aadb0547df6 : SUCCESS in 10m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dce5968c4384459293cb02e11f632a1d : TIMED_OUT in 2h 04m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/40c1761d69cd4a9d91cf63f3e4cd6f6f : SUCCESS in 1h 59m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2bdf200f9711499784322f81d5bb0b23 : SUCCESS in 1h 54m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7332d630d55b4cc5a3b0c71bde3b66b8 : SUCCESS in 1h 36m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69fbcbe481d1462c937019232c326e08 : FAILURE in 1h 56m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc3dacd3c8ee4739af9104b5cf54d730 : SUCCESS in 55m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73800efa29fb43129ed59b56798365da : SUCCESS in 2h 19m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/434180726cdc4382b1f67ea119f9b444 : SUCCESS in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"484212fbb75cae251be27cd32eab011e0105b54b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-10 04:26:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e800b71882a029642354549d6aff27644b0be35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 04:33:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7be059496c84905b60a1550b8d55de5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d93221f32cf7496d833e5c1e998473d0 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2800a32046ee4e94b9091b90dd84502b : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d442d61b784d4903ac1b0f507e9c9bed : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e6691ea83b877814d65eb36184aa3a2d3a5c1f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 06:30:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/877bf2516b1845028628a6b6864b82c3\n\n- grenade https://zuul.opendev.org/t/openstack/build/3b66c7787fef4adc8bfd530d05843261 : SUCCESS in 39m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b39d14df3663469280d50f060176e811 : SUCCESS in 1h 05m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab818b0a84f147f1a31e00a3c465fa41 : SUCCESS in 43m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/800b98da073a490ca16a82dbfcfb879c : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acda5181baa64a0e858ce00e6b5b447f : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cfc465ce902459cafb9ab95b05e102a : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20f8fe9c1f34409c80495686d74c318f : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a723ded2cf34cffa3d6bad21233bf60 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a71d60d740c74bd1b3004169a8b7d150 : SUCCESS in 9m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac1bfbec52454f7f92739d7b6050009e : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ccb17eddcdb84752a6a641251af2c95d : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a4234acfdd454015a7a3815a35c044a3 : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/92ed1fbd1bbd435d97c75c9591881b96 : SUCCESS in 1h 31m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bd960bcf81c2486daaa01414f82af23c : SUCCESS in 1h 48m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cfe71e93c3af4b02abcf90937f85e583 : SUCCESS in 1h 58m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/63a0c5f3d63a4d64bafb81eaee46f531 : SUCCESS in 1h 23m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c045d87d793d4f79be9e53fdb4566d34 : SUCCESS in 1h 24m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/70f9cff2ee2540c5977537da60cc7315 : SUCCESS in 41m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3ee2eb9cb7134d5682526944f22d1ad1 : SUCCESS in 1h 31m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e4a9d5168ab741718a1b8d721a7cec58 : SUCCESS in 21m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8efacbbe4edddd15a45a21d27f5ce980b73b3d6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 06:31:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"78925495a86c4f212b0ed2607528a6b5c0d895d2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 08:20:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd2619f8bd114da7a540520846f2c216\n\n- grenade https://zuul.opendev.org/t/openstack/build/d924a0acf1124aed9ec5bcdee3d71f78 : SUCCESS in 58m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5af18ec5d3f4ddea8be741d42bdbcae : SUCCESS in 1h 09m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed6b48a8560e47caabb78c592e929f97 : SUCCESS in 38m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64ef16a1ca34e639325987207f9cce4 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c262240c1a849f2bfca12559f8caee3 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/518dd5ea3c354dae94bd3d0b92d0392c : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7f4632c656445e69f357d08656e2e01 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc91136ca3e647e49027314582b772b4 : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e0fd52dfeff49ac9e5fad22f13a7f07 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0983b7edfc7c4daa826ed8a92e2d1102 : SUCCESS in 7m 51s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d5d20a1f419a4409b718ef25144240f2 : SUCCESS in 1h 13m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/40cbcdfdd07a4907a7df4986fa47c671 : SUCCESS in 1h 09m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7f53fd57a0e4426eaca3926e6e79e93d : SUCCESS in 1h 44m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/927d2b39abd842a3a099e2a6c3f917dd : SUCCESS in 53m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bc1b183fae5b4186b9d9238c08114431 : SUCCESS in 1h 26m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/39515d4bfbc64b85a306a7538eaca545 : SUCCESS in 30m 22s","accounts_in_message":[],"_revision_number":1},{"id":"69391f1362e8bc0b8560f35c8118addade977d1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 08:20:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e513c2d1a29dad8209caa3297dd7c100b8c89deb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-10 08:21:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7e987c377d94108952a69f0b153f0dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a212011b67b4f03b5d8beffc62dee98 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11061d5b5a11fec0fb2ca6d55b0f116b8c716928","revisions":{"11061d5b5a11fec0fb2ca6d55b0f116b8c716928":{"kind":"REWORK","_number":1,"created":"2023-06-22 13:50:25.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/54/886754/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/54/886754/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/886754/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/886754/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/54/886754/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/54/886754/1"}}},"commit":{"parents":[{"commit":"d47803613d501b18f9fa4f6aa3c899bd16f9ddde","subject":"Handle case for multiple cirros images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d47803613d501b18f9fa4f6aa3c899bd16f9ddde"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-22 13:49:49.000000000","tz":120},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-06-22 13:49:49.000000000","tz":120},"subject":"Remove the last occurrence of six.add_metaclass","message":"Remove the last occurrence of six.add_metaclass\n\nChange-Id: Ide428268f52527972165ab59ac6e3da0688968f4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11061d5b5a11fec0fb2ca6d55b0f116b8c716928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11061d5b5a11fec0fb2ca6d55b0f116b8c716928"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
