)]}'
{"id":"openstack%2Fglance~886947","triplet_id":"openstack%2Fglance~master~If94d7d1fbec07a49222b65aa3e530748a95eaf5b","project":"openstack/glance","branch":"master","topic":"New-Location-Apis","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-30 19:13:04.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-25 13:52:58.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-07-30 19:13:04.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-30 18:14:37.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"If94d7d1fbec07a49222b65aa3e530748a95eaf5b","subject":"Add functional tests for new add-location API","status":"MERGED","created":"2023-06-26 08:11:30.000000000","updated":"2024-07-30 19:14:01.000000000","submitted":"2024-07-30 19:13:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":110,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"886947-New-Location-Apis","meta_rev_id":"2d2b626bc0398c9ae3dc6b27dfc542c588d2cddc","_number":886947,"virtual_id_number":886947,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-30 19:13:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-25 13:52:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-07-30 18:14:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-07-30 18:14:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 08:17:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-26 09:34:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-27 14:24:38.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-08-04 07:54:47.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-08-17 18:47:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-08-28 06:25:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"df3f86b080acaa8fcbe218bc14680e4393dcba2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-06-26 08:11:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08b1126f3471468c5fffeb21cd7e9af6f184ee48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 08:17:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5e3f266d624ba2858a6b496c5d6c4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7f5b65449c5a48d0a494d20bc9eeeaf0 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ca951c76e8e4646908098d270c88223 : NODE_FAILURE Node request 300-0021541853 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4862c42807d44ad7b67e2a319968c9eb : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6c2293343c36dc94cb35271a843019f28174e44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 09:34:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1b518c5ec2044e8b3d0763406f3c3b4\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5ff643b5f4d45ce9eb9b400d7c66fb2 : SUCCESS in 1h 22m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ebbc83c0bc44028c156ff11d3562e5 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf89629f6b45437fb6edf61cde8a6684 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b61293761fc440a6a4e820fe2d01e3c9 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e911da16722440f80c042672d74aa4d : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3435b8404baa401f8a04d1ba3b334228 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8ba2883a21fb41399b39848104b41cb5 : SUCCESS in 13m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/45cf36661ff0404ebaf679d34fae5c70 : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7aae6a85567241618d15a56f4036bdd0 : SUCCESS in 9m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ee289e62dc94480a88b47c7934d9f381 : SUCCESS in 8m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f8ffea1ffb244e239d167dd3d98b1ae6 : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8a532aee9fbb768314abb47b113b053fa07568b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-20 07:59:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a6ee98a3f4f082b85369675cc3425aa1781571dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 08:11:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/571077df15b44cf5948f4a08ee0c5430\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a044f70c62b4e4eb51a1278ab287c6d : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd9fd10a6c444ce3b4e4fe0399a89757 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c945f2c2208c4416bc1165084a9aaf3d : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9e40a877fd1c25f70cbcc5b1c2b518f1f9df5048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 09:29:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158a2c846b3f4b54b230402503b0d874\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c7beb413c954ecc8a1dd2b60fe87f49 : SUCCESS in 1h 23m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a96ef83cac6406aa3910078d6c36c29 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86b20310731f46e69c008095a6ed0a2f : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b081b43524c8471a84970e118d7433f6 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b956a272d4b14dbbac8603cd30174c01 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9311c7075bbd4ec997237b6877348b82 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/623cc8eed6b74ae99057b042d626858e : SUCCESS in 14m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62a6a7ae49fd4222b9c42bf20967720a : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9a780d92b0164286a7a9df7d92d94ad1 : SUCCESS in 10m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9267ee23a6e343c1ab3632808a156bd4 : SUCCESS in 10m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/96dfb8f766c04135a6839e836cbc9286 : SUCCESS in 50m 02s","accounts_in_message":[],"_revision_number":2},{"id":"7b375f6e65d5ad5e3a1abb386a233bffa006c744","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-07-27 14:24:38.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"857bd594e331aab6dade1e6e4d62d72e566f3a33","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-28 10:22:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0f8cae7fbc5ae67b6c3e7f8887f34707a42b5bee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 10:30:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b2a0f716194e818a1ef49faba08398\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b1b52672b7ea47b6830e3183086ee2ab : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb5b8e1ae7f64e87a5a9c74d827b81c5 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c529a0ac29e493bbf4c24de1ec59060 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"93ed959790276acf75ab7f34663887b9514f3fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 13:53:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e266e5745a743808b5314572c406a64\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc3beaeeb0e5474c803be2c0c176e4e4 : SUCCESS in 58m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8a0f0227bd44b8a68c03dec9e6be34 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/277d698691ca4f9cacc22d69fff61802 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44305d314a3f4513a00ecfd90e03a09e : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aac9f8873d9243e290aa14db0475bfc6 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410729c2716748f6995bbae6751675e9 : SUCCESS in 7m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/64d310dd86174924848e7ba17245b36f : SUCCESS in 15m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a8bf6465a7048b18560d820f2d7f505 : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dcee193a9d804061ab80290e91b695f0 : SUCCESS in 9m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f2c585758d6b4646a7b2664db88494c7 : SUCCESS in 7m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9e456a79a6e244fca1110f5616fbc066 : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":3},{"id":"63f6359e8ae5289e6103570be73efa348f36c081","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-02 14:05:00.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3766fc0c6f3fdd1a3a7b4da28953addeb786cc6a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 14:10:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d13a518b3979403dad3cd19d8c6cea6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/00abb5252c7b4fcca9b6fd0275c6ed2d : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37f1caf44697474389f3310e8762c197 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95df3af0c4204756b443eedeb86272fb : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bb5f4d84add9571d43bcda9dc4d32a9d7c749844","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-02 14:13:23.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"8540a755e147324e8ff3d86ded933d4cd07f96a4","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-02 14:13:51.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8578802c2b450cd71f9dc34ee93e9d0ecff98b90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 14:24:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6a07ab8146e4644b122f6c93d7e2205\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20a9504c0b684f4cba15f147b78f867a : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56fc40f0545b43bf9e68f1d600ac4cc6 : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/15730961e94749d1a18cc04ea9eecc4c : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a8c91638a2f24d2415b33112be6a756fe75db82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 16:54:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e02883324e44b24921faf5c0d3addc2\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3dbd825c513443939e7e518ef275bb1d : SUCCESS in 1h 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2d8121bbec14aaea77cc9455275f3eb : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a07f0e1dd66a465ea556197e99a7cf52 : SUCCESS in 7m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a29f5d18eef7422786789c3c3d8cf24e : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/315f312996654c31b4840ab039bd87ab : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af5ae6b0a7384deeaf65f977aaccf7ca : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4b5bd6bd4e8a4e29b8213bf6c33b5ad6 : SUCCESS in 14m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3267b37f3a74ea89979484aed32c847 : SUCCESS in 11m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe1737c4f7a44e5cb562f396ade27088 : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/05ef59cad83a4aeda9d0b89e89c9afbd : SUCCESS in 7m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d7d2bda6bc14fb6a6349e06e231ea6b : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":5},{"id":"bc1a1315dd63a4f40e0dc6adc9bec7b9003dbc92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-03 06:58:13.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fd0946fbeefce0ebfdf68fa23a6b3e73a6f36eb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 07:04:13.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30a69c8b669d4b75b1b253fee2555a8b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4b87c77210d4ab8bcc9c881fc8fd75a : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4eb5ea8e450b424490555a422e0f29d7 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39bd1e946abb48dfb6372d44e41ea597 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"fb295c82fb198317d689dd02361a7dc17de76f78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 07:51:05.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cad8dca693044455b25068dca4c77274\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4743e86a8f7c465ba1a91ff87d9ebde6 : FAILURE in 50m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d2b93ba02c4ecab9cd87939c28bae3 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1df13055d0141e7a90b8f77bf619a39 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d30a9857e4a4815bbd117e737b4e4fa : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34d116b7672646adb273a5398d7ccab7 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c777395cc824613aea32f290ba86abb : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/372be20569474a888e4845e7268a8c1f : SUCCESS in 15m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4625dc96b8884cb498027d56dcd682a5 : FAILURE in 10m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2d56ed1198a94c31afe93b8e9ba413ec : SUCCESS in 10m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4cd8273e3ff741bf9f5e671691ebf8d0 : SUCCESS in 9m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c55a5f3ae8834c05902a8049ee2cabf3 : SUCCESS in 35m 08s","accounts_in_message":[],"_revision_number":6},{"id":"cb0f64f107a322015f6c248d98cd4ca1150499fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-03 10:29:48.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5113acfab4eb578762b2ac850b5d04cd5819b96c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 10:40:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e2cad81910f4f92b501bb60000266f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7998c1cabac741998b9ecfdbd92b1b47 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b01ee0e1aac0469fbb1e602fcf69cff3 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f011e29e2de644c6825d393c339b0a56 : SUCCESS in 9m 03s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d2153abdbfba33a6969c59924c9df44fbe5c7f3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 11:30:09.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2daa2bcf00874cdf92fe9b96ed4e6350\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f235dca1d9b4b91a656bbf0683dd80b : SUCCESS in 58m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8735049f8c7348618d1781822df8ba13 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c3c6e925504c86a289bcd39e5d32be : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1af11a5a27a42a390910ae3dd1b0f3c : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e71776a424c4e92a36b5a49cd955f00 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b35ed69bac2f45b28a31ab2f66460e6c : SUCCESS in 5m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/21ed8d4d864b4c9f817c701673881120 : SUCCESS in 14m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e19eb7d9f6184768a683f0e282f9d423 : SUCCESS in 9m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1147e9041bd4c09be238e6ead9073eb : SUCCESS in 11m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d992370cf07b4efe8e969134baa49f96 : SUCCESS in 7m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5a8eac6e435d4412a12fe0fc6bfdeadf : SUCCESS in 46m 30s","accounts_in_message":[],"_revision_number":7},{"id":"e716a5102a3b6dc20009c15409d1496c40cafce3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-04 07:54:47.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"9582ecb38496d71100a777be2dad80cfe58fabc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-09 11:13:25.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d5dd763c5c811057352b79d17140344e77312134","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-09 11:17:00.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2d71df1f02419bcef3b9fd30c998e2922099aa9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:26:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb2a6c5e05e44e6fac491ef461671eff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23d487a925dc4cf0a5b270c7f76a50d9 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2615c3a8cb1f44a9a449d507b1238c3a : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b763394037f4e2888438e697078ce5d : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c9c69f53a860c076f80f8f6c354714e12bdd8166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-09 11:32:19.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"5d76f851bf3cebef1b7c2b83a2bffa34be012c90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:53:52.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/873c99690f70455bb414c51987c01f4c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f868da3be9ac4bf180f25408dbfaf0fc : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70a6ceaa6e3f4487b5907cdb3a5553e9 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/168f4c4fd1054120aa4207733470ab26 : SUCCESS in 8m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2a8f56411077e37c8384ebcfe6166beaebf5e3d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 12:21:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2cea51ac7e444fa9b9696eeb193dda3\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f2609698a3740c1aa5613ed3303b4a9 : SUCCESS in 46m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd66653b9e14f9795de69c6abda848d : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f78fbbcc095344619c830f7ef54c53da : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3243336529864a2f924fdaa3ef4d0785 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6f94e345a8c49108e867abfa55a1da4 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a86a4a9a131472eb08dfd8e6e4673bb : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bbaec10dcd4d48a3b7e404b5f3b36f20 : SUCCESS in 13m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/975c9fb87c9a466b87f6c9a108a7a4e4 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1690c70cb90f426b876ece54f1105c75 : SUCCESS in 9m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2af1787522e04fc0ad7dbc36b5d69121 : SUCCESS in 12m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/776fef6b50d4465aaf5d9ccfc2665e17 : SUCCESS in 38m 42s","accounts_in_message":[],"_revision_number":9},{"id":"7c031cb73fff5f8edf485e1b599b01c16bd6286b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-14 11:50:37.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"25552571c0fe15923753d73fa79f3f1661544c90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 12:02:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a75a2d6b60a47fba0572f4b085aa5e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bc25a99a141442084558099ae2ceb42 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe06c91d5f3640f48c9694572da9ee0a : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fce11cf345a841fa9234996a4ab950ee : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"eaacb5337d5cea5c7787626241066c6d7592939d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 12:49:18.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6148dadf16f484fabdffd139ab12a30\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/013a9138a92d4a3898e692910646d47b : SUCCESS in 57m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a86d8672964a27910da20dc1901268 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7aea3c5f140d42d9abe16b5363fce1e3 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/457da8b3daca47128ca1c489f0184a9d : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ee2ee6423cf4fba9db6adce9f8a96ac : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844750d457af4abd93b1830345a646a4 : SUCCESS in 6m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/18ae8c7eb1d44126958105e0f86582dd : SUCCESS in 13m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6c32c5e5a06f4a3aa61105e052781024 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2fa2a5b278142a4ba9b551ffc86e69b : SUCCESS in 9m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/40301a8317a34ed2b74b313714e996ad : SUCCESS in 8m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a71e806daa6e42ebb0d4c736408670cb : POST_FAILURE in 48m 17s","accounts_in_message":[],"_revision_number":10},{"id":"17ff69fcefe14541da1988e4ab6714caa3f757c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-16 09:29:45.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"870557c73ae742e3d48104358ecc39a5728a0100","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:42:56.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17330fe385a44c3e87c35417b06e191f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5d421d0fd4a4ccdba73cfc020547f56 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7de0e7b6c7ed4a92a765edbfd6907475 : SUCCESS in 10m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3305da38350546fb84dcdb7442edeb9e : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3dd9cc9cb7e0f11dcb3a388d863c47627c384805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 10:36:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6788db07b2549919c0a16764261992b\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56132e9e28af48caa50236e4f776ed5f : SUCCESS in 1h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd253499481b43ec9eabca97d7645c9c : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba260fe911f44e7190e794fad452f8c7 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/675ef7329bd049d78fb4bdc775d04c9b : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79908a1b64c54e3f85b6adb8ebd9424c : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3485a50cd6452db87e3e2e7c46f59e : SUCCESS in 5m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ad491b6697784477b20183bb37204cd8 : SUCCESS in 15m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c6e4dd6b8664d9691d0f6e3ae2176a8 : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a584adb92e644ed182d728b9c2373062 : SUCCESS in 9m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e814be9aa92b429090d12ed5a46b79ea : SUCCESS in 8m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/119b4248131349dbb0899c1db03afc98 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":11},{"id":"bc5fdbd3e9b095403a3b8c1237a1d75b5b7bc392","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-08-17 18:47:52.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"bba7669befa03081d0cc374101737bcb5bfb9495","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-20 12:19:00.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"847d6f280d57339de9f92105fe8114f892481d49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 12:31:38.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aaaf642fb8341c58f8dcfc467d59354\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fae428fee7a1414ca6f95155936f78fc : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/465a6926dfe44ddcb7853f92f9d94df1 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e250827910214abe9f8ed0f5ab58deae : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ad887e36cd3e4578963429c01930c36a : RETRY_LIMIT in 1m 38s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"dd8a32d944a04f7ea57b534829538ba618e49e05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 13:11:39.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6857cea02f574743a5ae521a8bca671f\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/501ff30a98874c7eb74059cfb02571e0 : SUCCESS in 38m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e6c066ed5a4192915e25b191595727 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a477ea168ef34dc8a88fa99aeab7ed40 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea86273815da4aee82aff2ae9c92b56d : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c478637be64441c974d9b8021e5dd2e : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7cf58ae557049b6b57a6ce8b2010aa5 : RETRY_LIMIT in 2m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07589b8958054243a5ebd32ec7c37e09 : SUCCESS in 6m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/34dce86cbe014aa3b6256752566bb251 : SUCCESS in 13m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1970b957e8bd4f0698d0637224258203 : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08d9f096f9e14d388ec96a6f4f88f6ea : SUCCESS in 8m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/789df0fc67754f0094b2d5984e0c0f20 : SUCCESS in 10m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4cde30be948f4a8ab9b5406f2adea291 : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":12},{"id":"ae9f7170f9e017ec23247d1794a18c758d0fb7c1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-21 06:24:18.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"14f461bb480237ad5453ece76cc9642caccd53cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 11:40:44.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"62d34dfc3d74d326f53cf2ef30a510173e0ef5bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 12:04:03.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a59b868e24a3406a8dbef456ad6cad6c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b623b112f849420d8af9bc7ecdff1758 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce0ab7f0a6304493a1dde0fa04cfe0bd : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e40fdb62d29d46cda38ceaa3d69c324e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fbddb2f7976543f287aa1a9e0ea9ad20 : RETRY_LIMIT in 2m 45s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"92cd841a432bcebcb2da3feaed46141598dce3da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 12:40:17.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf8230216904c25964ef1602dc215a2\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8f30e03db9247dabfe67bc11815e20e : SUCCESS in 56m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3973e810ea0e4855a3a23ba5fbc773f4 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa31deb4a734439db14404a1159bd297 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10174cb196df42e98c3dba1f785cfd47 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbb7531a1a2b493cb2c35b15b2ff164f : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae48698331044240adb244c92ebe8a1e : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f25b0d3c4c4308a501abe55f8fb257 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/04a97eaa04eb4706ad4120bba0d70262 : SUCCESS in 15m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/71cbbafdbc7e4cab829532ded52d1b36 : SUCCESS in 9m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fbe17ec94ab64e66adc5f09789b36af3 : SUCCESS in 9m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8b97e20dc99442deb37046a1433e27b3 : SUCCESS in 9m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48a20bf6bf3b47e6a698e608982f2352 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":13},{"id":"f6efc89338f89e5c7bcdd457ed0e452cae032c93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 15:02:21.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"c6a37cbfad99537d41130dd4ef3fc3a1267d2ed8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 15:14:24.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e7eda34c4314556a1726e4cd4e86dbf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb56709a13c24927b16aa7b75eebedfe : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4c8f9439c954b2b9f254a1feb285430 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f211736605464ab9b5a0ad12ff29b2c5 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3acbc5aed62b47ea96f6ddadc280f2fa : RETRY_LIMIT in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a393e3408df5f108fdd306347d63f3a11d76644d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 15:56:35.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec79e1ca16594f1fb734b7fa5e8a789e\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e858502580c4132aaf118af1a529173 : SUCCESS in 51m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2a357dd1adc413680c9736a999f5f37 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b918eab3bc664cbfa116fb830616b52f : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e242e90e1fe409683161e9a7e7deea2 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4bc5b330a1084264b6e79d97416f222e : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d97aea48227412aa60a5ab1c6d413f4 : RETRY_LIMIT in 1m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e50d68a59b44f9c89e3815d8ce9faa6 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/113cf18b0a614fd8980ded8361571a95 : SUCCESS in 12m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30b66b7eb4a04b5c97765c41f6c3ff2c : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/19dd865d4b4f4d18affed47e75045fca : SUCCESS in 8m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0eb6563fb44e4fc886dd052031631d75 : SUCCESS in 10m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f917369014f64facb277204236dc17ae : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":14},{"id":"02b706099f8052d6bc93dfcde61f86dfd8316e1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-25 17:41:42.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"732d1dd722d71953572617f23c2ad497522cdef6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 17:55:00.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f74245aecdd4294a73f565ec140a98d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccd7498ca68443e9bc5bc409703bb8f4 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e2768932fc6476380f5ac924e80e4a9 : SUCCESS in 11m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3710abd9b2774693b28578b3fb1f351d : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/67364719f24945319d339e2ee7d245db : RETRY_LIMIT in 1m 58s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"5103fb3992cf72b43166326819f77cf31145070a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 18:50:03.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad4e62c5024e4efda4e661ed3ef88c3d\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1c2b02dffc048c6a2a0a0a7943a16a1 : SUCCESS in 1h 02m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f24abc3be5f44ecb89ff10ffa9b92c3e : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/386a842058744961974a8a19105221b1 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/740a206ae53b406b98e9dfb4ff0e8383 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4174f419178d40f9bed089944f0f11df : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/000e8ad78da549cb939536b7a91dd857 : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c3ea68b823a41ecacb3195e262a5645 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0338e14ab28b45ddbd0230b331c860fc : SUCCESS in 13m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5853b7a311714a28a0c4f6b73da5f9d8 : SUCCESS in 9m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d019b9038144af999b965e8549be9fe : FAILURE in 10m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/937acdd6ad964d7c97b8c92ef3cdef97 : SUCCESS in 7m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c56c53e85704497854676fcbab111e9 : SUCCESS in 25m 14s","accounts_in_message":[],"_revision_number":15},{"id":"34bd850a033e495f26a51dffbac7b87c88fd57e5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-28 05:45:27.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"c017a75918213e9c576f991f87d2bdea58847bfc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 06:02:21.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/757f665e35dc48bf81ede2d492339408\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a86b8b2a2f14a35bdcdcafc87957e15 : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b561dec99e0464d99a37822e1c1a81c : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d11adbb935e142c3bdf7611215727f70 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ee3e15f342f457faab9608eaee449a1 : RETRY_LIMIT in 1m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"f2592f22ec64fcb0d0bc74e1e04dd92f342b4028","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-28 06:25:37.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"53cea9aecd00a66fbb7ccca08dde4f264455e955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 06:48:10.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ee661a95ab448a1884526a7b39a2c05\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d90400853fbe4a679a4453dd8dd99b65 : SUCCESS in 1h 01m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e5ea9821dd4327af43d00aacacea12 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83051fc051604b189b24e2499b96907e : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce916f377d4a421ea33355194d66addf : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afca68b68f874dbd9adec8b84991b325 : SUCCESS in 4m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c882aff532b6468eb424d572b459935b : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcb6421403c443e90c38468e3142d15 : SUCCESS in 7m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0fa35c06ff62405f9a3caf59f60c89e7 : SUCCESS in 13m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec1995b98765431a99d24a0dea5bf84b : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84b3b79c3d354495a41715a210df38fb : SUCCESS in 7m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ec553bfcc2df4f5098ee59fa2ea00cb1 : SUCCESS in 8m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a8e12d8231f9419cb3f3cf83ac83ea2d : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":15},{"id":"971f70bc0fe9dac137d652e2c0614e3d3caa3cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-28 12:23:57.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"0c4f5787050f2a8e689bda95a18f3a0309191622","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 12:36:06.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95820963f37d42fc9b79334b13dabf1f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/853a7ebb8d084f9b8bdfe8a5f47e92b4 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb8dd04c0c3d4f49a1fd0ea7ef1da022 : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/894efa7d8d43468399c889b57f26adf8 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f78376d05a2245f7afab1a7136a4d6a8 : RETRY_LIMIT in 1m 50s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"863b1cb253169f6b63d0c38535fe75559ef561a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 13:36:11.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/080414abc3914a879be66fb5263686f4\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9c5d10f2fbd42a9a6ae4f90cc4ee94c : SUCCESS in 1h 02m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fec3ba9eaf16433092e70ed62f285818 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0948a886aa8d4f5cab0b10654959a641 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc43aeecf87a460f807d53b2a493d0b3 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a953da150f044b378379675c1281b450 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc53cfe4da5f41dba073e45e598b184e : RETRY_LIMIT in 1m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7caba64dba074b09bd679cb1404d73ae : SUCCESS in 5m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ff664bb53ac146d0b5eef360648e5ae0 : SUCCESS in 12m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6b738cd989d44bcea122f4fdca88910c : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c4e714fd86c4bda98c9c8847f1e07f7 : SUCCESS in 7m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4657297f3a484986959dde97eac41d1a : SUCCESS in 8m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c54030b33c4468b91915a161d62680a : SUCCESS in 27m 06s","accounts_in_message":[],"_revision_number":16},{"id":"803fc569ad51fd0aade564539802f28034ab4af6","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-28 14:36:20.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"30379a24795676242b97268f3a36045fe8300d5c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-28 15:23:36.000000000","message":"Patch Set 16: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"db69fba112259bd11b396b293ac8b53cbb204e2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 08:58:04.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"b28bb4a1ac8b89dac4ae983987c3e51fbacb4f00","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 09:07:18.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"e5c05c5d9617f0c5cbfa9a6e50e4fa68f5109509","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 09:10:01.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02833e926ad6451bb7b2491c70bfc88b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bc44abca7cc46a4b1f8da4a9b0c9573 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50340cad53784e9fa55f8d2d8fed7d6b : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7eed57f012245d59c4d92681fb5539c : SUCCESS in 9m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b2f46e0019247a0b19919e8a86b1696 : RETRY_LIMIT in 1m 46s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"83f886a2be3eb45d9139bc8a250199c5c0d413af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 10:11:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19ce4363d36f40af96dcc27342e6706e\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7da37afc1b7d43f88fe0458d73949e15 : SUCCESS in 1h 07m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f51c2ac001441fc959aefdd73a4cf2e : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/376746819de04357973ab775cbdcbba8 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8696a41afb7455d9843418b655ce4c2 : SUCCESS in 6m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e7d2d66fa3d4ad2862e99ea00afdafb : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02da7a0e50d14761b1b2c210fbaa85d8 : RETRY_LIMIT in 2m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15160cc6fc6645328a48ea486b2b0b78 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/dbfcde0e2ce04ad29f65c7f8c95c7794 : SUCCESS in 14m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2f077751483d442ebf60f856a88df141 : SUCCESS in 9m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d118ce6e7ff243958685725ab9bade30 : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a340da757dbb4f938c5652f77842ebb6 : SUCCESS in 8m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/77d29825188c4b9796e7775c25cad4f5 : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":17},{"id":"049036d30c856617e786715e0daf81a4bccb0a68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 14:21:04.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"2b423d63b191615607074e7b42d54ff172c07851","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:38:39.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e63f0a8a11a2490898173143947dc296\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0c7ed9093e64faf87c6723895159e96 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c33ad2c01b74e7f8e4687f25f6c3dc2 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eab35eab4ef24c359c11fc6f44393f00 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/91b891f59f38464c9276cdfb65e799d8 : RETRY_LIMIT in 1m 55s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0216cb27bcf941e634e35ab7a98639afe002cda3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 15:39:57.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aff231e0e01f430ab415c6bac5359603\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d99b1f88c3c4e5d8fa9360dcf582079 : SUCCESS in 1h 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f347983a7824a309ea9314497ee3735 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5f4aa14e3eb413cb5f82352764ae823 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8d06067fd0b4602919db307d2efcc53 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcd867a4e05e42cfbf1e4398c1828381 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc11083b291041439bfae41bf1afe5bb : RETRY_LIMIT in 2m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840e79c9bd284376b352aeff3702ce06 : SUCCESS in 6m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c0c96dfc81ab4b118d2c8db221ec2503 : SUCCESS in 14m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7efb2428792f4a059cb49bfdfb301410 : SUCCESS in 9m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76e2332e24c445438452bc5bcc1e719e : SUCCESS in 8m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/95ce93b0649b43949f0e94f481636fd8 : SUCCESS in 10m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b89508dcf654a1f84f47e6ed9b80a4d : SUCCESS in 44m 59s","accounts_in_message":[],"_revision_number":18},{"id":"ef48f74f6249528232c34c1c0e6351661e23839c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-29 17:12:06.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"fbd99a350f5d481f037eea052ac8990795fc1ae0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 18:40:41.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"7cccb56dfb027084e31c2a3466c6b15733052a20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 18:55:39.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01fedd3420ae482bb1a68ccd4718d3ff\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d644ae47c1a943359a7e33646b76f11a : FAILURE in 12m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ef22e6c6945419294207ef578b4c24a : FAILURE in 4m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a2226418d8a24e14bfccaae298529ffe : FAILURE in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2c084120ea5e4ddd98f8906405885450 : RETRY_LIMIT in 1m 36s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5a849b7ea0a0fb130bcb455cc27d226dd87aa6f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:36:24.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f03e9ab20cd54cbd83ae205f79123e28\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/882ba67f07b043da9de891298fd100d0 : SUCCESS in 53m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff11ffc387c446cb8abbb1194b27e25d : FAILURE in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28856238eb5b4259bc03392ecde043b9 : FAILURE in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a318b29a0c24782abab404d64ec3ffa : FAILURE in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a89708781fc8447686d37eab6a2a7317 : FAILURE in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d0a6ae3c83240a2a2d56a5205004801 : RETRY_LIMIT in 2m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba6046edcc841e9aaa03cd6c8b4cda9 : FAILURE in 3m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1d3e4f7d27994a33a0bc858c2fe0b463 : SUCCESS in 12m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ab75fe467424713a3a5ea5e691246b3 : FAILURE in 4m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7779cac2f10c4e5e8a015d0af9b6cd0a : FAILURE in 5m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9fc6d10223564e93bb68d8c251f094af : FAILURE in 5m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c48c0e4ab4734b5681e569b8ebae030f : SUCCESS in 39m 26s","accounts_in_message":[],"_revision_number":19},{"id":"24837e6831466b64fcccd83d3c6185fe12e9f99a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-30 08:44:31.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"4f241daccf5742b65a9dcbad5d4c0a282264f345","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 08:56:20.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10e91e56a61e41348a2151aadb027175\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f706d00ccb74a7586e7d69773f8f468 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac80ac3afd984678af32f90a50284eae : SUCCESS in 9m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d0697439d14f484391208c5052d877ae : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cae8cd8bc6f94d46ae8457adc8800526 : RETRY_LIMIT in 1m 43s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"db3476193cc87040a78bb58f4bb9063c3fd891e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 09:52:59.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ab230cdb97e43d28f5feb289a9f9bca\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5726cc52cb3c43eda2c00f71b26e6468 : SUCCESS in 1h 03m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1693786a7374804a09954ff876029a5 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d878245c686e4e638d8cbd993d411a4f : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d08c623ad7346e89f094a8cf5acc2a0 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/320a2175ea174ae7a1ec3a5b6956c4de : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70dedc66b87a43a6886a7a1c68ccee2b : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5222f9455a3447eaa9bdbf016aa7e8ba : SUCCESS in 5m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2d1d1cf21875463193bc298fa98130d3 : SUCCESS in 14m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f894fcb93464811981a7feabb8a0635 : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3203c0065a31424c93d5af695660c1a9 : SUCCESS in 8m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4d50a4c649824d55ae193ab2a8b366f8 : SUCCESS in 8m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0ab586d30b1245d4adf14cf3bcc85075 : SUCCESS in 30m 13s","accounts_in_message":[],"_revision_number":20},{"id":"86c6f28535e0007a373ef06dbcf8656cefcd57fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-30 10:40:22.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"7ff0ccfdc4b1eb2b50907f19437ee48c74137083","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-30 10:43:14.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"2e23af4407aa2e4bb645ef70d9040707a3531428","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:57:10.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/703b38ce7eb24db1a3f17482a71737a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb18005488a743cf95a3283e5a7fcb8a : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/23eddfc1f22242058abdbb1de80845d9 : SUCCESS in 12m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c6938e7a11de4b65946ec91e9ea5cee6 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ec9b1a4faab4515a15483f8ee243c3d : RETRY_LIMIT in 1m 43s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"d4d09b9b6343f86460a3c6086b9c78415e34647a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:48:49.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de50ce9b9c8f4cc4b4455678abf1e2c0\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52ef3937bd034d7b9ea2071f46982245 : SUCCESS in 1h 07m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47923fa1b5a54df585549377e90c63fa : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dd32e3ac0a8c46cea1b77d327a4d7e34 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec8a98fcdcc54348a6365d191db8237c : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e716f4b2db342909e272c044ee00cf0 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/450b5faea78e4adcb2765a46b5e427cf : RETRY_LIMIT in 2m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb31d7dbb284c958c3906a6a057a7f2 : SUCCESS in 8m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/09b1162a53604c4daa02000efe9cc003 : SUCCESS in 13m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63c8e5b03a6c49618d3dc8787a93c42f : SUCCESS in 9m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd7a5ae6f6874df2a9bacf2a7904cd45 : SUCCESS in 9m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ffe1e6fe38f64fd0acad3668f3e806d2 : SUCCESS in 8m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c08142464a6f43cbab01edecfa870805 : SUCCESS in 26m 37s","accounts_in_message":[],"_revision_number":21},{"id":"25c8c7451e1d0d819f413a831ab98de3a58e1563","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-30 15:49:29.000000000","message":"Patch Set 21: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"8e5dca49441d9f5a7edfff6f09c60a3bbc6cf02a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-31 07:57:48.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"4b69613cf67481023c9c20e6ac3521eafce13922","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 08:10:48.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59b76593edae4f8bb509cda981ba0f8f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/820d3eed947c4f4f9379f6643fb3ce77 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c4bff970a534cce973a389ab3178aaa : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a085c7b645494620af81e37a78cd18e9 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9737cc8e0164753aec1296ed020d3a3 : RETRY_LIMIT in 1m 35s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"5fe4fecc489b963eb268afaf0b5bffd53d54d249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 09:07:43.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1778bfe52a854b6c97ac2ff6bbd14e5e\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0249d2c8f7bc4272b9b4d48093a0283a : SUCCESS in 1h 04m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/880cb75aae1e43d89eb97771514cbef1 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4971311a34bc4c19a4d119ecfcfbecf1 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9475362c9bc485fa344e41e47e2e06a : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aea2f0720b4748eaa0c50b84ccfe079f : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d347db0760448cda27ac4296a70310e : RETRY_LIMIT in 2m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b0aeb7355945b69159e68fea8284d4 : SUCCESS in 7m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f327cb36503a4d70bfa3368bf331651a : SUCCESS in 16m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4251785d70154f659c2cb91b72b2228d : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24c9dd5ecdc84f47b3e6996d4187133d : SUCCESS in 8m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/76bae3cbfbad40ffb9e0dbeefdd05b1d : SUCCESS in 9m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49a1832c67bf4fbca646853117f450aa : SUCCESS in 38m 55s","accounts_in_message":[],"_revision_number":22},{"id":"a7901d826a46924be39f1ecf77b1c60629044ef3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-09-05 07:25:17.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"04d8602fee2dbecb8b44e837c1259a479702ca7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 07:39:38.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c22473a72aa422cb5a530d5609651bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76f8e2e5cf2c4d108b392ef9b896c4a5 : SUCCESS in 11m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5b494321a0b462eb12329ed0c904ac6 : SUCCESS in 11m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/773f412c0cfd431e9269a49010cd8ee4 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8a1cc524e3394d72a3298de515829070 : SUCCESS in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"c89d767ead7fc7ddbd12aa433d55968ef011d11a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 08:22:41.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0618f2891146769fc6bc6f81fb124a\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7bc39c4a8cb6407f891353614dc5ab0e : SUCCESS in 54m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dada93b354a49efa60b282e6961a52a : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74a63b75a8ef4bf58cad41ca5369754e : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d53100f08ac04621ba0bd78b5a0073f2 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9faea7f49c9a47dc9953a76b4a5f91dc : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/381c97b305a84dac998d901e6bc3d3ec : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b613ae5ad58e43d0bcc02f1a3f5fc754 : SUCCESS in 6m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ca1e944b054e427089d6d3382cdfb73e : SUCCESS in 13m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f54b3b4faee4d66ac71cf986f3b1e27 : SUCCESS in 7m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/efc71d1446ad406daa3cd3bf01ee2e54 : SUCCESS in 8m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6f67bda16d37472994018e7dd8a4d9da : SUCCESS in 9m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e16f7e9e90264a9b9f83635d6b4ff638 : SUCCESS in 39m 41s","accounts_in_message":[],"_revision_number":23},{"id":"ce1bfe4c1cb3bc74ea041925607df8e5c3ced623","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-19 07:06:53.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"1c226a3ef545a5d78bd771a5dc89a98cb2843196","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 07:27:21.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83904c90ab3d4e89a284d40d40c84e39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b7fd272ef6c48d1b74e97f19b3c95fa : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/91c6af50857443669fbeb69e6651c3c4 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/06039bf3be6a420287bf9dc9ed46973b : SUCCESS in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"20037ce7d5b6baa72288ba3ade2100b07fd5ba7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 08:16:21.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/998602c5c7de4c9f8be9b681157aa748\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e48239ad2a3845ec9c543e6b289e48b8 : SUCCESS in 1h 05m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63ea99cd030640bdb9625c67ffdda73b : SUCCESS in 1h 04m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/596e357923654919bf0eecabb10a16a8 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b64fb68874f4239bb57ec7799f102c6 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a6e0b385d7341aa9787f2730d87320c : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95fdde64dfd3444b9fdf90a8acf7c237 : SUCCESS in 6m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/51106489c8284ea293b00ee321a38a2f : SUCCESS in 13m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3660bf26accd47b7ad1c2534ddc3b22b : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98900c4d12fa4224b54717e125dc9708 : SUCCESS in 9m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5f09326e467245bb8879ea829a5f4b8c : SUCCESS in 8m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7f4630fb51c04ec9a7815dda4320e9cd : SUCCESS in 10m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/45e7957187a14470996bd8a7edff3b1b : SUCCESS in 39m 57s","accounts_in_message":[],"_revision_number":24},{"id":"df0d61bb2b088f9c2426597b38ba9757d7c4656a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-31 09:27:15.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"c229922f374f0e867d96f79ef094c53b334be19c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-31 09:28:11.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"57d8a597f138cd533606cbbf94f3fed2dc71ad3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:35:46.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b794cd07ca64959848c84be1f70f845\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2aed1590595744bca083271ac518ff54 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9e83d1c54ec5459fb19d0b95b50f4c1d : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2590ffb5d6ce44988dea39c7b193c7da : SUCCESS in 6m 13s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9ed0307aab6f1ace90d4daed9833c52d06703b30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 10:28:34.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0892332b602642ae8324af48dfbc7bd7\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b789a93c071048ddb07599f9647640bc : SUCCESS in 50m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/442a6116a5e04edb97f45df90237ed79 : SUCCESS in 59m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1268e9e79d74b039d421cb86b255976 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db2cd0cbc6454bc1bf9ca21fb42357d0 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0a9197d97094e48b0650e215a892d6a : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2014b5a701465f90b6681ec5e33a8f : SUCCESS in 8m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/49684c58f7774aa4b69f3f221539cc54 : SUCCESS in 12m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8551465f86c4c7da40d4dc93af60791 : SUCCESS in 9m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6449353279684819a8c4518d6c73bf67 : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c88b8d8c39d94c4ba3bd7b8722732577 : FAILURE in 9m 54s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f047c41800184181b4aa18e721c57ce1 : FAILURE in 12m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/736658273edc42dba2bcf7febdf93ed8 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":25},{"id":"a04d1fe35ef2a1b1ed4542fa2323945981060972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-31 12:48:40.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"b39ecf38c4f00d3ac1ff50c3c09674422a6d4a98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 13:04:55.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e2e5741cee94fae8f8e3dab2b5ab0d8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa26cb30a5c64899a468895f85c85a31 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bac02e6406744ce2bf2645c5263b6d4d : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/17a9c8f7a78a4784b4c0c5d17d3d4b94 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"e8e8a0237812cb6004146ef7d53777d315a0385c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 13:52:48.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6821ea70d057450faf4a32e08fcf157b\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a494005462a04fd7a19fe41b8ea510ff : SUCCESS in 49m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/337caafa76f14b51b36694f67ee30041 : SUCCESS in 1h 01m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcad6c57922a47cbbfc5e368e691f6e4 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e85fe17ce124e13bba1fa236c260e4c : SUCCESS in 8m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/028fa4a02bb341febdbdc48c1d0e01fe : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3c4e5ad15d45348820507ef63641aa : SUCCESS in 7m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0dfd4df6754a49ee82b58a9e9b0bb7d4 : SUCCESS in 13m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a2d48c25af2049a78858e9aa57892018 : SUCCESS in 9m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/352763faa9e5427ebbd51b04a6a13952 : FAILURE in 8m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5a796f3327b046cebd0c3fb38453e7bb : SUCCESS in 10m 09s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c8900d06105f42e3a406b91ec56028c8 : SUCCESS in 9m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0bc1eab48e0d46a382db8dfaf7da334b : SUCCESS in 30m 25s","accounts_in_message":[],"_revision_number":26},{"id":"292d9453bf685e6cacc29e6baf23821c5364b87f","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-01 14:18:29.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e579b08a1407b248f4bc108eb189562e322a9d60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 14:28:48.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f11f42a3edc43b487ba0f8d7ad186d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b6177c0d858485bbda79a10d920fba2 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/01531d243dd64df7b2e05fb6e4bec6d8 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3d60ec58be84437c9eb6b3564e32346a : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"8a94f45e9744dad9a44161e2e1d01fef7282b7f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 15:13:19.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9c9f99b4b1a44c4b14c28676533a8bb\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/59e86fff7c3446578941eb0408f5a390 : SUCCESS in 48m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90f49f3d4fd7421189df0c1814afc52d : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37f66a9358d4d04b967d14ef25dae1e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65393abe609e48f486067c58fdfd5b85 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7deb91e32a95479f8e6d9317347d5eec : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3e35433dc8434cb045e59c1707172c : SUCCESS in 5m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/64c419e438c840abb947d7728b81b9f0 : FAILURE in 10m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3c467a19d19491b8037132241aaac0f : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9cee5029c3b14f11a84d627bc198e70f : FAILURE in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c6206ba88f24d3aada426003f3859b7 : SUCCESS in 9m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/553f52d278594fd3a45bc7e21b3acab4 : SUCCESS in 10m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6587b216c2914cef8e53ab06eeaaa30e : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":26},{"id":"0c201b7c4f5b5f616f167ec1a1b3ad1caad4c44f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-02 09:35:31.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"0112cd2e395204766ffed95a0242b284df872f1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 09:46:57.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518b6a0255884f4ebb4eff918e0f4279\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e7f66e3401b4b3ca063be64a017fc18 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/efb83bc627f64aaabadb842984a1f96d : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce45714a890742fe96e4b1869dfdac2f : SUCCESS in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"131d746713c249ee94d0aeaaac25067bae40b93d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 10:46:38.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0df42ef475ba4b829995ed9d374e5db6\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0c64112ffc1943248163f0350cb22992 : SUCCESS in 1h 00m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a25530fdaaee48f29f479e6140709929 : SUCCESS in 1h 04m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95f46b28a38c44e4892b2b2cda32b995 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4689320336c4689ac44b33be07f4522 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ac575a7fd3c4d68bc3407674ca9d2b5 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb926df523c445e1bf6f711e4fc87411 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6fd36d517c1e44e1b76388f02aae3295 : SUCCESS in 13m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a7e925f833e9430b9d050d0d16e2a6b6 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7344861c82b4f7cb6a45f6746d31295 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6eb4ed8cafe6451a8906857a54540f53 : SUCCESS in 8m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/339c2a8b0b75488393aafb3b83916f92 : SUCCESS in 11m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2a91d62a92c4af397c30c0474010a47 : SUCCESS in 25m 16s","accounts_in_message":[],"_revision_number":27},{"id":"f1ff95dfd81f1ceea29b894e88e225a094745ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-08 11:23:44.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"c99d43cbccab3e14ea1686b258b3f77e2cdfad4c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:37:19.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28d9c36d946e4aa4863c7792c5963d5b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8f717299737449e80d9bcf067ba8660 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cf11ab07efbb477f894a4cc686e8a238 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2428663851d3493797deb9b173b374e0 : SUCCESS in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"c52ada618e687544d39f66a1b89b10e11898094b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 12:52:06.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bb9eefd64d04d6abb7a87bfbee690c1\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ddf691e1bfb84d819167783e6949e47c : SUCCESS in 55m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dc3ceb2e6c6c4f25a737f0c8160fdae5 : SUCCESS in 1h 01m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40ed6599e9fb4317ad52dfc00a36d2a0 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79e37cc050604e00bcf2db3e8588bca6 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/944b0fe8c51f41ab87c00a15aef21128 : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20fbb3ebecae4df4aa4288ed986bb2b5 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a30f989fa9784a6eaafce3d2d41b9f14 : SUCCESS in 14m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f34741dc7f4f4725844b6b9589817439 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/51d576e11f514994b57825fac68e3498 : FAILURE in 9m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ded5dc3006374a45939c2049653562c1 : SUCCESS in 9m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/263a3c0f29ff40ca9e4c64e8eeb314d3 : FAILURE in 6m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ebcdb066e7944f1a0b0c7a35df6d46c : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":28},{"id":"375b01d0761770434c7059af3bd96c68ece6fb29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-12-05 14:08:25.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"cf9e01378b17e3b7b0bda36b58cbc11c82c10a84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 14:16:17.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02bcea6c437142f99735c11ae3ecce0c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/61548bc0004f46bf8a617f2f5d20681b : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aac0802dc9ab4179aeb468eca336ae0b : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/413ec7b7b999495c9fb767fff6cf6d8c : SUCCESS in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"744353d5883e0f714d1eecbb6b38efe91a40fb15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 15:23:13.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d34490dcaf804536a0706c891cf856be\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b2d2710ecb924c57b0c4b8ce552a8e8f : SUCCESS in 1h 03m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ae2014bdbf340cbb04370fc2dc857bd : SUCCESS in 1h 05m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ea59712780d4ebb9107b0c7f397771e : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d1f7f349c3347b7937aded8c68feb93 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa4630d5222646939dc936187607235a : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a6039490a814c0caf3191ee0a50c586 : SUCCESS in 7m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/81f7ae5648af432c8bf4f000235f3d2d : SUCCESS in 12m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddb8724c380041b5bb6896613007ba3b : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c519b9f7e144eae867b4f1b683d74c8 : SUCCESS in 9m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/aaa41487935e4cf7bc35a8ce642230ca : SUCCESS in 7m 54s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/170ef26868734b9c81d86e5ed480aa1c : SUCCESS in 10m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de21b115e6b64f7bbbdeb5d5c34f6b73 : SUCCESS in 38m 51s","accounts_in_message":[],"_revision_number":29},{"id":"26001970d3ed0662a0794af1310adafc87df3f36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-02 12:01:29.000000000","message":"Patch Set 30: Patch Set 29 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"383dee16c9c08ada1ac45fb2190d3b229b6ac03f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 12:15:22.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/748bc93b419d448296387fc70e710b7b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6be1002fc24e4cde936dde6409931fe3 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b89b51e89dc84b4585f282652ed9f66b : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/671c8a79c31f47b88a0c67d06796242a : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"2d554cec5450b910e8c8b94fb245c5a416ceb6cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 12:46:52.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/86821708c2bf491391eec5df5aad0e5a\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6fb1e2e2fef04a9596a4cc06fafe6b08 : SUCCESS in 42m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4602814d236145179b5e66c09ebac413 : SUCCESS in 44m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351601d86416465b9742575b21fa701e : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ae9348ad8794e8b8deb3e4086179330 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/baa3824a3c2f400088cecfe2f682e410 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31eb78521d76470aa8e7d39ee88bf0c8 : SUCCESS in 7m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3306fea106ee41e5928185d3139c5859 : SUCCESS in 14m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aef926ec927c40a4a5d298eafa9faf37 : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4799f7747ebf4d2eab07cbfc3d7c8c0f : SUCCESS in 9m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7524f9cc908d49ca9e1c5d6401e98cf1 : FAILURE in 10m 19s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8db7ebbd960847d88c4e63593a0a65c2 : FAILURE in 13m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a3d6c5c3f654f4bbfd530632e226c2a : SUCCESS in 32m 35s","accounts_in_message":[],"_revision_number":30},{"id":"74e3733eada0939d96a3d97c996250a0d59ee9a2","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-03 11:38:49.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"14ba5b5cb286693333794581f62da96a6689790b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 11:51:15.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/747206a9197c46bea93e1e925de0db7e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd69c97f3e164187a7932853490c3dfe : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9124b17d80d14de88b8c62763e4802fe : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/be44c75036df44c3af3a1355a00a8c0e : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"a65e5f2aac0c6583f5244e839cb94036072bc18e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-03 12:50:41.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05e5cab2eba04fe28058e557884e4a0d\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/982f776d6a3040cf84920dc71445d347 : SUCCESS in 55m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9651691006314731b180f9e4930ac077 : SUCCESS in 1h 06m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a46ae81738e24b59b7fd3fc83e1a397d : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6fbb67fba747aeb7cfdf7e8328d9b6 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea76ea9f6f0f4890a54a0a1db6e66b9b : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d53fc80f4b34e62be8de0221fccae4c : SUCCESS in 7m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2eb31f7b7f0a47c9a4647d12c4ec927b : SUCCESS in 14m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c2ff57d684744f3aead3f5b1759df50 : SUCCESS in 9m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6d3e4d4ffb694e39ac2278b2bd273fff : SUCCESS in 8m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/85167b9352e740818631184e5782e203 : SUCCESS in 9m 42s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7e92d05b275740a6925fd4e8fc3a97fc : SUCCESS in 10m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ba702c9ca86d4b4896f93e2981c9170f : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":30},{"id":"9f3912007fb74463461e75682ff64208b7ad2d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-11 07:36:44.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"36025866d7567e1ac3567b84bbfffcf43816988f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 07:50:19.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f96935e4edea4d588101a2e55a31ad3f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/441b4c5f09fb4a7aa56f19e724a1a747 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba479043a9474bd1bc34fb60eadf272d : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3af8a2ed9bed48d78def4feab5d56677 : SUCCESS in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"c40bdfcf6ca4737f6fbd3f7bb2c2698adee385c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 08:41:18.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/778426561fd94cbdb9ac87fb6c27a747\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5c92bdb3231b44919bfa0a425678a795 : SUCCESS in 58m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65263157e79d49be877a1b88933c9f04 : SUCCESS in 35m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dbd92b1df3a48d5b215e5f698db7463 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/246a0087242e432fb343edf9ea9672e1 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/43dffaa3316b44b680a5cadf7cc3d014 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29a4b19abb76442097c5977a1e646de9 : SUCCESS in 6m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0402f1dc38ce47c799718112acd98811 : SUCCESS in 14m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85de9b25497f49e6a37a4d1e8c1210c7 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f1abb8c50784b8ca9acfb6e8febf18d : SUCCESS in 9m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fe902598db874b6b9efed559dc0068c0 : SUCCESS in 8m 44s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9056db8dc1a24ea8b2feb95b8047e9ac : SUCCESS in 11m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2be22c29bcd4d90867caab6b5a94fc5 : SUCCESS in 34m 58s","accounts_in_message":[],"_revision_number":31},{"id":"31228a42da514f00f835e7b9dd9bdf50b8342262","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-22 10:58:50.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"96ca0a86032ed9597f6650d58e93403ee1bcc35e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 11:12:40.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d4453cda2a346f2bf02e4a8ec32ecb6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c43164057074b35a5e6a97bea177dcc : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12436fd005294db592e3e48e85eec6ea : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/06bea586345940df810aa2f03564107e : SUCCESS in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"e9b624c0dd589fdc2e74a2beedec97c7a84a2c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 12:12:10.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e65ca031606649fba70022f3ac5c474c\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/81c953915dc041d3887c86551212feb6 : SUCCESS in 1h 08m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57e6e1b4d5c24e11b1449f44a7e95c08 : SUCCESS in 35m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bbdd025144c49ed8ad746a9f840a53a : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7fc5664fc644f51bf48b7a9756f4116 : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12d32e0193ee44e991ccb2d33641950e : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ddff508c12d4eed95c676057ea0b1ae : SUCCESS in 5m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0ce9f6711c6f46109f7646b09ee8287d : SUCCESS in 12m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/030af4a761a44861b2beaf34ce1df35e : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e1101fcc6af410696fc67827f7f09e0 : FAILURE in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f96e03a1ceaa49a8a2c3068f9a37d6a8 : FAILURE in 9m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/320cf700d4ea4b7e90a666d4acca1472 : SUCCESS in 11m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7d6bb491ae3f48d195d961807cbfbceb : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":32},{"id":"6183c498a1ed420ee23050e38ddc72188f960eff","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-22 13:11:33.000000000","message":"Patch Set 32:\n\n(1 comment)","accounts_in_message":[],"_revision_number":32},{"id":"bf48b8f200c6765af0899fbe6df7585c5ba50438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 14:13:35.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e27d4a995614c258c1dc876dd020af3\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/60387688135e440c9a2a0f90263d4a2e : SUCCESS in 47m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/491d713a01ea4c95a331fe0292f4f33a : SUCCESS in 35m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95c26fc7bce44343899fae450e29c579 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd6add3f02c54d5f8d0013fa391f7d08 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a7691c05ef946158d70aeb13fae190c : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f142393d924d858dfffa1416268073 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d32eaa2e6db54da5851249440f2328c0 : SUCCESS in 13m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3345b225af1449feabd83abce9f5d27b : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/587b71db13554df693621a980c13747a : SUCCESS in 9m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/08df135fa01e49459f521d8cfeff9b74 : SUCCESS in 9m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/034556c404034da680fc3e520c900388 : SUCCESS in 10m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1edb6b499dd540c0951a0f845ec27cf0 : SUCCESS in 39m 51s","accounts_in_message":[],"_revision_number":32},{"id":"32f646893e181dbeb668d669feca1b73078dfdbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 14:28:47.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05757adbaedf420fb8a146c3389729e9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9745d26c4fb4262ac026d57a900fb97 : SUCCESS in 12m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c9860418fde249f1908c352049e5fc48 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/665a98bb5e69460086e1ab831f9b48dc : SUCCESS in 7m 03s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"bf60ebda3fb3db85f337826ac66dd20ff2614d1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-23 11:15:20.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"383bbb24bdd082141e61d90b6d476eda19455ab5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:27:10.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b123c6e7d0164213be85178d816169b9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb880f4f46e94cedb515473ca6c6eca9 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d33b1efcf0074f05a23c437ca1706ae4 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/56bcde058fac4139a0ab69cad45237f5 : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"5ec338c84c6d696a1ecb22f1cf1332a5f011d866","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 12:41:14.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/018c1dd086af4b2e99e5b76ac7a7a497\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/45691d5fc9cb4bdea0af2198774ce0e3 : SUCCESS in 1h 02m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4dfb25db1ca4ae48e7ec85bc0326956 : SUCCESS in 30m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c981aae59aa48f99170535c0331d6e1 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1db93a72023f4f1990e17625ac05d0c6 : SUCCESS in 6m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0bed1b7626f49568beba6dd042d7a63 : SUCCESS in 7m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1de017cb984a889128070e725064d3 : SUCCESS in 6m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fd118e97551d49c89c67d058b5c76f9b : SUCCESS in 12m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0612c3f454e5473e8f9a6f4c0fea07f5 : FAILURE in 10m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e12b63e5698345d08b4c3e6eebca28b1 : FAILURE in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6e6b853fc4824df2ab9d410c73a56498 : SUCCESS in 8m 58s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c068517be66747009f5ece90669eed26 : SUCCESS in 13m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98ae38dfb3fd46b7a85f414dcf28bcd8 : SUCCESS in 38m 27s","accounts_in_message":[],"_revision_number":33},{"id":"56bff0a76f4f4fd0c990a3b23271c214cba95f27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-24 10:53:02.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"34932e5ef195bfff255bc3288641e52672a0f2d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 11:08:28.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3755544b03c429abf6b45b25d16b21c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f4a7d8f758764fb6acb582303c78a942 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e088f419d4cb4d4b9de2ed85e6dbcf5f : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bdf050885134257a4e2fdd58338f656 : SUCCESS in 13m 15s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"57f4a476f8233451d162418762589c1561e1ff40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 12:01:58.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ffad6378b1249049d78cfc34770af16\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9dd93cdb64034d2fa1b3394973f9a96c : SUCCESS in 1h 01m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1f193d6f4b0549e08d5fe90ec679c3a0 : SUCCESS in 37m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f0487d78a84a2993d03bbd01f91a11 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/561053476de241a693b21d4bc5eb3811 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/410eee023d1948d0ab4208bbe3bfbf9f : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/727e00b83d544ae1be15ba6df1da930b : SUCCESS in 8m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ff390b3aea7745d3b351d94492ddf395 : SUCCESS in 12m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6f2b81792c8b47449728791fda670227 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6341644a5a954c26a409c29d9a88f212 : SUCCESS in 8m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/43e377c761244ecab6973b2759687da8 : SUCCESS in 7m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/214a9afcc6a949f6a4ee8dcc777b2d66 : SUCCESS in 12m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bc3ca33be13540c4b9e5d1bb71748d54 : SUCCESS in 37m 07s","accounts_in_message":[],"_revision_number":34},{"id":"f52e43816a6f5995c37aab8a628f924254d675a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-30 07:21:39.000000000","message":"Patch Set 34: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":34},{"id":"217f93c8fc254c937b0b1885423271df06667d7b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-30 07:28:02.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"b03a729e0759e04d299f35cecb5037603f3f62f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-07 06:50:50.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"52b5ddbb4c95d50ddd09338bd326fa7d1f1a52e5","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-07 06:51:36.000000000","message":"Patch Set 35:\n\n(5 comments)","accounts_in_message":[],"_revision_number":35},{"id":"ffc23f1dd2398039183a6d99f0e6ea4980b53f5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 07:01:51.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/593ef87247284d409d2ce8f40939a3fe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41950d0d2f2f42c3b06c15c0cf5454ec : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2699f51ac8b646a59acae5642792668d : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/04b5c9452ee0411b90016ce5d1f370b8 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"2f58a243f1f8b7b305d9ef44b223ffece84b3a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:12:02.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcd1e484e79344dcbbfeaf6d427e4446\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/069fdc97b313474889448942f3e7e9ad : SUCCESS in 1h 04m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1343c41520e94bb4a3655b1e1926516b : SUCCESS in 29m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b4b53404f0545cf8d151a1438b9a1c7 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64fa3e67d22c44508b66982c249c1e7f : SUCCESS in 6m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/afc13e912e4942b3a394686b7759d297 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b15ed3770824e04a459b078068a1f3f : SUCCESS in 8m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/376a23d20ee9433c9a6b461d07d10282 : SUCCESS in 13m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be6499abe5a54f549697bbc2f765a87c : SUCCESS in 10m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/76a89660c8ac415d91a1cc51adabb376 : SUCCESS in 9m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7fde0fc4d14245e586fb1bbe6e7b68b0 : SUCCESS in 8m 11s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1a37caed18f248e983f234feb6391647 : SUCCESS in 13m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b66e91b618a44a7b970e4992e2ac1f4 : SUCCESS in 37m 56s","accounts_in_message":[],"_revision_number":35},{"id":"dc16f309b34a90c09d22be867e4141ccf7dc4493","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-13 15:27:50.000000000","message":"Patch Set 35: Code-Review+2","accounts_in_message":[],"_revision_number":35},{"id":"cf5d9e98e2607dda3cfae6c26b1b30088c161a92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-15 10:31:13.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"8091f4321d0517179be2c39be6e8e7c6804850ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 10:44:29.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b98f668af42544778acefbc06b354a37\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/497be4ac56884fd69e7c3a269cd36f20 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/579916f3076a4b88bf19883b5613acab : SUCCESS in 8m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cdc2138c0c6646599b9885fa0323a79f : SUCCESS in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"e89e4addab326cf0a9c267685e33f04df71cf9c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 12:07:30.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36a07a83c0444a98bbae3c39a6d426c7\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/280c53e2a8114a9fb0688a759824e6b3 : SUCCESS in 1h 26m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f3257d2b2184fdc906379dc2fc37b5b : SUCCESS in 34m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da788c6c6e9c49c6a0cb21f6adda9dbb : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6e9180614474b529da3dcb0263b2d77 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0a8ee1139f94f489d3b674ba2271a4d : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8cd97313a1d4c73af7eb710eaa6d24d : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/51dd67cc49f54b139f314bfd4ba41fcb : SUCCESS in 12m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2296a726a93b4efba82a827ab64c73b4 : SUCCESS in 9m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4a3f47b5341431d9a98ee836cd67600 : SUCCESS in 10m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/712b2050d4874f0c8e105dcde0d7ff57 : SUCCESS in 8m 38s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/321169ce09754b88b47fd1f0a58096bc : SUCCESS in 10m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d1a3fe4ce2b4c398878ca8da42ceaf4 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":36},{"id":"3186ebb2f72edefbda64fdb3ed08b3abc59790dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-20 10:23:20.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"db92b365d70b6e3243de28dcfe9dc74f0b1b1e0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:34:18.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0202a318cafc4cb8926120a3f3cbceb3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b727c45ccd514d19bfe98859cc64e4d2 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10b5094a82cf4d41ad32db608c658fa5 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3211539a11cc42799f8ebfed763d4b80 : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"b6f1377cb1be4d9abbf04f5779c6dcce14100656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 11:20:30.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8850ceb827a14942b571c4fd627d63c4\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0b553ece3ccd4d589cc64d277dbf0303 : SUCCESS in 54m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da523625e17f436e8dd6f719795af7c7 : SUCCESS in 48m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c3b257b119549fabf7dadeebaba1fee : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9736b58438f74f20b9e289bd12bbec54 : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/431f5f473b6c4c5cbd801ea52dda3346 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe48f8bfac0463b88a66a3df958fe89 : SUCCESS in 6m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d333d17de4b446beb7f79af15a43db1e : SUCCESS in 14m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8968d68f9a7e4c809972f157c1ed9052 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b29352f86a241d1bec92d01c5deeab5 : FAILURE in 7m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/52469eb0d609452593d278e33397c4ba : SUCCESS in 8m 45s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a496137d227a40d7b227de5501c08252 : FAILURE in 10m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a187a833d4d743439f69ea04f4ca9347 : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":37},{"id":"f085fe94b848475ba63e2fcfcb929eebb3bb2407","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-20 11:34:09.000000000","message":"Patch Set 37:\n\n(1 comment)","accounts_in_message":[],"_revision_number":37},{"id":"502cad57bc3bd13842d8b2cf019e2f77413f2fdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 11:43:22.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa5112da78e1468c9999887ffd70d05a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3022ad7256c34e0c982037012971fd8c : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a50341395cd943769b111e0b52826f2f : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/baf963c647e049adbb89cd1c9d7b31b5 : SUCCESS in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"4a9958e03b8879bd60e6d7fd7a68c1a83888fcd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 12:47:46.000000000","message":"Patch Set 37:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0fe7526918d64d689b23dc3af720d324\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/77ebc8ea85a2420b9c6c57ac9fef096e : SUCCESS in 1h 06m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e1600d6bc1f4bf2b790f00b09f9f634 : SUCCESS in 1h 04m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9686810acc674bfe896a6957ef15074b : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7c2734d78884652a6b9e5dd4f09fc0d : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/38b38eecb08f407b87c310d29f7047d9 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f107b364826a47d4be07b6cacc8c97e6 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/edb8601acc064d9197e011f5caa02bcf : POST_FAILURE in 10m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/52b06a45883f43dc8dce655a4039a696 : FAILURE in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3d5721c0291043aa8043a7bd2717f921 : SUCCESS in 7m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e45c8ab26d5b449fa7f46f74d08c56eb : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2331b972c58640a5891df8e1296b3862 : SUCCESS in 10m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e7612cc97d34594b149c7a7886b0953 : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":37},{"id":"3861cc4163a4a0f9243a083c13359661dacff3b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-21 12:59:22.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"5cd9331e98081e96fb58c49752218a6eae118e71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 13:08:27.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88241560c3d7485c9d95ab55421a87eb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/36a94af3886e4ec28404f912c372859f : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e387d4419b44901a93b728d190b286c : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/df9b807156d44f7fa08f80fbe230fc44 : SUCCESS in 7m 02s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"7d452f3d41acf8526cbb114b93d43d95dbf173a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 14:10:09.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dade8505d06458d8cec8670fe27494c\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9637a25cd08446b8abcf443a19b52254 : SUCCESS in 1h 02m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2467d95d85db4297ad0012093e558c95 : SUCCESS in 1h 03m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec683cd40940489e88c7c67235aa8432 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92299ef050614a8ead367abe7065e325 : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a90e30ef7901461185bf3a73d4c498b5 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab73b37866a4efd821a7cde3b3d061b : SUCCESS in 6m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/80413dc6591b41899e47151e3eec81d8 : SUCCESS in 13m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b4d0b56ca65a465bb8596b06087f1d59 : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85c93b3943e243a98cd0ca249280f001 : FAILURE in 11m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/05a6bb4a6e474124b8f27dfb27b82dbb : SUCCESS in 8m 40s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/303cc9818a8b4d3e9b50416ea5e8b484 : SUCCESS in 13m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff8bf9935ce24981aad8718ebccc7bc6 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":38},{"id":"65b5eab2c2fbdb4d59cd140cb38deeb54e758627","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-21 14:32:03.000000000","message":"Patch Set 38:\n\n(3 comments)","accounts_in_message":[],"_revision_number":38},{"id":"7f9f269661347dd82ee9bc122ee70d006ef52940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 07:41:38.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"3d2aa5dbf18a528b3b4a831be5202a284b0d8f1e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 07:44:54.000000000","message":"Patch Set 39:\n\n(3 comments)","accounts_in_message":[],"_revision_number":39},{"id":"968e735ab83d2e14b9d4674026b161033679dd98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 07:51:53.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aca754a7719d4bc2a6b8a9c7e71ffb50\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b1795d8a43b4a909e445b2412ecefe4 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f2ad82476934041871fce28eaa15bf9 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a83a7c5edc6d4326a2c591d3b3a907c1 : SUCCESS in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"81187a6e35f9a696352d50d8a0db4d16cc7afcc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 09:04:04.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4cd5459613cb4426825a2deb208437a2\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3ed53739cb7c40a4b5b6ceac96fbdbce : SUCCESS in 1h 03m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba59bc33b6214a25984a64b6332081ee : SUCCESS in 1h 15m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e49f3bc40f4a5690b35ef0b280d170 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb1921314c546d58aded94157879719 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e55d46a242d745d1a37ca599d935b30c : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877bc4432710470a98067b7be0299fbf : SUCCESS in 7m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c67276184a044d848601024fc96f7be2 : FAILURE in 12m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf9574078fe940a69fec466af681af54 : SUCCESS in 9m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1dfb8fe633864a529240b704e9e646e8 : SUCCESS in 9m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/811c22921e9f46d786893461c97666d9 : SUCCESS in 9m 52s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/aab1f60c8d8d4db68cc8f2a4d3f1365c : SUCCESS in 11m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d0a93392eff347b289801ebdfd4620a3 : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":39},{"id":"0e4a16976febe13d4452ae8cf8247baf05ded92e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 10:59:52.000000000","message":"Patch Set 40: Patch Set 39 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"230959d422f7d89a873c156f2e42f19991fb4af9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:32:30.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/858f18f34f1941c7a83487f98f97f97e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0447b31fbfd248e3868af4a188e7e85d : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e77bbe011c84ee3a17d1c89cd8cd1da : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b989be6af8204aef8164c0c25ec87ff8 : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"86cf1baa5c8fb985d4c9b92044f800fa64ece2e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 12:28:55.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ca380a1c8ae433fbcb31ccff3876eaf\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e7822c2815ed4f98a3ba9a424f38d0f5 : SUCCESS in 1h 05m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec161556c2a147d09a93aa7ee1905778 : SUCCESS in 1h 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a22d7b322c2408a91c424619368c024 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab0e6869656a4d859dc466cb09404306 : SUCCESS in 6m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a15e8d223aa4bc8bc524ad8d24154d1 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee4d9b7cdae4d9cbed25b34eafd1206 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8229fadd13c9427a9619e6cb38e9e588 : FAILURE in 12m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/000831a56c8c472b9bf1b7711cfa58c2 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83780ca3ddce4dc3affcfa3409190805 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c7d738df6236427e83933456e5aeb4bb : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/77bcc392035241ec92118772900526dc : FAILURE in 10m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c7f33ca38dd41679b5c4e0c30966114 : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":40},{"id":"f3b45830bf21e45697026eccef541d4b4f8fd612","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-26 09:16:40.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"cdef6a19536d6ad8a083e0d16065484c7fcb8a63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 09:37:31.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/569d76912dfb4aa2b755220c3d8a7abe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7ad5c85ff7bf401f8244a6c7a820154e : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10a5296ab485418cbbd4c3d649369448 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73c39f0f55ec4f7d97988033eb8cb1aa : SUCCESS in 6m 49s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"4973999553332405441cdd24fe345bc44fe3dbe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 10:25:27.000000000","message":"Patch Set 41: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e96dbfc24b9d45b7ad68668acba0f85e\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/812637ca078e466f9ea5b8239d768193 : SUCCESS in 57m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/033b3921fc3d45ddae175f2225e37ad6 : SUCCESS in 1h 01m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acc41681dd29413e9eec7019aef726da : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad714f2014ae4cdc92da76fbd2ee6e27 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56e7e6ce74af4e9bb96d4af8fcb6fe0b : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4ee54761d84a7cbc89199c7ef8a209 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2493ff5deb6040ad8f06e4be7c302ea2 : FAILURE in 12m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/037031a679594f84b59a40dc4424d6e1 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f51f934159e45c786de99c0c2261732 : FAILURE in 9m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/893ea43f0108479a8f35c738ea6ec641 : SUCCESS in 8m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/67ef5adf86784b139e17ea47dd925963 : SUCCESS in 9m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/101c738c55484b0198bf0d6472ee735c : SUCCESS in 23m 28s","accounts_in_message":[],"_revision_number":41},{"id":"c4d0d305da6f8ef3036f556d3024b2fc3b1b52e6","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-26 14:20:12.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"0f1bc015937cf1584d558c1b23b467df502715bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 14:30:56.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c836f9bea748d38c0f26f35ed2e8e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af57223996e74237b049d75714990470 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4880afeded3d433eaf531cc0d974197f : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3e1f7c89bab644598e30ef96fb66e5ab : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"4e02bec4eacb36d3423701cefc0ae4637e354274","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 15:28:47.000000000","message":"Patch Set 41:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c941fa68ddfc4ae5927de16f97dc9d8e\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b278fe1738f245279bdd5d3bc2788b56 : SUCCESS in 1h 02m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5007ce2f1caf45a0829b7393bd9ea7b0 : SUCCESS in 55m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a4e66c4c9842489e59e08421892d4a : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0180474e9ad148fca482871c93941586 : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f85854ee48ae4215b85e034fff883738 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/329b8e6c5954486cabc179545dcb85af : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5172a77f043246298f34afe9443c7608 : SUCCESS in 11m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ec505fe4ecf4299af9c4b7c0ff7050b : SUCCESS in 10m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f3a9b24370b44b4f891958c4d773b7cd : FAILURE in 9m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fcb7096637f04ce3a209833148888739 : SUCCESS in 8m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/26f9e9a45cc84e86bb0b200952dacda6 : SUCCESS in 11m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/871d63c846f2438c962c9b7b9a9953d5 : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":41},{"id":"a9dccb9fdee43677412aa6e45dbdae6245c8c061","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-27 07:12:48.000000000","message":"Uploaded patch set 42.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":42},{"id":"d43b19093c2840972e1d0f96c6eeb30fab99f858","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 07:25:12.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8f7b2a39b304fad85d57386f1f7dcf1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbd90706dc4a4bd58fba142934380f3b : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/71008506040b4a2aa31d7e558ae4182b : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bc7952a3192c4e1c959d48dca81cf805 : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"42aaa9207c1bead89a38bab85dc47d6a3ca04737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 08:32:59.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c6d4eb9c0f640129938bc73b113e2d7\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dfbc79f7a0514af6973a03262885953e : SUCCESS in 1h 11m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec150ecea4114ed9afb9f2e6709983fc : SUCCESS in 1h 03m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35cb003ff8944982a3f193837155f539 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43bcf521dfcc41b8845796e740e89717 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08260ac4a56e4ebe9f2d0c8739961292 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c83713f9320641ce99d195ddd15347ed : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e7823aea772445e7a87d660593903562 : SUCCESS in 15m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf54913e255f4d1bb6913bcc86c39e6c : SUCCESS in 9m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/068cecc9f6c64289836d73029384d971 : SUCCESS in 9m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0ba95d5de1774ebe9f174454662f27f0 : SUCCESS in 10m 13s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b3f9242cfda0479b8bdfbf3efed6a0c3 : SUCCESS in 9m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/77e10fb693f04c4aa200e38b5c05b547 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":42},{"id":"6c57577381e741424ccebab1ab2c5fb710b8abdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-27 13:25:30.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"2eacee2775577ebcd1f5fe3a6db39e9c3885397a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 13:35:54.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad658006cff44ffabb48123b77635c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22460e45ec524295a8e7b11afd96376f : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b88e8d23b849453080cdc7917c16cf83 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47c4e9100ee041569d7ed9739c4f5d6a : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"f8b666070a6c366d5dc7ba1fb9946fbf4f58af27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 14:33:03.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe83d394c9a0481ea9b8e9fbdb778fad\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/06796580a2e54adaa2416d398461ef40 : SUCCESS in 51m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da6ca1c07b354b37865e4128c7aab946 : SUCCESS in 1h 01m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc160622208a45f5b93d6a0a6c06f788 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9a528a8a3fc4d66a31b8379731c0d2c : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06fb453f6a9242b692cd833e7c879416 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/844dfda5c1a7461497a5d94f53fa19aa : SUCCESS in 6m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d5b7ba9c3a544895a3d04753f77df145 : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/74e3ec677145431f9439ccdc804c56c3 : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ed9815fecef8487abdd297d30bd9b3ab : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/08acce342a904cf89ce1e9fadb34aeca : SUCCESS in 9m 13s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/27a2584f7d4847be86203365e84db623 : SUCCESS in 11m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/18cab1ec34cb46cdab52f7eeb294b4c8 : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":43},{"id":"b821090c80de41e9ebb0c1cfcacf258eb91d088d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-27 15:26:26.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"2bdf9f591c67e847d8883977d5b79ae001ab7e2d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 15:38:13.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72d1e57a017a494c898a857110728e37\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0fe2c541a844f1f8de080badddbee5f : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/476ce726b42d467bb15562ebfa61871b : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/335203231dac4d9f9d3599d36e11d9ce : SUCCESS in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"b2d7e3154237da93e081bca478b1736a60b70cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 16:31:03.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/158aa2dedcd2424ba3f4195d4455f401\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7dd9037119724f72ba45ecd661d3a581 : SUCCESS in 57m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e5019beaf8a4964a22fc04461a7c500 : SUCCESS in 58m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea77ae114a043eba1e38d4aa11a5efb : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b62658d6ccb4b3b8595b24d7b95b5d5 : SUCCESS in 6m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8b8d7ba23484093a1624d72e250fbbe : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fee9a8e6d97494486510cfd2d67b9d4 : SUCCESS in 6m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8791805e1c974dcd8b924eb52e953a15 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d88f817f09ba467fabb0b02d2bcea7d9 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be327fa64e18482e89908bc4aa3eb727 : SUCCESS in 9m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e1248359af8d4a068416ffbb746f44a6 : SUCCESS in 8m 44s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/de418fbce7c043b38b4cf8dfc17d5012 : SUCCESS in 10m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8696c55c7eb54cf09be33665382448e4 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":44},{"id":"66a042e6cdc74d7a0560cd08f1059502947022e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 03:51:38.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"bb81ff3183f8bb0f10f74c2230de981a6aca5f2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:00:56.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f87b08d4a06848b48162acaedbbcf524\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff6cbf0f95ae4a72b4b72dec80196aff : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/faf2a3b31af04e218f1d333614aeef6b : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3c43c332124469280c2553bfaa8c162 : SUCCESS in 6m 55s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"fa971d3664be8f69907d348a5f0415bdef601f9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:13:37.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f85106479c954ee9bcb60bc459d14b26\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/819ac1c9ac654ad7963c274219d12f50 : SUCCESS in 1h 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb841e5795ca40988ed980c4b1a714a3 : SUCCESS in 1h 00m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3767382c5a46848d357bf194ffd359 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14ea9ebed63640f6a16d7311097bf812 : SUCCESS in 7m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/516a4c6792014ecda194517b10b97a6a : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50766bc4f54a465eb96b33e8f15bf4cf : SUCCESS in 7m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/74f5aab9c0af4818bc562b5e33364d9c : SUCCESS in 15m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db9216a780fa416ab5a9b03eba4aba1b : FAILURE in 11m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0a6d39614df4fbfbd08cb835d5fb893 : SUCCESS in 9m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f13d78ddc9fa4fbcb2db9ef9bdcde3d2 : SUCCESS in 10m 17s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d60121367af24d38aa7144c8a37e3fd0 : SUCCESS in 11m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/313c41caca63444ab1e394cee4c9b829 : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":45},{"id":"91c5bb7e48c7fb3c2d91e7c0279dfcea57ecbcb8","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 07:06:16.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"e98e7ffb30613830a6bf770754ea5e2d96c8012e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 07:15:29.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a53fe32b744a288ee233219e114f43\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e295967a42dc44c6b12abfc36857467a : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bab77303fff54c0e993db3edfd275d64 : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0277f9fed3f94538883ea9b42d0f0348 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"8dc32623fd60a7592ffd686b02f75a7b9862a4e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 08:04:33.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae3e2734ff8a41a0b93aff4504f0983c\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/78f6bdb1f68048f2a9f2b759b1880ee9 : SUCCESS in 55m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0050fcfd91e45b58c951523bbdaac9d : SUCCESS in 54m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2499c864e54443e19e5a1d9729176e56 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/259aedb1d4f94cfc892a355f5d0499dc : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a829df2f9c324b72a75988960335123a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8076548321e54d27946982427b1bf715 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9c12dc2bc9c143ac95a1caa25731278e : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bad09fc75fd646d4b231ea91dd1340d3 : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/385138f568c941f494abf118b1129fe6 : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/05e491e469f34a83a7e4949c143da33b : SUCCESS in 8m 46s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ed1292e1f8f0425b8df73efac79b6097 : SUCCESS in 10m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4f4131a48b2741dab6261791a9262d7d : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":45},{"id":"d1d0bba20c48e07f3ccac5a1791692bdf0c157f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 09:10:56.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":46},{"id":"85ba4936ae13a6e4760b5675d45200d299bb5138","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 09:33:32.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b7356f2f964063a482359752339def\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/337025376a664b179d3da21051e33553 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08f3e8bba4be40649ee45de6043bc275 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e2d0ec31370d49af8c7a44fe8cf79dde : SUCCESS in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"86af1772ebc9b3a10af48e4bd615aee445eb35f1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 10:29:37.000000000","message":"Patch Set 46: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":46},{"id":"87b9e9c5f28218f7fb7910299fa8a6460b1c5b01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 11:10:35.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":47},{"id":"804540cc3c413ee50b3e1dbc25189bccce1f6576","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 11:15:45.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"4bb219a65faa0af048326391c8ca800b9e14a1b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 11:22:58.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0ca18364afa450085f312e8866f530e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/690a115d3a9e4bfa8aa94782b31d03a2 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5387702afd14079ba2caf7ec9ecb7df : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a320ce82c4874e77ac5edf2b2df9f112 : SUCCESS in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"52563e5cee8b4c202fa6f763b258cd290d1278f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 12:32:56.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb30acef634a4f069882fbb949fd88d6\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/737c144c042b4bac919e40cde883299f : SUCCESS in 1h 08m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a9239cd8dc34d29b2ce4e5ab9841e83 : SUCCESS in 1h 05m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb60be589d9940618a6e6b3bbe1e55bf : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cbaa6260588457e8923b9f902ac68f7 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9eeedbb2e40a4422adfd5a3c3f48bac6 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65ebd70476db4813b4aa06065d7a7e01 : SUCCESS in 6m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/824ff91b9f4c4e3789067aee17ed4376 : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/714312b875554d94afea7da388144fb3 : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3d359b2fd6984e6fbea7dc4f8e47d5c4 : SUCCESS in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dcc98ec5f0a04dfd8c8fcb0b13bfaa4f : SUCCESS in 9m 22s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/144c7d59f29b4defaf24e9eee81fe854 : SUCCESS in 13m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4a34b8c3cb304b2da9fc5a26f2b0cafd : SUCCESS in 32m 34s","accounts_in_message":[],"_revision_number":47},{"id":"290036b9aa7d7e5a24423a360bbcde09d8ae3505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 16:14:46.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":48},{"id":"db5f32f6401983c9d60cbbb1a11d5aa68f571123","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:25:01.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be1ba7a4b159435db827b278c168b369\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64efa7d940b74c61b2105a85959c4ace : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d02a151dbf9a44c48a7224ccd14f2782 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/945339dc7096461187ebc5f6bdebec3e : SUCCESS in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"aa9e43213244d0ac550fc53144d35ef4dbe5f7e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 18:05:17.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e402354a768b44db8023122aba0a7605\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/679def271eb64a8f9c18ccb991301d72 : SUCCESS in 1h 01m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9906c7457b9647138756306ae7bff45d : SUCCESS in 1h 33m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0e92143620a4d19952a4ab00f858a1f : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fedb113cff9e4fb99e300e81b55bfa85 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b627815b4acf4ace9a60dd133c6ab146 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1427568803b548868fea6cd4d235686f : SUCCESS in 5m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6e78c72385e14c6ab5a8ccae6739ac50 : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb0bd2ea9e244312ba9d306b146bc52c : FAILURE in 10m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d2df288a59f24d6c85f35e752e439ab7 : SUCCESS in 9m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e174a04d06ff49baa1fc4173a7a444cb : SUCCESS in 10m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f72b32842ba14a848c4509cb306310e2 : SUCCESS in 11m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6f658ffcaee349a6b318da77c50c9998 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":48},{"id":"060c46c60934371eaf6c0508fd5b668e725b9fb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 19:01:34.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":49},{"id":"7bfc9b28836d55cad498c45a1f96487c470e8267","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:14:17.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d68a44c065204decb8d0dc7b3c97acc1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf85a577cb52429caf2ec5c8a2a6658d : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/943074db9c02499cac294c90276a9536 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b50efa24e6524d138afea4d1764e8a85 : SUCCESS in 10m 12s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"65bfc4807e14963e042783c8932aa407ee180a4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 19:21:23.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":50},{"id":"613d656faca989b6fe89f9f19bdf6a87be38b1ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:31:52.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19a01c8e0d89427a8afc4b5bf9f0da19\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12f7669a585444e3a4cc4ea05daad794 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/061ca5d395b64b2fb6bd673b56dfa724 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4abae47734448d8b231d72e46206156 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"33b74ae2d8d425d973940cf509b882319ca2f3de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 20:34:59.000000000","message":"Patch Set 50: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1a2f3fd641f45fe80a67125913ecab7\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8b116ba5d134deebb7c280010a52d5e : SUCCESS in 1h 04m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ccfb62b1117413890e550a8b87d7fa0 : SUCCESS in 1h 05m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201778d6a1f0480b98e07bbd91fef778 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c978ef30980c4eac81cd2428e212dcc1 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1416d721796e46d7a713ffcf64f67dfa : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0da390be35d4433cbc70b18d582b61f9 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/138b31e5466e4f0aaa801a6f3b3b7449 : SUCCESS in 12m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bcdec69bd7b34fae8fbb605021860a4c : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/72e6e173b8ea47c4b9cdf14bf8a150f5 : FAILURE in 11m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c8faaa9a54694f22b69d87c09a9cd0b4 : SUCCESS in 9m 42s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/404aa403eaf0475ea39996204c2cb8ca : SUCCESS in 10m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ecc6f8ed6c254802bc77beb635ed516c : SUCCESS in 31m 50s","accounts_in_message":[],"_revision_number":50},{"id":"539fc15e113572a1e5e476e147b990d4a40cde88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-29 10:17:30.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":51},{"id":"30397e9c07b5aa0693555288e1283ee6a3a12b41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 10:29:20.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83cbf49bf9cf46fc99bce3999104934d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32dbd42cc9464a1d986f240c367390e2 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2e5555408b81487381b5426f4b7bf997 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7de23b6d92b4be39dceb1fe7565d29b : SUCCESS in 9m 04s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"f45caffdebf8cf5eb5b7a5af3b6a127394900fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-29 11:03:29.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":52},{"id":"f366d0e4536049466d396498851500b3f1400aa4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 11:15:09.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ae4cc26447b49c493a4f4293b34e943\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/44ad8963137c4a1d963aa3fabe53e2b9 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/444fff82cfa8442a96d10f86a29b8fb8 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d25323119de4a7d995cb990ce8f1bc1 : SUCCESS in 6m 39s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"71193193a78b23bf2b7a550a63745c3cf69702ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 11:55:35.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e7ec7d76ab431ca388b280674a2708\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f7e581799e54414b908de8eb43baa80b : SUCCESS in 47m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac4461c2e48f42ee8f03e7fec4ccd23a : SUCCESS in 50m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adecbe53b8ab4ff19f3d681e39b7f64e : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77152d6b1ab446b28fd1a0d7a624ee4f : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6b47376bb86467ab791ee77b2a4f13b : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/133a6cd73a7d4a0b850eba745a8b48a8 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b4fbf87a31dc491b9ba4e20846c0cd28 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e2680a73e964cad9c7bf839b72a5d59 : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b39f4a77c129400695cf71236bdefeeb : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b205fcd47cc5441e86bae154f85d5ee9 : SUCCESS in 9m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/989b7b308da54f9ea2e809b9a79bf99a : SUCCESS in 10m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5dc23b40f9584c17ab0c61b0a80076cb : SUCCESS in 23m 34s","accounts_in_message":[],"_revision_number":52},{"id":"033d998ce8be905c529e6d1738447b694e2058a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-04 07:00:07.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":53},{"id":"fe56ad17fa0cdebc76420f9db372d5bf95e69468","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 07:14:55.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f5473d379545208fe1a25d98bcdcad\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e0232217b244f8db00f40a75aa4802b : SUCCESS in 12m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45cbeae246ab459786b6870505b7179a : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"b54baa73022615983966b399838b9552bef5924c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 08:16:29.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9267eb910434510966b5472ed76591c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1819cbf041174e24b88bd168d99eac7c : SUCCESS in 1h 07m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa47fa724fb4472dad8b73fedc4d79a9 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d416a5b9e5d41a9ac1d871caaf73c52 : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bacb8cdab2964ad3a0d371eac6cd7723 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e51c9269a40d4edd83d7a8f51f3e1326 : FAILURE in 10m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5abcbb1a3cbd4dd0a730e310370b5e32 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/abe541c918ac437f956b8e87264f8a25 : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c51ba5397b944d1486b132b583a9271c : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cab67b04d6334e80a24af30a73a544be : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/39a6a4225b6e44049191c11f5e0dda51 : SUCCESS in 9m 13s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/69821ec2a2ff4ce8b47e903d0efacddd : SUCCESS in 11m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6f8d8f30f5b458f88553fe43756dd0c : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":53},{"id":"2c3a8227a7054802cd08532f8c85f5e9b1cbbe07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-08 06:31:24.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":54},{"id":"a0963621119a3c7fbc033fe513f4ff1224578682","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:43:15.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f6bce24ef6946a98aa57e7282295ad9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9165013fcc694b5fa3fb7f9e08088353 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c03d7c3081a4392a231b01da26cc31c : SUCCESS in 9m 33s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"10c8876e8ec0a33077db9e0594ed330633db7cb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 07:25:59.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f804a83072f48a9900488ed64bfedea\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d20a4ff94134760a785c2318e4ee8b4 : SUCCESS in 52m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd279127e174621bdb015d7b5351260 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb76d35cc6a64155956aa6ac98185493 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a5a5f5424b6c48679304b0288fa94cab : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed9241d052db40cdb2fe5971863b5106 : FAILURE in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/666917e9127546aa87ce57605e1e496e : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/15932c60167746b59dd650468cf23919 : SUCCESS in 10m 26s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/436276a8cdb24fba8024bf99648ca8d8 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bcfbe783cace49a6b33b8956233956c5 : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f633fea5467f4520a6c200b50ec49cba : SUCCESS in 8m 40s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2618505bdfe1408baba2299939f0f73b : SUCCESS in 10m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7b05a290ff964ebb84631c16f158530b : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":54},{"id":"38fac776f5cd73990bf53b1aac9416a600b87d04","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-18 12:05:23.000000000","message":"Patch Set 54:\n\n(1 comment)","accounts_in_message":[],"_revision_number":54},{"id":"1f2a7b9d430179b47b6326fdf611f7efae9c8fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-29 08:04:05.000000000","message":"Uploaded patch set 55: Patch Set 54 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":55},{"id":"b738dc02979d907053deba1094069cd1e17936f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 08:14:23.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f823e35d2d1a4a6e8cce7287c04ca521\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d6ae374f0c644139740da7fb301e5f5 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ae704fed2c8b4c51b49d9e6ec5f5c3a5 : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"b95914c75288bd28cc66a5fc1b3c8e9006a5564e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 09:31:40.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/180b5f4d52e44f92ba25085bc3de6f6e\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3778db297b5544ad8591a4e61cffa09c : SUCCESS in 1h 07m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d96e8830c3e54bf5a983db72c8380f9b : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec517b8758374d988d7245955e36f394 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8da654565fff4227a384dc85aa3d65be : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dbd309d3c336431eb59f67532daa55da : FAILURE in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7be1cd9d78da4f99a7646e8851bae075 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/19babc737db8403d90bc29178fe8a02c : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79deceef0c1245df822ae7883a436ab1 : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fd70234244946aab824cca342f16824 : SUCCESS in 7m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2dab9e5618fc420a8b908e41d3f76b4e : SUCCESS in 9m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f4850300f8ec472fb98524f07502ac21 : SUCCESS in 9m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/edaaf3c070df444da447878811aca7f7 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":55},{"id":"ab88dfe76dc686a01b06039562009f830cef3410","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-30 09:42:34.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":56},{"id":"0b0e7a834ffc589c2e787e8f6b26cef9e17fc6e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:52:51.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a2d907b58874e218aaa161172c5d193\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73a5310cf63f4741ae6f6274e1f6d3f9 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b006389ad7844a3fa39ed4a8fa9150d8 : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"8eddd9891287fa123cbdda0c6604c508cad22f72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 10:40:58.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ca423b4892d4bce9aaa9197cf260a3c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/512aaf5a1adf493abafc59a1591f8c75 : SUCCESS in 50m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd20d83ad6aa45398f25996f8dc3d20f : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bc1801a1ae548489ddda0363e55838d : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d7a2cb4cff8e41799ae5480011620fce : SUCCESS in 5m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/86df7e6716a7465996dfa1e4fdb91788 : FAILURE in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f377fc81661b4e68b73b5a9ecc8878b6 : SUCCESS in 5m 59s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9d32d190b7a04fdfae9ba3110fe15383 : SUCCESS in 14m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7166208d15847f195c182286b616e3c : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e3345d6f7b19437fa26f501e36285428 : FAILURE in 9m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fe4698432d3847e089b0846bb010e8fe : SUCCESS in 8m 52s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ca920eb78d4f44e2a8a7165cfc76d108 : SUCCESS in 9m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/763fda9f95314dba85d5fbb851fc8d10 : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":56},{"id":"13d643d33034b40dff86c792be24e313e76f0d47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 06:22:18.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":57},{"id":"0b312fd5f0d96f637c4fa457c82cb32d8e399100","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 06:33:27.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a7dcbfb446449abe788b6cc31fb7fd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8ac505de4f34eabb702c537be16e523 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f5a20d458c8b44bdae4fc7764d2c36b7 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"ffb031ffad5bda34d28f9169e609aef4a1595a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 07:39:47.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c6dccc46e0f4d60baa4f782d11959e0\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3a5856c25814c47875ba9dac2eb519d : SUCCESS in 1h 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20d13dc288984a4c93f26608a01f043a : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e1ffddcd9fe4df1b7ec4b72d7d6f001 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44ddaa4916e64dd49e361e0456941207 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f60d204a49b1404e802169e511a5b590 : FAILURE in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840f98681fd444d280c4cf14b3382687 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f6c95003822c4829b1bbaed6073bf4ab : SUCCESS in 12m 20s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f0e0efeda450455d86c7c850f9fa0d5d : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/612d365a5bd741fba9ee69924f7b2c7c : SUCCESS in 7m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a661e5018afd4805b5de726f83eb2e5d : SUCCESS in 8m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/67d30305b38448eeb1b8bacdc4724088 : SUCCESS in 12m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb3ecb312b96494694f3ad8b8877656a : SUCCESS in 35m 40s","accounts_in_message":[],"_revision_number":57},{"id":"2415e6fae5bb5cf309d212343bf3bacfa7bc6d2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 13:08:53.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":58},{"id":"5ca6410c1e4c5a063b38d5546abfae6d02d414fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 13:22:20.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48f6057827b438ca77876fb1ea580b6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/305e739126a744fbba15f0ac2f835b81 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7053511361b44d0d9b4341e0d4e97606 : SUCCESS in 5m 34s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"2eb82bea5a872a02e85e36b45bff62f76c56bf4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 14:23:19.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d62b0aa453fc45f0a3d7645066ddd14d\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b95d500e9c2244c4bbf49aa24f1a35ca : SUCCESS in 1h 05m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f3cb3c20a0242c3ad5c1e102b0f835f : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e7234130a4848408769ec6ba907b675 : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8c45ea3d112467b99df1ca98b38e2c5 : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3df10ad4b8384c3f8d37bd8742c708b6 : FAILURE in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d10a35d73d324303991badf84ff54ace : SUCCESS in 5m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3b918a8222c5494180a986ff78105148 : SUCCESS in 14m 08s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58800e8558be45bca654bdf1eb09abc6 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/facf4b86b93445c7beb32fa5524ab2a4 : SUCCESS in 9m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ed35aeb116f443c8994a7bbd96d4e26d : SUCCESS in 10m 12s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/51f9b622333944a196c9d30b4556de46 : SUCCESS in 10m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2bfc0e35606849659be4fe895ccf06f5 : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":58},{"id":"76576e904d1c2b42c95c6c1a082ada31a7558d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-08 09:01:11.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":59},{"id":"7c4d85e4c369690c0073831f615b7e3722d351c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 09:11:56.000000000","message":"Patch Set 59:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d58f0dcfd5e443088cd589c6d5ffbba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37f6b961aa814ed5a86a4422e64ec795 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4f69595c049b4bb9adb79652bee0b5f3 : SUCCESS in 8m 12s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"10534a973141a88fdea9c7fb67ea501aa622b788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 10:13:33.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c47a6f911534c7cab97db5846f7ceab\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e645548cb9b4a80a4b0f2a5a57ae43a : SUCCESS in 1h 01m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a53a226e59214c89a499231d4aa6d0ba : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49d1df5671654d56823db12c1f5c1235 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9a78d0a7c224b03b5f14e89c7853b75 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/841d2cab97d7416c85ca2dee80cecf68 : FAILURE in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd01f13fcb042e89c905c23d1e3dca0 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2196dfcbdfd647b9b7594cb66d4c0506 : SUCCESS in 14m 07s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6887ba548924f98a9fe895c8dec8fb8 : SUCCESS in 9m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5babfb179ba74a94af81d31619e98e89 : SUCCESS in 9m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c36590ed34e2438696eb6d2e45e6dcad : SUCCESS in 8m 31s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cb62d0467abd428fa580a9973b9146b4 : SUCCESS in 10m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5e3332b3b8144f7aa0f97109b3ffd689 : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":59},{"id":"8978ee20a4a582ae5f62b82db9be294a5d78459c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-06 06:46:59.000000000","message":"Uploaded patch set 60: Patch Set 59 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":60},{"id":"43a4e5e7290a04e32fdd13152d6c673066e3fa50","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-06 06:50:08.000000000","message":"Patch Set 60:\n\n(1 comment)","accounts_in_message":[],"_revision_number":60},{"id":"4874abb10c6f6d26bbc259db6a528e90b06ead4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 06:55:12.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/539b87dbab4247cdb44cd5d598058c8e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2d8a839cf864d789083b5681aee1536 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3aa0b9e4679443cdbf21fc885fd9e2a6 : SUCCESS in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"72f1de1a9eb3259409afc61b52a7cb0996b21c8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 08:03:45.000000000","message":"Patch Set 60: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/281bfb1050ed48bea5b21399ab0ce2a6\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb3848c759f54727993b8610fb0f3fb0 : FAILURE in 1h 06m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c0c22a539454c4787dc66c71bd41c0d : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c67a4578c5a49ef9a092306e008f725 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a6195c287794a189ec996ba35d01c8f : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/02dd007e5f7d420489ee24c040760617 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23b2012b3b5a4222a64f500f81e90065 : SUCCESS in 6m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/51bd6978fcec48b49ab32e5a362fc8d8 : RETRY_LIMIT in 1m 42s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d30ade38154843dd8bde4987fc1f6dae : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b632a9763f3e4bf6bcbf4117e669e1a6 : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0f44501015034d46b7dff9e6fdab56d2 : SUCCESS in 11m 51s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c9f16021fdcb432991d7fe48eabd2322 : SUCCESS in 7m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6c87ff3f6e5b406dbed95539060a25eb : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":60},{"id":"dcbd3d818fbc33abcb3cfa6718652a506a2a768e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-07 06:37:44.000000000","message":"Patch Set 60:\n\n(1 comment)","accounts_in_message":[],"_revision_number":60},{"id":"b4698e124ec837129a4d3274b04b1737e9697f28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 06:44:08.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80f8af020bb149fd9cb031ac2ad40f1f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14304542c42345608fe60d6d3665f727 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b3d44eac8eaf4d4eab0fbbf9a6c8edea : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"db3f6ef2d3e3542f5a4636abf5287943c8def4f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 07:44:02.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4744475d30074f169fcc55a5b6d9fbe8\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86bb9f117c224d8393071321e6b102a2 : SUCCESS in 1h 05m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f255ce2b734b67ae364d60befb6e85 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0fe995bccf040f28a4e06e5f457c4cc : SUCCESS in 6m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f320e12bd44b4193bfd2170d80a7b8cb : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ea225a2862c4305994301da07e983bf : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578d35b7653741a79cd2c1edde54122b : SUCCESS in 6m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d3a4e01ee4364acca9454ffb9afd8842 : RETRY_LIMIT in 1m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4dc0319f3b4f4624a696a6b7235b3133 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c6a0429404db4e9abbbbd08b6de940a4 : SUCCESS in 7m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2d940614c6a64612b6957c34fb1c614b : SUCCESS in 11m 46s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1a537b24cbb44e8fa2124b28068a61cc : SUCCESS in 8m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6554016824f146dcabb3c4aee0a8b0b1 : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":60},{"id":"ec9b0c507b63b95da08792997d8b4f4199f3151a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-14 10:35:46.000000000","message":"Uploaded patch set 61: Patch Set 60 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":61},{"id":"2cf70a2ae53ea3a41918a914d16be19c2d01009c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 10:47:27.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6077cccb09274f5cb26a043d0032f2ee\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e893634c83704874b2e83782631d1db5 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4eb7a919e80c4580b7ba2bea55cb5f44 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"d3937be4b41f8a6341ecb5e4928f94bed11419cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 11:38:30.000000000","message":"Patch Set 61: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df201aeb0856487d87cf781750c5e305\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b7eaccfcccd4e9e8da5f0f4734e8c59 : SUCCESS in 57m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1f059d4b826499db894997e4332d424 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/065f69c05fff444a8879ec3a738314aa : SUCCESS in 7m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d09fbd6bc404e9cbcef72401d77b274 : SUCCESS in 6m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fdbe4c5ea17b44268230a55b20526575 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6ea2088958f45f2a1ef5e834780fcf3 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d561f16ebb4c41d1abaf68491003d157 : RETRY_LIMIT in 1m 44s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c3480ef1dc214bb493efbbcc9f3adaab : SUCCESS in 10m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8bbb83708ca44a7da030cf9a76b5e897 : SUCCESS in 7m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/869f176d0eb94ac9aabae5818e57f155 : SUCCESS in 12m 34s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ed53e6dd537f40c9af188ede0f0e904e : SUCCESS in 10m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d31161220ae4f8a8c951c20f90df689 : SUCCESS in 30m 39s","accounts_in_message":[],"_revision_number":61},{"id":"3a12fbd6dee40332d6323153a3166a93c7f1b154","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-06-18 16:51:53.000000000","message":"Patch Set 61: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":61},{"id":"1127e3eee27164107613701f66b863aa88a7690e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-06-27 05:34:05.000000000","message":"Patch Set 61: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":61},{"id":"bf40ce95fb4124f656150fa63dcd27100b9d58cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-28 12:46:31.000000000","message":"Uploaded patch set 62.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":62},{"id":"8dd9d1e829aba432dd7ff2a16386fcfbc030c88c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-28 12:48:44.000000000","message":"Patch Set 62:\n\n(2 comments)","accounts_in_message":[],"_revision_number":62},{"id":"86f4ed3781a336146c585c7320f962d4a49943fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 13:07:25.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/821c27b2ca224b2d9c37ba9f52ad0b0d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcdd3f0c8a07453fadea1a33c2d88bb8 : SUCCESS in 13m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05aff4349c0e4ff4bb0b2b0a629158b1 : SUCCESS in 8m 59s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"80d1e6499fe797947137a66f01a05e977b03ec74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 13:49:50.000000000","message":"Patch Set 62: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1d9f2da9aaa4d4685721e186aa0722e\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0814ef1ba3e4459691e6674731710d46 : SUCCESS in 57m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042e66957cf143db9f04351da3f5010d : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d44215ca61d49f7b2cc74a25b0a1f7f : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff0b1cebe0754c8d91456701c32d05e1 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2270c7b0f9904fa79731ba479fa7090b : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fd1e73ccd3c459fa0dec66a95af3e1b : SUCCESS in 6m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a281ff1b42aa4b6bb57476ea376a21a4 : RETRY_LIMIT in 2m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5201c09eb2e246e0a81e838e0196fb8a : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5958f26e7fd64e8a855cc5ecb2d5fa13 : SUCCESS in 8m 16s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/819709a60e44481abbfb233956ff324f : SUCCESS in 13m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6f5056ed499a40f889f098c16149efa2 : SUCCESS in 9m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/993ebc92209d4870bacd9f6943589151 : SUCCESS in 33m 05s","accounts_in_message":[],"_revision_number":62},{"id":"648b76feacbacb036dfb67c3aae8955cbe458956","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-15 07:07:33.000000000","message":"Patch Set 62:\n\n(1 comment)","accounts_in_message":[],"_revision_number":62},{"id":"cfc93ea5b9d22bf9b4363ac2c0530ec544c0313c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-16 10:22:56.000000000","message":"Uploaded patch set 63.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":63},{"id":"b87b38c498323f53e7e08b19baa9cdb1e8a318e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 10:32:46.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1c0f2e21248452b842a78770232ff0a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdf2d934790d4ca09177de5e4293ce0a : SUCCESS in 7m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8fe46f490c3944c1bbe3fe075ab7bbbc : SUCCESS in 6m 14s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"6fee3fae99604301f4e76b0cd1eefd2f1a8becbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-16 10:54:31.000000000","message":"Uploaded patch set 64: Patch Set 63 was rebased.","accounts_in_message":[],"_revision_number":64},{"id":"86b085d273540b12796256f3d9fb23ff52be59b8","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-16 10:55:52.000000000","message":"Patch Set 64:\n\n(1 comment)","accounts_in_message":[],"_revision_number":64},{"id":"001dda09333b8ebf11bb022a21860b81ae0f78e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 11:07:17.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/566b6627171e448c9b1a995da37b0317\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53ffd2206c7e4221adee92f79b4247e3 : SUCCESS in 10m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d150ac9710ac4ac5aaa7f20984795b0f : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"c6a27c9263e35d3bae9403f13acf235f30c38171","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 11:51:57.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf1d2aaed82c4adba740aa99b44c6511\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a683f45fa5c449f6bb157d31dc5b82c1 : SUCCESS in 56m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6602ac1975cd467594b4f842026c8370 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1611bb3b57ff4241b522d0c72bbac1bf : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a825f0c11d4043a9abd7c9854b7e879b : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56b7069899254bab90330ede36fb0131 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22467b2be1148b78fce17359af4ad3c : SUCCESS in 5m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7b3a30a74a08479eaed188264587df4a : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d387fc3f1cee492ba67844102a0db079 : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c0e20f88ef4a4ce685cd939fcd6f7aff : SUCCESS in 11m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a29468d77a6740c39bcd8c3c204739a5 : SUCCESS in 7m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f3ccdea01204beda09e3caa2418c74f : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":64},{"id":"1a7ba3e1f57ca09f7d54e3fa7d80091b2f5dcd88","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-18 05:40:11.000000000","message":"Patch Set 64: Code-Review-1\n\n(19 comments)","accounts_in_message":[],"_revision_number":64},{"id":"8ae2ae50d6e19eb2e9e1b776157a759a98605c32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-22 07:22:18.000000000","message":"Uploaded patch set 65.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":65},{"id":"759f0e728c65af4ba6ce98669ab6e475915b25ff","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-22 07:25:56.000000000","message":"Patch Set 65:\n\n(17 comments)","accounts_in_message":[],"_revision_number":65},{"id":"c481ffd6a66b6fd0b1d65f49ec93b4e68f8c5247","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 07:32:03.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eed217f833424e0eb267032fc0ab165d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/479cda8974d64a30953385ff0be66139 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fd0f5c488346479f8ae7809940bc71e7 : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"1cf248b99714ab86604d0d7f5238cc0c29d3b3dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-22 07:44:37.000000000","message":"Uploaded patch set 66: Patch Set 65 was rebased.","accounts_in_message":[],"_revision_number":66},{"id":"113e9970b1911817ebf2bb24e4aa7f55214c2df0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-22 07:56:07.000000000","message":"Patch Set 66: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":66},{"id":"b2a11fdd43b3cefd2da2927c86dbc583a8a763de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 07:57:45.000000000","message":"Patch Set 66:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7458bec8c524531b3f06ec05c4fc249\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a98690ee3b844730b2e6e62ae629f81b : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12e809f480ba4ed780cbc1e51a6d912c : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"f7ea639d6ff562c04ce5a64451b373568b0f00bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 08:49:46.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/88efeea786f8472191b8d0904e875a93\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5b6faf8d805140f389600bc8814a2606 : SUCCESS in 1h 04m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77830e0177cb4ab0b90f5b5fbdfdc13a : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdace0b8b2e94919af8c3a4642a3b773 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/067cad60b31a4632ac145bd0bf3139a2 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c122996646a64e46a43e98746dda4ac5 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc65d1a23933426db21fea54c3f33140 : SUCCESS in 5m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8c7d6d495b6447d98d741bea57462a91 : FAILURE in 10m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5185b9a35e1e47368c8292dab05f06cc : FAILURE in 8m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5f874e96838d4fbabe5d589a1784f80d : FAILURE in 12m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/decbb4b09e87469999e292c4d525798d : SUCCESS in 8m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/960c72fb148e4027ac16bcde8141a55d : SUCCESS in 30m 44s","accounts_in_message":[],"_revision_number":66},{"id":"745849d3d2c4ddd9ce254f324983ee7960124624","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-22 09:10:58.000000000","message":"Patch Set 66:\n\n(2 comments)","accounts_in_message":[],"_revision_number":66},{"id":"bb71872938a78428a646e688b9c3e3ef4a310917","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-25 06:29:22.000000000","message":"Patch Set 66:\n\n(2 comments)","accounts_in_message":[],"_revision_number":66},{"id":"8d7b3f2a85583090e781c991c413bdc6a07f72c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-25 06:53:52.000000000","message":"Uploaded patch set 67: Patch Set 66 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":67},{"id":"357de87780415a101a06326e0e8224858d3f2472","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-25 06:58:05.000000000","message":"Patch Set 67:\n\n(4 comments)","accounts_in_message":[],"_revision_number":67},{"id":"70d22be5e5f0623d179528cc701fe6d255c350af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 07:05:44.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99f423b6a4b4410889606a427b6bf30f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/befdd59c38224592b5fa1ca3be39a22f : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2dd767d8beda45d7b277f0b01a9dc2e9 : SUCCESS in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"8a2814e4e77e76ca9715438f7545e23707148b55","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-25 07:46:20.000000000","message":"Patch Set 67: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"a76c7d99d293ea1019d310042e55a6e9a0ab8c3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 08:03:02.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf159231a94449a2bfbe76c80f720b7c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/908feddc6489483e871a4a23af57bf30 : SUCCESS in 1h 01m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf186c4a10f48f4b12699118dde8e03 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c79c1095ef4400bab81d2e34ed5ca49 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/650895c326514b4b9369c5a44ac16a12 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1670f069cf514d529b5cb018dbacf7b9 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09c3d6797984961b1fab7f478496f06 : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0de69cffc5d8443487cfd214e895854a : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc197754acf844fb8c9585770db4f04c : SUCCESS in 8m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b0ca603d17d14d89a0fa24c06a0b65bc : SUCCESS in 13m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/009372f992164a0bb44e953a81c43505 : SUCCESS in 8m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fdd0045e2c8b413eb55c84f0df93b7dc : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":67},{"id":"28bc18edfcd46ee1129d2f5e0a76c51d67d3a828","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-25 10:32:00.000000000","message":"Uploaded patch set 68.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":68},{"id":"344032b91951c5a304a730d85bcc4a02be3d5517","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-25 10:32:33.000000000","message":"Patch Set 67:\n\n(1 comment)","accounts_in_message":[],"_revision_number":67},{"id":"8878c49488d81b4a824d1d8a9db2c71040221ac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 10:42:28.000000000","message":"Patch Set 68:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf1016868f5e46af98c1761ae63d5914\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78925e559eb746a0a8cfe74506534f1c : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9f8f75c838ee4e55999ac02a2bb1c213 : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"9c5608d2119105d1dbd9cfb1157388bcecfeff04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 11:29:15.000000000","message":"Patch Set 68: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8919df83f53443cb4fdc04a69cfbc68\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d09eb4d8671f4c25838785d727d8bc24 : SUCCESS in 56m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a74a7fcf41a4ddcaea5f76df948aec1 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd6c607ebbf844829990a282d42fc235 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfef4b1116db4651a6573120831eb701 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e94d2e131a74fb2af0d9f7688062301 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19e9980e465543dfbe9900bdf9c941cc : SUCCESS in 5m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/530fe36b48484142aeb0579ec26d10be : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85a8bbf018e045c4b261edcd02893044 : SUCCESS in 8m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fb460e6cab6340f0a26f109187c12190 : SUCCESS in 11m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b076416dec0746cea83612cce35d2d64 : SUCCESS in 8m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2861aa3788164e9d8ed3bfd74fb88d8b : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":68},{"id":"1caa115d8ea5b6ab0e365bc8e2d359e9c005fcdc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-25 13:52:58.000000000","message":"Patch Set 68: Code-Review+2","accounts_in_message":[],"_revision_number":68},{"id":"ee15787c2f61e01dfdad97cd0ccbd2c79e2efd18","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-30 18:14:37.000000000","message":"Patch Set 68: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":68},{"id":"7aa41111fbc4ee376e0ae952f3c9e22401f96475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 18:15:19.000000000","message":"Patch Set 68: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":68},{"id":"f0a9deb3a3d4874742ccf0a754ad9e69d91c4573","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 19:13:02.000000000","message":"Patch Set 68: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ee9ec998e342e7b61a1ed0f5d22c6d\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7587cc54d197495a9265489511ac7507 : SUCCESS in 45m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fe1c2cbe9f4eb5a56e1c9a4b1092b1 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e181fa8c17fe41d5a21d40a4421b9c27 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f122c9919754ef49e64b910b766f905 : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dafa65c456a048ffa912b91c0a6cefac : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/484e477814444ce28732e749fd5e43ca : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/001e388ca55e491e91afa6eedb6e53fb : SUCCESS in 8m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3f10157caccc4f4bb07dcf749c71ace3 : SUCCESS in 15m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aeb715d8f04c4763bb8a43ee5e5830e5 : SUCCESS in 32m 28s","accounts_in_message":[],"_revision_number":68},{"id":"c9d0126e16b20c6f08535378d6124b1e40dd96ba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 19:13:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":68},{"id":"2d2b626bc0398c9ae3dc6b27dfc542c588d2cddc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 19:14:01.000000000","message":"Patch Set 68:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd55ad6544aa4960ae0652e7bd30b6ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2e2705e3bf24e19bd0e487dfe176a34 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":68}],"current_revision_number":68,"current_revision":"b83f38cf25428983791017b74e92f3c44c41c6d4","revisions":{"5566e4c81db5a8d1ce0a71a95f51f0fbe9f7021a":{"kind":"REWORK","_number":1,"created":"2023-06-26 08:11:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/1"}}},"commit":{"parents":[{"commit":"ef4f88df77243f339d8e3199adf95c1169cd953d","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef4f88df77243f339d8e3199adf95c1169cd953d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"subject":"Add fucntional tests for new add-location API","message":"Add fucntional tests for new add-location API\n\nTODO: test for add location for multibackend\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5566e4c81db5a8d1ce0a71a95f51f0fbe9f7021a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5566e4c81db5a8d1ce0a71a95f51f0fbe9f7021a"}]},"branch":"refs/heads/master"},"cda9727da626dc2b3543a86b11c6d45aef5b54e7":{"kind":"REWORK","_number":2,"created":"2023-07-20 07:59:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/2"}}},"commit":{"parents":[{"commit":"ead111bb3ec23f9c907cc688a1841218913f89f7","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ead111bb3ec23f9c907cc688a1841218913f89f7"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-07-20 06:08:04.000000000","tz":0},"subject":"Add fucntional tests for new add-location API","message":"Add fucntional tests for new add-location API\n\nTODO: test for add location for multibackend\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cda9727da626dc2b3543a86b11c6d45aef5b54e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cda9727da626dc2b3543a86b11c6d45aef5b54e7"}]},"branch":"refs/heads/master"},"5d85c4a57f80858a73e18d25de6be2d810685f3b":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2023-07-28 10:22:33.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/3"}}},"commit":{"parents":[{"commit":"cd04b05b7f02d918621054f6bb0ade398f0032b2","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd04b05b7f02d918621054f6bb0ade398f0032b2"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-07-28 10:01:08.000000000","tz":0},"subject":"Add fucntional tests for new add-location API","message":"Add fucntional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d85c4a57f80858a73e18d25de6be2d810685f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d85c4a57f80858a73e18d25de6be2d810685f3b"}]},"branch":"refs/heads/master"},"098b2b198fd33c72b90f8f04217de65a95177fab":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-08-02 14:05:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/4"}}},"commit":{"parents":[{"commit":"d3e1d409ce8fdbba1e7450d2f165071dfc275127","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3e1d409ce8fdbba1e7450d2f165071dfc275127"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-02 14:05:00.000000000","tz":0},"subject":"Add fucntional tests for new add-location API","message":"Add fucntional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/098b2b198fd33c72b90f8f04217de65a95177fab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/098b2b198fd33c72b90f8f04217de65a95177fab"}]},"branch":"refs/heads/master","description":"Rebase"},"1ec053a7dbdd3da6a71b94957700d2472c158bf8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-08-02 14:13:23.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/5"}}},"commit":{"parents":[{"commit":"d3e1d409ce8fdbba1e7450d2f165071dfc275127","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3e1d409ce8fdbba1e7450d2f165071dfc275127"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-02 14:13:23.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ec053a7dbdd3da6a71b94957700d2472c158bf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ec053a7dbdd3da6a71b94957700d2472c158bf8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"f8474a91773b043a8e259f34ca5207230f00382e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-08-03 06:58:13.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/6"}}},"commit":{"parents":[{"commit":"33250e3de2451140aa2f5044a77f5adbdbc04f07","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/33250e3de2451140aa2f5044a77f5adbdbc04f07"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-03 06:58:13.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f8474a91773b043a8e259f34ca5207230f00382e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f8474a91773b043a8e259f34ca5207230f00382e"}]},"branch":"refs/heads/master","description":"Rebase"},"a67a6ffb0380b959820013214268da0adcf47454":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-08-03 10:29:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/7"}}},"commit":{"parents":[{"commit":"c9d8618f6ee524dd2a0c78f0f6c4dce5f685d4e1","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c9d8618f6ee524dd2a0c78f0f6c4dce5f685d4e1"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-03 10:00:13.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a67a6ffb0380b959820013214268da0adcf47454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a67a6ffb0380b959820013214268da0adcf47454"}]},"branch":"refs/heads/master"},"ebd06c49df72d9fde23a9924323bc969d7144f6d":{"kind":"REWORK","_number":8,"created":"2023-08-09 11:13:25.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/8"}}},"commit":{"parents":[{"commit":"b48fa211844cf5b12772e53cf5088730774250cd","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b48fa211844cf5b12772e53cf5088730774250cd"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-09 10:31:29.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebd06c49df72d9fde23a9924323bc969d7144f6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ebd06c49df72d9fde23a9924323bc969d7144f6d"}]},"branch":"refs/heads/master"},"3f2194bb0456839b4b2124bee80cb513b9b4b3d2":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-08-09 11:32:19.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/9"}}},"commit":{"parents":[{"commit":"279c010892c5f27790f6b8e4aa3bd0bdcb84d3ce","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/279c010892c5f27790f6b8e4aa3bd0bdcb84d3ce"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-09 11:32:19.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f2194bb0456839b4b2124bee80cb513b9b4b3d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f2194bb0456839b4b2124bee80cb513b9b4b3d2"}]},"branch":"refs/heads/master","description":"Rebase"},"295fa55f5e99bcc1ded22b38e6077d2f01669574":{"kind":"REWORK","_number":10,"created":"2023-08-14 11:50:37.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/10"}}},"commit":{"parents":[{"commit":"db13234a30e83f68aed844b6056758f1c6a19c8b","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db13234a30e83f68aed844b6056758f1c6a19c8b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-14 09:59:25.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/295fa55f5e99bcc1ded22b38e6077d2f01669574"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/295fa55f5e99bcc1ded22b38e6077d2f01669574"}]},"branch":"refs/heads/master"},"829b0bf6d113e61a2fa0af3eed43db9bd1c4700e":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-08-16 09:29:45.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/11"}}},"commit":{"parents":[{"commit":"299bf91255480c3ff43a7bd2008339c652df0c7d","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/299bf91255480c3ff43a7bd2008339c652df0c7d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-16 08:57:14.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/829b0bf6d113e61a2fa0af3eed43db9bd1c4700e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/829b0bf6d113e61a2fa0af3eed43db9bd1c4700e"}]},"branch":"refs/heads/master"},"f7bf6556711afe94404c6c057952230e9a40623b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-08-20 12:19:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/12"}}},"commit":{"parents":[{"commit":"48c08297777588fe31e2e55d471d7e6fd119105d","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/48c08297777588fe31e2e55d471d7e6fd119105d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-18 09:30:05.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7bf6556711afe94404c6c057952230e9a40623b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7bf6556711afe94404c6c057952230e9a40623b"}]},"branch":"refs/heads/master"},"bdc10ad42af4f15c221957945c5ed71ac06eadf6":{"kind":"REWORK","_number":13,"created":"2023-08-24 11:40:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/13"}}},"commit":{"parents":[{"commit":"f6c3d616004e66266fa2c7390101b2abc29c18b6","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6c3d616004e66266fa2c7390101b2abc29c18b6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-24 10:48:36.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdc10ad42af4f15c221957945c5ed71ac06eadf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdc10ad42af4f15c221957945c5ed71ac06eadf6"}]},"branch":"refs/heads/master"},"c1992947b5a6ed3fbeb44186bcffb8df8ee80575":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-08-24 15:02:21.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/14"}}},"commit":{"parents":[{"commit":"f642f321be0b5061da368a983bd296db29fce00e","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f642f321be0b5061da368a983bd296db29fce00e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-24 14:43:52.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1992947b5a6ed3fbeb44186bcffb8df8ee80575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1992947b5a6ed3fbeb44186bcffb8df8ee80575"}]},"branch":"refs/heads/master"},"69dc382c2a14b8aabad53ea15a4c389825d6b8e0":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-08-25 17:41:42.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/15"}}},"commit":{"parents":[{"commit":"c7ad2532e0c79de3b80f9ed3e7a07fcf053a48b5","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7ad2532e0c79de3b80f9ed3e7a07fcf053a48b5"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-25 16:20:28.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69dc382c2a14b8aabad53ea15a4c389825d6b8e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/69dc382c2a14b8aabad53ea15a4c389825d6b8e0"}]},"branch":"refs/heads/master"},"53cee05c07a4cf909694b681e5caee0dc7414d22":{"kind":"REWORK","_number":16,"created":"2023-08-28 12:23:57.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/16"}}},"commit":{"parents":[{"commit":"f56eac9bb3fa4fb810c8582c77f92f4253328b0f","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f56eac9bb3fa4fb810c8582c77f92f4253328b0f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-28 12:07:37.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53cee05c07a4cf909694b681e5caee0dc7414d22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/53cee05c07a4cf909694b681e5caee0dc7414d22"}]},"branch":"refs/heads/master"},"cd77ec379689ac929930b24ca9db846b31a80b52":{"kind":"REWORK","_number":17,"created":"2023-08-29 08:58:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/17"}}},"commit":{"parents":[{"commit":"9e3ff6b5e66b9b0aefe5bdfb2c9b5c119244556f","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e3ff6b5e66b9b0aefe5bdfb2c9b5c119244556f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-29 08:07:22.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd77ec379689ac929930b24ca9db846b31a80b52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd77ec379689ac929930b24ca9db846b31a80b52"}]},"branch":"refs/heads/master"},"d0d9f28b5fbd5f79fa82fc0b565ce62fad344e85":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-08-29 14:21:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/18"}}},"commit":{"parents":[{"commit":"d33c39088375aaab0328297d307eb3d0c74ef66c","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d33c39088375aaab0328297d307eb3d0c74ef66c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-29 13:19:44.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0d9f28b5fbd5f79fa82fc0b565ce62fad344e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0d9f28b5fbd5f79fa82fc0b565ce62fad344e85"}]},"branch":"refs/heads/master"},"3a81935740de98a9aed88db2dde17baa09ce250b":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2023-08-29 18:40:41.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/19"}}},"commit":{"parents":[{"commit":"a471eb8a0239a2e7774c3a394109b18d8f26d551","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a471eb8a0239a2e7774c3a394109b18d8f26d551"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-29 17:55:10.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a81935740de98a9aed88db2dde17baa09ce250b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3a81935740de98a9aed88db2dde17baa09ce250b"}]},"branch":"refs/heads/master"},"eab1ec5d2202de51c20e10d20877449dca9ae91a":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2023-08-30 08:44:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/20"}}},"commit":{"parents":[{"commit":"0c3a43202da44951f4ad4560e41640a6bcd29604","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c3a43202da44951f4ad4560e41640a6bcd29604"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-30 08:32:44.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eab1ec5d2202de51c20e10d20877449dca9ae91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eab1ec5d2202de51c20e10d20877449dca9ae91a"}]},"branch":"refs/heads/master"},"66c17ab5542d31593435868deb3d8bace36f1201":{"kind":"REWORK","_number":21,"created":"2023-08-30 10:40:22.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/21"}}},"commit":{"parents":[{"commit":"0c3a43202da44951f4ad4560e41640a6bcd29604","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c3a43202da44951f4ad4560e41640a6bcd29604"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-30 09:44:45.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66c17ab5542d31593435868deb3d8bace36f1201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66c17ab5542d31593435868deb3d8bace36f1201"}]},"branch":"refs/heads/master"},"8eed74a5f28aa270eb474181d26be669c716e66a":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2023-08-31 07:57:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/22"}}},"commit":{"parents":[{"commit":"72f4490e968570cacdff9a988f59f207783e3677","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/72f4490e968570cacdff9a988f59f207783e3677"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-31 07:41:58.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8eed74a5f28aa270eb474181d26be669c716e66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8eed74a5f28aa270eb474181d26be669c716e66a"}]},"branch":"refs/heads/master"},"a67778ac27c59b5ebb05147cf8af40b15009c313":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-09-05 07:25:17.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/23"}}},"commit":{"parents":[{"commit":"b4beaa61bbd22fed2676480ca1de48adc7c52cbf","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4beaa61bbd22fed2676480ca1de48adc7c52cbf"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-09-05 06:49:08.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a67778ac27c59b5ebb05147cf8af40b15009c313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a67778ac27c59b5ebb05147cf8af40b15009c313"}]},"branch":"refs/heads/master"},"a1efbbb22c7161db228e7e83fb3226bb5831faf8":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-10-19 07:06:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/24"}}},"commit":{"parents":[{"commit":"cf644faaf11c77e639091684d9e6f3aadaa12373","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cf644faaf11c77e639091684d9e6f3aadaa12373"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-10-19 06:38:14.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1efbbb22c7161db228e7e83fb3226bb5831faf8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1efbbb22c7161db228e7e83fb3226bb5831faf8"}]},"branch":"refs/heads/master"},"ee2d5040d97a3f757b5ce0d2f64e550c724a38ab":{"kind":"REWORK","_number":25,"created":"2023-10-31 09:27:15.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/25"}}},"commit":{"parents":[{"commit":"cf644faaf11c77e639091684d9e6f3aadaa12373","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cf644faaf11c77e639091684d9e6f3aadaa12373"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-10-31 09:05:39.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee2d5040d97a3f757b5ce0d2f64e550c724a38ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee2d5040d97a3f757b5ce0d2f64e550c724a38ab"}]},"branch":"refs/heads/master"},"c85b0b28f2ba2c0a36265ffba23769e0e68f93ef":{"kind":"REWORK","_number":26,"created":"2023-10-31 12:48:40.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/26"}}},"commit":{"parents":[{"commit":"cf644faaf11c77e639091684d9e6f3aadaa12373","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cf644faaf11c77e639091684d9e6f3aadaa12373"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-10-31 12:48:39.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c85b0b28f2ba2c0a36265ffba23769e0e68f93ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c85b0b28f2ba2c0a36265ffba23769e0e68f93ef"}]},"branch":"refs/heads/master"},"3d1d7d0e3bd4b8bf37e7daf323833a5ace4682b4":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2023-11-02 09:35:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/27"}}},"commit":{"parents":[{"commit":"7174edf1388f21cd08b2915fc6896f461484218a","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7174edf1388f21cd08b2915fc6896f461484218a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-02 08:52:05.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d1d7d0e3bd4b8bf37e7daf323833a5ace4682b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d1d7d0e3bd4b8bf37e7daf323833a5ace4682b4"}]},"branch":"refs/heads/master"},"b5e7e392f9f47ca298c3c51090fefd1137882ae0":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2023-11-08 11:23:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/28"}}},"commit":{"parents":[{"commit":"2053e2f46ede36b06e69c77633471e6238071701","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2053e2f46ede36b06e69c77633471e6238071701"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-08 11:10:13.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5e7e392f9f47ca298c3c51090fefd1137882ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5e7e392f9f47ca298c3c51090fefd1137882ae0"}]},"branch":"refs/heads/master"},"38c6521052f44ac304d55ecc6023e235d6152a99":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2023-12-05 14:08:25.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/29"}}},"commit":{"parents":[{"commit":"de9bd0d637c0ce50627a92d1f7ac89a988040a86","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de9bd0d637c0ce50627a92d1f7ac89a988040a86"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-12-05 12:10:34.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/38c6521052f44ac304d55ecc6023e235d6152a99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/38c6521052f44ac304d55ecc6023e235d6152a99"}]},"branch":"refs/heads/master"},"65c1b260ae02ea034e1b48a49421a22ccc0ccbb4":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2024-01-02 12:01:29.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/30"}}},"commit":{"parents":[{"commit":"d2424d0431ea8eafaf4a0d01427df1491af06fee","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2424d0431ea8eafaf4a0d01427df1491af06fee"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-02 12:01:29.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65c1b260ae02ea034e1b48a49421a22ccc0ccbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65c1b260ae02ea034e1b48a49421a22ccc0ccbb4"}]},"branch":"refs/heads/master","description":"Rebase"},"eefde4a8f4b5e1e75a7ac9b34e52aa115931dfcf":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-01-11 07:36:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/31"}}},"commit":{"parents":[{"commit":"b9ac2f56d8bf317c186300a20b64cc1576ba3c40","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9ac2f56d8bf317c186300a20b64cc1576ba3c40"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-11 06:56:53.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eefde4a8f4b5e1e75a7ac9b34e52aa115931dfcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eefde4a8f4b5e1e75a7ac9b34e52aa115931dfcf"}]},"branch":"refs/heads/master"},"5a8bde61544de768839e7d7595cbc6f285e6cb25":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2024-01-22 10:58:50.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/32"}}},"commit":{"parents":[{"commit":"77e1405f72bfaf02537ae3b6133e4104600b9fef","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77e1405f72bfaf02537ae3b6133e4104600b9fef"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-22 10:47:04.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a8bde61544de768839e7d7595cbc6f285e6cb25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5a8bde61544de768839e7d7595cbc6f285e6cb25"}]},"branch":"refs/heads/master"},"4743dd803dffbfdd55edc5f93b763d7322ca578b":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2024-01-23 11:15:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/33"}}},"commit":{"parents":[{"commit":"86eb10c4396b7b280f5ba7f400fecd36575e6c5a","subject":"Fix cinder location url compatible with multistore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86eb10c4396b7b280f5ba7f400fecd36575e6c5a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:08:41.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4743dd803dffbfdd55edc5f93b763d7322ca578b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4743dd803dffbfdd55edc5f93b763d7322ca578b"}]},"branch":"refs/heads/master"},"e2f4341a3a626696d17c3d8945b76f40b0ac7438":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2024-01-24 10:53:02.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/34"}}},"commit":{"parents":[{"commit":"9be49692618877c5d91f1cadcc73c690eb401d7e","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9be49692618877c5d91f1cadcc73c690eb401d7e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-24 10:38:59.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2f4341a3a626696d17c3d8945b76f40b0ac7438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2f4341a3a626696d17c3d8945b76f40b0ac7438"}]},"branch":"refs/heads/master"},"23d73adb274e13d2815837b49cd8604c860c3ba9":{"kind":"REWORK","_number":35,"created":"2024-02-07 06:50:50.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/35"}}},"commit":{"parents":[{"commit":"d75b7447775d4fe6dee5d3b121aacc4f5619a22e","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d75b7447775d4fe6dee5d3b121aacc4f5619a22e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-07 06:29:41.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23d73adb274e13d2815837b49cd8604c860c3ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23d73adb274e13d2815837b49cd8604c860c3ba9"}]},"branch":"refs/heads/master"},"66fbb56bad8b8ea31fc120d7ccbe98176204658d":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2024-02-15 10:31:13.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/36"}}},"commit":{"parents":[{"commit":"04f45f2219b8d8b56bd052dfe494972b245c4d37","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/04f45f2219b8d8b56bd052dfe494972b245c4d37"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-15 10:15:55.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66fbb56bad8b8ea31fc120d7ccbe98176204658d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66fbb56bad8b8ea31fc120d7ccbe98176204658d"}]},"branch":"refs/heads/master"},"29a22de838f152ef5916ebaa91fca7347bc2e273":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2024-02-20 10:23:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/37"}}},"commit":{"parents":[{"commit":"bbba4c28b756341ee0969658f6118120b86db44d","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bbba4c28b756341ee0969658f6118120b86db44d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-20 09:52:56.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29a22de838f152ef5916ebaa91fca7347bc2e273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29a22de838f152ef5916ebaa91fca7347bc2e273"}]},"branch":"refs/heads/master"},"b3097fb11f322ba37c55f677af4265140083afe6":{"kind":"REWORK","_number":38,"created":"2024-02-21 12:59:22.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/38"}}},"commit":{"parents":[{"commit":"03c126d4e22595c7ff64ebe7987da43bfb4bcc27","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03c126d4e22595c7ff64ebe7987da43bfb4bcc27"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-21 12:42:20.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3097fb11f322ba37c55f677af4265140083afe6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3097fb11f322ba37c55f677af4265140083afe6"}]},"branch":"refs/heads/master"},"f605e9ff464bc5abbf5dc78a6f004130d809e628":{"kind":"REWORK","_number":39,"created":"2024-02-22 07:41:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/39"}}},"commit":{"parents":[{"commit":"b4c49efa410ee89c943408647a41c4d1a12b24b0","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4c49efa410ee89c943408647a41c4d1a12b24b0"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-22 07:26:32.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f605e9ff464bc5abbf5dc78a6f004130d809e628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f605e9ff464bc5abbf5dc78a6f004130d809e628"}]},"branch":"refs/heads/master"},"fe7323516a55c58866d20fbab8c0043629b5d180":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2024-02-22 10:59:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/40"}}},"commit":{"parents":[{"commit":"1ec8667b42169687e56fda9b5ce8d5a9525379cb","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ec8667b42169687e56fda9b5ce8d5a9525379cb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-22 10:59:52.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fe7323516a55c58866d20fbab8c0043629b5d180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fe7323516a55c58866d20fbab8c0043629b5d180"}]},"branch":"refs/heads/master","description":"Rebase"},"39c5a11d934d38504025cbfaf4419a8b19bc2971":{"kind":"REWORK","_number":41,"created":"2024-02-26 09:16:40.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/41"}}},"commit":{"parents":[{"commit":"5f275b1766eeab764cf57a5fcc3c7ff2a057018c","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f275b1766eeab764cf57a5fcc3c7ff2a057018c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-26 09:03:00.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39c5a11d934d38504025cbfaf4419a8b19bc2971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/39c5a11d934d38504025cbfaf4419a8b19bc2971"}]},"branch":"refs/heads/master"},"c537f94f823de5a691b456d17d9fc96c06b3f1a4":{"kind":"REWORK","_number":42,"created":"2024-02-27 07:12:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/42"}}},"commit":{"parents":[{"commit":"e1a652c2961b7c145ee261c903c25f991471b72b","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1a652c2961b7c145ee261c903c25f991471b72b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-27 07:02:10.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c537f94f823de5a691b456d17d9fc96c06b3f1a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c537f94f823de5a691b456d17d9fc96c06b3f1a4"}]},"branch":"refs/heads/master"},"5f78cbe4c9154ac6f21cc9e845565abe7a4e327a":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2024-02-27 13:25:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/43"}}},"commit":{"parents":[{"commit":"c82302ef0731378257025cb23e95896e37d82a58","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c82302ef0731378257025cb23e95896e37d82a58"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-27 13:10:00.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f78cbe4c9154ac6f21cc9e845565abe7a4e327a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f78cbe4c9154ac6f21cc9e845565abe7a4e327a"}]},"branch":"refs/heads/master"},"5ed51eb0dac239a6309573e73562cc8eea0ea937":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2024-02-27 15:26:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/44"}}},"commit":{"parents":[{"commit":"3ffaad27b185c4681bddf2c5135b9a4339d7afe1","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ffaad27b185c4681bddf2c5135b9a4339d7afe1"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-27 15:20:38.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ed51eb0dac239a6309573e73562cc8eea0ea937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ed51eb0dac239a6309573e73562cc8eea0ea937"}]},"branch":"refs/heads/master"},"61bb93d9a359eb58fb9d9a9cec370e5269f29219":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2024-02-28 03:51:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/45"}}},"commit":{"parents":[{"commit":"4b6bc6d758c1567e20d7a3dbdb895c3462aa8e50","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b6bc6d758c1567e20d7a3dbdb895c3462aa8e50"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 03:24:07.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61bb93d9a359eb58fb9d9a9cec370e5269f29219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/61bb93d9a359eb58fb9d9a9cec370e5269f29219"}]},"branch":"refs/heads/master"},"a22df7f3863d068d55a836042731229f8f74c2ef":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2024-02-28 09:10:56.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/46"}}},"commit":{"parents":[{"commit":"4bd3a0d7783b0da38b7852b954bd271ac9065f6b","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bd3a0d7783b0da38b7852b954bd271ac9065f6b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 09:03:39.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a22df7f3863d068d55a836042731229f8f74c2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a22df7f3863d068d55a836042731229f8f74c2ef"}]},"branch":"refs/heads/master"},"3142180e716beb62d916d306a2eb744f6efc9d67":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2024-02-28 11:10:35.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/47"}}},"commit":{"parents":[{"commit":"1756de480dfd9c5c04a6f45dc9c4949c84737a1f","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1756de480dfd9c5c04a6f45dc9c4949c84737a1f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 11:05:17.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3142180e716beb62d916d306a2eb744f6efc9d67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3142180e716beb62d916d306a2eb744f6efc9d67"}]},"branch":"refs/heads/master"},"8032ad5a0eccfd5a200191f1688318723345e2e9":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2024-02-28 16:14:46.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/48"}}},"commit":{"parents":[{"commit":"da2c2e21a7997f5784c93d0dec199bc5c306b91e","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da2c2e21a7997f5784c93d0dec199bc5c306b91e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 16:08:18.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8032ad5a0eccfd5a200191f1688318723345e2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8032ad5a0eccfd5a200191f1688318723345e2e9"}]},"branch":"refs/heads/master"},"3434f1d3d3ddd531f139094db697f8ff8a1a1273":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2024-02-28 19:01:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/49"}}},"commit":{"parents":[{"commit":"52b952720fec2084777593fa6a0a5b6d00b04995","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/52b952720fec2084777593fa6a0a5b6d00b04995"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 18:55:18.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3434f1d3d3ddd531f139094db697f8ff8a1a1273"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3434f1d3d3ddd531f139094db697f8ff8a1a1273"}]},"branch":"refs/heads/master"},"b928f9671d09a449af70353800e900ea2eebc651":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2024-02-28 19:21:23.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/50"}}},"commit":{"parents":[{"commit":"2f806d3285de2591166f0056f32602f4993986b9","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f806d3285de2591166f0056f32602f4993986b9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 19:14:52.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b928f9671d09a449af70353800e900ea2eebc651"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b928f9671d09a449af70353800e900ea2eebc651"}]},"branch":"refs/heads/master"},"0a01fb76430b0c4ba65a60f6a139b84015ccec8e":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2024-02-29 10:17:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/51"}}},"commit":{"parents":[{"commit":"5e05c0f16ebff1ab21f6b39768208c16630e90f4","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e05c0f16ebff1ab21f6b39768208c16630e90f4"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-29 10:05:32.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a01fb76430b0c4ba65a60f6a139b84015ccec8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a01fb76430b0c4ba65a60f6a139b84015ccec8e"}]},"branch":"refs/heads/master"},"082e6f061c595502f385668c5593dfaf80b5be69":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2024-02-29 11:03:29.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/52"}}},"commit":{"parents":[{"commit":"de08233f8e264a496c681b53794a7cc878626c36","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de08233f8e264a496c681b53794a7cc878626c36"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-29 10:54:39.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/082e6f061c595502f385668c5593dfaf80b5be69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/082e6f061c595502f385668c5593dfaf80b5be69"}]},"branch":"refs/heads/master"},"3f85cf7d5fbd6e508268d026fe7a862be2416fae":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2024-04-04 07:00:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/53"}}},"commit":{"parents":[{"commit":"f34d2db9aa82bfb0a1149a10e9a42128ef36e98a","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f34d2db9aa82bfb0a1149a10e9a42128ef36e98a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-04 06:08:53.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f85cf7d5fbd6e508268d026fe7a862be2416fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f85cf7d5fbd6e508268d026fe7a862be2416fae"}]},"branch":"refs/heads/master"},"f4c31c74868585adaa566a7c01c2c8b77fdba187":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2024-04-08 06:31:24.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/54"}}},"commit":{"parents":[{"commit":"b71e2b6381f21cfd212921c172e46afb27354484","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b71e2b6381f21cfd212921c172e46afb27354484"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-08 06:21:31.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4c31c74868585adaa566a7c01c2c8b77fdba187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4c31c74868585adaa566a7c01c2c8b77fdba187"}]},"branch":"refs/heads/master"},"7cc98d201087babda285d6a62ca9c70af9c0555e":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2024-04-29 08:04:05.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/55"}}},"commit":{"parents":[{"commit":"4dd29158af83e7c90b242ba0b7c7952aed455b14","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dd29158af83e7c90b242ba0b7c7952aed455b14"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-29 07:56:22.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cc98d201087babda285d6a62ca9c70af9c0555e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7cc98d201087babda285d6a62ca9c70af9c0555e"}]},"branch":"refs/heads/master"},"9910122c582b0e177d78bd31380a367e6654c64a":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2024-04-30 09:42:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/56"}}},"commit":{"parents":[{"commit":"00cd18f8a046d44c7e8ca57bdda568b5b8bb92d4","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00cd18f8a046d44c7e8ca57bdda568b5b8bb92d4"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-30 09:30:28.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9910122c582b0e177d78bd31380a367e6654c64a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9910122c582b0e177d78bd31380a367e6654c64a"}]},"branch":"refs/heads/master"},"ba70c4a381b29b55596952b3a17fcbf67a4f8eca":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2024-05-07 06:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/57"}}},"commit":{"parents":[{"commit":"30a7301d6e82ca2580deaaa5a4e387d1b4f1ceb1","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30a7301d6e82ca2580deaaa5a4e387d1b4f1ceb1"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-06 12:53:52.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba70c4a381b29b55596952b3a17fcbf67a4f8eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba70c4a381b29b55596952b3a17fcbf67a4f8eca"}]},"branch":"refs/heads/master"},"1fd9dcda55d6071940a8f16f3dea1e42783ed5ae":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2024-05-07 13:08:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/58"}}},"commit":{"parents":[{"commit":"17472ed7ae395afa9f7bfa199d1f4c3f52642c7c","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17472ed7ae395afa9f7bfa199d1f4c3f52642c7c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-07 13:01:39.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fd9dcda55d6071940a8f16f3dea1e42783ed5ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fd9dcda55d6071940a8f16f3dea1e42783ed5ae"}]},"branch":"refs/heads/master"},"ade150ea6f717a78d851ac6878a9fe4c78af6905":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2024-05-08 09:01:11.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/59"}}},"commit":{"parents":[{"commit":"2b662078ab969223386e59d8b0226fb0aca9ff99","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b662078ab969223386e59d8b0226fb0aca9ff99"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-08 08:53:05.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ade150ea6f717a78d851ac6878a9fe4c78af6905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ade150ea6f717a78d851ac6878a9fe4c78af6905"}]},"branch":"refs/heads/master"},"c6ba6192077e7f8857b88a50bf8b191d0911f69d":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2024-06-06 06:46:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/60"}}},"commit":{"parents":[{"commit":"2daa78f28a74ed736dacaffff2c69fc8be3d4fa9","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2daa78f28a74ed736dacaffff2c69fc8be3d4fa9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-06 06:27:59.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c6ba6192077e7f8857b88a50bf8b191d0911f69d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c6ba6192077e7f8857b88a50bf8b191d0911f69d"}]},"branch":"refs/heads/master"},"a699d707deac16f0459024e2eb5680431ca7c43d":{"kind":"TRIVIAL_REBASE","_number":61,"created":"2024-06-14 10:35:46.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/61"}}},"commit":{"parents":[{"commit":"3e255374bb759c6e00e62b8c62e7fd2379791f8c","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3e255374bb759c6e00e62b8c62e7fd2379791f8c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-14 10:10:41.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a699d707deac16f0459024e2eb5680431ca7c43d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a699d707deac16f0459024e2eb5680431ca7c43d"}]},"branch":"refs/heads/master"},"8e1b2a1a2cce3b17418dd35aa8bcf0ed72c3c49d":{"kind":"REWORK","_number":62,"created":"2024-06-28 12:46:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/62"}}},"commit":{"parents":[{"commit":"b87637b604891ba9654dba16d05afda360665b9e","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b87637b604891ba9654dba16d05afda360665b9e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-28 12:33:40.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e1b2a1a2cce3b17418dd35aa8bcf0ed72c3c49d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e1b2a1a2cce3b17418dd35aa8bcf0ed72c3c49d"}]},"branch":"refs/heads/master"},"e3ff11d859e9eddf4aea55878b60cfe5fc7a557e":{"kind":"REWORK","_number":63,"created":"2024-07-16 10:22:56.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/63"}}},"commit":{"parents":[{"commit":"8c44c350c72fe9b5fc04250ce4fbcb78688ebfbb","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c44c350c72fe9b5fc04250ce4fbcb78688ebfbb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-16 10:08:15.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3ff11d859e9eddf4aea55878b60cfe5fc7a557e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3ff11d859e9eddf4aea55878b60cfe5fc7a557e"}]},"branch":"refs/heads/master"},"da1116eb78b00d0c01a22ab902b40c2c07b475b6":{"kind":"TRIVIAL_REBASE","_number":64,"created":"2024-07-16 10:54:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/64"}}},"commit":{"parents":[{"commit":"2b9967e75ac06711a2197f1357aeec34288310cf","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b9967e75ac06711a2197f1357aeec34288310cf"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-16 10:44:18.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da1116eb78b00d0c01a22ab902b40c2c07b475b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da1116eb78b00d0c01a22ab902b40c2c07b475b6"}]},"branch":"refs/heads/master"},"3055e417be0c30a701224e341ed60cf2f777aed1":{"kind":"REWORK","_number":65,"created":"2024-07-22 07:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/65"}}},"commit":{"parents":[{"commit":"20010512e87e0fbbfad168f077f39d40401d6a74","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20010512e87e0fbbfad168f077f39d40401d6a74"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-22 06:42:12.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3055e417be0c30a701224e341ed60cf2f777aed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3055e417be0c30a701224e341ed60cf2f777aed1"}]},"branch":"refs/heads/master"},"b5f93670e58459db12f2076599dd61b4e9924e7e":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2024-07-22 07:44:37.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/66"}}},"commit":{"parents":[{"commit":"688c69cd5eb5b8a10ddf5834718f05062c5f1f6c","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/688c69cd5eb5b8a10ddf5834718f05062c5f1f6c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-22 07:32:13.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5f93670e58459db12f2076599dd61b4e9924e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5f93670e58459db12f2076599dd61b4e9924e7e"}]},"branch":"refs/heads/master"},"504f4cd897f0e388ace60e7e8f5d9f5648a6b047":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2024-07-25 06:53:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/67"}}},"commit":{"parents":[{"commit":"63a20d458a26a0c56bd387883980f1e70642a44d","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63a20d458a26a0c56bd387883980f1e70642a44d"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-25 05:50:32.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/504f4cd897f0e388ace60e7e8f5d9f5648a6b047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/504f4cd897f0e388ace60e7e8f5d9f5648a6b047"}]},"branch":"refs/heads/master"},"b83f38cf25428983791017b74e92f3c44c41c6d4":{"kind":"REWORK","_number":68,"created":"2024-07-25 10:32:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/47/886947/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/886947/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/886947/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/886947/68"}}},"commit":{"parents":[{"commit":"4281558dff399dd9e7b462a499fa5a925245643a","subject":"Add new add location api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4281558dff399dd9e7b462a499fa5a925245643a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-06-26 07:44:37.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-25 10:22:17.000000000","tz":0},"subject":"Add functional tests for new add-location API","message":"Add functional tests for new add-location API\n\nRelated blueprint new-location-apis\n\nChange-Id: If94d7d1fbec07a49222b65aa3e530748a95eaf5b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b83f38cf25428983791017b74e92f3c44c41c6d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b83f38cf25428983791017b74e92f3c44c41c6d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
