)]}'
{"id":"openstack%2Fglance~887240","triplet_id":"openstack%2Fglance~master~I53c136c728fcffbca07eb5d0a67f9edc5e723ad0","project":"openstack/glance","branch":"master","topic":"correct-numbering","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-07-19 18:39:12.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-07-19 16:01:59.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-07-19 18:39:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I53c136c728fcffbca07eb5d0a67f9edc5e723ad0","subject":"Correct item numbering in install docs","status":"MERGED","created":"2023-06-29 01:32:54.000000000","updated":"2023-07-19 18:40:54.000000000","submitted":"2023-07-19 18:39:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887240-correct-numbering","meta_rev_id":"54c56a4bd1427af7a804c3bc211fb959a2fe0cc3","_number":887240,"virtual_id_number":887240,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-19 18:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-07-19 15:40:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-07-19 16:01:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2023-07-19 16:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 01:46:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-18 13:51:08.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-07-19 15:14:29.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"}],"messages":[{"id":"78942a9fa75da754df6e8998bc2d68648580e102","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-06-29 01:32:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"596358840a587e96094880ab590e5b70b81a711c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 01:46:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f7f673ce93648ffadb1f6ecb5ccbefd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a479d5a1c1d45d2923f5ce4b83fa3ec : FAILURE in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d4b73364b34e1c95b374dff7ef62f7 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b2c17d52d36a30751cdbe5bcf440dc6249da6c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-06-29 02:41:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dca30515d4b33a412c243374592d0ca88fa20ab3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 02:51:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9208006c217b405289a0375286e5e15d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9edf0486988b4c3aae5dcdca4b74e1c3 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5fcf79417454251b4e1f7f67414b2ed : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":2},{"id":"938daebd97127669ddaf1e2157c02a6a2b5898a6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-07-18 13:51:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"96f18f27a8a2633896408a84ff05bba99a6be2a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-19 15:10:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1b29d82315824ccda8c87a1495199392f62c308c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-19 15:14:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"aaf1176eed5c3a15b6d5ed16671e6f0797e722d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:28:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5578217b468c418b87e2b1915d119f6b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e2bd3e7dbd4d2aa86663d511b01c5c : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35f4c2f804b4deeb4170c19cd31c938 : SUCCESS in 6m 14s","accounts_in_message":[],"_revision_number":3},{"id":"14615932983447f9b1ce0ab4acab9ad35c9eaf73","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-19 15:40:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7e1f7db5909ffa27eb00d2f5a8d4fc33435b5558","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-07-19 16:01:59.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ff0b66570cc9f904bbdcef913a50b85bfaa5b6ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 16:02:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9d46409d111fb2a1984b3f5b5056a15d316e8d1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 18:39:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5e122590f1348a79a2e7204abff2f58\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d5183b3e3fa429b9b7a363bca9548d2 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c3cb3703c2f4d4baff70592df2371f4 : SUCCESS in 7m 32s","accounts_in_message":[],"_revision_number":3},{"id":"131576fd03b0ebbbbcf73b7082f4caa6023b14a5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 18:39:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"54c56a4bd1427af7a804c3bc211fb959a2fe0cc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 18:40:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fb0b13a1dba4dea9ed2025288f11988\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86732b7db1764170bdcb8cd5d389fa46 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2fd57d68f556a92221a626f71ecb19e59ed4ca86","revisions":{"6e56bb72286f7b40c38dbbc64eb6979849b6151c":{"kind":"REWORK","_number":1,"created":"2023-06-29 01:32:54.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/40/887240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/887240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/887240/1"}}},"commit":{"parents":[{"commit":"97551b953b6e69c2411578f5a8675e332450d71e","subject":"Doc: remove \"stores\" and \"default_store\" options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97551b953b6e69c2411578f5a8675e332450d71e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 01:28:33.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 01:28:33.000000000","tz":-240},"subject":"Correct item numbering in install docs","message":"Correct item numbering in install docs\n\nIndicating a new enumerated point by a \u0027#.\u0027 in an included doc\nrestarts the numbering at 1 instead of continuing the current\nsequence.  Move the \u0027#.\u0027 sentence out of the included doc and\ninto the individual install docs so that the points are enumerated\ncorrectly.  Also remove one level of indentation from the included\nfile and instead indent the \u0027include\u0027 directive.\n\nChange-Id: I53c136c728fcffbca07eb5d0a67f9edc5e723ad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e56bb72286f7b40c38dbbc64eb6979849b6151c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e56bb72286f7b40c38dbbc64eb6979849b6151c"}]},"branch":"refs/heads/master"},"eade518d543c775affdd97f1611117b4722ea6ab":{"kind":"REWORK","_number":2,"created":"2023-06-29 02:41:19.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/40/887240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/887240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/887240/2"}}},"commit":{"parents":[{"commit":"97551b953b6e69c2411578f5a8675e332450d71e","subject":"Doc: remove \"stores\" and \"default_store\" options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97551b953b6e69c2411578f5a8675e332450d71e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 01:28:33.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 02:36:57.000000000","tz":-240},"subject":"Correct item numbering in install docs","message":"Correct item numbering in install docs\n\nIndicating a new enumerated point by a \u0027#.\u0027 in an included doc\nrestarts the numbering at 1 instead of continuing the current\nsequence.  Move the \u0027#.\u0027 sentence out of the included doc and\ninto the individual install docs so that the points are enumerated\ncorrectly.  Also remove one level of indentation from the included\nfile and instead indent the \u0027include\u0027 directive.\n\n(Also had to break two lines in register-quotas.rst due to doc8\nD001.  Not sure why that suddenly broke now after I shortened\nall the lines in that file!)\n\nChange-Id: I53c136c728fcffbca07eb5d0a67f9edc5e723ad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eade518d543c775affdd97f1611117b4722ea6ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eade518d543c775affdd97f1611117b4722ea6ab"}]},"branch":"refs/heads/master"},"2fd57d68f556a92221a626f71ecb19e59ed4ca86":{"kind":"REWORK","_number":3,"created":"2023-07-19 15:10:18.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/40/887240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/40/887240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/40/887240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/40/887240/3"}}},"commit":{"parents":[{"commit":"97551b953b6e69c2411578f5a8675e332450d71e","subject":"Doc: remove \"stores\" and \"default_store\" options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/97551b953b6e69c2411578f5a8675e332450d71e"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 01:28:33.000000000","tz":-240},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2023-07-19 15:08:49.000000000","tz":120},"subject":"Correct item numbering in install docs","message":"Correct item numbering in install docs\n\nIndicating a new enumerated point by a \u0027#.\u0027 in an included doc\nrestarts the numbering at 1 instead of continuing the current\nsequence.  Move the \u0027#.\u0027 sentence out of the included doc and\ninto the individual install docs so that the points are enumerated\ncorrectly.  Also remove one level of indentation from the included\nfile and instead indent the \u0027include\u0027 directive.\n\n(Also had to break two lines in register-quotas.rst due to doc8\nD001.  Not sure why that suddenly broke now after I shortened\nall the lines in that file!)\n\nChange-Id: I53c136c728fcffbca07eb5d0a67f9edc5e723ad0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2fd57d68f556a92221a626f71ecb19e59ed4ca86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2fd57d68f556a92221a626f71ecb19e59ed4ca86"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
