)]}'
{"id":"openstack%2Fglance~887241","triplet_id":"openstack%2Fglance~master~I47e8ceccecadb469c27cbf203fe77528cc38d95d","project":"openstack/glance","branch":"master","topic":"docs-clean","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2023-08-15 03:31:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I47e8ceccecadb469c27cbf203fe77528cc38d95d","subject":"Clean generated source RST files","status":"MERGED","created":"2023-06-29 01:38:12.000000000","updated":"2023-08-15 03:32:34.000000000","submitted":"2023-08-15 03:31:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887241-docs-clean","meta_rev_id":"a503690cd691dc0a5150be8c6a9692eee3b0316e","_number":887241,"virtual_id_number":887241,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-15 03:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2023-07-19 15:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2023-08-15 02:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2023-08-15 02:33:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-29 01:52:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-06-29 02:27:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-18 13:51:26.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-08-15 02:33:48.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"0c9094320d3b74b335358fe64716636b47123236","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-06-29 01:38:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd908c492498c225cd5997fd2b2cdef19d5b0279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 01:52:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/870d6fac0d3641aab488d15bc2bea515\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6775d9e8f7904e04bdcafbc8f4638753 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3152175c29a4066939770ca36a5fc97 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3186b21a0e564f23a83882626ec84565 : SUCCESS in 11m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db79742f042d939d356185d45f674844790c6f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 02:27:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e20666930b07473baae4910a2f0dc499\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1dcb556510de402db3bc9f2b71064a7a : SUCCESS in 3m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3faca28fba8146ae96aaf65adafe7474 : SUCCESS in 44m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d836d9c0da25488b8de18220c1e083a8 : FAILURE in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7161652f26ad4f9087e9b56a2fd19d8f : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1436decb670743eeb3ed62c53c416c75 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/936a1ba468ec43828ffd49bf99da4055 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21eb4af9d84640febbf19b286c89af66 : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa2e0f3fc2b640a1b3455db75f75d385 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e896abf0d52f471c8d4cf25a6a00eed7 : SUCCESS in 10m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ccd4a6483bd44f5903e89464957361c : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8d6a2b7edc064affb6d4a414c1956bcb : SUCCESS in 7m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a1eddea302ce4e1b97af8988291efe43 : SUCCESS in 7m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3dbada6efee84d1f80bccf1eb82cde8f : SUCCESS in 22m 51s","accounts_in_message":[],"_revision_number":1},{"id":"3e626b43a34c80afb80497b29200907832a198a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-06-29 02:43:56.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fb8a17ea7eee79f3c5cc0f69cfd925efa190fdd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 03:07:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1174106c18d4566a3a099fe4fb46c74\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b203d2f1f4d8428e873312cfc0bf0695 : SUCCESS in 20m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8c126355132e48a991db306489136c83 : NODE_FAILURE Node request 300-0021572340 failed in 0s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/38ee6861810f4211b60788fb8f3f8df4 : SUCCESS in 21m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b6d76f3ad821b0402173cdcb78b1d6741ae848d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-29 04:36:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8b04bb079d741f7863d2de3d9dc675b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/eeca06865dec445faf087475981b1655 : SUCCESS in 5m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3df5d02e43f9447c8c003cef52ee37ff : SUCCESS in 1h 38m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1826ca62276f4bdea4a68f2719e2a05f : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4458548ca6d946368bdfd36b80c1aec1 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05ab522966374a8cb54cc35e38c63e8c : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30c4f55d5a1b4d19a1ebae17b85c1ff0 : SUCCESS in 12m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a99a349f2445e196f0cc9883276018 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1bd3624931f146cf91f06fea2ca439ee : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/184e7e89d735485595413e5ef7b129cf : SUCCESS in 12m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8196ae1d1d1644f8ac8e50c4c9856a66 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bf7a19d63d214a1ca62e4c71c60cc216 : SUCCESS in 13m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5510452045f147a5ac04d8cfd0f39451 : SUCCESS in 16m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7b2a589ee77f4fcc9e7d00c9760a6d8f : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":2},{"id":"77dd8903a6ee38623c63d3771b74a19b760f6606","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-07-18 13:51:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"952f781843deed2edb93c26aa176bfdd017816a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-07-19 15:41:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e821b104cadf91805ced1b3427151c31257490ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 15:48:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13fb02eed0744e46bdbdb2a08efaa275\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8ba91417aa74bf8a3549e25900ec47a : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/30c0e1c1eda04d7ea97152d643da5a1f : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13a80b7a023f4951a44bbe68914f319a : SUCCESS in 4m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5e06e0b9e68a3a469b0f5ddebaf8f511b0c26674","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-19 16:46:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c030bf394aa7430d825c5b52c6646648\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1ba0a6527c1f4cca91a4ef42482466a3 : SUCCESS in 4m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a7d6f804b4c49b2bbf2450515e88e50 : SUCCESS in 1h 01m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54346a24c3474963890054fbdc1511ef : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d17273cb35af4e3cab6d28f6716e71b2 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32412d2edc4d4fccab1785ca1a148d3c : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a79690bd5ca4489bc4026677676f41b : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050498195c0b47659fafd1fd2bcbc2d6 : SUCCESS in 6m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bb5ac01987b4b6ba9657f349e031a10 : SUCCESS in 6m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8a9bbee77cf14f1793d1416858adfe79 : SUCCESS in 13m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a6d7a9eee66489fb5e2563b36f4a1de : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a9a1e486d65347b6b824baaf727c9b3c : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/03f96e40ec7b4b2297d00474222399b1 : SUCCESS in 8m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1274e527d5ef456d811d46d3c33c89a7 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":3},{"id":"d5981a318ad5150b81aa2353ad6af7f7834c03d5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-08-15 02:33:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"818dd254074b4814f02847898f9cd1f6f6c30674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 02:34:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fc8eaf6b90b33908cc87976c11d4e83ca44f3389","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 03:31:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63578a1a925b432e86f321e9a1fa51dc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7ea59b63e77f4f189c4297d918defe2c : SUCCESS in 4m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2590e15afecc4e929e30590bc86c4464 : SUCCESS in 54m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae085215d9e48d98ed00dee7ce5d1b2 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62bda088f7904daebc3973a051a42362 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e2526dd826145349e4a5fdde46b5b04 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79b1c72fadf74d82bd736db204f0ae51 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd412a03f3a40c1be3c6ed726dc4ab9 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89036dd6c2044fa581d3b6305d5e44d7 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4dd3180f30114d4d908fd72fe10a68d9 : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2db3cd598caf44598f847545487abe39 : SUCCESS in 8m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c99b35b750fb475781eb930dedc08b2f : SUCCESS in 25m 36s","accounts_in_message":[],"_revision_number":3},{"id":"6d6548c34560552941269205fc56c69824a73c90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 03:31:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a503690cd691dc0a5150be8c6a9692eee3b0316e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 03:32:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edff329076c644e7a586f7650f983a62\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/703a1836efaa44fa9b1a4c48cad42bf5 : SUCCESS in 41s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4d0441564b44b9685dc7ce27bb09687 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5af2d93fe5014c089846b0c40fd6c404 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0bc55d4f830c49151c1f059888e8ef368e4ba4f3","revisions":{"a626a3fad194f05420a3b50046f242d92db3eb2f":{"kind":"REWORK","_number":1,"created":"2023-06-29 01:38:12.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/887241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/887241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/887241/1"}}},"commit":{"parents":[{"commit":"6e56bb72286f7b40c38dbbc64eb6979849b6151c","subject":"Correct item numbering in install docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e56bb72286f7b40c38dbbc64eb6979849b6151c"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 00:36:55.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 01:37:03.000000000","tz":-240},"subject":"Clean generated source RST files","message":"Clean generated source RST files\n\nThe tox docs env generates RST files in doc/source/contributor/api/\nand these aren\u0027t updated when modules are removed from the glance\nsource code, and can cause errors trying to build the docs.  So\nclean this dir out (like we do with doc/build/) before building\nthe docs.\n\nChange-Id: I47e8ceccecadb469c27cbf203fe77528cc38d95d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a626a3fad194f05420a3b50046f242d92db3eb2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a626a3fad194f05420a3b50046f242d92db3eb2f"}]},"branch":"refs/heads/master"},"ce025d88e5b151ad69a50ebd188b2ce4ec8f6baf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-29 02:43:56.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/887241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/887241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/887241/2"}}},"commit":{"parents":[{"commit":"eade518d543c775affdd97f1611117b4722ea6ab","subject":"Correct item numbering in install docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eade518d543c775affdd97f1611117b4722ea6ab"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 00:36:55.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 02:42:55.000000000","tz":-240},"subject":"Clean generated source RST files","message":"Clean generated source RST files\n\nThe tox docs env generates RST files in doc/source/contributor/api/\nand these aren\u0027t updated when modules are removed from the glance\nsource code, and can cause errors trying to build the docs.  So\nclean this dir out (like we do with doc/build/) before building\nthe docs.\n\nChange-Id: I47e8ceccecadb469c27cbf203fe77528cc38d95d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce025d88e5b151ad69a50ebd188b2ce4ec8f6baf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce025d88e5b151ad69a50ebd188b2ce4ec8f6baf"}]},"branch":"refs/heads/master"},"0bc55d4f830c49151c1f059888e8ef368e4ba4f3":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-19 15:41:53.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/41/887241/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/887241/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/887241/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/887241/3"}}},"commit":{"parents":[{"commit":"2fd57d68f556a92221a626f71ecb19e59ed4ca86","subject":"Correct item numbering in install docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2fd57d68f556a92221a626f71ecb19e59ed4ca86"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-06-29 00:36:55.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2023-07-19 15:41:53.000000000","tz":0},"subject":"Clean generated source RST files","message":"Clean generated source RST files\n\nThe tox docs env generates RST files in doc/source/contributor/api/\nand these aren\u0027t updated when modules are removed from the glance\nsource code, and can cause errors trying to build the docs.  So\nclean this dir out (like we do with doc/build/) before building\nthe docs.\n\nChange-Id: I47e8ceccecadb469c27cbf203fe77528cc38d95d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bc55d4f830c49151c1f059888e8ef368e4ba4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bc55d4f830c49151c1f059888e8ef368e4ba4f3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
