)]}'
{"id":"openstack%2Fglance~887475","triplet_id":"openstack%2Fglance~master~If22e04649fb9a5c237ca59e3844974756303d127","project":"openstack/glance","branch":"master","topic":"bobcat-milestone-2","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-07-09 08:04:47.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-07-05 07:26:56.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-07-04 13:52:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"If22e04649fb9a5c237ca59e3844974756303d127","subject":"Refresh Glance example configs for bobcat milestone 2","status":"MERGED","created":"2023-07-03 06:50:58.000000000","updated":"2023-07-09 11:11:39.000000000","submitted":"2023-07-09 11:10:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887475-bobcat-milestone-2","meta_rev_id":"e9c20d39e55466087a0964372c4c328d27ddd5a5","_number":887475,"virtual_id_number":887475,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-09 11:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-07-05 07:26:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-07-04 13:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-07-08 07:12:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-07-05 07:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-03 06:59:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-03 07:43:48.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-03 07:43:48.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-07-03 08:50:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-08 07:12:03.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"041d39fc782b43fc66bb1b5707308051e5059f21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-03 06:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3463c6c318f7eb8aa1a999095c2430e16b3b717e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 06:59:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbd0a8b2d4b04c269333529a03981c82\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a80aa31d27a47d6b81af9468e24fea6 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c08e03f05f545cdb7b1d6ab7a0f8bbd : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08bff8ef92554164967d457c7d4333f5 : SUCCESS in 4m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"53efec29a7f6d5a3e609b26d6ce04a1ad080d560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-03 08:50:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7895a2fa38be48d4a44cb36b36726756\n\n- grenade https://zuul.opendev.org/t/openstack/build/70380e94b7504492a3066545a2fc3784 : SUCCESS in 46m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c3c87a350b444e0aaa439bddca7e7a0 : SUCCESS in 1h 38m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bcd52d5f98c14e619c532a1ea494c84b : SUCCESS in 49m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b6e8c36a0af4c42b54ba8fd6d8c8544 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a115442c19984ae88e096c1ee3581d0c : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76db3631a60848feaba4e05ae5865596 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c168bc217c74c698a20bbab8e063bf3 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075a24957ce44167a6fbe99ac68dae17 : SUCCESS in 6m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7f23194aa07945fb87a1ffdd4042675e : SUCCESS in 11m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9fccf7ee30f4d57bb14ec317329aec5 : SUCCESS in 7m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73a40310c015491da089c2a91acbfff2 : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a8aeb164dd4a472eb9737c9efee5fdc8 : SUCCESS in 8m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fdc56173d894b7f9314c123ad4cabfc : SUCCESS in 1h 25m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e15058ef4fc41889cf714f5ac2bda68 : SUCCESS in 1h 34m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/89f6160946ab41ef9fa16e95f9decc0a : SUCCESS in 1h 34m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/989cbb89f8bb4abfa77e965b60ab7d92 : SUCCESS in 1h 53m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/abdc693517f6450cbb1cbb68b391b19b : FAILURE in 33m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9b40c8127a14fcbbf7f475a3a26c01e : SUCCESS in 59m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/109963e5832f4d9d9fa024c93395b04e : SUCCESS in 1h 35m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/99690a8f1f2c4afe8312ddeb86f1cf97 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2e793931d73bed800bd557daaecff476906ff7f9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-04 13:52:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c824f04e2ebe49e4b763591bc05a29a241515bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-07-05 07:26:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"95bd60664406c32f3215b6211a263104382b49df","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-07-05 07:29:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0112830cc16e9fdcd92eed2bae89d8d6a3a8215","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 07:40:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9727a0023af44d4ba912d42669a2419d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2295b0d1b75845fab3167040e511114f : SUCCESS in 9m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38f96e092f71419a8ce96549968bb264 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5858fa5382ec4c659c6f19c97110f2a6 : POST_FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b34def166f58c597e1cec361588f519c0958da9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 09:32:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5444daa02d3e4b8886e63178384704bc\n\n- grenade https://zuul.opendev.org/t/openstack/build/3cedd90b801e407fa8f03d03020dc104 : FAILURE in 50m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/623fadf9d1954691892e94616782fd4a : SUCCESS in 1h 40m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f58eb1634d554f5588512fb5f62be9f3 : SUCCESS in 1h 04m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f541f2c64ce840ceb0a1e5d76d3b6cba : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/426b1559db65404c88ef0eed5ad51367 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76b48a23421d4be798486788b9bc68eb : POST_FAILURE in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55f94aafa6fa47759887b0908f8b8ece : POST_FAILURE in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c69fda84ed43baba52d2ffa5ef44e5 : POST_FAILURE in 5m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5c23499055634fcabc5fe13dacf81cea : SUCCESS in 18m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9bb9d51f1344014b06a0a41453cf424 : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/30bef3b1cd174c108224406f057de4d0 : SUCCESS in 10m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a828ff8378ce4649aad76eba9307b900 : SUCCESS in 9m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8a680288e74d41d28a6730a6bb6a0acf : TIMED_OUT in 2h 04m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3ce1e6bf0a5542c888450b63c0953e49 : SUCCESS in 1h 54m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/47ce2a27db07489793a5632592c2adde : SUCCESS in 1h 47m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c91cf66d8499409f9d4626b08293b665 : SUCCESS in 1h 43m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/50606750089143e2afa55eb16a330628 : SUCCESS in 1h 44m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5eaf1d7ebff435b872cadb9483b5734 : SUCCESS in 56m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2beb64048bfa44cf8f31fc5085afcffb : SUCCESS in 1h 55m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bb65b47c75d24aa38c9ed73b7ffc0c37 : SUCCESS in 40m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3449c566b4152b92826208a486ad67c5ecd7f506","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-05 09:40:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6d08416926e5c0d813f4946fb20c6b80267a107","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 09:47:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9c78859a7574d04a8803d43d129e317\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b29bf8ff83e14f81a02ff7f9c537f0f2 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08fc89033e0140309a1d905c0d4211f4 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e61be00f5d8d4748af12e536850d6503 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc3b0d81866c6380ab775b0161404883ad5ea768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 11:45:31.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90595b0f144b4dc68a0035ed4333028d\n\n- grenade https://zuul.opendev.org/t/openstack/build/8406db41d48a413fb9ae6223dce72d43 : SUCCESS in 1h 27m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7830d2fddb20447a84f4daf177e63bae : SUCCESS in 1h 40m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b557669c1f34ea0a472b7f049c23671 : SUCCESS in 1h 02m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c5487bf3ba4be4b3ff41b8d1d1a996 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5633f773555b4ee49af44ac1f9760f6b : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17aac41467874215941c3b6281740ba6 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6432e06e780c4998ab1c3623fb83d280 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58713ef7db1c47beb396b4ed5ad3f73a : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/643e80db34534fedb259b10ff709e4e4 : SUCCESS in 12m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2520b742ae84ce3a2b0ff5cec6970a3 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e54ed0aaf6de449c9f0c4f3bac6a3eec : SUCCESS in 10m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4379e62b34644722864f4d3cebee1dfa : SUCCESS in 9m 38s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/de03df89381d4812aa392d54895b7952 : SUCCESS in 2h 03m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/79c77e7b3aab47648263a916f85d1fdc : SUCCESS in 1h 41m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41f8bd822b984225978547da30b73502 : SUCCESS in 1h 34m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c877c76eb2c546528c8e8e82d228481c : SUCCESS in 1h 54m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/07942fc705b64aa18dc62ebe542f5c56 : SUCCESS in 1h 40m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24c7826df17a44269aa32361af405e37 : SUCCESS in 42m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad135b3299fb46c4a7a2a37baac33030 : FAILURE in 1h 55m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0021f2fd89a3447a822bcd7321523064 : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":1},{"id":"21072a4aa9aa6ec51e26c1afe51ac9ba533df32e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-05 14:16:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"36ffedceec475b7d4f4bc9ceeb090399ba909556","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 14:24:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2d3477f0c5d4c56b5d74afdb5a1ba52\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/242e6ac6b66b448f96b913703f8464a0 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/776e0a07a3dd4c7b97784741ff5553b3 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74eba96a3e2947e8b6823c821b552b78 : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3a9d39e0308d5846d772cd421547fb8de193acc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-05 16:23:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d1c1ea06c2b4a9782326f4189b30238\n\n- grenade https://zuul.opendev.org/t/openstack/build/b721da2b116c49a3b0ae5d55b768386d : SUCCESS in 1h 02m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6176b9f3cbbf4651864d4845a9e8508e : FAILURE in 1h 21m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd45e98a52c9428bb57f3e38f7f1094d : SUCCESS in 1h 03m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85f236b140c44fe5bbc6d54b36a397d5 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20e62359dbf844a19a7455f9c88dd1ec : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c811cded58624ff2b85a70b89bbfcd3a : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/970a1ba5c9c94461be0552477560f8f3 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b012b9dc8674114ac736738ad43d55f : SUCCESS in 5m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/26a7fe363f1f4406af325b6ecc8926f9 : SUCCESS in 11m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/346bcf0fe5e84297a022a3eea231297d : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b3f0f5930e414402b4aa3b4a740fb151 : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d698f042673d4cc99f93e824d9e72191 : SUCCESS in 8m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a7b059b0d23a42319e9bbf3caba3711e : SUCCESS in 1h 58m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2814c69ef38d4391a46ba104212c6bc9 : SUCCESS in 1h 53m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe87e6abdfc047e690f48f785684ec9f : SUCCESS in 1h 56m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b11fcaf0ab7848c0b476978810ea2d84 : SUCCESS in 1h 44m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/407db0c5d8bb43548edfea993762430b : FAILURE in 1h 52m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98701c367c1841689a1ee58c4caed19f : SUCCESS in 1h 08m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/86ec07ed1a694dac8a680734de4e40c3 : SUCCESS in 2h 00m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fd01ed930aff457e9f53e93de10d9258 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1b0d4fbac98044983dcfe790bc1468f8b8e9fa9b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-06 06:02:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09fa262d658defa2e247e82d88111ff3db8a00bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 06:08:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22c77518d8d349499759ff3da824a76b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0d0a98b249a4ba2a0590cff319f0667 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1facf959643f4db289dcaeb01f54a156 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10c081da37e54bd181b58191bead0ab6 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ae4d257e6dbaad8aedeb9ff4e2ff9b655c00d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 08:05:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2a1b8447ff456ab594f811ae0ff60a\n\n- grenade https://zuul.opendev.org/t/openstack/build/194403b67cc04ee39d48b51fa0a933f9 : SUCCESS in 58m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb8249cf4c0040b2be8c6aefdeb79b24 : SUCCESS in 1h 39m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af76794f7804497e9a743d17589ea633 : SUCCESS in 58m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47906045bb2347af9a8735184dee59f1 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c4af650da0d411ba944b128822eb076 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2da3550e3a2f4ee1a4f2bd11dfdbccea : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4874caf6fb6432cb67ee971fa0c21e3 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78dbb4e53d594542b21ade3822f68232 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/519874373ec24ea8b124873bcd8dc73d : SUCCESS in 15m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da59c74c82fb40cab1fee05da3ec5689 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1696c4f8bc89419383ff942471ba5d61 : SUCCESS in 8m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fbdc1a2593ec4e7190aa04237daacdcd : SUCCESS in 8m 14s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a14c96b4f49e47cbae9070b85756677d : SUCCESS in 1h 45m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/362508e403c04bb3a15f5328a74da1c6 : SUCCESS in 1h 39m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f7f1f75c7ea34204b3f56c354ad49661 : SUCCESS in 1h 41m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/85c6554d384340a5aabab66d4a2b7970 : SUCCESS in 1h 55m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/78a6adfcbda74b869c35378f9ba9680a : SUCCESS in 1h 25m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa4d442df7044a0196c38edd7779af0b : SUCCESS in 43m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/50e73d4db7914ee8b5bcf5316f141681 : SUCCESS in 1h 56m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/041ce710c9bd461ea0a440d97205dc61 : SUCCESS in 43m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0893b76846d7f0d60858eb09e44d56e520bfcc81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 08:06:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cac61855f45a408ba590ee383463ea3cc385ad9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 10:50:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d3e5dad7d15446bbf72bf89cfbd0b2e\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e6a547895bd404e9acf4aecb98e26a5 : SUCCESS in 1h 09m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8e2ad9114e7b4e7dbf6c36b0c4547450 : FAILURE in 1h 39m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/443c098184bb4f4bb37b34240f41bc86 : SUCCESS in 1h 05m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7314311d3b94f628d763c4ddeaaecc7 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c8bffdc453e451ab942bb5394a8825d : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75ca2296fb9c411d99345ccbd9e7d78a : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54436052249e4eaa9c9052754621b9a1 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1b7ba720b284330b9e95065971f3285 : SUCCESS in 7m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e30ccf8016a54ea493fb92559f92b15f : SUCCESS in 9m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c6b3e034bda4feca2da06a51ff8f4f8 : SUCCESS in 8m 28s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2b2819b26ea14df7a6f6cd9741c7cea2 : FAILURE in 1h 13m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/258edc041ffb4c26ac05643d9998980c : SUCCESS in 1h 53m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6dbcdcf7d1a14936944433c4215c25e3 : FAILURE in 1h 08m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/746e47bd4c7845a38309db7b0abd0d6c : SUCCESS in 1h 09m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4745ab1134c4475dafb72c397a215795 : TIMED_OUT in 2h 34m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db356f6e2d794809beb2b107275b2fac : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":1},{"id":"f4b140c7ad4fb74653751b6f5b457ae4412ad621","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-06 11:59:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2917e72cdcbe6a63a944ade5ff3f1782741703b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 12:05:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae85949156bc4743a4d9f252ebbca6f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8821cc46b784d77abffe2b90594f543 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae8dadc9ca8045dfae6fc14ec5e2bc5f : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5eb6e81ee7145de8eb1fb0b0e4d2e5b : SUCCESS in 5m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db097fd6d5fa9cfac46a9661500c8d092cedfdc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 14:06:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e172a3a2f66b4e92adfe008c3e4e634a\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f24699ee03042aba1f0bb951f5950de : SUCCESS in 1h 07m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00a9170dbf624884bd987bf9c9b92e01 : SUCCESS in 1h 55m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44bc74d0221b45f99ba576d48b459e81 : SUCCESS in 58m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d795feaaa9874434ba2c905eb0a95963 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/942e594b59e046e6b854f1b0dde7452d : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8854d6e7c92645ceb3603365d2141bff : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d002215d4539496fa1eb1e52b06cf0e8 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6545e48ac9f44c6e816127f41c8c370d : SUCCESS in 6m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cd26678311ba43a9b66353f289eadcd2 : SUCCESS in 11m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a986a4b06c148bfbbc540a65adfc876 : SUCCESS in 9m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/669c9c7d864f474a84537841b3395c9d : SUCCESS in 9m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7f9dcf2b88d646f8bd3be0b4fa64401a : SUCCESS in 8m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4e30145fbba74a8a8ffcdc89370a7b6b : SUCCESS in 1h 38m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e349c75da8f54e5a984a6d96e0f1eade : SUCCESS in 1h 34m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/958c2b897be2417ca13867bcbf52a797 : SUCCESS in 1h 59m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/95886145caa2495a9a1f54f03cafe10d : SUCCESS in 1h 36m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1db9b09589e84acda8fe6850470a83cc : SUCCESS in 1h 42m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e82cce1da7ee4673a6db9f3a7320e667 : SUCCESS in 55m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c28ff0eb2c74022b3dd7da6ce4f54df : SUCCESS in 1h 23m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1eeb7a65f2eb43c7a8917568456469c7 : SUCCESS in 38m 13s","accounts_in_message":[],"_revision_number":1},{"id":"6cbf5c560df691e669aefc10152cced59b715753","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 14:06:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7372daa450173b59e25eca0f7dcacda18492e9f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 16:23:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1adae6615bc41d78483e2736fbcada5\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8e61470cc87486883b4cb9ed9d9a5db : SUCCESS in 1h 10m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a44f2710cfda430e9191b0543ce60cd5 : SUCCESS in 1h 49m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10f849e30cad41a49c853fed056646fd : SUCCESS in 1h 15m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f50ee1db9f44a8fb4d255bbf2c043e8 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06d3480167a34fe8b8b52aa9dc358671 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4754fd9a8a8d4931964134ec3338f5fe : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b59109a942d049a2b551173582bbd65e : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aebcdfc68184c34aef80cac0ab6c864 : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00eb06e88197479daa19d633f9a194b5 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/434df28faac041c681ba9d8057e0525c : SUCCESS in 9m 51s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba237ce20d9141d6b53f46670890fd4b : SUCCESS in 1h 37m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a4f08a5ea3d44d48572a3f061602a27 : FAILURE in 1h 47m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c35eca59a36b4c1eb1492b001f65d91a : FAILURE in 1h 33m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da96a824c09445329fb66725ffd78ad6 : SUCCESS in 44m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f084dc6a374149749e57fbdc6814f5ae : SUCCESS in 2h 06m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7edb57b97891417882bea1fcf860aaf8 : SUCCESS in 37m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b0e701b7b543bc1dd3fc1bc2f06b7075efec3028","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-06 16:28:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"596e879477b11a6de168d91f1c5de484bce93199","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 16:34:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7794a7b20c14ad7b8e2e7e3e9ae4bf3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92df95503baa42208e1604a05ea7a01f : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/250a851e14124f929fff6230eda935bb : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d5077e326ea440e1bee85c6acf5cdb76 : SUCCESS in 4m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4cd3beb3139687953481d4b9e82f75f311ab778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 19:08:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61d37faf473648ba9eb3b7e82de94ff1\n\n- grenade https://zuul.opendev.org/t/openstack/build/488f7bb559be4fc9a749aa1b99a65062 : SUCCESS in 1h 07m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef7c9846e458495dbd56178d8a0f58c3 : TIMED_OUT in 2h 04m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af13588a970c4e9a91029b639b1783fa : SUCCESS in 1h 07m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec031e5ef09444e7b262b25f727e3c15 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f266e32dc554f02a26b101057b8119c : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cb23f995902465f988856f16865ef6b : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/108a29ccbb244e6c9b5b61e6cd3f58e3 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033dcac575d444d29a97dba1d5d6f41a : SUCCESS in 9m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4176a77c7f2048bca9a091f9af289579 : SUCCESS in 14m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb69aaa2f74d4a39a9b2a48b7f2fdbfb : SUCCESS in 10m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/72d99111f72d454d971845c12b715493 : SUCCESS in 9m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cc049569634e48eaabf1b079b452f0e3 : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2f25c9930a594bd1a7b0e47804c607a9 : TIMED_OUT in 2h 04m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d676316fb4634a04bfed92ae7d925cf3 : SUCCESS in 1h 27m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2dd18505a7124e25a9efffe2bb12676c : SUCCESS in 1h 43m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8e2ccad2dbee4c69a01a3f7351307b0a : SUCCESS in 1h 33m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c93ec0f87c8a4ce090dc7a51ea73fe3b : FAILURE in 1h 28m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bfb431468c64f189bf0371652d8f32a : SUCCESS in 51m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2d68eb55edd843d7a01cf149977fe207 : SUCCESS in 2h 26m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/180faf84e1d64806b54858dc0ce44d2d : SUCCESS in 43m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fd6bfe8c91598831d67c75b858b8ce9f321b9c02","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-07 05:25:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81c90f61b7d7b308257788d28662da01980fc9e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 05:33:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da4cee22da604b89a879741a38ff11cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a4d2d94fc4043cfb8c59310316a7fb5 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/59b7f1dee2ce458eb498865fcaa60b78 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea75048853d945878292a6531d716c4d : SUCCESS in 4m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d36d17b5a12add4ac2ad783e61dc95fdccd0a9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 07:05:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c696463938894e39bf73442e3b0d7a56\n\n- grenade https://zuul.opendev.org/t/openstack/build/44b0fa6985544bd1a955f32d17db1aca : SUCCESS in 54m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7668094a51ab492e8163f071cc9dcc51 : FAILURE in 1h 23m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9762f4429224f0e805723192cc6f0b2 : SUCCESS in 59m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5144587251a34f60a3621ca791865816 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b45ca7c352b84b629569cc1ce5ea042c : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae4da2289df04e7ab6dfad08a1297f2d : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27baf702e1be4c599c4aba2691caa96b : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a30d5915f3c4f19b42fbeccbff5d21e : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ed61b04e6c9d413c83bc989227ea32e8 : SUCCESS in 11m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e62703f151d140c7a252efffc66a2299 : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5754ec3ec67b4dc3acc54714964c70a7 : SUCCESS in 9m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6d46e62436dc4168b68815afed97e83a : SUCCESS in 8m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/faf1e7b0d93649da85e73af923cce02e : SUCCESS in 1h 24m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1f01ccd32f0b407fb25be36a554440bf : SUCCESS in 1h 14m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c381ec0bc424eccb0d35b9f95c5cff3 : SUCCESS in 1h 40m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1e5923feeb414ee5a288c5a5607d00eb : SUCCESS in 1h 31m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53578de4a1c84a4ba6c07c88f429049f : FAILURE in 1h 07m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3aac7a9689f4ac68fe819e4ec07de35 : SUCCESS in 47m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ffd80c0e77c4da3b8b256ea349d551f : SUCCESS in 1h 37m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d98a8a22c4d94a76b5c52af25c468f57 : SUCCESS in 23m 39s","accounts_in_message":[],"_revision_number":1},{"id":"1d516f1f5fe7ced17e68bb0eb8823b4fff54075d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-07 07:20:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"25947ae3d2bd225d0a21eebf76e08e7a9de94c92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 07:27:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb76671f2a5548c4917248df0374b075\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/603eacf9558a4cfba9a032a3d65dfcff : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a16d2e6f8ba646fd90b87e285a61d5ea : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/996a7b6c5a9241298c006b21ad3f2698 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7aca19002fd6dac7ed339ae485719d429becc77f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 09:25:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29ce9445ecbf4ed8a38334843cb61b9a\n\n- grenade https://zuul.opendev.org/t/openstack/build/dadb5110c9f9452685572c646a937be8 : SUCCESS in 1h 05m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/837eacb5956b49f8b9aa8ce2928d87d3 : SUCCESS in 1h 39m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1309b6c80946452aa2493352d9326400 : SUCCESS in 54m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ad3751bb30248df962eefa292abfa95 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68155bb976244b9c94168d773f7c0978 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ed4a4869a3b40938fce34a8b37f246b : SUCCESS in 6m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed1194e418194bdc9531113cdc25768a : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf517aeec5ea4107b38d26fc733e6f68 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6e7e54ffcc484d58a3b66131d50f0edd : SUCCESS in 12m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39d3719042034aea83ca5257dc2fbb03 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b52c9094acb448fb214dbecc7c55eba : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/293766e271b7490eb978838349f81ea1 : SUCCESS in 9m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0a2a0afb576d445284c3cad9729e8320 : TIMED_OUT in 2h 03m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c819d5712694557b99d274284ee724c : SUCCESS in 1h 45m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ad8c175fbd8e4980801911da9153e5a3 : SUCCESS in 1h 55m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/639d2ded4d874f64ae7b157dbdf16b1e : SUCCESS in 1h 39m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/738bde500f0b4a299529333996ab0ce2 : SUCCESS in 1h 47m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/133f8ee88324449cab17fa27a87f0822 : SUCCESS in 1h 06m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fea18c3974ab460b82876bee3e961f71 : SUCCESS in 1h 43m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a1d91d6c83f44588851a1fa867eb6180 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d2ec54c5f8376f25cc879bd6e49da52fab055b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 09:26:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5da054095024f2d34418a3e92c987fa73090937c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 11:56:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd73642fe6194aa680a4e8a889d88d4a\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce615453f7da491e985a00c462ba6820 : SUCCESS in 1h 12m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c5cd700475d34284ad9deeef6e70b99f : SUCCESS in 1h 50m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/308c02ed3c2440828509ff8341133d96 : SUCCESS in 1h 00m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ecd2bc1a3b34321a0817d2a16cd2e15 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6ce1c7e93354847a69454eeaf8974b0 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c1b8a590afb45ab8c76619661a1c334 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/051431ee9f594cd8b6a86d98c438eb6d : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84ff9411dc5b48bbb9e84a6f9daa89f6 : SUCCESS in 8m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27df618dc5964008bb51604360379263 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/449246791a474c77aab66a489aa9f748 : SUCCESS in 8m 50s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a138bcea58c84de5b8dc41f5eab1a470 : SUCCESS in 1h 48m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9f2c774ac2ba40e18ab549481756a616 : SUCCESS in 1h 48m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9bb41ca8f86d4e1c9d35f97a8b545c5c : FAILURE in 1h 51m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f8757f3d1174f5db6aef84dbe8c8aa3 : SUCCESS in 1h 15m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/271cbb13a75946f28f94534a6226a175 : SUCCESS in 1h 47m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/59b2ccd0d6f94f2288ecd3c30259fb7c : SUCCESS in 41m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d333854d4daf4067e6aa4a5d0942dd5c50717f4d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-07 12:30:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34e1c4d6c9dcf1f8a807c1b38b3771fd8e0172a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 12:42:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58748286123e40f582ea33df663e29c2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8963fa22ec34406ba6f29d9011211cec : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0f4d5fbb2794afe9df628807174bc51 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f664ef98757d43578c8a3c985806cebc : SUCCESS in 7m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e12e1b5305699fc9da8b8a952113e8fd0df3d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 14:39:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/63e8e20bef184b18b7c722f8b4ae6077\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbcc3c92b8994d42bd92277ca1425b98 : SUCCESS in 56m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/709517d628db4d659d789391c38358f7 : SUCCESS in 1h 52m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02a222181f26401d9737654c04362164 : SUCCESS in 1h 09m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/487073fd320249e8979673fdeeeffbf0 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bc24bb6c7334ee69e250453b0f7816a : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0dea169aa70e4641bf0fdc6bffa367b6 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95574f3e8be1409090c314b3580a0b82 : SUCCESS in 9m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/853a5a2a895d4053ba4df742e982ed59 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/52ecd1732b79471882d2c6e6d113dd4d : SUCCESS in 15m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b909c751cd5742128492c218d439116e : SUCCESS in 9m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6a45430d67fc4284954c41dc537923d0 : SUCCESS in 7m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8b2923228498482e9c21ae410ab079b0 : SUCCESS in 8m 27s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4abc3228cfd846dea70eb92584c875a8 : TIMED_OUT in 2h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fce6bbb93da34a44b0ebbdba4d4a6b06 : SUCCESS in 1h 31m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/416fae95c3f44f0e99a9d5bf72b14bda : FAILURE in 1h 21m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3604407da4f140208a1eb77b80b1014b : SUCCESS in 1h 42m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/23c520cc7789467cbf9f105ffe9ab624 : FAILURE in 1h 44m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7a91da1924d4bfcb697578c4e2de14b : SUCCESS in 1h 00m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ba753c5c5e34330bcad4c8109f7126c : SUCCESS in 1h 40m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/50c129f3b8dd4de58f046345de1bf48c : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a2824f70331a4f1ba216db604fff13e2664385a1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-07 15:04:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e11cd12c36b76312bb57db4a00b177f20400164e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 15:13:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/224e80ca46cd407bb57b1df3c9654460\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cab5174dcfa4ef7a72082fe65cd0409 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1109712e231347ac855d91d4f4e78f25 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2472d2bf47eb457fa809132bb449b8e1 : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"68011236e238a786dbdff156c115992e1e34a787","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 17:18:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e4d3d92e724548fa9fabce2958fa509f\n\n- grenade https://zuul.opendev.org/t/openstack/build/9aaaf6294b8d4aa9855bd44ac51dcede : SUCCESS in 1h 10m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35355c24f10c426cb26341812fb2f852 : SUCCESS in 1h 49m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/830865fc30144a43927ff052b27b1068 : SUCCESS in 44m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c4f73ce67eb41a586c06d0e694ae978 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/732ac68e2d04430dadfcb95aa4e16c7c : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/407201cc3d2d45459b0e0e27d69eeb77 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/53fe965609a149c3868d417a1593be58 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d46fa0300a245608e3deb01a4092f8d : SUCCESS in 7m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3aadec01b7394c6882c80f098be35686 : SUCCESS in 11m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/28014711033f4b51889385f2006eff19 : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5f478aa8b8334fd9af0e833bc7cbe1c8 : SUCCESS in 10m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/82d04d631efa4240933bff0e812973f3 : SUCCESS in 8m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c0cb24da17a6460198351f2bfb49a7ce : TIMED_OUT in 2h 04m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a05cb3d0c487466dae78dd056c010d23 : SUCCESS in 1h 53m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c92fa0cab5324f90a7979b67b337ccbb : SUCCESS in 1h 51m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1bdc2f66a1e84efb93c9942401e8eee4 : SUCCESS in 1h 55m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2c342b79d6ce44fca671e14628d46f8b : FAILURE in 1h 40m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/200f6d5b89824be6bee0f24e15d38949 : FAILURE in 1h 20m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f72aa1f0f534724bb0f587964f65eb5 : SUCCESS in 2h 05m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a06000457c74e91805f493c9255bd1d : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e9e199aaecd5389f32bb38bd31ad6e165ec9e078","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-07 17:27:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e4df62fb385ffdfaba92a51be58310e6c5fc76e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 17:34:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03cb05cca1964f44854d319a36eca884\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c4f6af0ccfa47e6bb565abdc34a91c9 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb9de930970d4000b10c7a48e6040c91 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c9ac5c17c794552a881f53c83354a62 : SUCCESS in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c18ec997a838a4ae4a9e2e518c327afd4923180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 19:36:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8feed75eb02441cfa7930640753f53c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/702f615393f946709ac0c531c458d41d : SUCCESS in 43m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83a119029bc04917bcb83b879fb587c1 : SUCCESS in 1h 13m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdc3eeb4d42e40a8ba463be8bf395e2f : SUCCESS in 54m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d5b698a1364bb0b6789642618fe8f0 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37208e997c294d5699afd1a5cdd80aea : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6483b70af49441488f9125f53212083 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5d8683c891a45e99c3f82f83972d633 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef627fccb58e4c0da33f771da524ec0c : SUCCESS in 5m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/809008c2ce8a4b4891e5e747220e4d00 : SUCCESS in 11m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c9148ce209e471aaba2ace64332c1ab : SUCCESS in 7m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/659812307f554c549556b65b2b41614e : SUCCESS in 7m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8375298ad0054419b0ef1dcab4747979 : SUCCESS in 7m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/78b8fa6a95574f7689d97246753dbe49 : SUCCESS in 1h 46m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/18b3389f0af4458ab9bf5924003ccf46 : SUCCESS in 1h 59m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/62319574a7354d3c8e4a94a0cae24c08 : SUCCESS in 2h 02m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e2765cf76d73489c8611a44211d30937 : SUCCESS in 1h 29m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1a249e20092b4ab6bd17338ab94d45a9 : SUCCESS in 1h 52m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/970e486e4be54becae0feade4bdb9d1a : SUCCESS in 1h 14m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8af0dc2eb2b84f63b0530d5c1d71df9f : SUCCESS in 1h 43m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/43802a84c7fb4334866dc3607cfe03be : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":1},{"id":"24334094634d8a88616254b678099de5c6e0d7d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 19:37:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e5d211900bc371da28d0f399d17b5abc809d1e98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-07 21:20:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6fbe4991952d46edaff81ba8f736a060\n\n- grenade https://zuul.opendev.org/t/openstack/build/cc577ced176c462ea174ff418fb8ed20 : SUCCESS in 1h 11m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2849d48ac2a54822a21339ef019ddf92 : SUCCESS in 1h 09m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8f3348c2a734348b5ea61d49eae3673 : SUCCESS in 44m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/366e2f2a417940c5bb3088f70f8dd5cd : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3b1cbe95beb476494f7b8eb36505c18 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6be7cede4fea447290552801f63b774e : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/015c7423722d4d49abb4d0e09d77d54e : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bd99a6e94547999a49d3c4ce73896e : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c3298774b494402fa7d195740523b63d : SUCCESS in 9m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f61e0ef1137849398ab479a0829b1039 : SUCCESS in 7m 31s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2a36d1c77fdc4660b2e7099d227c056a : SUCCESS in 1h 25m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c4bd3960f4a34e88826fb4dc03d79909 : FAILURE in 1h 11m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/138a1d1668a64592b640fcdd245e20f0 : FAILURE in 1h 14m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5a20edd205b4c8db0b1ae2b39733618 : SUCCESS in 50m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/67e5ce4045394b50b3ec6c556494a004 : SUCCESS in 1h 42m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0738d5ebacfe489a8fa48605d56419e8 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"17a069c123eca955d655a3c36510be9a3d2a59ad","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-07-08 07:12:03.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"26c3117cb0f5ff5ded3c2d860c27eb0cfaf93f74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-08 07:17:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b85d7d0121f44f30ae2203c9b44e320e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8931f713c36341c6941e9ea71da83fde : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/783408f404e94ef1be4806bb3084fec9 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78cbcc2fb1b0444190e8f3a818407ace : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"06474c726c14468c634fda09e67a696069027cda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-08 09:48:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7631c03a2ff43a0ae54e509eeb4ecb1\n\n- grenade https://zuul.opendev.org/t/openstack/build/d3b34dc9b46b44b5bf3e1b1850388c37 : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cb0c61061fc47d081b776d53466c310 : SUCCESS in 1h 36m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d29416bf77764ccf9dba128ae2584f3c : SUCCESS in 1h 04m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bfd46d4d0df450b88855d06a8d55ab2 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9d8eaf6cc9949e5ab2adf6ab0d96165 : SUCCESS in 8m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55adf67e59a3417c91a501b5134fc7ab : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65bb6574cd7d4b3c86aa98bf8afe31df : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1776a80f53b44bbaa46550dea161c36 : SUCCESS in 7m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b21858e8e6284c948d266d261196d40b : SUCCESS in 12m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2e932d3c2fb4fde9cce4f198d4e7015 : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b60b183026344b08a82c310a3f9000c : SUCCESS in 7m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0b5d2684c80447fa8a20016f646b86ea : SUCCESS in 7m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/95d9d92682534322952a923e6742b112 : FAILURE in 1h 31m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d709c7c32b24864a653f1af2cb4d361 : FAILURE in 1h 07m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bff24f7390ba4b7a9729532ff0196585 : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/72521bdc574b47218ca6b26d5859ca10 : SUCCESS in 1h 09m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/60eb964c13c64568ab48c6742a96406c : FAILURE in 1h 32m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7acca3a43f944c79e06820905b1d190 : FAILURE in 2h 27m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf431db978da4b6694501a8eaacc1db2 : FAILURE in 1h 34m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f33236db7144c71a016fc4044c482ec : SUCCESS in 25m 49s","accounts_in_message":[],"_revision_number":1},{"id":"f992e41529a6e970c447648f75fadca9d4ca8068","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-07-09 08:04:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c361024a45208615ce232c7504aa5ce30291c81b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 08:14:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f93dadefe844a318415c910c531b761\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ced6430c82c14ff2881c72d1a538352b : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/616a4123f9d64443aad9aa9e1471e021 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/92a1506e2add43c7b21c54c356e65a1d : SUCCESS in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"49cc96d64507deb28339812adf96245241d96294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 09:50:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5fc729f964af46018ac4643a2d92a428\n\n- grenade https://zuul.opendev.org/t/openstack/build/225ffd80c9e6433f9d4f4ef297ba70e0 : SUCCESS in 56m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b44ee4418bc44f0a28a0da2dd202d08 : SUCCESS in 1h 38m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4fe5ed0251244778de23a676a411c76 : SUCCESS in 1h 03m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c328b2f9f4b430b8b611a1e6561b67a : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b7ba840ffb343a3ab080a44e641aa66 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54b6b6770f0b4daba81ef8d23455728f : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d26afdb0775a477d9b6e9ddf0093f0b5 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6de49a7cfca436099eca179b0da5909 : SUCCESS in 5m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/50e5189db8e0426e8d7054b6c27e2994 : SUCCESS in 11m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f94f748f4b144a6853aec927ccf3036 : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7c9751458af4bb2a198ed7c092d777b : SUCCESS in 8m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/95204490caed4f439b0e9203e924c95d : SUCCESS in 8m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/25d5cca2837a458b9ff50e83b02e83c7 : FAILURE in 1h 23m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6990a5d5722f45cfb67fa885ba8b32a6 : SUCCESS in 1h 04m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b489ed39b6f413a889349945d68654e : SUCCESS in 1h 09m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ee6313d12a5645ccaa96951f524f3b4b : SUCCESS in 1h 07m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1fc63577b231461d89a9a7cf8c75b3c0 : SUCCESS in 1h 15m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8eaae6efd03485aa42190f4652a7e62 : SUCCESS in 40m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0dcc4e033a0f4222b0d3e39a2d64a616 : SUCCESS in 1h 29m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6163f6424ee04928829be61ceb2343e5 : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":1},{"id":"48ec5ca0d8d2fead8c108b45934c1c5258767ac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 09:51:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d4cfd1029154f19f9977be5ba2221598e857645","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 11:10:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f8c1b2aec842e8b8cc1685bcd6c942\n\n- grenade https://zuul.opendev.org/t/openstack/build/0bd5183cf6ac4f25a643701a247d4751 : SUCCESS in 40m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/849a45e3a4f34ebb8c7edde586d8efef : SUCCESS in 1h 04m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fcdb9d6bdee4f28b6e1d967bd5fbd3f : SUCCESS in 39m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dd0108fdcb04c399a1c2bc7d22630ff : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ec397691b074255b56118e5a938ac5e : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f068711eeab495f96594c8c551faba1 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/514e879681e34811bd69e8c86f439791 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957a5a53de4141268ccd4cb62f596d8f : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d6cdcaa34fe4c3fa97db36484acbab8 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/435c6deea0c747f19fba75f60b4bc436 : SUCCESS in 8m 32s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f3604a3c00241d6b9b7ac244113dbe5 : SUCCESS in 1h 12m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/27590f620c1f46ea8db1edddfb1d1e7b : SUCCESS in 1h 09m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/db027189d4514f7ba79337772a93222f : SUCCESS in 1h 03m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f8d8f94ec31452788de75175430c8a8 : SUCCESS in 39m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/85d4c35ca3e04f6c898bc3666f2ad021 : SUCCESS in 1h 18m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8e3074c8effe4aebb464cb94d9eb14b3 : SUCCESS in 23m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fbed90a04b60eaa788d13c0020b624100850ef84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 11:10:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e9c20d39e55466087a0964372c4c328d27ddd5a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-09 11:11:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1c9523aebcf4636929991ec5eb5538a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e976ea107e4286a65238f029bccbf2 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d68f99a7384be677836a1309cb3b0f5d9f58277d","revisions":{"d68f99a7384be677836a1309cb3b0f5d9f58277d":{"kind":"REWORK","_number":1,"created":"2023-07-03 06:50:58.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/75/887475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/887475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/887475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/887475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/887475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/887475/1"}}},"commit":{"parents":[{"commit":"6d78bf16c5c122036e9d446af5d3b2d61272117f","subject":"Merge \"Bump eventlet to a version that fixes #632[1]\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d78bf16c5c122036e9d446af5d3b2d61272117f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-07-03 06:10:02.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-07-03 06:10:02.000000000","tz":0},"subject":"Refresh Glance example configs for bobcat milestone 2","message":"Refresh Glance example configs for bobcat milestone 2\n\nChange-Id: If22e04649fb9a5c237ca59e3844974756303d127\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d68f99a7384be677836a1309cb3b0f5d9f58277d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d68f99a7384be677836a1309cb3b0f5d9f58277d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
