)]}'
{"id":"openstack%2Fglance~888695","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I20e9ec459aa49937c74b1e40822464c9f93bed5d","project":"openstack/glance","branch":"stable/victoria","topic":"gate-fix-cap-virtualenv","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2023-07-18 15:57:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I20e9ec459aa49937c74b1e40822464c9f93bed5d","subject":"[stable-only] Cap virtualenv/setuptools","status":"MERGED","created":"2023-07-17 14:29:15.000000000","updated":"2023-07-18 15:58:44.000000000","submitted":"2023-07-18 15:57:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"888695-gate-fix-cap-virtualenv","meta_rev_id":"5b2e331d380400e6bb4adf622f92755964ef9870","_number":888695,"virtual_id_number":888695,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-07-18 15:57:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2023-07-18 13:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2023-07-18 13:43:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-17 14:42:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-18 13:43:53.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"43eda4e9a92f8d4fb56dc116fd7246162efd4232","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-07-17 14:29:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4859a392db2628109b42516c89ab7c66efe905a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-17 14:42:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54f197b85df44abab28c5e77de596c34\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c1143ab7b4422682a6f058a935b50c : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43b8b48f06af4d19b7db258344178d62 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/918242b2106d49aca224553c341f7db8 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ee8a9aff7b74656b7a2173332478ea3 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29751d5eebba4a02b23aa15d4c16e87b : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5bdd8e26d21401987b3000e9de4fc92 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f45eaf1aa47d4f1c8a013a3c7a28ca7b : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":1},{"id":"08e82c14c7b973186da3df38516751ecb1adbc75","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-18 13:43:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"700290c7b23def32090d14362d1c0aaf8af4713d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 13:44:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7b4394c5982882a6797942b89f1bb46ea57615f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 15:57:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a48d772dd0474bb1115857fb2dfb8c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36105d89edf64238ba3c5399ea5ad20b : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f0b041c4e4a4620a286cf11b1e8c516 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0066e8c7025042518eb45c464aa31dc7 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d119d98372a477eba47fa8a0d80a37e : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50d9ac17d350465dbbdd96026e54bc72 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5350d633304141e3952df2b94fd922a8 : SUCCESS in 8m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/50612e00a6574bf39a7c72bd2111a8ef : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"07e66c12c6bfc36cb1b736b9da302430cdab4de5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 15:57:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b2e331d380400e6bb4adf622f92755964ef9870","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-18 15:58:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32c7f076dc9c4a1780972f42bf324593\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d5220759e6c40e1ab88a710563a6f6a : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c443d59e08ed4ffca0141157dda58f90 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2bd78bd1c012ce8ade4706c3d08bfeb8d1460f2","revisions":{"a2bd78bd1c012ce8ade4706c3d08bfeb8d1460f2":{"kind":"REWORK","_number":1,"created":"2023-07-17 14:29:15.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/95/888695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/888695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/888695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/888695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/888695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/888695/1"}}},"commit":{"parents":[{"commit":"18f7557b33b8085f59a7ebd92cef537bd1f0bdd9","subject":"[CI] Add upper constraints to install command","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18f7557b33b8085f59a7ebd92cef537bd1f0bdd9"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 14:28:22.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2023-07-17 14:28:22.000000000","tz":120},"subject":"[stable-only] Cap virtualenv/setuptools","message":"[stable-only] Cap virtualenv/setuptools\n\nWith latest virtualenv release (20.24.0) and its bundled setuptools\nversion, stable/victoria gates started to fail [1], because setuptools\nis not compatible with the pbr version (5.5.0) which is in victoria\nbranch\u0027s upper constraint [2]. pbr\u0027s next release (5.5.1) removed the\nfailing code, this is why the change is only needed in stable/victoria.\n\nThis patch caps virtualenv to \u003c20.24.0 to fix stable/victoria\u0027s gate,\nand also caps tox for the \u0027inner\u0027 virtualenv, otherwise tox\u003e4 would be\ninstalled, that would introduce failures (those failures are fixed in\nnewer stable branches, but it would need several backports and \u0027outer\u0027\ntox is capped \u003c4 anyway, so no worth dealing with it here).\n\n[1] AttributeError: module \u0027setuptools.command.easy_install\u0027 has no attribute \u0027get_script_header\u0027\n[2] https://opendev.org/openstack/requirements/src/commit/636b6b3bde4b68ef27ac6b524cb888152eddf517/upper-constraints.txt#L44\n\nChange-Id: I20e9ec459aa49937c74b1e40822464c9f93bed5d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2bd78bd1c012ce8ade4706c3d08bfeb8d1460f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2bd78bd1c012ce8ade4706c3d08bfeb8d1460f2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
