)]}'
{"id":"openstack%2Fglance~889044","triplet_id":"openstack%2Fglance~master~I3e2227a01628f8f5d14d4b4ebf11a9374b52c228","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-09 07:55:57.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-07-25 13:48:30.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I3e2227a01628f8f5d14d4b4ebf11a9374b52c228","subject":"tests: Use WarningsFixture in all tests","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-08-09 07:57:04.000000000","submitted":"2023-08-09 07:55:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889044-sqlalchemy-20","meta_rev_id":"ce69e647d66f8167367597601c89a2a7ca197482","_number":889044,"virtual_id_number":889044,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-09 07:55:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2023-08-03 07:40:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2023-08-08 05:42:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2023-08-08 08:04:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-07-28 01:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-08-08 05:46:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-07-25 13:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2023-08-08 08:04:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:18:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 12:13:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 15:27:08.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-28 01:43:45.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-03 07:40:43.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-08-08 05:42:07.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2023-08-08 05:46:11.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2023-08-08 08:04:14.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"194dff9e390ea58cd779c6caef0e0d13e5cedf44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0dc1fdab9767ef6262408eac58d1296ba595686b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:18:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2904418178fa4b019fd4872725c32629\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/78d90e6ea1b141cbb47435854770c9ca : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cba42e1dbe55433a9d52316afe74f2e6 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41e385a7d553469c835df100237810d1 : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"da5a25ef0a33be9fe0a959682c3310d6cee26027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 12:13:19.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b523a972f456470cbb4029cb644c5e82\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/850e1816b4da4d31a228c59eb6a2f912 : SUCCESS in 53m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d4ecfb1c324f3abbce632d1a831c92 : FAILURE in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a6f27480e3c4e7ba0ea51b2d32f9ce7 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5baeaceb24d14715a118958536a1b29a : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62da52a5e3b84f1c99c3d49957c6d6fc : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1310de791c8349d69909250514bc8a8d : SUCCESS in 6m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3a9cfff725474882a2f07075338b9e79 : SUCCESS in 13m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e903fbd3cb004a919de2d8db61ca6ad6 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ff917fffab74ae287792986040097db : SUCCESS in 9m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7020e45e851a4015b1dc8d428955f649 : FAILURE in 8m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5702fa8d900e4c0289694ae193718bed : SUCCESS in 34m 36s","accounts_in_message":[],"_revision_number":1},{"id":"29fddcf8801b6ab2d4ab6aa69d4a27d160216cee","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 14:23:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4913cb645d03c7ea317f46dad62a656db8f34680","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 17:03:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f9eaec708609e16e4766ca3c370e03506a8aae86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:17:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0590ca9927346f9a793d80e00f297d9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4a9245717cc48098ed00b6efbd46c95 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2114b848cb3c4d588de330b9d2548a9b : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4925d1217754a7180b74ecfe30126c0 : SUCCESS in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f8e4eed3b7cd201c99c21fdff60f57f899411be1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 18:15:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b0fe23536294f828d66bb90027cfcf2\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/390e407b684247d1b3be0a7ea905d827 : SUCCESS in 1h 05m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ea70d4fc61c46c6bd65c806ccc05f12 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a40a1573237948d59c9d8fcad4f7847e : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98fde16f4e20419ba48922886fc84e9a : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abb37d5503be4ff5948a0fa1da8058d5 : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41188bd6279a4837a09a5fb1378ce08a : SUCCESS in 9m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/809de9f4ada14f7e8e6bdf38bffa27a6 : SUCCESS in 14m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/453b76d1cfa0444b81c23112d2d2490e : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae451dc344144677b3a1b19f6f52b898 : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f8b6207a4ed432ea6898edb9f4b0ee7 : SUCCESS in 9m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d79e20862aff440c882c333336cdbd29 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"243ba18180df921bac3742d975d4f7cad43c815b","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-21 15:27:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8c91e60cd81d198a1e4dc3300939a6cdd17e7acb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:16:49.000000000","message":"Patch Set 2:\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8e8b77e9231c79db84cc77e251dffa320f31503b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:55:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"f58461d86f6bf32d1d2e7207d09b273569d51206","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:11:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f428812641394fd8b992f787798a3771\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5be2c05c87649419068ad9859b53995 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3783e7a5eb574e508c2d10d9b46dfdee : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ee1a0e5b2e7547e2b3ec6e55bae196a3 : SUCCESS in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b65b71982304bf572154cd0a47971bfd568636d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:58:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cd59ae0310d4832921f12d5f035402f\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/256cb9f3d9524d6abf54b1dec367dce7 : SUCCESS in 1h 01m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/030c58b8fa5a47c18914d2f8c9982345 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22562d5b1594406a855d7b3a299985eb : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e41e57d86c147b6beb457661ade72c4 : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/869fab836e7b4e53bb3495653dfa9ddc : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6309ee6fec93436da8cf61bc443ec7aa : SUCCESS in 6m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c43b08d37f96494c809205cb12528a93 : SUCCESS in 15m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2ea664e5255e4f4d9b38138a4230a220 : FAILURE in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f734e7ea87bc47c397e91fea7eaa9bc1 : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cb0a5f2ffdb44072a56929ec8e5608b4 : SUCCESS in 11m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6dfb3a1495b54335bc526db390d8caa4 : SUCCESS in 37m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8f65e4464584f5847a86e8d8480f3d2865908bb2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:30:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b1ffeee83bbf67dbd004bb3e33bc3df017af0888","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 21:48:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b97e645913c4fa78f05e78842917438\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d73fb25b6f484490a666eb141a9c8b37 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e81b03e3b30466d8dd2af1f7e2a5768 : SUCCESS in 15m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6a305fd655264c168c5f0ff210bdbb7f : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b73f4879cea05ae65c885adba2e71b1ca139e8d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 22:31:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1cb875b9204381a2cef624e726246f\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44644f6caf5f4b9d82d5332ca47b784b : SUCCESS in 59m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c955c8035274269b1ed3364fe9d98a6 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94dda13b0a7840e5baa82434bc51dfc6 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62b76caa966045fea282dff80f5a5de9 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d832e8b572042a19fcb8bb6d563b460 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e822ed560da843e99c1d44ea2a618b1a : SUCCESS in 9m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f04e3f7a129044cb97b6f4fb64bcbcb3 : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac7e0a90c7d948e7b11dcc113f9f7c82 : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c882450279c14defa178b9f5481cfb40 : SUCCESS in 8m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c32dc5acb1aa4d569a451a7526ba408e : SUCCESS in 9m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a4fc2934e0174c9bb1b4549ad5b85288 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":3},{"id":"49e024e6af9ceb851da2f3bdefa416cf1be7b7f9","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-25 13:48:30.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86f27a7f261e5b0b551e123bb59fa428ee813f5c","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:43:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bdfd1957ae428a17196d7e082bbaa92099fc1468","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-03 07:40:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"95e9d96fe284172a80966caedd7eb065e61e66ef","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2023-08-08 05:42:07.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"16717f2c1cd3fc7466f61331109d3793a7a1d10f","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-08 05:46:11.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bbf58cfb74946c66681186388e81b8d6aa7f1b99","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:04:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e17d22097372b2bff2f3b1d76764d2644c59fa51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 08:35:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d952038cbec198f71b145b6249d5fc730f08439f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:19:00.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d30ac5e015364cfcb8e89d4f1d17acd1\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79f11b3d2dc54f9ea395d984a4a2dda8 : TIMED_OUT in 2h 35m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6862285d6c84ed4b81a0a0b383789c1 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/707c0facf45e4f7cb436c236a30e195b : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7189331b29ca4c9994f169aa1e7bb28d : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7af5beaf5d564792a5d54725c367003c : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ef5a55c47274a639c88f43e7d4e5291 : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6fc85b940a8a4e259b95c1523a6c8750 : SUCCESS in 10m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c74781a986244c195af585b3289db0d : SUCCESS in 9m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f6b90cf5af454b9382ecc1457d2866d9 : SUCCESS in 39m 09s","accounts_in_message":[],"_revision_number":3},{"id":"3ea063e949defad3fe79589430f2fa742b4f3276","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 05:55:24.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"df9e05f32a4c7336641794d190a3061dcfe5a1db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:10:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/996c2d1151d84fcd9696fae92bed1634\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7072612a78354caca6edc1d42a0b791b : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/188537f7d18443578d15fde66bfa18d1 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f69fdc98339c4c56b79180b271dd711a : SUCCESS in 13m 16s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa0190727f3fce2f6a338dbd4bdd6a2d19777099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:49:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/535f30dd530c4f64beb50f974c357be1\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ee30f0ac006494fbf0cdec826b7b61b : SUCCESS in 52m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48403a72f4854147b59436eb14f9666c : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a67854bbde9144ffb2d91b56eb96274b : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ecb7f9d308c4dc780b739301055e01b : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54986358d1154df6b02cd2912d934d59 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649b2aa74e5d4d0d834b331c449d8e04 : SUCCESS in 6m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fe9a8f30fccb4868b92aad785b507f77 : SUCCESS in 12m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb2cb0bcf478409b9340502c69ee82b9 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/10fe72058fae46988650bec173b88135 : SUCCESS in 8m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/85da758a3e554979ae48d34a80752d3e : SUCCESS in 8m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28a2e0de674d4690bd5624b7c4b878d4 : SUCCESS in 40m 35s","accounts_in_message":[],"_revision_number":3},{"id":"0e12833bab6d92b0d37f659453675c6e1d505332","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:50:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f51015e2084d0c590437e558c102162a822f7ab1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:55:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e79b28ed215245df89bd32944f4e2e95\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13e7acd16799475eba79b14df65a4d23 : SUCCESS in 1h 00m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d22baa571742f7b9d84929049e252e : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68d2bdacc2314e06b57a9b594efb563b : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c5a13d2463f41ed8c20030cfb9e38ae : SUCCESS in 7m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a04b7f611c6a4bb1b8abbfa3f0b8b754 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac80a54231a2427c9aa2fa179d9b3874 : SUCCESS in 6m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e8f70328b2ee40918ef27707e64a1106 : SUCCESS in 9m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b1beab7ac414e559013f5968d14f314 : SUCCESS in 8m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f72f8fbf134b47a896960ee50a0ccba3 : SUCCESS in 43m 21s","accounts_in_message":[],"_revision_number":3},{"id":"9aa5dce10ccb161951607467dd048e16c93f8902","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:55:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ce69e647d66f8167367597601c89a2a7ca197482","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:57:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbb5de126d6e4197a727e82b8d2252ba\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e00c7876474e1d921788d641d03b7d : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dc20d4a2a24df072d4a4bcb3277831dbd3aed650","revisions":{"af819043e23658add7f8be23b185a267a7ffd0d1":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/889044/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/889044/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/889044/1"}}},"commit":{"parents":[{"commit":"bf932b2efef21a8bdc4dabe4b93c787a815d44c1","subject":"tests: Restore - don\u0027t reset - warning filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf932b2efef21a8bdc4dabe4b93c787a815d44c1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:57:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"tests: Use WarningFixture in all tests","message":"tests: Use WarningFixture in all tests\n\nChange-Id: I3e2227a01628f8f5d14d4b4ebf11a9374b52c228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af819043e23658add7f8be23b185a267a7ffd0d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af819043e23658add7f8be23b185a267a7ffd0d1"}]},"branch":"refs/heads/master"},"de683a2a039d0578529dd33f080df9119b32fa16":{"kind":"REWORK","_number":2,"created":"2023-07-20 17:03:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/889044/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/889044/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/889044/2"}}},"commit":{"parents":[{"commit":"bf932b2efef21a8bdc4dabe4b93c787a815d44c1","subject":"tests: Restore - don\u0027t reset - warning filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf932b2efef21a8bdc4dabe4b93c787a815d44c1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:57:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 16:27:35.000000000","tz":60},"subject":"tests: Use WarningFixture in all tests","message":"tests: Use WarningFixture in all tests\n\nChange-Id: I3e2227a01628f8f5d14d4b4ebf11a9374b52c228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de683a2a039d0578529dd33f080df9119b32fa16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de683a2a039d0578529dd33f080df9119b32fa16"}]},"branch":"refs/heads/master"},"dc20d4a2a24df072d4a4bcb3277831dbd3aed650":{"kind":"REWORK","_number":3,"created":"2023-07-24 16:55:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/44/889044/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/44/889044/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/44/889044/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/44/889044/3"}}},"commit":{"parents":[{"commit":"cbabd5757da249c71d2f6b549bb31e4c0b9ff522","subject":"tests: Restore - don\u0027t reset - warning filters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbabd5757da249c71d2f6b549bb31e4c0b9ff522"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:57:26.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 16:14:15.000000000","tz":60},"subject":"tests: Use WarningsFixture in all tests","message":"tests: Use WarningsFixture in all tests\n\nChange-Id: I3e2227a01628f8f5d14d4b4ebf11a9374b52c228\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc20d4a2a24df072d4a4bcb3277831dbd3aed650"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc20d4a2a24df072d4a4bcb3277831dbd3aed650"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
