)]}'
{"id":"openstack%2Fglance~889045","triplet_id":"openstack%2Fglance~master~I3733f779075c189c9a630416181b7d4c14ba599b","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-09 08:03:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3733f779075c189c9a630416181b7d4c14ba599b","subject":"tox: Unset basepython","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-08-09 08:04:13.000000000","submitted":"2023-08-09 08:03:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889045-sqlalchemy-20","meta_rev_id":"256f28fb8eb0403ad2d0ab025048ee4515854777","_number":889045,"virtual_id_number":889045,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-09 08:03:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2023-08-03 07:38:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2023-08-02 09:41:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2023-08-08 08:09:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-07-28 01:44:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-07-24 16:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2023-08-08 08:09:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:23:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 12:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 15:31:05.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-28 01:44:26.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-02 09:41:59.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2023-08-03 07:38:53.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-08-08 08:09:09.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"f1999d5f078123205b6604c1464f0eb7c95d0d41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd95163c607876581b3f7277da44960a20d3764c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:23:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0424c9ea2c146e1b5cba91fa6c193d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e62d693d6924627a1a61c220a878753 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0fc00dbf30de4418bfe48091600d722d : SUCCESS in 10m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec23240d27ec4bf1bd1297a9ef290c65 : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40c577b620992e879990ef49e7735d6d71068f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 12:33:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a142345ad9342af9ff8fab5eda9f124\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/53b90eb724ed45389bfea9dd26e32850 : SUCCESS in 6m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/751eac236451458b9a1d56a197a58519 : SUCCESS in 1h 06m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a02c2604b0d944d3aee96cd8ecfa7790 : FAILURE in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/492ae8d90c4e48fa8d70066d1e9d061f : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e9990ab38b846ec80243bcdbb4ef915 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17fdb99a3d58492099b949ab05149438 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bef260bfa9974dc2bf031f8eddbe9387 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aee4dda271c472095855412e4d368fe : SUCCESS in 7m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cc361c5b886d46cea5f9f9e59ac51147 : SUCCESS in 12m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69288efaa6664289a2b37a038bfa1c8a : SUCCESS in 9m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36d1e84dd72e4ce9b5c14c20ebb58ab9 : SUCCESS in 9m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/889ac1fbfb5c43a787106438ca763930 : SUCCESS in 9m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e43dcb2cfe6240b1b79330649e53c9c8 : SUCCESS in 27m 19s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/api/test_common.py\n  Comments left for invalid file glance/tests/unit/fixtures.py","accounts_in_message":[],"_revision_number":1},{"id":"b8e1cefa2fe99d0b155694fbb7feff9d45513ff6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 17:03:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8129040d7928d949da0d4f2bb449ff810d5b552b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:18:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd190e2a0fd4dbe9bdab1167961d136\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60de45c6bb9d4c52b0b488919cf33186 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c251445ba9084ae19242471c39f2ebcc : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aba476161c6a451980273e1fbe52de5e : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2af40acecde12175c68690c602a78e03cf92ea6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 18:13:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74cee05fc84f437888b06d844669eb97\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/36d0f18419cf4a799de13aab9a635452 : SUCCESS in 5m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c63c21a75fa4d66b7c0ee251fde340d : SUCCESS in 1h 02m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30d297af55fb4033a217150fda758ee6 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee7224db15344a9ca9cbb7055cc73810 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4188ec0aae8646c8931027ce219334f2 : SUCCESS in 6m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78f352b0077c4ec2ada8688f174335bd : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed215bc7f8c7477988a2b765658062e8 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68fbe81ff1294247aec0c514e85471ca : SUCCESS in 6m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/551c59c832724398a17e56909f3a3a28 : SUCCESS in 14m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/64f725d393324d348d381d7051f6c439 : SUCCESS in 9m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bce572d74c944f659be2c41010ce5853 : SUCCESS in 8m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a324d5489aa94df490f8f0e131457fed : SUCCESS in 8m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5fd8a859fb1e4b37b1f53d458580ac0b : SUCCESS in 28m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a0b0589a2de2d3d8112cef8bbc50f4f4946a5367","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-21 15:31:05.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"daaecab45893fc567c009ad5f440e1cec8037108","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:55:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"61c50ffa2d043a59efa08f7dc7a96f2c9cdfcfb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:05:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e308ddc671634491871fcfef589e4c4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/518bff4be6c444b89d976dedc3c35baf : SUCCESS in 8m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/773c5974945346cda6e5e25b84e3dcde : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/57f0e1a7670b4b21bbcb9ac0947072d1 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5bffb511ecce224593f4f50b838f481dc55ec486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:54:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6f5c4c9d4614cfb8b5a396fe34ef419\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1749e5e3ca7b4e1da34d098afbeb6c46 : SUCCESS in 4m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/365eb4ecaa07435b9a4bb259df03275d : SUCCESS in 55m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64de03f4a254ac38b805bc94d3a952a : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ae4129c327a49e9ba840e70f010f4f2 : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46c61e88f4f94b4f93db3f3d2e223f2f : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f4451d50bfe48c6b28fd7eedb840e30 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab101b10af1242ffa807f61e5d6015b2 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbfcc481c9774610948fdbde442b4428 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a90252b9f4f843c08a8b6f6f03f9d4d6 : SUCCESS in 12m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc1ca306aabc4eccadc0bef0231ce2f7 : SUCCESS in 10m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d939e57fb09482b9e0016f92c21aa9e : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0bfa201a440f41919fde5a86d5c1ef63 : SUCCESS in 9m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2c00731a10ff46008c24bcd4aa892af9 : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":3},{"id":"df0aee81184381339b55de009783e32d3c22fa0e","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:44:26.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"28ba4b60961e51f5535921c0bb0a153dc3faeb31","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2023-08-02 09:41:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"88800c3b5122dd7d1da46a4a871295d7c2da00d4","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-03 07:38:53.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ddc602347aa47c26b4e46a8831b86fb73712fac9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:09:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"99943760836024019810228c3187cb49d9363b54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 08:35:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb1aeafc14c2b82971599ddc01f30e6a23b17fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:19:18.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/889044 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"835c576c19a5a3e30418fd17e44cebaae7082930","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 05:55:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d0b82e49b5d39bd8ec68b328400e384bec230d84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:11:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fa382a3f4a8411ea823ba7152d3729e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/769e573e05cf4f9aa60d93a9de397d66 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d48ba54ef85f438181b290f62f8b8816 : SUCCESS in 14m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/da9a6135246941babc85bd005ac5cb31 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ee35d813217b56de6816802b3c5ab0b88552345c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:56:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9909c390fc5f49329b4df3a7ee2e054f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dbf012c9859940c599e9da7b2f6fb841 : SUCCESS in 4m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8831e51a196c411bb3eb016db688a832 : SUCCESS in 59m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3ef8ad78664d32ba9b098acf7cfd57 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e752f4ffc4f042bc932e4a0be053d17f : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03c1380349fc4b6789dfc3514f5c5483 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39844120279d4377bfa79bfd6cccb316 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7758ef5179994c5b872c9a69fe8bee26 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17deda3f6ca949f99090c12d030d7bc8 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/91cbfa2a62ae449b86ba183549cf05f5 : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f81a64c65f7040f0953e1ecf1de68884 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ed91434274a4378964f4f679c8ec67e : SUCCESS in 9m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8feacae7f0e248e59d9770a7bb00afa7 : SUCCESS in 8m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b98e1af8ca64fa886b4e44916232c68 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":3},{"id":"feb218661b862c9b48f88bf8b1be7033e31dc15a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:56:44.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6575fbd27019da69a7c4baf404cc1bceb03a14e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:03:03.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4267fa4b0d344b9ea2e65012ffa7149f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d30d6abc03c54905ad750f8ab7e94ea7 : SUCCESS in 3m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c67a9e04043c4eb597e49b04118b2ac2 : SUCCESS in 1h 04m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd8e2cdbe794943a0a803b46cce66a3 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c125c2c6f6f4953b65b5218f2508654 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/916e2407f6ee4e57bfdd5ceb97f75939 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1cc318f88aec4de3a46ddfd7528c6a5d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4761c23f224fd385762d157a5ca42b : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e1ac3690aa34947bd70c058e66f464d : SUCCESS in 5m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d983b56382ea499a91eaa66cf6a982f0 : SUCCESS in 12m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/144ac22840c2437b8797e52aa6b58ed3 : SUCCESS in 8m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/df7401b8d5c44424a0bd49fe4c1ea16d : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":3},{"id":"f6fe2ae9746210c16e5621b7e6835493f1c6acd9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:03:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"256f28fb8eb0403ad2d0ab025048ee4515854777","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:04:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65dadea1daaa4e9180b0dcb4d2712a1c\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a05dfdf6a23542ed8a8370cb7ae0e418 : SUCCESS in 36s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d620b2660f27491da5adbe598964a844 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91748b83c2a94a3893f61a82cbccebde : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f136d103fb1beb96db1ad153e6568ebbb4695837","revisions":{"8553827e96c7ee04cc98e0c4cbe84cb805110378":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/889045/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/889045/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/889045/1"}}},"commit":{"parents":[{"commit":"af819043e23658add7f8be23b185a267a7ffd0d1","subject":"tests: Use WarningFixture in all tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af819043e23658add7f8be23b185a267a7ffd0d1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:58:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"tox: Unset basepython","message":"tox: Unset basepython\n\nThis is a common source of bugs in tox 4 and is no longer necessary now\nthat we are all using Python 3 everywhere.\n\nChange-Id: I3733f779075c189c9a630416181b7d4c14ba599b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8553827e96c7ee04cc98e0c4cbe84cb805110378"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8553827e96c7ee04cc98e0c4cbe84cb805110378"}]},"branch":"refs/heads/master"},"29547dd015946ea09c0fa78f7761f94ed102372c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-20 17:03:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/889045/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/889045/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/889045/2"}}},"commit":{"parents":[{"commit":"de683a2a039d0578529dd33f080df9119b32fa16","subject":"tests: Use WarningFixture in all tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de683a2a039d0578529dd33f080df9119b32fa16"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:58:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 16:27:48.000000000","tz":60},"subject":"tox: Unset basepython","message":"tox: Unset basepython\n\nThis is a common source of bugs in tox 4 and is no longer necessary now\nthat we are all using Python 3 everywhere.\n\nChange-Id: I3733f779075c189c9a630416181b7d4c14ba599b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29547dd015946ea09c0fa78f7761f94ed102372c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29547dd015946ea09c0fa78f7761f94ed102372c"}]},"branch":"refs/heads/master"},"f136d103fb1beb96db1ad153e6568ebbb4695837":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-24 16:55:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/889045/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/889045/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/889045/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/889045/3"}}},"commit":{"parents":[{"commit":"dc20d4a2a24df072d4a4bcb3277831dbd3aed650","subject":"tests: Use WarningsFixture in all tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc20d4a2a24df072d4a4bcb3277831dbd3aed650"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:58:39.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 16:14:22.000000000","tz":60},"subject":"tox: Unset basepython","message":"tox: Unset basepython\n\nThis is a common source of bugs in tox 4 and is no longer necessary now\nthat we are all using Python 3 everywhere.\n\nChange-Id: I3733f779075c189c9a630416181b7d4c14ba599b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f136d103fb1beb96db1ad153e6568ebbb4695837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f136d103fb1beb96db1ad153e6568ebbb4695837"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
