)]}'
{"id":"openstack%2Fglance~889046","triplet_id":"openstack%2Fglance~master~Ie3d30669dffd80aa21025a7a407c373b80b2e3cc","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-28 14:08:47.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ie3d30669dffd80aa21025a7a407c373b80b2e3cc","subject":"db: Update import location of declarative_base","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-08-28 18:12:03.000000000","submitted":"2023-08-28 18:10:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889046-sqlalchemy-20","meta_rev_id":"187d58c1bd6b473e1e8e4bac21a88c1aacbd8f9f","_number":889046,"virtual_id_number":889046,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 18:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-08 08:06:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-08-16 15:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-12 02:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-08-08 08:06:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 13:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 15:35:20.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-08-02 06:15:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-08-03 07:35:25.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2023-08-08 08:06:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-08-12 02:35:48.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"2463d4925c152d175e63fa5b1e2e0a347e0a8779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ddf82e72d8deaf9d1314a7d65f072432bd129309","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:19:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b4fcb8e677140c699afea4a7e862b18\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/464efc5549a942fe85f8f3f04e8300c6 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb56298844f34eae88bb83dde8a7227e : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/139164a98cf54c759a17aa0dea5543b8 : SUCCESS in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40ac6b4ee38f0b23c3fddfdcbef81e8ce0a67dc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 13:39:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6613b86e9c9c418ab3603ea7209dbb34\n\n- grenade https://zuul.opendev.org/t/openstack/build/8acb30b867424912ae362d8d600f0fcd : SUCCESS in 53m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6871400a526e4e9583c25659dd47dd51 : SUCCESS in 1h 48m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4cc57b4e2a1642c48666244c3cdefdad : SUCCESS in 1h 01m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c289690ebd3441d9f44091e50222b19 : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bbe8eac9acc34d48ad43558ae7be60ff : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d752e4d2a0414705b3c3a7e51fffefb8 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e035d7f912a4549a9ede8d5e6bf9484 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4fb9ad850864455b02a905777059c0a : SUCCESS in 6m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0c92f83f6d0f4cb3b889bda2e457b9b1 : SUCCESS in 13m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d8edb61374d74c2fbcad14042a4409a5 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/405bc0a4919c4b5390304ec1742aeaea : FAILURE in 10m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/222031175c904c6a8f4a70bc02e843d6 : SUCCESS in 10m 15s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9e329e17ebef4083852acd614c5bd10c : SUCCESS in 2h 03m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/692e0fc5bf3e40758b36c945d9b687d1 : SUCCESS in 1h 47m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ddb607800f874fbf9e9406253aae70b8 : SUCCESS in 1h 37m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5fcc8e9e72e34031bbbf809006a23119 : SUCCESS in 1h 40m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/afd1094cf2284a8e82e74d08e3792ea7 : TIMED_OUT in 2h 03m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c89e61077d64e609206f211dee90321 : SUCCESS in 1h 21m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1d5bdbf77884f8996670ad8bf5914c7 : SUCCESS in 1h 47m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8faf06f2b838473eb68673f843aa7bc8 : SUCCESS in 39m 56s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/api/test_common.py\n  Comments left for invalid file glance/tests/unit/fixtures.py","accounts_in_message":[],"_revision_number":1},{"id":"d63e5211a9b4201065dd94c87f746cf1ec56a3c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 17:03:14.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8c6dbd0ec792fb8687fecd7cb0ff34819656cf8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:17:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/959a19cc9e0c4d3c80668775965139dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4cd1c522a8544f4089c2d76d6689d1f9 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b26d57c9bde74f908bfe6f5afcca5fc8 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd81ba3d5d73472eba4a95fbbe2f50cb : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5e562bad146e4b854618bb789505143282866b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 19:10:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/884408dbfc074b508f70cfabcd41a53c\n\n- grenade https://zuul.opendev.org/t/openstack/build/93dc0bba37444e07affbc312201cea62 : SUCCESS in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f15b3426018a49cb9e11d21542012d02 : FAILURE in 1h 16m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b94782b598ed4f8899d973a02ea70d35 : SUCCESS in 1h 09m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba261c35dd8b4d9db76a1d15f774fbbe : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97187d86a39e417dabba5d2b9eac175f : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b88c58baaec435e8ef7ab350fdd6201 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/021de2bac8574a0cbee38b8014b8f173 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a80f2f81c245e1bf5260ad15d504dd : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/eec7603ac4514776aa513dc78521e8b6 : SUCCESS in 15m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f175ea3eac5b476dbcd0b553cbe237a9 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d15fb716437c427386967b1d22a8686a : SUCCESS in 9m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a11693db7ace41fba90a8a2e8bdd5364 : SUCCESS in 9m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3cca5d3f5386437ea9b9b1b7cf372a39 : TIMED_OUT in 2h 04m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/186a082c2fec42e5a70731a30f6e2b6b : SUCCESS in 1h 42m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c468ec6146804e01a9128e68b171bc92 : SUCCESS in 1h 31m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c8ac68d038334afd9fa16f9a57718be0 : FAILURE in 1h 21m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/290fe359595b4fbd926e61439e744d21 : SUCCESS in 1h 44m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/652f9a5c646a46938d8d3913c85d3a53 : SUCCESS in 49m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d74c48b30f543cf852a09a7a3dd5691 : SUCCESS in 2h 01m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28dfb677ab6a43d39988ddae91e0d5da : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":2},{"id":"b8db2c936a65d111510a7107d0dac6de7354e083","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-21 15:35:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"27fbfcefe4bd2cd691c838275d6a8f9d9daf04da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:55:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"25886aa40e963f4ae222a26a96af6561d69b2cf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:11:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33cb0a3fb8fa44de8d9dce0333a85ae7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c17fe01540e84baea250c2719234a1b4 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/49329e4976ce43a5ac150886d59020a0 : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/78b67e8124e845529119553bff4541c1 : SUCCESS in 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"df7ad724d5c9f01a09d5e13f812a488b08757051","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 18:39:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/606f144d98e043f69878a4442a111acb\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a927147049140ca98b02f9fcd8ab84f : SUCCESS in 1h 18m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/945d0c2a03a4456090fe1d09a446e5de : SUCCESS in 1h 23m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/891712203eaf40a4b507240576df35ba : SUCCESS in 1h 06m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5933f19b81d4413be2333a33f162f47 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27172934b02b48589dfc2c732bcb29e0 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caf060e688964dbcaa566d9001bc52ed : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5eb92451c51f4a20a6289cf09d219238 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ca2ae0bb8d45f8896069573fed9abd : SUCCESS in 7m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1dc79b31f1c54c64bd0f6c95da175dc2 : SUCCESS in 13m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5fda3517243b4987a41ad2e3c0cd9d49 : SUCCESS in 10m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4402a86f647a46f68b5da9e9c992070f : SUCCESS in 8m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5a2cba017bf94ec6a4da3670e0134cac : SUCCESS in 11m 56s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7348b1a0cd544bf3ac6ab66b190e67dd : SUCCESS in 1h 28m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/387d9552ffff4706abc1b265397eb311 : SUCCESS in 1h 33m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8c81c8787dd14f288f49601750aebe99 : FAILURE in 16m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/68156b20dd594cd68d3c3ceaf0b040c7 : SUCCESS in 1h 40m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4fbeb82996ee40bcafa86677ae2e9b8f : SUCCESS in 1h 39m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47dbb82760cf4d83843a8be0ade007c3 : SUCCESS in 1h 25m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3309c3777695407eae618f292673935c : SUCCESS in 1h 34m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ed94c069152e41218fc2d7e80c294c45 : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":3},{"id":"14eecb2b0c42f1901c7e189812bd723e00cf8707","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:31:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a1e9500019ccde5fcca1dd8f7aa744949c9b5884","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 21:46:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b964cfa67d4ec89e48b0edb40f6590\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6186d725ac8c4583ab47a65d9d332137 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/07ef07d83190434bb903d349a00f2c5f : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/29875b05ffa2430fb8492b16d17d3708 : SUCCESS in 6m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e606d3b32e4600582956706c628dffb86a32ab0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 23:44:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/49285d42e8424b808e58d58351abd3ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/59358860e5e44b928182b717e539d55b : SUCCESS in 1h 04m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b1173bfd752b4fe3862e51c4c7c6075a : SUCCESS in 1h 30m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96632ff60c944d00bb95108b7b9c44e2 : SUCCESS in 58m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93fd43774039407090b80a73cab7e8ad : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf75c05c20c84ab3a92be0f78655bf00 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd41788d18184aa78b963decde9f31de : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f5b5eb73db445b4b342004bf57ea0ef : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb39e98c64594e93901f88e6dc19b94a : SUCCESS in 8m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ac28984d7c6a448c93f8c39d38e89a14 : SUCCESS in 11m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2dc7ebc99d8a44228a4015a5685eaed6 : SUCCESS in 9m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90d8475463a440a796100f135c2c4b7d : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4b03e2e4d3524efa8d3f9b69b8afe930 : SUCCESS in 11m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2bd571bb4e754beab1b7add31ba2fcf5 : TIMED_OUT in 2h 03m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/35a6039745934fcb97da1f012de31c07 : SUCCESS in 1h 36m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c3c8680ff92a4b57a7be50b9a5201736 : FAILURE in 1h 55m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/63467b10af9444ddb1b6e55478e64567 : SUCCESS in 1h 33m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/be48baac48c24743b5e2da189d3e9c23 : TIMED_OUT in 2h 04m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/017566cc54de4a43a21a3d7c0993e54c : SUCCESS in 1h 00m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2dbea793818840cfb86f41276873680d : SUCCESS in 1h 56m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/023ffb55620742409e3e9c746b3326a7 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":3},{"id":"02ca7fd6c8a449051b372dd2b676b71c47de520a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:14:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4c55f4fb83a2612a1e902ed775d16f463abfd6ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:35:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/992a712c0f474e2facfc776005daffea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b43e81f14d0d4e8e97997025ba9da93f : SUCCESS in 17m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d824adb1e99f4813a9270cfb88b58fb0 : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c38080b53df343c782787842f8eb0351 : SUCCESS in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc326cc99572ca6e1f720bdc9c1888f7d84af533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:50:29.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af69856af5844926ab95bf29903b23c3\n\n- grenade https://zuul.opendev.org/t/openstack/build/8667597c77554d2ab63bc70ea36c8a18 : SUCCESS in 1h 04m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4dd74fb25c574d4f934421b309294642 : SUCCESS in 2h 01m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a9d190816784380afbee7202c4c78ce : SUCCESS in 1h 00m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72079949030043bb815b5abe8fdb79c1 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97c89b7e57084bc2a11c00f2923a62ee : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2531e65baa14935983b3e06704e4783 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/df7fccae77744e00ae406eaf76a88a4d : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f8f498e2f2488a94bef1dd1e20a7b1 : SUCCESS in 6m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/842dffbd362146699f08a66926f21ee0 : SUCCESS in 13m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7566361548ef41aea6707bef1c04cf14 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4977c48de2ad4ecc93dbb21165622dfe : SUCCESS in 7m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5f6a095ae1564c538b4c8da6203992c4 : SUCCESS in 8m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4523cb83ce144fe9a2a1ffc07f19663d : TIMED_OUT in 2h 03m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97a19077862541049469a4b7f7e7dc8e : SUCCESS in 1h 43m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/299a72ea6418419dbc8f6c6308a8ec34 : SUCCESS in 1h 49m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4c8887fb5624488fb034528fe6e2776a : SUCCESS in 1h 16m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e9cb3842472f4f6c867e12ca933be9f5 : TIMED_OUT in 2h 04m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/030f303d25e744df93d912360b70dc6c : SUCCESS in 1h 05m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9795636064c246ee80999163f4120862 : SUCCESS in 1h 58m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/16dc0173179344c1803bea49aea16480 : SUCCESS in 40m 03s","accounts_in_message":[],"_revision_number":3},{"id":"15d3030d9951011f53757a1a2f0e0026b2c392e3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:31:52.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f47b2a4fe7309a76c1fdec1adcb68acaa1a8c901","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:00:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235dadf6c1ff4974871a762f49527150\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25434b5c8381476f8094239fc3afe1fb : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ae9edb12db924807bacc8daf9a65e988 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f048c54834164597966dcd2929429842 : SUCCESS in 13m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b2cf8932438d07ddd0cb83896355f428c7d9f9f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:18:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7cdf6cedb2442a5b9f15de69a343fb3\n\n- grenade https://zuul.opendev.org/t/openstack/build/87910f610a0044878f7fe1350afe2282 : SUCCESS in 52m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/38d167924e644d858980d3344196e063 : SUCCESS in 1h 33m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/079524ae1a9c4daf8e7fd831a2dbe3f8 : SUCCESS in 53m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7a32682f8c4517bbf1a08e8525aa50 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff424bde8c2a4a4d859e57e78fea7c80 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/956590ff48394fa8bc293681a79a1e93 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fb162fc7dde4bd88903180c1919c7b4 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29057013dffe4f99ab6ed9368ebefb27 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e6843537c7584373b35e41a8682a2764 : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d3281a47ec96460485121672fb833033 : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1ddd56c760014e02ade3e7a937e0594c : SUCCESS in 8m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/db259b7b65f749989b693c2bdccd471f : SUCCESS in 9m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7b3c25927cb8423b805454832ec9c8b0 : SUCCESS in 2h 01m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/842afd55f32340bebb278759ee3b310a : SUCCESS in 1h 32m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9816cb35b9034357993be206c5839d32 : SUCCESS in 1h 37m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88c07532aacd456e887812f8880affbd : FAILURE in 1h 31m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/70c9794b3afc4daab5cd4d164a6bcf20 : SUCCESS in 1h 51m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b7d2d638e0004da3bfc3e4792fc82688 : SUCCESS in 57m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4323d0a5161446cbb7ffdaf89423b733 : SUCCESS in 2h 12m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/72f5d3bf9c7a457197f59bb70c00f5ab : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":3},{"id":"4a8870f25afbe2870054654ab23aaff7bc78f1cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 08:59:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27e8b2a24224882c9a457191227994ed028b7c11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:08:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c966212c5e643bfa863298e5fdb69dd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/529b6315903d481d973fe30d2a1191f2 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a16e1912319a407ba9db5d95aaae224e : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/82ec671714384697a59c4aa69e51d28e : SUCCESS in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aad506e9a0b20c57320e1c87e6209d674036501c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 11:05:56.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f99f5899f7d949e7a66d5166d7a632ce\n\n- grenade https://zuul.opendev.org/t/openstack/build/19abbe0230bb47faa182b233d5f45055 : SUCCESS in 55m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c42803a28d14b288101ae6581910f66 : SUCCESS in 1h 30m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3603413e0173453584096ab62c9388fc : SUCCESS in 1h 20m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5351af4bd4fd4e78aa1d0393aac71082 : SUCCESS in 19m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc139f81e9d94168937d7d896fe21a20 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e36fdfec1c1f40f1894745c8043fe338 : SUCCESS in 8m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06e81932eb3e49dbb3c52b3efaf61757 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc0aa5af3a242619aa0ee92d7fb4ca6 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/81f5a478d87b427aa14de6b53f491c80 : SUCCESS in 13m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d73c95a45db544679c3aa9aedcc67276 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73b4f6a1a0b448db8a84992c1b72e15f : SUCCESS in 9m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4034b406cc184004a72d60f50956762c : SUCCESS in 8m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2db21a8d934749cf856c168726cff7fd : FAILURE in 1h 34m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0033d2b102c2495aae7ccf4e63707837 : FAILURE in 1h 17m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9ee6ea492e574e0fa20d1261d9e6690f : SUCCESS in 1h 16m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b1015f9863de4ee7a0b6aa965bdf8c93 : SUCCESS in 1h 58m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3ccb7196186f4b628147f5152760fced : SUCCESS in 1h 34m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c4c10ec42f74e0688ec8e0b37062285 : SUCCESS in 1h 09m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44d89fa2cf4b469ba2aa1bf05cb8b092 : FAILURE in 57m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5697e1dbce75411ea0446ee7c634c4a8 : SUCCESS in 25m 06s","accounts_in_message":[],"_revision_number":3},{"id":"455ae1e5818e8ac50fd233ab75a4d96998cb8a39","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 15:17:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"13e9276d6de1d864b46a1539ed0748ae35898ec7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:28:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fa13ea3835648fb8f68e374c87c2286\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/79b5fe55122c4366a2b21816bca7c7b3 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a55b27d40a2e48029e6950c8753afb9e : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9ead14155474b5e93d1da80d7790d92 : SUCCESS in 6m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"691aa278e3d7c4c0353a8954646332a338eaed7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 17:36:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da8724a8b15346fa9a607073952f1f75\n\n- grenade https://zuul.opendev.org/t/openstack/build/d6eac93c0aec4337ba31a3f0d3599719 : SUCCESS in 58m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1dfd8bda5488497bbc12d5f4adca1848 : SUCCESS in 1h 26m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b79d05111afb42dfacc993603833b4df : SUCCESS in 1h 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51b910b9b80e4f8bb8c60d4e19eb1735 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbbe55b2587748c7b0a493f1e8501f2e : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/086b64fc4030435281d17aa4a5e3cf34 : SUCCESS in 7m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5489ae7dc1741708c3a5df77d1b5b79 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9633d5126f1245c687270efd4ef2855a : SUCCESS in 8m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1118013c070f4802bfc0485fa92f32a6 : SUCCESS in 12m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/779e94002a864e77848f0206b6bd96b9 : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d26b667b58c499582fe475b2ca964bb : SUCCESS in 8m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0f3bf77fdcf745138f850ee856a8dfcb : SUCCESS in 10m 05s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/daea6ec2fda84264987a95e5ede8a676 : SUCCESS in 1h 31m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/776e32458a8848f3808a8dbfc52b37c6 : SUCCESS in 1h 30m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/009ebddda98b4d8298cdbbf8b6c68db0 : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d2966dc4bd2545958de0cdc8fe347d46 : FAILURE in 1h 36m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0ae8ed9260f94624b0f2f7188d76e845 : SUCCESS in 1h 45m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae4df52455164eccb6c20ef26eb00a61 : SUCCESS in 1h 07m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/647fa36a26f7407d936e06561a59e485 : SUCCESS in 2h 10m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/80eed1faadb84c5b947b554e337e5857 : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e26b00427688d346344b8f3258d07d63617d9eae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-02 06:15:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"513d7f2ffbf502e9b37beca2cb47dac7ce2b20fd","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-03 07:35:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d9a1491b8b2f049237333dee038b6073a8ed9dc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 07:41:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ae84dfb84f546acb285cc339c0f13f8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b44e951a8eed47afae5869f52a6c56b1 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2a64d5a1ab124952adab5fa4d2a7571b : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e5b5835ab28b4dd5869f8040f699c76c : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9af5e503bc6c48da97c58e4ea8fca348576cca13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 09:30:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0987b805be564326b27aed864477bf1f\n\n- grenade https://zuul.opendev.org/t/openstack/build/c162a71a95ce4e36b498aecdf22fd233 : SUCCESS in 1h 11m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4f5d5c878fd84fec8698b27de669f444 : FAILURE in 1h 30m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/be85d8c86d6f4ed0b7cda194f372addb : SUCCESS in 1h 08m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0ce426c7e2e4dab94b61f9aac489413 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f13d581364714165a37975d083df4244 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d0471ba6594b8493d0282fdc518a8c : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c29d9e7c7cbf4dd1bcd20a7bb84b38fe : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6802f2463f14c819ab56662068b1dea : SUCCESS in 6m 23s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4b308d3e3fac417e88d83e43e64d7f6d : SUCCESS in 13m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aa3ce83e9ee44b21b6f7e1fbae7e2925 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d36bf10949594614bda364140872bcf9 : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7947241b064e445d83717338c3d5f710 : SUCCESS in 9m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aee5d45819384d88b1cc75a29c2f6904 : SUCCESS in 1h 41m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/33be512f401749c9825320633d2f9f1d : FAILURE in 1h 36m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97d4741e6a574724824e33f64ddd7911 : SUCCESS in 1h 45m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/69e40f4a99b347b0806328b7a858eb8d : SUCCESS in 1h 26m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/077fbc19f5244487874690da6581f177 : SUCCESS in 1h 15m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/534b9448a03442ff8af2f024abf13152 : SUCCESS in 43m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f68e4aa6901406396568f8a98f8a1c6 : SUCCESS in 1h 16m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d4bb3c9712541228e2db7ba651e40b2 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":3},{"id":"73484833ce4c0ec247d8cd1cf75d94aeae678f2b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:06:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e230a83a6175ba0d4dc096babf1f08d413da073","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:06:33.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6042f5932da86e9b87c9b3d28a6cd9f0298110f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 08:11:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/199c295f6f254f9bbd70eabec8608371\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8154e2834434c56be107b81c3721fec : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/851cc9b8c7bb4cd3877534dc4ac0a90b : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca03da318d714ac09c4613a793376944 : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6457e3ead1cb2dd7c8d6c7c72f057425f9399c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:11:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c33d1059ae4b3d832b9d3636364648\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a6c7fb60e0d4c29aad2b896038c191b : SUCCESS in 1h 05m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0cbeb0572923488aa144beec48bcbf8a : SUCCESS in 1h 22m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c11ef407468a4e7cb5638493e2263902 : SUCCESS in 51m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f777e0d1ba43408fb8189952eb2a4827 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f57d64314402444d889567d91abeefcb : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c4d828e03864f15a0070997746ba4d9 : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/083cf43e6bd74c31a2abc7e5295965f4 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baacd8471c3c4ed49a0d18739fd5338c : SUCCESS in 6m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/da7b367b561a4673b9ee638077e97e7f : SUCCESS in 11m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/62a3ee528f4c4bd685d077858884a3c6 : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/acfb8b768a454114be736935484552fa : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/52bca5ecb3274a53a1896154214eb73d : SUCCESS in 8m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/864b0857345144d1949663ff9a87eafc : TIMED_OUT in 2h 03m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/181575706bf14967a3be80ed12c5c743 : SUCCESS in 1h 26m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/801c9e61a48941f2b8d4114e717741a6 : SUCCESS in 1h 23m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7eb37575456b47928179eaf524af0164 : SUCCESS in 1h 34m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/818ca1e432154bcc89e32c121d72a815 : SUCCESS in 1h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bc073cd4369443a9bf85dd6e33d7d8f : SUCCESS in 43m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9cc12fed84384bd1a29c09f2af74f0ef : SUCCESS in 1h 18m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bea6a706743745cf9e195b90675a4206 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":3},{"id":"53d1b0b7b1fe8e0279e3125f434971929c9bbbed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:12:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dce46f260ee31e85d2c7610c96fd0dab9d308cb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:19:19.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889045, https://review.opendev.org/889044 are needed.","accounts_in_message":[],"_revision_number":3},{"id":"26ff5e2d91aa9cbc049604cca615505ec3a4af1e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 05:56:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"906e829e76ddabb68a842826aeb550fcd9bc86f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:10:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c78a8df77a004502b1d0470c947a28d8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a2e992a994f405a911b176e305cdf96 : SUCCESS in 12m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8535a3edf7bb4d2dacd579172bbe3659 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0193330f326946a5975e4acf05206572 : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"671eeb35ce9733bd9b5288ea9ce169a4754ed4ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:00:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccdf9292c3254775b78af3e8ac2d125b\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e990ec124954986acfd74dbb54d57bb : SUCCESS in 57m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b14da4b5b8af40529cfaea83bbde76bf : SUCCESS in 1h 16m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca688321083d473ca7aac1db8627442a : SUCCESS in 59m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91c85cd7ecd41c886c50fac1496d059 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f389301747a841c7bc7670df4ec7a76e : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe6b831bfc2e4e8197221f738f7f6a19 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/023d75f9a94144e0bc0fe0127c12f4d3 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5ead2be9114f95a535c44c735a5078 : SUCCESS in 7m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/996f7081a94c46559037a89ec516a59f : SUCCESS in 14m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c10a742d1014772adcda920c077e049 : SUCCESS in 10m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d3334da9d7ac4917880de4738f74df71 : SUCCESS in 9m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3731c22622ed4fe2a7ab7c8faafda1ab : FAILURE in 8m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ce12076a21e84d6182b72f77f1dfedbf : TIMED_OUT in 2h 03m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1f5e5de300534072be7c0b258eb420db : SUCCESS in 1h 32m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d85c0b2467cd4478ae81311e8d9df8d3 : SUCCESS in 1h 23m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/03fb7ed911d44fd9be9b924b8d762bd9 : SUCCESS in 1h 14m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cff1597114294a28bbd0e8b33afbf309 : FAILURE in 1h 50m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc4478a5a07045f4b3d743a1161b8ad4 : SUCCESS in 1h 06m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f87406d945c241dfbab55ad787795736 : SUCCESS in 1h 31m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a7a516ad59194a929f9dfa1908319d68 : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b7869080ba6cecde7ffa7b8e3bb87501a88ad3bf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 08:08:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a328a0ca5248f8cdad25203a2595f0bb9da28d1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:24:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f027b5ced9498d8eb2c43d6c537ceb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/756290896b8b43ab9936e672dd36b236 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c445a3daaf2b48139822ec9301849a73 : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af9144020ec6408e86125ac01df6ca42 : SUCCESS in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a44500171b02a9958ea00a59c8ca9aafbdb1af2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 10:19:22.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81c28afe89124d009e0f63568e81c55b\n\n- grenade https://zuul.opendev.org/t/openstack/build/bfc336a3f5624ddd988b6c6273054757 : SUCCESS in 46m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/66502896289e43a4bce68271deaa80ce : FAILURE in 1h 10m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16f65c0e5ca54eea9c591233e28ee215 : SUCCESS in 57m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63891bb785934c7bbd9a02a3e5698a01 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46af97c31501489c98c2319f83dc9a0f : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0f781000ac74390837577082ccae795 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b85c4a074964ca2bec268cfcc504eda : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e338c1e9f3f4773a27e7a49a517ca87 : SUCCESS in 6m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/27a3ef0e4dda4e86ac5a1e349c1ef36a : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8cca7aed8db54798af1ac0439b635958 : SUCCESS in 9m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01df71497fb74aa2ba7d2f510a3123ec : SUCCESS in 8m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7bbc4c7a3f534e828384d4992e44b522 : SUCCESS in 9m 05s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/25e892a862b44459b654d648f5042b1a : TIMED_OUT in 2h 03m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8746fe39f6294e149065451f259b6f44 : SUCCESS in 1h 13m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/831a4789092d45a0a8f28b8a821d36f4 : SUCCESS in 1h 29m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/115d3c88612a45b7b01a3f3df8c057e0 : SUCCESS in 1h 42m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2b590769e3ce4d4da0169c76e8cbe106 : FAILURE in 1h 32m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/877b4b688e78446a90268a6f0a3de954 : SUCCESS in 39m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca6a17c8052042a68e4686b3817607db : SUCCESS in 1h 52m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6aa5c50a760f430590db4237d17d8a47 : SUCCESS in 23m 21s","accounts_in_message":[],"_revision_number":3},{"id":"efa9aba934934f99e91c9e3a6b7b55ebdc675f75","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 10:53:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6936563d85c5a63312e6af44c871ce70ea4a05a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:07:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c1002965ec455ba44116d1e4536c42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aaae501bfd3a42e89951f4b1878edbb6 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65b04a5bf7784c86b98e04d71fec6a1e : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a1dfd47e52d044cb8fb7b3ff73958cfd : SUCCESS in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f656617860abaca0a08ef9f6134673b9a357a01b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:09:01.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66fba446e02043ecaaac59cb2b3b453b\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b892a6522b744c39305de9df51aad12 : SUCCESS in 49m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d753a17163354258b57e3859504e94bb : SUCCESS in 1h 24m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77fff148ca104b1a8266f1416b9b170b : SUCCESS in 57m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4373ea93a8d34938a6bc42180a4be59b : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe82a3a9685e4a8ba316a70dae32e38e : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5e84b6c532041bba4e5da347bba6323 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/955cf67f3fdd439ebec03e65081a041f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7833c19874c4b858b227c442af6da7c : SUCCESS in 5m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1ac62b2565cb4cd9a773d638b0777e6b : SUCCESS in 11m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9904176357c54e31818564e16820561d : SUCCESS in 10m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ed3f23039d524bef8d2ea1d948bb2922 : SUCCESS in 9m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a1b8dfa199804a17a88d35a8a68ed074 : SUCCESS in 8m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ac9e785aa7c04da0b3916bbbd2d01d28 : TIMED_OUT in 2h 03m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ee947638af54498eb923edc1c1fe8a96 : FAILURE in 1h 21m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/440ba4f625084115b6ade0da0d76e561 : FAILURE in 1h 21m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/22946b00d26c4ddc9d46a722f8985172 : FAILURE in 1h 24m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e5b8a141c56d4ede9f12cd389a05ae04 : FAILURE in 1h 47m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af6191a648a44f6ab297110825b90ec6 : SUCCESS in 50m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8214afa0e2a487a93c9849b380d30cc : SUCCESS in 2h 07m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28e1f8b8b8514fafbcde9f681fb71fc4 : SUCCESS in 41m 29s","accounts_in_message":[],"_revision_number":3},{"id":"74238fcebd63ad89752d74053d9556da25795e8e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-09 15:58:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"995c9e94947373e3a3c3ea9652dede82bebc453a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 16:14:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8d4d4e48d3e4e1e8e68cafd6b236760\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d0d86d90cbe4070ae88e18ab7415b02 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3767f29430ad409eb54b6d2d53a62a1f : SUCCESS in 14m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb105c8ef9c645bc82254982947a15ab : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6194a6ecd70fd606b10e6e632cbd349c7017e56f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 17:43:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82fc5a684ab64c0d8900a7a8d207114c\n\n- grenade https://zuul.opendev.org/t/openstack/build/9db3f4ab12124bf8850e2817226a5dfe : SUCCESS in 1h 09m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/984497b9f5b54c53a84d6615b689684f : SUCCESS in 1h 29m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b29b750a9316412284b6ac4bc0b0c57b : SUCCESS in 1h 02m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0181dc1b847c42d0b60cac4c69fc07b6 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c14b6b81cc04af19bfb614cbbc6bf14 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7933a8b26324da796b8440ca343ebef : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f07fe649a2b47b9a6af4a83453ce004 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f25f67611e84497d8b1636ac77619a91 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2644c4fdb9bd41d1a3beeac6e34bb91a : SUCCESS in 12m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b7d2e061663f4c15b6a8a397699ae8d1 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6fd5a104969742bd80bc12491e0bc058 : SUCCESS in 9m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/faf008214da44ed0b79905dfd13a6fa0 : SUCCESS in 9m 19s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8f594aaae5254031839b7c7db8bd0e84 : SUCCESS in 1h 42m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7293bf970e874eada46046c9b3b9992b : SUCCESS in 1h 30m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/63d7e3fab5d94aa69ecdbb0209c0d8da : SUCCESS in 1h 07m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/74d74c39135e409a881af7192334983a : SUCCESS in 1h 30m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/81ea7368e16947c8beea8499eeffba8e : SUCCESS in 1h 32m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb67222b6c6346a6b3cac732fcb318e4 : SUCCESS in 1h 04m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/64b2900e86b54d87bad957c00ce1b11e : SUCCESS in 1h 35m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/336792f1a8314a20aaa845aa01b60204 : SUCCESS in 26m 04s","accounts_in_message":[],"_revision_number":3},{"id":"965f16293cf324290630eeea5b98ba4a18f6abb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 17:44:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0036eba566452079c425b2964758ca1e926d0b3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 19:41:50.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f3e94a16992f4cb697842e777c480613\n\n- grenade https://zuul.opendev.org/t/openstack/build/9efa7f95a4554535924eeb074bc20e78 : FAILURE in 43m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70c45276a47d4ce19aa4290f4bb4e0e0 : SUCCESS in 1h 24m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cf58aa9d54f473988216d6223014294 : SUCCESS in 50m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d373b52c704079be79613fb70bd87f : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe38e82edaa04c8d8618cc1fa591c89b : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de47dd4a42cc408da5ff9b7e8e0b3e7b : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aeaff72e532349a5b8371e85d240924c : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d626890e634b46488d20367cb71ea509 : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/585650a558d9455e816e55467f376e6b : SUCCESS in 10m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4971946c71041619ba1040f4187ed8e : SUCCESS in 9m 06s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6e62ea37a4444900b0bb272d514917d3 : SUCCESS in 1h 17m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3692a1b57faa4ab49f7560ee1f2b46af : FAILURE in 1h 01m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d981bd30e59445688fc6a094d9ae32fa : FAILURE in 1h 10m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae4c94799e2d4117ae310acebb0e169a : SUCCESS in 1h 01m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1fb78c2bdec45f3a83b126733c88202 : SUCCESS in 1h 48m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/18bd9ae87c144bf8a423ae3275edd2c7 : SUCCESS in 33m 40s","accounts_in_message":[],"_revision_number":3},{"id":"5649247cdf7705c3c2b41b98d3f32fad525f9a08","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 09:17:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0cab9b2f16a8051b1a18f9a1d9cecdeda6cf2319","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 09:22:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a3a32fa03b94ccb8969f5ae4ac0b409\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31b01563c4414046af64b084375b2cd4 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28092e66e8ff41f1b4c7b7a414cb3cfd : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ede94ec09de54e3fbca6896a11c726d5 : SUCCESS in 4m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2cf7617bfd0358e8ff1e363b9fdce659f0e59f6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 11:33:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e47043912f26411d8b93862895239f2b\n\n- grenade https://zuul.opendev.org/t/openstack/build/799063e7e5b94a4c9ff7cb0f6b307566 : SUCCESS in 50m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/79674a4145c244c986e244a71e73f056 : FAILURE in 1h 41m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/429223e5673848589a2b69acb3520151 : SUCCESS in 56m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b5dde523a6647f88635d75496fc2c9c : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc8e40300dd44ac9abed12cfeb951599 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d89a063739345879442d1a20eaf155e : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/74cba98312f949e28e6338e6dcbe6975 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51ec58e4f5cf4eb889c45b3eb5d1df34 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7f6b1e6e5cc449109ed2986936533f0f : SUCCESS in 12m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c4e751b3ea7b40a695397ea65b50f28a : SUCCESS in 10m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d62ff32ff9964a848a05370fa462ad52 : SUCCESS in 10m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d2d7ab39537a4881aacac869d05bf9ac : SUCCESS in 8m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/87f6cbda1cf748e8b2ba34edca687e4b : SUCCESS in 1h 51m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3833a48ae1004e0e9a7205ef89924f75 : SUCCESS in 1h 35m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6587e7c91f524b4f99561c8b17ac257d : SUCCESS in 1h 44m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e9d41c042d64489db496451a2cd06f1c : TIMED_OUT in 2h 05m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4770b190af5a412889258da6df5af007 : FAILURE in 1h 37m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f93cd229393494687195e41322ca3ff : SUCCESS in 54m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/57ad3a2333e944d68f80d2a576513bf1 : SUCCESS in 1h 52m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/83aa3e8ebb164acd8bc8cbbf1920527c : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":3},{"id":"cbaea95a97f7a8306655ccb2e71178a3db93f995","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 16:42:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"760e4e3abd4bc749bf9ee7810d2c86907707c65a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 16:58:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ffa87673eaf400687b69aa25e0d6b18\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba781cd6ab6349d1a5fb8cd2232d59f6 : SUCCESS in 11m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bae99f6cbbec4752a9fc9549af4cf9a8 : SUCCESS in 11m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5fc30d6163e24c74ba18f2d487747fec : SUCCESS in 13m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f324db8e583d07193fd5ffbb6438eff669570e4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 18:49:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6decc4b6bd14ab2aee61befee55805d\n\n- grenade https://zuul.opendev.org/t/openstack/build/05c14fc6ce0e4293a2cbe944909e3113 : SUCCESS in 1h 08m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f8ccef5772a4d5fb939d07bf6c3bae4 : SUCCESS in 1h 25m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e110c4eb3616479ead9a443727c79815 : SUCCESS in 51m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3facddea70c4232a65302d437ed5b7c : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3539d31e1c89443792ef5ba57664b7e4 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac9f6d4a146b45ddbdd7d6ef2975ab30 : SUCCESS in 6m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d7f10f890b24bccb510d746f1f3b2fa : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c953f4fd8d44faaa3d8d248afcab6fe : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/55f67f88222245ed9b5815432381eb46 : SUCCESS in 14m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/450ba2ec903b4c3faa0811317fddc226 : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45b892ddd9cb47369664d26950a36491 : SUCCESS in 8m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b28db1f3841d4bde9a74b59a0840c988 : SUCCESS in 9m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1ec4817163f44a8690d381764ed7fea7 : SUCCESS in 1h 52m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/90ce5c5d78dd4e328d1a91bc9992bca0 : SUCCESS in 1h 36m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/16bd49327000475c9e314d339b51426d : SUCCESS in 1h 03m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cbb4cbfee4f54391bd97545d6e9092cd : SUCCESS in 1h 16m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8c615bd69bb7427da4329f67fd45d22b : FAILURE in 1h 18m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78e9858d4d414976838b573d29a0c04c : SUCCESS in 44m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/96f14403314e4cd2bb7b70ac88ff2492 : FAILURE in 1h 59m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/62f1ae6047ec4667a07956884205d6b8 : SUCCESS in 25m 34s","accounts_in_message":[],"_revision_number":3},{"id":"58aa96d6aa88dea1fcbab4fec1384ba978e9c175","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-11 13:11:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"07a5e8b4dfd0beb049323d10729dfc7fc0f479f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 13:17:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b548784fdf843a0bcd6948d17f2f7b6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4522076c131444ad8a9cee8593fcc2be : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7581d63d4c0f4264ba284bc53c83769f : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10b58689bf7c4facb1c403fadca7ca74 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fc916f49e651d1efe4a2611d0c2908ed88eb0f1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 15:23:32.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bdda699b2074b2f907980655ffd865e\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe1c7d9014fa4497b2e62a1965bfbca2 : SUCCESS in 50m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/458b951702714011b0e81ecc82d44ab0 : SUCCESS in 1h 06m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38169014b6004d0baa45465b35dcc439 : SUCCESS in 1h 03m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d766f8d7ac7432b8a1f2a714cd1a1f8 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2588a53943f402ea1a9f9e8466c049b : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a224e943e834da19ac13c410cd97b03 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bd63bbb9c7f48e48056f78d3e39cf35 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033589659d834ee089187ed4f257f59e : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bad3eb694cb243a481fb4bbab990ef02 : SUCCESS in 11m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/16fa24f6b0de480989d222f1e9aed689 : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0fff20dfa4b34bbc838f9632cfcb126c : SUCCESS in 6m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9d31973b5b994c439c97ec07832e6ad3 : SUCCESS in 8m 19s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/188f20db94c84fdb81f49bc0aea74a98 : SUCCESS in 1h 48m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2348d5a4edaa43a8a4cc9d34b9603667 : FAILURE in 1h 14m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/64874b378ea54eae827cdba86aa58dee : SUCCESS in 1h 09m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ba6a4ea0ef4744bb93793658b07736a8 : SUCCESS in 1h 48m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/11c807f507ba4ff5ace322fa6308c210 : FAILURE in 1h 40m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6916346fa4248d08a8f49c2f8c42bf9 : SUCCESS in 1h 06m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/691a2a79c21d4ac7a63e3768e5a9ce3b : SUCCESS in 2h 10m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3574000439934b11afa6fd98697a38ae : SUCCESS in 27m 26s","accounts_in_message":[],"_revision_number":3},{"id":"55cd428ff5ef56f9e2e1da5ce57baf384035a839","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-08-12 02:35:48.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d243fec4a8c326a65b15f79779568b2095441da4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-14 09:09:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb2f67aca0679979910bfcf5533005e9986a1aeb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 09:40:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f39cdf8837e94221bc6621bb4ac7c784\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e8f5f1c6db640748f6d97e78d372bfb : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f27671d2e63473383107dd0f5dc60b5 : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f15198b719ed42398400dd6276d2d942 : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4ef0c8efc625633c2b44851d6c52db070afec955","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 11:50:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4a29667f3e74cc7a34ff138215dc603\n\n- grenade https://zuul.opendev.org/t/openstack/build/e50d007b5dc5430ab109609351f22bb8 : SUCCESS in 49m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36da273a6c68442e8e234cf1202c8764 : FAILURE in 1h 27m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e90c0911411c421f932542cda87605cf : SUCCESS in 1h 03m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7f4a445a9b4bd48c28c681ee13cd21 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e68b78b9d544089b94ab6bb968e5eed : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97ce333c9b3b45f7bd02dbfdd4ab3dc5 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7ee22b9c83f540ab897bb8305dbe00e6 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/092bde23836d48b5a2a630641de6863c : SUCCESS in 7m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d1693215a3284c05b14c24d5f15ee688 : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a33b54b7847a4fc3ac86a57267c48092 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be1a8bdd85db454ba1e146dab7b724c7 : SUCCESS in 7m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/43a960f8a850415c8ed61f4ce994d63b : SUCCESS in 9m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/61452d476f7e4181b96d14cf588d9f6b : SUCCESS in 1h 56m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d07da002e9e14492afe0ed439a675bd5 : SUCCESS in 1h 36m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0cd4ceab2244a06a80529703bf05a90 : POST_FAILURE in 18m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a2ea46ad6c9643798c710c80bfbda556 : POST_FAILURE in 34m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2c0e23ee5eb349f596f7b3b560378b78 : POST_FAILURE in 21m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08271d3c15b943df975a32119386c2a8 : SUCCESS in 41m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/202e80c2faa244fea5867700ff187d70 : SUCCESS in 1h 34m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ca5f89e230df4fcb921e55f300862d02 : SUCCESS in 41m 57s","accounts_in_message":[],"_revision_number":3},{"id":"407353aa4e795015a88f0b9e5ef3deeff8eec460","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-08-16 15:24:17.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"97399c930530c57b33491d8774f3438ec555ebd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 15:30:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ead96f2c771a489b9cc01b3360820ac3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce307ac7556f48eea1714b19f060c640 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e1116ab5de3f479e814cc8337cd7cbfe : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0bebec052467456bbe08e6dce660fff1 : SUCCESS in 5m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"70e1e7de40bbb6202cf01debeea0dda860aebb99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 17:52:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db132ee6014a4eb8a7609396bf756d7a\n\n- grenade https://zuul.opendev.org/t/openstack/build/7517d877f0c9480b90d621e506a109f6 : SUCCESS in 1h 03m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e998b4102c2d413d801be951118c12c8 : SUCCESS in 1h 57m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e06b85be3d4846a6a905ec088c06b09e : SUCCESS in 54m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf26a0e65404deaafa020949354d49c : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/16c03f5dbe0448909611791f634323f0 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46ed83f9f2834f59a726dba01df90975 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0d1d46724ad4797bda9544d87bbafcc : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aee4587e49045bc903d14cd62cd39d0 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3e63cdad3bed41aa95086684f904ba6a : SUCCESS in 14m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09c65899206440bf9c1491fa60dab87c : SUCCESS in 9m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5d9df568f8644e11806569ef77a99a5e : SUCCESS in 10m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/48ea1ba4cafe41ffbcd6c9cac8deefb6 : SUCCESS in 11m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4d7eca65b747444d9e4710582b3fefae : TIMED_OUT in 2h 03m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb17b96237414ae59a2cc8b36adb46f3 : FAILURE in 1h 44m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7f5d7deda96149729bd32abb3e8f48d0 : POST_FAILURE in 21m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fafa17d86eb94e5d94eadf99bdc25954 : POST_FAILURE in 38m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc5263b099a64e64ace972fc8881c359 : POST_FAILURE in 25m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4cfeb50fe699446abbf2ece4e4c33378 : SUCCESS in 1h 06m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cc62add6cbf34aaa8ca12c3b5084d4b9 : FAILURE in 2h 13m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5805398211c542b7838440ce4f17fc9b : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":3},{"id":"8921115c4635265adfce0ed2cf91704f055cdcc1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-23 14:57:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0ad31127ebbd2a38ef00a1ee5b722c25fcc51bc8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-28 10:55:37.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"327b255d06718d3aa9b03f56310bb95e9a4d90e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 11:07:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8156068d277e4be69bb10a7a564adcdc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9453eab0ec5467cb083d6d34060df3a : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebdba8b5399f476293af302eeee81eb2 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8948d685de434cb0965209c8c7d5f1f3 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ae822c4dc5d4739a1223df686bc8c3e : RETRY_LIMIT in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8b647bcc014dd0d9a1648849332bbdc335c46d2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 12:58:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67ab0b82c3724d0d9732fcd773f409fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/f3501e296f114bdf8aca1924078ef720 : SUCCESS in 48m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf3135b96a70400bb884365a17417013 : SUCCESS in 1h 22m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/800fc227970e4115aeeb6f1a85b9430f : SUCCESS in 51m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eec4ea590504d27b85cb4abd7999171 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8931864ef9d742c1bf0cd71d342a86a4 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52f28ad44a43400290f01d497584d2f1 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a4b82f0c5574619a47a615fd2416f3c : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ea26fa8f62f41a5a3af9a779f50fae2 : RETRY_LIMIT in 2m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bea5980231f749439ba4204840149787 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/53c8dc5ba9674913b0f0dad4413bc438 : SUCCESS in 13m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6d4bffdd212f4ef9b6a73aaaf457fa0b : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/527ee9c8f4d142ba84b34d7c55bf6389 : SUCCESS in 7m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1451468b46d9426198b3fea6f3786bcc : SUCCESS in 8m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fd181ba9b4b14209b764aaf10591c027 : SUCCESS in 1h 39m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2d1686729b7c4d9b9052eb79f7b45d91 : FAILURE in 1h 27m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4ef36b93f3c4e4dabb4f16f73b15cdf : SUCCESS in 1h 49m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6036f0bc1147490eb62c7b1ffd45facb : SUCCESS in 1h 16m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/141356861a334877a5015fc85bad2b4e : FAILURE in 1h 43m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29cb9bfe29c74159830e0558e3bc83e5 : SUCCESS in 51m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa6c1f6bf1594adb956ffd29634ee161 : SUCCESS in 1h 59m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9b50153eb5c544bab5c180b72e68b574 : SUCCESS in 28m 50s","accounts_in_message":[],"_revision_number":3},{"id":"758e4e8a5a0d6e50a4fa01591b1a1a35f9210951","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-28 14:08:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"173f28eb0ea0680518463d09551dc2d7f4fd69c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:22:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bf3b601e2f94dbebb8529d457b9eeea\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/41c1fbc4b9e44d4e942193adc4eb7d70 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d32c06e54e34492b114146cc4347134 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8a128aa85f845559b1f66cff3d6e9a2 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc334dd8b3ce42a0bfa793a28d97360e : RETRY_LIMIT in 1m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e72e4347ee07d6b53758ece651ef35228850f8d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:11:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9343093770846fa846ea30c058aa33b\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a659859d20a4ae78b93786b42647465 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a0f7152cd01472d87603acf05249921 : SUCCESS in 1h 23m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd7ac72ffc75455fa51e788d217c9137 : SUCCESS in 1h 03m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f31ffef83ad34770b7b6dff8d9b904b1 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee06dcf79c441b8b1b2ffb100abcda2 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1da9743a636c41ed9e71b5fd51a4fe6f : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e896ce097b5841e2a6d16abbc193ea6c : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/64e5368d9cc44163981ce265480ecd94 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a19fea98c46e498b82355a20592689b4 : SUCCESS in 5m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a802c0d385ec4f2080b00443e7083826 : SUCCESS in 15m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf9a1a6d24424e5ea9747d6f64e50ab9 : SUCCESS in 8m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce590dd57adc49b8b04c430370204c57 : SUCCESS in 7m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fd14fb373416495bb2110b9c08d295d1 : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eef3b91758414e8bb918f483ca9d076a : SUCCESS in 1h 46m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b975376f470b48759838c60ecda3d093 : SUCCESS in 1h 14m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a45cbcfb7d9d46febf257d6980112a49 : SUCCESS in 1h 23m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cce68d3f294149b7830021c5d0059a1e : SUCCESS in 1h 27m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f6cc0f635ab04c938b32ce1e4f2c6aad : SUCCESS in 1h 35m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ebd5d58876743d4b2179e0cf69a3a7a : SUCCESS in 1h 11m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ed1f3599351485a8431c27391d96931 : SUCCESS in 1h 56m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d185ae0eff447cda955a725c9ae7097 : SUCCESS in 40m 01s","accounts_in_message":[],"_revision_number":3},{"id":"0b0a117afe04e7619272edf65224744c4c7e5467","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:11:59.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fbd4d678e74e447e78dfc46aa835099e4e455070","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:10:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8368b90bbd24535b0868a0e8c4a0042\n\n- grenade https://zuul.opendev.org/t/openstack/build/e8a38adac9054d45aff13ddd48e5ce4d : SUCCESS in 1h 07m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b9915d00d0f44a9a6dba15fb7b6da1c : SUCCESS in 1h 38m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/018c162ad7d04d5ea190f5735aa035e5 : SUCCESS in 1h 09m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/769fbf4a57154b32b3a3ae06c8d4cd6c : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b684a2856647efa15298fb026ff1bb : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec87ca2c6a244065ae6a49a2accfb9ce : SUCCESS in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d29f4f6bb4b0405ab1341e460bd38c4a : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2904fe3f70bb40e787b9d48482452bb7 : SUCCESS in 8m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ddbc6dd467ec4b408efb2295f249f2b8 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0ebb6c7dfd1427ab4561153284e4ae0 : SUCCESS in 7m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/03d90d723c084be699d17636a25be09d : SUCCESS in 1h 21m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a2ff78019b4d4bd593d33bd188656cfc : SUCCESS in 1h 46m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d8c12916d95147b4bb89a09ee0f81d1f : SUCCESS in 1h 23m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6786584306934125bc4b794f838ac694 : SUCCESS in 46m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea1a417253754cd1974629db915b112a : SUCCESS in 1h 57m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2c403378003b45e38d34de0dec836ac5 : SUCCESS in 42m 46s","accounts_in_message":[],"_revision_number":3},{"id":"9d091da3e18cfc6d7d43f4daaa5a6302e2f76760","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:10:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"187d58c1bd6b473e1e8e4bac21a88c1aacbd8f9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:12:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09ab28a720db40178b4881c914a52886\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7161b9f70c486297b9fe25d50e79db : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5636008dfe30beb7311f7f65b500cb852cd394e4","revisions":{"7caad1cec0444a8389bbc39a42baeb8129118de2":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/889046/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/889046/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/889046/1"}}},"commit":{"parents":[{"commit":"8553827e96c7ee04cc98e0c4cbe84cb805110378","subject":"tox: Unset basepython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8553827e96c7ee04cc98e0c4cbe84cb805110378"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 16:10:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Update import location of declarative_base","message":"db: Update import location of declarative_base\n\nThis moved in 1.4.\n\nChange-Id: Ie3d30669dffd80aa21025a7a407c373b80b2e3cc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7caad1cec0444a8389bbc39a42baeb8129118de2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7caad1cec0444a8389bbc39a42baeb8129118de2"}]},"branch":"refs/heads/master"},"c4b4679649e81ff86a9fb270a0ed3b0b3286ab86":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-20 17:03:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/889046/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/889046/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/889046/2"}}},"commit":{"parents":[{"commit":"29547dd015946ea09c0fa78f7761f94ed102372c","subject":"tox: Unset basepython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29547dd015946ea09c0fa78f7761f94ed102372c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 16:10:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 16:28:05.000000000","tz":60},"subject":"db: Update import location of declarative_base","message":"db: Update import location of declarative_base\n\nThis moved in 1.4.\n\nChange-Id: Ie3d30669dffd80aa21025a7a407c373b80b2e3cc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4b4679649e81ff86a9fb270a0ed3b0b3286ab86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4b4679649e81ff86a9fb270a0ed3b0b3286ab86"}]},"branch":"refs/heads/master"},"5636008dfe30beb7311f7f65b500cb852cd394e4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-07-24 16:55:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/889046/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/889046/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/889046/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/889046/3"}}},"commit":{"parents":[{"commit":"f136d103fb1beb96db1ad153e6568ebbb4695837","subject":"tox: Unset basepython","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f136d103fb1beb96db1ad153e6568ebbb4695837"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 16:10:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 16:14:22.000000000","tz":60},"subject":"db: Update import location of declarative_base","message":"db: Update import location of declarative_base\n\nThis moved in 1.4.\n\nChange-Id: Ie3d30669dffd80aa21025a7a407c373b80b2e3cc\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5636008dfe30beb7311f7f65b500cb852cd394e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5636008dfe30beb7311f7f65b500cb852cd394e4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
