)]}'
{"id":"openstack%2Fglance~889047","triplet_id":"openstack%2Fglance~master~I6b2cbf0b97bb3af7354f2c10a8d83909db963d43","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-28 18:12:47.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-07-25 16:29:04.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I6b2cbf0b97bb3af7354f2c10a8d83909db963d43","subject":"db: Avoid import variable shadowing","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-08-28 18:14:27.000000000","submitted":"2023-08-28 18:12:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889047-sqlalchemy-20","meta_rev_id":"033736ccb3df4e9e935e4004d6580eaf0bb5ec70","_number":889047,"virtual_id_number":889047,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 18:12:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2023-08-03 07:41:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2023-08-02 09:51:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2023-08-08 08:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-07-28 01:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-07-25 16:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-12 02:35:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2023-08-08 08:08:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:35:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:16:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 15:41:16.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-07-25 16:29:04.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-28 01:43:37.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-02 09:51:19.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2023-08-03 07:41:55.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-08-08 08:08:38.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-08-12 02:35:31.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"96116810926686c9415b181e614bfcbe91b6d9f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dcfe6fba0ecd3525074850699a4d3b169fbb7da6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:35:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ddaa28ad4fc457b9fba88d3de3b5b87\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f34c1015eaba4a43a44cbf2e454761b3 : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8b91b8d23c64d42a666f099971ab5cb : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bbcd8435bb6d4810a23d18b90af02735 : SUCCESS in 6m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61e04fd69dc7b0b58c11f7a7ac3a2d8082ec86ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:16:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44cf7df3db2b4993abeae9f5455bc39b\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab067ca6dbf34ba4b2f1ff767bcaa1a1 : SUCCESS in 1h 19m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d46fa3ce9888429e9dd761c4d261fde0 : TIMED_OUT in 2h 04m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df280e9d45e54246bb55bce78bbea3d2 : SUCCESS in 53m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c0bab0258de433ba066bfd1d207b204 : FAILURE in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0509c2fb90a4c9bbae4af0ea3763c21 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeeb0b318eb24922941f8baf108b309a : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ef50cc34caeb4796898998078fd0f4c5 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e630e3607f64148a55df17b4c782624 : SUCCESS in 6m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ed1dfd1fac894dad8c4658008361111d : SUCCESS in 14m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a26a327e694484f833035b353fb28c5 : SUCCESS in 10m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/00d1c1a9b1b64f77a7cd85e10a059c9d : SUCCESS in 10m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/654772df32a347ecb65d28f693abf2b6 : SUCCESS in 8m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2b0c0ef7e69d45f88aceeb0aebfc8072 : TIMED_OUT in 2h 04m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/070a7c98786541ec824ceca8ee116de9 : SUCCESS in 2h 01m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c1d6dde3df4645a09ee0b1106fcd5445 : TIMED_OUT in 2h 04m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/adccfe3329aa4b4d89841e8b2d5d8436 : TIMED_OUT in 2h 04m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9b4afff71314f3bb0ee726e0942b7b2 : SUCCESS in 1h 17m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da18465347304b30984915ea7e07202a : SUCCESS in 58m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c964e160d41d45f3b584d1cff10f3877 : TIMED_OUT in 2h 33m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/31c58d221b8a46f0a6aae93f17ad7d3c : SUCCESS in 51m 27s\n\nWarning:\n  Comments left for invalid file glance/tests/unit/api/test_common.py\n  Comments left for invalid file glance/tests/unit/fixtures.py","accounts_in_message":[],"_revision_number":1},{"id":"10ebe416c077ffaa183539ae58a0fe322d09eef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 17:03:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d3d4c22fa08e1c10d241329788ac2d02c7afea50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:18:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f243e8c960948fcb07fedb650f0b1e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbb483986b2f4f0baece66c304a8efeb : SUCCESS in 13m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f1baed66698d45948b0bb243540c66d5 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/75207c01650744c1a3a894b931f459f0 : SUCCESS in 12m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7addc57a45ca27840fff58c888c51159343e680c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 19:06:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/628e75aaff1143e780c9bb3546947105\n\n- grenade https://zuul.opendev.org/t/openstack/build/a015d4659c644480aa86468772a8f351 : SUCCESS in 53m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1feec4dd4d234f348fdb8cf2a69403ba : FAILURE in 1h 07m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0f89f9a2b69444b8c37e11e78a83047 : SUCCESS in 40m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c97243cab6e9440782ec7e1cca91766d : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9aca63e746bf4aa389a03560e8bf045d : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33246c40d3e84cf4bf09ab7e0862dc50 : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58269214e01246f1b00b1b22ebf468ba : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec99d90d4994750adf800d015b91aa1 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1e5cf0fec79c4f70ab49fa24eb7e3d56 : SUCCESS in 14m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0ff3cee2466443c9c50290ba9c97bc0 : SUCCESS in 9m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2dcec1ff0a7498b972e04971c7b575d : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/74513958867f494595c1b49309e2834d : SUCCESS in 9m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c86e7e5f57cc463c806f50489c51203d : SUCCESS in 1h 47m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9a7feacf1ca4647abe26708b12eda38 : SUCCESS in 1h 29m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9dc625d27f04ea1b21c8a8d7f739c5b : SUCCESS in 1h 51m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f74a4ace4129497c835dafef7d790614 : SUCCESS in 1h 55m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c307e73222d34ffb9fdc8b1b9e469e47 : SUCCESS in 1h 48m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6444b6f62524622889685db9517707d : SUCCESS in 1h 04m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/25cb4973066441d6b3a26e60f6ea7c0a : SUCCESS in 1h 28m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a14e4f4b87484466b2383970c9fe1942 : SUCCESS in 39m 50s","accounts_in_message":[],"_revision_number":2},{"id":"44ed65013b4e62ffcee0faa8a9707d6bdba4b4c5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-21 15:41:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"39b686d307c81cf879399be7b85cc7e05e7d27bb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 13:52:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c02a0b5f668ac5edab11c51cb9eede1bef0c607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:55:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dc74d4bb765318064885eaa01f674a3994692a3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:13:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0d1429b69a4faf9a24336a3fe7df32\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73a2138cd772442da01a43bc21794900 : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0cfb75892a734d0d94b3240dad784d80 : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/393fcb02f85348c99e35271ad64aad71 : SUCCESS in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0decec189bf7d39e863321b03635e9ae8ea795a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:36:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c86ba5b9884860bd5d835513144ea0\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb8925355bd744f195716d1e2c207bc6 : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8351ce6136024a3c8142979122a5b1d4 : SUCCESS in 1h 43m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c1c6beb1e944d21a7797addf0f5a0e4 : SUCCESS in 1h 33m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795942cbb40149e49f9040d365dbbc53 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2c236fe74954047ab6f0b140b0f9883 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d4556a45f144d3694d1622d73dbb976 : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/25b2de8fb5cb40a2a04326df6692da2f : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c9b085e5d6463fb47c15a0240cd50a : SUCCESS in 7m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d4b54fcdb0754215b9ef6ec6ed2869bd : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ad3f858528e4573bc95cfc6c92e05c3 : SUCCESS in 9m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53120b329ec14b60b6801fe927cac155 : SUCCESS in 9m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c244134dc1844a10bc6798d8cc60aabf : SUCCESS in 9m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0045aaeee527403999a3b455aff44f36 : TIMED_OUT in 2h 04m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5bbbc93bbc7f4221b49021aed0c4c808 : SUCCESS in 1h 34m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cf9b1355cd5d496a8bea7ce1abb12a58 : SUCCESS in 1h 30m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f360052f87c1492e9b57cc551c0451f6 : SUCCESS in 1h 30m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bf45847b8151411293539d99c8bc12a9 : SUCCESS in 1h 36m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c14e97e3a9294995ac83cd7553efe07c : SUCCESS in 52m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/70d80b850d3a40fba717eeab1b462a42 : SUCCESS in 1h 52m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00d8006d997a45a0848970de9cde5200 : SUCCESS in 1h 11m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a47eb08bf882125f60039d5b004774f06dd36580","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-25 16:29:04.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb671cd6ed1eb850e5dfefe85ae2c5b09b2861d6","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:43:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"4f0f26144c6a9a67615abfa65f09bd4a04e0d1f0","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2023-08-02 09:51:19.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5bdc6a39392889fab50586fabf9657beee03dcb1","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-03 07:41:55.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d24914c9a832df00fb90f1b962d8e0d95b703218","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:08:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f8d19396404a12cf5894e786abc0a28e796bb0a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:12:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ea19230c6147295d284951366636ade11be2e3c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:19:19.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889046, https://review.opendev.org/889045, https://review.opendev.org/889044 are needed.","accounts_in_message":[],"_revision_number":3},{"id":"082c925983af2d51b9d8f100f7a4cc489572703e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 05:56:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d3bb19644cfa03bd1adae9549ac7cea859faee56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:03:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86baf02e9e8b42349055ddc9953027a0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72d6603e153949b59c0503805e1fd33d : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ad67a65f42a4edf84ac9db6871ec42a : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4dc15905e91844809f134f7f8fdf942e : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"596896c01e56a03a4e6d48739f67526408b87838","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:07:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/10438209c79a4aaca19d0c37012f8ffb\n\n- grenade https://zuul.opendev.org/t/openstack/build/59befb54e86f40dc86190f83f9f181d4 : SUCCESS in 1h 10m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a05e16dfaa0e46f3ad2dd9aa7f4dafbc : SUCCESS in 1h 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af7a812bcd4544f6aac4b1bc9898f882 : SUCCESS in 51m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b158d255f2241d79cf0663e24df33ad : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8c4a37579d0482097cc6b0b31a1fd7e : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f324c3544024383a644ba81db27f0ae : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/782b992d791f4821aaade03599648405 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc82c0cdf6a433481cfef6f44389746 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/01a1bfd2a49c435abf243b70712aa78c : SUCCESS in 13m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/416bbfb94311482c989c491e2d0dc98f : SUCCESS in 9m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9816a5ae0754e3093adb3eb9ec4831a : SUCCESS in 9m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8cd804a3cdd640e4a92d8c0359e489d7 : SUCCESS in 9m 46s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9080de5d2ef949d38f525d28d6f7c16c : FAILURE in 1h 32m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1312106c75c1499ca3a3e22cbcc16044 : FAILURE in 1h 28m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87d331fdd9fa46dfa45be71d034350ce : TIMED_OUT in 2h 07m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d9741d017e5419da47a9446506068c8 : FAILURE in 1h 30m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc7d6b09b11b41bd982b466e9901b3c1 : SUCCESS in 1h 27m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dc1d86c01744ca48fc0fb136f95a58b : SUCCESS in 1h 20m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a8ce2f851adb4a78b1a46e6ddf47437c : FAILURE in 1h 29m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cabc6bdd3a734e33a543597fc0bd57e8 : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":3},{"id":"9a3f9cd9c382876b2520a7bd78e535a5900cfb8d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 08:08:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69ea8a25465f4ba5d1d5b19691ceda4e692e79ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 08:28:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae5f1db062849a4b7a26adee6a35456\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbdccc6800304f63bbca9b8325a0f760 : SUCCESS in 15m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3bb2b0f9cf74fa3aceaca941acdad6f : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1bf15b01482449f59bdfc72ce21cec4e : SUCCESS in 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6bd4503f70e76bb4c2c02f2392d5adcc968d49b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 10:03:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c89a506ff1bb4abdb4b2461e48aceefa\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1cc7926debd4634a5d8d81e2a803eeb : SUCCESS in 46m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17ca3e4f89584bed99fdf94135bcceb3 : SUCCESS in 1h 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2e4a9e045a544d8a5741a8e287a86f0 : SUCCESS in 42m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1599c4322d149978f07dc19ecbaca26 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0206569e54c8451f8b0c6aa21b63e33a : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aa2af8db08a4456b3eccff832e6a7b0 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1191df2c855748bfb9734df4f1e613e5 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b25f7ca63ac4c9e832b820bf3bded76 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6fe1b1e35c3146639e852446fff318be : SUCCESS in 12m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4e79f3ca77644e47800dd56bd9c5a830 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9a9b24fb371d41219fd17548fb857f09 : SUCCESS in 7m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/399e00cb2a8344faa695e1a61a5431a0 : SUCCESS in 9m 36s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a49491669fe4407da3c3f05366fc8441 : SUCCESS in 1h 13m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3a26298cf5c84eefb8efaa5a0c8e91d6 : FAILURE in 1h 21m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/760ba26b2ee444018dc3e9fd9011ba50 : SUCCESS in 1h 30m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9396962ccdf1456988351e5f24cf910b : SUCCESS in 1h 30m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/829c851678cc4be1a84b32a6621d59e4 : FAILURE in 1h 44m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ae1f9e7d7bb4090953afc61ca1627ea : SUCCESS in 45m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4f7d56b3a993407c92cb7317ecc983e2 : SUCCESS in 1h 33m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f8a5e3c4c592457da7c88f3f0b6342f2 : SUCCESS in 22m 00s","accounts_in_message":[],"_revision_number":3},{"id":"10a5c7d22b4bc767d5f38bd3bb265a745cfd3c10","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 10:54:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"de9829fe9bb00ecf35f8ad3520e4be385be718dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:07:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/041d3807d7504390add90b63f6d764bd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e99b8bd42dd3459cb89f70cd4b8a9a4d : SUCCESS in 10m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ef7a7d1ceff48cfb52241ccbee7a82f : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fcca9596f80041aebefa419782953c2c : SUCCESS in 11m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"053f1095e093aa9160659cc6782cc611a0a8c3ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:08:45.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc7798bbd075484bbb29fa1f257fc846\n\n- grenade https://zuul.opendev.org/t/openstack/build/67a40ccaa7be45bdb313673a48be2625 : FAILURE in 48m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a19f9428e814d5fbd86af7048c1a85f : SUCCESS in 1h 28m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a36cee0c47ed40f4a58e90ae79577101 : SUCCESS in 58m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80ff3340c07a480da0e3e505c3eead98 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1172fe08668439ea60e7c4c02bdbbfe : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48ae763732b145fb8034d0a2bca7e82a : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f120695a9ac439eaafbff9256071da1 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b9375a7adb40bebd5a9768ce6762dc : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8da8820814e945aa88f02b57bae5899a : SUCCESS in 14m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/faf0cfe28a9b4dbf8f34beb88c5714af : SUCCESS in 11m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a889785a4be34e5e80778ffd20b21091 : SUCCESS in 8m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6ac0af4d2d2e46c382116dbf28769986 : SUCCESS in 8m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe1542308e154f5f9e0ccc225eac50da : TIMED_OUT in 2h 05m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b7dde196747e43e49e1f2628c66e9419 : SUCCESS in 1h 31m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c1fa43a602a74a279f85b59c513b0f5e : SUCCESS in 1h 46m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/10e89d48659940a4861c1c51769f6b10 : SUCCESS in 1h 20m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9622a14cc496410988c218cfc26fc82b : SUCCESS in 1h 59m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dc6e966c321403bb2b88a38b1ec7c39 : SUCCESS in 1h 03m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/de2c879174504090bdac16ef928cebc8 : SUCCESS in 2h 06m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/99c9e68184fd4a8ea4f7069354fe462a : SUCCESS in 32m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c41fc03a60abe50961fa2b6f7aed2e2fa2668a8c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-09 15:58:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a8ab03218e3a16c1a2a0a9bbad4ec3d56410d449","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 16:12:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4053d0fb67b0490b9ec7816a401d419d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a28c11564764c8da68328b10db6f2eb : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e3291b90fe648aa9ce56cb5b07e20f3 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c908b11fe7d4c1fb0d5a9d8dcb486d4 : SUCCESS in 11m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dc2766c1a1b9fc2d5fae27f6cf345a6425a176d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 18:21:31.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/92727d7940594489b57ba1110331fc65\n\n- grenade https://zuul.opendev.org/t/openstack/build/9599abffad9441238368428babe6f1e6 : SUCCESS in 1h 03m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/affbfc42f9974e908a38272c065affa0 : SUCCESS in 1h 10m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fadb2336062475c897f5dbeb1d06d9d : SUCCESS in 43m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc8343c9e274bd884d0b9686b986cf6 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/056c043f02384762b3bcb1ba21eeff02 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8249a24e20a459fb7b9b39354878551 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c3f4635ac244f929864f86395a4c431 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8b466a4aea24474b4885fc7fc63ad8a : SUCCESS in 6m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4346eaa37fab4ba59e13a36811e7303d : SUCCESS in 11m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8758abaeac3d4ae19b247b5d0bb86958 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8ead639390c6462eac94b736c0988a5c : SUCCESS in 8m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dee219536f99456dbed1882584901b38 : SUCCESS in 9m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6359ccee5f27422c92e9b0eeac0b2d54 : FAILURE in 1h 58m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5194d06e16414db3a6d9115619ac72ba : FAILURE in 1h 07m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/af6a2a4f035c4973b425543a9dac7403 : SUCCESS in 1h 20m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/90c831dbc3ee47949a6af4ab12d5c3a3 : SUCCESS in 1h 18m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cf32b4438aac40dc84ccda94be193cd1 : FAILURE in 1h 30m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71099593053a40d79e7cb449284a37e6 : SUCCESS in 1h 10m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/710f3fc510f24cb39ef79a0832a35190 : SUCCESS in 2h 15m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/185cf8d070b449ccb591bbc57d4137d2 : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":3},{"id":"3aad4936668bed6201573f53d25951d53557909e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 09:17:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"30e8d3e7519dba62d583c772ce50c9e850c42687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 09:31:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd4da4fcb7e24dde89488799e55301fe\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2acad23f042e43b98a5b0407c99fc0c5 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9370de30724d4ffa91697a49e3b29cfc : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/53c7fdb484a04fe3b83e7ccf89cf052b : SUCCESS in 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1b8508be528d5e9f937098f00dc6a66daa6851b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 11:20:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/345da1dacb734670893abd2c0e310fb5\n\n- grenade https://zuul.opendev.org/t/openstack/build/21734d16d751477aa4514ac4916a7bde : SUCCESS in 1h 01m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbd9380c791441beb11d24d552a9f7dd : SUCCESS in 1h 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5acbe85022974fbda85e004c1cb93da0 : SUCCESS in 1h 03m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b193771450204f6297db7cba211abc6d : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfd2d2ddffff4d35b927f70c5dab5273 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4ddfb642da0460ba5de1a84d4c011e1 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdc2f11c2afb4692ab443a8633ce044a : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e41431715294cc49349304a104e1cba : SUCCESS in 6m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/770669ca82f84e1aaf8b1a229f034ee3 : SUCCESS in 12m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d9b18b34bddb417c9a18d2deb1962c90 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea5a2ab4a0204c20ba2ba2481d15c35d : SUCCESS in 9m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d247c21b699940fe814e4b9f0352c597 : SUCCESS in 8m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe620afc5f194a89a6166a126564a84e : SUCCESS in 1h 54m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eefc2b8c38ee466cb59c9ccc926e4858 : SUCCESS in 1h 17m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/822f528239224d039547e5954e82b093 : FAILURE in 1h 16m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/30235c49f18c47409da3afeedeea44fe : FAILURE in 1h 12m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/811a23812ad94f6686ff65ada3e16632 : FAILURE in 1h 25m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49d45e64fc94457daeefa146616ead29 : SUCCESS in 59m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0303c7b74544a368cd72eda5b61f510 : SUCCESS in 1h 47m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7ec9708f511b4c35b6457f9099de737d : SUCCESS in 29m 58s","accounts_in_message":[],"_revision_number":3},{"id":"6fcfa6957e654d51c63c55979953a11ca5da20f6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 16:42:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"184a74cdb5e50b6b11af9357eb2879664080484c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 16:57:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/695e6301cfb6402b852a1fec17e477b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea4d3affae744b2287013e0225d34aa7 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9edc208811344e99102eadfd8de6acb : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f63bc110ea5b49838ce20bf1785fc282 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92fe95a479a348a767764b8f28299ba4c23abe49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 18:55:12.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ee6fd353b5a4d1b8949eb209c0413ee\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ea5012a55584baaa81b7dcbc31a1743 : SUCCESS in 55m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f5667f8adb8482a89052df7c81834c4 : FAILURE in 1h 10m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/634ab49b49924d2ead8ffad8a005844b : SUCCESS in 1h 01m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8675b81b8f7a4186b6075fc80650997f : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ead62732328f4a17ab454e05e3c974b8 : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d721b9e768a1477ca4d5c13736b75f52 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b44570066884594b4b3e0667c23888c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/447eb9da5f00480ca0be80cc83f555c9 : SUCCESS in 6m 38s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cc1447d6c8ab40ac8249541028d34554 : SUCCESS in 13m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae311e4e3c3843ac9eff8068888df4f0 : SUCCESS in 9m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ee7d3d7cdcf4fab9c84da06617300e6 : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4da4ce7564d3499d86299cdeb3822bbe : SUCCESS in 8m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57b4775bc1de4e15afd0aef1f4236c2e : TIMED_OUT in 2h 03m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c6ec36253be14af3bfdbbfde5e3fd797 : SUCCESS in 1h 22m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41b826e41e1a4ec1b1fa943b6aa9ad55 : SUCCESS in 1h 23m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b04b13319334295a20e100e85018ed1 : SUCCESS in 1h 27m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8c1cd33f9a4d458cafffefd0b3948bc6 : SUCCESS in 1h 41m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b255aaaece44cbdafa6cea77d986e4d : SUCCESS in 51m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0a5ba91b3ad4f3382d329c40f122bc6 : SUCCESS in 1h 55m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9b1d034705c49e7b59bb4331d7987c1 : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":3},{"id":"dd0578ce01cd5de79e7812323826e5398cead116","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-11 13:11:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e00ee4d78a4f5621a5d68c69306e18dc525123e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 13:23:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56da6a78e8a743bdb937de8d310a509a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fbcddec55cba42c29717aef439339e58 : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a15e74b7189045e0b35df0aef6ac9fbb : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f431350bee4464eabb2d41e2cbe09b1 : SUCCESS in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c0bdfc80a6c65038fa3608dc9e73e219cabfddfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-11 15:25:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c9c0977eb945bdb43bd2c22eb57abc\n\n- grenade https://zuul.opendev.org/t/openstack/build/d668ed12fbcf49e0aa61e9d2dfba81e1 : SUCCESS in 59m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f2d054b40bf48eab5722def6a057b5d : SUCCESS in 1h 26m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94db3ed2c0f145f390241687e6a661ee : SUCCESS in 1h 02m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21fb95ca99934896b4281da370f46e43 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49afb20c1f2f400bb5520aca76f61355 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3d70d5570f84a30a5b9eb158096c6da : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c2bd292f1a44032b2d819f49e4ddc80 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b7263dab6741a69a6b62944179fdd1 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cd9d0090bc7640e7817173f84ec52906 : SUCCESS in 10m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/556c4f1872cd4ca9b38948ba95f2ff59 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f34533ed65e40a99a2b622bf904a89d : SUCCESS in 7m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8cc9b135efef424aa411fb3906b4907a : SUCCESS in 7m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/66d0c840805b45208b790cc060678d71 : SUCCESS in 1h 36m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9a5e1588e5fa43839801a1e9b0582889 : SUCCESS in 1h 23m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fc7e551401ba4e62aae83a7585cc8013 : SUCCESS in 1h 33m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/762ea4d1d5604ecfb2ee23a08af07a9e : SUCCESS in 1h 53m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20f0cf80f8de4e09899a917f48cb2215 : SUCCESS in 1h 52m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f39a998adf174a5eaaee6999aa457d92 : SUCCESS in 43m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/768a5e7fda7b44089a14bd3b845b73bd : SUCCESS in 2h 11m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f67b87073f24b32880d0d3a54c232c6 : SUCCESS in 27m 32s","accounts_in_message":[],"_revision_number":3},{"id":"8b072881c9e4a69ef0c6ff108f193beb8e7cd55f","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2023-08-12 02:35:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"1750023058b2240c241b1cd371a25c0c9cbce55b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:12:00.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9bdbebc16709a7f2629c54334c39b22876706df5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:12:47.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/906237b868624294bc4ce9e619602238\n\n- grenade https://zuul.opendev.org/t/openstack/build/98e83b48e9554317af5511fe4c8139fb : SUCCESS in 1h 20m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df892be5231b498993581ba8e422071b : SUCCESS in 1h 39m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c14c6bdcdde04ac6893b4bc5235e81c9 : SUCCESS in 49m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3dbc60b306e45a89244d83e3bb4989b : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35790dfe06f449289bff86f276106f12 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/916b7a3690614ec0863c2c85b618c708 : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0f97517494240f4a61d7be0a4b99e15 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c321e4b7e7472baf51295fb5142d35 : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e97f5b0e77674b0db7938c04126bedf7 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ddc0aec7ae14c8c92ca859a5d4c1e17 : SUCCESS in 8m 11s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/18da37dd904d4d57a94abec219dfb606 : SUCCESS in 1h 34m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a8802d1ce3f7446ba04eefe8a426a609 : SUCCESS in 1h 54m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9c6fce4c70fc44f5b12570212e95ca34 : SUCCESS in 1h 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff0a604e449742d48ca450d7ea5a1363 : SUCCESS in 51m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56f6ce1dbe9d4864b259d46adeecbe2e : SUCCESS in 1h 48m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4c7bc052dc16455eb90e4c9109ce5f9f : SUCCESS in 40m 02s","accounts_in_message":[],"_revision_number":3},{"id":"bbe89f147687fcb9359c6c3d589161b8b7600028","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:12:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"033736ccb3df4e9e935e4004d6580eaf0bb5ec70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:14:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8116c205bcc452fb248ec8b0c9074f6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3063fe233f6e4e66a3f4cdebddfc7ac7 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1557eab3687713f05a7711645c68c04e13512847","revisions":{"bfcda6536bc3eb94ccdd10fccfac8b8f5ba35a23":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/889047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/889047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/889047/1"}}},"commit":{"parents":[{"commit":"7caad1cec0444a8389bbc39a42baeb8129118de2","subject":"db: Update import location of declarative_base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7caad1cec0444a8389bbc39a42baeb8129118de2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 16:44:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Stop aliases import","message":"db: Stop aliases import\n\nThis will make our lives easier later on.\n\nChange-Id: I6b2cbf0b97bb3af7354f2c10a8d83909db963d43\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bfcda6536bc3eb94ccdd10fccfac8b8f5ba35a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bfcda6536bc3eb94ccdd10fccfac8b8f5ba35a23"}]},"branch":"refs/heads/master"},"d14d527d59e3f9998fd8696af94daff014938126":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2023-07-20 17:03:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/889047/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/889047/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/889047/2"}}},"commit":{"parents":[{"commit":"c4b4679649e81ff86a9fb270a0ed3b0b3286ab86","subject":"db: Update import location of declarative_base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c4b4679649e81ff86a9fb270a0ed3b0b3286ab86"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 16:44:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 16:28:30.000000000","tz":60},"subject":"db: Stop aliasing import","message":"db: Stop aliasing import\n\nThis will make our lives easier later on.\n\nChange-Id: I6b2cbf0b97bb3af7354f2c10a8d83909db963d43\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d14d527d59e3f9998fd8696af94daff014938126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d14d527d59e3f9998fd8696af94daff014938126"}]},"branch":"refs/heads/master"},"1557eab3687713f05a7711645c68c04e13512847":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2023-07-24 16:55:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/47/889047/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/47/889047/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/47/889047/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/47/889047/3"}}},"commit":{"parents":[{"commit":"5636008dfe30beb7311f7f65b500cb852cd394e4","subject":"db: Update import location of declarative_base","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5636008dfe30beb7311f7f65b500cb852cd394e4"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 16:44:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 16:14:22.000000000","tz":60},"subject":"db: Avoid import variable shadowing","message":"db: Avoid import variable shadowing\n\n\u0027session\u0027 is a widely used variable name in the glance.db.sqlalchemy.api\nmodule and this is only going to increase over time. This shadowing\nprevents easy identification of e.g. missing function parameters. Alias\nthe import and update the one call to avoid the need to rename all other\ninstances of the variable name.\n\nChange-Id: I6b2cbf0b97bb3af7354f2c10a8d83909db963d43\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1557eab3687713f05a7711645c68c04e13512847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1557eab3687713f05a7711645c68c04e13512847"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
