)]}'
{"id":"openstack%2Fglance~889048","triplet_id":"openstack%2Fglance~master~I7a59714b104659b64f46f4c1437cfc4018356815","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-28 18:12:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7a59714b104659b64f46f4c1437cfc4018356815","subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-08-28 18:16:46.000000000","submitted":"2023-08-28 18:12:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889048-sqlalchemy-20","meta_rev_id":"956bd46f2ba43a0d46ebd7f80fb75750a6602e4b","_number":889048,"virtual_id_number":889048,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-28 18:12:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2023-08-03 07:08:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2023-08-02 06:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2023-08-08 08:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-07-28 01:43:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-07-25 16:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-02 06:01:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2023-08-08 08:10:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:22:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 13:01:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 16:32:35.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-28 01:43:33.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-02 06:00:07.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2023-08-02 06:01:19.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2023-08-03 07:08:42.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-08-08 08:10:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"e750619e76d925a3e7e5728939531b4b27a83082","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7143e3680e3acfd0d47208ece102eaac00604eea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:22:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9441679f1a94a87be1e5373eeddefc4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc7558ee7821473f8ccfb20525944cfc : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff0314bdea9445eba5cc8596290ca60b : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9a349c4bd2b7416c85b29f14df535c28 : SUCCESS in 10m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"311ca7c76082ab73b7978a99e745a977cef8e782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 13:01:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7aa8afd62cf4ac28a5a856bd9cf86b0\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f0a55a20c7a14525b20436296e8c7e47 : SUCCESS in 5m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aca6675e722a49fd9a4e57e95bdecb32 : SUCCESS in 1h 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32688dea94664a65827d8ac249398b9a : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dda6e4dcb52a41529c6394ff94604d3d : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c89d56d654846b082bb407c3839497e : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b640b09896e24547bd927fb842b7a25f : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be99772d7ed4499b9979bb3979b70ee : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52c87b217bd74436857b8a2278bf4ff7 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5bacc8dd9e7b4c2abf2325a7abaea07c : SUCCESS in 12m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a09c3d0ed6f04e0c9c0fc0a07d1eaa0b : SUCCESS in 9m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ac4d483c67a4a559b5f1e3fe3e38ab1 : SUCCESS in 8m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/34d9694147074f369ded55ca97631bc8 : SUCCESS in 10m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f17b66132c2d42ac86a9d073bd759344 : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e908497d3e9f3d7fae6f9d97b37753c47f8052c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 17:03:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5a367419273783335d20524dd3dda3925d3feba2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:18:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d51629d85d04c87ade593264ecf4246\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a5f12c0d8c5417fb18b153db04b8495 : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b621f75a9ba7402193ecf22c0b5f55ca : SUCCESS in 12m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3012c402721d4cb9b2d1b965a5692743 : SUCCESS in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"880d07b14d13cf017b8be118023dca41ca3b1416","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:57:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff2b4270bd324066945dcdae908fac58\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f0f3ee2e49a04c4eb132dbd28b6ca975 : SUCCESS in 4m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/585b8f2ac46a46398feb42dbc6d43719 : SUCCESS in 46m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29d20d7f03b04a0a8d681642a0d63d4d : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0588858cb1d45979168b26a8723e5fc : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaac88dde90e442caa42f15ed0d98b5c : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e94af4be69b34b5591f964f14b6be86e : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f836f87e03f645acb74ca7e38640b59d : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39844172fca94a13b6898398dbc20a33 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9a27703f4829460a9acd8a9fca7b1d43 : SUCCESS in 13m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/611beaaa26df40cea0dcd7a43348c643 : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0df3193b959f4efab08cb3c19b063765 : SUCCESS in 7m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4ce8ae8faf28447cb517711fe25d61f1 : SUCCESS in 8m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e5c9e84961294e77b3303d41df3f6f29 : SUCCESS in 24m 26s","accounts_in_message":[],"_revision_number":2},{"id":"6ebad4130a87af5370216ef89fbac6cb01ba3f9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:55:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e76acce08a1f0d16af41775e6e3599038f1378d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:09:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/019021836c02443ba19998d32fda18ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4c4cf9fa481468aa62a858c60884fb3 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74838c1bf422429ca34364e9508c3d82 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f9c062c4e974c92b2abe06078a221aa : SUCCESS in 7m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"adebedce0f730923decf1019a00cc165c4b5518c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 18:05:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9361b40ab084409a9074f7e51d9c90ab\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ec9cebfc394b4541a6a31ab3395d0fc5 : SUCCESS in 4m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42149be45991479f9264d4ecd38b4001 : SUCCESS in 1h 04m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7744e04f07407dba6ecb1bf422daf7 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6d1710becd340b6b2461de1dfd41bf5 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e5b7cadb3684aab85eba9200cfbefd4 : SUCCESS in 8m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26b474735c184919b5c70bc001c1cff4 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d85f8b682cbd400db24d11727c7d7251 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e66e4d1773db481aaaff2f8eb7a7684a : SUCCESS in 7m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e99ce34f09174490bc90df80a4b3a6ed : SUCCESS in 12m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b1cbb30d6e041acac778c5bb6176c28 : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2aeb2d16c2e4432c8828b23286d3a0d5 : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9f082144a1064503b10764b34f58603f : SUCCESS in 10m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/abd334479ef04962bd0f09280effe1fc : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":3},{"id":"26b1801fd6a90d45439b9f206120be923a9e0913","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-25 16:32:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e1d1eed94c2048001a12f930b41c52f11b70e84f","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:43:33.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"abadd71c15426df0d434e89df7a5f5ca003d4103","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2023-08-02 06:00:07.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"87cc26796ff1ca36b937fa6d55a9bde149a31d3f","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-02 06:01:19.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bcb140f8ae7f8686e5bff4b6bb07ca4037f6f7d7","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-03 07:08:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"054300914723c67e971e28c80ba5575d1c4c4714","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:10:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ec0f3eb146c8339e3af997a98755addf7ce3ea74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:12:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de0cace9c8ab419952b62e264d01b64cb3096577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:19:19.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889047, https://review.opendev.org/889046, https://review.opendev.org/889045, https://review.opendev.org/889044 are needed.","accounts_in_message":[],"_revision_number":3},{"id":"33c0651e414f4f0273363a5106ba4980f58e236a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 05:56:40.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"22c2b89a0bc22b278c5cd9bed7190c5fe954845f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:09:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59fb47071d1447f0bc0ecedc58e22309\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f15a915ae6944208959c2f004186bb53 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/98e9104b6b824b4d877a442fcb509298 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/417fbc3e82b74782815ef645b3b39845 : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d2d624da59089efd377303ae538244071f5d1537","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:52:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65dafd2fb9b346d1b7863ce6008296c6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/11829841c477422ba22dd9e4d5e15ad1 : SUCCESS in 5m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e62a1acdb047423cb19e42f0fab77e19 : SUCCESS in 54m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c89e46e7043463581791a59dc45ea25 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e68130830bab4e639a6c7261fdc43b29 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03c754f25bd2427b983c89296f99729d : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8920fbc22a24c1c8001cf8559583d7e : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa53d0c470ad4cd7a38efa1fd137b032 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c88bfb30637e4bb78bab894f02ba77b4 : SUCCESS in 6m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d934b82e581d40d29a29e30daf519df3 : SUCCESS in 15m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc0fe9595c29490ca8940cd689e616a9 : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ddfb80278f694613b09563e7b02bf8f6 : SUCCESS in 6m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6b58ddfc950e4605b71e2cabc95adc84 : SUCCESS in 7m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4eb1ed1e37c143adabafc08f25103244 : SUCCESS in 23m 52s","accounts_in_message":[],"_revision_number":3},{"id":"14d6b12991c47c5dc0fc01ea3c05b071374937b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 16:11:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b52fe041f013eed5dbd7f76f08dffc22ad9c6d58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:12:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ca22c20eb9e45cba24fb3aaa7717512\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4146421997ad4f189e0522c3ce6da40f : SUCCESS in 5m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60d93d44e778462f82f358c03d44ed25 : SUCCESS in 1h 08m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a599eda8f1df45f08a4ffc746c556e3a : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1292c8900e448cfabe07677ebc93a09 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6246254320b54039940332abe4c69e56 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e193647cdf44957abe5a7b01e5a6a47 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6b7f01f14947618fdbcb7bd227aa74 : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a903de8cf9944f2c976e6b1f26909714 : SUCCESS in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9185372bbc474d4894027a06f36f275d : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad153f2a527641e791adffb08c4bc936 : SUCCESS in 9m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0bd6413132ce441384a95c6b00e6d2a0 : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":3},{"id":"80695bd983d02c363030c083a4274f2bb599445d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:12:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"956bd46f2ba43a0d46ebd7f80fb75750a6602e4b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 18:16:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82bf90c49e0c4c0ba298b32344b59916\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/5f6d5295b3d74e828fd7001f7a3df787 : SUCCESS in 1m 00s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9db6577e0494e35b0ca97eed060a923 : SUCCESS in 2m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30d6024c22624752a09ef318bccb8ddb : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d409b27a793210e975b07243ac6795fc939e7c81","revisions":{"da2b6216889a3b6df8d008babcdccd75ff46c1f9":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/889048/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/889048/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/889048/1"}}},"commit":{"parents":[{"commit":"bfcda6536bc3eb94ccdd10fccfac8b8f5ba35a23","subject":"db: Stop aliases import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bfcda6536bc3eb94ccdd10fccfac8b8f5ba35a23"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:57:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","message":"tests: Enable SQLAlchemy 2.0 deprecation warnings\n\nWell, sort of. We enable them but immediately filter out the ones we\u0027re\nactually seeing, the rationale being that we can address these in a\npiecemeal fashion without the risk of introducing new issues.\n\nThere\u0027s a lot more to be done here. However, the work done in oslo.db\n[1] and other projects should provide a guide for how to resolve the\noutstanding issues.\n\n[1] https://review.opendev.org/q/topic:%2522sqlalchemy-20%2522+project:openstack/oslo.db\n\nChange-Id: I7a59714b104659b64f46f4c1437cfc4018356815\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da2b6216889a3b6df8d008babcdccd75ff46c1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da2b6216889a3b6df8d008babcdccd75ff46c1f9"}]},"branch":"refs/heads/master"},"672e1f2ca061e23dd5ea88a4c1de364bb4712a29":{"kind":"REWORK","_number":2,"created":"2023-07-20 17:03:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/889048/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/889048/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/889048/2"}}},"commit":{"parents":[{"commit":"d14d527d59e3f9998fd8696af94daff014938126","subject":"db: Stop aliasing import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d14d527d59e3f9998fd8696af94daff014938126"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:57:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 16:30:05.000000000","tz":60},"subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","message":"tests: Enable SQLAlchemy 2.0 deprecation warnings\n\nWell, sort of. We enable them but immediately filter out the ones we\u0027re\nactually seeing, the rationale being that we can address these in a\npiecemeal fashion without the risk of introducing new issues.\n\nThere\u0027s a lot more to be done here. However, the work done in oslo.db\n[1] and other projects should provide a guide for how to resolve the\noutstanding issues.\n\n[1] https://review.opendev.org/q/topic:%2522sqlalchemy-20%2522+project:openstack/oslo.db\n\nChange-Id: I7a59714b104659b64f46f4c1437cfc4018356815\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/672e1f2ca061e23dd5ea88a4c1de364bb4712a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/672e1f2ca061e23dd5ea88a4c1de364bb4712a29"}]},"branch":"refs/heads/master"},"d409b27a793210e975b07243ac6795fc939e7c81":{"kind":"REWORK","_number":3,"created":"2023-07-24 16:55:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/48/889048/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/48/889048/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/48/889048/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/48/889048/3"}}},"commit":{"parents":[{"commit":"1557eab3687713f05a7711645c68c04e13512847","subject":"db: Avoid import variable shadowing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1557eab3687713f05a7711645c68c04e13512847"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 15:57:36.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 16:29:01.000000000","tz":60},"subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","message":"tests: Enable SQLAlchemy 2.0 deprecation warnings\n\nWell, sort of. We enable them but immediately filter out the ones we\u0027re\nactually seeing, the rationale being that we can address these in a\npiecemeal fashion without the risk of introducing new issues.\n\nThere\u0027s a lot more to be done here. However, the work done in oslo.db\n[1] and other projects should provide a guide for how to resolve the\noutstanding issues.\n\n[1] https://review.opendev.org/q/topic:%2522sqlalchemy-20%2522+project:openstack/oslo.db\n\nChange-Id: I7a59714b104659b64f46f4c1437cfc4018356815\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d409b27a793210e975b07243ac6795fc939e7c81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d409b27a793210e975b07243ac6795fc939e7c81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
