)]}'
{"id":"openstack%2Fglance~889050","triplet_id":"openstack%2Fglance~master~Ic956980a03e1a0e1b6f90c492d8a03a64ea275cc","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-29 15:31:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ic956980a03e1a0e1b6f90c492d8a03a64ea275cc","subject":"db: Remove use of \u0027bind\u0027 arguments","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-04 08:30:43.000000000","submitted":"2023-09-04 08:28:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889050-sqlalchemy-20","meta_rev_id":"a500d698b44b280d2c5bdad748576d1473c30ce5","_number":889050,"virtual_id_number":889050,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-04 08:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2023-08-08 08:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-07-28 01:43:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-07-25 17:41:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2023-08-08 08:36:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:33:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:13:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 17:41:13.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-28 01:43:24.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-08 08:36:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"edcdafbe13764caa47ffa7c2800df9197aaab04e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52d0b493e21ac0968d2512e1c4f05fb96f359c57","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:33:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33c5c331949c4d3099492da55f5ace0e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1530570604ac478d936f6c69b03ece29 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff2600ccc04842ce9b04d5fc3dc8d319 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f233475266b045e4aabb2a51ff1305a0 : SUCCESS in 11m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b816a1900f843e0c8aa7cf936bf43aa61e666da5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:13:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb504e16f8704a9db7230b4e0b47ff77\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ddb11f918a942cb8c6202dcdfb1e05e : SUCCESS in 54m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/355c80c746e74525be44e7fdcc2b6507 : FAILURE in 1h 07m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3fad7ce4828446195aa7d2dc7cffcc5 : SUCCESS in 1h 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82c7cf9079904de7be74b19d6719b055 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f9adae757d3451fb10949d000d9c416 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c28874412fb4390a4b786a38b8a8bfc : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eac9ba1e09884e68be9194d4bb87cb7d : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5faee80c91214e0fae2d04eaabc3a906 : SUCCESS in 7m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5b277b17903141e8a48389d0cb1265df : SUCCESS in 12m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65136fc55f9e4d7b96f837cfd030c271 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e6fb5ef1f30e47cca569dc9c4171e2d6 : SUCCESS in 10m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/26cb89733e474afdaf80ed7f45f964ab : SUCCESS in 10m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f19349308960452599934138c21e331a : TIMED_OUT in 2h 03m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eecaab00e199486aa5a8877732162438 : FAILURE in 1h 58m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ae299277d0945b6b0eece471fafc57a : FAILURE in 1h 54m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b06654ba000943e5af8cd2a3ea093d02 : FAILURE in 2h 03m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e36bb5209732413fb57c8eac1dd0a169 : TIMED_OUT in 2h 05m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91771ba4a9cf42459605ea41530567ab : SUCCESS in 1h 07m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9fa7ba6cd927428d8245f757f3713201 : FAILURE in 2h 24m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/447d8477c1ba4c5fa2e98cee2c8aed01 : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e7bde7675a36d56fbdf1ee5ef081a334f5d5b5c9","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 14:27:22.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b8d20234187a5c97f89fced261bd24749b9b300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 17:03:14.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d958a9feaf9fce1018f7e6f316a35bc46f5e86a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 17:18:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6c9722f9ce400793a445a73f2e3e8d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdf2908f54264c148eef2d3bcea4fba3 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/87fd7b7bc6e24fb69760b77c48cfd020 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6cfcbf8adb0b452a8b16debb504543aa : SUCCESS in 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"58c6e0a838cf7354315fb005bcfc262755bd3ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 19:15:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d7ac780eb8f8447d9f5b218ab777fe61\n\n- grenade https://zuul.opendev.org/t/openstack/build/f057234df45745bb9d5fe1de1fae0566 : SUCCESS in 1h 01m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b9b199817994f97bddcac8c74e3bc28 : FAILURE in 1h 34m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb8c504de4a447209183013554da7582 : SUCCESS in 1h 01m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1093c50582d74cafaf760a872be24578 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/492cd469839346cdb81a79f21c732513 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b36f624050e45279de05fffa9e413e1 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/05c1cc927c814d41a2fb7a1a5d43ba68 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9516c43c151c4d20a7f1afcd4fb40007 : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/609050fee3f14390a53b2773854a87f9 : SUCCESS in 15m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c3e4580fc17c483bbb863ee83a9a54ea : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/438754da856044eaa038042032dc4262 : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b586852ecd2343c8a1d6cae7973d4868 : SUCCESS in 8m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/15ec125f60c24125853163f73d8ce09c : TIMED_OUT in 2h 03m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7b6c4acda1ba4839b998fe0e63f1bfd6 : FAILURE in 1h 11m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b51c66006b004775a65c7c2c8df0d7c7 : FAILURE in 1h 30m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f4c1862c25094d2a9924963232e90f45 : FAILURE in 1h 18m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/47490bc08d474bc6904aa7028773ea3b : FAILURE in 1h 19m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edc3f6bd3ed943f0b41a286bb1571320 : SUCCESS in 43m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8b75bc39a2ba4d57990516f67deaee38 : FAILURE in 2h 00m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e87e77c59894a5e9e2f80673169a167 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":2},{"id":"81ff3dbc0cbecf8c55621ab6d3dd346cc5f4a5d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:55:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4533a3f1f4565088f7658d57877d9c79de4cb71b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:24:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bd02288ae3c43dc9f71f836caa51f17\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/760957b00a5c436591740eb506b38990 : SUCCESS in 12m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/527b6820c0814453a29a798433fddb87 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95de5ab748984a38bfd03152ecec4bb4 : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f81797c78e0c7b47cd8f1b6ab5b2d5bc6b97a6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:16:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f94e9ecfc7dd48c8b5c1d40959398332\n\n- grenade https://zuul.opendev.org/t/openstack/build/704370aff3b248d3bda6e15a0c481119 : FAILURE in 1h 08m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/170a8d28d26646289d68cd97b436ffc5 : SUCCESS in 1h 59m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce5c821a04a74413b864f8d131178fa1 : SUCCESS in 1h 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d14fdd1679348fe9e3cef40755f6209 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf2feb4afb6a432686fe24e4e3d7eeae : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf961a7223094de6bf547c020496f7e4 : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e056dc5fa452498280ca53754fba26f5 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b09cdeb4b734dadaf31bc16a76cfc5e : SUCCESS in 6m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a665b6daa6ad4388a46263066686c1c2 : SUCCESS in 14m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bd6a24ec0e10497eb4c296d12623a058 : SUCCESS in 11m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98960a9ca47242d4a5a87545868b4703 : SUCCESS in 8m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b3397a2f1e8a41439fcd82dfde580e05 : SUCCESS in 9m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fd1f569d68294c0f9325cac07925a304 : SUCCESS in 1h 53m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf7ab067d7d24b868308bdd72d4c95ac : SUCCESS in 1h 49m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cebb901a92a740519ecac4c6026abbcb : SUCCESS in 1h 41m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9b6552ae77d940bcb4ce4d97e2950321 : FAILURE in 1h 46m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6e6bf0019250492e8b277ad5764b1735 : SUCCESS in 1h 55m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eee58d6963d64789bda2e67bbfbee602 : SUCCESS in 1h 15m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/685dca04fae04a4a9766278a494a11ae : SUCCESS in 2h 11m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d07a220673914af182ca78e105430a40 : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":3},{"id":"efee08b71062db564dd17bdf891a85e69402d191","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:32:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"67c5eeea92e963c6a420598bdcfd10ac3c7baedd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 21:49:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce6383ba1d2447ea5d6747b05946a6f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9393108a2bc64760b9a2a45b61e65e05 : SUCCESS in 12m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1667640fbf904c279a3096693800e8f5 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0aa52bbd2f34e3f9abb3dabe58f951b : SUCCESS in 14m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8dcc07b5130aed723c2b63d40590f23f043c569c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 23:15:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01b9951150f94f6696fa4bc30f8e80a5\n\n- grenade https://zuul.opendev.org/t/openstack/build/0cd63fe874c749fc8158d9d9484c50c5 : SUCCESS in 49m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f057141f21cd475b8f7e020459dd0f50 : SUCCESS in 1h 20m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f62cc7c20944c41a284770292f627c9 : SUCCESS in 1h 03m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24d4b7e4ba3b4c93b01659a632cf4e5a : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6292d99fe3b8441abb81bb1693917cfd : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/385404ae2d8a423e93d06efc7ca9490b : SUCCESS in 7m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3f0b98395e7242069fbdc70c0a782433 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bb34e6364a437ea82d8a0960ac87dc : SUCCESS in 6m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/894a955b3f8e4c45ac4d6631d610d35a : SUCCESS in 12m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4bf84770e7084f66b50af24ba617f17c : SUCCESS in 11m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/73ef6d67fe8047a3867efb8b186246c6 : SUCCESS in 9m 49s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bd4d70c9504e4bf1b80a23145fd8f3f8 : SUCCESS in 9m 38s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c2eebc88b87e46a39b600e0e24cd3b81 : FAILURE in 1h 40m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2da2982130a74bb48f2c612529e16449 : SUCCESS in 1h 20m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d6ebbc542a9415684feba5969971260 : SUCCESS in 1h 34m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73f5ce0ebcd545e596fd7afebd702912 : SUCCESS in 1h 23m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/54d4a22f2a0b4d0fba70a7be33ac8dc9 : SUCCESS in 1h 37m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccce16c0d7f943959b4ae9685800bf82 : SUCCESS in 48m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/17b12a7f8e4c40519eae29a815ab629f : SUCCESS in 1h 25m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/698cd027ebe2404ea17843ff98f2ca91 : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":3},{"id":"b278e7d1fb8237ab69c856cca67b86f941bb271f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-25 17:41:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c0cb3f0a2df2db704034b68c2704fc0d2de1d8c5","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:43:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bbab9dabe17d7cba39b233f67ead7ccfb5346e4b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:36:06.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ba06913875c25bae968cc18cf67ee23b09caa91e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 12:32:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8f2259ddad4a4c715b7674c4409653c736fc3b66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 15:20:22.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/889049 is needed.","accounts_in_message":[],"_revision_number":3},{"id":"c16f3ce4a0d201b6cc6c9f4fcd8274383ccf8e8e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-29 15:31:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d4b9115754622cd2ee8cfbfa6df0674d075f7052","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 15:46:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aff6ef09f5c461f849da89b02f7d82f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/19c8d166d543420c8bc3633525ab1212 : SUCCESS in 11m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d341b37f5ada4b3bbefb80fadc80954d : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dba152b3b58b4268995084894285ac67 : SUCCESS in 12m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c9e95183f504f86af6bb8a017ffb0b2 : RETRY_LIMIT in 1m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be2df95e7c1a036eb06d04063ae928e5922a7d91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 17:24:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/faa029d4d932460a9ae9cd200cdba8d2\n\n- grenade https://zuul.opendev.org/t/openstack/build/e50a35e6be2a48b0923a6d9fc7a2d128 : SUCCESS in 57m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/421474a7ace44cc68f91b75f04efc780 : SUCCESS in 1h 16m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82788f4082a6469690fb5101c91409e5 : SUCCESS in 1h 05m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603f43af26c641c19935af107c4e775e : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7468b22eae1645af8a11fe9c3a6c845e : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da9258fd269d401a82c2eba90c016f39 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3905b789796c4f73b6a67e717ade8fee : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/769698c7896e490b9615dd26c90224a6 : RETRY_LIMIT in 2m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3654f936f9a44765ba45933c61b4a1e8 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/09bf856124ba4ad9a188ce71f857267e : SUCCESS in 15m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb70a7a18cba4ef68ebb15d03ba7f56f : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9113e84393a447ecbcb2bd03b1168b85 : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0896a53c9b0b465d8c47c80cd5e8058c : SUCCESS in 8m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4e61cbfdf67c481790c448fa55e0e677 : SUCCESS in 1h 51m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/91b566ab05d1436f9149d8ec348ee1ce : SUCCESS in 1h 22m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e8289878a07f458992e40fbd59fe62d9 : SUCCESS in 1h 05m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c7b5738b3f464cc79dcf5e9b98d1f895 : SUCCESS in 1h 31m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/97454ec09bfe4e27b9e57ab309604b49 : SUCCESS in 1h 43m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac16a9c28b3e44c489ff6090eab0171c : SUCCESS in 1h 05m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd5daf57e6fc47a5b242317e72d0ba67 : SUCCESS in 1h 36m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9d1d3f8af3414c20950502046a1b957c : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c87f844762b604e2b90ba71cd7fcf09b090f43c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 06:35:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"41e8e6a87af563a23d8e6f8979c167803f5ccdda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:28:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2139932cb49f4b71a71545e4fd28b23e\n\n- grenade https://zuul.opendev.org/t/openstack/build/df4823ad2a824181a39488ae10dc279e : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/30117c0582804245ab4dfdf2769fdecc : SUCCESS in 1h 23m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4671a1be734f4d069897d9da7d2c63af : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fcae0ac8a504012bba02a156c261761 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61fa161757e6467dba44b661fcf34f70 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56db78acf0a94005b0521e0b89b36897 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d79107c92f0e4fdcbd4c73b31ec39be3 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b247d1dedbb4faea18811104083f0b3 : SUCCESS in 6m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/634d471cb6ef450bbfc099eb57b2bda2 : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fdc2c15bad484239850161122322fe14 : SUCCESS in 7m 05s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9ba439220e3b477fa629cb2583b746d1 : SUCCESS in 1h 11m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ad89135eef8b4787aa823f8bce3b4637 : SUCCESS in 1h 09m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6f6c3e3f0e7f4f06abc38ca28c9f7c7f : SUCCESS in 1h 20m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29565843957b419d8eb1d566a1b53dd3 : SUCCESS in 44m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4db0f703522748cfb11c998e5200ad12 : SUCCESS in 1h 34m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48b1e0f7c6d14e1d8f05d4420cec63d6 : SUCCESS in 24m 13s","accounts_in_message":[],"_revision_number":3},{"id":"fc3e1c0f5c18f90c8c5b499720673a740bed3539","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:28:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a500d698b44b280d2c5bdad748576d1473c30ce5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 08:30:43.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5edd635da0346d088a61f85303ce6dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3695b8f42b3d4f808250c3caf2848f09 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1f5bbf0bd1715b0004d966b7cbb31519691aa0de","revisions":{"932cbc5da97f6d557797abf4839e49e60939265f":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/889050/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/889050/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/889050/1"}}},"commit":{"parents":[{"commit":"2a135029902de4b745a9254f10aa414c6de3ef96","subject":"db: Replace use of legacy select() calling style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a135029902de4b745a9254f10aa414c6de3ef96"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 16:33:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Remove use of \u0027bind\u0027 arguments","message":"db: Remove use of \u0027bind\u0027 arguments\n\nResolve the following RemovedIn20Warning warnings:\n\n  The MetaData.bind argument is deprecated and will be removed in\n  SQLAlchemy 2.0.\n\n  The ``bind`` argument for schema methods that invoke SQL against an\n  engine or connection will be required in SQLAlchemy 2.0.\n\nWe are also able to resolve the following RemovedIn20Warning warning,\nsince the migration away from the bind parameter requires updating these\ncalls.\n\n  The autoload parameter is deprecated and will be removed in version\n  2.0.  Please use the autoload_with parameter, passing an engine or\n  connection.\n\nChange-Id: Ic956980a03e1a0e1b6f90c492d8a03a64ea275cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/932cbc5da97f6d557797abf4839e49e60939265f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/932cbc5da97f6d557797abf4839e49e60939265f"}]},"branch":"refs/heads/master"},"4e5cb96e7290eec0613eeb8a5fff3b1820a76d6c":{"kind":"NO_CHANGE","_number":2,"created":"2023-07-20 17:03:14.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/889050/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/889050/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/889050/2"}}},"commit":{"parents":[{"commit":"c04858c528fd002c8ba416e92a1f06a6ec1d3ae7","subject":"db: Replace use of legacy select() calling style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c04858c528fd002c8ba416e92a1f06a6ec1d3ae7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 16:33:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 16:31:02.000000000","tz":60},"subject":"db: Remove use of \u0027bind\u0027 arguments","message":"db: Remove use of \u0027bind\u0027 arguments\n\nResolve the following RemovedIn20Warning warnings:\n\n  The MetaData.bind argument is deprecated and will be removed in\n  SQLAlchemy 2.0.\n\n  The ``bind`` argument for schema methods that invoke SQL against an\n  engine or connection will be required in SQLAlchemy 2.0.\n\nWe are also able to resolve the following RemovedIn20Warning warning,\nsince the migration away from the bind parameter requires updating these\ncalls.\n\n  The autoload parameter is deprecated and will be removed in version\n  2.0.  Please use the autoload_with parameter, passing an engine or\n  connection.\n\nChange-Id: Ic956980a03e1a0e1b6f90c492d8a03a64ea275cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e5cb96e7290eec0613eeb8a5fff3b1820a76d6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4e5cb96e7290eec0613eeb8a5fff3b1820a76d6c"}]},"branch":"refs/heads/master"},"1f5bbf0bd1715b0004d966b7cbb31519691aa0de":{"kind":"REWORK","_number":3,"created":"2023-07-24 16:55:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/50/889050/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/50/889050/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/50/889050/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/50/889050/3"}}},"commit":{"parents":[{"commit":"6c150194c6fda112c1758abfca35a8f0582569b2","subject":"db: Replace use of legacy select() calling style","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c150194c6fda112c1758abfca35a8f0582569b2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-18 16:33:16.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 16:47:39.000000000","tz":60},"subject":"db: Remove use of \u0027bind\u0027 arguments","message":"db: Remove use of \u0027bind\u0027 arguments\n\nResolve the following RemovedIn20Warning warnings:\n\n  The MetaData.bind argument is deprecated and will be removed in\n  SQLAlchemy 2.0.\n\n  The ``bind`` argument for schema methods that invoke SQL against an\n  engine or connection will be required in SQLAlchemy 2.0.\n\nWe are also able to resolve the following RemovedIn20Warning warning,\nsince the migration away from the bind parameter requires updating these\ncalls.\n\n  The autoload parameter is deprecated and will be removed in version\n  2.0.  Please use the autoload_with parameter, passing an engine or\n  connection.\n\nSimilarly, we can also resolve the following RemovedIn20Warning warning,\nsince the calls we\u0027re updating here are the only ones affected by this\ndeprecation:\n\n  The insert.values parameter will be removed in SQLAlchemy 2.0.  Please\n  refer to the Insert.values() method.\n\nChange-Id: Ic956980a03e1a0e1b6f90c492d8a03a64ea275cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f5bbf0bd1715b0004d966b7cbb31519691aa0de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f5bbf0bd1715b0004d966b7cbb31519691aa0de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
