)]}'
{"id":"openstack%2Fglance~889051","triplet_id":"openstack%2Fglance~master~Ic1bc4ab967913a8adaac559482d330d72e47a4b5","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-06 10:05:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-09-05 18:56:52.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Ic1bc4ab967913a8adaac559482d330d72e47a4b5","subject":"db: Disable logging by default for alembic","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-06 14:02:52.000000000","submitted":"2023-09-06 14:01:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889051-sqlalchemy-20","meta_rev_id":"ef47b77ff8c1f2fdc54017f2a9d948d14b22fcd1","_number":889051,"virtual_id_number":889051,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-06 14:01:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-08 08:41:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 18:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 18:56:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:18:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 15:49:47.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-08-08 08:41:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-05 18:56:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-09-06 06:04:40.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"009ede39bb7960b9ea8222c9fbca060c9288ec6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b75af38e93aae377be3c8d750c565c32ff6fe13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:38:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30054ecb6b1a444e9e1aad72f3771c60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/777696aef0c34ec8ba0db7b4f5edd95a : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b35c6f3e0304541bb85a5f1b6072566 : SUCCESS in 11m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8359544d63b8444b9731a5cf912890da : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c211c4b3c1aa8f1c50d129c325545d26dd2a1045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:18:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d478ea1ec384f99aca6010582f200ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/01824be5303844a0865e72e3d2c960e6 : SUCCESS in 1h 04m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fed9bea220af48819230ce481fba6268 : FAILURE in 1h 51m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d973c92582a941919e6d4819ee5b240f : SUCCESS in 1h 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d881d8a2901a446b991772687605b0c4 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52dc60cb439c4d1984977ef7da4c3bb0 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e731f54efe495885607177df158fc7 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0618111c93394b798fee3ca9d609cdc7 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2b7fe8f8704be6bda61eacdb8b027e : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1f7d598ee7f0487d9d0d6908bd6ef98d : SUCCESS in 13m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/695640c4263a41c283163ba85dc471cb : SUCCESS in 9m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0be0e636247246b6a327970e692ba009 : FAILURE in 9m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/954fbacb1e5841debb3828cc11672815 : SUCCESS in 9m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5690019c655a436290f1a6ea14afaae8 : TIMED_OUT in 2h 04m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1603a07ee759429b9b4b7d2b4442c6dd : FAILURE in 1h 35m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dc1cdbc2455e4496b9d50eca2626bbb3 : FAILURE in 1h 48m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dd57e1fb99b54567ad3103f6622c96f4 : FAILURE in 1h 16m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27b75f8cc829436c993db3e6ad25cb43 : TIMED_OUT in 2h 04m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38ea2eef59114caaa000253a35bd1e0f : SUCCESS in 1h 00m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01f82f8cdd694be2ad0c31a413a041da : FAILURE in 2h 24m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0fa9b032413846ee849226623212e74a : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"b607b2d30303dd818823eeb286087783ab87650a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-21 15:49:47.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2aaa0329964fde0754ce450e019fed499d03da7d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 17:04:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4fd60f38c3404d918c62f60dd75fb44938ab114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 17:08:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0883e5aeeffdbc9680b4202db11dbed5627f588d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:15:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc3ea59e5b6421abc9e1f88766df840\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5f39021ca4f41cdbcae504516713339 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d949da0e6c4644eb9bdaecc69db0dd54 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/926dacf85dc643d18ba03ca8d6345649 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"35c43e609399263ad138749aef8d57c509483bbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:37:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17f3915a4aef443e91450454451efb2a\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab9720612fc24f7ca9297d6aeef88903 : SUCCESS in 56m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cf0a7064b494c118d3db807b3f48926 : SUCCESS in 2h 03m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f6a9ac22db54ad1b6515333b904bc5e : SUCCESS in 54m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10cd8b09e89541f389160712ddfc6c20 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87aa6b248c894613b5a7999db54e59ec : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e385a1440bb4d9b895fd0dc4f6d52dc : SUCCESS in 6m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/174fcca75d6a4858a98811b669f4d8f2 : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed1b67c40d3483eaf3ab6e8c61d2731 : SUCCESS in 8m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c085f22bd5bf41d99666bd179036dd15 : SUCCESS in 15m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0752c755182641cf8dd64574a3d3a3c3 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12033583d4dd45b5b312f27b47d1aa23 : SUCCESS in 10m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/634ec94525ec4cc78aea63ccebb93d60 : SUCCESS in 9m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/acd9d0bb197d46e3bfbffaed85901d58 : SUCCESS in 1h 36m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae9a196281394b169cc0b328fa0b1c00 : FAILURE in 1h 59m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8a861c2241d1441c91c788b479e8f23f : TIMED_OUT in 2h 04m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b063ed427ec24bc3bb5ed9085f62cf19 : SUCCESS in 1h 59m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/19ac25619e744b09923094d6f7465475 : SUCCESS in 1h 46m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c619c6f4854b4b458f12cec19965262d : SUCCESS in 59m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a10db82833324ab89f37f351eebeeb5a : SUCCESS in 1h 46m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a732a9107e144d138bb1d5a1188ef100 : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":2},{"id":"8fed50c0873e819da7c2edb85552da4fb09d7190","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:32:13.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"971a13bc7a23cdd01837a08bb5b53572807bf771","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 21:40:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/143b011d2cf14d068a264b71383d6f17\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ae46e429e71498e9a1df75c849c0eec : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b576cde255e9438387417a08777288a3 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1aeeb5e78bf0420b941d4462a3fd2afb : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8dc20f687a8fe4b90736e1d88b2fed2bfb5713aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 23:15:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87ba7ca317f54ea69bbdb8e1f5b36c6c\n\n- grenade https://zuul.opendev.org/t/openstack/build/e117571e4eb0412e91c78d80c2f54311 : SUCCESS in 1h 10m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a056666fbe80486d8966ffaf02e6b31b : FAILURE in 1h 32m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce5bf3c3eee7493a8b93702c78c035ef : SUCCESS in 1h 07m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05568cdbf1af47c6b75a511af89151c0 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa6580c6ffc2495d991cbc05697e083f : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfe63af540ba41a585dadb8085f97b60 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e54c9a9212484885b31827ae0cf6442a : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/474c637991f24c16a2378983a8a598e7 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c11e8553ee0b4787992aa4c02dbe22db : SUCCESS in 12m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abe3b4c3cc3c4f89bf074a859c91a976 : SUCCESS in 10m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d840e9127033459facaee390aa252d16 : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/69b633daecf548afb9c107822c059bff : SUCCESS in 8m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f7b9a274b2b04554b3b058740ec6be89 : SUCCESS in 1h 41m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ab3943f064464b4d827cbf4b9e75c424 : SUCCESS in 1h 15m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1ad0a3427eaa4f549a23028e82e8cae9 : SUCCESS in 1h 10m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ebe705b1f57a4ee4a1dac91534136e37 : SUCCESS in 1h 17m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f4e9006ee2c942568195d8b4cdb60242 : SUCCESS in 1h 33m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc2a16a22c594e7ab60237c51130f415 : SUCCESS in 50m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f5aac00043943a39753b378e5284a84 : SUCCESS in 1h 32m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/980019f368334605b48adef0980ed4cd : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":2},{"id":"255359123f1d192f2431010f5081fed0b139c023","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:15:06.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9e11b3c70db675205e18e03d1dbc3bf1a44eaecc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:15:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2635ba7f08837b44f2e01d0b87ff0ebd94fb0290","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:38:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daa07bd48a9a4b17837e57ff7876c20f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/48e28e40ecf44a0485049c7acb80bc8b : SUCCESS in 13m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73a68223377d4f82a02825f7750924eb : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2fd8f454e5ec4aebb841b74204346c7a : SUCCESS in 6m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"52627e1ce40a09d689c09cdb9738a708d3c90e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:53:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/418ab42dddb347378ad728fd4bdf899d\n\n- grenade https://zuul.opendev.org/t/openstack/build/e68f7a03a4684e6094af3b8661d8a1e9 : SUCCESS in 1h 10m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7898939c2c0f40d1b4cae8ea98311ba0 : SUCCESS in 1h 32m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f4b91e321914c3a920020d4d90b400c : SUCCESS in 1h 16m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1646ed37c53d46a6bf7ff439300cc7f8 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfbfb3996666438595cee4ae41d69099 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f349c94cea643abb6cbd8eb94f381ec : SUCCESS in 6m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fff2657a7f5471495dbe3391358ecd8 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a546aa52d4e45809197e7de91c300c2 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1341318730cc49c3bb31eb1ec59e8cc5 : SUCCESS in 15m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/141c85cf830d4e3b82826a43b4ccd579 : SUCCESS in 9m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9706de1a0de444388f4efc594c880550 : SUCCESS in 10m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c4cccaabd2cb43c7940f7c502a4448d1 : SUCCESS in 8m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e84102f8e8e14c52b8c3928fc81a0ea1 : TIMED_OUT in 2h 03m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/db843139e5334ceb9d428843a1e65ce3 : SUCCESS in 1h 53m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9b4afc4c08d48be86525908e1776043 : SUCCESS in 1h 43m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ab49fc03a71344e3a64b0dce78367b48 : SUCCESS in 2h 02m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/272ea0c76ebb4fb2b176b309e426c1a7 : TIMED_OUT in 2h 06m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/629fd1a26263475db553983a3a14f7b7 : SUCCESS in 59m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/433130702b5a49f7880c3632d195561a : SUCCESS in 1h 42m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7bb11f5889eb40199e654f2ef8a741fa : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":2},{"id":"d999cd99ceed7dc1c7bd2b59d9a74fbbb890f5e7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 09:04:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c98a9f5b43ee5555c7d755c42d3123f29690a948","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:20:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c0ee7ad450a4a619a69466ee8f07526\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/084184f9a1114774a6a2ec537a832b1e : SUCCESS in 12m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3a8b76d4def4e88a313722bd6a4864b : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/366ea0c3497845e5b49faeb37cf375b4 : SUCCESS in 14m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fdc6ee88641703184a49ae5bc91c055810f27142","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 11:17:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cea29b9362504fd7bdc917e640952a51\n\n- grenade https://zuul.opendev.org/t/openstack/build/b2900fd7cac6427f9faeec7a7d9edce3 : SUCCESS in 1h 08m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/77b14ce96bbf429fb82ca8fb36c6202c : SUCCESS in 1h 32m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a39eed5e4aa4429d8671438c0a064c08 : SUCCESS in 1h 05m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1c75916d45a41f88e3f3666ef525d30 : SUCCESS in 15m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91818f22669e475bab20e2f8c6bdc53f : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23e541d0e6e045b985eff83f2762b009 : SUCCESS in 7m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fb7c8b888104f01a38242d1812c83c3 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b6e9b6189f4c828c1872d7923d7208 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3045f5ee285e453c93768c94e5eb1773 : SUCCESS in 12m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40fe1349e015484aa598b745f7f9f5ec : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b964bf126e646a6a9ea92275c997e27 : SUCCESS in 11m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/711920935b8341dbbf0449f7aa4667f6 : SUCCESS in 9m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/588f25d14712495e939d1d5f8ede6e5f : TIMED_OUT in 2h 03m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/77a86c44f62e450b9c8ee00783452ec5 : SUCCESS in 1h 29m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/43c9e945aaac4931aa8378872f8a7711 : SUCCESS in 1h 26m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/44881ef00a25420d820edaf9283fc222 : FAILURE in 1h 37m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b47046fe85254d97bfc0825ef4e50c94 : SUCCESS in 1h 39m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8112eb08bf3445dbb6e50c8bd52c3f8b : SUCCESS in 1h 18m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/29f87fa30a1f4b2a89c6f254d049071f : SUCCESS in 1h 29m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/36177e14876b4ead9a31ef10e1c5fb78 : SUCCESS in 44m 44s","accounts_in_message":[],"_revision_number":2},{"id":"146ec183a1d4c41ad9451d5f792b5e53d384d9d5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 15:17:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90eaea2a94fdb373b429d581067eceb6737df251","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:33:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ca3047b21074a0aa58fc4cabb20b831\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/384b3241814e43648d0ebdfb9ddef694 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b58f1317fcee40f6ab5ba6690a4852a8 : SUCCESS in 13m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5b5bf7eb98074314a699689ecae04cc6 : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67ef24176969ec19503fac69d16d8a58f542c58f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 17:33:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e747670db7d447980a38c1c325344f1\n\n- grenade https://zuul.opendev.org/t/openstack/build/3dd598035aec40d29623aec8012732cb : SUCCESS in 54m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1895fcbaaf804523ad5f584f0c904e38 : SUCCESS in 1h 52m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8bfecce8ef241cf8c9747a8ee917879 : SUCCESS in 1h 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef7fda861864637b597b8fa61d104d8 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f712828b259744b0bf13be38ab17eb81 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/705b9bf0597449f59d9dc8a82de2c6ad : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1769503028a94ece8d66742049f75377 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0180485cd9fb4dea991db5435ef6b0a3 : SUCCESS in 6m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d60e57f91b23428f8af8ebb06299dee5 : SUCCESS in 15m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53e8e5e4bea74bb6ad556c325826508c : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b479433930154b4da7114e37b09523b0 : SUCCESS in 9m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c0b05f8a790b4caaa012a06d413368b2 : FAILURE in 9m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/775e0fd3a41b41828562c521ec7c8c43 : TIMED_OUT in 2h 03m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5bb744b3e51c4e258deaddf5a90f80ac : SUCCESS in 1h 19m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d58b14ef381e46b38849b4f7acad14cc : SUCCESS in 1h 20m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5a1919fa3a8149808710cd074a0abadb : SUCCESS in 1h 42m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3ecb879c4e5049cd952a3f340d97bbe6 : SUCCESS in 1h 36m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38654cff9847472d8a9f9606ec45de37 : SUCCESS in 1h 19m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0158bb1f96da4cb5af65601619aedd89 : SUCCESS in 1h 28m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/81660101c8a941339cd0d10af58771cb : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":2},{"id":"4de2671339edb1a8f457e8bab71a12764f37de98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:41:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"5653e8359f5a5e335eb03ae358d107eaf71acd18","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-09 17:50:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f07f916b7b7a18ddd9fde05682c11011e5e60a25","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 18:56:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0d921617372c0e1475cdf22b5cdd0f152cdb0913","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:57:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"15c32d6a6e249309c628fb26f32e5953dd2db672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e4eb86d556e4f55a19837dad625cba4\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce7a12d5450b47988cf557f4ce7ecc9b : SUCCESS in 1h 05m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/058e6fac6f4947b58b905108f48725aa : SUCCESS in 1h 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8af9f22a80454ba888de77c2bdb68508 : SUCCESS in 47m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b04cdc269bb4d5db39625c075e2e6b2 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/702e9e1539504f91aaa998165a52c810 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6fafbce18ee4a48a4336a5030748b07 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2613efcb7d34a6f86d4ff5a1b4dae37 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec273757f2e4ad39eb9d076c379bcde : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f4dae3997e41451f8700f8bd38da94e8 : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c38ce3ea8284eb882072722d24bc9b8 : SUCCESS in 10m 33s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/18bd591c0a07417fb53c0835a19467ae : FAILURE in 1h 02m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48c4eb26145d4b9ba7f710e737bf41c0 : SUCCESS in 1h 05m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9131d11273bf47dcb46ea34ff33e9dcc : SUCCESS in 1h 17m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/915d744e0b6243bb83fa12f1c136bb6c : SUCCESS in 1h 01m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2a94a1e79d144dfe877180c0e59a83de : SUCCESS in 2h 00m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60254b86d5334b5a8bd09a54cd2f740c : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8637ee94b7c69aaf2b8e16e08f3806b99c19b84d","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-09-06 06:04:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d8aebbd22225e792272d366619f3c80c7a486515","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 06:17:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c901365e18843dc9b70a79880d1383c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd70bf87651049f291fb429669e85db6 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ace6c60231044cdb898e32694d9a3a2b : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/abf47ae58f8a4a7c8431ce3803e2feda : SUCCESS in 8m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d3a3f1804344e82b1680fdd38ae39d8 : SUCCESS in 4m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5e37780e3b0f35a220a15beb0c92cddf7e0fc692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 08:22:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/663489e2ecfa40dc8fb3eeb9dc8f3089\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe8674bafb0e4f608f87bbeee2d1798c : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91212203cf03422aa2cb5e2306114538 : SUCCESS in 1h 06m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1edcd787e1e04b6c91af3614aa7e332b : SUCCESS in 44m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/298012180e474f80aca1a8de20afa17f : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fc5d898042349ad951f1ac303deedf1 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7528622270bd4b2bbee0f87937d5582d : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c76d8984dc264268818ad0231d5cd5d3 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/74ecc381d35744f79e2e50d8aa352ec6 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2799afc601344757924b10bfe7824c9c : SUCCESS in 7m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/dac549b4835b4b6ea3e5063bf86da57a : SUCCESS in 10m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d76bea5fdcfb482b98d78a6cbdcdf7f9 : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5770c582e94a4bc7ab20f5b9a1edd15c : SUCCESS in 7m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1c8c115a30fa4b859746d5668d7120cf : SUCCESS in 8m 13s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fa89b2225895474b9430167fe004fad3 : FAILURE in 1h 32m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6c4fa484b52142f0a4704ff7b2601448 : SUCCESS in 1h 32m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/affc3bbc10d7472293d01bd3b60fd0a2 : SUCCESS in 1h 29m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f312d75a293944b2b72ce9ddc6b6795b : SUCCESS in 1h 00m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80be605693104f8088d251bbffa352cd : FAILURE in 1h 53m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d120dc4a740d4c1bb8778e4dc399f51f : SUCCESS in 38m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/190cb638a87747158aead0ec1268e962 : SUCCESS in 2h 11m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6f406f6751004397acff68755c6de6f1 : SUCCESS in 24m 12s","accounts_in_message":[],"_revision_number":2},{"id":"2904d34a960b715b6a71b78ebf86fcb52f41cdb7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-06 10:05:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"758e226ed2f38fc18ec4ddec686abd09bb286499","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 10:13:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc69bf0f207041a1ab1f4f78205cd5ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0308316a631740b294f12cba2c88518b : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/27c42ca53ec3459e8be90b34f223e7e3 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7cb33964174421f85d10b9c47cd8eef : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c886a5e41eb74e76866ab879facb07c8 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b3da4342e292a8d837e25d2675c88b2244c8813d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:10:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0ed97c40a5b4bc0907b4f5ade71fa3f\n\n- grenade https://zuul.opendev.org/t/openstack/build/65a7337a7d1e4af1957c7413303a6ef4 : SUCCESS in 49m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/20200eee556d43f8a561711dfba47ed7 : SUCCESS in 1h 18m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4425d103739b4b69a0b07590f560a2e3 : SUCCESS in 47m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdf8b2ba0cd47b2ba66978d1d90aa76 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cf84c09271e450cb194914728a38128 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9121cc7785c84d33bb3e307f6108d9a0 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a031650617b402c890d42a99049ce33 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b7880d73bb94a9487a412d5f7df2575 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aad58fb01f848938bcd3f85da286444 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/83bff0835f634b4db73aff34f910a061 : SUCCESS in 12m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb885799799a44b6ac36f3cb87fee5bc : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/787ea7017ff641fbb9ef56bc80e771ee : SUCCESS in 10m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e1e39a896d6c4e7c9bb7d1a466164e8f : SUCCESS in 9m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/51ea8d28849346c6abfdb3a0695cad51 : FAILURE in 1h 58m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e93c2c06287e4e1fbb578ea6ba161e27 : SUCCESS in 1h 50m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/611dbf0fe37f4ac8abb15e12623e33b5 : SUCCESS in 1h 34m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/34b16726583e4f47abe93b6071c589c6 : SUCCESS in 1h 28m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a78bbe0dcdb44bd8102f64a6a19e72c : SUCCESS in 1h 44m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35e7beaf09a84df0b6676e56ba39b217 : SUCCESS in 46m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8ac9939b832749a18ceda08d64a3d0a8 : SUCCESS in 1h 42m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b00746d6614c492894f21ce5980e2a11 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":2},{"id":"f1480a147d09bac2070052c4138f63e188cd7c0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:11:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e442580b8068c2827c539570a6111e86eff72550","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:01:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed34658394274337aabe0235ca98bf44\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f8cf6b1202c4f3f96646ebd925f0be2 : SUCCESS in 46m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16e8297ebbec4255acabb12ac195fa75 : SUCCESS in 1h 11m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1aab3656694a4ea5bbf2f441117837ea : SUCCESS in 51m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9396179175e481399f0f7f43f4d8192 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d85b9f7d073e4cf5aa923b21f9de9faa : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f850b81b09e441a1926999b54506a4e6 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c0c05f9c8e745708b4adcd90d4c1f56 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6cc68df03694d6e88b11e9a5d54e2c2 : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db62ae2f32c4406aa6a18279942b75a1 : SUCCESS in 9m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95e29853fe3b4665809caed622d4a6c3 : SUCCESS in 8m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/50576e84587548d39fa87d4379afec5a : SUCCESS in 1h 25m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ab4248ee16940089eb64cf689a82357 : SUCCESS in 1h 40m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/67c338389284487496ae0b94696b2715 : SUCCESS in 1h 23m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1e86ebeae2ad45699f48abe0d556d191 : SUCCESS in 1h 04m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a3d8fe6bcaa84c37a6a9c29ebda74aef : SUCCESS in 1h 48m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5035c7227cd748ad9a4fd3772663c2f1 : SUCCESS in 42m 43s","accounts_in_message":[],"_revision_number":2},{"id":"5333201bfdbc842a745b36bf52fcf98383e36bf7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:01:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ef47b77ff8c1f2fdc54017f2a9d948d14b22fcd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:02:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8eef26651d04bfaa69d5534128bbab5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4bca2e40bfe47c78b8486da5e8b258b : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b9afcaa29572c2e16b2b4ddfa9d70b99de0efb1","revisions":{"62663ddaec2847898e2015606a325cc9b1670671":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/889051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/889051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/889051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/889051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/889051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/889051/1"}}},"commit":{"parents":[{"commit":"932cbc5da97f6d557797abf4839e49e60939265f","subject":"db: Remove use of \u0027bind\u0027 arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/932cbc5da97f6d557797abf4839e49e60939265f"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:17:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Disable logging by default for alembic","message":"db: Disable logging by default for alembic\n\nWe don\u0027t want alembic overriding our logging: we can configure it\nourselves. This removes a whole lot of noise from our test output.\n\nChange-Id: Ic1bc4ab967913a8adaac559482d330d72e47a4b5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62663ddaec2847898e2015606a325cc9b1670671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62663ddaec2847898e2015606a325cc9b1670671"}]},"branch":"refs/heads/master"},"9b9afcaa29572c2e16b2b4ddfa9d70b99de0efb1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-24 17:08:23.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/51/889051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/51/889051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/889051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/889051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/51/889051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/51/889051/2"}}},"commit":{"parents":[{"commit":"1f5bbf0bd1715b0004d966b7cbb31519691aa0de","subject":"db: Remove use of \u0027bind\u0027 arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f5bbf0bd1715b0004d966b7cbb31519691aa0de"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:17:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-24 17:08:23.000000000","tz":0},"subject":"db: Disable logging by default for alembic","message":"db: Disable logging by default for alembic\n\nWe don\u0027t want alembic overriding our logging: we can configure it\nourselves. This removes a whole lot of noise from our test output.\n\nChange-Id: Ic1bc4ab967913a8adaac559482d330d72e47a4b5\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b9afcaa29572c2e16b2b4ddfa9d70b99de0efb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b9afcaa29572c2e16b2b4ddfa9d70b99de0efb1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
