)]}'
{"id":"openstack%2Fglance~889053","triplet_id":"openstack%2Fglance~master~I7f34d8079975e57df904e59e325bee21ca1417ed","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-08 09:47:00.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I7f34d8079975e57df904e59e325bee21ca1417ed","subject":"db: Don\u0027t invoke and_() without arguments","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-08 13:53:24.000000000","submitted":"2023-09-08 13:52:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889053-sqlalchemy-20","meta_rev_id":"fb400daadce36f6197b5c8647828819b0303c24d","_number":889053,"virtual_id_number":889053,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-08 13:52:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2023-07-24 17:08:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2023-08-08 08:34:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-07-28 01:44:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-07-24 17:08:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2023-08-08 08:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 10:53:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:11:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-21 15:53:03.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-24 08:58:35.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2023-07-28 01:44:21.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-08 08:34:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"a708dfd68715f9a9012ffe519355dccc91b0a0ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb8ca02f0544199a24d2d2ed4e77bc183a37f2a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 10:53:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd44dc1369ee4e438b670e296cb5c0d1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e3e42dccdd742c9b9575bdca6125c31 : SUCCESS in 11m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8234740bd25d4565a37f5020bd365880 : SUCCESS in 15m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f19ad1fb165d4caa8ffc1d0c774102cf : SUCCESS in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ef8492033c994edd87ad8c5d620bbb22adb53b2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:11:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6eba4fded964159a80b5072e9556048\n\n- grenade https://zuul.opendev.org/t/openstack/build/46fb88d273b84e0f9549a3fd235e75fe : SUCCESS in 55m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/27d723e2ce5d4bc89612650f6393a4f4 : FAILURE in 1h 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4698d9e38304b57912516e6326e29a9 : SUCCESS in 1h 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea845143c6254e028a4020adb79e53c6 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d992673960304cb391032b8056923301 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16c6d06335894dd79629726b01f5675f : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cdbe66f04be4408a635ab697e61b3e9 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a4555e26374761bc3986e2ea6a21b8 : SUCCESS in 6m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7cf58d0deaed4074b271e8ac12395a74 : SUCCESS in 15m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a0ab741474b40ff9a6bdc8cba0ae998 : SUCCESS in 10m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/979f6548bf8f4d56b4ad7cfff96638ce : SUCCESS in 9m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/112195d2f3af4d679567a464b498b7cb : SUCCESS in 9m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/833f2b482b1c46bfb2d5ff7ca94720ea : TIMED_OUT in 2h 05m 08s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/747a12155c7d4c968483803494dd7c31 : TIMED_OUT in 2h 04m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0773a77d997e4419a61fffe720155957 : FAILURE in 1h 55m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bc5207d6088740bbba754c66b67ee3be : FAILURE in 1h 13m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/029d093f9ee044b98522cd368dfa9fc8 : FAILURE in 1h 26m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19ae9068e73843c4a74b46aa0e9d12f0 : SUCCESS in 58m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7b87310acfbd4ac5a8def147a3569124 : FAILURE in 2h 23m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fa862de49b9c47a1930cf6f9d5a766cc : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8e84c8bd60bc73410432f5c1f81ed89a42cf2338","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-21 15:53:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"96a948088d0b95d443387ac6c7b6b0de3f6f138a","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2023-07-24 08:58:35.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69bd3ea81dc66def1afa238eb8f8ac05da01b370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 17:08:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"afbed62e1715ec5d8d954f1fb77c9c037bd05d44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:28:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/938033446d8c4560b8167a516f42dbc8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0407a53083ab41c4b84fc3aaa7cd6ac7 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8848be6563864dffa1dffbea118d402d : SUCCESS in 12m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ddc7af0434e84269ab05b92d0ca90922 : SUCCESS in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b5a55f249465142d372d1148465716722f9563b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:34:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2ba90e8f83c4166934fb6d5a26b2aeb\n\n- grenade https://zuul.opendev.org/t/openstack/build/726518721eca4630a79bac2c7fe2a978 : SUCCESS in 1h 04m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4faecb60903841d1b9b6a0c241dadfd4 : SUCCESS in 1h 52m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac149beb000742b3bf255be5ddc888f6 : SUCCESS in 1h 07m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/448d60f6d8cf476b84560bda1c5ac59b : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f903a0412cd24250a137e21df096b8ab : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c3babe5231541d79ae054eb41709841 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1619de8a1adc4826a35752adf472c5cd : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1187d754dc44b9baebe393bd296231e : SUCCESS in 7m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b06f4a29eb524c54b967da3b4a569a4b : SUCCESS in 14m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e168ad45a1f47598047d7ea400f7e69 : SUCCESS in 9m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9bc20f7cc41843269bb5ac7d3161e5b1 : SUCCESS in 9m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/25bea6635e924346af51dbea3930a2f7 : SUCCESS in 9m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7e7445a972994873afe5575e3a0c83ee : TIMED_OUT in 2h 03m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4343b9375e5d4f03941907fc90b1607e : SUCCESS in 1h 24m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f35b4e5a821a48049374d3d8a97d14a8 : SUCCESS in 1h 11m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c2c93b7596b947508fe0e1ec7596fcf0 : SUCCESS in 1h 48m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/130d281dd3994771a913de79f49974d5 : SUCCESS in 1h 37m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd5807ab1e1a4556ab21f568362cef41 : SUCCESS in 1h 02m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/115c8ba135dc423f9047f04d692e8cdd : SUCCESS in 1h 15m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b9be9cfe9cab4babb53be13a69e3d3f4 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":2},{"id":"7207e0893dced2e0e8d962b5726d6c49f3107af5","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:44:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ecfb1d6c35c52a565f277908d9c37eccb66775f0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:34:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a33dad9dab1e3193fbe5b628d621ba3f08f30e7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:58:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d1879841d250d9908cc1f4a709810006362df75d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:22.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"c51befa126a9103193746d15c5e51d256d329ff3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-06 12:06:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d14570b695643494371fb0c0219f26e481c8059","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 12:19:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4375b2f8bf3748f6a91d3da3822e946d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdbef038308b488c93892635cd67f5d3 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a2cb7b370a74c9a94de592095b419e8 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/14fd6174801b430282baf7a1d661ae87 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1382785f739e424597d9f33ca890f28a : SUCCESS in 10m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d387c6a8c6d1af6a35002db0859999447c921431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 14:19:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19facba694a3415c8237e3d7bba1d513\n\n- grenade https://zuul.opendev.org/t/openstack/build/12a1f8ca6343433f984f8cd3f2e41ef2 : SUCCESS in 43m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6b92f20183b04a76a4449289f37583c0 : SUCCESS in 1h 29m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aab92abf236d43bfa2962782a0c51313 : SUCCESS in 45m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718b8445564548dbbc034ef34ec01bd7 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc40231881844e4bea49369d2a44468 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b06fcae6c0e42d5b482d844b8b1c980 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a3cee1f5a2e4e4394b195488723c9a3 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f1e04e6b00048e7a862db25772e6932 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7428ffec8f584ca7ba12d424741fab6c : SUCCESS in 7m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5938d4cb29354b8bbeb8e8355a5e4b66 : SUCCESS in 15m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac6ca0f35b854b38a773b0005f8af65a : SUCCESS in 10m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aadcfd9a81b8490a8335b2bd71168a5c : SUCCESS in 9m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f3899a06f11b472ca760a7c1152f62da : SUCCESS in 10m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1450afa8b0d4c2abc2a1886024f7f7c : TIMED_OUT in 2h 04m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/42be8f0d6a634c3fae665d1c9b7d1636 : SUCCESS in 1h 30m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/84a6ff67035d4763a7cdae03df9165e6 : FAILURE in 1h 16m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ebb2b43acd0142bf916b4bad22ebd8f9 : SUCCESS in 1h 42m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8030e797a68541469cba14c5b7b68294 : FAILURE in 1h 47m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1f861abdd084a998799ba8b4b344aaf : SUCCESS in 1h 08m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c07c5cb7d664bdd9df2797fd0354994 : SUCCESS in 1h 27m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c36cba13041e47b79bf33c1ecbc231c1 : SUCCESS in 26m 29s","accounts_in_message":[],"_revision_number":2},{"id":"c6f45d7bc244363ccfdd1858d7db9cfd81c28361","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-06 15:53:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a1f1d30909306a2d2ea40e57068919953d35200","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 16:04:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4103c01b7ca4026a07295e84b4c2c5b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4d93619bd494fd3a9a5d25496a10358 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/747a5e1fc89843389e4bee70218487b2 : SUCCESS in 8m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c8dc578ae72b46ab811f1391fbd6c176 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/43a0fd03fe564b6991a23568aea0097e : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64850bddf0ff69b9b75f5fa6f34c3944a7f0f285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-06 17:50:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/50d28b2605e843cebf81f082f5c0e2a7\n\n- grenade https://zuul.opendev.org/t/openstack/build/55e938fd05cb439daec261f12050d034 : SUCCESS in 43m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59f56d1b49fa43b5a28cf18b9d8de88b : FAILURE in 1h 16m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2dd9644ff6034484b319e8acdba0af97 : SUCCESS in 49m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699a553ddb754ced9e64c7327514e14a : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/528e648012cd422fad4ede75d0f1df92 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69981fde711a442287f03f1d1ba7454a : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9042d3d39fbd45fa8640a635ddecfa93 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3e9b6d7e47f647ed842d0721ff9a2daf : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38f6f2aa6bd94c0988dea0eb82557cdc : SUCCESS in 7m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f6a7cbe6b67e4e21a2557a2a3d0b32ba : FAILURE in 14m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5bff9ea1d28c406d9dc8b9332391d0d6 : SUCCESS in 10m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58ddbeef45374c008ef4bc32ce2c6866 : SUCCESS in 10m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5e6d7750b7224af2806939a31b76e30c : SUCCESS in 10m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8561d7225b97499bb045981256e52555 : FAILURE in 1h 54m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f1a6912cd55487dbe00ce10755105a3 : FAILURE in 1h 05m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/388275c9d9d049ceb35b480fc435f11e : SUCCESS in 1h 26m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ea492de59bcf47f6a3f40b84a13b4853 : SUCCESS in 1h 12m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/832f4f4c860a4e1695121e37c24bcd55 : SUCCESS in 1h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc6551d123be4a2a87ba9a7931f2a7c8 : SUCCESS in 48m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/481a02d3c6b2471aa3212bb3fec69be7 : SUCCESS in 1h 36m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/55e616031aa1431da909faa39fc56c28 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":2},{"id":"35f5e39de14ba1733d7c7468b62f8367db417b5d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-09-07 06:33:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0ce00578dcc1e8ed2fe90b94dd8d5d0bc3fb268e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 06:47:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/675e67e26b5f46c5bb4419c093910552\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/394c5117d6f24ee2927ee24a3a495903 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33dd0b7a51624b88a23cd20d74da5201 : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e9e3ada5ece45cfa25c4fab7cc9b525 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e9c98406edf4351af61fad363fb57c1 : SUCCESS in 12m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1bd532ffd7f34f8abc1b962815266973405ccc20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 08:37:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d215622a0de4f1cbac570073bfe66d0\n\n- grenade https://zuul.opendev.org/t/openstack/build/048bf7791382446cb6fef5f92b4879a8 : SUCCESS in 54m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6341402b73974825aa2a661f7b4144d2 : SUCCESS in 1h 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/494d3c266f37479f863d7c85e59894cb : SUCCESS in 58m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b696e331f7254261b32d5777e0e578fa : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77532505223d466da43a090c63f840c9 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9e955bdd61c4d7d84b6c8c06a29d020 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f50bce1a7da45bbb8b9b54bc8d4d771 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e3c8a9aab41492ebd8f08762e8edb0a : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029a53485ed84d859758eacd460058c6 : SUCCESS in 6m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/51910565bf73466aaaa218199e7a82aa : SUCCESS in 14m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ce3615bc6ca5409fb5dc732bc8ab14e4 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5d4a17c6c264a84b80598b215b5d5d9 : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/35001084e7ab43e2a3cd50209e368902 : SUCCESS in 8m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ba61e8c069ac4ef191ccf204dc94e21c : FAILURE in 1h 49m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/be1e6448245a4c108fa216c4a3cf74a5 : SUCCESS in 1h 25m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/60ac4aedb6fd4e058aee846ff34405a0 : SUCCESS in 2h 02m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f596a7d64524458eb5a111615fc8e11e : SUCCESS in 1h 15m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4b27f120678d4e31b5b1c88569dcfa2c : SUCCESS in 1h 26m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/158625c8fef7479995391d45b74f3ff1 : SUCCESS in 1h 02m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/caa4168ea06e4329904f215b163ba781 : SUCCESS in 1h 54m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ebb995df05554354b7ca5a7e7ae5701f : SUCCESS in 23m 32s","accounts_in_message":[],"_revision_number":2},{"id":"0347fd43966656f1e90d4a71e112d90d46e3d3dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 08:37:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e36549f2d8a9fd2974115bba8351c9a50e87f57c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 10:06:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/41c1827bb38447acab0e549ff24577cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d843608ce4d42cea57e04a64ca4881e : SUCCESS in 38m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8669dbc3e2ba4b42bae172cc6cc8dc34 : FAILURE in 1h 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64cbfd5770f34cedbe9cb52e38765407 : SUCCESS in 1h 02m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cbc596c6b204041b25b31ed9dd5d6ac : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de2f995d414f4163b8f3920c24f2aa78 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d6f95f9786148ff9e0dc213e935b105 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fdf42548f8f4ef3b35c91ee8321d099 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c61f0de7d644be8b90e72d989f95fe : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4635dea713bd47be82531252fc27d7b1 : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a3f09ad0d77f4964850b4b961ee3d202 : SUCCESS in 8m 37s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/12c72759d5884b9d98b9305013f16112 : FAILURE in 1h 23m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/837dc41283fa45a5a907ab6a35b85ba6 : SUCCESS in 1h 17m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c75408778ca14b99ba5f8c73bb1c0185 : SUCCESS in 1h 16m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fca3a68fafac43e7a988ec5fe4f7dfd0 : SUCCESS in 46m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f2bae5a399d4e19adbaacef5459e693 : SUCCESS in 1h 11m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b6c903478cb46fcbec4639738f7cf68 : SUCCESS in 27m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b468fce038bb69609bc5c3e811f54de71c6c755b","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-07 15:51:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0efdb51326f36cceb5b6a8c2d8de4a50c5affe6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 16:25:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae4c5266f195453ca56a358359feaec0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1633f407cc2c4c0bba7fa6daf896c058 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c59b3dcf1064e4285dbf86f3abff464 : SUCCESS in 9m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d899ad32a2a9491783d8cc5037bc2482 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/13117cb7cafd4a2fb30780445d048dca : SUCCESS in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1c71757eeda06ce266935c783f12164bd8874f31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-07 18:08:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8443bb1dbed1478a831064295922d34e\n\n- grenade https://zuul.opendev.org/t/openstack/build/433bc1bcaa7b4662af194493cf8c92e2 : SUCCESS in 1h 12m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/948df45ae4ad44179c3777107bcd92ab : FAILURE in 1h 30m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/996291e8a2c84f04a327ec27a6c54559 : SUCCESS in 1h 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed37d56874b475aa7f6c9a5805c3aa1 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f87aeb1ac8eb4d8d88721169ca844d64 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21d2db731194404d937ef5b4c5ed378d : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67c5243c4d3243ffb6fb928eccb1b067 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/14cc020b114544dfbbf75eb6ec0eb1f8 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9723ccad683c459d86dcbb23648020ce : SUCCESS in 7m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6f69da11d6584471baf07c8804b47e60 : SUCCESS in 15m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/018d128ab63943ee976b00be8347a677 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d616395a765548f9bc68f608ad3086fd : SUCCESS in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/876dee81de244313bcd6bb250ef15abd : SUCCESS in 9m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e470e7793b01478db111171496b18d48 : FAILURE in 2h 02m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c5f148ea0f9541bcb513b2ee87dfa752 : SUCCESS in 1h 24m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/843c803f901949bda3373967d4da65b4 : SUCCESS in 1h 43m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/99dbf2628dc04570a9b0cb9a0677022e : FAILURE in 1h 28m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6f6290c8963f40d193039f568bf27c13 : SUCCESS in 1h 26m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f0b350457a94a62a878dc38cb6c1fc8 : SUCCESS in 51m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9dc5552671e4f13a1cfe31049a234cd : SUCCESS in 1h 37m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/988af8eec12c41bb926457db2b8cacbb : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":2},{"id":"adcf1a36e3a7c5905c17d8d139f1d3ca8c589c92","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-08 09:47:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed5a46bc727dbf9eda5448f0318fd669bbcbbc2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 09:58:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89df230db9b14965a71b2b73d4a5bfaf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e226292938e4824b71c080d60cb0daf : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4f6b6ddda3d443286dc7790ae933041 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/47d9ad87cc734769b8512627f101000a : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/723352ff2bf3446fa87b0abf67a97f0e : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f54c0657304a8e938195b4baab5560cdd664f0e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:49:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3d436ac4c0945f3bb7e4e6ca5b7b516\n\n- grenade https://zuul.opendev.org/t/openstack/build/031bd0ed17f54f668b5cf7024acaf8d3 : SUCCESS in 55m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c97c1c87aa546e3a11eb4081327c51c : SUCCESS in 1h 17m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f06ee43d885640ab894247a5879d0b28 : SUCCESS in 50m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b40e52eb264b27986efca3d21245fd : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f31dbff907f4dbba88555131463d51d : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/552deaa23a094c95908d8f7e38d661ce : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a509a598f654ce2a389e4d74a49a84f : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec24d4d4f1bc4c52b01d770ecdee4fb2 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e94a6624ee42769afcefeac8f1422f : SUCCESS in 6m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/66e9619bcc894ed19b637c3ae4d8182b : SUCCESS in 13m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27563f4491ac4aada7f8234a3d43db1c : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1e202af405b4bd09ca11ad94f6eadb1 : SUCCESS in 8m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/86c7654955a149c484b0c7e5dbd8bdef : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ea4e77197a1d464d82cad68c4ba4ad30 : FAILURE in 1h 51m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c6c0bdb420f344b6a67c6d1a1e2dce3f : SUCCESS in 1h 27m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/96530e185e3146279c18c1822b7b7fb9 : SUCCESS in 1h 31m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/39f25d60a4664b5480bd552a190c7a4e : SUCCESS in 1h 27m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f762adfa57b345d4a1424398ea38ae58 : SUCCESS in 1h 56m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62012e0f0cb941e2bbfa8ed3c7b193af : SUCCESS in 1h 15m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/81379c238b8f43a29f6c2f2e2676725e : SUCCESS in 1h 46m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/50ab45a927fe4d05a7321f2234a29333 : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":2},{"id":"28586cbd748aca2ae3770c52fcc0ee54818b5c03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 11:49:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6ae2d690c9f66b2196ff00ea4ad041ff53d3b354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:52:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfc0d637d2f940949742b5cb9f5079cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/75e9396ec87d4c169c7b2b46a12d11e3 : SUCCESS in 54m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ee26ece5e4b444e795b59dc92bbd1ccc : SUCCESS in 1h 30m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c880f87837343619497e84430bdf0d0 : SUCCESS in 1h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b6894e6eae94cafa62f0bee1e70487c : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dc371674db7416c8539ede9ed6ba75d : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9c1ad2daabe4d82bb903b4449db0cde : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1a767c1040c40b59a24b8379411761b : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cf7303e02a148b5a24acb0d3e9e568f : SUCCESS in 6m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/070b23e6a8d542c1ad988f7315f1a71d : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69ec068846534ee997e5512e06d46f48 : SUCCESS in 8m 34s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96e2ef46580840b4a5beb8b10c704e9c : SUCCESS in 1h 47m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bf530ef33159499c8e67c19392a58b62 : SUCCESS in 1h 40m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cf0890ff48ac468ba03a624345c2b1bb : SUCCESS in 1h 44m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67bb9b4149694b15b1531ec2ea195eb8 : SUCCESS in 50m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d0bd5407f148499fad77a150bfed792c : SUCCESS in 1h 57m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1eeb1fa9c1b44a089fdc7788b6771bf6 : SUCCESS in 37m 44s","accounts_in_message":[],"_revision_number":2},{"id":"2740811ece97745ecbe8ace54bf05abacfee3568","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:52:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"fb400daadce36f6197b5c8647828819b0303c24d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-08 13:53:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c58f08159f214451aa1fb23762b421d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecbea8c346b24fa8825a9e661227086a : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"36329c0a755277c293ee8dba5686bdee30d938a5","revisions":{"88051661beb6ca6143d69c5408cb6117161c3516":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/889053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/889053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/889053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/889053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/889053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/889053/1"}}},"commit":{"parents":[{"commit":"1db496c6ce0acd23b11eb1213cee17039cfe02ff","subject":"db: Enable batch migrations by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1db496c6ce0acd23b11eb1213cee17039cfe02ff"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:21:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Don\u0027t invoke and_() without arguments","message":"db: Don\u0027t invoke and_() without arguments\n\nResolve the following RemovedIn20Warning warnings:\n\n  Invoking and_() without arguments is deprecated, and will be\n  disallowed in a future release. For an empty and_() construct, use\n  and_(True, *args).\n\nChange-Id: I7f34d8079975e57df904e59e325bee21ca1417ed\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88051661beb6ca6143d69c5408cb6117161c3516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88051661beb6ca6143d69c5408cb6117161c3516"}]},"branch":"refs/heads/master"},"36329c0a755277c293ee8dba5686bdee30d938a5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-24 17:08:34.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/889053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/53/889053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/889053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/889053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/53/889053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/53/889053/2"}}},"commit":{"parents":[{"commit":"abddf333add9c3be5938e10ae4b60acb6177dd19","subject":"db: Enable batch migrations by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/abddf333add9c3be5938e10ae4b60acb6177dd19"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:21:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-24 17:08:34.000000000","tz":0},"subject":"db: Don\u0027t invoke and_() without arguments","message":"db: Don\u0027t invoke and_() without arguments\n\nResolve the following RemovedIn20Warning warnings:\n\n  Invoking and_() without arguments is deprecated, and will be\n  disallowed in a future release. For an empty and_() construct, use\n  and_(True, *args).\n\nChange-Id: I7f34d8079975e57df904e59e325bee21ca1417ed\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36329c0a755277c293ee8dba5686bdee30d938a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36329c0a755277c293ee8dba5686bdee30d938a5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
