)]}'
{"id":"openstack%2Fglance~889055","triplet_id":"openstack%2Fglance~master~I4df3a52333d71809d6ff305a786ba6ae03a230ca","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-12 09:38:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4df3a52333d71809d6ff305a786ba6ae03a230ca","subject":"db: Pass case.whens as positionals, not a list","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-12 09:39:37.000000000","submitted":"2023-09-12 09:38:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889055-sqlalchemy-20","meta_rev_id":"2132a2747083af7e33aba5f1a7ba3c2723b96991","_number":889055,"virtual_id_number":889055,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 09:38:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-08 08:36:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-07-28 01:42:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":1,"date":"2023-08-08 09:18:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2023-07-24 17:48:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-08-08 08:36:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:00:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:20:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-24 17:48:57.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-07-28 01:42:57.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-08-08 08:36:55.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-08-08 09:18:10.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"}],"messages":[{"id":"148bcc4024301ab715c1d71d73bba6bacc22c724","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a4bb8a452539d360ef56fa6eaf5eaad9dcc98f09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:00:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16bbd900f1c488a96937ba560c11cc7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e981fa8eb5d248d2b145b3a876a24d0a : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f56183d19a5457e813f2dfe17e621f8 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d4de049c9b644fa8a1a0f416d8cd7ab4 : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b905a03f1ab04fe9beb9874892321bf67e13303","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:20:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd932a028a66412a83de2ed7da4a8b18\n\n- grenade https://zuul.opendev.org/t/openstack/build/535bd4328f11485ba3a166514dd5f303 : SUCCESS in 40m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91e45d055b8d4cc29914b6f20ba3edc3 : FAILURE in 1h 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d67910374fae42209c7c9e0e448faf31 : SUCCESS in 1h 07m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ca2792bfb74bc8b3ab305a88cb6baa : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94837d95b48f4d528881d8bd43ec3ab6 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7009d7595a95496e8b923ee6a27907f5 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc83f34f303a4c0d8cea4d4cec41f1d4 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e52ecf360fb443d94f3dd3391bad902 : SUCCESS in 7m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/63d3b9d9fc60493abc3f9167d593d998 : SUCCESS in 12m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e87b3aafe68456686e3b79f21b0362e : SUCCESS in 9m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/756d0c9dd4044dd09fa28ffb49272d18 : SUCCESS in 9m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9a373e31fae649ddbb5f3134b7415a43 : SUCCESS in 9m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0386f46503ec45f78c5c88b64033b076 : TIMED_OUT in 2h 03m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8ab6f63ff754453ab058920b24b29e02 : FAILURE in 1h 25m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/827f3845aa204005a610cd456ee2eec9 : FAILURE in 1h 51m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/92ccebfb92d547339c6bfb21b8fc7019 : FAILURE in 1h 39m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20fdb8ae1dd844f19abe1c2182c70857 : TIMED_OUT in 2h 04m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/345e72f354484d31acdd6e25ac5076b8 : SUCCESS in 57m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/88c9fb7f343e4097893a64fa110d362c : FAILURE in 2h 29m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ecfb5bac5f9b4e5196dd0833b805a86b : SUCCESS in 40m 39s","accounts_in_message":[],"_revision_number":1},{"id":"da8d6bc69bf97f445ccc502872ea1d212618dbb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 17:08:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1d9b3ee6814e504a64c4679ca26550623701fb08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:38:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58419ca0559c4f3299fb8a7038e31db4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8838ba54b856489b98b506c12637b5de : SUCCESS in 12m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83ddbd9c43ca48b7ad1da85d813a86a8 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca3d1d0965944e2bb8379662ebf59ae5 : SUCCESS in 13m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"48135f574ef1b45c825c195d188e54c1fdc1bdd0","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-24 17:48:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fa99301a69e05d3e4fdb55d80716df7ab0685a97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:49:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/669ae6f2479349258d33d520b49d8d48\n\n- grenade https://zuul.opendev.org/t/openstack/build/a16d1b39258743038e1f0b0910cc90ad : SUCCESS in 1h 12m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18d488d7c5974c4ba6c2fd851970dc4b : SUCCESS in 1h 45m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/521193dbb2b84f24a35c6411e2744b36 : SUCCESS in 1h 07m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/317e310d35b74ae69da452a127973070 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02c61bb3a4d34ae6be48aef6bf1fa971 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/017caa2c1ac24b3ba0f8f176c2a9ba52 : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c062fe55d074863bf6edc91f4d8a7ac : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f21182578c4a4681cb051045c024d3 : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9968d931a4a542f5b14997d1d0aadbc0 : SUCCESS in 14m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a455061f39d44d1aabe6cd9101cc543b : SUCCESS in 9m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d1f01a11cc34c518fe0099b82aaba62 : SUCCESS in 9m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ce40d60af2af48f8a32ac40e20ead797 : SUCCESS in 6m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10f20f6ed2614c6f83d1acc5baf071b2 : TIMED_OUT in 2h 03m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/483b900739de4b7daa2ee60a83ca5600 : FAILURE in 1h 11m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f1472f413efa4165a5e7a5881b760c92 : SUCCESS in 1h 11m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/301aec97123848429f6193a622889ba9 : SUCCESS in 1h 10m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e41e406c51e94ee98d4c8c6b8de6eb3b : TIMED_OUT in 2h 04m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a14281a1c3e41abbac412321efc225f : SUCCESS in 48m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/10cb74796c7f4bc3b96bdc20c4f098ea : FAILURE in 2h 06m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/21d66d4a917d4835a081494869474ace : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1008e7732c39121a8df6d5cea5bfdcdae465d568","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:33:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6038f0bab758f265b55fca9f393b9178b5b76eb2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 21:49:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36a1e47eeb9a442f82ed2a040808f214\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0168f068d654d23bcb15b3ac078c5e6 : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4f10a34bd6e4f19b1884826b2a892e0 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/05f8fa583f0b4f7eb13b14c318effcf4 : SUCCESS in 13m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78493f48a21f02e44dc68858128d454ac1bcaf8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 23:33:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fefd5fc4c7c44ebabc149a0bd9c0336d\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b558814d4fa45e9a34fef38d087c78a : SUCCESS in 1h 07m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7fe43e93ee0f414c945285eef1ec8343 : SUCCESS in 1h 34m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0933bd59359b413dbbaf2a7d7b8d978d : SUCCESS in 58m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c662f9fd513449038678c1187f531ceb : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/43d964fd2f8b4d20a2680f61b1c404f3 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4ff6e4c04f44f82b485eaad09981c58 : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc1df86e3b6f4e90af842e0565e03402 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f321c4ff4cca43c99c16efe89a661d3f : SUCCESS in 6m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ae36e0404e03434c9e7dd0faab59fbf3 : SUCCESS in 11m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b83efde33c264f0cbeaa674854213373 : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/452d01fa54294dac9a711de586f494c7 : SUCCESS in 10m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/67f63aab8f844683a41cbc1b640af147 : SUCCESS in 8m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/98c91b90cbbf472983f88e4899970a39 : SUCCESS in 1h 47m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8b6924842b944753bc5f8ef9c1e3597f : SUCCESS in 1h 17m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1b6c880085784808a31bb3aa4d094d37 : SUCCESS in 1h 16m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1dd7fb68ddf14d1eb2227f2b890ff997 : SUCCESS in 1h 51m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5b5f4861414444258ff3538d8854e749 : FAILURE in 1h 41m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f80eff1f8ccb45beb769e8efa5dd6512 : SUCCESS in 1h 10m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4080d5aa609148dc9b07ec3663ed1d68 : SUCCESS in 1h 52m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/334e193308174592b45603b0a1a8c841 : SUCCESS in 40m 20s","accounts_in_message":[],"_revision_number":2},{"id":"34f1e1bb1159c984bb48afb4bcf18ea431bcbf21","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:16:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5dca5b0cf281337680409659fa6dd3ced16412df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:44:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5c3d1b94d644bab3dcd6321d380380\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/283edad1641c4dda8c4906c88066e67f : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3887499406df47a98e0cc1b148ebbf1e : SUCCESS in 12m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/28705f81ede546e9a12023c91520f428 : SUCCESS in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2bc49f9938f155a5364fc4af66fa54a9a9ef9727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 12:20:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/69d9c5b615734fd4ad1e6d4a9e1ce0ae\n\n- grenade https://zuul.opendev.org/t/openstack/build/52839e76f6774bf7a5a02710d6a14f6a : SUCCESS in 57m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/948918c355614b708af2505af07529af : SUCCESS in 1h 47m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57ed1e9ea0fd4e8e8b35b402b3b9bde2 : SUCCESS in 1h 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d42332e3484a0996e9a80952b90953 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04830a5067fb4460900876f3d6e7597c : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f401777369a04dc790a1490ac2baae4d : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7ada1a823b548c5848c0c9b00ba6871 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9686ff48338d48f7bb3ecd7c2cdd8a15 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/938e98a4215841f8ac959e8d81fed1e7 : SUCCESS in 13m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9470a5a346da4124a75558fd647908b6 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce76e3c3a191408b9e305477e5a602eb : SUCCESS in 9m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c04fce0942724438b963d67b72b69850 : SUCCESS in 11m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8360d62980024e4b89b6083d02416bb2 : SUCCESS in 1h 49m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/09fcc047a9ff47789373c389a7630845 : SUCCESS in 1h 36m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eae5672efcb84034a64091b1a79f587e : TIMED_OUT in 2h 05m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa5a86ae9f1440cfaf128e43fe4319d7 : TIMED_OUT in 2h 04m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c90b3e6ef1704db18ebd289133d21c40 : TIMED_OUT in 2h 04m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6b73f4c113d4687a7bc40c398f677ac : SUCCESS in 1h 23m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/45f1f1a74e514390abcf4534985fc35d : SUCCESS in 2h 24m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ee39146a8ca1465a9cb33b4545de627a : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"b2c932cf6c63eb5f015909b40f56dd18787d8dec","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:32:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3b69f0334dcad07709770e737d69d7f688341fba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:02:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/075f72d99fae47fe8d9618a40554eff8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f04478838d7641109deefecf0dab68e1 : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7675a1a2c2e94d1983ed6662a32eda14 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/29387279a5e645f59cdeb14712fc3b06 : SUCCESS in 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de6a2e64b138449fabe742ac0985295c84928ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:37:46.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4aaffdd617de4301a43a27f8f91c09b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/6933cbea251b435c9eefaa88a5c68b26 : SUCCESS in 59m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/574634abb7694a2eaef4323502ced3e0 : SUCCESS in 1h 24m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33adc1ed35bf4f7f97458b13c968101a : SUCCESS in 1h 04m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91ac603abcd44ae8a7baaa05d2deb066 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/186e77ce899b422a9e643a650841420a : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e5e0525b674ea6af0f708e4760d6a3 : SUCCESS in 9m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ec7774be86b4002b113d3e267d23eba : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dfb4f43c45948ffa8ffdb2cd042fc9c : SUCCESS in 8m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/027524cb6b2846b887136574df244200 : SUCCESS in 12m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/baebe4c9309347b4bd2d5fc7827d6da1 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a079a09f140e4268b1d4ca9e75a6696e : SUCCESS in 10m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/049eedd2922d42c5a07d49aa7695ad6e : SUCCESS in 9m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b467d4b946a4929b1d7f5ebe806ae80 : SUCCESS in 2h 01m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2edd7bc4a27747e784617c41824f6b01 : SUCCESS in 1h 47m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c68951b353f402eb8e6064402605e9d : SUCCESS in 1h 41m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ad9fa1036fb3461a94da2f7164d37c66 : SUCCESS in 1h 31m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ffcf5afbb4a643ab9cdd86270e9fb257 : FAILURE in 2h 03m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba0138f47c464e6492c71c0e242ad0b8 : SUCCESS in 56m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2db06c7a63fa45378fe660f2e1e2a4fc : SUCCESS in 1h 38m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9304b4d81ae94cf386b34543e559dc55 : SUCCESS in 51m 04s","accounts_in_message":[],"_revision_number":2},{"id":"7d4f1ab88e7b0c95c1bfef1f90b97efe4120e8ca","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 09:04:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b66fd336a9379e990b340c6d662686705db8f6e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:27:20.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb5b01d5ad0f4bbf8c2b2087e62a05df\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6f1473c20144fbaa85629e763780c30 : SUCCESS in 11m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a31cf42dfe14448b9b5269cfc30bf51 : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8329436f17524d9bbfa38cd21f227a7f : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"10ae4442dded5e2d20284fc4ef65986241c94b8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 11:16:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dec7dcd48a04859b841524b2993c411\n\n- grenade https://zuul.opendev.org/t/openstack/build/0513ca0a6629482e9e6e77122f6a1ea6 : SUCCESS in 56m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/acac595fe9e74edd8940b562a78bbdb9 : SUCCESS in 1h 28m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07a63fde4d2f41759e237fa471a88b35 : SUCCESS in 1h 05m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e45a74058b52436db53e8503253e3350 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e4b13cf20764b37924d5674107e3c70 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a031bb35845542ecae30a747493f0f24 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/090116e669914ce8888c3fc18c06dae9 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1370f2e77ed949aea541d89b08f30b67 : SUCCESS in 6m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/696a1bfdc102424a9c74ec6f9632a860 : SUCCESS in 12m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09a5b26dc0a14e7eb77ba855c4aff900 : SUCCESS in 9m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8767b6ededd643bd9211445865c4d192 : SUCCESS in 11m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a46cd99416b3494bb34bfd2cc6d4514c : SUCCESS in 8m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f7f7c55459bf4cafab4df5c92641b739 : SUCCESS in 2h 03m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b4fc72747354a62b3b47fc51030a5a0 : SUCCESS in 1h 39m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/58868d7e5e0342c5a96e041c017cabe6 : SUCCESS in 1h 24m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f27a2696c05b4613933d2e520c2b75b9 : SUCCESS in 1h 52m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69387f4561ca41ec960cdb7903c60e44 : SUCCESS in 1h 49m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66e9148452bd4b6bb346fa3ad8319c71 : SUCCESS in 1h 09m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2cade27638294e2196bca8e4f1e9015e : SUCCESS in 2h 02m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0dfdee4fa9f4adba7719dbc19fb370f : SUCCESS in 44m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e4747b431613a3e9f0a6a3a5463814dc6e8564a2","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-07-28 01:42:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"dfd804307030aa0d23c6551041ffe5240c200cbf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:36:55.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"59573304fc0115e07e8dc8d8bcad5c5157194420","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2023-08-08 09:18:10.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"e7beac78482ba817509aa6c59d35d74c987d4582","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:58:06.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"87f51ac98becb846f96deab6e871ca7f7c9c702d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:23.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"38e2dc34553800d35512cc4d4498f6af3ac91922","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-11 07:44:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b60cf1f1a09fda916c87c7984e6e2d68f45a9700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:40:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/536448e02e4249ae8abf2b84915166cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed6ec639799b436f86677eca1be13afb : SUCCESS in 1h 03m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/520f5fa7a3ce49e4b488cb12e76b0746 : FAILURE in 1h 16m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31896402091645e69b32b44d50eca229 : SUCCESS in 59m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af14dbb2dce14a95852ad782558d6c06 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2d9e1c17271431db347bd4f61d3fd0b : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f206709320114d2bbcb3ff164825f573 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9b9095398fb48a284815b16b232f790 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3ff6d016f83486598a8f477dff7a4b5 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee9fd3285ab54e02a636c848fb77dfef : SUCCESS in 5m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f315441df95d41b5be050a12a72d8a26 : SUCCESS in 13m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c62dc327bba43a4883941adf9101e02 : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/975989ffa8cd48109f212c94bc630982 : SUCCESS in 9m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/11cb39b3c1654a5b9dd9ee6b68de1311 : SUCCESS in 8m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8cd86442289c4176914c39de90c10cef : SUCCESS in 1h 50m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/645fd12da02149e69af62cfe5528b21b : SUCCESS in 1h 30m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d2eeaaad946b4e6ca6e49d0b6ba22393 : SUCCESS in 1h 38m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b2c7c66311254769895b678e876bbbf7 : SUCCESS in 1h 37m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1ab106356c543d0b1574296671dadc3 : SUCCESS in 1h 38m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e47539e58a01498ba9eff67b183979a1 : SUCCESS in 54m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f305cbb0a7264d219979d9b28cbe0bc5 : SUCCESS in 1h 32m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ad27874792fb48f3bd8c1bf2eff17117 : SUCCESS in 27m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ada9cdcd830a55f45d4e9b132648478cb079f5d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:50:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fac35f94c59486e8a7e96024f15ee28\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdff0c2de85a4383b08d01bcdf97bd39 : SUCCESS in 10m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95ac23c4380b433c971cf2cedf73ee2a : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/39495c5d297641b580e882f786526438 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21eae7ab22324abcba4d94dc65e65dfe : SUCCESS in 9m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"07f4df73b8627152f7e89b25d67bf81b2c9e6320","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-11 10:05:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9fea7cc1bebfae520921ef2618ccd2bae13871a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 12:18:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/57c25e37c46a4f6bad4920d379024304\n\n- grenade https://zuul.opendev.org/t/openstack/build/1e2328f1152441d9947e16d33accde0c : SUCCESS in 1h 03m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73086979128243e09b1b2f9cf24cf6dd : SUCCESS in 1h 23m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f99dfa09a8da41d68b75adf1ab367037 : SUCCESS in 1h 06m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b079264a080427f99b73dbfd21fe83d : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac45996a81eb43aaa4720856530beb68 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16e34ed0b426442a9763011c0a47d561 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/246d40a5729d4188a9ab1d6f42de04fe : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/470fb399b2164851819bfbe4e77d01dc : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a95ce29e804933bc39653112a6cb68 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f314086a3a5c4f069a5e118a5ec74027 : SUCCESS in 14m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0478fec6af47493a933f85ba987dd7e9 : SUCCESS in 10m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b04d0b004067494987cc75209f10156e : SUCCESS in 10m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/70c1434a154d4efc97e9dddf0674a37b : SUCCESS in 8m 27s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/49d1750c5c90458a81e65760d8f7d94a : SUCCESS in 1h 55m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/111a793f252d4c529b96670622baab01 : SUCCESS in 1h 51m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0f918fca7a8e4239b135231fa74b55e8 : SUCCESS in 1h 38m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/945b7bea77aa4d5e82d782d96fbca536 : TIMED_OUT in 2h 04m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3c1395147b6748e690b30a51dd4a769a : TIMED_OUT in 2h 04m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2546d345fb24791978e4ca8b5d5802b : SUCCESS in 1h 02m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1645068bc06348b8b1c13ccf22cca7a5 : SUCCESS in 1h 48m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a95509f7e3a64aefb7c9a08122d5a0a5 : SUCCESS in 26m 06s","accounts_in_message":[],"_revision_number":2},{"id":"289a558ec0c95f38a8a0d1398dd779bf9954878c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-11 13:42:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5eaf144ee4b13f43b0fec944f43d95d93c175233","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 13:51:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/512df354b85e4e868fab5d798afdcc5e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/896ddce0e3c542c781d66cfc3a235670 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2105469db45f45c9bc206f6e8bf93253 : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7a83ff0306c841eca35922e5b7cfd1a5 : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bf6849bc3b04cd4a96f582459e5d586 : SUCCESS in 9m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9f631b2293eff9ae85873747f12ae22f64baa7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 15:36:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45096faed13d4d8f9c4eb19cc49299ae\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ca70d9ce03249c7b97e3d6cc415406a : SUCCESS in 55m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/98dfe9669f8f475d9136ffd0f2db5728 : SUCCESS in 1h 32m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/850c72439a2c4663891e891b6d6ba3b3 : SUCCESS in 1h 02m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b36efeb0f1842dfb8739f90a5f8ffd4 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01e87159cdab479c9bd2f39312da933b : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce18a274a95a4c4eacf89f37620a94ce : SUCCESS in 5m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4b3b7f86d5a456a9456e48dda5bc07a : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30f194c346404e71b1d935e5bf03a925 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa543dfef4dc4b64a90b50174d8b3b90 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7e3479a242f140c29f2ab655c14447ac : SUCCESS in 12m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/033a9df55c614cd3b7659656cf3d98a5 : SUCCESS in 11m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03ade1d8db3248508467dc77c02b9f69 : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f655b97acab342349152bf99bbedbeff : SUCCESS in 8m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b9367652ef7c48818c619a13d031f806 : SUCCESS in 1h 42m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/46094f72e0d44a03be1c4c0dfade72b2 : SUCCESS in 1h 41m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a7c16447b5814d76b15808bf5cf2dcda : SUCCESS in 1h 48m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f82c40161b974635a050b57eaa504287 : SUCCESS in 1h 42m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/617d4d9a598640b9bf1010c719365139 : SUCCESS in 1h 50m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/268703e272264bb294bd32d0c975b91b : SUCCESS in 1h 10m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8aa8771b7a7a4a568dd0435ac866ce0a : SUCCESS in 1h 46m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1ec692c726cf490498028eaa8c9771f9 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":2},{"id":"97887298287073b2f3ae202f46c72cc756700173","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 15:37:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40d734a70b2d835dbe5170fd4c6f7d789dc461e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 17:21:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b502e93cf4784e7680e154141684bf78\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9fb05e63ea24857b430b57b5fae5021 : SUCCESS in 53m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/280b91cd0f454c98bb27ffc687810892 : FAILURE in 1h 08m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/076323dcc4d04ec4983aae038c906702 : SUCCESS in 1h 01m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/059497f472a3430d90c979c965676a31 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f905a3d705f428481d0492a739c9d6f : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6847b44a759144eca617230525246211 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fa37399aad8e42a0a3a92630e15e3ecf : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09970b9a6fe34cb09cbcca8c8c7e2868 : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39a52e116a2a471ca550692c038ebf32 : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9d1656cda44140528a69af56d5d79d3c : SUCCESS in 8m 47s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/047d5243491c4481b0efc87fc983445c : SUCCESS in 1h 38m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/562668d5b0e24c7dbffbaf57c7f767c0 : SUCCESS in 1h 39m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d438bea2c444488bb347a1792051cc96 : SUCCESS in 1h 39m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5d1e7ac77c74a9283e7af6c855b0098 : SUCCESS in 1h 04m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/240d00ac914b48eabb3c7f0b1dfca1dc : FAILURE in 24m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eba6345975f045f5ad191b2f1a15fce3 : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a606ce52f0079de1a786f3c6c9a2b9f5359506c7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-09-12 05:19:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d3604af00b1ec12cbd84c6e628f7ea6837d4101d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 05:32:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f82894622957495db9058bc1c718a256\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6f9627b2998451ebc0b085d2297333e : SUCCESS in 9m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8affc1fd5c6543bb8bb9aadc1fc3140e : SUCCESS in 11m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b169cebfb97c4e27859f51ea666badaf : SUCCESS in 10m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/51a273e3350047b1a52b5087e1a2cd1f : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5a4e2cd9d47e06e5ad9d6ba4e28c1b63b3cee2aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 07:18:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2692ea11ef045138d7034d3627d5227\n\n- grenade https://zuul.opendev.org/t/openstack/build/30446a19c323445a846df1a7b2421947 : SUCCESS in 49m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/726bb6fb4db34d81ab28832ec5bf5b48 : SUCCESS in 1h 27m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e0209f1d4074ebdab701123d991c464 : SUCCESS in 1h 05m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61fce191ddf34de48959a62e141872f1 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ea932cf654c4302b4a2d71bf39ec540 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b12ed2724234a8ebbbeb374e5c7c59e : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57dcb8f382d1482c8f0dd282d9b37128 : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a2b9d3c8f21492a86d73f04534b21e0 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6006ee96510743e89e56d0debb1b1f69 : SUCCESS in 7m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e62c4384b3f64d00a42c5398dd029b70 : SUCCESS in 12m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/64c3961e89154df6af24eacba494fc32 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fab3929137ac4d92b5e05684a73fe587 : SUCCESS in 8m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f035dcfd4bf54b76911ce94c2317a804 : SUCCESS in 7m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8f7c55fda04d4214b19b7327c7a46d1d : SUCCESS in 1h 23m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9881484483b7462ab1dc8752713bde6c : SUCCESS in 1h 40m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/02eb65e90a4e41b09ef6120f6646e596 : SUCCESS in 1h 37m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/665ad0429e994266b28761a6f9b9a152 : SUCCESS in 1h 32m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c5f712135ab49d9b39035a23e731878 : SUCCESS in 1h 53m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc4b2159d1b34393a3baeea7e2ae3e83 : SUCCESS in 1h 00m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee5b40cb3d4e40e0822d362f45092823 : SUCCESS in 1h 50m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/06a9e3367622473e98ae62f685cd5a5f : SUCCESS in 23m 06s","accounts_in_message":[],"_revision_number":2},{"id":"6becca4cf538dd9edc6374b0583a01823b5129e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 07:18:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"431d3f9ce2150d5cb71db1374f568d82a674ba14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 09:38:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/244e843569eb4a81854c2b571d392147\n\n- grenade https://zuul.opendev.org/t/openstack/build/4a276a82930c4a46997d63493354bb6f : SUCCESS in 1h 06m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e5f7758190f44a293d78b45a764499e : SUCCESS in 1h 35m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a081f5cc15a24b0091dac59c6b44feeb : SUCCESS in 1h 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d58fe1c49654a32a500b0aac25ada23 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a10a844b9e4b45d78f76e5b3a7c7745f : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faef5ef5035b4b7e908f3a8f8de5a0aa : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de5c7c308250479dae9530a0a0a63258 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9135be489f24791b54ef7df3386c005 : SUCCESS in 6m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2dc2e9486e15430a9d9c5b0dbd5476b3 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2129a2e9c3344e709628de29da3484d0 : SUCCESS in 7m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9545b180e3fe4fab9db16a2d4344d399 : SUCCESS in 1h 38m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0793e34ac37c43f2a9a9d1bb1cefb46c : SUCCESS in 1h 41m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/012379ca3c5b4633b361b5b1a723bd46 : SUCCESS in 1h 59m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1bac69a63e5c442aa4768bc51bfe03ef : SUCCESS in 1h 04m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/618e53b2d41d492e952cf2da9010524d : SUCCESS in 2h 11m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4f80306175aa4bbbac7c44910dc01744 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":2},{"id":"33865f2daaa4836a7972e5bd06b19a5d78a33107","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 09:38:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2132a2747083af7e33aba5f1a7ba3c2723b96991","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 09:39:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bbdb939b000448e80c75d19d6b557ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa4b2d0a4df41599bf0540c26b75ccf : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e5a8918e8708e00d0ffc5179164b9965b22c5f28","revisions":{"c104c367da1d3a26371388b08ca082b337b8a95a":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/889055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/889055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/889055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/889055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/889055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/889055/1"}}},"commit":{"parents":[{"commit":"4b85a413a2211979344ea7cb1f7241d767e9aa51","subject":"db: Replace use of \u0027Query.values()\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4b85a413a2211979344ea7cb1f7241d767e9aa51"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:38:02.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Pass case.whens as positionals, not a list","message":"db: Pass case.whens as positionals, not a list\n\nResolve the following RemovedIn20Warning warnings:\n\n  The \"whens\" argument to case(), when referring to a sequence of items,\n  is now passed as a series of positional elements, rather than as a\n  list.\n\nChange-Id: I4df3a52333d71809d6ff305a786ba6ae03a230ca\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c104c367da1d3a26371388b08ca082b337b8a95a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c104c367da1d3a26371388b08ca082b337b8a95a"}]},"branch":"refs/heads/master"},"e5a8918e8708e00d0ffc5179164b9965b22c5f28":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-24 17:08:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/889055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/889055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/889055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/889055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/889055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/889055/2"}}},"commit":{"parents":[{"commit":"c70349d8234fe65c74a01a5005b6c5e1f9006ba1","subject":"db: Replace use of \u0027Query.values()\u0027","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c70349d8234fe65c74a01a5005b6c5e1f9006ba1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:38:02.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-24 17:08:56.000000000","tz":0},"subject":"db: Pass case.whens as positionals, not a list","message":"db: Pass case.whens as positionals, not a list\n\nResolve the following RemovedIn20Warning warnings:\n\n  The \"whens\" argument to case(), when referring to a sequence of items,\n  is now passed as a series of positional elements, rather than as a\n  list.\n\nChange-Id: I4df3a52333d71809d6ff305a786ba6ae03a230ca\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e5a8918e8708e00d0ffc5179164b9965b22c5f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e5a8918e8708e00d0ffc5179164b9965b22c5f28"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
