)]}'
{"id":"openstack%2Fglance~889056","triplet_id":"openstack%2Fglance~master~Ifb818df8ca8500adcf1354eff0671f5762473110","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-12 10:57:43.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ifb818df8ca8500adcf1354eff0671f5762473110","subject":"db: Replace use of aliased keyword arguments","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-12 14:54:08.000000000","submitted":"2023-09-12 14:52:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"889056-sqlalchemy-20","meta_rev_id":"c2c0d71e808fa2bfb96fc7e4a59794a58b3ecec3","_number":889056,"virtual_id_number":889056,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-12 14:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2023-08-14 07:25:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":2,"date":"2023-08-08 08:30:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 14:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-11 09:49:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 14:19:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:03:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:09:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 06:14:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-08-08 08:30:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-08-14 07:25:54.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2023-09-05 14:19:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-09-11 09:49:20.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"120125b496d31c821a1e1ed0889e18628ce55bb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aabc83bf8d9ad4cd7be4953499462a88671de269","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:03:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f37e55779545a7bf27c2507441db0b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/132f882cc4aa4ea6b944bd6b547f121f : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efd11f51e66843d2b4913df440099461 : SUCCESS in 13m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/988ea03f08414ab4a39b10d1cf5bb3fc : SUCCESS in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b593ef929fc20221faca6dbdc60a4ceadb03e868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:09:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/59808f33f1a547c1bd8b420ed817244a\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e1a664ff89846f8b1062d091185dd45 : SUCCESS in 58m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/19950d36dcda42bf8b5187bfa939c542 : FAILURE in 1h 58m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/142061b260224002b6b6787865954efe : SUCCESS in 42m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d072deb34784becb1a6003daa4e5c92 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a05b3d93c6024e99843c8b19b1ab2d82 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3e0fccdd9f64c10be954691ba26d5da : SUCCESS in 5m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02e62e75e98240348cd42506e3751f3c : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a9b0cf7c2464821a942a6dbeb76cca9 : SUCCESS in 7m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5946e65ebff34b01b78aaa32e86c093e : SUCCESS in 15m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e8209984f7fa4e399502c7e4acd2f843 : FAILURE in 10m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dd63db99b7d9497280f9785ddb53912c : FAILURE in 8m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/15d398ca7f344e7a8286cc132b3872f6 : SUCCESS in 9m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c1b21f4e4d8c42a680033aa3ccdb4a7d : TIMED_OUT in 2h 03m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/727e992ea56e43f5a94a279444674872 : FAILURE in 1h 21m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ee3917e228e46bb9b5bc040ee460ee5 : FAILURE in 1h 52m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0f232e541854d95a8137de22df60b7a : FAILURE in 1h 59m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d703d2bd0db04395b9019583e16ddfd4 : FAILURE in 1h 46m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ab69a6b8b4146669f065f34581dd6f2 : SUCCESS in 51m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d361ac3d95b40ae80dbe3111684e175 : FAILURE in 2h 11m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e33c8578f6442b6819120c5d4ef91ae : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":1},{"id":"764d89c332e5af3b5cd5f485d4432afefaa02e55","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 17:10:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f84c676e10698c7e022cc049c7860ef0f0c7fc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:33:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"da211ab00248d622cc7d1cc3478abd3bb8d59c85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 22:00:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6711423c4dec4f7c87bd8fc5937499d9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/831563f1ff4b4c2a97374394a7e6c371 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/760da42741ac40c89044ac1a78d048c3 : SUCCESS in 12m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bda12ba654a941c0bd71b966000bc96f : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6de4c423d12f59f105bbb09ec5f660187a945cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 23:47:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/343dc652359b4e83a030fce9ef1992bb\n\n- grenade https://zuul.opendev.org/t/openstack/build/5a920d1267164318a3ae91be43e7f831 : SUCCESS in 43m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5bc0b75038d4d5ba40ae61127ec1a21 : SUCCESS in 1h 25m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64673682304f443e8bf5cba0e198a4d7 : SUCCESS in 1h 04m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b192388a6cd54912ae35c6f63a045b43 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3af0d2298e54736ad8d387c7e91ca39 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4391fb75798418585748e941c324fe7 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9fa151ae72e4084942c50735b461f8c : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a84cc6e743b492eb1a782cbc8e93d33 : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6c0fb7df161e4cbaad4d9dcfb12d67c6 : SUCCESS in 14m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bfeb35907b6540df87850a92601952f8 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0206c5b016df40cc98a7cab84502fb60 : SUCCESS in 8m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/024dcd24048747f3bcfe09758e8ccba5 : SUCCESS in 11m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/870edefa948e4bc280eac9684c15b007 : SUCCESS in 1h 51m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/95edcda56bd54b0fb128d0034ee6bda5 : TIMED_OUT in 2h 04m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/443bd4038a3a44c5885327fc2c20ddd4 : SUCCESS in 1h 18m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1618b75c54f64a94a4ce160ae6f7e696 : FAILURE in 1h 35m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b29e4e2c3abd45a193c27ed6dc6c68c1 : SUCCESS in 1h 22m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a819e65ebc5f41a09c77d17530b7903d : SUCCESS in 1h 10m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8b96d2beae8240a68df0fa79a5394053 : SUCCESS in 1h 25m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/426491c483114f058cfcc920461c0039 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4eadc1bda2b98249cdce26df0f9b871282d59fdc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:16:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb2c2db9b3d3f74854a9b792768d45cd8bff5738","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:50:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/274875ae35854f418499978c7e671824\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9050ae5629154ae992da51a2df859d29 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6086bb68314d4736a3f60e054a900730 : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1ffdcd80891c4f488f3d99d292c5ec5d : SUCCESS in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"dec4e9114d3af74f1517558744b5a0b554ea7d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 12:23:33.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26d0cec1aa47498cb01c21057780f987\n\n- grenade https://zuul.opendev.org/t/openstack/build/6af4f5f440a6468fae28f49de236852d : SUCCESS in 1h 01m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d295f3ce2254fa7b8ba93adf733db87 : SUCCESS in 1h 26m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4bcfdd451034ae8b416d0d421a87bd2 : SUCCESS in 47m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29ae282b8db84de2a68ccb673c7c978e : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/065c475134d143bb95d3be67f51ee61b : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c5fe78a7f2249d0908795f5b6bcfef7 : SUCCESS in 6m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/573b6048d6c54b0d9dd6a6e01473b959 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e90349d236d4683b9b76025c362a78d : SUCCESS in 10m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b92e16296e064fd39bf500641325972c : SUCCESS in 15m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/45241ab0dcac4e6c94da487fb84ae146 : SUCCESS in 9m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fae899ae03df4cc0969cd738e2693751 : SUCCESS in 8m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/17eb54fcd0df4ff2a6d0feb9e801c56a : SUCCESS in 8m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/67a9f6856f43453aa21d657e076eba07 : FAILURE in 1h 34m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/95995a6ac9e747faa44e69104eebcbfa : SUCCESS in 1h 18m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5427d20669f44fb08a9278c4249c2b66 : SUCCESS in 1h 30m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f19c6fac89184132b821647724b53d71 : SUCCESS in 1h 53m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f08a398923d3411cb7d2ad0a9172a7af : FAILURE in 1h 27m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/494ea23301eb4f9891d7180bdecf52ce : SUCCESS in 56m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1d02afcac7f43fdb4afb7de4a69990a : SUCCESS in 2h 22m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f6598416111b419d997d34a9a03c673e : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":2},{"id":"dbe3baac6b56852794b0a61d473318f1cb957c20","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:33:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55c0139374e240d3218168131336f71e6d27b943","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:12:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ab8d46ae648485ba7e0a2ad1450b861\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f6f9eb062e643e0a59f7af2a22a6588 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63ea7bd2d0f64999afbfea7fc6ace308 : SUCCESS in 15m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21af9db865234beaa0aa1b47a77860fb : SUCCESS in 12m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4800a691f30d79f50a3a888412f05b107d02e9dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:46:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/40ca93dd83fb4146afdf3d3d9456a3d8\n\n- grenade https://zuul.opendev.org/t/openstack/build/d9e66b0415df440fa037951bd14afbb5 : SUCCESS in 1h 07m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/12b606785d104949b0683570285e9479 : SUCCESS in 1h 39m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00f5683623a342fd8988fdd3095ee2ae : SUCCESS in 1h 05m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501b4287f1e44ba7a3e28028836b8d14 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4b9df02542a40a1ba2205568da1f2e4 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77025ff394b14a77b5352d7c89512746 : SUCCESS in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/700b5dacc7524ae6bcf09940b6555613 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6010137bff7542869000cbb7897d840c : SUCCESS in 6m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/df0f935f5203411eb6d47958063738a8 : SUCCESS in 16m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/31e0eafdba18437cab874f135f259322 : SUCCESS in 11m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/abd8f59e009345f1a3117b076cc1525d : SUCCESS in 8m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cd2961962d694cc0826a8bb06a7a592e : SUCCESS in 13m 00s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2991aefaaec04ccb924e77a7e649a74b : SUCCESS in 2h 02m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/562d76d18b3946aebdedcc4ac1a4baf8 : TIMED_OUT in 2h 04m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/77c58dc554fe403eb0b84cfbbea225bd : SUCCESS in 1h 30m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b0e016c258fa4c849e7a073cc7f7b324 : SUCCESS in 1h 34m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/450d28a8a85b41a185a30fe7db194e0c : SUCCESS in 1h 37m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9ab18db00fb4e38bae8efad285f71a0 : SUCCESS in 1h 23m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/47d22bbb3baa4078b1f709e54f25d5dd : SUCCESS in 2h 01m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d1fb2cc23f9b4d0db1e20924d687b26c : SUCCESS in 35m 37s","accounts_in_message":[],"_revision_number":2},{"id":"cd2ef258a8ba97c79dfe25c66f57c0464e5ee6b8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 09:04:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"54133e7ccb8a8d8fff7eacbe3f5e5ec0b13d3fc3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:24:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86795327128b419eb4f01627757537f1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/293a318913b049d3a8edaaa2dfa3234c : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da1e4f103a8746a4ab4c3c1ec55eea34 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4c7a43612bf4b5f8da5b9e0bf20ab72 : SUCCESS in 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"80ba2ae75645529cc8e0cd5d90ea3a3c60338d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 11:20:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb0f33803e2b405cae811f101589aa4b\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a1b6676638e40ac98161ffa26c02593 : SUCCESS in 1h 08m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/703a41383c644063b29f5b519acd1425 : SUCCESS in 1h 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f797e760f7fb4e7590f11e62f78d3b10 : SUCCESS in 1h 21m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/136ce63ae5c54b7fb33d76309d0d02c8 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e51831e05cce4741865247b00818223a : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08739fffeef74a199331b6b5b0d43142 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f6998c731e7498ba8ae06f94f981620 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b30d7664b33433bb3a2f6d5212a209e : SUCCESS in 6m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c20239d768944b53b5d910c712acb64b : SUCCESS in 13m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d45f1c34bc0549eca3194daa35562692 : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/914b0b9dfa1b437798269e865b8a4024 : SUCCESS in 7m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c549c242692040a58716fae67530e766 : SUCCESS in 7m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/20a7449b55824b75a97d6e999b4f8f79 : FAILURE in 1h 32m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3913529b5b5d47b1905fc11c99d54f08 : SUCCESS in 1h 17m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a6d0a22833f9433ebed0f5a76020843e : SUCCESS in 1h 54m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dede67f4ef2049cab52023aef7d11553 : SUCCESS in 1h 21m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f653cdbbc3634a5da2d1b06e94986628 : TIMED_OUT in 2h 05m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/378f071956c242d8a02a98cc188d4059 : SUCCESS in 52m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/805568ea2ed0414082fd6118302870d2 : SUCCESS in 1h 58m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38d5bd5da90946a9a14fea1e3827c3ec : SUCCESS in 48m 10s","accounts_in_message":[],"_revision_number":2},{"id":"1e3adfd49df7040eb94124c6d46f71429c96e255","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 15:18:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93ee226024090ed7e952bd89056b58a9888f8411","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:53:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0e467ea75b84ae3b5ee1904fb0ed1f0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d93aac23fe4c427b9ba7be02f3e71f2b : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3779cfccfec436592ee1a469a7a999f : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dc033b0ca065491eacab1e77e00279e1 : SUCCESS in 12m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"87cc1825ebd5c023d0f43f1ea1346c52f5c5b294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 18:04:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d57201283f92470f9dd43761ff2736c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca01e48abfcc4eac8535c783868ceb59 : SUCCESS in 55m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2a483325b0af42d1bde5fd1ea9781c1d : SUCCESS in 1h 29m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1cc22644ef3b4e6ea3b2dbbcfe4dea40 : SUCCESS in 57m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfb80ef743d1431aae01b55704b0e09a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69a02241726b466497018b442dfaa1d0 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea1e1ac5a18e4359b12bba6487bfcf2d : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b16eab9de884c83b3709bbd0a38d3b1 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac0db624f274b58a1d79a44bcefcf0e : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/71075b8dd1df41729fcd6b1fe5aacc6c : SUCCESS in 15m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a96392ff9ab4ef29b1c873c9225a98c : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/25242daadbb447a7af516dc389886cbb : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8b2bf71549eb471687022135ab0ba91f : SUCCESS in 9m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eec3dd8643a34cb6b0f37fcb03ef0bc2 : SUCCESS in 2h 03m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/638c067ee72a430ea6ef86ae3bd180f4 : SUCCESS in 1h 47m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db292573fb3e40a98f647091f1b22940 : SUCCESS in 1h 51m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/87811ab1d205451eb3974a6d440c2def : SUCCESS in 1h 39m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/90044c6be49c4371a49ab0600f7545b5 : FAILURE in 31m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/703976f17d9c4bacad13276b2606781b : SUCCESS in 1h 19m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5ad68f3d18bb4c4786922373c1b23739 : TIMED_OUT in 2h 33m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4d9ab605295b4b4d84aad67fc4f1434e : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e15ef203ab866430b52605d85f0d93d4823cc187","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-02 06:14:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3a809c83e9e43779f5829c8b6da9d45f35212a37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 06:30:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61b53dabee3f4cd28543d84687ec0256\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fdfbfc76b42c41328c4c59e6963348a1 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c19d799781e4eb896804ced28141d34 : SUCCESS in 13m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/869221598cc2485d8d40d5703f0c0fe2 : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0da68752bdff3c8866b0355cf64563a12d29f89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 07:49:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b5aabff1108247ee803e7efb0c95da90\n\n- grenade https://zuul.opendev.org/t/openstack/build/b4cb8ddc6274404f95d5091c0fe30711 : SUCCESS in 57m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/078fdf1a96d74d7f91e8edd5c1ed1147 : FAILURE in 1h 25m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89927e2b92474fc2b5da61582a419ff7 : SUCCESS in 1h 01m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4957c6544c0f431f9136f578013d8969 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8562092df0f14436b07ace3eb009bcac : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f638ce10e9ac4601b05d6053e5e9636b : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e5f9acce358e4d4b8477a0b7cd8f5f91 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf468fc4b26c4bc69d1710de09df4746 : SUCCESS in 7m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fb580765501946d398ac60d3a1aa8071 : SUCCESS in 11m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/85215f5eb68244c6ac057e46cd31b1f2 : SUCCESS in 10m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb949712aebb4818a69d1035f8c8cde6 : SUCCESS in 7m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fdec09c186ce41368ca3b50ef5b09e91 : SUCCESS in 10m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5b45380e46914c7aa127b3a09b99cd7b : SUCCESS in 1h 32m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d3e18b2e098f450395c799453a1ae3d0 : SUCCESS in 1h 26m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a84284549cb44921a89bf1e54ec4d4cb : SUCCESS in 1h 18m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e84df634fa00497f8cae09b38e146b40 : FAILURE in 56m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53610eddbc944aa3ad5bfbeb4797ef09 : FAILURE in 1h 18m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f7789dfe6854c96abd1119d4fa4f1af : SUCCESS in 46m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78c69db8597749be97890753beb9e6ec : SUCCESS in 1h 22m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b034ce233486494fa79f70606a5ca42d : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":2},{"id":"1ff1d7d6ea49bd68f1256ee87ef97e0334f216a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:24:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2820004089d732b6d966b1126f18ac1c9c6999fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:30:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c342363a8561803b2ccac016036c93da05ca2cec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 08:37:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6eff57b7c9054818bc4ac3e3a55364cc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5fccd2ab7c1b45cb84424fc8cf22e302 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8bf92c4c21854a638c86d93378e676d3 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/08b2835ab2214871afbd4da011d26c1d : SUCCESS in 11m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"74136d698192f67b3a8ad2d201dd443586cb489b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 10:46:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed01a27c32de40868d0cf9592213c463\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c91f8b119ec402aa552fb885363b158 : SUCCESS in 48m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b9caab53ba1475ca08aacf7744e2222 : SUCCESS in 1h 23m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4699feef9864873b4e5a23d523a4fd1 : SUCCESS in 51m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56bb84f211048979667fc596fd5eef9 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4c61cf53ccd4c53ae51cab4cde62755 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/422b879536d44432b3e64847ddb470a7 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a52a7e5de7cb4a4e94a39d420bb8ae65 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0053c1c3c44de3bc4792b7e2127800 : SUCCESS in 9m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fc85fb68d30c46f2bdf8e052d601f9a0 : SUCCESS in 13m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/610f4afa094f4b93bc1ae19dac3ef8f4 : FAILURE in 10m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f0a4a7f61ed3490d811d6bc0908a6eb9 : SUCCESS in 8m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/264ee20eea48475b9ece6c07134aba3d : SUCCESS in 10m 09s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e74b76ad570b4a5fb7dd5a30287a50ca : TIMED_OUT in 2h 03m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/17ea2e82970d4704aaf17bbc53eefadf : SUCCESS in 1h 17m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d8cc58c0bb0a400cbee167e4ffb8aa5c : SUCCESS in 1h 17m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7d071c02694f4e25aff40385b67de1e1 : SUCCESS in 1h 51m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bd30c4b21def4384a7f1c72456863f37 : TIMED_OUT in 2h 05m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23587d091e564b34a859f79a83f06e4c : SUCCESS in 54m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a6e41143a0b84daea32662d97e2759cb : SUCCESS in 1h 54m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5d4012a3e0a147b182e8339a1d0ceec9 : SUCCESS in 44m 55s","accounts_in_message":[],"_revision_number":2},{"id":"dc51a54748c27095a50e065e0276bd52d5897741","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 06:00:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"25b9751258066d8aaeb1b6b543dabc7af82ba650","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 06:15:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/111e3bdb243e43818517ebacf310da38\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d4a5e6d9027d43459b69e5a9843fe183 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0657649ff773425c8e2bb346d5672710 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/59a61872a36d48dcbddae9a771937a69 : SUCCESS in 12m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8410d6e4291c3f0f94a9eb1cfacd2d4d3e35ab0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 07:50:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eda5fdc42054a3db8cdbe13fa3a58c0\n\n- grenade https://zuul.opendev.org/t/openstack/build/8911b94233f944258ba52b4ac45668c6 : SUCCESS in 1h 10m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84ecfbf768fc403c894f0138d3889a9b : SUCCESS in 1h 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bae09eda008c4ed7aa3103c8c223b56c : SUCCESS in 1h 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2672fd80a9184d048b72e2f7654ffe26 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0eeb09b9c2b94cb4bc6204e558d4bd91 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcd2d2e253aa4cdf8b792766dbdc074f : SUCCESS in 6m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fb589b9e861480eac8f9201cec8e698 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b1e56fe270a486ebf685d663734ac2e : SUCCESS in 6m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b1660ba7bb8c4b238e40fcefb01ff820 : SUCCESS in 13m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/555c1c47b1c844d699dee33938c4a323 : SUCCESS in 7m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2e7be5ac44584abda4286c10af6fe2a6 : SUCCESS in 9m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2190273910534b9da92ed1ab5e5e8fa7 : SUCCESS in 9m 41s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/474b20c05f104fc189d04bba021c7828 : SUCCESS in 1h 42m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c771d09f3dfd492989e6d2caaa95b374 : SUCCESS in 1h 18m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb2c78d37b13455cb26b85df928f11a6 : SUCCESS in 1h 06m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0f94c27817d47d4a9a6ec85ba52637f : SUCCESS in 1h 16m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d3f98022bf7749e396ff203040d2cb43 : SUCCESS in 1h 31m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59927aea6b9f4e4c8bf84a804048daed : SUCCESS in 1h 12m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48ce277b7bf84c4b9d36a38027de1bd2 : SUCCESS in 1h 38m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9788f73ac06c4cbbb26f479ed83b0ecf : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":2},{"id":"69db116e29e572de4a972d8826f43a33b2fa5c19","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2023-08-14 07:25:54.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"176ea666a841caf26b379659eae436720cd0e44e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 14:19:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"162999bb6c808211b8a16e1be3ca9f87fff97eb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:58:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ca2b5f8bb482b1f539dc391cc03213561ff46a4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:23.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"6a8a86e2c524215c2f1b4a4df358e1971deda295","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-11 07:44:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a88993d20de53ed501bdba038728acc7fb2b6d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 09:47:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c257a8f9b2f4809b79f66d6af381a8f\n\n- grenade https://zuul.opendev.org/t/openstack/build/bae286774625428a992eccbf4f375e15 : SUCCESS in 1h 00m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7dec6f4a0ae14cc7bf822d89b0092a91 : SUCCESS in 1h 20m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0edfe2847f8b4d1f860c0d9635509394 : SUCCESS in 1h 03m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91f21231a437489fa9490b56d3fa8a1c : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5a17701cca94d2cb696be6fa1f3978b : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c049ca017b9f4cc1bf16e47775e3fce9 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d26590cf84c4c5b96ff8c088a9692be : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/71930f92f10046339c20738aca77791a : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8924267e001449f8be6b0a64ef3a5b56 : SUCCESS in 7m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/33bcc2f59c7a4816b9e7ad7ae65588a2 : SUCCESS in 11m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7b2e5803ce094189a44ccc7369ddf356 : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4baf7f509484b70bd29a23f7cece4e7 : SUCCESS in 8m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/56cd8b74b8aa458da635b2e1c93ef03e : SUCCESS in 8m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8e5f2da1f68a45ee9844edbfe146ca73 : SUCCESS in 1h 37m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3ae8a6fdac44b14bc792ff3c01c0dcd : SUCCESS in 1h 15m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4186ed50882b4c23bbca7b46692748b4 : SUCCESS in 1h 36m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5fd304dac17546f59e00802ad11f097b : SUCCESS in 1h 42m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d0f52c9176745aaa260c0da1c1aee68 : SUCCESS in 1h 57m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7080d403d03948b2b2f7dab0a2205713 : SUCCESS in 44m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ccbb2b596b749a982b620115ab21932 : SUCCESS in 1h 57m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ff8b0354cef4bbdb2909c333b9ccbb0 : SUCCESS in 25m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a7c4a6e9668e57e90c6238236ae34791f6c2cb16","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-09-11 09:49:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"60f6e616d2e1999c9773dc72b7edb3b52ec3075e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 10:02:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da98e63be144408b875b260dfbec24c3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/26e81266e87d4fc594b2b263f5d6f561 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7aa2d42882ac4238adaff2e5d5f44a31 : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f2ccfdec355f428491f67db205542918 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6708d0f421ae46caa32c190f193dc932 : SUCCESS in 9m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1878b06e201e083c2201ad520e5ecc514ecddb46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 15:37:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"251e63bfcddf5c252c445bad824b77ec0893eff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-11 17:21:14.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/889055 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"2167c524f654222c436a178178f95dbe1516a4fb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-12 08:20:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3dd442fd8787019116b11b1bc8db333597f31e12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 08:45:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0475300defe54c2fbcfc8bd17776a890\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bac4895a3cca47ecbc3d51762032a72c : SUCCESS in 14m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8319f631c40490c899adc2196ee9472 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2fb29003a30e4d57b456fb1dfa024cc5 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d23178da8a1e4eccaf6f7ff4bc1f6fa1 : SUCCESS in 13m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c72d0b8b81c463316113da9ed48a944c698e0583","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 10:37:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/90e38ca17f0f4576937660c0c0fd47b1\n\n- grenade https://zuul.opendev.org/t/openstack/build/da5ffe5c707e4d7ba1fcf12927949129 : SUCCESS in 1h 02m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/162091e54443487a9334aba511f2451e : SUCCESS in 1h 30m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b5147ef109844b9b2675791c8e14769 : SUCCESS in 48m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5056949a50cc48c98173496226387124 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/236360d326844c128b73c13076d19778 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f189c3e766c485a86f27f5eac163ad1 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba7437394b9644ab9f6f2b6ae847d856 : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2c3d10026be46efbb7b6424250aedb7 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f1610d4cc14c049ed506fc859c20ce : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/52e95e3556d446f786a48f338a1bb22e : SUCCESS in 12m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1b93b31808a547e182b875f6fc0f253e : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c14258fff1a641939ac30a11b7c089dc : SUCCESS in 9m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/145869f5260b4866871ec499fbcff299 : SUCCESS in 8m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/003e13cb8d3a4f54937bef6ad5685080 : TIMED_OUT in 2h 03m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9b2a8d5c960402a8994d78441a3c859 : SUCCESS in 1h 38m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/96b30ae340fd4453be1e0f8cd28f7e84 : SUCCESS in 1h 22m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/99b819ac324347548f2505aa3e5fbd0c : SUCCESS in 1h 56m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/be4c834715d84e73bbef25ca299d0e78 : TIMED_OUT in 2h 05m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47f7f0db00b84cbb9a525a066f9b72b7 : SUCCESS in 56m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9cf7946b80c346979fcbe8f0b2224c6b : SUCCESS in 2h 10m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f27903876290405b911c012e74c10d0d : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":2},{"id":"115f449cb8187dc6c1536d4e6726fcd6ea38d888","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-12 10:57:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a320cf4c360a9a6b2692ee265f9e2f2c8d4234bf","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-12 11:03:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb99050bae7f5766c73487885e3209b96abba7e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 11:22:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7f4a5c27315449ea6dd0d6885cd148e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8915fef5311d44e088ec77b19ae1747b : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/953978ef153d493eacf485382b140ff9 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/529c21d5ea7e4525b9b72306c4149819 : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a106c9bbf55a46c782f7842b96efe6d8 : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"48aab9244d9cc9cdebfea00a4751ee8efca4aa66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:06:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d73ccc88aa4f999a326ae4e4830e5a\n\n- grenade https://zuul.opendev.org/t/openstack/build/94a61891481144b2a61ae243fac506b1 : SUCCESS in 47m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7325d55d2f9f4cca80520bdcd97277d5 : SUCCESS in 1h 07m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a26a345348194b26aea4e24221772fc5 : SUCCESS in 58m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd93ac7a08b483783de2cbc748792e0 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e5cff2f21ba4c1ebd60bba182007f1d : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7f799c8001448a4bdd4a629f2eb5a9c : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9bfb9a74827442a6ad119b8d7654993e : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3352d1ac85374613a75cd1d980bf051a : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ee7ed9dcaf14d88a522dd7e94698469 : SUCCESS in 6m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/74c2cd41cbc54b449ab864a3cfb6ce5c : SUCCESS in 13m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff02d8d076e94f159b0de5e883d956ff : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec65d6fd80274f37bb4483997b91db80 : SUCCESS in 8m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c9cf268598734682889a0a70d40ed765 : SUCCESS in 9m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c526875054254206afac56e9beb756a7 : SUCCESS in 1h 29m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/90770a79cb1c4189ad8edea3047780a2 : SUCCESS in 1h 39m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8222acb594444bb1b1c7a37a70b4604d : SUCCESS in 1h 22m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b3975bd703274d4dabf80b22fcaac396 : SUCCESS in 1h 23m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d394fa6b15047e393f1a9137ef99a13 : SUCCESS in 1h 43m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3670d0cfaf274daf8efa3a34e61abb8a : SUCCESS in 50m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d4a9e5dbc2b40c4ab251c53c013dd86 : SUCCESS in 2h 02m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/34a5af418488434f8129bf9675f30b69 : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":2},{"id":"b83b98244f562c9143b7ead0480d4f969c73d82b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:06:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f4bb85c09d6bf267d03cc6b15f56a5d66f1de53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 14:52:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae79ff6b828b499f8cd5d7a71d3e6d04\n\n- grenade https://zuul.opendev.org/t/openstack/build/84033dd1c64f40118d65590240a432e5 : SUCCESS in 1h 01m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3c44884db6b4020973617e8d55f7b92 : SUCCESS in 1h 40m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecfbb443166d44aab89a71bdc2478ec8 : SUCCESS in 1h 08m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5429a9d3ef984c938903ce02a93c2877 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54d428725e1e4b97b241aa70374ed028 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b3a2545e6b343ca9ab06162bb41cee6 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a0a554b4480427ab62e54032a40db9b : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e96fdc606ea94ed5ab4f05aa8028e5eb : SUCCESS in 6m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5f685037bfa149939900c9a74cfdb711 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc24be1350ee4d24baf84cecbf51c36b : SUCCESS in 8m 31s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c0fc546202a341fd8912bf26264208a5 : SUCCESS in 1h 30m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/073ec7af49de41a19a70a4ab8756d234 : SUCCESS in 1h 32m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d45a533b2d0e4b00a44007553aa743b5 : SUCCESS in 1h 42m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/106b7c8c156c495eb04252837543689b : SUCCESS in 49m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5dbffa035b794fada1c857463af8d29d : SUCCESS in 1h 43m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/514d03d95f234bc89f717b48a0bfa037 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":2},{"id":"b2f5e4e65da890a4cdd8074a049a5200f598cd74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 14:52:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c2c0d71e808fa2bfb96fc7e4a59794a58b3ecec3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 14:54:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/030ed005f4234bd5b4cc52351fd79533\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e26ef77a4ec4034b4a0c96ace9a69d8 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44050a2f753e4a3af76475c27f3adea719819cd1","revisions":{"2d4b891906d1b01ecf52da70c16e1b32cb69a693":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/889056/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/889056/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/889056/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/889056/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/889056/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/889056/1"}}},"commit":{"parents":[{"commit":"c104c367da1d3a26371388b08ca082b337b8a95a","subject":"db: Pass case.whens as positionals, not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c104c367da1d3a26371388b08ca082b337b8a95a"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:39:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Replace use of aliased keyword arguments","message":"db: Replace use of aliased keyword arguments\n\nResolve the following RemovedIn20Warning warning:\n\n  The ``aliased`` and ``from_joinpoint`` keyword arguments to\n  Query.join() are deprecated and will be removed in SQLAlchemy 2.0.\n\nThis one is a little trickier to grok, but the SQLAlchemy docs explain\nit rather well [1] (Look for \"Legacy Features of Query.join()\" and\n\"Joins to a Target with an ON Clause\"). Essentially, because we are\njoining e.g. the ImageTag table multiple times, we need to create an\nalias and use it for both the join and resulting queries.\n\n[1] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.join\n\nChange-Id: Ifb818df8ca8500adcf1354eff0671f5762473110\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d4b891906d1b01ecf52da70c16e1b32cb69a693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d4b891906d1b01ecf52da70c16e1b32cb69a693"}]},"branch":"refs/heads/master"},"44050a2f753e4a3af76475c27f3adea719819cd1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-24 21:33:58.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/56/889056/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/889056/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/889056/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/889056/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/889056/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/889056/2"}}},"commit":{"parents":[{"commit":"e5a8918e8708e00d0ffc5179164b9965b22c5f28","subject":"db: Pass case.whens as positionals, not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e5a8918e8708e00d0ffc5179164b9965b22c5f28"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 09:39:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-24 21:33:58.000000000","tz":0},"subject":"db: Replace use of aliased keyword arguments","message":"db: Replace use of aliased keyword arguments\n\nResolve the following RemovedIn20Warning warning:\n\n  The ``aliased`` and ``from_joinpoint`` keyword arguments to\n  Query.join() are deprecated and will be removed in SQLAlchemy 2.0.\n\nThis one is a little trickier to grok, but the SQLAlchemy docs explain\nit rather well [1] (Look for \"Legacy Features of Query.join()\" and\n\"Joins to a Target with an ON Clause\"). Essentially, because we are\njoining e.g. the ImageTag table multiple times, we need to create an\nalias and use it for both the join and resulting queries.\n\n[1] https://docs.sqlalchemy.org/en/14/orm/query.html#sqlalchemy.orm.Query.join\n\nChange-Id: Ifb818df8ca8500adcf1354eff0671f5762473110\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44050a2f753e4a3af76475c27f3adea719819cd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44050a2f753e4a3af76475c27f3adea719819cd1"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
