)]}'
{"id":"openstack%2Fglance~889057","triplet_id":"openstack%2Fglance~master~I547bd2f441085fcdc5dc4a0e58d27d11e8a8b223","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-13 09:34:00.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-09-05 15:17:35.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I547bd2f441085fcdc5dc4a0e58d27d11e8a8b223","subject":"db: Don\u0027t pass strings to Connection.execute()","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-13 09:35:16.000000000","submitted":"2023-09-13 09:34:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889057-sqlalchemy-20","meta_rev_id":"51101f657d506dd02794cce1fabfd5e63dfe6f4e","_number":889057,"virtual_id_number":889057,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-13 09:34:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-08 08:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 15:17:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 15:17:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}],"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:27:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-25 17:13:20.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-08-08 08:49:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-13 05:46:52.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"}],"messages":[{"id":"355c9d3ea899ffc4c86dad302e9ae3a138f32b94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84a0ffd13e06b6f87ce0dc04a99267c59370b44d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:07:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b05b8d77450549b1ab4bd8525e400ad3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8970a43271984f68a6ebceb3f5e858ed : SUCCESS in 10m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1bf690af8954b6e85e48698223b09eb : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c45735aa2b94a2cae5c255047c62b98 : SUCCESS in 12m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc109a2fa4a0b7f3b347a55911cca9ff5ad13fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:27:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac06cf73326e4076801125c8baba3357\n\n- grenade https://zuul.opendev.org/t/openstack/build/204174d8c0c741b3b090ec16ce190861 : SUCCESS in 1h 10m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/541506ef44e94396a202d7f30024e274 : FAILURE in 1h 34m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b713b7badea42a58001ecf51648ecf9 : SUCCESS in 53m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8f3aa57b6a450dbd1ce27b3bce5b21 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/341dec6c09514588abc0acff3f4d5ffe : SUCCESS in 7m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/760cff8bc2b3413888ca82a7bdf5e4f5 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daa2878ad36d47eead44bd703a029c7e : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f0797293de4025914381abf542abee : SUCCESS in 6m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/aca3be12c3244808a5628fbcbd860467 : SUCCESS in 13m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63a25f2a9af84947bd550ace9a1ce49f : SUCCESS in 10m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0de177d6f254527a0b6f6351edc81b8 : SUCCESS in 7m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/707eeff2d5a84fd9bd2e4f087a656209 : SUCCESS in 9m 40s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4d1cdffe97df4cd39c20f794956b69c7 : TIMED_OUT in 2h 04m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/08f8f7798bdf4835ada1008974dd3390 : FAILURE in 1h 40m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/add177a46ca140ee9d84598d0a01eefb : FAILURE in 1h 42m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2902c036bfb5491fa6f823eb5e29a0ba : FAILURE in 1h 28m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e510a3271c0642698e47cbefb804f8e2 : FAILURE in 2h 04m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de3b86fc50774ef1a4ad14e48a62bf92 : SUCCESS in 59m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d01f727f911c442683c7a8958e50fdb8 : TIMED_OUT in 2h 34m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ae92b9cf8e8c4a448635a0821dcec422 : SUCCESS in 27m 26s","accounts_in_message":[],"_revision_number":1},{"id":"7332bd53acd5af89623ebf4eff23a0e2c05a3bb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:34:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a8794dfb2ab47dfaf8598a1cb13db02f092b1b73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 22:01:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/633721d21e504ed3b064089087851865\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6e3accc98af4bbb86e19bc3d95c1a6a : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4c274b0797dc4e6dbfbf8f2df14e9d02 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f3f4b69d4bfd4f8dbd088ea5378bab07 : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1845c5ffb49864889aa7f1be0fd16c85f2ec2103","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 23:42:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/769c9e2ec6c84c10bc5b829a08625ccd\n\n- grenade https://zuul.opendev.org/t/openstack/build/0a1c5a4d73e649f5bde08505590fe980 : SUCCESS in 47m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e3053fe2bf04e0097b4c7f81d1d3b82 : SUCCESS in 1h 31m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e503dc98ddb448494dc6585762dd0e7 : SUCCESS in 44m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab161404692d4966be2da7854d44b650 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cd2ea43bca64f7a91421d7752aa5a5e : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fec27767634043579f4adabd286ea515 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d63c45c7585e42bd922e8008bda2cffe : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f84b225755aa4ad18dd9b72dbd279290 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/10a70dbf226445c99e82a8fb45726ba8 : SUCCESS in 13m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1df0f1225a804374b138eb4e34ce0b50 : SUCCESS in 9m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af34ae807e67464caf88c2b0a461f348 : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9ba0ce4f510e4a49b08c1b585eaaa8e3 : SUCCESS in 9m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a094b9e1289c4582811625cb0eb5f712 : SUCCESS in 1h 57m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c3ebfae699d749439ae88ab8e7eeb4d0 : SUCCESS in 1h 22m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d90f077735224c50b55421bc0be85c61 : SUCCESS in 1h 32m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e58aad3ab1aa422f9bd03fb552c5540a : SUCCESS in 1h 23m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4836173ef8ca416183dd307d3ffa397a : SUCCESS in 1h 31m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/066206c3e6c44bcdb6c5e7f827d0a00c : SUCCESS in 1h 07m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7bef613e26b64ffe9dc353a4931be8e2 : SUCCESS in 1h 38m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ab1cd108ffc34633b2a761fe762a1ad8 : SUCCESS in 33m 36s","accounts_in_message":[],"_revision_number":2},{"id":"84fabb43d368dab06bd33310e2b071321e7e9c80","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-07-25 17:13:20.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9fbe1bc1898cfd711c33f5389440fb1751d127d1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 09:03:38.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"28efce8eaa906aacdcfe2a2bcca9e2df7d31bb50","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:49:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c6cef5e0c8c635c3638358a038a31b54d0272d8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 15:17:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2cbd8a0e3611c700151f9082a86ce79ccd3cdf3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:58:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e94d0f8c5748c17afa6d7ba4856367a3f9b30708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:23.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"e428e2e592b690ca830aa538426524ff31c47352","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-12 11:03:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f950f868b4f89125eeb6b307a70e932c9c427381","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 11:24:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ea9356b6eaf4224a57e5847fe3f8299\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cddd6dd840e4e1198466f6b5cca7733 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2184a59c0c640f588f2a34f330e8610 : SUCCESS in 10m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe77ee2349104a41b837a01bf82c2fa8 : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d7919c2020634f239cf0982db1e3a7a2 : SUCCESS in 9m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c85760f56b3c2ed9eec7077ec4c0e436a9a8235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 13:03:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4f9d89d1aff4e4e8f18d71936098cca\n\n- grenade https://zuul.opendev.org/t/openstack/build/e6263b16c39f4bd98ca1176bae61246b : SUCCESS in 57m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ced31e145fa542c8892f7d514e2c57f2 : SUCCESS in 1h 20m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2496673b3c7044a7a0a74bc6248fe541 : SUCCESS in 45m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39bb76a71a9f4bcf9cfe5bf91459a9af : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d12bcc28c9a54e25b4cea8b7b0b8c198 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0dd5981a59d4a91b12f35831c2647f7 : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a9bebaed52f64d9683430c8a467a0edf : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5eac89df49504e68b60de742c321b075 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ddc9c744d2f41859f0f6c1adc6aab6a : SUCCESS in 6m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9ad88c2e7b644749a28b81c866a97607 : SUCCESS in 13m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2b26e2271ca4b58b70412d23ddda167 : SUCCESS in 8m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1a39f917646843969785b8da52add432 : SUCCESS in 8m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7f0a0536a60641f59d8783d5fa85b895 : SUCCESS in 9m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a59caee05fc243dca4b08e7db58821db : SUCCESS in 1h 44m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/01ed5a33473a48598d05caca7ad84a09 : SUCCESS in 1h 40m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/112c7dcb362a4732bc2afef84dd480d5 : SUCCESS in 1h 09m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/af16be75e3a846b890067c53ee2a58cb : SUCCESS in 1h 46m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd2035c07ba647d085afd055493420c0 : FAILURE in 1h 53m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f93d986cb951452f8436b5954c55b59c : SUCCESS in 58m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/59a38e346f2e44a1bb41a79dda1f8740 : SUCCESS in 1h 48m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b2ff18203b1d460786977c4d21b1524f : SUCCESS in 24m 49s","accounts_in_message":[],"_revision_number":2},{"id":"7593b16be528dbf154bec3a22b0140cd7c4c5652","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-09-13 05:46:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ce2bd82764d282a95c741e3f6e3a1504dc5b16d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 05:59:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/707199323d7649069f38a6e4663b9298\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31fef62d0fec454ba0dc817c5fc549ff : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d644159b76324261ad60a411e87719ba : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ada7a8ae17b41e39b6fd1c8eb671873 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e0bc365895964206809d50dfa36ae7e6 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"62dd7e9613e71f593737170005d17256bf8cf17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 07:40:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d59f99f9e08041138b35468e7a4cc5bc\n\n- grenade https://zuul.opendev.org/t/openstack/build/65b49afc29c045ab9bf9b297cf1525ea : SUCCESS in 41m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/093f8293c43f46309b244b8a9262b02c : SUCCESS in 1h 14m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec4f5c1b790441fc92773378c8781efa : SUCCESS in 42m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418b8909a9e041e29c0394d8020d6ba2 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51d8137397d14655a2f26f3b72b156b7 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df612f4344094beab197a4df48d8c618 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6c312ddc2a784a199964b2f8db0920ac : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c74c47b6f2243c7899e24fb6cc958ca : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb732184a20b4d5e87229099e35a8a6b : SUCCESS in 6m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/394e5e6e63ed411c8df2b73baf22a06a : SUCCESS in 10m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ee3080578f54e8180530cf3809d4781 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6956b2a83c1c4d6cb1ddc6d09a4522ca : SUCCESS in 8m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8bb9a4529e3e4cadb36d3e64e0bb42e4 : SUCCESS in 8m 14s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9aaabd9d4bab4f3ca6b27094e2e63ac1 : SUCCESS in 1h 31m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f925da5becdd4f17b85dadd6bb2a4142 : SUCCESS in 1h 15m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5b537fdf1fb74a7e8828af17787b0615 : SUCCESS in 1h 46m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/74b46b8f0b5e47d9bf068eadc0866b0b : SUCCESS in 1h 31m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4303be10a5fb4b2bb33a03dbe007960f : SUCCESS in 1h 34m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dde1a95791534ec294d78953596c1371 : SUCCESS in 1h 05m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b1c3c8ad8bad420884a980d5af356514 : SUCCESS in 1h 47m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78ac19bf7c3745ffbca268ea87699266 : SUCCESS in 22m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1c1bcafcc5d9edc3bf8b123232a0d375d0426dcb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 07:41:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d780988662396b6d5b607bcc6085a72e047abde1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 09:34:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a235daccdd2b44d8a99e88635843d8de\n\n- grenade https://zuul.opendev.org/t/openstack/build/ebb3c3bf9f274584bf30508bf7a93886 : SUCCESS in 1h 00m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ba1504ef636546cb9583c4de129c74e2 : SUCCESS in 1h 31m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6510d08bc7f34a55bd27f80c42f12c01 : SUCCESS in 1h 02m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99f183517aa54ccb92c429472ee60e2f : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5227aecb962b439882dd7aab242103cf : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/076acfdd16cb42c59a4d3e40f3dbb12b : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9b091078612e4cd89c211ebced58f165 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db44cda3ee114253874722aaeab4df5e : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e4ea7626c574c38b8717fc94b9c19e0 : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/344c330999a842bc9ca9d11da294e2b5 : SUCCESS in 8m 12s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a5d008b244644531b61347efd8873c6f : SUCCESS in 1h 14m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2a56a18f48b94a8891a38d83611e5202 : SUCCESS in 1h 40m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/67a997ec4bf5451eab2fdccdc0c1de31 : SUCCESS in 1h 36m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7780270899864872be3d8102a65561d3 : SUCCESS in 1h 03m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9982bd2d3c2144ee8b29cd20f00d403c : SUCCESS in 1h 26m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/379a0e6da6dc47568ada1aef3e303b20 : SUCCESS in 25m 53s","accounts_in_message":[],"_revision_number":2},{"id":"cb2979d36c2934a2fb87aa84e8c843f2d266490f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 09:34:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"51101f657d506dd02794cce1fabfd5e63dfe6f4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 09:35:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cf9c39f64ae413892db10f329cfcfaf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fafc9b9127e54fc1bfff493d96965eab : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"25dcacea7c389819d988f9340d93c69cb11ef298","revisions":{"d4cbf0d03deb7773f8d6a6519d94910205373040":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/889057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/889057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/889057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/889057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/889057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/889057/1"}}},"commit":{"parents":[{"commit":"2d4b891906d1b01ecf52da70c16e1b32cb69a693","subject":"db: Replace use of aliased keyword arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d4b891906d1b01ecf52da70c16e1b32cb69a693"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 10:28:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Don\u0027t pass strings to Connection.execute()","message":"db: Don\u0027t pass strings to Connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nWe also resolve two additional RemovedIn20Warning warnings along the\nway, since we\u0027re already touching this code:\n\n  The Engine.execute() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session.\n\n  The Executable.execute() method is considered legacy as of the 1.x\n  series of SQLAlchemy and will be removed in 2.0. All statement\n  execution in SQLAlchemy 2.0 is performed by the Connection.execute()\n  method of Connection, or in the ORM by the Session.execute() method of\n  Session.\n\nChange-Id: I547bd2f441085fcdc5dc4a0e58d27d11e8a8b223\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4cbf0d03deb7773f8d6a6519d94910205373040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4cbf0d03deb7773f8d6a6519d94910205373040"}]},"branch":"refs/heads/master"},"25dcacea7c389819d988f9340d93c69cb11ef298":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-24 21:34:03.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/889057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/889057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/889057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/889057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/889057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/889057/2"}}},"commit":{"parents":[{"commit":"44050a2f753e4a3af76475c27f3adea719819cd1","subject":"db: Replace use of aliased keyword arguments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44050a2f753e4a3af76475c27f3adea719819cd1"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 10:28:18.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-24 21:34:03.000000000","tz":0},"subject":"db: Don\u0027t pass strings to Connection.execute()","message":"db: Don\u0027t pass strings to Connection.execute()\n\nResolve the following RemovedIn20Warning warning:\n\n  Passing a string to Connection.execute() is deprecated and will be\n  removed in version 2.0.  Use the text() construct, or the\n  Connection.exec_driver_sql() method to invoke a driver-level SQL\n  string.\n\nWe also resolve two additional RemovedIn20Warning warnings along the\nway, since we\u0027re already touching this code:\n\n  The Engine.execute() method is considered legacy as of the 1.x series\n  of SQLAlchemy and will be removed in 2.0. All statement execution in\n  SQLAlchemy 2.0 is performed by the Connection.execute() method of\n  Connection, or in the ORM by the Session.execute() method of Session.\n\n  The Executable.execute() method is considered legacy as of the 1.x\n  series of SQLAlchemy and will be removed in 2.0. All statement\n  execution in SQLAlchemy 2.0 is performed by the Connection.execute()\n  method of Connection, or in the ORM by the Session.execute() method of\n  Session.\n\nChange-Id: I547bd2f441085fcdc5dc4a0e58d27d11e8a8b223\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25dcacea7c389819d988f9340d93c69cb11ef298"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25dcacea7c389819d988f9340d93c69cb11ef298"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
