)]}'
{"id":"openstack%2Fglance~889061","triplet_id":"openstack%2Fglance~master~I0de54294cc311b32f0956942a62deb2e47729a4f","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-18 20:41:14.000000000","reason":"removed on reply"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-08-30 09:48:42.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I0de54294cc311b32f0956942a62deb2e47729a4f","subject":"db: Drop the convert_unicode, unicode_error flags","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-19 00:41:39.000000000","submitted":"2023-09-19 00:40:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889061-sqlalchemy-20","meta_rev_id":"9db6fb0c765367850f2357a775bb06e992d483a8","_number":889061,"virtual_id_number":889061,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 00:40:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-30 09:49:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-08-30 09:49:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:20:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:11:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 06:16:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-08-30 09:49:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"7d44241e9731e06f793bd5eab9ac211d908b6cbb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b6e38673fe77843248b7c2350390a7578a1a9f68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:20:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ce81ce9e0614344b14a220d03ab48b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3bbbb9d66cb4a888e3f9968839de6f5 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f307f03d7e5a4ff6bce071de5600b938 : SUCCESS in 12m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c898f92966644f98bf478bdb569ca7df : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ac4387e2b20085e52e110416e782a8cac7e6843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:11:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3779ba37e1e3444583d66d637445d2a6\n\n- grenade https://zuul.opendev.org/t/openstack/build/228bbeaa2d0a49a5830c7e8981cf6c6e : SUCCESS in 58m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ebb0036617944b48ba7fc872714f823 : FAILURE in 1h 34m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79982610eb81443a8630fcd9b6a2eed0 : SUCCESS in 59m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c36c9adc134ff4b988d567b76ae1a6 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/154ee8c389ed4037b9899f97c8ff29fa : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2194a506df374811a7de57bd5b432670 : SUCCESS in 7m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f010e6a6a7144dc9566f50faf1092e5 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7c7b062df649dda4450b20e8153d55 : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4e3888ded891491cb6327cfb4d018ddd : FAILURE in 15m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e60f039d53cd4ecbbbca11a0fd4430dc : FAILURE in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/093af03686524410a3dda3e1cf878973 : FAILURE in 9m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/92b6b6235eee420c833c6996f9059017 : FAILURE in 10m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/21204806d67d4a4e893ab6130f55d02d : TIMED_OUT in 2h 03m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d3f3adcda0fd4a7caad44c6c89aebdca : FAILURE in 1h 43m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/089017d6084f49fb9a0c07189f54b74e : FAILURE in 2h 00m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5503d6321b2f466fab705a22d1d94bb1 : FAILURE in 1h 54m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/12eee47ddb514a39a8e6a19e7fb0f7d0 : TIMED_OUT in 2h 05m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/876dce7173fc4d139464f3fb47915212 : SUCCESS in 50m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f7a784c0129d485482521f542a20cfd8 : FAILURE in 2h 08m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c8a14218bbb74cd0a61154d55c024403 : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f203bc43d234bf924e10e594489bdbeb8c083c52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:05:52.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"755558891469137a24c0d4bf9ea197560ae9a546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:35:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fe6ffacd54a41ee8598e54b417fe136\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6406afba02534598aa1d73dc859efe9d : SUCCESS in 27m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9d225bb9ed240c896e5fff3a5a32129 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67d50d7eddd64f329e4f01875fc02672 : SUCCESS in 16m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fdd5df4e1f6b5917f6bcc3ab2b0fda5caf888adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:28:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/04f60fc00166430881e090b9976ab58a\n\n- grenade https://zuul.opendev.org/t/openstack/build/49de9453bc20469d9d71bf6dca0c3247 : SUCCESS in 52m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9867d9662e074b48b1181e5f4dc09fe8 : SUCCESS in 1h 32m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4378c6203ad47eaa6c3f1c047dfe9f4 : SUCCESS in 56m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/138d2a02a51643f5a71e81406e596cb4 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6863b099424e428db054d94ff63ec685 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8a89c179685427bb8ed1201a7e0949b : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7c03499008c41b5bbf73096d7e916fa : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ede05f9f5a2426c9947df0d1506f5b9 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ddd7bdc5a2d84f3d9e3726933a80aba3 : SUCCESS in 14m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/489c8e99e4b640ab8913790b3139c287 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c14573dd73ae4300aae44f48549b20a1 : SUCCESS in 9m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f7e489c2be83487c8b2fda11c29970fc : SUCCESS in 10m 09s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1cb59e8edca3489ca40df2e06f0d5077 : TIMED_OUT in 2h 05m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d0602505674c4adf907874bb6f8de384 : SUCCESS in 1h 26m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5f091f0d8f9840b7bd0aeec134c45c25 : SUCCESS in 1h 39m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4232c64c6866488d9e58e809867012c7 : SUCCESS in 1h 35m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9004e5a92ff84fd6a83f1c3c2a9fcdce : TIMED_OUT in 2h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eaa6136773647bb8b31bb0f47f1c99a : SUCCESS in 1h 06m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ca0a67a191e46c98dd230c253a8439c : SUCCESS in 2h 02m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c3c4796d07114d9293ded04379e0c716 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":2},{"id":"bd79fd55768a04772c431fa81e652bab51a85308","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:31:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f681397118d5a20d6b817fa79eb374d76ee8dc74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 14:47:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85c607f4740e44cb8c989c9dc0b0dfd6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17932555be5c4b0eae33a8772575bbab : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dfdbb244125e4d45b1aa68def6f370a8 : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb955cffc99f4d24839ac2e2d7d3884f : SUCCESS in 12m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bace9747edeed086d28978f2f93cec7270cf1cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:16:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab879c21ce824ab1a7e7520201c245f1\n\n- grenade https://zuul.opendev.org/t/openstack/build/ec4540b9148746308d12329cb45f967b : SUCCESS in 52m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64170cce8c9c4d29af9034e4c1c25c83 : SUCCESS in 1h 34m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04d330736628453ea236b0304ed27392 : SUCCESS in 1h 07m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c2e5ccaeb449a89abb13612fcb834f : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e91de8a385d94e7d823cdbc788f54b3a : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dc6355187d94611b5b20ec0722fa4b5 : SUCCESS in 6m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6f2a4088c364b169f42484230c5e399 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eef4b6c8aa8462f94e372068c36e32b : SUCCESS in 7m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/dfe310e921e2406aaff2047fe3b99692 : SUCCESS in 14m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c7ddfec56c24597b22431d50723a118 : SUCCESS in 10m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e875776f302467db58fd30942029819 : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b5838c2bccbd43478dcbc49034c6b46c : SUCCESS in 14m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e8da7ecc054e44b3a29fe5dc6f81ba7d : TIMED_OUT in 2h 05m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fdf33c034b284671b419bca7f7344f5e : FAILURE in 1h 31m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe669788f2694aa1a0e7f532fd61e013 : SUCCESS in 2h 00m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b6987d8cfa0b44b7a9d6b5cb383fa6e2 : SUCCESS in 1h 42m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1313f542df454f6a803f996b9a854117 : SUCCESS in 1h 50m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a50c1ec9f8242a5bc2518664e66d23b : SUCCESS in 44m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4a5ac70a469941b0916e3a729a31a624 : SUCCESS in 2h 03m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b33c49290b144078f81804775257a06 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":2},{"id":"eb3f7ace787e0d6fda70e850717d7c5fb6abdeb2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 08:58:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d576ee4c738a33e0423e5e041908fc0c948f0580","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:15:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4316027c499497bafd1e58fe651c129\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c71ab31ad4614673bd1b4ca8fa430a47 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d56b6063ed094bad944b5a1b743260d3 : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/04bb13e9b64945baa8cefea212f76aff : SUCCESS in 6m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6419f6b33f2ea7570e6af8ef2046509a63579800","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 11:01:22.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d70dc484285c42f69705cd1ecd2bdeb4\n\n- grenade https://zuul.opendev.org/t/openstack/build/36c64cc375b446a6a31b78ab32c34025 : SUCCESS in 53m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/295b42d65a2a46b5b03caff79ce88c84 : FAILURE in 1h 51m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2fec6c0d65941af9ad9c030914f8f4d : SUCCESS in 51m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dedbcb21149473184f4c7196fc55e8e : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac70831fb0974e2b83fcbcf550a1a4f5 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5df580c9a184e45b624988fd5034e87 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c285649e3d3d42dc87504820bbc5e417 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15deadf41f742b299c08a4dd089fc14 : SUCCESS in 6m 58s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/28625e92a6304970a79c798b94e10839 : SUCCESS in 14m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07e1a816921949c2b896768f14b764bf : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/464e2ecf22f049769836ff1f6e21c708 : SUCCESS in 8m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e4d0e391e61b49cca797cbc7ee66d0f5 : SUCCESS in 8m 56s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d8c3f5735ea34b2f957601c2f350ac09 : SUCCESS in 1h 34m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e8d3c6336c9c4b2782e4cf9b8e62f5e9 : SUCCESS in 1h 23m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a03a3310d99a42ebb67bb543eff8e9ac : SUCCESS in 1h 15m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d596817ca6d46b29e8fa6ef0ac5e3aa : SUCCESS in 1h 36m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df11b5164d184d8cadac733a60ea2f49 : SUCCESS in 1h 52m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa0ee3188c6348a68c44547e95a78de1 : SUCCESS in 1h 06m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4ad5349f8804154b66a073c8a517014 : SUCCESS in 1h 57m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8841d99e074040c6a0ae964ddbf02942 : SUCCESS in 46m 51s","accounts_in_message":[],"_revision_number":2},{"id":"78842da93570b57b96faa6bc1aea39afc2aa2072","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 15:16:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6e49c7852ce1e387e6764cda6c6a59e193c6708b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:37:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81158be5f7c44bc59d414aa507015f4f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ca5d5151917453e9ef137d0d722b325 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/02f19a1850ad420ab0496af7a7451abc : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4cb11c072c5447ec90d597b19361bc1e : SUCCESS in 6m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a4fb36093915acdeea7e5424a5ecbeac81a623e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 17:33:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0bae6fe2eea4e4a9da8db2a8f957507\n\n- grenade https://zuul.opendev.org/t/openstack/build/75cb68c4654f43da8ab5ea461e6a3335 : SUCCESS in 1h 13m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ca6c68edcb846079a49a8dc89f1e4a9 : SUCCESS in 1h 17m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f342f898162e4b82a6f33a7ddcbb123b : SUCCESS in 1h 05m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/645291a5d1c14f5a8736eb97b5b08ade : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40fecfd14c444b4f878d9259c6f8ed06 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44a464bccc5d47648b040072e857f292 : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43fdc4ee91c441469111ab7a2f74bb7f : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fde9801387d4ce3a7b650f2b93d0c21 : SUCCESS in 7m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e1f1dcff51f74f2ab0cfd29b520ed290 : SUCCESS in 13m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/61b988c87cc14ef2b3a29e12a643faa8 : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ef5e6de3e0d44ada3f7dd2ee3ba7641 : FAILURE in 9m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6a046553cdf140c6a80e3669a19bba37 : SUCCESS in 9m 15s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/605382dd3fe74213a0f82d950e25bf76 : SUCCESS in 1h 57m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f267316b05df4d6da0174e48c9a6b70c : SUCCESS in 1h 26m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f90eb3a02d994059bde56a94685331e0 : SUCCESS in 1h 18m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73801c38af094b659e22c4fcfa4fa941 : SUCCESS in 1h 51m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/796efa857cd848ec9c7522aee687f559 : SUCCESS in 1h 50m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77028d99d8d44b42a8156c655ed287bb : SUCCESS in 46m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/781a3acd31b64f788ef959b14698f6d9 : SUCCESS in 2h 07m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a080cbe18e5c40538e826818ae2a7b4d : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":2},{"id":"637ae35d6cb08166d7e9eefee457571f08bbdcb8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-02 06:16:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e4d8452a5fa75bea61e270475f6a7ec860a2e39f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 06:35:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95a0dbcdaa2b4d46a50efe3c38f7cf08\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cba3289a378d4fc9a78354a5ae59311b : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/83897e0e436047d3bd432d40fc8219c7 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/74dc76e1acb840c5af4cf21021524d25 : SUCCESS in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"42458a40818dd3d14236117d5c4eb6b5a8515dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 08:40:15.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81361aab767e436cbc433921a8dfc799\n\n- grenade https://zuul.opendev.org/t/openstack/build/fd7bb7876d524109b269ccf0c0f963b9 : SUCCESS in 49m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/384011a13a6042718ac18790ac89257c : SUCCESS in 1h 31m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85812795246f40f9823d8607c842cb3d : SUCCESS in 45m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c91608584d64b86b06907e1aadcf13b : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a3077b5ac3e4e539fd5bd24ab58314c : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef0aa0935cc04dd2b15997585a5e19fc : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0be3b6a7a314b3f9eb295cfeb193100 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b933c0bd9a5847f98ff4bd0815eb7504 : SUCCESS in 5m 58s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/db00c416a36448a19644892e0a65e22a : SUCCESS in 12m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/431227723bde4ef691d695963c1b5bb9 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/209d103665bb4ac590a68538fa5734b1 : SUCCESS in 9m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f1f25ef4ba1d44fc8d5517292bf29a0a : SUCCESS in 8m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1dac9343d5a849dd9fdbbc4c68d54121 : FAILURE in 1h 55m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bdc258afbd50417580d9b4ba6565494c : SUCCESS in 1h 28m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b82476d2eb944702a11fa115d2c302da : SUCCESS in 1h 08m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/27e9a36b0d5e4d558ee9fffa4c67a0d7 : SUCCESS in 56m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8670db8337a24dcfbe3f32adaae46d3e : FAILURE in 1h 32m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3df48816fe554882b2bd672b34c55e04 : SUCCESS in 1h 07m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c6508d6c41447e4a8bd7f0c48e9a043 : SUCCESS in 2h 15m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0847ad2d0195466b9c1ebf71244cd85e : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":2},{"id":"02110dc7df1c8224ff7f837f9f847ecd4d6c43bc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:16:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab7ae43c307303b2332d52c8ddcd56ba8736bf0c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:21:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9036473c3c4cc0095ff2c818a4078e0a8868ca97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 08:28:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8fc5c5749874a71944029e6cf994dc4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3846842ef78c4b7996fbc6aa8847212e : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4533d6c13a2411d9d7ff5cae0994104 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fdd5699cc9147c1a1d82bc25f69140a : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0bd4779a23a1148a48b3f34ef36b5d5a04181374","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 11:16:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f87f1e2a697345c4bf0a9fc832b033ed\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd4df88c28274692a43910e5e781aca3 : SUCCESS in 43m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1929cfab8c7b478395d5481fc01000cc : FAILURE in 1h 17m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c55087fd26541da80f4024eb4be2732 : SUCCESS in 46m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a803726d1ed44d92b88f0849c65e1842 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5af4023e48e413392b15ee6938293b6 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/958ec2053139452ca950f71f4755f7a6 : SUCCESS in 6m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3b5f7f5915f84fa693e626ce240d8f30 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e44db6cc604e08823e4fac2c4f6fc4 : SUCCESS in 7m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/84d26b4de3b9457e8aef40b0a3d37d66 : SUCCESS in 12m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e05b12b3f8de42c2b384f819c1b0629c : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d39d377ad997421992bbec3eab5054f8 : SUCCESS in 8m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fc12716ef1d2423d909b1f8a76978004 : SUCCESS in 9m 00s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3702c02e54274a908b84548795155830 : TIMED_OUT in 2h 03m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7c0e2b309fb340a59117636a07246168 : SUCCESS in 1h 35m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/81d0b8c460b24df29d66c1ba4b5107dc : SUCCESS in 1h 35m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/99fa6eecd1ac4eaf99da222bdfd21bdc : SUCCESS in 1h 31m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b0b39a65bfff4189b3fcc6e3b1991028 : FAILURE in 1h 19m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f626845bbfc47199c02137488f09371 : SUCCESS in 36m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1babd93799e4409193a1ffcf737754c6 : SUCCESS in 2h 06m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7142a767d4a1482a844dfb8ae51a9a30 : SUCCESS in 34m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9fc384bd69fa493e480c8f3397bf91643f235d8f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 12:27:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6a39f833120d573c374d6af1327ce23f27d74456","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 12:27:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"150968ad2b8b1523e8506f033ff23402af69b743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 12:33:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3282847a52334fb9a53d51c8d7a4614d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31a61452d36941fe9910342e9254ef4f : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d8bc78bc29aa44dd84ccec94aadde379 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4529cf6de8b54a668547d400172ca61b : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53a2e0487ce9b98202f05d27c7dd81bba5500d61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 14:22:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6673cd77ea5b40919fb518801ff094a8\n\n- grenade https://zuul.opendev.org/t/openstack/build/01ad2ea2f6934f97b37eb15604f82f93 : SUCCESS in 43m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8d31603d74e4b1386ed3dd27855d8fc : SUCCESS in 1h 15m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1468d25fab4b40af816936f909f0b370 : SUCCESS in 50m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/427d0eff5d914bf8be8cc05491ac156e : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba79acd0c8f443ec96ae44c08a48b169 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb94c9e6ae5841619937eabcd0d39341 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba82442e9b204ef2ba2110e872ca928f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ad8be00ae9b4d618f34f533dcf67286 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c869613e4b7e480eb680e4393e887352 : SUCCESS in 14m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d7086024e41a4fd198022b6d5b2001c8 : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/859d7d258ce54dd8b4180cc2997763e5 : SUCCESS in 8m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a1d8e410f36a4ee6a26ffca106f0bed1 : SUCCESS in 10m 57s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4067a708d3d45a8b783cd42dd852377 : SUCCESS in 1h 46m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d98a1e554aa42d98a57607ab7857eac : SUCCESS in 1h 28m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4c83c4462277439b857e8c43af80139c : FAILURE in 1h 17m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/99cacc1a94e94fed810cd8eb455a6e0b : SUCCESS in 1h 24m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3264b816ff494a1482346f7de17dca67 : FAILURE in 53m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/356cca5f2a104aeda2423cbdc0ac4806 : SUCCESS in 44m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3cc3664f3d3b4ff2937f6b518b92faab : SUCCESS in 1h 48m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1399f559fc2c4659a321461ae36a9acf : SUCCESS in 41m 30s","accounts_in_message":[],"_revision_number":2},{"id":"f18bc53788019e2494f5220a6b1fb7ff6676e8b3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-10 16:42:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29cb3b259bbdfc159d2c2cba988b9eeb1d1f3cc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 16:57:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c88d7d2ff9a4883802174b9c0ac1fb2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d47032b518004b90a71a40756d4ca427 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eeaf24fb16fe48aab04ca48c301ebace : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9644a3e4b044b94aff96219cf4b7390 : SUCCESS in 12m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7c93846b0af4932e02520cc17e783c4fefeeeca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-10 18:54:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a02141e90f604492836b31e08c34ca4d\n\n- grenade https://zuul.opendev.org/t/openstack/build/d641e132a08449ee8a0d7e8037284b29 : FAILURE in 55m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc6632ce81164a539397f579ece54e7a : FAILURE in 1h 24m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54a160ab054d4e0284c026acef3644c4 : SUCCESS in 45m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/023a87145ef341b79df89c41dc4bbdd7 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6614737de244ae889e3b1a778a9f586 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7d37db812304208b3d5579dab9edd9e : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd59c630d32a4322bd0707d33b7cf5ec : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32c6f3d71bc7417abada4bb9f4ee0efd : SUCCESS in 9m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8bed25e85ee34ab5bbab6d1525bd670d : SUCCESS in 14m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f35fd8bd89ab430ba0fe6a3007e8ac3e : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d0c683f658684032a1067d4ca3ac8a97 : SUCCESS in 10m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/125a7548b8ad4fe2aa8439240a68ecbd : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/64a617c4ed264580921f3eef581bae5b : TIMED_OUT in 2h 04m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b53337d1b434a5aaea050ebd6b642a0 : SUCCESS in 1h 15m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3383f551ec194ded841e7837d76c2ebf : SUCCESS in 1h 28m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/abf28f9cc8884816b4a4f36ddb9fca92 : SUCCESS in 1h 11m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/344881c454a643c9a8dba694054e19f6 : SUCCESS in 1h 43m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/964b072dd892442c9aa5c603d800daeb : SUCCESS in 46m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/823d74334a2b42fea3f685e55fa71501 : SUCCESS in 1h 54m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e17dc9c68c904818967cbcb61930bbc7 : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":2},{"id":"b771e7cbcbf5716f97e96a363c641c142ca521da","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-23 14:56:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"786a56a06cc293718401fd77723ababaf08cd704","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 15:11:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45738ac639ed411eb6b8c519ce818986\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7a40e936a9047e3a130cfe762c99bb3 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41fcfe559b854aa182b9826ea3d6fe9c : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a7be0eafb5ca4a26b0700e5d2023fe29 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3f0cde8b12204309a152a346cf3762da : RETRY_LIMIT in 1m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ffed24938d8a635d56d6457a490687a42941352e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 16:52:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/045341b3246441bd9709bf56ef68f2f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/9cbbd7a319624623aa89b963999289f5 : SUCCESS in 1h 01m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e3ff316ed59340f49a203ad6dc2c8acf : SUCCESS in 1h 33m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/944d0787bb384bcf94c1f59ac5d15233 : SUCCESS in 50m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fd7fecbd7854b7c809b45456f48274e : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ebe94674f648d49b488ffaf13a275d : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa045f62326f4cb891a30f53eca2486f : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0b25cc9b03340948822ab420fdbef43 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e2dcacd936e4baeaddfa82bd7c5dfcd : RETRY_LIMIT in 2m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5e0e8e9f56b41c0b01798db703986ee : SUCCESS in 6m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e2ac30482dfc45f5ac47bd0057802701 : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5468db265a7418498803f7bdfbfa628 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/86ad3d29980b44da92faaa1468e1c14e : SUCCESS in 8m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/067fe4f544704fa7ba9fbfd65af764ba : SUCCESS in 7m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/abb3791b7cd748ea8554bc30a85f13bf : SUCCESS in 1h 47m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f4947b4962e14aed87dac8b8d3440198 : FAILURE in 1h 11m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2601403281cd45c69a5c198d564071e3 : POST_FAILURE in 21m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b7898a7db4ba40339c28cccd28321374 : POST_FAILURE in 31m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c855e2aa975749209cbda377a50bea03 : POST_FAILURE in 21m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1b21751bce94849b1b402adfabf0a77 : SUCCESS in 51m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7e641ec140164ab886c8f3cb668ce84c : SUCCESS in 1h 49m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/62829bdcffe544a69da54250905af31c : SUCCESS in 42m 17s","accounts_in_message":[],"_revision_number":2},{"id":"22b3c9f5fe0960e0e4e6f88d390e592e33980b1b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-30 09:48:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9fc72ded43b6f1f401ed595bc5fa9c1d0f22f36","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-30 09:49:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f2868674b054a58226404535fdcae034eb30b62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:00:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24768f60ee084487b463e3dc92915ef1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43dd177239334f3fb37a1ef3df96000c : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84365db05547491ea85cea7a4cd03143 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ea8ccceaf76f4b96b4c298a68730c2ba : SUCCESS in 9m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2ae25213abc44a9fb6ba461921062d6e : RETRY_LIMIT in 2m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be0d475a615d4cffbcf8a4ca226780760fab22e4","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-30 10:46:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1cbd6e50f979ef74ca1c7ced530f5f1c979d79c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:05:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6980eebacae4d21be9e521ce7285844\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9c620c2f7a842fcafc6646114f2e39a : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0f2a2bbb51ae4ef19846d6eeac6014e1 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80bfd545171e4edda17282ba6f31f53b : SUCCESS in 12m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/06ba79dd0da0458d99412e602ecbfded : RETRY_LIMIT in 1m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2baaa2c10bfa70d2129ff1575f5a50af1ff6a5a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 11:57:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/563425c078904c4f9142d6ca1df01524\n\n- grenade https://zuul.opendev.org/t/openstack/build/c2cdda3ec4034b5dbbc84e6380de981e : FAILURE in 51m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1319e03706db4622a884f63c2cb3326d : SUCCESS in 1h 15m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ccea92c3e6d4f6ab62f8767e4d64c0e : SUCCESS in 59m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9da2bb0c017467d9c876c5266ef9270 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44acdf34c9bc4bc49b9741019d24e473 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6670066640641a4bd3394d7abf3835e : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bfac3893de14432bd4061873d54a7ee : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03fbcea335954e798dc6b935c2d94cfa : RETRY_LIMIT in 2m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b7f42b41bfb45899826b89715365435 : SUCCESS in 7m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bbb7dfd3c046488c840092200d3c89c4 : SUCCESS in 15m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/530fc653e823499e9abe19f96e41e9ff : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28b635fb5a4744db8a36195afa743d27 : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/92af59b433554d7bb76615fc60d91f44 : SUCCESS in 8m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a731da0b669e471ead48640942bc055d : SUCCESS in 2h 01m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ebee580618a54937be2784e89eea64ee : SUCCESS in 1h 19m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c449cb319f1742ed93dfcbf2523c2f86 : SUCCESS in 1h 20m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/13816f86d73b4c7ebc4a9343548bccda : SUCCESS in 1h 14m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e0926ce4279646a6929cbcce97de1670 : SUCCESS in 1h 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24f041558e954b3a94f2ba942e1be6d9 : SUCCESS in 51m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c149575786294b319469db0671c67695 : SUCCESS in 1h 55m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8834fb0c16274321b6a1cb81c44c7272 : SUCCESS in 25m 31s","accounts_in_message":[],"_revision_number":2},{"id":"c5ec1d3de043aae1b6b4ee091352cd95bd4bbc94","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-09-01 05:34:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"548bf17942b376e814a7e2d9ac6a67ef9c99510a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 05:44:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39950b8fe79f409f92fd23a60bf3e9fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a49ea294819347febc120fef6c09fd0e : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91b23f0f646f4a04ad0fcdcad2f7b01d : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0029e01db5fb454ab66c049c915f6a44 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6b8e1d6e762346359428ca204dfe2b26 : RETRY_LIMIT in 1m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a45b8bf12f51502eb9ca236ea93b760864f5a3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 07:46:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61fc41f4f619434ea9454b66642e7685\n\n- grenade https://zuul.opendev.org/t/openstack/build/e8c0e14559b44b7584cb976b5faf2d66 : SUCCESS in 1h 02m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/056e0919fb1344bda372f729f4d17712 : SUCCESS in 1h 28m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94fe41fc8d3e4fde9e88bc7af9117718 : SUCCESS in 59m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b67207b82ce4fc3a00465f93e501f7c : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/667a778e66fb409488b96c07096efaf1 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d59c1e4e8494b6badcd8fdb580a53e9 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/12150381014e43d7b75f2ad6868ebd93 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/971ee8f302a24ac4b29f41d879354c49 : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de8fbf66f28f4040a7eacb367f7e8bd3 : SUCCESS in 6m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e990c2ecd0754f6c8beab78b328fd3cc : SUCCESS in 13m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0084f622be124dceaf2cd23bfa74a44e : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4b83fc5443b4d48b93732238c4035c1 : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/19634860690b403a9434ef5da088ff8d : SUCCESS in 8m 05s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e545dcfe22314e21b316b9e59c0c72aa : TIMED_OUT in 2h 04m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7188fca56ccd4b7e9135daf9cf67ad32 : SUCCESS in 1h 34m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9b4779b3f94b4e6e8f12c4da1d2db34b : SUCCESS in 1h 32m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f558e05cfcf64c4eb4612b02b342b03b : SUCCESS in 1h 27m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a7f7cb3e97d44312bd8d3aa57d7208b5 : FAILURE in 1h 34m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2917d13ac494617a9d42f9928eb5382 : SUCCESS in 1h 00m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f63d23d1757b43f998fb0f9c2427ee9a : SUCCESS in 1h 33m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e7327c2d7c9e4a588093c60794775b21 : SUCCESS in 35m 34s","accounts_in_message":[],"_revision_number":2},{"id":"dcb65290622c0bd9eeaadf9387abff17d9f88bee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-09-01 14:33:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4605eba6b34a412b363d7993fff91cd45397e43d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 14:49:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0816088fe1549b7bee6933c5f970e85\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a6f831384c746a48eee3acd94f46321 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b8665ca20ed4b82b63fda3026d911b3 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/34765c42cc4a46848c8938900d1b038b : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1988fd7a35aa4671b54829f63dc61653 : SUCCESS in 12m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ef8e26771fc82cb7a8877dcb531827004558d8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:44:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/99e6dd13ed804126b21c56647bdf8fac\n\n- grenade https://zuul.opendev.org/t/openstack/build/35403011ab594c7b8f507ca66ac9e867 : SUCCESS in 1h 05m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/35c749af2bbc40af96e986670dc4441a : FAILURE in 1h 36m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d15386a3f6174f91a686c6777fa96646 : SUCCESS in 1h 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f07756ff93d48ff9e44f6191969beb6 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c4931772b2e42dfb02e504de6f1e6b9 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7261f831f20643b2aa5f7d1869c9a254 : SUCCESS in 6m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21a2bff9f549435b859685aadca8083f : SUCCESS in 5m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0786cbc017214a3090deef97f54d43ee : SUCCESS in 8m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7577a3b4e524353b0075a6749a116a4 : SUCCESS in 6m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b5c0d79c005f44ad9057c327147446b3 : SUCCESS in 15m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13afb0ee604946c2bc06089b4cf73240 : SUCCESS in 10m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f91d2489a864a7f8c1cf9a83c009ae8 : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c391e6be97934087bbcbccda3c5cc53f : SUCCESS in 8m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b16f2e73de054767bbe6ef27365509c3 : TIMED_OUT in 2h 03m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/44277b37443d4e2690348bdb78d813d8 : SUCCESS in 1h 16m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8cdac944eaa0473092455c9f61e79394 : SUCCESS in 1h 53m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/24bffd4bddef4ec097f848aee8ea365f : SUCCESS in 1h 30m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6122c4e36053439186e13018e89328fe : FAILURE in 1h 58m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad32cd5693104f63882483a04ae6c4ea : SUCCESS in 49m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3fa33bc4c8424ec68dc76a88a994b116 : SUCCESS in 1h 57m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/14325d07ba0543f8bc36f5361a87492a : SUCCESS in 42m 57s","accounts_in_message":[],"_revision_number":2},{"id":"10d093f79fabf5082b01e99f2fa366f6addd813b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-09-04 04:42:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6cf0028c1b57d4098d493df9842b6562851af080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 04:56:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a2afc8608054df9862e6fa082dc7c09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e078a0011ab941cdafc767bdb23b82ad : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/662b1ee2f4934e7496807766fb915ef5 : SUCCESS in 11m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/356c91ddb81a424987188e077c46e74f : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0f3cf1afbbfc40ee8d78d6505a8d0e8d : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9a836eee76abb68663a9ff23bf89913ab8ccd6f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 06:42:30.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/053faa59393b4fed93067d195a728d14\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed57171ae47d48e38b8c613e055d44a8 : SUCCESS in 48m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24dcae03a52e475c9f5ed285d4cbcdee : FAILURE in 1h 30m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dedfc34dc17f46fe83480655458cf1ee : SUCCESS in 52m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c2cbc99208a4c33acbc454c8c84db1a : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/732f435217d8437e86971e097146ea38 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78fe8bd99f964699a7a360f3433411fc : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d115463126e843e49bb5db82cb87a9a2 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f205fe88ddbb427bbce67c387cb8af8d : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a03b9758b58422f8e191dbd16632da0 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/840a4aba70c84370bf8369bf75ab740e : SUCCESS in 11m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6cd30fbb8944a4c8f0613d919314a47 : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b21d96e29dd143f8a7bbba0293fe2c9a : SUCCESS in 7m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9a43195e82674b2aad99980c0ad47530 : SUCCESS in 8m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/adb1ed95922a464db37b29673ea18f07 : SUCCESS in 1h 53m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dbf3aa398e414c68ab17caa6ab1117a8 : SUCCESS in 1h 25m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/836b2bb59af24542bc1cc6f70beb42da : SUCCESS in 1h 29m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1eba136b64ed4a0a8bc25226fac94d7f : SUCCESS in 1h 23m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2bb71458e154443e9901b706faa6e07f : SUCCESS in 1h 32m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3c655f7677a4504b27a56939ff3c5d5 : SUCCESS in 58m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ae83dfb15744ccc98aa119b5348d678 : SUCCESS in 1h 47m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a79cf91d49394c429cdc8c234bb33c4e : SUCCESS in 41m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6ba3ac7cf1b8217a529ac400917c569469ce3511","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-04 08:55:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"216cdaf345750a40537820bc5e8eae985ec842b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 09:09:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0eb1633287443a4bd13e0b9e5333f09\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b77efd49ed1f4e129fba355edcc0f676 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5fcb23c009bf4328a8da8f3eb33e194d : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bc0b86ca80c480eb17320eac5aecd51 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/03c7c850501646ebb79e50698e0310ae : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5dec09809065966a4fdb5aa803fd173cee57d771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-04 10:55:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9097ba5ebb25428281b153e7e541902f\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd1a5696f9c24146b4d2a42c6afe1ae0 : SUCCESS in 1h 03m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d100f8b5da3b4f938d39aafccdad2ac4 : SUCCESS in 1h 30m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c958e55c7e984008b94b74b0b9c54b28 : SUCCESS in 1h 03m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b950c62f219b4b1e8c78e5468b3de95d : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed8f29ff5ea4e7fab7b1b7b6ce0136d : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e730775029684b07a5ed793838e0298d : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21ec25cd836a4590957a9257c641eeaf : SUCCESS in 5m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2e428d894334d24be3d290d52156ff6 : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790ab7d50c9943079ed7527cbd075555 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4a57c984e7a14f15b6fef03760d3708f : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/824810c783a94171be571559762afa5b : SUCCESS in 9m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ef648ca94c0a40b6998479b021d773eb : SUCCESS in 9m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/69bdea9fff30464aacfed974d5bf8a02 : SUCCESS in 8m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9169360e077f476cb97c6e0c71279cf5 : FAILURE in 1h 54m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88a3c2d9ba2247e6be69239997d59692 : SUCCESS in 1h 20m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0113e53ec34f4a1e9cc8154136618b1c : SUCCESS in 1h 20m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/53464fcb1edb4ec888553e0a8872ab73 : SUCCESS in 1h 07m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0d7931321a154be3b72b30d9c2510af7 : SUCCESS in 1h 36m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d260e597e29e4ef7a1568e993b492ba8 : SUCCESS in 51m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3b460fbc0dd4dfe9d5edb85c4d59fb1 : SUCCESS in 1h 45m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/194740c4b7424f74a599ce0ac9ef7d23 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":2},{"id":"59d451f3191e6d7168c18ff8fe27f75a6d0fa1fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 19:02:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e57a5dcdceecf2bbeb3ce5cc8f8f774aa136d82f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:26.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889060, https://review.opendev.org/889059, https://review.opendev.org/889064, https://review.opendev.org/889058, https://review.opendev.org/889057, https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"db88f825c85bcbbbc7ebd9afe638620aa486c0cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-13 12:25:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"33057f73fdc097df264045105df6d7ed6b77b13e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 13:43:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c0b52bde9e406280b4a353767e5cd9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52302e1398b64df0b92f22af3784312b : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab4391bf2a744f99a36f9b66a0246fa3 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3ade8736def843b3aff9fa9d64bfc375 : SUCCESS in 8m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9dd17fb879304eb1983c9beefe426b5d : SUCCESS in 10m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a60c82b574785339fe6caa9f7a595ca9cdc2cab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:38:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94a0b76ca69840a4985abe7ffc1f792b\n\n- grenade https://zuul.opendev.org/t/openstack/build/8bf9305ba91d4118bce4480794741134 : SUCCESS in 50m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/49f64d3789ad4f6e8aab85a711842d5d : SUCCESS in 1h 35m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12708349bb644042bcf9aa0879af24f8 : SUCCESS in 57m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f59ebe57338c437aae03e96d71d10bc9 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf1d6aa44f7246abb988804cd50cca68 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/824d9564d22a41138204c7c1e6cd87a4 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95635edb6d5b48468bf49230b930f20c : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/216a67e448cf4b658f80dbc942962538 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1574278105a443090b24012d2ca7fd0 : SUCCESS in 6m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/04dcab94abbc40988f7d0daafd79fa8c : SUCCESS in 15m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b8403e4aa6c44f5ae2446c4d24bca92 : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e8ca32e7bc040318e3029badf437468 : SUCCESS in 9m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4a5032b41a9d409aaeef095973972b9f : SUCCESS in 8m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2940547ec70041eb87e10fd27b04604f : SUCCESS in 1h 36m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fe0e27225d5e4738bf11e194099ed33e : FAILURE in 1h 40m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e97b0b7e066846e09f17fb7ff0fa9bc2 : SUCCESS in 1h 45m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9bb1247ccf104fa592ed57bae7778f7c : SUCCESS in 1h 46m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3438590878c747cfa18252fbeeb616a3 : TIMED_OUT in 2h 04m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/389f1f64e07c486f9fe01e783b810b05 : SUCCESS in 1h 00m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bd131141bc7e4078b2dbc111fb4a4004 : SUCCESS in 1h 59m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f568093d6f046edb53451ef6c4e99d3 : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":2},{"id":"60b065d30682c3db4599b67a9e5e3eb01a63be5f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 10:10:54.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"cd0cbceaf8da45ee85301d8839affc9d59faddad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 12:22:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1c2023e4758548279295ebc3c46bdaca\n\n- grenade https://zuul.opendev.org/t/openstack/build/1356d5aa081946578ed86a8068f74814 : SUCCESS in 54m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/afb542724011475398ef2182cc27b228 : SUCCESS in 1h 45m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c47fa30cfd24493b83de215bcc3f72b1 : SUCCESS in 1h 03m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a321dc8da3a042749217b45b19e949fb : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b31e85af0ca45db8c81c4ebbab2038b : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b7717ae906140288c34b78c7142aa7a : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7caaa61b51804b6b94fd60972af47506 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6f04d4492b343ada1cbe67340564ded : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22499127f0db44f09edb7c456c04d2f8 : SUCCESS in 7m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c9af4e47c4834e1cb3bb7ed45cf52d4f : SUCCESS in 12m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00e9630049494d37b7a71e2b8ff7e38c : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e8f6c9d0a2dd4dabad8ec306758f126b : SUCCESS in 7m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2f660ecdc5784cb69d988b6a4acf8131 : SUCCESS in 9m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fa943bd3dfe440beaf45dad1aa03b101 : TIMED_OUT in 2h 03m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/48de24ddf8e9430884a8dfaf0b2967c3 : SUCCESS in 1h 42m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef6df146664b497589c67c4cf428f0b2 : SUCCESS in 1h 54m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4f2af97af27e4bb2ab154054a52c3805 : SUCCESS in 1h 59m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f3093c3d325a4b25b8de6ef66b3b5736 : TIMED_OUT in 2h 04m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/560391ede028462fabf0ecb4ba53f2ff : SUCCESS in 1h 00m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fd052eb484c44523963d63cd8f2c47a1 : SUCCESS in 1h 40m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9fb56aaeb514f2aa709e2c3cbbd529c : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":2},{"id":"061d293511ffbe15009765ea7518f52e099af8a3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 12:41:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4da168010ad28d3c6505f4b2e0dd351524e3803f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 13:06:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e6a120c2f454a80ad5fcebab9863358\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73039c288d3f4b988f16e9f7f4026dbf : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce4b371872ad40c3a60ba2afd5829b87 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f0de9f5565af469a838c05e9b21c0870 : SUCCESS in 12m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f24b6228a9fd434c8520d3aac2e4d617 : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f18605dadb983c93b2a648c1c8b783c1ee901b08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 15:15:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dde9498d9908477d8f3411fd9fd4e715\n\n- grenade https://zuul.opendev.org/t/openstack/build/b195e199160247d1996954dd9a7d4f42 : SUCCESS in 56m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2324fa39fb5b41ad8f66e5bf73f3d043 : SUCCESS in 1h 18m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/712be4599e394fe7a5894a0d93933af4 : SUCCESS in 1h 04m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b1badf6f24412db3b27a9946cb0c9f : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea13015d6f334ad9bce9446672b7b812 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58ea0da1ac14410089c2821017e19dc4 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/488d022da55c402094cea05c692cec7b : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8e920f6bc3b4e7e83a6ec0e07659f04 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8748ab02f1d94e51bcc79b144ec964a1 : SUCCESS in 7m 15s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/923f2abdfb0c450c821add5695087320 : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc7f4d1af4d740f9b9fe8ddf6af1e4a1 : SUCCESS in 10m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8f91f9422d8412ba866049b4f2296a8 : SUCCESS in 9m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b6c44290b5f046edbfbcc0fedf471c0b : SUCCESS in 10m 00s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/839b2f013f874e399b8b22ad5fabe3d4 : SUCCESS in 1h 50m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9380fdbec064c67a000873f40047567 : SUCCESS in 1h 34m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9759892d3c1845f89d2cd9500b9a310a : SUCCESS in 1h 32m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a746f2b051c8418ebc8baf944247e757 : SUCCESS in 1h 26m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fecd65be10774b59ae7bc3e37d1fb18a : SUCCESS in 1h 47m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a20c2e0a2794c7ab363661ba2c85638 : SUCCESS in 1h 00m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3fc06fa91394d39a7b898cf786d8523 : TIMED_OUT in 2h 33m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/781eb16b81d6431d82798898f483f07a : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"019a375945dbf9a2a909203a1754f96b797466cb","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 15:41:05.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bb0d615628a099bbe9289a5cc67eb1097b1ce329","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 16:02:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6e8ea2beffb47169883b67a6bf2210a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac2398571b5c4301b28da48940cfeef2 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e806224c3890490297bef87c64e86d78 : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/94087d34abda4123baa576656ade3bd7 : SUCCESS in 9m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2984cce168ed4d2b981894ee43b145bd : SUCCESS in 10m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"810a75abb34df5eafd1bce958e94aa6208e2d785","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 17:52:24.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ceb105623d68443199c87db1ab63abaf\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f4ecae572644da5927712fb87aad40a : SUCCESS in 1h 02m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e3d236ceab34520a2e539672c5398b4 : FAILURE in 1h 24m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0b0cb1730244c1894624e7652546ee4 : SUCCESS in 45m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e52e97124ebf419eaee0da8a83647d32 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df67bb722e934e59bb676b9826700e35 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f08dd662c25643d9bc8b9eedd768c145 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c5a0c0c76e540c8904ef1722d0992d1 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66bb2fe486eb42bd9ac3c81bd297e9ba : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1116b29d9184659bc9d71da9e9a6262 : SUCCESS in 5m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1bfbbd6f98ad480cba7e57c628390a65 : SUCCESS in 14m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9a4af1590d93466cbceb5e43ec646694 : SUCCESS in 7m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41c34245d2f54d189e3bb130f176acdd : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/61699d5f96744179819b962869ee07a2 : SUCCESS in 9m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/90d4c036005047b0a9b2caa990c38bb8 : SUCCESS in 1h 31m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/624f53208dbe40169161596f647b3899 : SUCCESS in 2h 02m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8673aeb0cba944919b5106344c997b58 : SUCCESS in 1h 17m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bdcfb2f7a41746f4a4a8670a95c148a0 : SUCCESS in 1h 29m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/acbdad438a9146ae955d649e9dd0d2ad : SUCCESS in 1h 43m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96baccaee51f4e20aaf187596ec0cf8c : SUCCESS in 59m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/97a0e8895cfe4d9d83e2b5101dd02951 : SUCCESS in 1h 50m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1d1733f717774989b1ccfc9bc0aaec93 : SUCCESS in 39m 17s","accounts_in_message":[],"_revision_number":2},{"id":"43f43fe94f3d3767e1ea697395888bc3eb5b0379","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 18:11:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9ab19202487766352038d6b1362e63b7e9c3c65e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 18:29:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bdb796302434063aaa76cbaf655221f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/52b56eb092c84aec9b6230035e11bcf6 : SUCCESS in 12m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc60529cbbcc48c48778d5be7391c8bf : SUCCESS in 15m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e1342e7754b43a59094bdf47957edf7 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2effa0cb1c37413fb0912b0558f644a5 : SUCCESS in 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"15908a7358026bdd798fa538c5d72cfa733beb4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 20:15:47.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/396f635358b0464ebda8b68c41074194\n\n- grenade https://zuul.opendev.org/t/openstack/build/cb5f6513debc4716bcbfa99935ce375c : SUCCESS in 1h 05m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8b625ec12984d0eb2b76cd01429f62f : FAILURE in 1h 09m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/765834a6360e4412869044ffbee33ec0 : SUCCESS in 40m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9e53e97585446118561b7c0251cc6ab : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13e9e34386864c4da5614312d17de3ef : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64039f010d1742549e2c1757c69974e6 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82b2557ee6f648be84583b0b524ed5c9 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c78801d7c7a3426a970817993cdddc22 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8de1672f83484418a49caed8af632954 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/25f4bb46d5444049a9a8b8845e1f2c62 : SUCCESS in 10m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4101aa2da29404798ec032a8aa25092 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2703740dc3074643b9d89eef8c1a3404 : SUCCESS in 7m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0ab7be39cd204373b4c14d8733d818a7 : SUCCESS in 8m 00s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/03ea0b5818e9426587e04855198df1d7 : SUCCESS in 1h 32m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/56e1a3f878f34efabd0ffecd6c89d402 : SUCCESS in 1h 29m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f5187c85f1247a7818fadcd3d550c9e : SUCCESS in 1h 12m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/72d542d4c78b47e8a44b9de0bdf36142 : SUCCESS in 1h 19m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d9b60bcca6bd498094bd7749155af333 : FAILURE in 1h 57m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20b699944bee4529b6fa3b18fbedf300 : SUCCESS in 1h 01m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c3ff384f490445d2a61aa4d0f6985d55 : SUCCESS in 1h 43m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/184c4be4fb504b12845162c07baea3f7 : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":2},{"id":"d1478639044c49d59791b50e0aebd9dcea80d267","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-18 20:41:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"344d4eabb1df35b35d8f53cdc37380e0e1e19f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 21:06:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3159d3b5cd7f482db92a0c3662a6fc4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9a580a31101413bacf83d3ed24053c9 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4891a7cd5de847fd8c134c59c2fb0929 : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f1f2c5257c074e659deb49d27cce6009 : SUCCESS in 10m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4eca90722c024f518d98766316f125a6 : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1d468dfe90fe876827a084f7984809c4ecdbce0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 22:42:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9634a4c30c3d4aa8b0e0e83df66dada2\n\n- grenade https://zuul.opendev.org/t/openstack/build/16b8307c394d4d7eb0bcde0ee814f3c1 : SUCCESS in 44m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94f95cc93251432d9abc0bf3bdf7db60 : SUCCESS in 1h 01m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1627c42f6f9e4ccda6e961c2e7f4f390 : SUCCESS in 44m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf95cf7b33f4adeb03b80fbc0640ab6 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41bd0295f41b4aa3874cc37219b0fba5 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7817c0c7aa9642fe88642c94f91b2332 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b2a9a0b72ec45b4aa0e5e1ca9cd2573 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36e28ea9eb5b4fc3a22158c1c1bdf828 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2a199fc59243c089e48d545f8f8ab2 : SUCCESS in 6m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6488d4e1c0cb4bd7b1697bb7ad7b650d : SUCCESS in 14m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d85f0355fc144dff9f41504fe2e00166 : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1c476f4cb064b7fb3e6acecaca608b0 : SUCCESS in 10m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2273ddb76737495cbda35dee864ce138 : SUCCESS in 8m 17s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6234ebe9fb9b4ec3b06995aa6f2c00cc : SUCCESS in 1h 56m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9a78e57ef6f848f994d93598352b9f68 : SUCCESS in 1h 42m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2320bf7795e741b0bedeceb71d73ecbf : SUCCESS in 1h 52m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/38ac5ff15c5e431dbf6970dd6cb1d43a : SUCCESS in 1h 30m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/04379179f1044f3da684bc477c063090 : SUCCESS in 1h 34m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc3776a85e01424d8e03821e913bf80c : SUCCESS in 43m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/69944d36af2940e49217a5690a3ad939 : SUCCESS in 1h 19m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/88e078662839478fb17877e580254f4e : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":2},{"id":"0f80342fa1ad0e049d4c5e8f81231715c56fd2f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-18 22:43:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c127cdfe228c3fcf7d16b99685a9a4b502a8453f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 00:40:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51b9f4be59f34c2f87f9bb1a5b473a62\n\n- grenade https://zuul.opendev.org/t/openstack/build/50fa4a5501c247ac8e30024f4ceeabbd : SUCCESS in 37m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c20a18d43cf4fe8b331cb206912d49e : SUCCESS in 1h 06m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b52ea025203c45a3b2b291fdbee9ea86 : SUCCESS in 1h 01m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65bc192acd3743368d871f2f17a3c745 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/356982bb35ad4d52a7fad59907b06927 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d136065845147128eced8439fc5407f : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a971caab1154bcb94167f1257b62f49 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b24b51daae84ba1bfd08204f7d40b13 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/66d9fc23d282441f807a9aed1fddc17e : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/01b53421fe5448d0a1c36afd2478e6e7 : SUCCESS in 8m 59s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fefe1037804843e6880c5ca5ac24d270 : SUCCESS in 1h 09m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fb9146e28fdc4ac3a6531a940e34e100 : SUCCESS in 1h 12m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/161ad502044445a38e66e2a9e3160848 : SUCCESS in 1h 20m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72ef7bdd51564a33ba63e9a4d82b6476 : SUCCESS in 56m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6cd61cad337a4d2b826a2f50d28661c7 : SUCCESS in 1h 51m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ad1e6a1378a546faa328ba622c184d14 : SUCCESS in 22m 30s","accounts_in_message":[],"_revision_number":2},{"id":"fe6a4d3df33520f54664eb8543df166cc3852081","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 00:40:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9db6fb0c765367850f2357a775bb06e992d483a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 00:41:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a6980203d784103968dc5e0651a0274\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b6b88122c940abbce5f8a28ced68cc : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e1c9d6803052b9b1848ebc197cde2418008e1530","revisions":{"95453f1779fb11f4ac5f1887701e9ac8f3649bb0":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/889061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/889061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/889061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/889061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/889061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/889061/1"}}},"commit":{"parents":[{"commit":"176821704bf4b70fc5c4d95c2029982e468060a7","subject":"db: Don\u0027t inherit cache for DeleteFromSelect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/176821704bf4b70fc5c4d95c2029982e468060a7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 12:02:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Drop the convert_unicode, unicode_error flags","message":"db: Drop the convert_unicode, unicode_error flags\n\nResolve the following SQLAlchemy deprecation warning:\n\n  The convert_unicode on Engine and String as well as the unicode_error\n  flag on String are deprecated.  All modern DBAPIs now support Python\n  Unicode natively under Python 2, and under Python 3 all strings are\n  inherently Unicode.  These flags will be removed in a future release.\n\nChange-Id: I0de54294cc311b32f0956942a62deb2e47729a4f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95453f1779fb11f4ac5f1887701e9ac8f3649bb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95453f1779fb11f4ac5f1887701e9ac8f3649bb0"}]},"branch":"refs/heads/master"},"e1c9d6803052b9b1848ebc197cde2418008e1530":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-25 09:05:52.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/61/889061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/889061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/889061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/889061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/889061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/889061/2"}}},"commit":{"parents":[{"commit":"ede0f9c1b454ece0fddf0ab5b52c5b05925e7cdd","subject":"db: Don\u0027t inherit cache for DeleteFromSelect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ede0f9c1b454ece0fddf0ab5b52c5b05925e7cdd"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 12:02:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2023-07-25 09:05:52.000000000","tz":0},"subject":"db: Drop the convert_unicode, unicode_error flags","message":"db: Drop the convert_unicode, unicode_error flags\n\nResolve the following SQLAlchemy deprecation warning:\n\n  The convert_unicode on Engine and String as well as the unicode_error\n  flag on String are deprecated.  All modern DBAPIs now support Python\n  Unicode natively under Python 2, and under Python 3 all strings are\n  inherently Unicode.  These flags will be removed in a future release.\n\nChange-Id: I0de54294cc311b32f0956942a62deb2e47729a4f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1c9d6803052b9b1848ebc197cde2418008e1530"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1c9d6803052b9b1848ebc197cde2418008e1530"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
