)]}'
{"id":"openstack%2Fglance~889064","triplet_id":"openstack%2Fglance~master~I9dd70c9976f83a00a8cb270422e545ff2c77f12f","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-13 09:46:52.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I9dd70c9976f83a00a8cb270422e545ff2c77f12f","subject":"db: Enable foreign keys for SQLite backend","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-13 14:11:22.000000000","submitted":"2023-09-13 14:09:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889064-sqlalchemy-20","meta_rev_id":"ebdb41813fa6179b5b4eb9fabc399d2a7973394c","_number":889064,"virtual_id_number":889064,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-13 14:09:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-30 10:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 13:51:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 13:51:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:27:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:23:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 10:08:30.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-05 13:51:21.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"8a749437f45ba05095917f3e615b24862afec631","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bc0cd7d15499fbacfc8fb1aaf64e839aeff54f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:27:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9e718cc9f274bab9e8932056d6bad31\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/925a24bc8049474881a7286421f1a4e6 : SUCCESS in 14m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9239748494ed4dd28ee210eb979b2af1 : SUCCESS in 12m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5edce582a03047458ff328e34ee358a4 : SUCCESS in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bacff501d0edcfe4ec05a3b17547512db5a8227d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:23:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6487ad686a4e4496950d163fe3d998ac\n\n- grenade https://zuul.opendev.org/t/openstack/build/b066cc3fc9b54e659a15706038aa938b : SUCCESS in 1h 24m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/08ee66ba849b4ea69d2eb2df1e46deac : FAILURE in 1h 42m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7d385a876d54aec92aa8fbc9ef2850f : SUCCESS in 1h 08m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d86ae191b194449d91dd9a2045d23980 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/780911b7810d4df492b62ca8d7099861 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f35cf57b2ab944358791468685225448 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/485bd99530f049699c90cf10f66c697a : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/538e9c7cd216414c9bf463c5cb3a3506 : SUCCESS in 7m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4b90c620aa08466e932da98394305c1b : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7fe428ff0f0d41cc91063889c48a0243 : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2aab2ea581f74a9c9ac6b7acff5ec138 : SUCCESS in 12m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bd6f381600474cae9876f752867b0194 : SUCCESS in 12m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca02883707d3493c9e3282da1e86453f : TIMED_OUT in 2h 04m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a62b0b6a464c48bba363b90edb1b7ed3 : FAILURE in 1h 44m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/77a13920d9a540fba3d8569b38a383ec : FAILURE in 1h 28m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0d243a446254c67beb88bfea99effef : FAILURE in 1h 47m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bf81562814414c80bdd613c2f505f92d : FAILURE in 1h 38m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e21d68c90a2546a1919d454b69955680 : SUCCESS in 1h 04m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7f3db58d79304850a3b386a145db6cd8 : FAILURE in 2h 14m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a40455099be945de82dda49df067625f : SUCCESS in 42m 54s","accounts_in_message":[],"_revision_number":1},{"id":"ad7d5dd8bd26194ff414f9f150ec6b82146aaac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:04:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"40d33501e28e0cdff7a0b44e360ac27d71126cc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:13:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eb9bf29c5fe4452862b3fb871e6d9df\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4228e7dde9524471b9c864c0f270c477 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e4583a7711ba4fe99bec48117e582125 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f85298b581354913ab0056da2ba4782e : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46d5cd45caece93a8fa4a021c78c68881e5bd71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:20:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89454d4b9ea648dda21a080dc16c97a1\n\n- grenade https://zuul.opendev.org/t/openstack/build/e4f42ac91c3b41fe8c1464f458c9a8f4 : SUCCESS in 46m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c0bafd61004a4b03a4c107fcf4cb0e7d : SUCCESS in 1h 52m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adc04484c0314bc091785dfc1106c40e : SUCCESS in 1h 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f6519003849449c86bc623f71db5c7a : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83cc4e7bb92241c99f9e0d5907cd6fa9 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c62e67143f8484faca9cdcb3cd0f578 : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da1d026acbb3489a942d4333e8ccd48b : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f3757b4bcad4d16ac3f0445e75443a0 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d5f33c087d014bfb8510cee9b6bb54ec : SUCCESS in 17m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f626139559a34e0f8a5ea9c69df2cba4 : SUCCESS in 9m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b02057c849264cedba15e7b3a807d454 : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b0af245b994d4573a3fc7f7053dbf0a5 : SUCCESS in 9m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2d6ba4dc11584e038f6fb51188e37617 : TIMED_OUT in 2h 04m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c99e07741cbd4a4694d8dee0f00253fb : FAILURE in 1h 47m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d38341615c8843c08f485397b138a238 : TIMED_OUT in 2h 05m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c46a3ed81c224e03afbd64b9d274347d : SUCCESS in 1h 37m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/68a5f2d023e6447aa46d319647cb8368 : FAILURE in 1h 42m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edabf8bc5cfa41fc9a4e1be170420c66 : SUCCESS in 1h 00m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bda0e81562f94b2b909fbdcbda20a58f : SUCCESS in 2h 03m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f70a512dd444f188f71adf108910470 : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":2},{"id":"2064cf43166462ed92113464635c75db63ee7d28","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:31:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f3b5bcea4fef5c3c1b6a4c8d832aa3b8f325d23e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 14:41:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdebef3cacfb4fab8c2a94d1b7d5c0aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1972765430a84d2dbfa2adf27064fb68 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/541082ff51bd4d5f8ab6258c216fdc9e : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7cb63d5f5cde4ce0b756b056c1b222f3 : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"006c8872332b6bf7de524178bf7283b9fd3f48ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:22:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b2b4762624b417c81b4dd151c31a4cf\n\n- grenade https://zuul.opendev.org/t/openstack/build/50492f8a4a924cadb63828f8eb8834a6 : SUCCESS in 58m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d3a90be374604c34a48d53ae2ee2fe92 : SUCCESS in 1h 35m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d39362471ee4264a609b12c4496eda0 : SUCCESS in 1h 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ad0383ad4e43d5a1a75a00984f9185 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fb849d4cd5344e98635cff8033b7798 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5916ab1783f541b3b2dbc2a28f3ddb2e : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8402ac6f59c9491aaf04d75347a5ed88 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f484e394f04b909c85768b84f5a8d5 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/07e08afaa5244ce1b17d98561123a67e : SUCCESS in 15m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb4abceb0efd46398c3739e1e98204fb : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/66acce65a5144d509939b7d914371286 : SUCCESS in 9m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/189dddd547e845c5884f4f952286e030 : SUCCESS in 9m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/08e2f3e2607d499a807c5a18e2945957 : SUCCESS in 1h 59m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8e35d90696e04097907714dea02ab06a : SUCCESS in 1h 47m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bba7220803a14bb587053fb97e251536 : SUCCESS in 1h 57m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cf108eb29dda463aa770702e054675bc : SUCCESS in 1h 30m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/37de9420dd1b424ea8c13182b00ac201 : SUCCESS in 1h 52m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14d30263147242adbe889b20e2f1dda0 : SUCCESS in 52m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c18956a6c9147298536819d0abfa844 : SUCCESS in 2h 04m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/70dd54b496e649819fc9708a40f22e43 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":2},{"id":"c2f15f75b7975c6b154d960359827cf048f2b64e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-30 10:08:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"611c0dad2e2ab06208d0a92479a03d7e8be8a988","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 13:51:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e04435ae8274ebe49475c963e96fa214b9c332d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:58:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"64ada54f07a92d266997e7cc2e247f069dabd667","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:24.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889058, https://review.opendev.org/889057, https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"7b780da2f36058e69a27955efd012001aeb125ff","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-13 09:46:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f5d9fd9b3729bdadcf748e03f692125f442a1277","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 10:11:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2522719059a1483a92327fc415204a11\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/57f5b2b8472440edad4286cdae50a9f7 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ec01dd6753c4045b40c64810de2e3a4 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5709c904ef4d4a3a83b854906ca325e1 : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1379bf01ce2e4ef597d50ec3f71e5dab : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b66ee9829bdccafe6b11b8dcb1e6edba394546bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:48:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50a7195870804220b5db0b79736b9cd7\n\n- grenade https://zuul.opendev.org/t/openstack/build/9fb6cd06834c4ae89ec8e647f75d50cb : SUCCESS in 47m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/64525ff713e14528afbd82993a892b0d : SUCCESS in 1h 42m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9b5b147e86d4a3a8d737fcbbd2ccef8 : SUCCESS in 1h 01m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd6406d26044a2a8b3168a200b01344 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/152e82ccd7da4186bc9533d2da9f6956 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd99da3f65b24fd3b4ff92bb46f477fb : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbf26290ed0e48dc8ae05d5167d3a111 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41329107c9ab479bbbb5b04f7d6d1aed : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e656d4d5ebc4727aee353d4a361462b : SUCCESS in 6m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0eeb40524a054e0987d42360835edcbc : SUCCESS in 13m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/15004b0a26bf416ab7749a54dfea2513 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be9b7900d8f84b669409138a049a5343 : SUCCESS in 8m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3971679752c44aa5881e13fc65172ad5 : SUCCESS in 10m 03s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b964354a203432caf0979a12d759b2a : SUCCESS in 1h 31m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4cff98fc6af74fa1a5731ebf21b3311d : SUCCESS in 1h 42m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2447e9da9867447d9c7090f234f5a83e : SUCCESS in 1h 47m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/035eebb6aa2f46839f48366c168d8366 : SUCCESS in 1h 45m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/165eedf443a24aacbbd3368bd65adf4e : SUCCESS in 1h 40m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75b4e62c18ca46a287449a627c005182 : SUCCESS in 45m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01d035a7a7b34ca6811cb6da7c704d94 : SUCCESS in 1h 53m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/93afc1bfabd64f85a0b83791b2791a66 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":2},{"id":"14a917116adfa1a31045eb89f9a504f473da2668","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 11:49:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"444413f79c72b5d548f160d2c379951f48a22bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:09:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0054a0d18148b6ac09bcffa714955e\n\n- grenade https://zuul.opendev.org/t/openstack/build/7668efb6d8eb4a649936f0e959b51fdf : SUCCESS in 46m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eed44ccf5cdc442a92528d2dbacbff48 : SUCCESS in 1h 26m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b64bbb1b6e54feca21324d29702a5ec : SUCCESS in 45m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41f4c1eb4e824f9f9b1d2af57eb11104 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffc92a767606418fba4cf548ec43d9b6 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b17020e68c543aeb59eebeb6559e2e9 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d39c84999a74cdb8263715d361aa2b7 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f5cc54ec2154553bc92b08ba632ebf3 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f1510dcb2564eb78f548709ff2b2249 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2b50992f3934646b838da086d1c5b4a : SUCCESS in 8m 55s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5e91c843587649a6acab99343a18e13c : SUCCESS in 1h 48m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9ab613efd0a6478a8e0489e96d0684f2 : SUCCESS in 2h 00m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d32854143aad4a1fb8fe0d712dd8a2a5 : SUCCESS in 1h 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b6652c2cfd04e439d385a1213d35756 : SUCCESS in 1h 01m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2b5241c4688541d396ce40da2314474d : SUCCESS in 1h 43m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5c3f6a975c6d4473aec07b74c9cddf99 : SUCCESS in 23m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f17ae1fb77a0531be21a14ed84c59d1ad5f53225","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:09:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ebdb41813fa6179b5b4eb9fabc399d2a7973394c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 14:11:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fee1c143617e47138df67194e01bf676\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67403ae97a84441fbabb5a55b4475108 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e716b53f3622ab3f8e44f6c5aa0e062d58386f7","revisions":{"aa090aa01c5a43157e5bb6647d1e45d4e0e0df58":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/889064/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/64/889064/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/889064/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/889064/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/889064/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/64/889064/1"}}},"commit":{"parents":[{"commit":"deae700624cc2cd4db0410d680c88867ecbbd5d5","subject":"db: Replace use of update.values parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/deae700624cc2cd4db0410d680c88867ecbbd5d5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 09:13:24.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Enable foreign keys for SQLite backend","message":"db: Enable foreign keys for SQLite backend\n\nWe were doing this adhoc in tests, which doesn\u0027t seem to be possible in\nSQLAlchemy 2.0. Do it everywhere instead.\n\nThis highlights a small issue that we have to workaround in the tests.\n\nChange-Id: I9dd70c9976f83a00a8cb270422e545ff2c77f12f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa090aa01c5a43157e5bb6647d1e45d4e0e0df58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa090aa01c5a43157e5bb6647d1e45d4e0e0df58"}]},"branch":"refs/heads/master"},"2e716b53f3622ab3f8e44f6c5aa0e062d58386f7":{"kind":"REWORK","_number":2,"created":"2023-07-25 09:04:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/64/889064/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/64/889064/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/889064/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/889064/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/889064/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/64/889064/2"}}},"commit":{"parents":[{"commit":"ce5056ad3a5698faf222cde06dc1bdcc88e74e2c","subject":"db: Replace dict-style access of Row attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ce5056ad3a5698faf222cde06dc1bdcc88e74e2c"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 08:56:29.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 08:56:29.000000000","tz":60},"subject":"db: Enable foreign keys for SQLite backend","message":"db: Enable foreign keys for SQLite backend\n\nWe were doing this adhoc in tests, which doesn\u0027t seem to be possible in\nSQLAlchemy 2.0. Do it everywhere instead.\n\nThis highlights a small issue that we have to workaround in the tests.\n\nChange-Id: I9dd70c9976f83a00a8cb270422e545ff2c77f12f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e716b53f3622ab3f8e44f6c5aa0e062d58386f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e716b53f3622ab3f8e44f6c5aa0e062d58386f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
