)]}'
{"id":"openstack%2Fglance~889065","triplet_id":"openstack%2Fglance~master~I5245c402699e179c219039514b626a52fa0028d8","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-21 18:43:56.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I5245c402699e179c219039514b626a52fa0028d8","subject":"db: Remove explicit autocommit","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-21 22:36:18.000000000","submitted":"2023-09-21 22:35:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889065-sqlalchemy-20","meta_rev_id":"88ab35f5f6f8fe6e71dc05e46bb2149c8d57d6a2","_number":889065,"virtual_id_number":889065,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-21 22:35:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-30 09:52:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 18:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 18:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:33:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:48:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 09:52:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-05 18:49:04.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"4ef501b13fbaa302572c02b7bc88a44e75c5612a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5937029b5a0316030ffab250da5f7fb3334dbeb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:33:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a86b73885b534823bb75f2cda280c3bb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb18e1540560484fbf3d8132595ab052 : SUCCESS in 10m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d32d5331ddcf4e7baff7836bcde655b7 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/89cd4faba82f434dbbf7544c3262063b : SUCCESS in 14m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c7c102e928f755fcfb2a676f74bf83191f2679c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:48:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d170c08926149749140afb1838c4e5d\n\n- grenade https://zuul.opendev.org/t/openstack/build/9dd6a6c390504c849fc7bc68f02e16b4 : SUCCESS in 1h 02m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f59c56ca5f84cb38f4ddbf3c1caf3ee : FAILURE in 1h 25m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/863710526bf2482cb36a5e8528d0776d : SUCCESS in 1h 05m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c5acf02e15e428da1409749429e06ab : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/babdd756dcee4093a590b5304d3557e5 : SUCCESS in 8m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/997602ff43a14eae8c1311be6e6a3d61 : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2646be4a01b4479e945fe7a31af8c83f : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f36fc1e572ae4cd482cfa8df8ac571c6 : SUCCESS in 7m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9a8c49a91f6e4a43baed63a67ad14f88 : SUCCESS in 15m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/78447add9fd249a58b829f274acfb5a0 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b63de7402d51466ca91bbabff5cbf7ba : SUCCESS in 10m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/98bcef90c84649b2b67d84d13ffe27d1 : SUCCESS in 9m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d3617abdb1dc485d9c29705019d4c1d3 : TIMED_OUT in 2h 03m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/72e1c7dff04945158060803ece141b8b : FAILURE in 1h 56m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/63fac4347c86424ab751078b625cbfd2 : FAILURE in 1h 26m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0aa76fb9790f43289ca2ed3cdef8f97a : FAILURE in 1h 38m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ecd5c1d3d0c346af875cb760d056dfaf : TIMED_OUT in 2h 04m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b884bd70664f4849983521aed6291212 : SUCCESS in 1h 07m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ba7c84aad8344feaa0b510c63721698f : FAILURE in 2h 32m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9dae3ee7c5b0417c831d65131ccf45f5 : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"8a452967f4556ebbff575f68e5af68e393de995b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:10:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1a61cbcd7aa786d77bc0f12b4bd170ef6306ab15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:21:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57416f8a699c4e0fbbe7cc0e54518e4e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89a1c8dca730483abdb92ffc36f8174e : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffa800ee0c4540898a1ab460365de403 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/572d1545a28847b793acaa037fd48383 : SUCCESS in 6m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d999f7833ed27b8651ebd8914eddda8ab9286649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 12:02:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74c44a184a2d40c78e7a53c1a1d99dee\n\n- grenade https://zuul.opendev.org/t/openstack/build/5824278a95484e898c0ce7a081e744d9 : SUCCESS in 1h 08m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42e58922d0484538b713fa04dfffb092 : SUCCESS in 1h 41m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a71f4afbc54450d97b8b31a5f3a04c8 : SUCCESS in 1h 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6860c0393e2b46ad85897578d47c8704 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a73c1b1c8e74a1aa719f7d1d17e39c1 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7e5f11252cc4c5193e17dcbb495d6b4 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2204434a1f9b4db5b6a25613f67708bb : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1a173a15bb47a5ae551b846aa6089f : SUCCESS in 7m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0d3cde12fa72429191f0e4b3d799342e : SUCCESS in 14m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0a118701f1d3434d9ae8f53661b061b3 : SUCCESS in 11m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95fbddd29fd84488b1eab99fc737eab4 : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0818ba3f6fb94f0da2748d1362b06eb0 : SUCCESS in 10m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/642a36d079f34e8ca8ad82758b7f478c : TIMED_OUT in 2h 03m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b81981e9a106497b9403316ce2419646 : SUCCESS in 2h 02m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cce5c36a03f344e28dad9dfd9e544c4e : SUCCESS in 1h 31m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/396a29999bcf44c98b4f463c527ff5c2 : SUCCESS in 1h 36m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53f07b5f48af43d089cd1271b1bbdc0b : SUCCESS in 1h 41m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ead2881341f944b98c0abee8da8875d5 : SUCCESS in 1h 11m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d483a0b7f7d941b482c9d429ab499780 : TIMED_OUT in 2h 34m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/50d916ac574d47f6bb53b861b5b9f91d : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":2},{"id":"fb86e81720ae69f1b28ba7dc895b488b6a64a4fa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:32:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f98a465c0375dbe7439033a5bb7211a8efea5550","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 15:00:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09550908ce1f4cb481906f324769608d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c18f2e501df4b5bae094bbb7238f5b8 : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ff40f6fc6614b1db0ef22365288117a : SUCCESS in 11m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd7d7299dea34e938016ffc48515a3da : SUCCESS in 11m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d76a3d3c8e73bfa3b28d43de78276037bd1ddcb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:23:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87cddd8476034a0b99fb68a05bd34347\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a095e9383744d97bfb5c6cdb58b11c7 : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26d6d5fc15a24bcda48bd1b67c9cbe59 : SUCCESS in 1h 36m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/079939f568594981a0eb25fc7ec6762f : SUCCESS in 1h 01m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eaa79dd33d948c39b1856a15bcf8c45 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7da959a03bff4972a93f311d632c88bb : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4b07648e97d47938984b4bf5ae316c1 : SUCCESS in 5m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a2e01ed266b47389139cb9affbda728 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6bff365055d41f6a83bd5c7984d7b66 : SUCCESS in 7m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b901106af0c44324af8a0e197b0efe53 : SUCCESS in 16m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bfdd9bf94954e58a82f6c49950e3e57 : SUCCESS in 9m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28b0446e134a4343be033151270112a0 : SUCCESS in 11m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1202e85ba11943a58e0c1cf97ea088ec : SUCCESS in 10m 47s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cda483baa2344e75930ad533a7ae0150 : FAILURE in 2h 01m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae24849b077348cab567295ce42a8bae : SUCCESS in 1h 29m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ddcd06e57c8a46128c6e34e7a24e15e9 : SUCCESS in 1h 32m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d16034a6081046178557fdc5053ac9eb : SUCCESS in 1h 58m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1b54f99de314b239269fb75f9626e20 : SUCCESS in 1h 40m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fb2b3e68293468488ab4cd90e0b0d3f : SUCCESS in 1h 03m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/01e030b9371c4ab481f899ae9ceebb72 : SUCCESS in 1h 42m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b424f622b2d2431d8031a884f1694c51 : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2612c265793a1764d8bc36258b666a4ca6870f8f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-30 09:52:42.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"92b3a56604044a83b2df0ad7b703a2d7b956d28d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 18:49:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"90785d7e9fc5bc1c84ffb5503b5762f40d36176c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 19:05:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a310954f66609c4aea358d31aa0ff59107d0a567","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:27.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889063, https://review.opendev.org/889062, https://review.opendev.org/889061, https://review.opendev.org/889060, https://review.opendev.org/889059, https://review.opendev.org/889064, https://review.opendev.org/889058, https://review.opendev.org/889057, https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"d03c0e8885da7bf395cc7eb7a6f7fed34abc0851","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 08:43:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"18759ba3cbaec5953572e7dc8f7ae7907ac234ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 08:59:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/580828a18ad440ccaacdc7fa597fa060\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d0ccd1978b74273aaa17d0170c5c52a : SUCCESS in 12m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f89dcc7211f845c59b0f708d456acabb : SUCCESS in 13m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12d89345eff944ea89faa04ced5efb5a : SUCCESS in 12m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9d4678ca430c4305a76ff9dc8667980e : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ec46c1ca216fce1d23dff40a715efe477379bd33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:52:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd83eed3e8d640a48be2321d5f320be2\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0d72c68bb9448ee8762fae84a386d84 : SUCCESS in 42m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c94f6f739c64a388e2ee4455013ab1d : SUCCESS in 1h 16m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2444bdd8cb9f456e82b51612768f3b6f : SUCCESS in 51m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29e922f18b1486484693666af5115e6 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2991c9b1629745578bb729150e7f4520 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0585d70e5ea44d99f4608dbd7fed22b : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d13b0b82b6043759a86c2985d7e18c3 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/846d8670bfe643db8e9dccea682623d3 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4289101772104ab492fbe4921b97bdf4 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0dc6da8f33524a04b826d8b05d4cbe7d : SUCCESS in 11m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/115b6055f1324d1dadbaa9fb2b246cb2 : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d3e420e5453432383b078180c8b16f9 : SUCCESS in 8m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9164a8467fed4d5d8285d243453e0845 : SUCCESS in 9m 00s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6f78204deb344382b15c818c6e7b5f73 : TIMED_OUT in 2h 03m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5e2c927c1f0449759acd2100873b311c : SUCCESS in 1h 49m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d06b675de2a4f1889a91f1d01898629 : SUCCESS in 1h 36m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7765dd3635c54c348b1bee00a51ffb6c : SUCCESS in 1h 24m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b94e31511f164b1ea3203eaa9171b368 : SUCCESS in 1h 47m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b412c2f8c37042e983233e254c73c2c5 : SUCCESS in 1h 00m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b287454ec78f4234be0b13e76ad9332a : SUCCESS in 1h 47m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2f5eee0607824010b5e81ee8a42b5721 : SUCCESS in 23m 25s","accounts_in_message":[],"_revision_number":2},{"id":"1a75c7901f1632b56d8e0bf058d5a1750b687fe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:53:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c373bd25b878efc8381fb62c547ac196e205c3d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:13:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac60aab0cff74da3a960240c7db60c21\n\n- grenade https://zuul.opendev.org/t/openstack/build/4126a725b2c14ef29f6924817f3c55f7 : SUCCESS in 45m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8ab989df255b447f8775a014e1bc2fac : FAILURE in 1h 12m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34ecbadd2efc4b209cb2e796de711b80 : SUCCESS in 1h 03m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0fe01ed46544bbba8a6cfe030cd268b : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a379c77587943e6aa7ba63579c8bb5b : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4965aa07a3764db288dc9ca87c106c05 : SUCCESS in 5m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/410ea0db922f494780c94ad7eabae12c : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed7c34f14b944059a2a2f8d3a2677a13 : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1c69e8068a944d5baf4185a0134b92fb : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff4273b8efb742f5b0bc3fc0f4ddac04 : SUCCESS in 9m 01s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f922ba946104fb6be07058b6cb49f9f : SUCCESS in 1h 17m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/04872eadaf524f56b84b9f833657d6a5 : SUCCESS in 1h 43m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/857067fdfc064a03a1746affe8e9c5f6 : SUCCESS in 1h 39m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02fc3f8c65cc4912b5663aef7c8901c5 : SUCCESS in 1h 06m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b8fce2637cb64ad4802690565f4d4f20 : SUCCESS in 2h 13m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c6247d884e89468c98528673d20b38f0 : SUCCESS in 25m 11s","accounts_in_message":[],"_revision_number":2},{"id":"21204961ea14d03ac36ac599decb095f1d94a141","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 13:43:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b87603be71f4f5b3fdc73c2caba26cf558bd503a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:55:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e54a12edd444692845a3d38cd2f75a8\n\n- grenade https://zuul.opendev.org/t/openstack/build/da209509846c463a96f7815c9fcca47b : SUCCESS in 57m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e79f81fcc06c4a9ba33680cbb51b33a7 : FAILURE in 1h 08m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4772b05f0d246f6b44e8ecb0f0a9bce : SUCCESS in 1h 07m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e899160c2dd840fcacc95fc31f2d5b5b : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99d55fef27f14608bb8dfc0f0cf16f60 : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c88594bc4bb497595705ed8c47dfb35 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d1f5fe75785f4c7aa66dec403d771358 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c9c2602cbd74da4bb5a6d86d76e1c25 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebb5eaaca5f43bd9ca993285a887b94 : SUCCESS in 6m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f3a858012e5a40a3851dfefaf20d9008 : SUCCESS in 14m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0e1ec36b9c9b4c59abc8f14f1f5a8759 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fcd6e71a58884f57aac63b0f67c4b18f : SUCCESS in 8m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/874c873bc7c24fb3b44bfc637f822c0b : SUCCESS in 8m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f11a32e9dac24523a10afe125cfa6abf : TIMED_OUT in 2h 03m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ceb5d2c6cd947e09aee14e099840833 : SUCCESS in 1h 37m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5b35414110554728bac110a60232231c : SUCCESS in 1h 45m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0667dcbd63f6436f8bddf6a69009f209 : SUCCESS in 1h 56m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/784cf9a0a7cb462e933caf111ee27cdc : FAILURE in 1h 44m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74fe45eac9ba48d080d7baf2adc57f90 : SUCCESS in 1h 02m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f261d4f9e70e4c67b35103583d4020bb : SUCCESS in 1h 52m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b279f8d908d04f3aa40b95eb09d4eb6e : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":2},{"id":"4035228e9a8f5463bf3b351bc70bcf14eece53d5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 16:13:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"48fd38a33ed0087a5b0338f67c8480959c18b1fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 16:33:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0e3fe540b2440e79bb4f7226aeb1957\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4fc435609c614a98ae5c8592470e5b01 : SUCCESS in 10m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad5bb89aca85474dbe941875732f2313 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65ef3b5ad2864271b5b7bd127f600133 : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6cc8d4e79853455088cd69a63c4024ee : SUCCESS in 13m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b0e2da4b454c83421e7e8ed53e588a1291a1a23d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 18:24:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/683c680349264964b4b2f396aff29113\n\n- grenade https://zuul.opendev.org/t/openstack/build/2c35052484cb43ea9567eb9463193c51 : SUCCESS in 1h 01m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71524207bd8b4bf5b2ddeff9e937b188 : SUCCESS in 1h 25m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2f07a951cf241199e28bb3a865ee900 : SUCCESS in 50m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01ebe18f2cf420daa7717e2e94528e1 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/773cf080cdb14bb2b093b88170a50dae : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe5d11a35b244a4b8e24566ff377ecd1 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62abb26f45594cc28695fdd111e5d49f : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b0579b8e91d84371bd5f43f299eb0a1c : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abfa27b38a4e48bd98bebed85afc8662 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/829c508dceff47f2b426c76a8f293fea : SUCCESS in 13m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0f804612f3f49429042e1dd599fc8c3 : SUCCESS in 10m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b411088e0b794ac3865fdf7885fb8c14 : SUCCESS in 7m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/10567a4d8d464c9e80d96d4d08a63ba2 : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c0b9aa13727c411a9d91848143a4a7bd : SUCCESS in 1h 26m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/58d9748875534b8a95d0d048fc58632c : SUCCESS in 1h 56m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f7f04e65dcb4667afb035c0ecd88119 : SUCCESS in 1h 29m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6eb9570741cb4c10a4c1a8727eb26109 : SUCCESS in 1h 39m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96372df39c7044b3b29090a9f28377be : TIMED_OUT in 2h 04m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8506dbd23c9d4e019a3f285ffd1c8581 : SUCCESS in 1h 06m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9af717e082864c12a6ef16244cac317c : SUCCESS in 1h 56m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e43111babf54a069fb97f2f4403546e : SUCCESS in 37m 38s","accounts_in_message":[],"_revision_number":2},{"id":"863f170c07f35ba8677ce9a2aab629738c8467aa","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 18:43:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"054b04205d7bd6876226dc6b9b663a5480e5bf83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 20:46:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8ae4ec9ab347819ca1af99d3ba26ee\n\n- grenade https://zuul.opendev.org/t/openstack/build/51e4966e7c6e427b9e496f1062275818 : SUCCESS in 1h 07m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0bcf7b31edf4cf39c7856bcd4c5f088 : SUCCESS in 1h 28m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40210513459a41f7ab870d71216b4441 : SUCCESS in 1h 03m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c61ff0facdfb4fe1bb262148c5f79c5a : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6b8b4600dc34c27850e47dab5ee714d : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1187d2bc52545fcaeb8af22dac04fcc : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ac2af4ec8844c5aafb1334b483f0bf7 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ba0eb4450224c7f93db900016c25675 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe8333f7ee6c49ce8f4f451acb10c8fd : SUCCESS in 6m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e2b49e0eea6f4042b68f2f39bac1b3b5 : SUCCESS in 13m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f74b62ee2b94834bd8c8c6989eb9dc9 : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc59568e719a401dacf79968430c486e : SUCCESS in 7m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1d8dd64aefd7455692437649e595de01 : SUCCESS in 8m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dffd9a8461934d9fa4e7f347f6abb2e7 : SUCCESS in 1h 43m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bb5bd5838eb04e6db95803131acc8139 : SUCCESS in 1h 52m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1455658acdd46e0a3d6e3a14139b5ae : SUCCESS in 1h 47m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9305ab2c75a64e45857347f542b0c7bc : SUCCESS in 1h 48m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/208c160385e84d3491c264455cd2f3b5 : SUCCESS in 1h 56m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a09e3d7a7f8e4d5ea10f00c25e85a186 : SUCCESS in 1h 08m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83bd005dd1354575a5f135a2bcaa5f6a : SUCCESS in 1h 50m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be8a43e1354b45e1934126a9e27fe93b : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":2},{"id":"a61ade10aef4686592707ca84c1cafab54a1089a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 20:47:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"32dd6ce5ef9383c72a73cfa495ec1d61bc7abe41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 20:48:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b9e7a4bbdc409188702639293af258\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c90251b1f86497da85eeed5de107372 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bf641b73353746d4af42b680b7c4fd61 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/249d34105df34d9b989db76d1b9f2f15 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/91d1197e0213476190572bd90024d999 : SUCCESS in 9m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8c098568c00cac736243c23fd4bb92e926cce03a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 22:35:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9f1db3442de435ab18f49f108413f00\n\n- grenade https://zuul.opendev.org/t/openstack/build/57927c600a7649b88c0323c31d852908 : SUCCESS in 1h 00m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39f6bbf5a5064a92b4e8e68b6bdf75af : SUCCESS in 1h 35m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c5cd39eea4c4b7eae76df0e497d4cad : SUCCESS in 1h 08m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d631e6f5a1a84afcbf79c9ba24d8790c : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3669d11186c84684b20f585881117aff : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7596966b9e3543febdba4d005ebb4472 : SUCCESS in 5m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfbcdcf00a304257bb93a77bfc08839b : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/318068efd0a84745886e98602c3b3ab3 : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2874190d55f24811a5fb397cf16ac957 : SUCCESS in 10m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c035a891df5d4323a4380216d19e3b03 : SUCCESS in 7m 36s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/70ec900499ca45c1a7a7ea44d113968b : SUCCESS in 1h 28m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65e43e85d9784bae9a75f49edc197a5e : SUCCESS in 1h 20m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7725b0849e754378ba0ef19ee42321e1 : SUCCESS in 1h 41m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/849d5bc1635648438fadfc1490b7894c : SUCCESS in 34m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14be3061131540e794de6df13c5e0eec : SUCCESS in 1h 31m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a21e386423cb497ab210350d4428af50 : SUCCESS in 28m 16s","accounts_in_message":[],"_revision_number":2},{"id":"745f0990ef953be3091d5a23f0735dded70ffdd4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 22:35:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"88ab35f5f6f8fe6e71dc05e46bb2149c8d57d6a2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 22:36:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f56bf44417e430abdff2bff5a78ac61\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518ed2ea22914a7f98ab42f5cbb35ead : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d1368745c5a75c331632408ce284e4c782aa23e9","revisions":{"e2e39a379a4fbbb4ee38ce0b04602b191c8ef42b":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/889065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/889065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/889065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/889065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/889065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/889065/1"}}},"commit":{"parents":[{"commit":"aa090aa01c5a43157e5bb6647d1e45d4e0e0df58","subject":"db: Enable foreign keys for SQLite backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa090aa01c5a43157e5bb6647d1e45d4e0e0df58"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 11:44:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Remove explicit autocommit","message":"db: Remove explicit autocommit\n\nWe already got rid of implicit autocommit. Now get rid of explicit\nautocommit. This is far more involved than the prior patch but the basic\nstrategy here is that no public method should take a \u0027session\u0027 argument\nand instead should call \u0027get_session\u0027. In addition, all private methods\nshould take both a \u0027context\u0027 and \u0027session\u0027 argument and they should be\nthe first and second arguments, respectively. Finally, public methods\nshould only call private methods. Combined, this ensures we only ever\nhave one session in play and the transactions are correctly started and\nended as expected.\n\nChange-Id: I5245c402699e179c219039514b626a52fa0028d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e39a379a4fbbb4ee38ce0b04602b191c8ef42b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e39a379a4fbbb4ee38ce0b04602b191c8ef42b"}]},"branch":"refs/heads/master"},"d1368745c5a75c331632408ce284e4c782aa23e9":{"kind":"REWORK","_number":2,"created":"2023-07-25 09:10:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/65/889065/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/889065/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/889065/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/889065/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/889065/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/889065/2"}}},"commit":{"parents":[{"commit":"707ac699144394e7de63652693253d944f167e3b","subject":"db: Replace use of update.values parameter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/707ac699144394e7de63652693253d944f167e3b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-19 11:44:07.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 09:08:52.000000000","tz":60},"subject":"db: Remove explicit autocommit","message":"db: Remove explicit autocommit\n\nWe already got rid of implicit autocommit. Now get rid of explicit\nautocommit. This is far more involved than the prior patch but the basic\nstrategy here is that no public method should take a \u0027session\u0027 argument\nand instead should call \u0027get_session\u0027. In addition, all private methods\nshould take both a \u0027context\u0027 and \u0027session\u0027 argument and they should be\nthe first and second arguments, respectively. Finally, public methods\nshould only call private methods. Combined, this ensures we only ever\nhave one session in play and the transactions are correctly started and\nended as expected.\n\nChange-Id: I5245c402699e179c219039514b626a52fa0028d8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1368745c5a75c331632408ce284e4c782aa23e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1368745c5a75c331632408ce284e4c782aa23e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
