)]}'
{"id":"openstack%2Fglance~889066","triplet_id":"openstack%2Fglance~master~I1036b7b0d64c91f0f64b37340490a2e52ee0baea","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-21 22:44:01.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I1036b7b0d64c91f0f64b37340490a2e52ee0baea","subject":"Add job to test with SQLAlchemy master (2.x)","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-22 03:04:03.000000000","submitted":"2023-09-22 03:01:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889066-sqlalchemy-20","meta_rev_id":"177cd4861e47c51a6ab67355e3f731da52ec8d8e","_number":889066,"virtual_id_number":889066,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 03:01:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-08 08:22:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 15:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 15:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:35:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:26:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 06:21:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-08-08 08:22:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-05 14:38:21.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-09-05 15:18:43.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"b7d21a77ca7cd975075585477904d30df39ed084","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2c87cda26b4fd8924583dacba2f63bdc11659c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:35:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7741b4a92eb4bc080faf24069dc6684\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b1cf148964c45e6a55d242dbaaec559 : SUCCESS in 11m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/376f548c57924e578ea1857c28d7ef06 : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be3f3ac00b0e408bb2ceef825fcb60f8 : SUCCESS in 12m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aed0dd1b0019983f19cd631c92ab91cc24779d45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:26:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e9b3c381a964668b35dfe6776512d96\n\n- grenade https://zuul.opendev.org/t/openstack/build/1efcba69a7314fafa5868c8d84a4c4e2 : SUCCESS in 1h 11m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fbab862e566d4ebe8673dbd55d20c777 : FAILURE in 1h 49m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/99372c0e83c14f47a12552bb3dd61dd0 : SUCCESS in 1h 11m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b47432d2ca346f8be4fe2c1d5904cd0 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8b02f8bdb5f4cd38068de2ec8e839d8 : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89009aa0fb0449bf8913d791876651e0 : SUCCESS in 7m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3813129a56b343eebd53c42d8767a385 : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d53bfc162384f76b3cc4573fe5474a2 : SUCCESS in 15m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6914071b9bbf49d684a10ae7a9d8ab44 : SUCCESS in 13m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7cf39885ce2745f4b9c34ff8ce6e45f2 : SUCCESS in 10m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d0ae1641b704d5f9fdbee84cb96c055 : SUCCESS in 8m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7d5286b5ea744eb9a1f8427dc4ee04da : SUCCESS in 10m 00s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f21a9c024ca648c693fd76e5f297d878 : SUCCESS in 13m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a29b5a3ded3840dc9200b462d5460153 : FAILURE in 1h 59m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/487b4ba3b2df4be68078f6af01fd113a : FAILURE in 1h 35m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ecf0d2a17fbf4a1d9d3d09cb98a36659 : FAILURE in 1h 51m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/46882b8aaed642cd9acf16453f1062c8 : FAILURE in 1h 51m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/425c09a4011f4dbdb86df37fc3458cce : FAILURE in 1h 51m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4998bf692eb4cf2acd3d3dabb449164 : SUCCESS in 1h 12m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9170557bc6cd4d70b44233b184c6ebba : FAILURE in 2h 07m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de32041ce6b047128c44023bef204a04 : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":1},{"id":"98013ac76e0c886b9ba9c54879f7fd6ee58cb55e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:10:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"188a6306938d3e0e8dc13489ea864f838a7161a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:21:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a8fbb07e315496d9d11fe07f02e02d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d71897b25be84aafa1b68348102bd5b8 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6f17a1a95f242928dc9dc30aa288ae2 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18dedfaf0bd54fde9ac29ec3d6a96d0e : SUCCESS in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0fffbf2eb6d3de53b188234cda738b35b956a465","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:22:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"03d183528ed591f06019ad8d1bb47da10db2ee14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:40:26.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/904841eec8cb4e7e88399c120863b035\n\n- grenade https://zuul.opendev.org/t/openstack/build/1032fdf9b2f34d04a1697b4b4cc62b8b : SUCCESS in 58m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0273a9738b43498281d4c88e35e47992 : SUCCESS in 1h 35m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2b5d7a390a7347818f027447e362c6f4 : SUCCESS in 1h 17m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d32e3cf092841098662aacdc69f0757 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fccd03cc55b43b6b7df25f44ae7d6bc : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/954d81db0f7a4429895c9623876d66e9 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9161f596de9449b29324d69ff67d9c9d : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dc2368e12424f92b239ff6706f9b7de : SUCCESS in 7m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a1d68e51702e459e89585ccf5bb3e26b : SUCCESS in 14m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b881de801b9245bea6f4c3dbf0cc2a0f : SUCCESS in 10m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/175c0d0779cf48d1afec279c6826846a : SUCCESS in 11m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/87b4d4031e3d43aab9c3ba03b29b749d : SUCCESS in 10m 05s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9e7985cf10ff429ea62935a4221529ec : SUCCESS in 11m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57461bf524ec47e5bb1c541ea5bda0a2 : SUCCESS in 2h 02m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c2a045754d624dd69faca692f1558430 : SUCCESS in 1h 48m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3bc60ef179e4b9fa0caa4a7e7f757d3 : SUCCESS in 1h 58m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a8190ad166b0401aa5c3961778ad4ad3 : SUCCESS in 2h 04m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/07126ff8a72e4a9895b3470b5108b78f : TIMED_OUT in 2h 04m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8553e4b9d664da1982a3b11ed971f18 : SUCCESS in 1h 18m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c890338c25744d2390fccfff60feeddb : SUCCESS in 1h 59m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9170efcc53de48e6a30cd0c72cfdca47 : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":2},{"id":"3eb8459c4ee3949143bfc96f62b3d4efc5fd077d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:31:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27be6e251ed366fc85c43e778e68980fa32d74c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 14:46:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/731fbcb6105344849c4dce9aa0322666\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc0908031d7a436093d6ebd0d0e3d75d : SUCCESS in 12m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a4264cb040c4fc1b9a6d3f2c9fb7849 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae541a57463f4189a17136b7bfb06dad : SUCCESS in 11m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"535ebba50c48ee96ee73ff7e09f6426bbb1db795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:11:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4cdd67dfd1b4d4a9a5de45c9afc7c76\n\n- grenade https://zuul.opendev.org/t/openstack/build/b514a4fc64cc4ee18e0f943b75705662 : SUCCESS in 58m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8e9153b91ab44f3a74fb391b2afd24f : SUCCESS in 1h 24m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e7948c86c994e62b08efc9b3c3619b6 : SUCCESS in 50m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fcf8816871241759c2e99b9d2dd2f58 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef248ccd3a954002ad6e858aa25cf68c : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/844beaa265124caf835678f296aa178e : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce70d219778a45798656bee2bc3d9ae5 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92ce3e2b4b74411d96a5446684ac16e8 : SUCCESS in 6m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c5eb348bb1c342899a252fc3618d635d : SUCCESS in 15m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc05164a9df04775bcf8efefdac5a36e : SUCCESS in 9m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c510dba493004d658753ebd6943d09ae : SUCCESS in 10m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/91212e7fc42c46fb84bf70894ab73f3c : SUCCESS in 10m 51s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c1d3a95c3bc745efa36a1757f8e3c0ed : SUCCESS in 10m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/11d8bccd795f4765acf0843cc096cbf7 : TIMED_OUT in 2h 04m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a1e50cf827d0479f81bedb7f347339fc : SUCCESS in 1h 24m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e5c057c38b0f44c1a0b34d45d7efe06a : SUCCESS in 1h 43m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5e6fdb79efb941f4b78addd02faa6d6a : SUCCESS in 1h 49m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/35e0f59824ce4c8db654c55a89f120da : SUCCESS in 1h 29m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3506bbb37cb3442b91a3a022e30d672c : SUCCESS in 1h 11m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/950223d16efe4d829557595cc06b8219 : FAILURE in 2h 15m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c92c351022334a0591d571e6de88cd68 : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":2},{"id":"88d9ccfb0cdd13c24f0529bca29e87e51d8af043","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 08:58:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7e56fe441d8cad1330ca0432ed4334156738339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:19:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e8cf946583d4b808f3d67ba0332cd79\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/47c511fac534421185b6d069a8346aef : SUCCESS in 12m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/39c1760dd13a4535857fad681503d741 : SUCCESS in 18m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/439e71ba6d234ae0958b6f19d600ee5e : SUCCESS in 6m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b3eeb4bbb0a374047c569dac48353cfe147b14fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 11:13:18.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/263586bd4f5240b0af250340eb310aae\n\n- grenade https://zuul.opendev.org/t/openstack/build/41fce7a36c04421b962316f743f12a25 : SUCCESS in 56m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63f25d36a4204b72955d92bbcd21ac43 : SUCCESS in 1h 30m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b12fcc84ae0e4aeea4b916c6b12de12a : SUCCESS in 1h 02m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd7bdc5b52214d669da3c1369f52a369 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3155da99cf264594aaa24c73ca46a667 : SUCCESS in 13m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc368d0baded4b4faf268e8b0ef77ea0 : SUCCESS in 23m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09b454e6062643bcb7f4c6554054d075 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb75cb667cd46a18c693593cd4fa9f2 : SUCCESS in 25m 09s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5063d99cddc042c3a6056b7c72f617d4 : SUCCESS in 19m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/267252afd7e844f3a400b52c777462db : SUCCESS in 11m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c369a7165da54578a2cb5742967b9a11 : SUCCESS in 9m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4f56b580f1c341a1bd57c8fa11dccfe6 : SUCCESS in 8m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/72582ade989b4b43b7449f1b56399b71 : SUCCESS in 10m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2c438b70048043448125fe273352e1c6 : SUCCESS in 1h 45m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b4c10212ef441c0ba76f709e567b41f : SUCCESS in 1h 32m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/88d364b09de74ef18e9de05e458f83f5 : SUCCESS in 1h 42m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d03a76a9fe304c95841a5b78211d5e77 : SUCCESS in 1h 31m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/029661565ac24912958c08a13b4ea4e2 : TIMED_OUT in 2h 05m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19503e4082af4678a374f09893c9d66e : SUCCESS in 53m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce44e22363174b0a9b2ab87b59a3c55d : SUCCESS in 1h 56m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fd97cbed59ef4ba286157b4063dab2e0 : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":2},{"id":"44e8bd4abc1301a9eac7813df2448b2f5b38fd4f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 15:17:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"18de140dc48bcf44a71dcd1081e138111b7da2b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:37:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dcc6fa4724b4e2eb88fcd3cf7a0707e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac12579fb07542b58c4257f74abf4609 : SUCCESS in 17m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e12ce814c604b58bfac2650d25299ce : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9ee1d79d9da34f51b783e549364cadf3 : SUCCESS in 14m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d68e25a87cba8fec53dfa4c9bfb3fe70b2041216","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 17:19:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c072d07e9844620bf5aa1f48a539179\n\n- grenade https://zuul.opendev.org/t/openstack/build/61bff8541b0c4a8086572197a94b3aeb : SUCCESS in 1h 18m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cde4f5e3b12b4230ab3cbeeb1a53fdcc : SUCCESS in 1h 47m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6395620d1cf546ceaeacb175914afda7 : SUCCESS in 1h 10m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a4e8cddf73458096f5286e53204bbf : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bed6d3a19d342178ed94ce6bc8c9a4e : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdedecd35695459785541d7592f4cfec : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f65a7036e4054776b3735a4ea200a311 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c99d72f9f94086bffb7ce43ed62f75 : SUCCESS in 9m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/42b3146e25d041f184978467e6bd676a : SUCCESS in 15m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bdbe7eda7ee04989a9fd3993b870b3b6 : SUCCESS in 14m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/760991775b0f48d9b09ae0e61d748331 : SUCCESS in 9m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/55d9d21ebae245288a8a4501806e6b5f : SUCCESS in 9m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7bd302e349164e2794424c76004af465 : SUCCESS in 12m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ae0ceceab3004d3b95f95a7b6a7a9e4b : SUCCESS in 1h 58m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0add6f3d15044797ae01a15bbbb682b5 : SUCCESS in 1h 49m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2724e762474446448b852a7a7768b5ea : SUCCESS in 1h 47m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7cfdfe8676a44ea1b0361aaa17debee6 : FAILURE in 1h 51m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/21c755b1ca5044e397bf0d2a81ac3570 : SUCCESS in 1h 28m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5bd9a1b7b9b477b83fddab047e7d0e9 : SUCCESS in 55m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4c5bb056b8424e89870c1b656deafdb9 : SUCCESS in 1h 52m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91c698af0df24a53b84cd1a991fd37b7 : SUCCESS in 26m 33s","accounts_in_message":[],"_revision_number":2},{"id":"d369871605dc7a5aea8bc1acc0fcf201f62b1b8e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-02 06:21:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1741ff750ff9a0f6e64f9a3e8d543c4b5ace9ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 06:38:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/600aaf07e5194ea38003dcf80ad6b988\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/66e7bd82409d4ebc8f1dbe1b80640b4f : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/773c13e422d147afb9af34a268542920 : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96a9211a25bd4cd494109c0376e9829b : SUCCESS in 9m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6ef2e8b652fbca4501cfa40b858071cb50feb0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 08:18:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8f00a11e784fed9ee422c77de487fe\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee97e4dd3a1f4f2e8e624628a2309ddb : SUCCESS in 1h 00m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d8acef781964de895ff1b6b0f599b7d : SUCCESS in 1h 31m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/beb1c3675a0d494d97adf53c10594521 : SUCCESS in 57m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b65714f481814a63bd6f1053efc2d5db : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55ee8c4f8a134856b24f5f2318a19534 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afc959fa5f4040a99bb93a50d8e3edbe : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ce1a98ff8d047d89ffb9edf28b73e0b : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37827f77b54434895a3a13103f07b98 : SUCCESS in 6m 12s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2dc01fad91e64308a6115d5b9b9954cc : SUCCESS in 13m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b6ed41dd0a54ca1a40c6352b770887d : SUCCESS in 10m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e32eef5b71424fb28c78e34f12da20c1 : SUCCESS in 7m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9ae901b5336a4d3faaacaa074b4b5b1b : SUCCESS in 8m 33s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e54a583b21ad410da5452d930679575a : SUCCESS in 11m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d71a99ce834242578662c34320aefa85 : SUCCESS in 1h 32m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9c1639f336b4473dbcdf320b028923ae : SUCCESS in 1h 23m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5f8933ba96ac492f8fcdf380741ef30a : SUCCESS in 1h 06m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/62c23753d38a4badbaa30c774deac2d1 : SUCCESS in 1h 47m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a7b08925c66c410f91cd12844978c8b4 : SUCCESS in 1h 44m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9455866f440f4a2996a8f942caf1f8b6 : SUCCESS in 1h 08m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8bca190b7655460aacb8baaea6128b1c : SUCCESS in 1h 29m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cede952f52cf48459efa8296253616a1 : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":2},{"id":"0c1370c279d42d4c2b7a4cb1935352c38fe0807f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:22:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"2781d656e5b98a2c8101490447d12f7fb07b6820","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 14:38:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6328463b54ae035c4094c3b9e6a6cdc189f55c0f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 15:18:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e550b3996b9114cc3eaac147aa3440224b40028a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-05 15:43:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eaadd17d4d257a292ed82d0e3e0846615688109e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 19:06:02.000000000","message":"Patch Set 2: -Verified\n\n(1 comment)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/glance on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: glance-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Glance py39 functional tests vs. sqlalchemy and relatedd project masters\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/glance/.zuul.yaml@master\", line 35, column 3","accounts_in_message":[],"_revision_number":2},{"id":"90c35f3e76fadf320096a62bd33b0c821bd68c33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:28.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889065, https://review.opendev.org/889063, https://review.opendev.org/889062, https://review.opendev.org/889061, https://review.opendev.org/889060, https://review.opendev.org/889059, https://review.opendev.org/889064, https://review.opendev.org/889058, https://review.opendev.org/889057, https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"466f2c148351c24bb182b37b33b68a39ab7984e0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-21 22:44:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9ea30caa735df18f9f76a53a54301c21f4a40891","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 22:58:08.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92485bf866694bc9b6f8a84d0e20d40d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/437f2fb1d0bd4adeb863b0f49b950587 : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3050c2baef6744d6a713c59d8e8c61d3 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4849d827bd04f5cb7a069841d91e2c4 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/525af7e961e04b65abb71a2c326b8b3b : SUCCESS in 11m 17s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/glance on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: glance-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Glance py39 functional tests vs. sqlalchemy and relatedd project masters\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/glance/.zuul.yaml@master\", line 35, column 3","accounts_in_message":[],"_revision_number":2},{"id":"0de766d48d4ea3404112fd42c53397680291e436","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 00:53:59.000000000","message":"Patch Set 2: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df7404f77ef4cf2b91e70cff5f9f08a\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e06a6a5a71e433a91efd8fd2fe23082 : SUCCESS in 1h 00m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70d5557c19cf402982de72289a6bc2c0 : SUCCESS in 1h 29m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7e53ba9428c43359db281b56964de23 : SUCCESS in 1h 03m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0ab7bf0410d408094317d055364e9d5 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9b46ed6910d4aed93a43ba83b4becb8 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/843b39bad9f844ce919ab5f0d170a9d9 : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/951be41f447f4dc1b40a042e771d8d32 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1dfc05ba7e25430592169a1cfed4e70f : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e2e412b8e144f4b863b6e6fa615b86f : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cf51e1afd0214e9487d76b2f216b6b8f : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c7ad82b651d64a7fac340b43d7852d69 : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b6b631ec56b6493397d1f25c09666033 : SUCCESS in 7m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/44a8345f25d243d1977367502e1e7563 : SUCCESS in 9m 13s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/04c999aa60d94ba694893a7b3fa3f918 : SUCCESS in 11m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ad1bc9bb983e469aa3cfadce0011557d : TIMED_OUT in 2h 03m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d9a7572f4e5d40c6bfb9850c5e0a5413 : SUCCESS in 1h 03m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c3f643c2b5544842931f8703fe8266e1 : SUCCESS in 1h 18m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e3c849e0fbc545d99df5b40fc8550872 : SUCCESS in 1h 08m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/107b430cf9fa468a9db712ef3044bf9c : SUCCESS in 1h 21m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/19a041b7e0bf4c6a9c8f2eb234400991 : SUCCESS in 44m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f9f1cf579a144afdad44d7710c179227 : SUCCESS in 1h 14m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bd60dd07b1324c17be3afc08783b09a5 : SUCCESS in 21m 53s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/glance on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: glance-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Glance py39 functional tests vs. sqlalchemy and relatedd project masters\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/glance/.zuul.yaml@master\", line 35, column 3","accounts_in_message":[],"_revision_number":2},{"id":"b79d92bcad58db50db2548b6c54175ddb0f1f82c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 00:54:42.000000000","message":"Patch Set 2: -Verified\n\n(1 comment)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/glance on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: glance-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Glance py39 functional tests vs. sqlalchemy and relatedd project masters\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/glance/.zuul.yaml@master\", line 35, column 3","accounts_in_message":[],"_revision_number":2},{"id":"8800af4cd50edaaf1c208a628afaef4c1f7dfdb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 03:01:15.000000000","message":"Patch Set 2: Verified+2\n\n(1 comment)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb1e3fd0c950418fb97adab7a0eebb02\n\n- grenade https://zuul.opendev.org/t/openstack/build/413c3ec7a61e4bd4b44b21b3b7f08fe8 : SUCCESS in 43m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d941fef72dfa4044b2d3d74e8a72441b : SUCCESS in 1h 07m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/122d3e58024141f3af2dfc71d8e23aae : SUCCESS in 54m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7f7ec4f446542798ef711f987111341 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66d7cdfbe28c4defbeaa58a511e14e99 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7fd5f41e9d42441f9db34e8b21b2a440 : SUCCESS in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d5215cdc169f4e1e9511cdbec8099a9c : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde188c9f8ad49e1824cb2e815ebf719 : SUCCESS in 6m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5b528787fdd44e8b87a110c3716e2d68 : SUCCESS in 9m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/325086ec901842c2b1696f1754aa9e7e : SUCCESS in 9m 10s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ae8df8d3372d4a58b2670ff1b6daacc4 : SUCCESS in 1h 50m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48e498de4a1c4b7fa301ac4e190926f3 : SUCCESS in 1h 55m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b5f8ce89972473eb6424826f354d177 : SUCCESS in 2h 01m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa313b8746494012b4b253623121bba4 : SUCCESS in 1h 02m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/feb0999a1f6842aca7f185f247e16cc0 : SUCCESS in 1h 56m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1c88567e58874723b8a560ce17ada832 : SUCCESS in 38m 20s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/glance on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: glance-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Glance py39 functional tests vs. sqlalchemy and relatedd project masters\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/glance/.zuul.yaml@master\", line 35, column 3","accounts_in_message":[],"_revision_number":2},{"id":"812037693fe390be5d7e36a76b9b405a45aea7be","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 03:01:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"177cd4861e47c51a6ab67355e3f731da52ec8d8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 03:04:03.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb82c3dee750402aaef868badb250df0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aa785e561eb48f4bbb0bd688e2aaebf : SUCCESS in 57s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/glance on branch master.  The problem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: glance-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Glance py39 functional tests vs. sqlalchemy and relatedd project masters\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/glance/.zuul.yaml@master\", line 35, column 3","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"02440705ba6ae3120e24b1af1c5b97c8d1c73a26","revisions":{"d9645a8e4fb8117dcc3b9c2b3bef49d821f5c70b":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/889066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/889066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/889066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/889066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/889066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/889066/1"}}},"commit":{"parents":[{"commit":"e2e39a379a4fbbb4ee38ce0b04602b191c8ef42b","subject":"db: Remove explicit autocommit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e39a379a4fbbb4ee38ce0b04602b191c8ef42b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 09:35:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"Add job to test with SQLAlchemy master (2.x)","message":"Add job to test with SQLAlchemy master (2.x)\n\nWe add a wholly new job rather than modifying one of the oslo-tips jobs\nsince these jobs are run periodically rather than in the check queue and\nwe want to prevent regressions.\n\nChange-Id: I1036b7b0d64c91f0f64b37340490a2e52ee0baea\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9645a8e4fb8117dcc3b9c2b3bef49d821f5c70b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9645a8e4fb8117dcc3b9c2b3bef49d821f5c70b"}]},"branch":"refs/heads/master"},"02440705ba6ae3120e24b1af1c5b97c8d1c73a26":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-25 09:10:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/66/889066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/889066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/889066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/889066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/889066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/889066/2"}}},"commit":{"parents":[{"commit":"d1368745c5a75c331632408ce284e4c782aa23e9","subject":"db: Remove explicit autocommit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1368745c5a75c331632408ce284e4c782aa23e9"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 09:35:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 09:10:13.000000000","tz":60},"subject":"Add job to test with SQLAlchemy master (2.x)","message":"Add job to test with SQLAlchemy master (2.x)\n\nWe add a wholly new job rather than modifying one of the oslo-tips jobs\nsince these jobs are run periodically rather than in the check queue and\nwe want to prevent regressions.\n\nChange-Id: I1036b7b0d64c91f0f64b37340490a2e52ee0baea\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02440705ba6ae3120e24b1af1c5b97c8d1c73a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02440705ba6ae3120e24b1af1c5b97c8d1c73a26"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
