)]}'
{"id":"openstack%2Fglance~889067","triplet_id":"openstack%2Fglance~master~I78e005b51485a224e59c075a00798de5fa5a3896","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-22 03:14:11.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I78e005b51485a224e59c075a00798de5fa5a3896","subject":"db: Trivial parameter swap","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-22 07:21:18.000000000","submitted":"2023-09-22 07:20:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889067-sqlalchemy-20","meta_rev_id":"f5a1fdeba2bb19392aea7189fcc9c384adf2b7e7","_number":889067,"virtual_id_number":889067,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 07:20:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-30 10:10:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-08-30 10:10:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:34:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:26:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 10:10:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6f275b3fc96f4345f5e37d21bc53920333c77dba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20652a8eecb625e2be3ed65de0406c38c6297042","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:34:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3422ebb819a34ddbac4d5fc51470d19a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/faa9f470b77a496cbc116e9128aff330 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f942f29e25e41b79c3c65bf50896998 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1769e359570a462daa7224eaf19cf9ed : SUCCESS in 12m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e0b73fef49f107297151fbbc60d8be2450f7d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:26:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37b7779c2949433c86c316848b8d3546\n\n- grenade https://zuul.opendev.org/t/openstack/build/a63c5b9246554607b73e7e1021f191b4 : SUCCESS in 1h 00m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/381b9657a83a46749c7c34e88d28bb3b : TIMED_OUT in 2h 04m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/687e24eeba6c4f9ab88235c90e859537 : SUCCESS in 1h 13m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0606055f745492a88f509426c7e19b1 : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f11051e66b7649648ea049e4ae36c229 : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dddc215de03b4821ae8fa3eb3f288a10 : SUCCESS in 7m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7a2f1bc5f654b92971f927bfbb6f1e1 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c68590e34d4941b306f46d1d0b7698 : SUCCESS in 7m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/676112e59b354d2086e6c6092b8930e3 : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8a543b3510f4ea58e1ab691cbc4198e : SUCCESS in 11m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1d85f358d4524d90830a0db7d9a08ada : SUCCESS in 9m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fd2bbf7bd8054634a355c069ba5c5ee4 : SUCCESS in 8m 49s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d753f149e2e240fba5bca142073457fd : SUCCESS in 12m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0823c4bd4d204687b438e9cb370d4640 : TIMED_OUT in 2h 03m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cb3b5b17dab34015a902241998ab391a : FAILURE in 1h 22m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/51e95b27d0fc4c7c862e756ea8d37160 : FAILURE in 1h 34m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d80986faf624251b27b60a208b5063e : FAILURE in 1h 53m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6b6beb8c34e49e9bb8658707bb825ea : FAILURE in 1h 29m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/875aaa6c2ce644878cca56b7ab16bf82 : SUCCESS in 1h 09m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/95f1b862d8df4db08ef0db8dc8a5f667 : FAILURE in 1h 28m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/615dab066c254e8dbf9a07856581af13 : SUCCESS in 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"73073e2b0909a9811f71aec55b16e8a51010e72a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:10:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e943c32014ab63583933adbbc8d1e252b3354250","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:28:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/264362a914524afcb8232706d17650e5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b662a3a261504cb588234e489d335234 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ab2758ff36c4fb78e859032ec6f06f7 : SUCCESS in 13m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1bff167686d444ea910561e5ee5eaaf4 : SUCCESS in 6m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f061d43b8bc817e75575200b5c345de1b6f89ae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:39:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/168f66ea7fad43e0ace6b24b67a685d6\n\n- grenade https://zuul.opendev.org/t/openstack/build/2a1299971ab94d9d8e2b43bd92e63b42 : SUCCESS in 1h 40m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f4f2e8af0304c60b0c1d40150b8e197 : SUCCESS in 1h 52m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17debef9ede64896be982d185b1fe5f0 : SUCCESS in 1h 10m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d62808836f40f190be6387409f832c : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c19031059f4242d596ce5ba176ba795d : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9c4b23241e3497daa27c37480b807ab : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e56a479ae56b45c88ca68642d5b0e11d : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae1e00172c7433786a1aac4379de9cc : SUCCESS in 6m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/241ba1db62e54912b8b97856468609f2 : SUCCESS in 14m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19790b9833a54ff481a5ba719f2efe34 : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cbb391e14fb14b768d5897e9e20a8bab : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f5152bb43e434617a9dc30f89134129b : SUCCESS in 10m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0a54380a63f948bfaa537b3d83caeaeb : SUCCESS in 9m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a9aaec00277142bb8991c8bd5af247ba : SUCCESS in 1h 57m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/25791446802644b6826ef3f13e2050cc : SUCCESS in 1h 31m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fa1b9289a8b44701b1a5963d657e55c7 : SUCCESS in 1h 35m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa38d1cd86d84b78bfd6d5bbf3af21e9 : SUCCESS in 2h 02m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/900c1622b64047a28d1a0a36e5a57af9 : SUCCESS in 2h 01m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a429501417624effb009574276f7e07f : SUCCESS in 1h 00m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/86f317e4567a4e999f9a6b59df0ffcee : SUCCESS in 1h 38m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1f568568305641b489dbfeae0f4e8860 : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":2},{"id":"2f5374103b4e764b01edc44cbbb8664b3168e585","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-30 10:10:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27021780ea84d02d1b22f9aa278e1177062422df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 19:31:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"03f92e91221b6ff6bc25a35ca4424a3ddd4d453f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:28.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889066, https://review.opendev.org/889065, https://review.opendev.org/889063, https://review.opendev.org/889062, https://review.opendev.org/889061, https://review.opendev.org/889060, https://review.opendev.org/889059, https://review.opendev.org/889064, https://review.opendev.org/889058, https://review.opendev.org/889057, https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"78239d3b61929d0636bb74698d2d3fea37de9c9e","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 03:14:11.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fcec98b132b1b35639e07b33cd67cf538d74e809","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 04:34:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d9a98651c3a4b2ea050fff0238fd4e8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e4978707613f466eb291d1dd13ab1298 : SUCCESS in 26m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0da1be6d20d4958a918d869592aebb6 : SUCCESS in 26m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aae70a443ecd45b28a61d1c680c38b39 : SUCCESS in 23m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7aeac8c144824d578e985007a01a4081 : SUCCESS in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6561f04074c13ade5d7e165f444c94df1ae99ecf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 05:21:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4469efe5cf3e40a4a536b86fbc2c3350\n\n- grenade https://zuul.opendev.org/t/openstack/build/738ad141e1154cea8b212198e6c0a24d : SUCCESS in 56m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8126add0909c4dc78b630e16d90cb7e3 : SUCCESS in 1h 02m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6ef6da52e0b4738ab4adfabc2a2fdaa : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6688a54ca834d47b6409a992f1fbccc : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a0f9ef2c3ac94428826be5766c80cbf3 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22e5903dc8194abda462b501c2ef8f11 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6953961d255c4880b5003415ebcbf1c3 : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f720f47a2dd442009b7005f703a2059e : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87360a7255f64e089c63c34189659565 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/dfd7eb1ab1b74837b7cfb89b946f396f : SUCCESS in 16m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3064e98b879e4e90836d4c0da80f743b : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1c6711095c9a4b18bf9d73ea86389ee5 : SUCCESS in 9m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a4a70276d7494ceaa61359df282f39e1 : SUCCESS in 7m 46s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/75e5c016ed714ffea021cb8371fd957e : SUCCESS in 11m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a12292827c84b8ba63ee78bf65fb57a : SUCCESS in 1h 58m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21986c52dd3043c38a8075597935b236 : SUCCESS in 1h 58m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ba7d96f9969e4eb89a21ec9b01639464 : SUCCESS in 1h 52m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/03ad9f1840354a028cf497364c08f09b : SUCCESS in 1h 53m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a9e901ee76d44496963dcfa46d85a28f : SUCCESS in 1h 42m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c725d9dfc6846aea2acceb51bcf2fe8 : SUCCESS in 46m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ec5150e9d6a4f76bb427849fde956b2 : SUCCESS in 2h 01m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2fd53215a5c44560926aabdf9e1aff0f : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":2},{"id":"4245259707c237e8f17352f5009f1cdbd935d01a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 05:21:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fd93f45b55597a489328633dbd7929c06211fc9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 07:20:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13d97b500733499ba571b7c685aa075c\n\n- grenade https://zuul.opendev.org/t/openstack/build/5086bb66471f4875949ba50d1ef11aa9 : SUCCESS in 47m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/263307660c1744a0a9e9e9e6eaae7c08 : SUCCESS in 55m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bec0c0d841964f48bef8c7cbf125b548 : SUCCESS in 59m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a89050dd0c34f48a7b2d31fd1054720 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6c6810743fe4803afd2f6beb2b9a0f7 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61078ea43b2e43ee8dc6c6371978ca1d : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41880553fbea4c899ca144762274d625 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177078236bf541bf84729d93451d0809 : SUCCESS in 5m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fe748d79964431498ece59c253215b9 : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0046d8ea23934dd1bed1f468ad5cee92 : SUCCESS in 8m 17s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d04e77aeb7de4e729568b01fe5fdb0ee : SUCCESS in 1h 17m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/98f5ff6ddc2b4f75b5dd3e6185a971b4 : SUCCESS in 1h 20m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0705ce7734d842978864d8af03695356 : SUCCESS in 1h 16m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9e47bb6667049dfada361c0432deff7 : SUCCESS in 43m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40377097c94042918b59ffc5acd97d92 : SUCCESS in 1h 51m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a8969bbce5c74a91997e7bd5dada703c : SUCCESS in 26m 29s","accounts_in_message":[],"_revision_number":2},{"id":"096c693427d030364d8f7ffc14fda7898ad3ed22","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 07:20:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f5a1fdeba2bb19392aea7189fcc9c384adf2b7e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 07:21:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fca70746100a491c92bb8a4611f7c36e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/399785482faf4061a96e064654cdc924 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"62893e9373a9659957aa5b0263e7a19d57f549d2","revisions":{"071c115873a29bca7fc36d96620678b943827d57":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/889067/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/889067/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/889067/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/889067/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/889067/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/889067/1"}}},"commit":{"parents":[{"commit":"d9645a8e4fb8117dcc3b9c2b3bef49d821f5c70b","subject":"Add job to test with SQLAlchemy master (2.x)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9645a8e4fb8117dcc3b9c2b3bef49d821f5c70b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 09:51:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Trivial parameter swap","message":"db: Trivial parameter swap\n\nAlign \u0027_image_update\u0027 with \u0027image_update\u0027. This caught me out once or\ntwice.\n\nChange-Id: I78e005b51485a224e59c075a00798de5fa5a3896\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/071c115873a29bca7fc36d96620678b943827d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/071c115873a29bca7fc36d96620678b943827d57"}]},"branch":"refs/heads/master"},"62893e9373a9659957aa5b0263e7a19d57f549d2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-25 09:10:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/67/889067/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/889067/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/889067/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/889067/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/889067/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/889067/2"}}},"commit":{"parents":[{"commit":"02440705ba6ae3120e24b1af1c5b97c8d1c73a26","subject":"Add job to test with SQLAlchemy master (2.x)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02440705ba6ae3120e24b1af1c5b97c8d1c73a26"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 09:51:25.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 09:10:19.000000000","tz":60},"subject":"db: Trivial parameter swap","message":"db: Trivial parameter swap\n\nAlign \u0027_image_update\u0027 with \u0027image_update\u0027. This caught me out once or\ntwice.\n\nChange-Id: I78e005b51485a224e59c075a00798de5fa5a3896\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62893e9373a9659957aa5b0263e7a19d57f549d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62893e9373a9659957aa5b0263e7a19d57f549d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
