)]}'
{"id":"openstack%2Fglance~889068","triplet_id":"openstack%2Fglance~master~Ifc1605e8e89c48b071ea505e40a9488df208fb37","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-23 08:44:49.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ifc1605e8e89c48b071ea505e40a9488df208fb37","subject":"db: Use context, session args for metadef DB APIs","status":"MERGED","created":"2023-07-20 10:09:08.000000000","updated":"2023-09-23 11:46:07.000000000","submitted":"2023-09-23 11:44:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889068-sqlalchemy-20","meta_rev_id":"0e6ace7487018fceb0b4b9b366e0641c14f236cd","_number":889068,"virtual_id_number":889068,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-23 11:44:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2023-09-04 07:27:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2023-08-30 10:11:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 17:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-22 09:45:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 17:20:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:37:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-30 10:11:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-04 07:27:54.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-09-05 17:20:42.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-09-22 09:45:21.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"6ee9c4821ae5418a1e7806059f1b22d8894c13f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:09:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d4107e4ebdb886886685aae41fc6396829b72ec5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:36:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2cbb840f12c149dc99953d72619aa06a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bb028b103f1b4528aa0dbfbcacffe449 : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cbc4eb2af7604a33b5362f7ca3994f98 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/965575d41a4a4741a060e7200e8ebacb : SUCCESS in 11m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"453513243178bf1a4d28cf88bbd87975d9153a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:37:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f70eda05b6f487491aed39396bcaae4\n\n- grenade https://zuul.opendev.org/t/openstack/build/987a7e646b8345a5a00c9a28f7da0cbf : SUCCESS in 1h 01m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7e15a15d3af4f5e98c4219d329f99c5 : FAILURE in 1h 39m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b2a44284b094c0699c04a1f8f1e8f59 : SUCCESS in 1h 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ffbb4b3abff4f20a41cb3908c67110d : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d8adfe87e9d44d2af9489ecde1ab359 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a44530766b2d474389167ea071858efe : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30827dc73b9d4f21a9ce93a5a6df69f1 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba1fa94eabe0493c9d080629eaf69237 : SUCCESS in 9m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c601d05a4c42455a9dc8ac3f0c8edb7f : SUCCESS in 15m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94aa689f7eaa4acebeb78b068e3222de : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c881c2ae08b4466ade391bf8fd4660b : SUCCESS in 8m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/43f6729ca22b4e039e159e73d65b3ee0 : SUCCESS in 10m 40s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1ed22e3023bb47ca86df423efc257f79 : SUCCESS in 13m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/85c7d10005eb4ce8a3a5e37c5a4d3f18 : TIMED_OUT in 2h 04m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/94120f83be6c4a259e17cb789b825e63 : FAILURE in 2h 00m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c5eab138092c455b9cb908357842329d : FAILURE in 1h 22m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5c96aa10ca0b4eaea83b3131fdde0896 : FAILURE in 1h 50m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ef7674ec417e4ac7bc67a92bd32c5cbd : TIMED_OUT in 2h 04m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b72012c3f92b4462801d5b0f0af0f41d : SUCCESS in 1h 04m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/09ccc339c3fd4a0cbb7dda907297040f : FAILURE in 1h 57m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7259dbcff8844f2c9c51e9173a847b83 : SUCCESS in 24m 00s","accounts_in_message":[],"_revision_number":1},{"id":"dd58acd5bf6468126cc957c6c5167eb14f8913ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:10:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dd49d1fb1e7cd9dab8b888c8e5f7588cadde7477","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:30:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f900f0c2fce44b3fb993cc0712197b87\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04867afaadf44b1e897e5dd872bfee39 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d134f5ea935a48cf81754afa56365d2d : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69e1d3ee017c4b3da08402668612d656 : SUCCESS in 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"88eb2987a6456b5d4af60ac166fa6761f6a50198","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:36:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90eaefcfb3b34642a10ffb2621313ddc\n\n- grenade https://zuul.opendev.org/t/openstack/build/f4090d1df1044e238a23967a6c89ba80 : SUCCESS in 54m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4cfe79348078411985b39214b8e7adf9 : SUCCESS in 1h 37m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6251d4d0d5d14a939ab78cf8894f643a : SUCCESS in 1h 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c03564f9b0d44a9ea487c3de02a39102 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/25dddf803a714b9a99e389a42774c4ea : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ae76b1a311847bbbf8173a25c9eff5d : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd802fbe3fca4d289333a6702698862a : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a6c0c4a68149748811bb82bfd9ae8c : SUCCESS in 7m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3e631f42e0a04ab68337fd225b38a245 : SUCCESS in 14m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6172754d253242998775c06c3f9f0fc4 : SUCCESS in 11m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/47ec54716c2448a88b0162b286cd52bb : SUCCESS in 8m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a930389e770e4f75b4430545e5309096 : SUCCESS in 8m 22s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c1eeec249d6443a28da6b86de167c387 : SUCCESS in 15m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6dac89288963414aa2fae9ee165257fa : SUCCESS in 1h 52m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/18ed444343294a73acf643f3870adb68 : SUCCESS in 1h 38m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b870b8823124fbf9ff60b4ba9fd0e78 : SUCCESS in 1h 36m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/44d0497156c44b0f8faee304756e23ad : SUCCESS in 1h 39m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7770b55ab9b446459f8be332d048ffc1 : SUCCESS in 1h 53m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8e6395b85d9478a9ec866c128fc36a7 : SUCCESS in 1h 00m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7aee65c1077411c90848ed74b92c666 : SUCCESS in 1h 29m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f5802c25f8d469181157d27b589124c : SUCCESS in 26m 20s","accounts_in_message":[],"_revision_number":2},{"id":"34381585dfc02a100fce759f0f0f544f88fde1ab","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-30 10:11:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1784b46758a39d058706e3bb39fb2c33b8915c01","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-09-04 07:27:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"457600618382374982848b56d96a4b191f977164","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 17:20:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ae1f7867f6c132d75a6008971ed34b95a4f5b9fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 19:31:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"85aea9d3d250eee77b70f785d7c3c4c4da73e224","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:28.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889067, https://review.opendev.org/889066, https://review.opendev.org/889065, https://review.opendev.org/889063, https://review.opendev.org/889062, https://review.opendev.org/889061, https://review.opendev.org/889060, https://review.opendev.org/889059, https://review.opendev.org/889064, https://review.opendev.org/889058, https://review.opendev.org/889057, https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":2},{"id":"8e33701592a41c2dafd99deda07df935c3cc2518","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 07:44:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9c5ccaef44b36506f78a83eb1aabdc4e4cd31ca0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 08:00:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdd7413f2b2b408cb56bdec7096d8e7c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8d211659401414694f684c3f45ad1c9 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/535bcdbc9f1b4e588aedf26c3364956b : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cf56045f7cb4c23b9e332e8b30ab9fd : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/415f020bfb194821a683cd69c61c301d : SUCCESS in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f2d15e4b794e7286544fb0aebc1c08b77b8050d","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-09-22 09:45:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"43d571a9ba3f03f01969fa29c03a7c32388af39e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 09:56:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8f81cf1c9b949bfb666cc33a7d32d5e\n\n- grenade https://zuul.opendev.org/t/openstack/build/61096640f6424375ab1365d206d560b9 : SUCCESS in 46m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/280547337e554200a0da311e46ebcbb0 : SUCCESS in 1h 14m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/36f0c4aa10d24924bd7e493cf091d1b8 : SUCCESS in 50m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17477cd860ce4d60bdbeccfb90663214 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0895b7a22ab849908c3d3487183787c2 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd758c3b941b429a8a4936db52f96953 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/924f6243f9f9407cb51fd9040a4994e3 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6ef4328b0f349d0adf29900ab2e6b4a : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8efaf1f4fb9548098f8e5f86a96e8d12 : SUCCESS in 6m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/92889a1f2ae74395a8253ae13af40334 : SUCCESS in 11m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8e0b9650d5b84fa286a2a5d636d70c2b : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a7eef6d915c54a76a3d8e41e13da6a3d : SUCCESS in 7m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dc3efc25e8364419a0b97dd464f9fb72 : SUCCESS in 8m 59s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3e1f0792d7974388a541d080ada9c9a4 : SUCCESS in 10m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1283206dcc134e5e855585a6bcebf5b3 : TIMED_OUT in 2h 03m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0fab7869828b4815847143d9968ce109 : SUCCESS in 1h 42m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/333ddc63bc09400dad2b065b1cb3ace8 : SUCCESS in 1h 39m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d55e6e24d60f4cd9a9d1a3595eb9ec75 : SUCCESS in 1h 46m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2b9322f444df4d3f8faf26eb6e6ae3e9 : TIMED_OUT in 2h 04m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c1c45200f094287afe86d2ed24818f9 : SUCCESS in 43m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f13cdb2cb81841ef86c93588586b59a3 : SUCCESS in 1h 27m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f32c1b975f6049098c2d786ee5509e65 : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":2},{"id":"579545acaac5761dacccad417f1d4b397ea05451","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 10:14:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"63db0360b346b390deb07902d26f8832f080529c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 10:40:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b05a1aee49f84e09b1cc5543efb14d42\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3591b92e4e034c43ac0ce301fe070c73 : SUCCESS in 13m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5064c038a3f04849a8b2c415f3efda95 : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/816d1de61cfd43d2a673a8bd506bd84a : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/478452c25c8144a0b70a066dd6277316 : SUCCESS in 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"84526dd610e9b3e8852fcb25b090fa7666c365b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:16:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8856efe793904ccfa65fb1c3325e8bda\n\n- grenade https://zuul.opendev.org/t/openstack/build/74968ca646df496db44bc943e2f60a8d : SUCCESS in 39m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/477866b9d4ce4b51b9073bae7355842f : SUCCESS in 1h 32m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7c874542aeb41978647c24c0770fade : SUCCESS in 48m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9044f3c220f42a191ab8b605d647ec1 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a70a077ee5ff49fca531e4bae7297cda : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe8e17c56d574624adffe44aedd0d6a9 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3029beb067ce4ca3890649caf4380be4 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbde9bb651dd47848245e8cd8c93845e : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe1a739e0ed4f2db0e3aafbe5e2c814 : SUCCESS in 7m 10s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c91236a1455b4c3981e6bded082e92a2 : SUCCESS in 14m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/560330d793a24dcb8767450d5b8aed7c : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8a36b5f053e94857a9aada79d51a247d : SUCCESS in 9m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/841d3973bca3483c97604a730ba56e7b : SUCCESS in 8m 46s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c88234be752c4829acd2013a7ac5f406 : SUCCESS in 11m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a7b794d1ea2a46d0a8c67c1b3553ce12 : SUCCESS in 1h 46m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/83ffc79eb07440ec93dd32579fbbb601 : SUCCESS in 1h 47m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ca0ceda8109c4776853466fe78488001 : SUCCESS in 1h 41m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7482bfffa5324fba9143e230966e43d2 : SUCCESS in 1h 42m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a99b515552574126a505829963e8ecd4 : SUCCESS in 1h 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43c2077acf284b36a343b36b3c723a62 : SUCCESS in 52m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4b248958fbaf44b3bd8b15baa9517031 : SUCCESS in 1h 53m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ff336d13cd4e43279ad7667f9323d21d : SUCCESS in 41m 53s","accounts_in_message":[],"_revision_number":2},{"id":"4d2d3c21f3c1c3897960c8d2ae2fce2765b46275","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 12:16:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2244ae05865258268d39874de6349e50ffbea358","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:28:48.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cb044558898d4bb982a66994899e1014\n\n- grenade https://zuul.opendev.org/t/openstack/build/a93c2e19f09a44e78d6f223c7981323a : FAILURE in 54m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e7c3b9798cb4fa284fc59f934ab4488 : SUCCESS in 1h 24m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3e7fb787fc74724a0a8a975065a5c76 : SUCCESS in 1h 00m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7210862fc35a431092393388794dc587 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65a1203577314000bc3212cd33a65ee3 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53d300a2f9ac4a7cb4f4129b84360514 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95a39e156cd34f64890460fc48c3ccbe : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3316f3a677c04e48ab741dbfb93cc0ff : SUCCESS in 6m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/254f40c74ac543d2bd247a4007653c19 : SUCCESS in 9m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b9371bb227f24149a973cf3c46af0d6a : SUCCESS in 9m 02s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9f4f803bd2a64b30afbdfdd40926cb9e : SUCCESS in 1h 43m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7fa05f70e5a64e9e91c0e561d10cb9f1 : SUCCESS in 1h 29m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e8c4c3675eb94016b718e45465af4cda : SUCCESS in 1h 56m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c352b7a9a222418bba91982453fd1043 : SUCCESS in 41m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e65560784df74a609179e3e5074277ac : SUCCESS in 2h 05m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b978d11512d143c3bcf1f0ef1fcfb8c0 : SUCCESS in 24m 33s","accounts_in_message":[],"_revision_number":2},{"id":"63cbfdbb45f2c4d6b46950299005e70550413be3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 14:44:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1d07e51253956a9fac00ad79afafe1851c4337c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 15:33:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63a1ddf74fcf4acb96a1d65b8441e14a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0de194100af2479ea5bc9f5fd6866bac : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3364ca6925454a888c5aadce0c2a634f : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27fb8a7770284cff922bf05002fa9eb3 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/251caa8425b34e0ba02f0918dd8dced4 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ab1c4c15e9c0beb1d7328360874102908b04edc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 16:55:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b34bd3198c44885a8c30eebca2e946f\n\n- grenade https://zuul.opendev.org/t/openstack/build/6930275295cb42ccaa9261aafe668e9f : SUCCESS in 1h 01m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e733a54526e49d6b83c8c720ba70c81 : SUCCESS in 1h 17m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5329aafcc8744af865896898e3eafc5 : SUCCESS in 1h 02m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6bc874f49e249f1b4814955f7564e6f : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de3f113e61b64559bdeac7f9c9c34a1a : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eafedd9ff36f45cea97383e1ad3abb36 : SUCCESS in 6m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aec8f909b67b49e2b5f45ba4bf3a9274 : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5677086540514c278d8ff21aea69f24e : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab789793ea7e4ba6a3a46c0184ec9d25 : SUCCESS in 6m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7d28cb33f62d4b1095c42130670767d7 : SUCCESS in 14m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e8e45a460de447b081c18ebb30186e09 : SUCCESS in 8m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7049bd00d62405b9c95240b471b6208 : SUCCESS in 10m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/10165667aed3413fa2eca77175f9a845 : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4ad9f44134be4c6ea5c5a5b742a9def4 : SUCCESS in 11m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/26b73bf0335e46cca0699c14cc98fe02 : TIMED_OUT in 2h 03m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/11c524e1051046598fa123ab7ef72d31 : SUCCESS in 1h 49m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6498c3f529f0455da792ca9f97187694 : SUCCESS in 1h 36m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/690d0dc4fd534f1e804b817739a315cc : SUCCESS in 1h 43m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9c3b0617d33240c1a163aad2adf352fa : TIMED_OUT in 2h 04m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2733799f5d3748b9b67080f5ee54b8af : SUCCESS in 1h 05m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e4db82405e9540df8f8c955c89e89170 : SUCCESS in 1h 49m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/388128382b2444d5a31b04a58ebe5e72 : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":2},{"id":"3fb022c426652ad690184dde90f5b2d78ed0e8b1","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 17:14:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"893247a5ab6b91dc3e69483c05a0d42351a16079","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:06:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3711b26a93645529d904e14e2f6c70c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d7668279029463da625e066b4f53a48 : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dde2d2ce7ae14563a199617c518c47f3 : SUCCESS in 16m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c521333f12a94d71b193ed12cf9e316e : SUCCESS in 15m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/114288f10a9d4a55a6b0119d7a32d7b1 : SUCCESS in 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f672256b8c5c255ebb577d67f010a01c1ae785a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:19:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b0911d279bb45588486db56c654ea19\n\n- grenade https://zuul.opendev.org/t/openstack/build/7cd247d66fd445c5b43acb7355a518da : SUCCESS in 1h 07m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d6eca80e6d64842b74b45b0bbef368e : SUCCESS in 1h 03m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2601f8e49a134b6a9582def608540db8 : SUCCESS in 59m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1d3d172b9247cb8c31b563ecdfcc57 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d0df4551831412b89bd18757a21d3c2 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d06d5e024dc4e07ac5185938193f33a : SUCCESS in 8m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02650f32d1aa4022be7a287081e304c1 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2b17e3cb44b4d7e9300e81b642873bf : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5dbef27cf064e5f85a4e0b331b5270a : SUCCESS in 5m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c85584ee800f4654a94a43e53806f7d4 : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/09123f806d264b33b007936e18dfee9e : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/187891b2726a487ba4f4cdde1b9ec83a : SUCCESS in 8m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ecbc572c62a5406dae6aa09acc8522ab : SUCCESS in 8m 47s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d663ce40e79f4c8dbcecb2000e0b6c6b : SUCCESS in 11m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/629accb849224ddbb1aef9b5fdb59113 : SUCCESS in 1h 59m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8e3c44a428c24cf5a77e4604b3c15231 : SUCCESS in 1h 44m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b5bcb1be00714b508210c1f6b3ab0724 : SUCCESS in 1h 47m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/393cb0c57c8a492982ebc98a10deea4a : SUCCESS in 1h 57m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4b7e51e93b1142348c87e4f2f77d3a7e : SUCCESS in 1h 58m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7669c1062a964e78b1ebc5fb14fba4c7 : SUCCESS in 1h 01m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5e3f6a56675445258a5cef426d6a38d4 : SUCCESS in 1h 46m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9a44d9a9bc1d4b4ba8d35358c0a18b47 : SUCCESS in 46m 19s","accounts_in_message":[],"_revision_number":2},{"id":"687ecd1e0328bdbff3b9b03ec2744348c24af636","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 19:20:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9352fda8b7271f8275f3bf63c7bebce3b60e2ce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 20:56:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a7b19e190774616b585be724034cabc\n\n- grenade https://zuul.opendev.org/t/openstack/build/fb021ff00f9c4dc8bf388127877e6d22 : SUCCESS in 42m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f829df6b3b24496fa73432747519e163 : FAILURE in 1h 03m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/490a7aa841ba40f1a366ed9f940e760f : SUCCESS in 41m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d012c22e174cb7bfd270a7877424c0 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba5579ece11d454eafb9137f53b4a6c4 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aca84452f2cd4a4eaf1a56b3f0da411a : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/650e2ede6e234a719e37bdf81887e32b : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd9987490614e0881ed402eaeededb8 : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e71607679b2406eaa193e2f1d029910 : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3eea30e4e2664b56914d094a157bf7b3 : SUCCESS in 8m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b6fb3740357a44fe98376a883d535c54 : SUCCESS in 1h 18m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e3f558e8bec043d7be035a7ba4692f41 : SUCCESS in 1h 27m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3b491aaf569d4db0a7713493ec18808c : SUCCESS in 1h 21m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d0c33bce828f47d48f43dd2b7769fe6a : SUCCESS in 1h 05m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ff08d9160184aa081ed17b545744cff : SUCCESS in 1h 35m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ec1baa8fa6e64dbc99cf14d75aaa452b : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":2},{"id":"d1e93f5cfb414b8d8dd6f2965aa7d20d9bd62b88","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 21:14:31.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9920523da598ecfa6d39f996b219f9b1ca50536a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:01:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87f13d59c9744273b42b6ac61594cb19\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ed8553d26e94736b82dbe780411f1ef : SUCCESS in 11m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a9fc44cc07445358c0c9c2fe09c5e71 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f987974a5a6948508dd9df4fce11cfd4 : SUCCESS in 12m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/98b1c1dd7b114796a45b50cf4a4de356 : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"259a92c5dabea8bc82a55c2e76d27c116ce32afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:27:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4ca1afcdda34d38baf04b1184cc5e49\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b8a970c9e364e3e972fe894849a6fe9 : SUCCESS in 45m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9942018829cd49b1ad049366b7548f49 : FAILURE in 57m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ff49a162351430f9d07f67b668ca1de : SUCCESS in 52m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af7907ba0ad040d6b8d9cc5e262c1344 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13f7ef5270c84755ae7244b35f1d4e05 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68e9e7a76ede40a4a30a7b0f8ddcdebb : SUCCESS in 5m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2442c99b1e42460ca68ddd0801ae45ea : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd87ef764c5e414fb3f7f994e9966d36 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761fc099ef54466188e6c1e393da0f45 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4e1306a848f945648e43deec700d8535 : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb6e163156b24f02939ed06f8bc14691 : FAILURE in 9m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f524c98e058f4430bbf37421159569cb : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/daf9f0fd1c484843bf91ba537f5bf493 : SUCCESS in 8m 39s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2266d715052c4fdf88e8270263b805c8 : SUCCESS in 12m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7a697daf8579408087f981d6a5c82ed3 : SUCCESS in 1h 56m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/35b11c609c534077a8f0cf9d01643d2b : SUCCESS in 1h 48m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/faf3447be92944fb9e6c89f578f7013b : SUCCESS in 1h 38m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1d7013990502445aa4ef1f345a36e09c : SUCCESS in 1h 29m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6373e3df6eed4fafa782641e1ec125f8 : TIMED_OUT in 2h 05m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7df2d268494b47268f206572a873180d : SUCCESS in 41m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a6a2b2e89b834cc78f734341f2bf9f5f : SUCCESS in 1h 52m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98829749a4254c4491de31b63daf76a4 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":2},{"id":"db5d77c2ccf85b4e1097700b68eb7f3978920152","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 23:44:35.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"23a9067bff4b3ec2d1d4596172aff73d7b2c4295","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 23:59:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a47f3a48e04979b225758613ea89b2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/005f11f4e0a5491b87c9ff2563165372 : SUCCESS in 9m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95c8c450fc094de9835afb95c4b30946 : SUCCESS in 12m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b69428549fbe4aaf97b637ed153e0a94 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ee5c0c2f28224070814c252c48bb9743 : SUCCESS in 10m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f7ea2e661706c981f21450ea594ccd9980598b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:32:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2613b8e331b4d52bbdd6b4f76544045\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce11d39e6eec4577b8e39871aa0a1a6b : SUCCESS in 1h 02m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1b61e4776874078bf68e3152d6f43b7 : SUCCESS in 1h 39m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/71511678a3324f8485e098e726f957aa : SUCCESS in 1h 02m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0290b051ee642c68b6944fed14d213c : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4dc144fa3064f75bb2a78813890a8de : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fce889937e3d47efba780fcedb11b692 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f75a24e3ad024cd5b7c42d526f687a20 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/263609b80f12468a99788c534d3ae479 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6213ee29d974480d9affdd44e65842d1 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/27f48ada43364a20b3dd5b6fc843d770 : SUCCESS in 13m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c54fdc7474e4d5d9b7d5b8335617d15 : SUCCESS in 9m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4e53d09ce5f4cbe977229bc097738e8 : SUCCESS in 8m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/92657c8747b24012967fdc7908abd773 : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/aade67e4bf594d6b9e0b0ee50c8564d0 : SUCCESS in 10m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/752c39e483b44c279c439919016ac8c6 : SUCCESS in 1h 29m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d0f90966c5f458d8be8cf40d4d2c54c : SUCCESS in 1h 27m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2594de7cae544d28a26326d80da72cfd : SUCCESS in 1h 21m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bac9fdab785946b0849ccf61ca8ae681 : SUCCESS in 1h 20m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c02812d121624b0c8401f0a99606300a : SUCCESS in 1h 26m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c264b85cfd0f4a16b22daae6a1153230 : SUCCESS in 58m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dacf15b156eb4a888071b70e55577a52 : SUCCESS in 1h 28m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/94efa5d642544cb5ae7d7fc65100db62 : SUCCESS in 40m 59s","accounts_in_message":[],"_revision_number":2},{"id":"40cfd928c3182713ab5e32c98f709ff50a2172c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 01:32:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a502e0eabb08c8bf18bcdf444398c3b5dd4d2cc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 03:51:53.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/157013ad8ed24219b812e0f2f427a965\n\n- grenade https://zuul.opendev.org/t/openstack/build/967f0d33427f421587dddcecdb97fbe7 : SUCCESS in 40m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e768d785b7264f5693475d1ec045872f : SUCCESS in 1h 05m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3a4b0fb22e434bd8b1d1c31166c39098 : SUCCESS in 41m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a2d001b7de45798cb42db8d6715938 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2c522314f624d719adf954419ac2d78 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2aa4fece43c4023bdadebea1b66ce3f : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94c46dea23734ba7b9727e6af72b1359 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c63b3c57844472cb9dad270ea554069 : SUCCESS in 5m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/119fb6d9f8a84df09a1d40e64718bd44 : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81e8762bbbf6489498ae5c66ee4be922 : SUCCESS in 7m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dcca9cf59e7b48e989d5d1f36e1a6979 : FAILURE in 1h 29m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d272ae031cb54fcd816f5f66a1a871c0 : SUCCESS in 1h 45m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cf6add0b67f64b64b0c09a636539cd2f : FAILURE in 1h 14m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2243af210d7b4068837645507a456813 : SUCCESS in 38m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3ecee4348e6c4eb7a080b57ebac3d70a : SUCCESS in 2h 13m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/44e45f559d96420d996b0702f8eb2be2 : SUCCESS in 23m 07s","accounts_in_message":[],"_revision_number":2},{"id":"242b40193984b9b8c39a05ce6ae8d58089c94d5d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-23 04:14:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9b244c13260183d69725901743cabcb1dd58a133","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 04:27:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a03435d39a1342e7bef7dfd885373da0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/326ad8ae475f4baa97f07838470af265 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44ec0343d8ea4357992464548d4f80e0 : SUCCESS in 10m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97f4855c2d884c9dae0ebc57e24ff2d3 : SUCCESS in 10m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1f52563f37414db0ac675904d462ae48 : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"68630d8f1beb329e222de82d12910cb95d6f38d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 06:14:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d25588e6dd428d8c4c94edd80f85eb\n\n- grenade https://zuul.opendev.org/t/openstack/build/b037efc143f1486ea2e04fdda06a215d : SUCCESS in 44m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/72162cadf3284c0d9c68c8d14ccdc0af : SUCCESS in 58m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dfc319fd1c7142118de893cb837c0eea : SUCCESS in 1h 02m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f95c617ce069462da5dd444179a247cc : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd8cb643786e4d21bbaf7bcee0fae6a4 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39728e94c6614987acaff298f62e7602 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6139d70bf244de79523a3c76be24ceb : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/231edc2d98424bf0b5d71d0f7567cd71 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec5365d3f40d44f1b94b9d4e05732bed : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/803b976a9a944c4aa5b9b272f4433a27 : SUCCESS in 15m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bcd8818aea15468d8e48e6c484c95436 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1708c36485ab4a31b89acd917823d1f2 : SUCCESS in 7m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e93eda4889094443a3e760d3b9fdf916 : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/00027a50191f47008ac20f77e1aa24a1 : SUCCESS in 9m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4564843cddcd47269412e54935bf4999 : SUCCESS in 1h 37m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7ca2f4f6a369400cabae14fe2ffca646 : SUCCESS in 1h 15m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2cd5b45faad54027a4d5f2e11dca2846 : SUCCESS in 1h 48m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/de908c18ba764715a25cbdba218754d9 : SUCCESS in 1h 54m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fdd0f0a172e74b7c8a0611e936c54aab : SUCCESS in 1h 22m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/338cd64e4a814e54bf02eff4726b36d1 : SUCCESS in 42m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f434e0654c114a528bd6da6b95063ff1 : SUCCESS in 1h 49m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a4154f5dde5f426a8fc270e829b78f34 : SUCCESS in 23m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cebe6db31e7e11706d0bcad559511b0ece943392","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 06:15:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ec8f72716b1f859fb242bf4161036759495a7d39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 08:20:22.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d67c287c8384ce69f977382c39346fe\n\n- grenade https://zuul.opendev.org/t/openstack/build/d66dc9209fa6423aba2514309d109302 : SUCCESS in 59m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebaec7bab79745728edc865cccf8a8a4 : FAILURE in 1h 17m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aee87fe7a3e94fdcb564966077341b9b : SUCCESS in 1h 03m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ff1497efcb4853b2f12b69cf627f62 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/275c965f41c64585baabc055c8593fee : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c434904a28440c684e567fbcac06460 : SUCCESS in 5m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/946cc463f65342ab9095bf9982096914 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8a5bf6588441b588ccb2b1d7433dbe : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/049d03dd758443d8be388c94231ba3ae : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5cd416417bc343309386a3f6bfddca5d : SUCCESS in 8m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6b8d800bfb85444da3124bb36f8ebbd8 : SUCCESS in 1h 42m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4cbe4f211cf64050823d73f643c4c2b8 : SUCCESS in 1h 42m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/786d5420d11c492a80aaaa3dba11bfdd : SUCCESS in 1h 52m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f57041dcbc7b477cad9b5a3bc3ec3a0b : SUCCESS in 1h 02m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/378ac4d248c541fab817bbdb15ee96ea : SUCCESS in 1h 57m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d2b48f4861c4bb1bb9c30075df53bfd : SUCCESS in 37m 22s","accounts_in_message":[],"_revision_number":2},{"id":"8a9186509641ff42db0c679b740255de46963511","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-23 08:44:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3b01b21c363e8d653e064f89168a1ef5a8bdcdd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 09:00:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a8265be6be44e6a834f74b553313a16\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8a58c2ac1824fee951ffce2057e6c30 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26f70eca68224a76a523aca58e385827 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6154d9e2623848dfa838199dd4b87e64 : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79fc909085fd4eb19d24a3e394201c3e : SUCCESS in 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c6e0076aafe9ceef581a1acddd15e4ab8b69c6d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 10:23:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b72a7301a7143e2932fa645c1cbaa3e\n\n- grenade https://zuul.opendev.org/t/openstack/build/899859fe382e4b21bae648b8b32df238 : SUCCESS in 1h 03m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/baae381b64fc4f61aca159e5ab801c89 : SUCCESS in 1h 31m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e7dd52a57d24741a593879a828e6496 : SUCCESS in 1h 08m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6a518d0bebf4a298b254850efa2dd1a : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9118f83437e4e009e77d3c89e05ba71 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6acb3a438bfc423d9795d139bdba9758 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/edb4017112284f2ab437118b1d6c322a : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9dae0dccb0c34bc48b4fa6d2f4867fcc : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309f53b5cc9f4ebb82d716111fdb29a6 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0854251ef3a947ef8903e697bc045d9e : SUCCESS in 13m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e7f2b96eb4549399ea3c3218ff44d04 : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e88f5bee8de446bfaad23b18aee25fa2 : SUCCESS in 8m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/558b3954a94f4c88953f73e3a288bf98 : SUCCESS in 8m 09s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/85a2506cab69414e994b2ab1ba415597 : SUCCESS in 10m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0b9e9dbbe4204bbcbae4107628ced82e : FAILURE in 1h 34m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/39a6fe0b529941f6b733a4884c2bd8f3 : SUCCESS in 1h 31m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1d53a4f3747646f88f18d2df11213e3b : SUCCESS in 1h 25m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/81ee2fd451a244d7bf770a74421550b6 : SUCCESS in 1h 12m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69befa61f62240b9849fdbd0dcd1548a : SUCCESS in 1h 37m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fec25a02e5a544b0b289124c145bf879 : SUCCESS in 37m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2a962cffdfca4ffa97a0cbda34930fd6 : SUCCESS in 1h 34m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ec0588860af844b58ba93c0aaf7f118b : SUCCESS in 25m 31s","accounts_in_message":[],"_revision_number":2},{"id":"93ef4ead047ea5e9dac329483a18538b669d74f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 10:24:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cd369f907e2cd010ea167a640b2a2004b2d3cc69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 11:44:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/245d376e45d143a99795aedaa4a6bed7\n\n- grenade https://zuul.opendev.org/t/openstack/build/f43836237c074731a5e845cab1f6bff6 : SUCCESS in 40m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5e5c278ae3294d0691839327121dbb07 : SUCCESS in 59m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a043f44d9da04dff85d894eff68981cd : SUCCESS in 47m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05f647b75e78437d878d7dd8e3b3d1b8 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b22d507c37de4a309a7b373b8dfbae02 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cdc085eb809417caffa958d3b5bc487 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0b7d9b8bbd541ba93ec59f63fa9d684 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe975ce421d94cc5ad7d786b52712fb7 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/046287c902984d31a6e83776ab8f5aa8 : SUCCESS in 8m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/711a0f767e9f48f487eb006d974ab31c : SUCCESS in 9m 02s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/108be7cfc1f747e18dee6a3d28a12978 : SUCCESS in 1h 15m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1fd9a632fca54f4e8aaf73e8603f4e0d : SUCCESS in 1h 19m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d6cd1b1a14af4c6e8b9932f77c24dc4b : SUCCESS in 1h 13m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8be234d9190d487db4fb6b0d3c287a86 : SUCCESS in 36m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3436cd2e5aa46308b41c80ca464e5af : SUCCESS in 1h 15m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf1071c20c084e14bc26c075658d1176 : SUCCESS in 23m 59s","accounts_in_message":[],"_revision_number":2},{"id":"06a88de587d6a67d14bdf2c3dae628907dd4680b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 11:44:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0e6ace7487018fceb0b4b9b366e0641c14f236cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 11:46:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ab50f6a2b414de3a1e040f4813cddfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1849b0a32434487bd70ae6f2aae28bc : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9558093b3e56d5c743b2e4cc03483383394b85c8","revisions":{"4ead0bac2c9ad0cea8ad432c173e36ddc114d171":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:09:08.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/889068/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/889068/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/889068/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/889068/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/889068/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/889068/1"}}},"commit":{"parents":[{"commit":"071c115873a29bca7fc36d96620678b943827d57","subject":"db: Trivial parameter swap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/071c115873a29bca7fc36d96620678b943827d57"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:00:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:08:40.000000000","tz":60},"subject":"db: Use context, session args for metadef DB APIs","message":"db: Use context, session args for metadef DB APIs\n\nThis aligns the metadef DB APIs with the rest of the APIs and simplifies\nour life when we decide to move off the legacy engine facade in the\nfuture.\n\nChange-Id: Ifc1605e8e89c48b071ea505e40a9488df208fb37\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ead0bac2c9ad0cea8ad432c173e36ddc114d171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ead0bac2c9ad0cea8ad432c173e36ddc114d171"}]},"branch":"refs/heads/master"},"9558093b3e56d5c743b2e4cc03483383394b85c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-25 09:10:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/68/889068/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/889068/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/889068/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/889068/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/889068/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/889068/2"}}},"commit":{"parents":[{"commit":"62893e9373a9659957aa5b0263e7a19d57f549d2","subject":"db: Trivial parameter swap","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62893e9373a9659957aa5b0263e7a19d57f549d2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:00:17.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 09:10:25.000000000","tz":60},"subject":"db: Use context, session args for metadef DB APIs","message":"db: Use context, session args for metadef DB APIs\n\nThis aligns the metadef DB APIs with the rest of the APIs and simplifies\nour life when we decide to move off the legacy engine facade in the\nfuture.\n\nChange-Id: Ifc1605e8e89c48b071ea505e40a9488df208fb37\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9558093b3e56d5c743b2e4cc03483383394b85c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9558093b3e56d5c743b2e4cc03483383394b85c8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
