)]}'
{"id":"openstack%2Fglance~889069","triplet_id":"openstack%2Fglance~master~Iccfdb7fd0bd65318997c598db7e6cf59d4ac0df0","project":"openstack/glance","branch":"master","topic":"trivial","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-25 15:16:21.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Iccfdb7fd0bd65318997c598db7e6cf59d4ac0df0","subject":"Remove \u0027egg_info\u0027 configuration","status":"MERGED","created":"2023-07-20 10:10:44.000000000","updated":"2023-09-25 18:56:45.000000000","submitted":"2023-09-25 18:55:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889069-trivial","meta_rev_id":"7d4d867268a9f0292eda7c46b4064921e84e4951","_number":889069,"virtual_id_number":889069,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-25 18:55:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-09-05 06:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-08-15 03:39:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2023-09-05 06:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-20 11:37:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-20 14:31:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 06:23:03.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-08-15 03:39:51.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-09-05 06:52:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"17645877dca503368af22f1baf6daf35a1702efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-20 10:10:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1037e98a513d941a234fa7719ff762db4c3dded","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 11:37:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e141593830a455e9e56a1280d78c19b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa94fb21c6b44e499365a546df2a9585 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4509ffaa994a49a49a22b7b6caa0b272 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/444144ffe5624ae987bb117807996652 : SUCCESS in 10m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"232ddcd84976d7066c9dcf72732a4f820eda0c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 14:31:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/51268748054b4f3fa66583e8a8fb6eb6\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fdf91340b1534c2fafcb063e80df7843 : SUCCESS in 3m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/42c734d2d20d4ee490a9277af65c0e37 : SUCCESS in 1h 06m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/980f3e5162a2448e82addac1b8c26319 : SUCCESS in 1h 33m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8537586c8984e048d7121793753a1e6 : SUCCESS in 1h 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fd2a127698848bd881ddb7717e2585d : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45f5ccf39b2143c68318b523fe0b806e : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b05ae8191e75461da9d62f44aa0d4f7a : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d51f77804c934754a732906b8eda5d32 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be625bbb06394e048e2cd0cea7f6b36b : SUCCESS in 13m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3c7abca036cc48f0a47a3fe85527e912 : SUCCESS in 14m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9bb7acb6db604f09ac80546e9df0370d : SUCCESS in 9m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e0c26cd664b465496d1792e887b1b64 : SUCCESS in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/71d31cf5f03944a2a7403d4f1105579b : FAILURE in 9m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1b7006f506ac4a8fb2fdb2031bce554d : TIMED_OUT in 2h 04m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/baac116b82514bbc8ee21bc1485fe17c : SUCCESS in 2h 00m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2f3f5243f30f4555a1d19cab8ccacba7 : TIMED_OUT in 2h 05m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b56a83830f4c477f978c9300d73fbeeb : SUCCESS in 1h 49m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/96e6f4873211411ebc32bda42258e766 : TIMED_OUT in 2h 03m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b154e5e5af04dc7bea4d14761f40eac : SUCCESS in 1h 27m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/87faa4dd27a648d9b5a5bb9c110e30bf : FAILURE in 18m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/85739a1e5f7142eda8ed53da43d39394 : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":1},{"id":"49552d657f21b7926b7bc5406ff774450033cb75","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-02 06:23:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec91aa38bc48a2a2bede09cc6a7404fa69586204","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 06:38:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f44a1666245d4ebb879ec637039f5c90\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d2e143de8e44750b46fea247f2ba197 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82c01d8a733442989bf082eeadeff596 : SUCCESS in 13m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62880b763f734ec09cb2e0fa9dd68687 : SUCCESS in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23899b761e7e73cd850c6e71bd7de1cffcdc44d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 08:34:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e95a22110874d86bf14b7665313dd11\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/22d115899ecc4139a39350edc67c7b4b : SUCCESS in 1m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/c8fc968ed3a24c80bd94706b6cbc74da : SUCCESS in 50m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67f57882867c4b9fb23e982ec457f0c9 : SUCCESS in 1h 08m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0fcc5256b318480e8e239387130fb889 : FAILURE in 53m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec995cc5c114756a881097e95d6dce0 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6cbe7eef5564f4fad42e8b85a60feda : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cf0c2bc1ebc49c4837be52025aaedea : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b92dc1e7ec54673a9b70e6aef1ceb72 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/554be849587443768bf71de7f7e4aa8a : SUCCESS in 7m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/10c92f74f72d4afa8e26aff3bfb5fca0 : SUCCESS in 11m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/460b5ca3d78e40dc844adfa7852fc0ee : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/601a8100b7af4ff8ba624d53e4e8907f : SUCCESS in 6m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dfcb98ab8c674cee804529e09380bdb3 : SUCCESS in 8m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b18ad9efbdd44928fed9662515302b4 : TIMED_OUT in 2h 03m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6c130141e9344595876997a985ad5714 : FAILURE in 1h 18m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/390bd48e9d5b4735ac53d11973ae27f3 : SUCCESS in 1h 53m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7a4fdf16cdc64c65a7fda3123368d1ca : SUCCESS in 1h 36m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/46bec6e3a7ad4a2f8ab7041a9b2198c4 : SUCCESS in 1h 26m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbec8712cdaf495cb88d50778c3d50c5 : SUCCESS in 43m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0be12e8b05bc4af5ae9930d565426736 : SUCCESS in 1h 34m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f260bce4de9f4dafadf27d7778dc209b : SUCCESS in 24m 40s","accounts_in_message":[],"_revision_number":1},{"id":"4c63b406457117733097ab34da3411971d3680a8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-08-15 03:39:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"39cf478ab34874b36d7da53f017b557fdb5249cd","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-08-17 10:44:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6dec8fc66d1b5119edeaea777e7bedb68962e52c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 10:55:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53dcc1d6f277415dad205870de26d507\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c298b8e1ebf1495cbea03a73c77b2143 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b74fd31f19c0499c97532d0b15fe9fd4 : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f661bfa5d03449d92fec0a238d236c2 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"528a035022c9280d7f0eaca09ed6612a977924d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 12:57:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b48f507477c9435092e27f5a5d97d6b4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b42fb8bff2f747b5bd27991796e39544 : SUCCESS in 2m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f255c9d2e3e6450c83c6fff5a09bbeef : SUCCESS in 50m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/16e2cb17e98b4ec2938c681b0a57e473 : SUCCESS in 1h 16m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a90443473f9745de81bc874d56d5b5c1 : SUCCESS in 54m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8dde4ec449a420193a36e276694db4c : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cca521a8558a476a8554122a0d3aca3e : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5cf322407534f6e97fccb4ae1dde9c7 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1698ba1b1d44457da221e95932fa9c1a : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e6f5e439eb4683b4d8103a4f8d544e : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/06f390e3f83c4150971a07d0e5e3fd6d : SUCCESS in 13m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef5e894cee7c43af9c05b1e3567db4eb : SUCCESS in 10m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3ab6a33b93814f8fa7ac0ffab8817432 : SUCCESS in 9m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/76616bfdd0ca40c49afba19c733fedcc : SUCCESS in 9m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/560c390834d24cd997afce2b572970d7 : TIMED_OUT in 2h 04m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7c653d829aff463e97f02b6b2dec660a : SUCCESS in 1h 16m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b9f23189ae64bb7900206fafdecacca : POST_FAILURE in 21m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fc55020ca2df4c4eab7c36ae707dd3f3 : POST_FAILURE in 19m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7385516aac6043b1a9bafa68410330c5 : POST_FAILURE in 20m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26092446735f44d38936fc150b9177a4 : SUCCESS in 52m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a552a48eeb6c4816ae24d2daf1f4956a : FAILURE in 21m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9b91b6fba1464caf8908bf93330388fc : SUCCESS in 46m 05s","accounts_in_message":[],"_revision_number":1},{"id":"c6c3e938ac595954d6bb0cba740c8af7222e560d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-09-05 06:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44d1bb3aa553d3df684248fc88a395db7b8352b7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 07:01:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c4f37f5ca147abb7987011f2cc4157\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/471b5c04362142cd809c3b67a9c6ec3c : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7bbcdd0fc8bb4f39b43ae2b557ff1084 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5e1a9850cd1b46ff8a758bac08f212af : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3be6a84788564f168e796f27e1223a27 : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23c746134b1f9178cb820e9690ba48b4da24093f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 09:02:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94b52197fecb438fb70abdb34699bda7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cb336efdd6aa49e390975ccc77cef318 : SUCCESS in 2m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/9ee4996d0ff74663902a8d205065dd29 : SUCCESS in 44m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6899acb6c5a340a0ad4f62fa3e4786de : SUCCESS in 1h 20m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d6527678fcf4646b87716d6894c3581 : SUCCESS in 46m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f048c3a0f284f61b7f6af9af65378cc : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6d8371208084322907ba4b4d3928cd5 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c85a66d122cd4c27a3664fecc2d0ffbf : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22cd51e084294b3fa1f6fea7f2471fa5 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83c14e932a7d4a879a5faedc1cca78a8 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0008e6149d8348d683bf7206766f7415 : SUCCESS in 6m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/921525134a0a4cada02de56a4b13356e : SUCCESS in 14m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f19ceaeeb69c4e64a1122b9637a0c5c1 : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a4a21b6ddbde4497b5eb19e981460bd0 : SUCCESS in 7m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/47a90ace4ef842fa87bf0c6ea8d8b3db : SUCCESS in 8m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7c95c312b78d4256ae05a99723987781 : SUCCESS in 2h 01m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b7590810d424491291a587bb552bbbe0 : SUCCESS in 1h 32m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b32c9a7933b4475e8e3d0cbb971efea4 : SUCCESS in 1h 29m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/94a5def6526f4f589dc4d2ae062b3309 : SUCCESS in 1h 23m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9a0abdc854a6451790a217a1a5577f84 : FAILURE in 1h 33m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/955f24f9e8584521a7588b9f7acbd9f2 : SUCCESS in 1h 01m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1f2053910974abb82ac94d470908035 : SUCCESS in 1h 46m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d0f0a545999c4fa2b1240db60caebce7 : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5aaeda5ce2a5021d3dbf873bfd509b694ee81b4c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-25 15:16:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2445fba24214942aaaf38382238e900dff596c10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 15:31:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/192a1a20b48346bfbc72132116f78cdc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cc19705f6a6476fbe6b61882cf9b090 : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c9de167455949e68224bedbb067ab9d : SUCCESS in 12m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3757f9eb7f92498791ec6ee054bb9138 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/33ace9b194b24bee8ea809185b3bf333 : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f56ded30474e1a7aad3c5bf3bf8a94c47077188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 17:07:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d5836f9f8874ca4bbaa3526ced002de\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9f39d58cb2c14d298add968ac67eab51 : SUCCESS in 1m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/8a654fa0941449cca5c3dd03d9a291af : SUCCESS in 1h 05m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fe59c7020ee5490687ea71a8852f334d : SUCCESS in 1h 37m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23d51b6cf3284fb5849872ff32310fd7 : SUCCESS in 1h 06m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/241ffd9a1d0a4a23bc51c32ec0aa4606 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9707be953494ceba023319d2dfe3188 : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19e2e22802bd4a7080096a76e99b5bed : SUCCESS in 5m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6d0bd9352d9944da829480cd65579b51 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2ceef8553f341858ba9a61be71f79df : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb74e7bcb416476e95007264caa0b524 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c392e3270d9840f19587787ab11a1ce3 : SUCCESS in 12m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f7f9c051e90944389dacd4512b30ab4a : SUCCESS in 9m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7775119047f64a5fb4770cf9197cfb20 : SUCCESS in 8m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/55e05cc9d05e4fe7b6a5173037dd2f46 : SUCCESS in 8m 12s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1da52804e8e543f2bbdbebfba3566f00 : SUCCESS in 11m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2880c8c99ebe412ab8716fea8b09c008 : SUCCESS in 1h 32m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4b620b26c31c4df9912aef2e77a9307d : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f0c9f72c79ea4ca8a7f8bf9df455ac12 : SUCCESS in 1h 26m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8f3c3f3a80e745a792e2fa673bcb3e32 : SUCCESS in 1h 23m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8843e9043f2641ec8cbfb53b4d30bd43 : SUCCESS in 1h 38m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c28eeea51da0465f8696f723e7438187 : SUCCESS in 42m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c771c9d1e2724686956e73d735fddaae : SUCCESS in 1h 46m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e9406c006fc64673915eb37ad1f33780 : SUCCESS in 32m 56s","accounts_in_message":[],"_revision_number":1},{"id":"4f3b03055145e6eb6ce34473e6e832faa9c79505","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 17:08:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"beb1a553f7bab3c9adaa2471d46c76921740711d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 18:55:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a93ab1fdb3b425c81267ef9607edde8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/08e51a0d38214b1db2c20457910f903d : SUCCESS in 2m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e06fcd62845947648a97220e1fdb5820 : SUCCESS in 52m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a420c075c904966a8b78926006da657 : SUCCESS in 1h 08m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2de5a43f9a9f4afdbf7c6d09332129da : SUCCESS in 58m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0962a11f962d436eb4e91bca379a661a : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1696f8071534f318cb573e59b16e291 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34ab0054e800423bb55f8699b8da14e6 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2d9206972bd34d7896dfee0cde5a1f13 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66917542099b45caa7f0120595c56ba1 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42aab06ad2a04d1a882a8f73fc2403ec : SUCCESS in 9m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5c8308a6a4f4ece81e1c30bee0c0f20 : SUCCESS in 8m 50s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a4c199057cfa455190e3d5dd21538bae : SUCCESS in 1h 12m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8042060f741b442aa87f8e941c11ae64 : SUCCESS in 1h 07m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/62d44a1de87a4395953dcc3c296c7820 : SUCCESS in 1h 41m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc77b125634d4060920e14815acaa343 : SUCCESS in 1h 07m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/377fb44340e247238703fd89114ed315 : SUCCESS in 1h 40m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/958accd3a5704e499c3ba64ae33f682a : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9b24d720bd491ec03493a082d6425aec9d626837","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 18:55:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d4d867268a9f0292eda7c46b4064921e84e4951","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 18:56:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca506622f0d467487ebbc769dcd0c49\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f7512349c784ebe8a6ba972e61e8b89 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f1823b5876460496821d99fb4a5709a7b8bce9a4","revisions":{"f1823b5876460496821d99fb4a5709a7b8bce9a4":{"kind":"REWORK","_number":1,"created":"2023-07-20 10:10:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/69/889069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/889069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/889069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/889069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/889069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/889069/1"}}},"commit":{"parents":[{"commit":"5f6f67b0f473e6745e80e8301ca7711818945d69","subject":"Merge \"Correct item numbering in install docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f6f67b0f473e6745e80e8301ca7711818945d69"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:09:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-20 10:09:51.000000000","tz":60},"subject":"Remove \u0027egg_info\u0027 configuration","message":"Remove \u0027egg_info\u0027 configuration\n\neggs are dead, long live wheels. We don\u0027t use this and it can be\nconfusing if one builds eggs since the version isn\u0027t taken from git\n(using pbr) but rather from setuptools/distutils.\n\nChange-Id: Iccfdb7fd0bd65318997c598db7e6cf59d4ac0df0\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1823b5876460496821d99fb4a5709a7b8bce9a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1823b5876460496821d99fb4a5709a7b8bce9a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
