)]}'
{"id":"openstack%2Fglance~889330","triplet_id":"openstack%2Fglance~master~I6e76dcacc32f05a70757df28ab2f119b25da440b","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2023-08-29 15:20:18.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-08-29 15:20:18.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2023-08-29 15:20:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6e76dcacc32f05a70757df28ab2f119b25da440b","subject":"tests: Add tests for loading, unloading metadefs","status":"MERGED","created":"2023-07-24 16:55:40.000000000","updated":"2023-08-29 15:21:37.000000000","submitted":"2023-08-29 15:20:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889330-sqlalchemy-20","meta_rev_id":"9805a79331885598c06a7ff1a02466683d2dec1b","_number":889330,"virtual_id_number":889330,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-29 15:20:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-08-29 12:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2023-08-09 05:58:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-08-29 12:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-24 17:23:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-24 19:11:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-09 05:58:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-08-09 05:58:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2023-08-09 05:58:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-08-29 12:31:18.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"08761e7800132d72147683a54158fd516fc1cfee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 16:55:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c5aba4c5857389bedfb545cebcf1c2ce9cbe7e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 17:23:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3cb6c4754d14c758bc79540b57e11f5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72f978db569048a0a58f590195ba8731 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b3ef4f7f612f4512b3355f21e209e58d : SUCCESS in 11m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b4e204915da645968312f3274550d267 : SUCCESS in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"679a95af706b06d71fde57e78d5579504250370b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 19:11:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/573643792d5d496ea0f74e5f6f674833\n\n- grenade https://zuul.opendev.org/t/openstack/build/c14e95e712c24a10b90480ce0c4eccc4 : SUCCESS in 1h 26m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af57024fce134cf8b5a0e8cd1f4b075e : SUCCESS in 1h 38m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0828ed1574c44303b5de3af4622f4aaa : SUCCESS in 1h 06m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdf5b00cd6434119ad7fc90e057b9253 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7c77b70bfb24bfd9f10dcbac62a2a9d : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58efec4cf51b4c2eae14c2b0bb3dc645 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15808fd0d8a94d60b3ab4e29b194a8ec : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c6133baaaf54090b729e72d46fcd845 : SUCCESS in 6m 59s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/82a92c8ef75a40fb86e2e04e5f0fc5f8 : SUCCESS in 14m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f31c78479fd64fa4aa7b92959498fd65 : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/146d2a37ae984d118fb68465c0a950b3 : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/374773b1a253422db804f4c00a68727b : SUCCESS in 9m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6faf8f86f2ce4c1ba71f044d372edaaf : TIMED_OUT in 2h 03m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0b8a2cb7271240cab21d8baaf8db738c : SUCCESS in 1h 24m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/046bf3ce079f45349693962b70af571d : SUCCESS in 1h 49m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/275c9ccd416d42b8b3d6f6391ed09c9d : SUCCESS in 1h 46m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4106004d360d4091a5b853331ba7eb82 : TIMED_OUT in 2h 04m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a753ff9200c417d96e00ec4f3de1b55 : SUCCESS in 1h 01m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a9820fca9a62448e8ea9a668aefbc7f2 : SUCCESS in 1h 47m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be12a30ed0624c04b5fc583e7638da37 : SUCCESS in 38m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d7832e5bfd5d95b136545bf24ffbe4c9f2cb6875","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-24 21:31:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"901425b64f0d4696a67e8e8f9041d2c8cd369839","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 21:48:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2732207b25e34c16b97fc263d98bcef9\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a6963ea057744ffb9136c1c8efaf617 : SUCCESS in 13m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ecea734dd0747a7bf687e1cbb348752 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0de011585d1c4d97a244e424ce17e642 : SUCCESS in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83e92beaf943938f04e5833931fc2976d36c92df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-24 23:46:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/229e6122d92f44e7bc32d09980b99861\n\n- grenade https://zuul.opendev.org/t/openstack/build/192f4fdc83ea43d8a6aea67d5aca9a78 : SUCCESS in 56m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d4114634fbf420da5ca82a050e5bbf8 : TIMED_OUT in 2h 04m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ebcf774bc89459c9645fffa352ff441 : SUCCESS in 1h 00m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8029b80b14974a8ba1a4d1744b73671a : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03609883d1444582a0fa1f030499bc4c : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf25e04e1f094ea1a4e8b54a6fd8d310 : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1be073878b6e4a94a05a6da2a1ce0686 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5213672b5a4c4246baa71051761d5009 : SUCCESS in 6m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fbd99286bf78480dacb1aadce9a42397 : SUCCESS in 13m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7a7feaae0e9b4ae0a2a0462f246c2571 : SUCCESS in 9m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b2b8518a49594b999f6ee1cccb07cdb6 : SUCCESS in 8m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ea45b611ed9541618fcabe02aabcb9b4 : SUCCESS in 8m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f256d89a5e73495190cd69d1aeb5658a : TIMED_OUT in 2h 03m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9079c43be2624fd59aa427b55408d485 : SUCCESS in 1h 22m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/60d365e4a74c4f44a832c92bb9011180 : SUCCESS in 1h 19m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/24d390503c4246afaa1e07a24ec44e01 : SUCCESS in 1h 24m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/241213ad512c4d1abcca5c09dc60a5d6 : SUCCESS in 1h 36m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e463b8c0d4d406bb5ad4297dd88e834 : SUCCESS in 50m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78554bf20ad3436ab4a99f6cad09d3cd : SUCCESS in 1h 37m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fae14ce61fee47d09151aa63cea8b653 : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":1},{"id":"c5299d61da187f55ce5f10f2c86547ada4c00024","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:14:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87eaa28063ea1daa40b9250e7123098a74d3193e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:35:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5a5a484306c4448b06696d3961fb677\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f330a1663230447a8d6319bbcf9ac93e : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdea23cd88a3446eb2b9e1509855f555 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a614efaa5df45989c736bee496d4d34 : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4c95cebcaa592dab5c941feda7187f6abed447f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:47:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd841b2421cd420181418d62d13428fc\n\n- grenade https://zuul.opendev.org/t/openstack/build/90264d1a7288477f993114290a1ab20b : SUCCESS in 1h 05m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5658c3aed0c4471da8325fe62bbb39dc : SUCCESS in 1h 53m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9da1a0d7552d4279858d3957fc9e62de : SUCCESS in 1h 04m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a1d2f20dd848fe9ea282d911b5b9fc : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a462d6807474da5b09c433f4cd4a768 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acd17930c926483e86d1fe33b2a94694 : SUCCESS in 6m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/daf8f2a4837b4fd79cec8ef8cd789d68 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483f9bca1d554303b41ae870fb7d84b5 : SUCCESS in 7m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/00d8896c3c5f4b1fb32466852a5770c4 : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/84de6b9b357c404fafd1eaa6b510d620 : SUCCESS in 10m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dbdd196b8e5d4af79b3a25b830c013c1 : SUCCESS in 11m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/74c2b0d9eeea4a0fb22fd3f4b87b02db : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3418344c92aa4dfb9d7a9274d1af4e4d : TIMED_OUT in 2h 04m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/db68540abe464d8ba764930ff24189fa : SUCCESS in 1h 38m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3799816cd9c84f2f8bc922088ad45da8 : SUCCESS in 2h 03m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2f0c3b3af5064199b196b5f782ef6c41 : SUCCESS in 1h 50m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d70a9e3ab8084044a020567ac2b6f243 : SUCCESS in 2h 01m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93b2d9a4164b4351853c02c5ef571219 : SUCCESS in 1h 05m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/53dda1ca82f14633b80309337f65840d : SUCCESS in 1h 43m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/15370c0f03b4441b85f8ce64c1e828a2 : SUCCESS in 37m 43s","accounts_in_message":[],"_revision_number":1},{"id":"1881a945377e8e3b22154c48a987c2b3fa0fb704","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-09 05:58:21.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c8d5228d16f990296b1e6213af6507e17a2342d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-08-29 12:31:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1e3e48c656c88e0a80e1da5f117997da4f7160d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 12:32:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0bec915d5e708eacf019a45452af3cd08b5e49c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 15:20:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60f7b1e940fd4abaa6fdfdda3adc269e\n\n- grenade https://zuul.opendev.org/t/openstack/build/1b894a123f554e8fabedbeee98c91bc4 : SUCCESS in 47m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d95c34c1b2584e4c82d78b65df1349a2 : SUCCESS in 1h 18m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82988bf82c3d498da5ca346d6f835144 : SUCCESS in 1h 13m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51099c46f57742a891355ff6afd61b2d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2b48a72bf944973bbb64b9df2b025cd : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ba7c07a37d34b0ab0124ff643d5ac16 : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bcba95b4adf64c60a746f0bcb65f03dc : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe94524d8754d11bd2e45b21ec2b93c : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/87a628b082514444b164f5a8f08e5683 : SUCCESS in 12m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3cab35a49b664a73a2c8199193133975 : SUCCESS in 8m 17s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/745a92cf3bb148209efb56a5e2d052ac : SUCCESS in 1h 12m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/83f878e72e1e4ff0bba5a14e781db704 : SUCCESS in 1h 21m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1a799b861f694a9ab71d8b474c9e09af : SUCCESS in 1h 30m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66ed35ccac4d436ab30469a71056d87f : SUCCESS in 57m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e09fb03d005749d782217ea77f66cc6d : SUCCESS in 2h 06m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1bc73a2fc82b4aea9b0dd88c904abf9a : SUCCESS in 42m 14s","accounts_in_message":[],"_revision_number":1},{"id":"abcd4c046b766ec453063a2306a38f3de2efe5b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 15:20:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9805a79331885598c06a7ff1a02466683d2dec1b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 15:21:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be3d24f2590d4509a6263a0fe323877b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b8bea433c1490a8ae63499af066b53 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"604ffd390ca9750db3d015bbf72dae4c3c46fdd6","revisions":{"604ffd390ca9750db3d015bbf72dae4c3c46fdd6":{"kind":"REWORK","_number":1,"created":"2023-07-24 16:55:40.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/30/889330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/889330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/889330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/889330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/889330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/889330/1"}}},"commit":{"parents":[{"commit":"d409b27a793210e975b07243ac6795fc939e7c81","subject":"tests: Enable SQLAlchemy 2.0 deprecation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d409b27a793210e975b07243ac6795fc939e7c81"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 15:30:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-24 16:33:43.000000000","tz":60},"subject":"tests: Add tests for loading, unloading metadefs","message":"tests: Add tests for loading, unloading metadefs\n\nThese methods were not being tested. Add some basic tests to validate\nloading and unloading of our stock definitions, preventing regressions\nduring e.g. the SQLAlchemy 2.0 preparation series.\n\nImplementation of this test highlights an minor issue in the DB API,\nwhereby the default values for some methods with \u0027sort_key\u0027 and\n\u0027sort_dir\u0027 parameters are not valid. This means you can\u0027t call these\nwithout overriding the parameters. This is corrected.\n\nChange-Id: I6e76dcacc32f05a70757df28ab2f119b25da440b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/604ffd390ca9750db3d015bbf72dae4c3c46fdd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/604ffd390ca9750db3d015bbf72dae4c3c46fdd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
