)]}'
{"id":"openstack%2Fglance~889380","triplet_id":"openstack%2Fglance~master~I02dc85729c3fbb6d775a32e3c2924e79abe3761c","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-25 10:12:53.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I02dc85729c3fbb6d775a32e3c2924e79abe3761c","subject":"db: Re-use a single connection","status":"MERGED","created":"2023-07-25 09:04:36.000000000","updated":"2023-09-25 14:22:31.000000000","submitted":"2023-09-25 14:21:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"889380-sqlalchemy-20","meta_rev_id":"97cb88973d079d9159e5d95e02bd7266eb137cc0","_number":889380,"virtual_id_number":889380,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-25 14:21:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-08-08 08:46:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2023-09-05 14:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2023-09-05 14:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-25 09:27:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-07-25 11:44:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-08 08:46:16.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-09-05 14:55:12.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"a417e9f398a634cf9a8f4845b0b3e238eecdbe64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 09:04:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ff32f9154027d812e40f72e79173b0cb852cda2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 09:27:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aa36ec220b74c2aa5d8b118271f8611\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/73300de65ce14e9db78e58299661f4c6 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad5cbdfaead74687912cbcfd7fedf242 : SUCCESS in 19m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3bb16ba65eb54df9bed32a82de699c59 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2f48b414a73c46dc699ed61c10da7518e6f65cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:44:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7597ddadedf54654b255b78cf18daede\n\n- grenade https://zuul.opendev.org/t/openstack/build/ef1bcfd2cc424abbbc82fa348cf68115 : SUCCESS in 57m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bc883c14359b4acd9dfdad842001b771 : SUCCESS in 1h 42m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f538897c5a47431d9599d9853a6e47e4 : SUCCESS in 1h 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2758431a30b647fd8ebfb428a89c1728 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f033cc140ca44ef093f2ae40f9814ec5 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c98e687ddf6c422689a0d70660b7c0b8 : SUCCESS in 6m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4498fe68ed004f86bc333fa6f98d1940 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e54ba5c547d41bcb870b5d9e37ab595 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/527c37c2399545aea905f91008a20d12 : SUCCESS in 12m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a76effd698424c6ead06d20eb4d4a1b2 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f8ff8a9b48c4c698cfb51e0d3bc57ef : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ad733063f35a4d3a97a332a0cabcc390 : SUCCESS in 9m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/57d12d7bb5c440c190e9fdb538baf7f5 : TIMED_OUT in 2h 04m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9052f1fa59f942778aae956a67654fd5 : SUCCESS in 1h 31m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ebcc8fd36d7b4251bab2b15483f695b0 : SUCCESS in 1h 46m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b645567803d347ff9949ed4462c34c99 : SUCCESS in 1h 56m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0dcf36cd60bc4a9eafc19ce402a58a6e : FAILURE in 1h 49m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/729cab492bc84ef288e38b53d8b7740b : SUCCESS in 1h 17m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d49e67074fa04b36b1f311ccb921ee64 : SUCCESS in 2h 24m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/13b9cb24f990486c94742c821b692a04 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":1},{"id":"864cab4c5f87d1cba74b980a1da0d2709009a5e5","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-25 14:30:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b386fe91afc03c8fa7f276043e6e8eb1e965274","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 14:46:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6131cf6d60a45168af732c7ae35f2d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aed956258e9c470da305154d48b82bb3 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f8967e51e5b436cbdbf83ff533b2b0a : SUCCESS in 12m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7287f413028a4cc183f6c284c8fc79ea : SUCCESS in 7m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"160c7869dd9acd68cb61ad601a3f2cf17e84d61c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 17:08:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/05873dcdb167498cb0aadc5cf470c132\n\n- grenade https://zuul.opendev.org/t/openstack/build/9125ad95d2e14c5c936a77935c8fd681 : SUCCESS in 53m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5b57faa1648f43cbbf2207a49ed13b5c : FAILURE in 1h 45m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8a4152ebd82424390967b3ddbe33777 : SUCCESS in 50m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19641d5b47d444baa8b845cca6826903 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6a59b6006414d0e85bcc93c3c742fcc : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9982efa006b490d987e8cf3ed27a09c : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e119ccce82d488dbaf2d566b856d658 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d2539859e074f8db56c05c2f906f2ad : SUCCESS in 8m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7f8cc244d0bf4754abe15a26500c5d46 : SUCCESS in 15m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d7dbc317ad43433487b5b9428959816a : SUCCESS in 9m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc3c6084ea114a2eaa4884700a9e88ad : SUCCESS in 13m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7dc1d7ee9bc242daa668f7b790cf68be : FAILURE in 10m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10eff012104642e18b4f4527367adff8 : SUCCESS in 1h 46m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ebf96e2567b14da0b37f0e2d87e13b4c : TIMED_OUT in 2h 03m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b5882b8c209e475b8f0e811808742c2d : SUCCESS in 1h 34m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/35f3349ca8c74c0eaed576f3f3c97bfd : TIMED_OUT in 2h 05m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b2b004d792174087a105407a4784c953 : TIMED_OUT in 2h 05m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ada32d7518bc43f2a775cea036e9e5f2 : SUCCESS in 1h 08m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c856ec8c91a342f998f076c74e93c9f5 : SUCCESS in 1h 51m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0323f15d9aaa4ccd8e04e6e60f210c74 : SUCCESS in 41m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e71e125f5f1fbf30f0c1ced573339e7435191fd7","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-07-26 08:58:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ece99f360740f92fe849256d89400c7330a21f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 09:15:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc6a7554a140430ab5a54433fb85e7f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ef030a9a0b148399399c93434831680 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53cc9ac780604f6eb39990edc23cd767 : SUCCESS in 14m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0750cdb190b54bd38292b0fee9d05e3e : SUCCESS in 7m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"827d83f88d9ac0c7eba0fb9a1fee2de116708e42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 10:58:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab9a714a5a604817afbf8eb5919e50d8\n\n- grenade https://zuul.opendev.org/t/openstack/build/6388d9ebfb924b0f9ea6dddeb89b077c : SUCCESS in 1h 11m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c1ce63bb91c4c8681cbd2619c0dc0ab : SUCCESS in 1h 36m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c79856ea32ca460fb5678c99fe31600e : SUCCESS in 1h 07m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b32e3ad04c54c88886c00d452acdbe3 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fda8fed7ebb445838132ef03fb6deb61 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0866ce47ce4ab49f892c99f602f1be : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/540d4adfafe5421ca68ff198c01a1e75 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e7ae9a438340f0a0f6eb0e3c00abd6 : SUCCESS in 6m 32s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9d6649b35d9641b789e42c41175f7723 : SUCCESS in 12m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1a2bf33184944cccaf3be094a11c04cb : SUCCESS in 9m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df56af4c81f8467eb5c823b3dfb9b100 : SUCCESS in 8m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bcc39b0842884ab9a762d92826826e70 : SUCCESS in 10m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6023a32bf6854c14a0822e543ee32630 : SUCCESS in 1h 32m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4824a2f0266a44b68b303a76b8a0682f : SUCCESS in 1h 25m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/01ca079d8a95491891172a71a75e94ce : SUCCESS in 1h 40m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d58e52fa3334a8d92f0683fb048396e : SUCCESS in 1h 32m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/88011ec74707448dacc31267d11a7020 : SUCCESS in 1h 48m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3d919dae2c64300afadb6ca04462734 : SUCCESS in 1h 00m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/46d738b726424040abc83b23a3999de0 : SUCCESS in 1h 57m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d06d7a9bfbb04d2fb0d88dbd03c0f9dc : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b43a1e50f288b5775dcbeeda7017685a850ba55e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-08 08:46:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06031ad576237c31a92d8e676b6d26f700ab2a53","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-09-05 14:55:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6331e686d237aa6389620da6df5d1346bfe0d4c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 18:58:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28a93cf2a811a4c1a6c55533a146794480612082","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 21:24:25.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/889059, https://review.opendev.org/889064, https://review.opendev.org/889058, https://review.opendev.org/889057, https://review.opendev.org/889056, https://review.opendev.org/889055, https://review.opendev.org/889054, https://review.opendev.org/889053, https://review.opendev.org/889052, https://review.opendev.org/889051 are needed.","accounts_in_message":[],"_revision_number":1},{"id":"b0f639885130b2260106b4ea15c29f3723d27e93","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-13 14:25:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ff8b0c3c5a839ca68379c52ee34035b782efae3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 15:17:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/435b3a88d4ee4bf39afb35255f9b7036\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dfe1652abb494321aabb98e132c6e773 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2780e7c493cf4517843ab9ddf91f9b22 : SUCCESS in 13m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6fb5659034704db3805026465660aa33 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f567447ed3334ad2840a1b15e5a29968 : SUCCESS in 11m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84c25e03085f26ad9c6a9295d26ea454f340e418","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-13 16:39:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/957d56d0552c41d49b8aa0908d0456a7\n\n- grenade https://zuul.opendev.org/t/openstack/build/b9f296298ff8443196574f185f360e3e : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69f09fe474f44a409e1ea6430dd5f283 : SUCCESS in 1h 40m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c208fdc7e6e48c9846b3256a9b7895a : SUCCESS in 1h 08m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06d2390cc397474a993aae987fd5efba : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baaf701835e4473a84ff9ca5a7aeee00 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3b5d57798654e8cb7bab8a0e3fb60fc : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d441e0b4781345a8bdbd0898ec87f81e : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e49d092d27b742f68486ff5a1ae3c4a3 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f773ffd4f1464e4d8ddd46a3c5e33496 : SUCCESS in 6m 19s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4ec5f1e772044a8fb99394b50970d319 : SUCCESS in 12m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a378cd02b904f85bffac5a69aac4a3c : SUCCESS in 10m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ad41809b5582477fade21c3c2dc3b82d : SUCCESS in 9m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e76f042bdc744dde931e68477a67ebd1 : SUCCESS in 8m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a8572400df9343a39d11c9e303d9c8f1 : SUCCESS in 1h 42m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ce18b9bc123543b58432e5fef8eb648d : SUCCESS in 1h 49m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6fdad1274385498bb39b2fa73d84d371 : TIMED_OUT in 2h 05m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aa8cebf45a9b4698a125f1994338cbe0 : SUCCESS in 2h 00m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dbadfd70d646485f8a360d60c79a146f : SUCCESS in 1h 48m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efb86c0fba9c4e6e8f4e62f60979275d : SUCCESS in 1h 02m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1d67dd9f19a840f2acde945c0af9c288 : SUCCESS in 1h 32m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a76099b5ad1d44a39a7566fdb068a57d : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4a21f1531dc3786b7b38e79729773bf1ba7f69e6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 11:29:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46330b3b0ea2acc2e7a0255eb8c583c5c7140748","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 11:45:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e7d0e9b14bd4746815edb18e8e3b5f7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2d7508b8ed72436fac7f1add5db162f9 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f01dc1208adf49ed8ce3f444abb635d2 : SUCCESS in 14m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ae16e7e94ec4385a012f4a4df27b0b2 : SUCCESS in 10m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6fb21060b20549dabe22c0b17a4525ed : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4eb1ef6e444a6dcad6aef94bcc4acae61f4e845","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 13:25:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5917fb601b130b75b75f6d4f689ecf8fddbbf4f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 13:42:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9386a54acda54245af449471fdf6106b\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9f133925f3d456da2ad7fc17f821840 : FAILURE in 25m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9b29c222199545b39b1662e44db6e445 : SUCCESS in 1h 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1394e5edc2a84cf999efbb59a94aa42e : SUCCESS in 1h 07m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c31784382e04ec680741ef31d43dd29 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2736707ba75d4a088e027252608a1336 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bab3693bea104b1096edd540e4b0b0c4 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42c9e84a522149efb7af77603455770f : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bdad941121df46beba7aa239770020e1 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2e8e640a5a4302ad2b04c10586a72c : SUCCESS in 7m 01s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a4ef746473184272b2d60d39f12b1d07 : SUCCESS in 10m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/470bac9140834b0c8e1dac0e693fd9e7 : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c0e96f7b1bfc4e269555dbf5ead2e53e : SUCCESS in 7m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e98e7fc171104f2c9b649a7ece8a33d1 : SUCCESS in 7m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5175152829d24d2ebe015f4106233b8f : SUCCESS in 10m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bb4daddf47b94de7b8a0ad5e54d65169 : FAILURE in 1h 30m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ccc0161f68e44afbb7febe8aef64b616 : SUCCESS in 1h 43m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/16f4acce307b4dbfaccaed6a7bcc45d8 : SUCCESS in 1h 36m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a5c3d63c04d841ca849257361cae3b92 : TIMED_OUT in 2h 05m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bf9a19ef4bfc434b82fc194f392ca8d1 : TIMED_OUT in 2h 05m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4aad992819246a9b5d32da124e12e94 : SUCCESS in 48m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/06ef8e1754474f2b86c7e816b4cdfa0b : SUCCESS in 1h 45m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8a250e9115c8460c83fa54ee6792be2f : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9e0d7364f11442796ec8766fd6d76a7b9abfe2e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 14:32:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64ea37549bb24cdf988dc10068a2cca5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d3b78db8d4442fba1998d30b29cb1f3 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/77a32db1bbee4988abcaac9c1bb59fa9 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35695bc0e0804c5db40d024a434e4228 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7649d253112d4b119945f0d5c4e1fde5 : SUCCESS in 15m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3752cb1d653d450be9bba4e14f66824f73d1e1c3","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-25 10:12:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0a6596f078302f80eb697024a8c1d66216e9250e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 11:27:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33318be488ad4c9494e44baf110bc6e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb223ed3fbc54d17ade9a07df0dde12a : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3e2cd10cb1b46b790698047bd002188 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27be9a06455146dba8e112e51031c59f : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6671db68bb814d3c94144f3930cd49b9 : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65d7af76ae35c6cf195018ca246737b7c3b67492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:19:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b3164b113b1440383d92d9ff6e4d72c\n\n- grenade https://zuul.opendev.org/t/openstack/build/7aa98d9fdaa140f3a932c8f744f36071 : SUCCESS in 1h 00m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f011f34503142ceae595bab0ec7f1aa : SUCCESS in 1h 23m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c94a0e2e418407a9b69d2c565859f05 : SUCCESS in 58m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9e620d011704e8ba648700e7287953b : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/808759afce9d4d9fa2a213a7d3805cff : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8579bb868ae54928a179869845e306ab : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2603d720876940018922a912ab0dae44 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbc8ff08f2af4c0ba203abcd92ccba22 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21cafaf8bf384784800ae10943cef104 : SUCCESS in 6m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/227cdf5d30874f06a187cc4a6ef0445c : SUCCESS in 13m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e3589dbd4c44a448f52807da95fb966 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18edac73c1824f9f9465e79cfd9ff7c1 : SUCCESS in 8m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f8814d7eeee848cea89e1d5dee72b8c2 : SUCCESS in 7m 32s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d21f7c780fad4b7ab30f4388620c9a0c : SUCCESS in 11m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f90762bc44d840eeb5bcc11a18ae82ae : SUCCESS in 1h 52m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/194b2a3b651b4ffe8fdb40953e82d2db : SUCCESS in 1h 47m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7a8f55fe45064fe8945902491b438ab7 : SUCCESS in 1h 31m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0e97ec6c8f9e4f1780de5fc42ab26971 : SUCCESS in 1h 13m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/91f80bc1c7cb4720b740834984d34ac9 : SUCCESS in 2h 04m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ca174d55544401bb0f41995ecb53a85 : SUCCESS in 50m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5711fdf6c2ef4fc58de967ceacf4fe02 : SUCCESS in 1h 46m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60e645176f35473398e8d4905c278f0d : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5a85bfd583b9bafd5689ae1644f3679e059fcd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 12:19:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a7441cc7984514cf161ff62cde4db62503e4e6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:21:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71140b0fc4c542048a08d3aae717d62a\n\n- grenade https://zuul.opendev.org/t/openstack/build/7225ca40613748e18e5445e87491c105 : SUCCESS in 41m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ddc732d112704c87bf035fb5b1377bc2 : SUCCESS in 1h 16m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b305d55622c24d21b3904a23427aa592 : SUCCESS in 53m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51d605a4c1e47f99e91d943473afd96 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3880309b58b4889af0847ff8edddbca : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/831b930102144767b696b253aa55543c : SUCCESS in 5m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bf9630173fe462684ec0387be5aeeeb : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc286656e884596abf6309c91aa64a3 : SUCCESS in 6m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/52f0bfc128ef40039259ed6ba694f545 : SUCCESS in 9m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/81c72175557044f9aa66e56650d7ff3d : SUCCESS in 9m 23s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6da13daac12944a1ab0324c18ed5518a : SUCCESS in 1h 39m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1a01f8b993e644dd9d76cc6b0f0431fb : SUCCESS in 1h 42m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9c22050550d348a3948e64726f6e7424 : SUCCESS in 1h 56m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3377331e0fa349c5ae80ee64fc401a8f : SUCCESS in 38m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/be08e3035b9f4b0b9746ccbd7ec563a6 : SUCCESS in 1h 49m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49501bee5d2249f98bb62388ff26e647 : SUCCESS in 40m 41s","accounts_in_message":[],"_revision_number":1},{"id":"5e05596aeef80a8939518aa3c6a07967095313c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:21:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"97cb88973d079d9159e5d95e02bd7266eb137cc0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:22:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77925a2d79a24900b9b4465d205077b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63c13e0fed649f58ff06dbc30e7c597 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"99e3e5e3891b0b325f2522c2f048f43a5c91d966","revisions":{"99e3e5e3891b0b325f2522c2f048f43a5c91d966":{"kind":"REWORK","_number":1,"created":"2023-07-25 09:04:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/80/889380/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/889380/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/889380/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/889380/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/889380/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/889380/1"}}},"commit":{"parents":[{"commit":"8c5091cd48d95a8059959585261dd085252bbcc0","subject":"db: Drop use of implicit auto-commit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8c5091cd48d95a8059959585261dd085252bbcc0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 08:54:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-07-25 09:03:33.000000000","tz":60},"subject":"db: Re-use a single connection","message":"db: Re-use a single connection\n\nRather than creating multiple connections, use the same connection but\nmultiple transactions. This should be slightly faster.\n\nChange-Id: I02dc85729c3fbb6d775a32e3c2924e79abe3761c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99e3e5e3891b0b325f2522c2f048f43a5c91d966"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/99e3e5e3891b0b325f2522c2f048f43a5c91d966"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
