)]}'
{"id":"openstack%2Fglance~890381","triplet_id":"openstack%2Fglance~master~Iaef25a4691406c4e0998a81a0e0f99e1b99fa96a","project":"openstack/glance","branch":"master","topic":"setup","attention_set":{},"removed_from_attention_set":{"32029":{"account":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"last_update":"2024-04-30 11:56:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaef25a4691406c4e0998a81a0e0f99e1b99fa96a","subject":"Add python3.10 support in testing runtime","status":"ABANDONED","created":"2023-08-03 05:38:58.000000000","updated":"2024-04-30 11:56:51.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0685176ff96c016f20bfc3ce9a0981a492c9e37a","_number":890381,"virtual_id_number":890381,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-08-03 07:43:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-03 05:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-03 07:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10dbe329bf1ba369bd7f8ef468a8148dbbeac19d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-03 05:38:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac14d34dd9956df7654ca967b0e3e23790f9f651","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 05:55:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96000c71b0df495b905d3df049888d03\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0bfa02aaec5548a091aedf425e9fed5a : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58334a206a664390abc07bb6d4ee09ea : SUCCESS in 13m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c0f6aab4e50425a9255afe2fc2ee228 : SUCCESS in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e78e2b11a6b0949c4c8ca35da6d0546c06110ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 07:43:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c7ceacb0db3c4d4bbdb994c76f1a8840\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2b564355e9474fc390ea00cc051a5afa : SUCCESS in 2m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/d1a74c56cdce4d76aa91c33884daab7e : SUCCESS in 41m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd326a89cd7f446c95ac230fe79e2c5d : SUCCESS in 1h 50m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5286ba5c9674e0fbb2e5670b463f59e : SUCCESS in 1h 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f896677240dc4edf85034c73a99dd10e : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ab1c136c93f40319e7786530413589e : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa95110f11124a99a5e03ef2ddd4b518 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0efc7038d6d94b2fb4dccce51969b09a : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d06f3b29f9c34659bec5784fae97eb04 : SUCCESS in 7m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/429c0451fc0c46459fc3e37f5e5bab9b : SUCCESS in 12m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/444f5aa898714578b289b5164d17b905 : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/553fe411a8394e6e8ee56bbfb2188bc8 : SUCCESS in 7m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4a19078d78eb4996ae710b29184267ce : SUCCESS in 9m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/47a896b1d4154302b42f61b43b83f7fa : TIMED_OUT in 2h 03m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bfdf35660a3e47f79f02ac55e1129cc9 : FAILURE in 1h 37m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c0eff2ece2c74791bde55a38c08d79ca : SUCCESS in 1h 40m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/64c5315d305d46d7a90e8461dcac569e : SUCCESS in 1h 27m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/41dd4b7dbc9347eb98f96cc019fd8cec : SUCCESS in 1h 42m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e11af09f221470bb8f445600c6c5117 : SUCCESS in 1h 02m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce2bddb16c0a42c299728dc3539d0899 : SUCCESS in 1h 49m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cd97f03e77224bd48c6c6088a1d992a2 : SUCCESS in 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5b42eb71b8ae4ccc09f7766abf9517b167add288","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2023-08-03 08:02:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8b9813897f07836435c21ff458c6ed8cb6995735","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 08:13:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/654ca8034e584e1681387eb75b60002f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3dc65bd7ace44a96aa2a1a4cdafecb99 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffa2216b44f24efd946fa310c95d3703 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9882a79700cf4c2ca625d5e8cf6265ed : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d81ce601bff90f76457e64b964bffd626023fd78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-03 10:19:55.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0e72d4b63684219ae6b050ee80eb20e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a14051ed2afe49f795f4b1e1bfe7ea16 : SUCCESS in 2m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/26aef814384c444591ea0b6fd226224e : SUCCESS in 1h 07m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9655224a3a98492b8f978109d3df9f4c : SUCCESS in 1h 17m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01f729af16f5455b98f74686b2fd2fd8 : FAILURE in 1h 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44e667e5270f4ed894056767c1986498 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f0d97f18ee74889979c122ebe9824d4 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f0ba6bb843f455ab379b169339b03ae : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b1e144d9ea1454a83bc9733a6288c41 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ad168b200a6498db2a9344fdc4a16ba : SUCCESS in 7m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8cc05783d68442368cdd990f69894690 : SUCCESS in 13m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/405bd097cc204150a58a3514d21064da : SUCCESS in 10m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7566d617702a4970b52114d83c666dd4 : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9cf4d4734a9247e2bb7e2c04320b0c14 : SUCCESS in 8m 09s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1173f34d2e5e4741bccada392a6a6f14 : FAILURE in 2h 00m 58s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0d5f53382f52401a808fb257e6d99392 : SUCCESS in 1h 53m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/84d6bc6d7d08478d8c7e807edc0d5d94 : TIMED_OUT in 2h 05m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8da6bb27278d454f841d8d2624579b7f : SUCCESS in 1h 40m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/859b36c2f92b4562b902d251ce74c324 : FAILURE in 1h 32m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86d4465148824ff78414b2023ee611de : SUCCESS in 1h 07m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/28c790ea01ea4e53a3cd938f4269e011 : SUCCESS in 2h 08m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8412d40231df4260a2b561b47168185b : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0685176ff96c016f20bfc3ce9a0981a492c9e37a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-04-30 11:56:51.000000000","message":"Abandoned\n\nThanks but this change has already been merged through another patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9e3970e3e4e4283ac3cba81e0da8fe74bda0666","revisions":{"a9e3970e3e4e4283ac3cba81e0da8fe74bda0666":{"kind":"REWORK","_number":1,"created":"2023-08-03 05:38:58.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/81/890381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/890381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/890381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/890381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/890381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/890381/1"}}},"commit":{"parents":[{"commit":"46c30f0b6db6ed6a86b1b84e69748025ad9050c6","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46c30f0b6db6ed6a86b1b84e69748025ad9050c6"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2023-08-03 05:38:28.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2023-08-03 05:38:28.000000000","tz":480},"subject":"Add python3.10 support in testing runtime","message":"Add python3.10 support in testing runtime\n\nIn 2023.2 cycle testing runtime, project test in py38 py39 and py10\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2023.2.html\n\nChange-Id: Iaef25a4691406c4e0998a81a0e0f99e1b99fa96a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9e3970e3e4e4283ac3cba81e0da8fe74bda0666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a9e3970e3e4e4283ac3cba81e0da8fe74bda0666"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
