)]}'
{"id":"openstack%2Fglance~890509","triplet_id":"openstack%2Fglance~master~I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6","project":"openstack/glance","branch":"master","topic":"New-Location-Apis","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-31 07:29:40.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-31 07:29:40.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-28 10:30:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-07-31 07:29:40.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-02-20 17:01:49.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6","subject":"Bump Images API version to 2.17","status":"MERGED","created":"2023-08-04 07:04:53.000000000","updated":"2024-07-31 07:30:52.000000000","submitted":"2024-07-31 07:29:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":29,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890509-New-Location-Apis","meta_rev_id":"3911d5c73b08b9a4ed8a49e78f91233abc1d1536","_number":890509,"virtual_id_number":890509,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 07:29:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-07-31 05:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-07-26 06:43:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-07-31 05:20:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-04 07:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-08-04 08:57:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-10 06:11:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2023-08-16 16:45:38.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2023-08-21 06:35:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-08-28 14:48:59.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-13 02:39:21.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-02-20 17:01:49.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"2b0b105efba98320f6cf556ad81c346691edf7be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-04 07:04:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6fc40c5eb4cdfc25538ca3bb62de4b5912ff32fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 07:16:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6c33cbb82d6476197f1ef8230259dd3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/527a2f5d966f4272b0b7aea389fc25b9 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eba2d93eba0b4b31924ed8be97792737 : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/54da5051f6a845359dacbfd1c028a7d3 : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e0f3e4cf595fa0c2d2543aa9c1fdb1cfdb367b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-04 08:57:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e848e40f17043e1a6b4177d0f30cd0d\n\n- grenade https://zuul.opendev.org/t/openstack/build/fea2037855ac4a3ea38cd91319fae5a1 : SUCCESS in 55m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/03766ac20bb7497e8d1150259c8e2359 : SUCCESS in 1h 25m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/490db6ae90804beeac6526a9c200793e : SUCCESS in 54m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d01753c028e64619b999d700c174ac47 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920f830e5d4242b19818b62ff9f42095 : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9ac0ace4d5749d095658f853db6c844 : SUCCESS in 5m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5172e4f3e03d4df7b3dbde6100bef99d : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ada7df7ddfa4b06affe418d21bcd675 : SUCCESS in 7m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f192824f2eb64c97a02a1ca3e5356374 : SUCCESS in 12m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb8333b3bc864628b450591f61e0a110 : SUCCESS in 11m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0913d1ea1f8341a6893d8a5cb5d06594 : FAILURE in 21m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d619279da3b1423aad650dc4a7ca955e : SUCCESS in 9m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b9520f4d8d2549548b4f73cde8c5a2a9 : SUCCESS in 1h 44m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a2c0466518894b9eb6bd905d2f960940 : SUCCESS in 1h 29m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef883bba22914cbe9e04f92e818aef73 : SUCCESS in 1h 07m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5aa678d31d8647fe9f1c0eda679183dd : SUCCESS in 1h 17m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ce989610e5cc4af592a6e800ac8a0228 : FAILURE in 1h 17m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fa3a9c800734a2ba056e264a259a14d : SUCCESS in 48m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1b03f0b852424dc7bf7f6865ef18795f : SUCCESS in 1h 33m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a595941fd397429ab00a58ca798f2945 : SUCCESS in 30m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c2715ac3d79b6d2bec98b7b496ec6c8b62ffee35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-09 11:13:25.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d3de39b8d9cef157893218367e36be8561ab9e9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:26:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25be5a2e255441b6995b6c011f597897\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/58db6e14b031490187ad3263bcab2e99 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/62e6fce179af4d709665c7d242590cfe : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10c3e06076aa4e2ca4e4208d7181546d : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7351531dd7feed1df084cc553a138c34cc9ae122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-09 11:32:57.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"617467fab5290418ee24118610eb14af67ab78e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 11:45:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ed87d68b846458d94d53d8886c30379\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ecf9d6b6632488daa6354f6d2e5b159 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66d2c0bfa1084d179b06e548a380c04f : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b8fbe2074c574f30a38c48a2792a5a05 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f10de4b6f8e5eecf6007d54c5745d1eee80224f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-09 13:44:36.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aeb0d548395f47a28414af66d8cc97b2\n\n- grenade https://zuul.opendev.org/t/openstack/build/babbb1aaed324a888c4b9038b7b18b7b : FAILURE in 44m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7ffc669839894ea994003d367c1808eb : FAILURE in 57m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63a788e1507144b8af36aa0aa4ef1c9d : SUCCESS in 53m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a344895b8048a68ab7ab4d576453ff : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/310384ffe42f442dad66d1b9c307b18a : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39d0fb4d649146bcadffe465a96a13ab : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63a0077e5cb5434bbe72fe763d965039 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1914b2a80fa243a78ff515221089f812 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f881a3693a6e45f89681630f36563274 : SUCCESS in 12m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/065d1a0899bb42329fb29a3a8982f5d5 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5955bc045fdb46f7ac1b077655dbc139 : SUCCESS in 9m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fda2bbcc98be47e7be49a0dbf40f245f : SUCCESS in 13m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/060d00351919473e80a13d043597d675 : SUCCESS in 2h 04m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6b5b3fa304ec49bf85fc8bc9a91b7527 : SUCCESS in 1h 29m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6031e55b4dc94f6c8340c2e062a383ac : SUCCESS in 1h 33m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b4f988b49d1d4d089ed8f58afacbc393 : SUCCESS in 1h 28m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/703cd35e80bd4b61ac9f2ad276725c3a : FAILURE in 1h 11m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcb5792a75194f51996a6bae44c01269 : SUCCESS in 41m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d2d7dd4ded7c45819b010dad7986e5c0 : SUCCESS in 1h 49m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9567c42b0eb04af28a3a19d85138f442 : SUCCESS in 27m 43s","accounts_in_message":[],"_revision_number":3},{"id":"2f141bc7c53e6eb49561dc2931edb20ded80ef59","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-10 06:11:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"984bf8f30bc94050acd990b64f001e49904f3da5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-14 11:50:37.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"4643b98c0efedbdd9e281ff9b96b613177f09429","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 12:02:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3049cfe64544905a190501c191fe9d4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/da0bcb32ff7e45eca8b9ab8e0e53cafe : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6044da15f3ac4efb9578656ac4754871 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93c00b7c6d9f40dfa8355d958daab830 : SUCCESS in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10a2bf9f406ab94ebd08e4e6c20e6ef00c5c3b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-14 14:03:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d10e48cf69a54725b4af8a504b06c553\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f9132919f3e4b1ea675b844e8f1ff2a : SUCCESS in 1h 15m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/733b79afd8c14c67b6327c0c2c580cbb : FAILURE in 1h 10m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbaffaca24f44b5e91d2cc55b6be11cb : SUCCESS in 1h 16m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d105af5d6ad481395ad798419290557 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/456d7c83a36a4bdba98eb00cf15a119b : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94b9e002b7fc47028443b007cb270f30 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94fe60f9efc144899ac408e49ec318da : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69381f60c4e4e6e92cb428ad0da5e36 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ca8cc3c8ede74e05af0eb1fbae3f95a3 : SUCCESS in 15m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9154d478cca342238921823797145e1f : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ceda6b6a7f944142a37bf218886039c1 : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f4ec2219a93b4a748e8c4550d65ea479 : SUCCESS in 9m 14s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4a906e014fa940f49190899abe0034d1 : TIMED_OUT in 2h 03m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fd2cc54c5bf6423a88d84b7d9115efda : SUCCESS in 1h 46m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/830e1e1d25f4408980dd60d8db8e71ae : POST_FAILURE in 40m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f1e94137920640849ca7fd53c7a79679 : POST_FAILURE in 18m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53122d9b90274f208e08bc55c1696d6e : POST_FAILURE in 18m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c124a6d258cb4c5ebd9aaa7d42085143 : SUCCESS in 48m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4cd80a33f6e64d21acdb1ca650e2e56c : FAILURE in 14m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/158608fb5dc8412b8176230289116431 : SUCCESS in 25m 25s","accounts_in_message":[],"_revision_number":4},{"id":"f23a8bcb7e572738b8a83f0c6d42c5f913bf3685","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-16 09:29:45.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"ce60114bf0ea01e80d2985c9ae11805b8badf80c","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-16 09:36:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3795f88ec40331004f8bd23d364982a002aef79a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 09:42:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66c3688c490439ab1894ced17553ebf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c0dcf27f2a04dd7ac01a1b4067d6b0b : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2e56509b9fe4c4fb618af0f36703d61 : SUCCESS in 10m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ed39a505bcde4cfea0773b1b7e8adae3 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fa73f71de6c90b559893ee48c511da5d34f23954","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-16 11:34:56.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6519353c06e64ed689e44f5ee4aad368\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2ac67471a5f46408558e570b28f5baf : SUCCESS in 1h 01m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/952777f7481f4029a6c492d7977c6a38 : SUCCESS in 1h 48m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2962a1f01574442281ede1704a9bc0b7 : SUCCESS in 1h 22m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41e8344e446b4c09af81affcf1bf4aa7 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09e6ba451bda44eaa7b76d7dc386f08c : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9973ba1164474f7ebd5756b8e4c6ce5f : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b0c3503e05324e33a44d0c8110f8151d : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df90871672494079ac7f0fbcbadb5257 : SUCCESS in 6m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/244e58bb45f242018ca545e5620abd29 : SUCCESS in 13m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cf895eb84b0a486a8b2391d8067f0a71 : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/836e94f0176a4fe7b3f0564bb03cb716 : SUCCESS in 9m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a664bcfc15a84e8cb82c02361c834570 : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d9583cffb57e4f12acac0683596f514a : SUCCESS in 1h 56m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0d6d6db4410b431ca655eb0f48d4e549 : SUCCESS in 1h 20m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07d59e4120094cfe8f0315120bc6b54c : POST_FAILURE in 37m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5646a6d94e47485cac5805a5ac343ea0 : POST_FAILURE in 18m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/28c3625f9bd344fbb6e44e77ecea96df : POST_FAILURE in 19m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eeb9c6cf71cc4a3ead3cb2d7a883e3b3 : SUCCESS in 1h 02m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38cdfed169c54c6ea25e0ac8d7ebfdd3 : SUCCESS in 1h 37m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a4e526e0d56c44e98dd4a0776d5a5b0a : SUCCESS in 38m 53s","accounts_in_message":[],"_revision_number":5},{"id":"4dd649e349303e37d691bf15dc55e8cb66102d2a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-16 16:45:38.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9e4ad3f4f91baa5f0aa9217d754b10c6f091a04b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-17 13:10:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4184c6b6c7ad9f25f142429472333d9687e1155a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-20 12:19:00.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4469a0ea9c0f09bc1ccbaae094e0bede5ab1e395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 12:32:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bc82c63944c4ff985972f9003ec3826\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/76c9a392dd744baa96a5e085c129be32 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2272255f5c4e4bda802f95a5565ab07f : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ab393646765544939b6482716c9ba5a2 : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c095c1b1659040feb795d4dfa0873e7e : RETRY_LIMIT in 1m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1ca3149ec7d070d893b47149315fcf1ee3c22b8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-20 14:31:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d50a2d58b991438cbad829325fda88ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f38fc17e43b4aa9b9486305c7a8fa51 : SUCCESS in 40m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6a459170f6d049ff98b67c6e2d4aaa63 : SUCCESS in 1h 09m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fdd3e3c9879c4acea23a9ce8afa6306e : SUCCESS in 45m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25b417382392418c854b5b2aa61590d3 : FAILURE in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44b5f02709204a8a9fca24209b39ff2e : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f4b3d5627830419bb15898d38c63d8ed : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/572df0a2ba2a4c2ebce6f422f90b902e : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cd4ac7febfb47769f6c7806493449ae : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ca5afc1060a4286ae06e138ddf3838b : FAILURE in 5m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/23babbb419d045c68106f8ca0b4ba908 : SUCCESS in 11m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a51a8b567d034b39971c70bb8c3c2235 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5bc49f15e23c4e96ba67ac15c08caea1 : SUCCESS in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/605d28d1d4104bf781b69a97e76c4495 : SUCCESS in 8m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f6b0ef8d077d4cc58aec003bc7b5af71 : SUCCESS in 1h 52m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0777872c85e8401d84b08ea293589c4f : SUCCESS in 1h 07m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/46ea611561ad40f4966ce58af4a19c1a : POST_FAILURE in 18m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c60c04af6d6e4c88a5714c2df2edafb5 : POST_FAILURE in 15m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d146de938cfe41c88a0cc864375e8fc7 : POST_FAILURE in 30m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5fd47695ff84c6e956c03364197cdcc : SUCCESS in 40m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b17760803c52412fb675042f3994150e : SUCCESS in 2h 04m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9e790425ecd4f39acb2b98a34b0f66a : SUCCESS in 23m 16s\n\nWarning:\n  Comments left for invalid file doc/source/admin/new-location-apis.rst","accounts_in_message":[],"_revision_number":6},{"id":"9f87d4901f297be6b0f3ff78a2b1ef539db5e74c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-08-21 06:35:33.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"318fed2c61ad19fb682f47547b71ada6ccd48304","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-21 13:50:52.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"260ab649df366cda2b18f4221b998099ed2c8e02","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-23 14:52:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9e3a5747fb5dbc9e297951c76b3a0a92c162d243","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-23 15:16:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b9c038b1d42f6b5725ad2a91eabf97e4836305d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 11:40:44.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a9dcabcd3fc19ede8b6d42e175e5f2127e7b025a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 11:44:22.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c83af9beb4f36862f336c358b9388a2a04e592ce","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 11:47:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a3c12e444325bab34524986c2cddc09c368ea19c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 12:00:55.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"554c5475e3c2667b9b6fb5b1f9776e1e62f40fdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 12:08:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3d0efbeedcf462bb5a8aa996e16ed24\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99d54c6e18194b6cb7358d1739283d95 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01558a35cf764ce0a48ed6e5809bcfff : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d7c087ccf444c198991eb7104945875 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73243824dd2147b0b75c4d92920ad4b2 : RETRY_LIMIT in 1m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"30eef5e1a369eedcec5e86431a3f899142925210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:05:06.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5837ce9ee4af4f5e819aa7dd13b48ea0\n\n- grenade https://zuul.opendev.org/t/openstack/build/0e9369dd621c45678bfd57f6c6816285 : SUCCESS in 40m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c77645a681314d039cf5e1c593844c74 : SUCCESS in 1h 23m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c5082b78f4f847c1b53bf39873719c62 : SUCCESS in 51m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/909fd311322b4584bd990f09928aec37 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f49eeedc8fe84641915ca6d28d951340 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e60c35068c2c40fbb54b313cde622a54 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13930ab4740c42dfac3f459c7e2e9cef : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b155eec57a854649af2e987f94c0676d : RETRY_LIMIT in 2m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dbc292b42474269a6e2bb1d3a7e87a4 : SUCCESS in 6m 23s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/81ac816e5b5b4a72b2400fa5552d9ab6 : SUCCESS in 13m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e506c71706f147f2880e1fb6f8fca712 : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1dfa5dd9ee764c5cbbe398e72994e048 : SUCCESS in 9m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/28f10c3aefd94d808a76982d874ed4cd : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d246dd715e79409782602e9b534fb170 : SUCCESS in 1h 58m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c98d12facd614ab5a83bae4d0020e0cc : SUCCESS in 1h 07m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/65e85e897ef34137a37040890b7d003b : POST_FAILURE in 22m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ef03bf5e6c6149fcaad9090df655395a : POST_FAILURE in 23m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc8ab430d92e4c85a1d84dd5fea2e581 : POST_FAILURE in 21m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0705770464f24dc681ce96fbc954e5ad : SUCCESS in 47m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5a673c27669849dca6f72e738f95b9a3 : SUCCESS in 1h 59m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0ae4c48d64764117875b44de415c16c7 : SUCCESS in 27m 37s","accounts_in_message":[],"_revision_number":8},{"id":"e583d0745e8dc08bc45ca040b30f0b4d78e6d451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-24 15:02:21.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"c9ce49f76d67c0c413b0fea4475954eec007d939","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 15:27:43.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b27c9247eb843c789c62fcdff26e58d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c627526e78b495a855d01a961181109 : SUCCESS in 10m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/542aa2b2b82b43ec99e939b282fc7bd7 : SUCCESS in 11m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/059c6b8572e44510a614741c07e5de1f : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/15e91fedfb874d5d94c51fc394bbb3fb : RETRY_LIMIT in 2m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"381c293e2c5b0f9983b367d462fc5da81838b038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 17:12:14.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cae3e5d46df74fbd974a3b584c342c30\n\n- grenade https://zuul.opendev.org/t/openstack/build/b90db6c3e61c4923a8478a48e24bd97b : SUCCESS in 1h 01m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7de8d291f484b32a100a0e1e250a245 : SUCCESS in 1h 24m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd1e723aa25a4732b0f9e497f197e2bc : SUCCESS in 45m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3688595881440a0b4ff8d862f28ede6 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8eaa7fd9e4f04474a8165cdcda0a89db : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b7d0338cb5a4ac293692753b43f9dd2 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bf695369dea4e61bc131564baabb471 : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15307476820149009ebdc82c80872f54 : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38c214922ac34a5abc2e6a88aad5cb70 : SUCCESS in 5m 21s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f591f81058264df39d0ba846511480d5 : SUCCESS in 13m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b098a8c5e7494ec1aac26daf9b4e0045 : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4416b74ad4874f59be8960bef9c6ff85 : SUCCESS in 8m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c522d23c692c4a2caa3177dc2b8025d3 : SUCCESS in 8m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/262868d7ec62431fb7e92638a09f93e8 : TIMED_OUT in 2h 03m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f73be02fc12e48a089716680ba8ebe60 : SUCCESS in 1h 16m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/86c9a213f1ba4c3e95a728836cdc1b92 : POST_FAILURE in 18m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/315364df29ba48318d2fc3f59ee59d3b : POST_FAILURE in 31m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3e913795c4e4fe6a3bf7723c2880be0 : POST_FAILURE in 33m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e183a10506ba432299390859b2fddd6b : SUCCESS in 1h 07m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a7893acece2e4ccdb24edae4a4afa8d7 : SUCCESS in 1h 34m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/20416c3c08004f0fbb32bbca87c21d9c : SUCCESS in 29m 55s","accounts_in_message":[],"_revision_number":9},{"id":"9ceb28c5f24abb26719fd346972ce63b84398f52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-25 17:41:42.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"c0df78ec3ed1e4b680a2aaa5d20653cc869c0ac0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 17:55:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c9c15c4bad4027a9592b03c3b7d8e4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e13b3458e46c4a8ca7caf912f62f9fef : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a599f21244942cabd3e2bb33a1b07c0 : SUCCESS in 10m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f757e1641d74305b6a92ad9066105e6 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/77313c10aa2a4d70868af82832051636 : RETRY_LIMIT in 2m 20s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f913383b8d6209f4922dbc87ebe166294de83cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 19:47:21.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb9ae6f1c7dc464ab4ce319450b1710a\n\n- grenade https://zuul.opendev.org/t/openstack/build/7325187d8ffa4ffa90760b8099cd3afe : SUCCESS in 38m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f34ca536b5ee453b9bc1c03d92f378d7 : SUCCESS in 1h 17m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3808c83816046a89be5535216cdce01 : SUCCESS in 53m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aaf11c695554a49b97e37d2f2ac7b4e : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac16c88a0824d609b98f704f51365aa : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fde821552dad4590b88eff5d53785816 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d7a4ced107d4e70a92a74f60442de5a : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc9b67680ba4469e884f9a1845d2509d : RETRY_LIMIT in 2m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c908902b44d24df78f6d3d27b0036bf1 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5786f3dd636b4fb88bcb779b244a217d : SUCCESS in 13m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5812f312bf464cfc914feb6bc79169f9 : SUCCESS in 9m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95fd5aaee1054cdaba9fc87439a0296b : SUCCESS in 7m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7e0df112437a4d3188b399f1bf406f3c : SUCCESS in 8m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d587f1b3570e48a68c3499a9ff853731 : SUCCESS in 1h 59m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/67cb1000a6654e3dac80da7c868c3758 : SUCCESS in 1h 24m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f05ed59bf3fc4b6186d6f27f82ad5b66 : SUCCESS in 1h 22m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec78fa91c74341b9884be36cb706b5e8 : SUCCESS in 1h 28m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/db02b739bde94221bc32036d32b97c10 : SUCCESS in 1h 33m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2420ba5be0b14b16b6b7ec006e422967 : SUCCESS in 1h 10m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b040d058e41c4817b9737ca30c818db5 : SUCCESS in 1h 42m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d24c0dede2244805a956d366e4bb7de8 : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":10},{"id":"c71eedd9170585f7090ad88eee3538111a7da933","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-28 14:31:13.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"cc874a8c9f0f616b783c21206436acbf879c7b88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 14:44:04.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c1a42c6dc974cfdbdd4b2184d0c936d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfda1f59a15441b1b94384959988d1c3 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f90fcdc760d247ae8dbf6a582b31ca38 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8989277e86d44529b86909ec47aca0eb : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c97e2678350649278a59ee2fd6c858dc : RETRY_LIMIT in 2m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d1c1a4bbe43320bb8ca818e2fc5c41b788e87299","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-08-28 14:48:59.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"1a31bd172bf4cc7d1eafd9cbe3182e6eb8b38af6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-28 15:25:36.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"a04793a3e570b855b5c97dd75e477db96b599307","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-28 15:26:07.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"1897d7d01d0607a41b4abc0afff1c95d6bbcb3ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 15:44:24.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a140bf1343b34660b22f193b8bd66d6e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7aacd43583dd440392d5451998560f71 : SUCCESS in 16m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f9fa814738554ec98cfc68490eed1bdc : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/755e8258c266488bb93b074a48cf8a68 : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1954d13a57ac4c24877d7ee67c454b93 : RETRY_LIMIT in 1m 30s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"083a2d9ea29ca3926c54e9f8bfe78a8e7160e79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 17:37:12.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5383d04b6fea4c30ada3fd7b8ab67c21\n\n- grenade https://zuul.opendev.org/t/openstack/build/153489b741164ad4b27cf5247da22aa6 : SUCCESS in 45m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b284d6aa64254869ad31cdcf2caca210 : SUCCESS in 1h 20m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f0dee15298a4864868c9191445c9252 : SUCCESS in 58m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76d2de86b6f6484f85139a58cbb7761e : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07aa03fc088c497fa8589112bc319570 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05094140c54d4cce880a9b5752c72f79 : SUCCESS in 6m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5be2374ded1b4110ab2206d73922af81 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7646abbd8e654e54b70801d4a6d4393b : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1ae5de972014304810f06266907e083 : SUCCESS in 5m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1748337fce584980aeb176981094314f : SUCCESS in 13m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4257442978db4c0fb2a90240555bcf89 : SUCCESS in 10m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/362199bbaf894c049adeed35373ef4a0 : SUCCESS in 10m 03s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0ea84a26ef2b4ebe86165aa602fc1742 : SUCCESS in 8m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d801e14268a24f25adc643f0b9dd81a9 : TIMED_OUT in 2h 03m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d756f870c5e490b88055d0093cced14 : SUCCESS in 1h 21m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ab697c9057314d1ba5cf208598ee7201 : SUCCESS in 1h 26m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88db63a9f6064a3ebf92c82ac919d85b : SUCCESS in 1h 22m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1ee14ef750c942d4ae2d96e3df254d78 : SUCCESS in 1h 32m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/23be1e789d524b56a515ee43749005e6 : SUCCESS in 57m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd2904b3496e44df876a4405f2e7ace9 : SUCCESS in 1h 36m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3efecbebac9b40178543c8dae4fca113 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":12},{"id":"f054b69fa53c6fb493d4541f95f15c87e872e2df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 08:58:04.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"ea57d57afb56b2c332bfacc08c6534158320ebc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 09:11:20.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e1dc63f96d34f72aacded55596975d6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0695117abdb5454c876e6ee71084c05f : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3edcabc957274e85b2d0fb84f14089d2 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/12094889c6cc46599540701346af4542 : SUCCESS in 11m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44a3efc1d0e442b6be106608a4e690d7 : RETRY_LIMIT in 1m 40s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"66974586ebf8e5c17a1f9ebe1e88d8b641baf5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 10:57:07.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d872b0f4f92e4a1ea0b4726c37e559a6\n\n- grenade https://zuul.opendev.org/t/openstack/build/114300d0d24b45958fca0b1b8959a635 : SUCCESS in 56m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfb919129e97460e8f94908f18b340cd : FAILURE in 1h 38m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3280e6e5d6274b2497e792891bdf2313 : SUCCESS in 1h 04m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de608979b5994eceb9048218d4f885fa : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bc673256ddb40d2b7595c00b79572a0 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f99a7313a5f4c67b59fa2207efff8ce : SUCCESS in 5m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8085e004019a4a55b51ab916988a8ef8 : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e42bfe59bdfd4574983193ee76de844e : RETRY_LIMIT in 2m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4089dc8559c4153adfb8c6a4340869a : SUCCESS in 8m 23s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5f91d9557c304a77884dce1141fe5f92 : SUCCESS in 14m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1fec29e0ac9e4c6ba1d20d994a4ad78d : SUCCESS in 8m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/128d2cdd28924dea8f62d56c3a4b72dc : SUCCESS in 9m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a3aa7749c9224c99a546237a55e4506d : SUCCESS in 9m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a69a4d7f6dca4704b702a20c26cf374c : SUCCESS in 1h 53m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b32214c595d74069b28f6b2c182f4e93 : SUCCESS in 1h 23m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8acd6887a3414a7797144b069d6befe2 : SUCCESS in 1h 20m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7d06c134e2134afcaaffc0328a3304f1 : SUCCESS in 1h 33m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d28532e779e4afab5b97a0b78f7ce1e : FAILURE in 1h 29m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07bf822565ab4375a6a5384877560461 : SUCCESS in 1h 00m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6028e3e231f44d13ba862e8fdd3d1ad5 : SUCCESS in 1h 48m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5a9c90c7a7674b3ea9f19079276f9fa1 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":13},{"id":"2dcfb15a524be89a1c79a9a857785bdeb3dd08b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 14:21:04.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"df00eab9c6d92101eb8442e72ce51623aaaf8623","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 14:34:18.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36ed2841301246e4a1914dfa06f309aa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ae231b26c4047d3895edfc963d2d028 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fde85a75c56a44b8a20eb238b54c64d3 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/83e393c031a640e3b2bf3f9e9bbf7cce : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/00795352b4384a83b6a4e48a15e01660 : RETRY_LIMIT in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"246c82a362a4cc094fc828194bdeafa39346727a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 16:20:26.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e3b4570d17a444dbc4a1f420bd5a4a8\n\n- grenade https://zuul.opendev.org/t/openstack/build/d65a40fdc55d4d64a2ee0a48d654be56 : SUCCESS in 58m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/24d5d953da7e4a18a575aee520b78a51 : SUCCESS in 1h 18m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed99609878ab44e3a445d514496b2026 : SUCCESS in 1h 04m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3ae46bb13046d8a401961603f6807a : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2251de82adc46df89079d6810177bea : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5d6ea944af54568932c46fd65770866 : SUCCESS in 5m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4290061eadf54be4aa3050307f76da81 : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/188375276ffb4fe2a0b4b7883a262d68 : RETRY_LIMIT in 2m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbfb55801ed44d68b5a2b3ef109c50ae : SUCCESS in 8m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c409f5f165fd42c98ed2be2e64ae2ec3 : SUCCESS in 14m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/333d8f52ed4b456a9b47730567a9ea60 : SUCCESS in 9m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2cdcabef56a34209a092f2288d6ff27b : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0ee1029fe093474481f8d46203c7c398 : SUCCESS in 8m 45s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/369a9bea66084751a929758e462bb325 : SUCCESS in 1h 45m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/997d461231cf4082a4d1adf1f2723994 : SUCCESS in 1h 30m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1200000188cc483082a1e0f074daafd4 : SUCCESS in 1h 47m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/855c238ffb5748e9bbeb7cfb299d58a7 : SUCCESS in 1h 37m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/89a90d079b744549b5791613d09396de : FAILURE in 1h 17m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0061e60194c84eb7bb14dd2019efe701 : SUCCESS in 53m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00af142f473145f597d0f9481c5bfaa7 : SUCCESS in 1h 51m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2c1587a3461d40aa8a0509f1c5d8961e : SUCCESS in 29m 56s","accounts_in_message":[],"_revision_number":14},{"id":"f4a4c3b21f62e46cc9d176e688b3912a2268e025","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 16:40:58.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"a71dd1caa733143572309d98b539aa06871165e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 16:52:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfa3a9a03cfe4860850a5902b2484707\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7d78ae35f8441aa82a3553bf9e0429a : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ff4be93dfc1842a9ad525f672b98ce51 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1099d678fe0249a395d1fe34fe1bab7e : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/377c078369f0406cb37c885201784ee2 : RETRY_LIMIT in 2m 19s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"931ab5cc3fc4eb4290ecd5b994c98e369d4d820a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-29 18:40:41.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"64e7490d72e53220cc022b09e5ee446d5a4048cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 19:00:14.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/004fa0d6259840868d935c52e9b0c74c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d071fabdbdc40ba917f32a778837573 : FAILURE in 4m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7280d8a5309442c789a11f94ffda6646 : FAILURE in 17m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae7b6901ddc54c5a994e598348e6b85f : FAILURE in 4m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ac262b9266b41e2bc758ca2cb9c3987 : RETRY_LIMIT in 2m 32s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8ada5589ee58c90677a01cc9912b88fc19e24ed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-29 20:36:38.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccee7a49b7294a34ad0f74a91e0a798c\n\n- grenade https://zuul.opendev.org/t/openstack/build/717ca4ba3da949c992917c53591c1857 : SUCCESS in 59m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/74872c777b274000be943ea21b7d6459 : SUCCESS in 1h 07m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0bd84f53eab347b6aad01d011f722d42 : SUCCESS in 1h 03m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c273a938002046b895d81cdd7df88a05 : FAILURE in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d92304b9bff47e2a0863079b368c23a : FAILURE in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/117a3d0f3f81428ab436bc655f663164 : FAILURE in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff642fd265a342e5b3cdd53e7668a9d2 : FAILURE in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6995062afe80418fac3dad4922a7b172 : RETRY_LIMIT in 2m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c9d0b7ee944795bfd5e64d93bcf50e : FAILURE in 3m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3237b51bd56a4281a5355515960fd989 : SUCCESS in 14m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0d8f83734dc14b6fb5897d7254e1a079 : FAILURE in 4m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2bb3b89961404e0f92f4d92c15c7413a : FAILURE in 4m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bae8cccd144e42e9bf11fe38b83c0fa4 : FAILURE in 4m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/16e220b0e2114d7db8ca491417e0cfa3 : SUCCESS in 1h 50m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5ce61e32d5c1445aa4773ea902f17765 : FAILURE in 1h 18m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/441c3755b1934e08afe1a8e8d89abc95 : SUCCESS in 1h 23m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/42970a6c21e24839aa4de73416d9786e : SUCCESS in 1h 20m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/83e90bcb1bd74bc3b9e8e8d1273977b4 : FAILURE in 1h 39m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4aa9071140954ef38fdabd7284cb5a23 : SUCCESS in 1h 04m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/27d5e1c0251a4b5fbf6aab19a6e00fae : SUCCESS in 1h 43m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f91a4b4a7db345c9aea4ea9ef93c5962 : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":15},{"id":"e31a98a972d14893f33dc0488c0a8000ddcdf483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-30 08:44:31.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"4cf1a3e3ccffe8034d502bb179331261671ebde0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 08:56:31.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13ebb086640f41849f9a900933eedf21\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95feac6cb11c4d56a5db2aa7d6acc8bc : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbe19d286e274539badd8022fa0dd793 : SUCCESS in 9m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/32ff5b26739545548cec5afb18fad04b : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0527793ac87445a1acec9f7d4daec51e : RETRY_LIMIT in 1m 49s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"30a215a555938631e0244d29f69f8c73c328ac7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-30 10:40:22.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"5de530743d68599570dbfcd693734f9c2799b442","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 10:58:45.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce94794578d4a10b8a1e27d50ac4e84\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d80ca423d6d64959a4b24d8fbc02ed2a : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c4e2f54a3298439aa1407497f1796b9e : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cae2384a0d944604bcf5955ffe759bdd : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02326a7bba6443adb2b715509e1108b4 : RETRY_LIMIT in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"ad47deba92fe49db8a3aacc168497e710dbfe88b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 12:39:00.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3c88e216574f40f69569d226ddcd3b90\n\n- grenade https://zuul.opendev.org/t/openstack/build/a81b7a1b44b7472285578d50142079fb : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d3655fb220b4522bd6f1359351bb4f2 : SUCCESS in 1h 23m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f59675c4c3d84c02a2b467e56c6b3b63 : SUCCESS in 59m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a6f498db2304f5b98fc5be747b799f5 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05bc856496d54d218198e1d48fc69c2c : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38735505ca0e4fdfabe6e1b2c63be016 : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f0640a3d33b4f30aa0cc1efe7ab935b : SUCCESS in 5m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08b40adb50b8423f8cc788f7265989fa : RETRY_LIMIT in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d94342711b8d419ab27f25c13b07ec52 : SUCCESS in 6m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9bba2361224d41c8babae0f9c6526df1 : SUCCESS in 15m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b65d993f5055478391f5a8cd0861085b : SUCCESS in 8m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9dc99e70cf22441ea7fa5696cd8529f2 : SUCCESS in 9m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fb0bc111a667426c83a3e44a08a3dee7 : SUCCESS in 9m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b82da123e8dc434286e209b92ccb3b4e : SUCCESS in 1h 51m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/20cfb708212945dfb7f7f660cc978475 : SUCCESS in 1h 33m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f256f384a0084f2289e1e7304359066f : FAILURE in 31m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a804b26a459453ab4b1d4fa8b0a65b7 : SUCCESS in 1h 27m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2684e73b1f3349b98a89577d276cf8bf : SUCCESS in 1h 41m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1128c8b83fea45dcaadfee16b840042f : SUCCESS in 55m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4b9a5ce90d34eadbe5e7fc1b03a6d28 : SUCCESS in 1h 51m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d8303fa3c0934e5191ab245ef7e08a13 : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":17},{"id":"ed1674241f83deb818e4e0c3f0203f006542e1cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-30 13:57:53.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"f3c96e43db2ce7719c1839006aa529925168ecab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 14:11:00.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26f0ff808e594dd4be0a40ff277f6b66\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8e62e67eae241a786e4959de0b28ebb : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad987f18c0874e01ab9ee96829cc04d0 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/656c8b6ecf6f40ab8cb26e5ef84b193c : SUCCESS in 10m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/297493246a57479aaba8b80fbedc05e5 : RETRY_LIMIT in 1m 54s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"e995784329d3a5eec809ddc9c0f12b19974aaaad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-30 16:19:41.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec9a775cad5348ddad267a2de4c652e9\n\n- grenade https://zuul.opendev.org/t/openstack/build/873f99d7a29b40a5b4777bbcee1c8a88 : SUCCESS in 57m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/51b645fba6a14c129d13573c395e35a7 : SUCCESS in 1h 17m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3a114aa6bfa4ed7b19a83e70c497def : SUCCESS in 1h 09m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/154428f818c94efba209eb4274e2ec3f : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ba73553aa624ee6954e894b7c8c0b6d : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/182c6f82eb824dd997e7a600c6a3749e : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e28ca79335ae4e529b07567d857c118d : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc77c31b8605411bbd79513e6cf47a40 : RETRY_LIMIT in 2m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73658f2b930148019933f6f20e71338f : SUCCESS in 6m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/36d65b1a07db4a08b230fcfc1071aca4 : SUCCESS in 15m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ea28b74dce14e878220bd81a6b8ea0a : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/039e75091a8b43d2b187d55176f377f5 : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7f2ce6131f2a4fd5bbf9f479ff900aac : SUCCESS in 9m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/77f7c6bba2bf4f54920d269ddc1d6ccb : SUCCESS in 1h 54m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/97c5e90eac6f473086c9023ea136c233 : SUCCESS in 1h 19m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d58817a5ab024653b0bd7ad38fc8acb6 : SUCCESS in 1h 18m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5bb465191cda4d539e6ab0fe13808d16 : TIMED_OUT in 2h 04m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/073e3fc98b5b41a5a6e19c2b7e734964 : SUCCESS in 1h 45m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02e5a2d90ef54438b32be166aa3dd1af : SUCCESS in 1h 07m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20ac561b222f4d009a411f37ee2b7201 : SUCCESS in 2h 14m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fa9bcc7a01094578bef90dba420b19b7 : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":18},{"id":"12b7f91e7f3ff40a3edc8af494767a029ef7b683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-31 07:57:48.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"5e49ec7459aae7ce521ff54d2c8ca10df9b63b45","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 08:11:51.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8524a792649946a39c026d793cb976b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1cdc6fc1afc44ed95b9865c1e0182d1 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24566267ac814e57a1b868149b433850 : SUCCESS in 11m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6d69196285e94491876117a466ff1370 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4b0211684a4455d9b09e909d6a119e8 : RETRY_LIMIT in 2m 23s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"5ddcc0a4938f33f50a9ea15188d019e0d19637f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 09:57:39.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0c4f808d079844eab1eb245efadbac97\n\n- grenade https://zuul.opendev.org/t/openstack/build/8d8e775481c94f71be9077ea49d528ac : SUCCESS in 48m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/364fdd246577425c92d6500487483806 : SUCCESS in 1h 17m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7857f59c7134e898158b889e3c016f4 : SUCCESS in 59m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff7e799ecdaa4205927e553f850243d5 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe689a16b88a41a7bdafcf2cfede9030 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc6431c7ea1c40198d5b0a47d8c575f1 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0176bd6013584a6297a3bd49ab76b9c0 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dfe62239fa5943049d03fad737174c52 : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f18f11a9d144824b9d3a3d562340d82 : SUCCESS in 7m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/79aa01a1c63b48c986379444235f6f13 : SUCCESS in 14m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a31e0a1eed0342dca7148af6725f13f8 : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/af85441eea434fcab70035b299d237c4 : SUCCESS in 8m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1ede0c08b7c34a02a0170cfe668f1d92 : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1fbef46875bb4d6cabaef2337799640c : SUCCESS in 1h 50m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a0429b4497a74398b81d83ca0458d7ab : FAILURE in 1h 08m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a8e1306abe244f2ba708ba59db13575 : SUCCESS in 1h 25m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ba55eeef113f4c80b33ad131b71584f8 : SUCCESS in 1h 30m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9f1d6736d7f548e7832064eabf31dd3a : SUCCESS in 1h 30m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/337c46147f7f41a49a4c02f82ba504fb : SUCCESS in 55m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c937d84359274428865e0bba6b1cb356 : SUCCESS in 1h 54m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bdb3fb541de3434db369db9969948f00 : SUCCESS in 27m 58s","accounts_in_message":[],"_revision_number":19},{"id":"151930fc870cf7814fa633d177e6373dcee61256","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-31 10:08:08.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"f294617456641b831488a874cedec03388781f86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:21:45.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fce562de7e584c4daa8e4ac464a9bc28\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1611d0849b243599b709b6465e32a06 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ea52a73783248a9b7a7ea2fd8a33137 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e9d8c78712db4166864acbabfbb9a732 : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a9f5139a45674f0d8ba5e02190f8b932 : RETRY_LIMIT in 2m 43s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"6631b3563bd4b80200518d9fd175f73bd25b51c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-08-31 10:27:37.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"85611025d96cc236613b04cf69f0c14ab4a8a8c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:41:30.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d1c0770ca8a4787944467c8e4b34643\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc8b8bdce86d40f8869cb0814355ebc9 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26dcf25dadd449d599d97c9bec622261 : SUCCESS in 11m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/10b9327aae6b46a18f880ebae7d445f1 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0073c008e6074f5cac2f2881ca47e42b : RETRY_LIMIT in 1m 42s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"534450bca35fe90c1ababdaf00d6b82a646c5d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:11:52.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9770669d98714f719b602f9cfb81ef7a\n\n- grenade https://zuul.opendev.org/t/openstack/build/94f120eba3c04170b2c7479853951f55 : SUCCESS in 53m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9abfa8dbe31947aa85889fada101155c : FAILURE in 1h 30m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc0bcc801e3742c2b8c0000a09c67192 : SUCCESS in 55m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8946be4917440adbf8d06c91f95da0d : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/034976ece5f54bc88ad12262316929b5 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/705744ffa9834820ab7a48eff5daeab6 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4dae066a80ca4039ab2c640b397962d0 : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d17dccdd1bd24a33b22edc10dc5da1cf : RETRY_LIMIT in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/261f73f0141247acba35ce94f72c8577 : SUCCESS in 7m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ac9a2a559f00435fa156aa66babf4c0f : SUCCESS in 13m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8756b4360ad14cf9b0531da16f2e56f3 : SUCCESS in 10m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36129a5ead6846e2afb570a383716ee1 : SUCCESS in 9m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e724477121784f1ea1593ca2591756a1 : SUCCESS in 8m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/37105ea51980434abca27bce6e9ee0e0 : SUCCESS in 1h 40m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b259619ff41442cd9ae2db29088fe8bb : SUCCESS in 1h 24m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31cd7abecf1c407baaefdc10054696bc : SUCCESS in 1h 31m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/46529af2a4b0491b92587d5031a2ea27 : SUCCESS in 1h 19m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/875fa3dabebe4f69a128582fd3067237 : FAILURE in 1h 35m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f590d1b182d543ab87b24798b84b4525 : SUCCESS in 1h 03m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/46eda4986ff34e978fb3215d76145b35 : SUCCESS in 1h 33m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3f8f2420860e426296ca340687bbd30b : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":20},{"id":"4c9b8113ab6ed5b9c507d962d68b95ee79fa4f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-09-05 07:25:17.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"84eda7cc8a969e128d91322963511a12d95939e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 07:39:52.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce952fdf919b47ecb4789b43d377fb6b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6904fae094045049013e6c5c8a7360a : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/89ae62e2c1134c879952c7389c639dc6 : SUCCESS in 11m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1510c1b8176c4398844c91869968178c : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b4fbc48ca47d43b3bd8aea6cc71095a0 : SUCCESS in 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"d70c7325b44e53c4bfd1b0d31152d4a546ef4ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-05 09:17:09.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ad9919842e94bc0bb713c0ef7ca65b9\n\n- grenade https://zuul.opendev.org/t/openstack/build/58dc34ca199a4a9ba9d2fd3b13e9a807 : SUCCESS in 38m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a3170a373784925b361a1e8f0e1f2f9 : SUCCESS in 1h 27m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80e69bcf13a94582ab8968df431d594a : SUCCESS in 1h 02m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92a7820eccfe44bfabb9e3d9fc0fc51f : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8dc95423e054c28ade957016fdc8619 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcbab82df0794d0ab40665f35981d2b4 : SUCCESS in 6m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fff2b4139e648ee9c27f4e567199e2d : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/955da535c5a041eebbb4bab2c32de63c : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af7c8c4847544b5cbae7c75dbc10adf3 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/35b3e8d901c44c68b48b40fe20d6e6b3 : SUCCESS in 13m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b47dc0a3ddce4bb4affc3afb97b3b551 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dc46565d1cc246dcb568da05d8c4342a : SUCCESS in 7m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c5b08c2066c410f910765eb967dbaf5 : SUCCESS in 8m 35s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3f90f733ba26476682e83a60398ab6ed : SUCCESS in 1h 46m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6eea4ad7041d4de694c160168fdc451d : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6c3b2e92a788461a976756f1361a8f20 : SUCCESS in 1h 26m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4a31454f5bdc4e07bc10c9af6082971e : SUCCESS in 1h 19m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8959243207094da7a7dcd538fb27513a : SUCCESS in 1h 13m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17cd1b5051074486ba12ed2d8ec5d68a : SUCCESS in 56m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c2daaa4f10345dbb234e7c02cae6752 : SUCCESS in 1h 45m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/278d4fad5e6d4ade9d0ed25dfde53502 : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":21},{"id":"8128208399231d4378ce8cc8cf4bd7a3e95c56be","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-28 13:04:25.000000000","message":"Patch Set 21: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":21},{"id":"c2c861c5cb68eefcb8ffb220e992052b553c247f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-19 07:06:53.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"5f7ac3854f8de89411f2daf28e9f8af58714d585","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-19 07:07:24.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"06d65a621a36229ce193268e84a33ca39d846d82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 07:37:30.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf3eae384ca54593a42b8c8116b89a96\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc7ffbb9b1134b199139ebb93d246dd1 : SUCCESS in 12m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d625dfb5a8e454baa00c5eb71019661 : SUCCESS in 9m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb12d1f92481451fa143567f7638e7b4 : SUCCESS in 12m 31s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"75ce71d31942d785991ecad25afed5179bbfea24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 09:28:04.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/595814d0e0d041c2bdd28966d71835ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/6d645be02464404d83557c546deffc7d : SUCCESS in 57m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6e110555ec0f4d65b802dcd167c83a2b : SUCCESS in 1h 03m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3d4939d9f65e44868e138dbf37c905ac : SUCCESS in 1h 16m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c23f67085484cb8b3965ca9e348a87c : SUCCESS in 59m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baaf239f087045d3b717648335a001d4 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c2da60def144909bd0695d0c984151c : SUCCESS in 6m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a687b3ecb844ddd9589e580363482f9 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/badbd0de65394030947120f7de18cf1b : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c968f4b428544cfb985a650a89f8275a : SUCCESS in 13m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0340e6c2874d40ed997a61f0156dd442 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8a4b4d3b1f74b219d21fb150fb6bc09 : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/49e370fb5af24bfb9b97d50988da7416 : SUCCESS in 8m 42s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7589df98fb324ff29108a642497d4d22 : SUCCESS in 11m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4a9bc4216d2549bfa9d43634a1a2bb19 : SUCCESS in 1h 52m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c5829f98f87943eabeb40ed12afa65cf : SUCCESS in 1h 57m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e17d74809e6a44499ee1109edf864467 : SUCCESS in 1h 34m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/25fbbf9f4d0242dfbdf4f352dcba2508 : SUCCESS in 2h 06m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6675cb37d0aa4d6796c5b0c0905f0b4a : SUCCESS in 1h 52m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c673bc6126dd4d78bda8669acff5b274 : SUCCESS in 53m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48d465481d5e411890fba07166160c7e : SUCCESS in 2h 16m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be1eb75ae82b4515860633ab0e93caf4 : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":22},{"id":"47d97c778eec5fbaf47499aa7dde985594c6d319","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-31 09:27:15.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"9c02c1558c6376015849cb552b231a36fa169f51","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-31 09:33:22.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"832c0be6468bc033b1569310a1711d6e874f4b75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 09:42:54.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ad7c5fa5b544efa6f1b92cab59c5a6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e1cd746fbb64cf8b3ddca0d0903b6c3 : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c8546e609c2540f2b1a9623c30ebd463 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0583c9c1ee42410291675b4163314dd3 : SUCCESS in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"9f466a2a84f75dd57f18b2492d1f8618d9c92c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 11:29:51.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dafd7cc2dcc94a68a30c2c415a1d931b\n\n- grenade https://zuul.opendev.org/t/openstack/build/811e71619adf41868ce5924db2b8e3ea : SUCCESS in 56m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5d0df420c47d4c6c81994656d0ea763d : SUCCESS in 51m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22ff7c76adec4e159fc2e80b1783c022 : SUCCESS in 1h 19m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7392049cc654ac7908f09c792ec54cd : SUCCESS in 55m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f647ad3104c14b63af25623959061878 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59facdbb704c4e6eae00ddd5b14508a8 : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bf67e4ddb2084919abc616ee1f541b2a : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d172ba1ac8e94721a03ab17ec1875370 : SUCCESS in 6m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0678a9a5e60140dcb79abed5c88b39fd : FAILURE in 12m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5c4b66345c8d450fa161885930e7c018 : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33ccb482d7db4363a49a598be4aa3e30 : SUCCESS in 8m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/df4993c5d67e43d395f3e120fe8377b8 : SUCCESS in 7m 47s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/27c1cbe057ba473385627666a114adff : FAILURE in 10m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f9a861b366624e8fa4c5fa344bc0a479 : SUCCESS in 1h 49m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0cb5eb80f0eb44bf803e17181f7aca8b : SUCCESS in 1h 36m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dcfb378e08b141f786a51bec5cb38a6b : SUCCESS in 1h 42m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8803ba794c214e319cbe56f6120d4acf : SUCCESS in 1h 12m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1ca71916fb14f919de9241ade01d54e : SUCCESS in 1h 57m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61b383a9cba5497795b92583bbecd868 : SUCCESS in 56m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2656e3d3f9384ac08dbe1b7c968ca1f7 : SUCCESS in 1h 41m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a6ada86771e840b8a48de5cc8f470d16 : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":23},{"id":"a02b6671184fd772655b3befe55981d7d025b231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-10-31 12:49:39.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":24},{"id":"819a7dc3b8cd1a31a6ed73de269421ad46235027","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 13:05:20.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f1011034234e8aa1a8936490a36f96\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5393a83490214b4ba8b5c679139f3304 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2df4a8fd9ed14e8398c7527ccc4e28db : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f2b66cddedcb47ac8c066fca75c64447 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d637cf47d3e63f92cd303b44289e4abc3c176358","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 14:57:17.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36b87690b9e54bec8959f1d04c399f9f\n\n- grenade https://zuul.opendev.org/t/openstack/build/c16dfb3cf14a4591a8dba86b71afea72 : SUCCESS in 55m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9c80ab9a2531452080736893e9a53b4d : SUCCESS in 46m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ae64feb15924b32803ba42718fd3d90 : SUCCESS in 1h 16m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a378397305f34b97bf00b061148fdece : SUCCESS in 54m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c72475f907409aaedac9559e7c0eb6 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30767b818d94420a85dd107df5f23288 : SUCCESS in 8m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66ca78219d3c41dabc3e337ab9e8ee1a : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/188983f5348b49ef84b55162fd5ef8e3 : SUCCESS in 8m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b82af437353c4c2ea60d93e55e192662 : SUCCESS in 12m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/82ae21cd07b94d37911a1bfd3812c344 : FAILURE in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bcee9751e863444bb4245099f183c49e : SUCCESS in 12m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/227d5c49c59a4052b5b00348ecbe8f97 : SUCCESS in 9m 01s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b9d1c754baaa4cc5945e1af615aec5d5 : FAILURE in 13m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8e3aadf8a3f243deb8994ecd79b5e519 : SUCCESS in 1h 55m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba0ff3e41faa42ff8146842d8a8e9312 : SUCCESS in 1h 32m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4941141aec5040199cade38f402f6aed : SUCCESS in 1h 36m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/22c70fc818994291bb67a2fd174090f3 : SUCCESS in 1h 52m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4cbe45a0a7b44d97abb9c5dfa6d1d8d6 : SUCCESS in 2h 01m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49488c43f9fc41dbbf22dc4463dabac3 : SUCCESS in 45m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20d9b21eda204ac7b2f23056a217b852 : SUCCESS in 1h 51m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2fd44f963574a28ba4e7ca91ca9fe89 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":24},{"id":"d32240f6d34269fc912d057c47083e7cf4b471bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-02 09:35:31.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"63d69a7bdd6b9b1a71aa19e1407be52e208a4715","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 09:47:17.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3a5cde4c8d43feb44104ae65217dcb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d06bb657c6a4853a3724ba14bea8791 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e0038e1d9d724e7bb79d7b92504429df : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/375a431001844b659cf4944849eba108 : SUCCESS in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"084e7f54871b962b1fbb63ff9ac80bb873c9cf2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 11:40:26.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8806c5cb1a8b44a897a5d7c27ab8d80a\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e90c1aeaef14480b5654d2ea499a854 : SUCCESS in 52m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/71e5b82feb1346b18a6083f434ea0ed6 : SUCCESS in 1h 00m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/29d86b21593a49acbe9c6b50d5bd4557 : SUCCESS in 1h 45m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a0f5c4da2fc48fbb62f8cf36ac4a7a4 : SUCCESS in 1h 08m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ecca9026c04443b9e53b73306547c74 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f657f0c3178c4a9fa7a4b37998ccd181 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be5e5fcbb1f94ec992527f17150e05df : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e442140f6114444aa091d5721a1ed41b : SUCCESS in 6m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1d37eec9517542e0af07209f654e5f91 : SUCCESS in 14m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eea9be4358cb468998d0dd4ca9b834cb : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83e6108ba81a47e9b794a2df0dc9e23d : SUCCESS in 9m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c0c89f4da174d1e8afe0296e702aaeb : SUCCESS in 9m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6fb1182c87ed47c38d182193311b2754 : SUCCESS in 10m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7ea005b152e648008099cbe7e04dd4db : SUCCESS in 1h 50m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8d2dcc6db27241a7aecced2bcc856810 : SUCCESS in 1h 37m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e5fbe8e6b3684316974132df97561083 : SUCCESS in 1h 43m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2cb1176453014f45a9eeedc678ebdb5e : SUCCESS in 1h 33m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9dd61066907c4d4686491b4699ecd5c0 : SUCCESS in 2h 00m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea4d7ef0174f436a84b9267c56216a47 : SUCCESS in 51m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8b8dccfb8054349bdc685f0814358d6 : SUCCESS in 1h 47m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/282b908680a04d37995931d51e7a5b0d : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":25},{"id":"72dbfe0f64b071ecba5e584409c7b0f443f42022","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-08 11:23:44.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"78ab8e1e1e42251c974165eb367a12700200ea7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 11:38:14.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46e08c1926484250bf17b04495cf2564\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce4854e91ad44db6bbe407ae71b02083 : SUCCESS in 11m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/984799392be144a1b0fdc3359316376f : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/68135c7f2a7948a49c9de3cabf4c8782 : SUCCESS in 10m 03s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"08a2bcab711a416b6bef729deedc2601d9259c4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 13:32:01.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a6b37cd0ffb455795fd874825adf0dc\n\n- grenade https://zuul.opendev.org/t/openstack/build/c16720718c304117b082a8c63b2844ba : SUCCESS in 58m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c862aae086c84d93a1d3ec5a7107dbfb : SUCCESS in 1h 02m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a2c0e9062d6492c97f261cc0bff591d : SUCCESS in 1h 28m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f7df71068c6e4010977b5d83afab298b : SUCCESS in 1h 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c27f47136e3c459a892d411b6fa2f71b : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/880fc6d4a030400f80a3ba7cd33dab0d : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee24791c6c0d4d2e834551e432cbe5eb : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e09046dd17df428b9d9b550cbc4b16fd : SUCCESS in 6m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3e4f502862ff4feb98e48c739ae900b4 : SUCCESS in 14m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f140b768b7da4f5288256c820b4ddbfc : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/46ad621969004d9d8bf24707d0528a70 : SUCCESS in 9m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/69886e91f54447bb922df38060016729 : SUCCESS in 9m 02s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6ef47814a70440ac85c8c9d5f5e1cbc4 : FAILURE in 6m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1430858250c74cf9b67484819cbf8874 : SUCCESS in 1h 56m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/289216d1d1494dfda27880bb4a35eeea : SUCCESS in 1h 54m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a4cbb3f6b3e24efcae6c6495eb6eee55 : SUCCESS in 1h 46m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e278b4e79f534b43ace4ff97c5143088 : SUCCESS in 1h 41m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4c8df731aaa34f1d9f7b7830afdfec49 : SUCCESS in 1h 58m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/befed872e4d0448b862d9ba301499215 : SUCCESS in 1h 13m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec5192b1798a4e60ac6c7e279fc061b5 : SUCCESS in 2h 00m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/feaedc7ac0c949c180c33734df2dad9f : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":26},{"id":"520190b5f5352012221f8b962d856ade83b632aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-12-05 14:08:25.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"f3b642305d4a91956479fddb62c6f7773c915d1f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 14:33:52.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/013baf4651cb4c499088fa9ae3c67dd4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a280b0c550b54f0f820a059cf2954145 : SUCCESS in 11m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/48c12f56d1b247c29ba36907c3a3da2d : SUCCESS in 10m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0af1e6d48b994373b7e9c3eaad09cc2e : SUCCESS in 10m 45s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"135726f1f454d28d15373ab299e1273a1a57b444","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-05 16:23:01.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6f84ce4bc3db4d72a3797c1f9b55fc2d\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f2698f4754e4ca0b23afa19fa10cd31 : SUCCESS in 1h 06m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/12a370055de74262a7ac73962e6b9677 : SUCCESS in 57m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4e6af02691334288b0f15a52bc5431ef : FAILURE in 1h 22m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6132f4996db04404861a8a10e92dfec1 : SUCCESS in 1h 06m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77d8f80b2fc5432f80948d9a57034f87 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7d546462d594a70848b9e1b04e33de5 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4c3fbfef64694be684957d8da79e5883 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94a64a9fd08a4ee0bb9183b899eb0720 : SUCCESS in 6m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/459b0adc809748f79586cfc4cec54916 : SUCCESS in 17m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7e63dedcb9f147c1974a6aee4f3fb0bb : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12bd51f94fcc4974b0a06af08ce37c9f : SUCCESS in 9m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/753dca3d5bf94f93ae795879e2063eb1 : SUCCESS in 8m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1265314a20b04554b79c3ed566327ea8 : SUCCESS in 11m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ad71d7fbedf14bd5b3ca0e25f87dd2ea : SUCCESS in 1h 54m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7cb35b2e5cbc427fa4697639f27f2304 : SUCCESS in 1h 50m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a6aafbe18cd845bca29c26f4b0bf6d46 : SUCCESS in 1h 49m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0451757e8c6437e96d7cfce18e53364 : SUCCESS in 1h 48m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bbac8e59747a4b93b8a1f6de6ec32769 : FAILURE in 2h 04m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad5fedeed2aa447da699c2a905193907 : SUCCESS in 1h 09m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eff38309e1ea4f4581b0d7d781b40341 : SUCCESS in 1h 41m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38be0c12232d4ca09a35bc07e8173b1b : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":27},{"id":"ff4e65ea427de604dc1d2fe9320bc3c467aa3da6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-02 12:02:35.000000000","message":"Patch Set 28: Patch Set 27 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"e475e55ff9bc106b3031c6c10c285c52d9da412a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 12:13:34.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1c778a954e8430c9a07cff3f338c604\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f60e1f3986c45efa7c81577c4891ad9 : POST_FAILURE in 8m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49b43867bd464ae7976fb89047f26492 : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aa2176da43174e6b86ad79abdbab4e15 : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"6c8f24e0bb137c7ebbe74f3b6f32f904329292c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 14:09:13.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95d7580728b14cbaa68be7cd251dddea\n\n- grenade https://zuul.opendev.org/t/openstack/build/7b926ac277de43289698be582240d9dd : SUCCESS in 1h 03m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7eee12023cc74489bb8786fb310acbaf : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9bb10d2277a44ec38337aeaab13884a7 : SUCCESS in 1h 22m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9abc2b49ff94ee4b391e18eeef62b0c : SUCCESS in 1h 05m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae441de44fac485797fc2711dc827c02 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3243e3be406b462d96d145bdb9f03b5e : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d0e33ec52092443792422092e6cf5157 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc2fcd69320482ebfdded27514f7e67 : SUCCESS in 5m 53s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d07b91c44a2047219fdfa6b2a094b8cb : SUCCESS in 14m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6f156d3a49104b659369eddfa4c07beb : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8e05f30012b749968590dbf02156ba0f : SUCCESS in 9m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c85c18f50390447dafb0e51bb1bb2b95 : SUCCESS in 8m 03s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/53a233f64166478db6ad71b40c180652 : SUCCESS in 11m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bbf9b83598d9494c834f54e49ba29e4c : SUCCESS in 1h 58m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/39c069d6351f48028172642782939ed9 : SUCCESS in 1h 11m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9f8eff25f4bf4db9bd7a4694e59899dd : SUCCESS in 1h 59m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a099e88c80694ef79b692b6e213b415b : SUCCESS in 1h 18m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d607157c6d4944569fba16520fcf222a : SUCCESS in 1h 26m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26a016bd9da845188acb3f86fbe864c1 : SUCCESS in 50m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ada8cc9f85b45628750ea76f0ee943b : SUCCESS in 1h 50m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4c6e05c0bb5f4ab1bc2d04b00ca2f279 : SUCCESS in 26m 24s","accounts_in_message":[],"_revision_number":28},{"id":"9484e209e5ba3c1b25989a1cb7c7fa2c9556426f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-11 07:36:44.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"7942d2235b73e02b5557730c40bb61357edf55a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 07:48:31.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71d756d63dba4589998c0d9450da4498\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d002ed589f446038baccad599705ba7 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4939fbe889364d268df1ee81fa35900d : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/45c8c702583f48039d5c25698e8f28dd : SUCCESS in 5m 46s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"1e9c5d48cfa550c9deeaa8939720572b3ec3947c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 10:05:20.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f137f6f477414531a7cd686b0540bb6a\n\n- grenade https://zuul.opendev.org/t/openstack/build/cf92014d130744ffa7757f773d024408 : SUCCESS in 51m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9dcf33d517b84a55b8a93996b9573a8d : SUCCESS in 48m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b11a36358234b2e8b52a8caf9b2e411 : SUCCESS in 1h 23m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad41a836521541c5bb59d2cb99a506f1 : SUCCESS in 25m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/686c52effc3a40cf9a135e700028fed6 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5f6221b1a0149a48fb5ce5cc98069ae : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e234a712109641fda1a80f6e5db36dd9 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dce1ce1d59245cda7b97fcb5133001c : SUCCESS in 6m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/41df5b5eb8a04da68e91723dc097fa3d : SUCCESS in 13m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/950c6c7ddb6849e1833eaf158b48e3fa : SUCCESS in 8m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3632eeb97066412588b4325006ea5405 : SUCCESS in 7m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3bf49fa24a7b4cd3918b07527966c8e0 : SUCCESS in 8m 58s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/53d66e9db9db4a418d20784e269fdf16 : SUCCESS in 11m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a0e73681c234f2aadbdcb257a96b76a : TIMED_OUT in 2h 03m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0965552d1e4144e98ad2dc9312eb8a01 : SUCCESS in 1h 33m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/369af3e96c7f44b0b2999271fb9e2b0e : SUCCESS in 1h 52m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1a0cd3bc681f4cb8b8e1978a18d00920 : SUCCESS in 1h 47m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e54421918d05494b814dc1831d3f6552 : FAILURE in 2h 03m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c0e3359c7d943f6a5308facf06d81ae : SUCCESS in 56m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/599a1af310c04974b40030e4595cc8ce : SUCCESS in 1h 44m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e76454dae26e4ec881a44d3815bd882d : SUCCESS in 41m 02s","accounts_in_message":[],"_revision_number":29},{"id":"06a8db69197daaa33ada8b040f7c588ce904c78a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-22 10:58:50.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"59b07ef44fd03a1de23d74ffa32e55566f6090bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 11:22:39.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93968e573f804e0b814e866afd3b9e17\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b45dee46cf874b748c5e22fa55d00f31 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7fe714b6ab114b88bc0c44445df5758e : SUCCESS in 11m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b1470d6fe33f42019a646cc212fdac06 : SUCCESS in 9m 43s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"64816011498d47c0afbe684798ca3a24f6a77de4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 13:25:11.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/83b715fe9a094a7fa4987ba8ac19a2ad\n\n- grenade https://zuul.opendev.org/t/openstack/build/9783c74136b94fe8b09c81ea312c49fb : SUCCESS in 1h 15m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0d8fe0008c8f4ed985ee178d52e728e1 : SUCCESS in 1h 09m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52a615ae58d746dd93739dbe8de9a5f3 : FAILURE in 1h 41m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a9d8f8cab7a487c9cc0413165841f4c : SUCCESS in 42m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/821c0a9b231848329e1aa7cd96c906ca : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/630e3416259544e7acfe034b1dea7373 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/280cece81536486dad577e2bfee7220f : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99cbe7f8ea5a452fbbb31105ff7794b7 : SUCCESS in 6m 42s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bca9388682134daa81f6c8a4a9696bc2 : SUCCESS in 13m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb8b1b1465154b31970724219c837bfd : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c1bf388006174fc783266ee9cdedaf79 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5c40504ad30b4e05887ca7f1f553e817 : SUCCESS in 9m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/163402fd76c24cd797d7c24e06cc761b : SUCCESS in 10m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2e84e3e2db3844f4a66d7868355d7c18 : TIMED_OUT in 2h 03m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/721fdbd4d27b4be7810111d99698e3cb : SUCCESS in 1h 48m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/288b59e4e7314bee89ff22197df8ea26 : SUCCESS in 1h 40m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/19b7dfda65fc4aeeba9c78b9d264a1c7 : SUCCESS in 2h 01m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cfb6cfec70954036bc195a5919f83947 : FAILURE in 2h 13m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/768274e0662c4b23820c377900c3b0f5 : SUCCESS in 1h 06m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/966bae7a497e4770b4cbab1dd45ffbf5 : SUCCESS in 1h 57m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b4916001f3c64fb3934f97059e7c8cb0 : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":30},{"id":"59275cfac5e4cecd9fcaadbc7e72222724160792","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-23 11:15:20.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"c5de324dc763621874f970c67e038b4608c43a8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 11:23:33.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc62120bc204afb9257e87157c77c08\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8f114ff51de4827ae5c98dcfc4f9477 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7dbdcbdbc48044eb87cb85ab188bb844 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fcb70bf2751a437596e214cc66c262b0 : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"95686d1b566fe701d42214250502c43decb22cd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:29:29.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a943227acd645a4b1ddcbb028075dc3\n\n- grenade https://zuul.opendev.org/t/openstack/build/a46bbcf953c24573b966abcff7f95d18 : SUCCESS in 1h 07m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0a32c82bee5f4d72b9aec63ff42fe6fa : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52e205f6cdfb4d1ca2c87eb8a6883cf2 : FAILURE in 1h 19m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c122b25871454b8299d3514ae7880419 : SUCCESS in 36m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13b2394d6c74d11ac56d5996f0fc915 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62d7af2b66164a7ba7197305412f5fa6 : SUCCESS in 5m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fde0f4b5f6c94617bac9c6ef13f3298d : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f50a7282d8f468880065ae0ef2aebcd : SUCCESS in 6m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b639231f3d12480d984c4d2b96435b05 : SUCCESS in 14m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abf5c06f587e4effb58e65a82f5af42c : FAILURE in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a4058fc72b694b3cbfd727af84b287eb : SUCCESS in 8m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2d64281e1e6641188c9b2f35781334e2 : SUCCESS in 8m 38s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f8ff3698889349bcb4eaf2b749d23cce : SUCCESS in 12m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4d6aee0a6437444ab9467dee52b4bf06 : SUCCESS in 1h 55m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c8eb8d06a3574e73b4cafa20e2f1d959 : SUCCESS in 1h 48m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7ae077aa31c34a78b91453a386e01c23 : SUCCESS in 1h 54m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4cd4fc4e3fb745f5960d3eb354d85b2c : SUCCESS in 1h 44m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dd0da4905f124ea8951341a68b566fb2 : SUCCESS in 1h 55m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85ae4bc5d18c4b38a3f2c99bf9a5c2d1 : SUCCESS in 1h 00m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c2609cea60843f5a5996682d08672b5 : SUCCESS in 1h 52m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0d46374bda6c4dddada7b8bdbb49de11 : SUCCESS in 22m 01s","accounts_in_message":[],"_revision_number":31},{"id":"924329f2bcdf13f187f6fe8790f5f44993ef6a35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-01-24 10:53:02.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"2b187c4488d4d8532a81209d8fc0ff7fc8dabc63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 11:06:52.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50944dc659b24db7b763e376f8c86b8f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/730bee941c86437d9762e7bef8bdbe39 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72d74db6e67c47579bf4ff8a0c6696ba : SUCCESS in 11m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c82327b978649a1ae40338150a6cc20 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"e9b5ecec5a06f2fd80f0f8f445ebe2c06a798b9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 13:49:23.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a0a217d608f4aa88e9f54eed2bd216d\n\n- grenade https://zuul.opendev.org/t/openstack/build/7b5b2d2867974d26b923f24dc45c8673 : SUCCESS in 52m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3f8c2c1c47274c0a92a3dfaa3d62f2b5 : SUCCESS in 1h 15m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7da124456a7d4c7bb93062c35ee60cf1 : FAILURE in 1h 10m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e2cbcf6abd843e98d375dc01fe73be4 : SUCCESS in 26m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42acc22b0ca947a19e226e42dfb3754f : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a1d4625f2494355875ae1b36c4803f0 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28e52b5bff9e4038a3e834a9fd17f214 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545abeeaa3954ae2ad9dd0ee4f645740 : SUCCESS in 6m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3ae6388431074fc8b8911d6ed62a8ccb : FAILURE in 13m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/44f07db4d66247df938b4732adf9e51f : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c960e22d40ff4514852fda8208bfda6b : SUCCESS in 8m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2042b571c51141258a1220f49f97df85 : SUCCESS in 8m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2fec2110851f4dc7bbb010075020a96f : SUCCESS in 11m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3f350bd5b5e74d6595a35aae306c6d33 : SUCCESS in 1h 56m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d445702dc13e4b719937c479a5943242 : SUCCESS in 1h 26m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8f4c35d5a78947db95475af9eca4ceed : SUCCESS in 1h 19m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fc90fced03514d9390853bdbf499ebb5 : SUCCESS in 1h 40m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c346ff1ed36b4f0e87a99580574b2c17 : SUCCESS in 1h 33m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8f44c14f52a4a3a9259c3f27792652e : SUCCESS in 54m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0433745c8eeb4e41b054fecf1e3bdd77 : SUCCESS in 2h 06m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0063b7a6b4d343889158f7ed30440973 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":32},{"id":"237e9e72b48ce74df221cf5882d3e72cb741b619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-07 06:50:50.000000000","message":"Uploaded patch set 33: Patch Set 32 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"b98eb3cfa34e6c7eeb5ad140cd5964519a076fb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 07:02:19.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5be853c7901e485ca9aec50a135b69ba\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d8633ecc2ef443c894981562bf15b114 : SUCCESS in 8m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae87d5b8df564d04b63a55cf382fdfa0 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0cdd398843e34612875235ffaa72e8e7 : SUCCESS in 9m 25s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"c3762d2e642fafcd191896cabb942c59c3388597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 09:02:03.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3680c47123bf4d27ad415d76ecdf2461\n\n- grenade https://zuul.opendev.org/t/openstack/build/d9f09743ebca4569a917fbf363b1b138 : SUCCESS in 1h 00m 12s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3c8f875fe50645848eeae609860d4e61 : SUCCESS in 52m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/05a4394480764b10967470aaf5c7ac4f : SUCCESS in 1h 25m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba952013d69046c5a7d581bfbacbdd66 : SUCCESS in 25m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d5c74288d847e5ba12d5b5afb3ea39 : FAILURE in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b56b4258f3864d38bc6d109882000a0c : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7afb02aaeda44de93a3c314021ece0a : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1e794e1d2145adbc6f62488068c8ad : SUCCESS in 5m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a0a16dbbc1914ab58340534e360f12f9 : FAILURE in 14m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19ce67af4b45461b8f75b21f166a4bab : SUCCESS in 12m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bebf780a3afa441088ccd047e2580537 : SUCCESS in 8m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3bca0625663b4ced91041d9048f405ae : FAILURE in 8m 36s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f7aec2d5ac83458a82de18a00984948d : SUCCESS in 11m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fcb8ccd20eb94bafa6dbeb9891822af3 : SUCCESS in 1h 50m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/61b3d85b94a449ada01e72a888211f77 : SUCCESS in 1h 31m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1074a16861d94a95939c8073f5101e22 : SUCCESS in 1h 58m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/316d88a551be4b5689d734b2bc5d37a8 : SUCCESS in 1h 56m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5493e4132cc0447c8da1c8794689dee1 : SUCCESS in 1h 27m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b697b4f89e4b68966e0faf3977a9f3 : SUCCESS in 49m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5bee5acf5684b7aa10ae22b641b6a01 : SUCCESS in 1h 21m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a2f01606b9244510806bd53e060033f8 : SUCCESS in 27m 26s\n\nWarning:\n  Comments left for invalid file glance/policies/base.py","accounts_in_message":[],"_revision_number":33},{"id":"3b5c17ef5dbf84756c802d63664a3095981a58a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-07 09:14:43.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"4fb4c20c8c9187c1c76f0bc6de4d131ebaf12d90","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 09:26:46.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0876367eddc446b28a083a1fdfb26cf1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/df509292ad764c228790254a6f5eed4f : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ac1343de51774727871f81372dfed13b : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4843b484b54e48dc94d0682845d43783 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"b0d0373c14fb518b7b219083f3fe146364f6845c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 11:33:00.000000000","message":"Patch Set 34: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b0d1e59d2d34f31a37ff9a8e4812f9b\n\n- grenade https://zuul.opendev.org/t/openstack/build/555a9ccc51d5495b83ce9157c9396dc2 : SUCCESS in 1h 13m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/80ccc75e4b9d45cc8a5bc14526bd7201 : SUCCESS in 55m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0f6f9f7eb494bb09ee4ac45407763b0 : SUCCESS in 1h 08m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da9910b5dea14c71b50c27de3bd40831 : SUCCESS in 27m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892ee7a7fe4041b88c0cd7d6782e65dc : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/568cdcf4a1bd4696933ee86fade38b6e : SUCCESS in 6m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d35c5fc84e44466ba1cec9eb72823704 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8801ba95561b4819bdc6636c7677e8e9 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/068df83d3e634217b02d4a9fcad1b135 : SUCCESS in 14m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f3d43410b884d6eaf8bda25756b0821 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e6f39995baa9439d9542e59bb1b3a95b : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/60b2e216b82f4ce59a05959cb20e045b : SUCCESS in 10m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5c357076728846718f890512b70cb4f3 : FAILURE in 12m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/df747f777efc4ffea5d0f044c135fa2a : SUCCESS in 1h 26m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b666e61f249f44c996b80c8cfbfcc5f1 : SUCCESS in 1h 39m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb850e9d11d04fb9ad603b80faa8652d : SUCCESS in 1h 25m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d7e28b590fd74d9eaba0cb80ba2a5d56 : SUCCESS in 1h 49m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5dc97d98e42b426381d14cb2b98fe5ee : FAILURE in 2h 08m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b619bccccb5f457ca580ff0c3b09536e : SUCCESS in 44m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c63b761bf7b64635b93c9b25563718a8 : SUCCESS in 1h 46m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3413dcd8250d4583b42cbefaa3eba4eb : SUCCESS in 26m 13s","accounts_in_message":[],"_revision_number":34},{"id":"c51fd0e59cbda9cb06d936df4a38ec7830769451","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-13 02:39:21.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"dc6f49708a0c1cd5d8d397eca16e82b07f2dcd19","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-15 09:09:39.000000000","message":"Patch Set 34:\n\n(1 comment)","accounts_in_message":[],"_revision_number":34},{"id":"e1d5bc689f802bd73cfe2a7a54ff32b479774705","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-15 10:31:13.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"5076cab7ef70c2ff3cf6134cedc32908fb5da46e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 10:44:18.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50624b2e01d74de69724ca75a46a4714\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a50503cd956040feaa91b32704157201 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d2e69b211eeb4c01bfe780447072c3c8 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cb536040bbab47ccb5c00a63856439a7 : SUCCESS in 10m 24s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"c284f1dfabaed6720814c3ec49dfc310b56590ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 12:46:28.000000000","message":"Patch Set 35: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2db942c93b71415fa05b30b604f98682\n\n- grenade https://zuul.opendev.org/t/openstack/build/8428cd1875d74e5bbc5e1e59fe6501c4 : SUCCESS in 57m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2468b2576e0d4a40a6be85145924170d : SUCCESS in 57m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4d1ecc2cc9e47fea671ed34548ecbee : SUCCESS in 1h 10m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ec20c6b6af64df79aeb61923302e7ce : SUCCESS in 32m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e536b9fc109540d486c7c6dd98ba7a1f : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a676dc256ca745ab9080fa72b2cf3be4 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d02665cacae343758424dd73e960ed37 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05959ddda6424bf8a7062c31a53395a6 : SUCCESS in 7m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a27c80b6ca124897b2af5eaefd922658 : SUCCESS in 14m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6771ae1260d44ffc8d2702ff34f50b15 : FAILURE in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b0271033b7124f1db93ccdc63a3586ea : SUCCESS in 10m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ff0f58cd82734e0db0f9af3d5dcde1e3 : FAILURE in 10m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/06a8d2769818488f89da6365bf85679c : SUCCESS in 13m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/501008bd28b24fc0b00fd2c897c449a3 : TIMED_OUT in 2h 03m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/81030ca831554bf58bac60368d5ceaad : SUCCESS in 1h 34m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/92f1558180c8429fbb6a21d2aaaace1d : SUCCESS in 1h 34m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9342a25f5911445599cc69d4a6fda043 : SUCCESS in 1h 52m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/69cb6c0b678a4f29bbdd4768ee00f25e : SUCCESS in 2h 08m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/428a72d1bb0e489db7d160fb77c75035 : FAILURE in 47m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fe02fdae0f534b15a92ae43d0e0de128 : SUCCESS in 2h 08m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c981b281d4345ee8f178ed40511eeca : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":35},{"id":"e84427f7b28e3f5b3d4cb69a454104eac4981c21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-20 10:23:20.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":36},{"id":"af963fd498dfea1a27aa251a947bbafafe099be3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 10:37:27.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b086fbf1b2f6422cb364fa95a19ea4fa\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/500f05cd8efa40fa8a573d1018da332b : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d47e959ae45c4bed86321c796141d275 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/86ac811eb6d74395bd9cee1e18168ba3 : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"5022b31291ee4c9d86dd43df79761f716b48bff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 12:27:49.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/60f411e51b554a7fad67b3d9065983b8\n\n- grenade https://zuul.opendev.org/t/openstack/build/87a4932bc7b84d6b94619b9efefb71b9 : SUCCESS in 41m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/43dc6a4fa0064d1ca22f35f84f835522 : SUCCESS in 1h 05m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1dd92ee4e7d46689f3a09fbb4fbbfff : SUCCESS in 1h 39m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51cc83191c664506890de64b22711ef5 : SUCCESS in 1h 01m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dab94859d81f4f3bb1174c5a76a949a7 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2c482752db045fe8e424dac5d454ebe : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27d6a97a8c2b482f9c90ca77e175969d : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feeede30854b4aa19fb274f8553784f3 : SUCCESS in 6m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/087a37716f1546c5879072624b17adfa : SUCCESS in 14m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef8bd1310f46445f979bbf1e3d23acd9 : FAILURE in 9m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/94d44a145c5341a68ebe4950e503cea4 : FAILURE in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0727addbd9034bb7b1ca1c98f160ff1f : FAILURE in 10m 07s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/001122bd93d64cadb17e8db35c755b40 : SUCCESS in 11m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fab9cbf660024a84b48fdb65ffb19b7a : SUCCESS in 1h 58m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7472d250f6e14d37b661c07dbad6c236 : SUCCESS in 1h 33m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/74b7cff0b90d473a91e2c35685b4357a : SUCCESS in 1h 28m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/08662a9b22f84bbd95b68f4147375935 : SUCCESS in 1h 27m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/27bd4b048adb42e0a2af1299e0f1f5d8 : SUCCESS in 1h 12m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c74000f20a894994abbf382202bf759f : SUCCESS in 45m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/171a91efe5f84eac994d6923683be9ea : SUCCESS in 1h 44m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7bea5e6586584d32a633598ead58ed9a : SUCCESS in 32m 13s","accounts_in_message":[],"_revision_number":36},{"id":"311dc9d2aecb60fa9c87bdd14f3be4cf6e9d267c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-20 17:01:49.000000000","message":"Patch Set 36: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"59700203736792d9cb720ab85e377091277d3234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-21 12:59:22.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"7ea88b1476a5a746c5089c155ffb1be0e89db42f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 13:08:48.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f998a72b5c4b4f2fac638635d5656ec7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/699ce083e2b14eab8bfa2dae22b9e2ce : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4aedecaaddd746db8c94e8c08ed8631b : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8fbc08e35d8a43a6a6db061a6ee40791 : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"28be5374f38f4eb2e15b7ef10c6ccb03ca970dd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 15:06:52.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d53d1231d90481eb7a4acdd91ec5808\n\n- grenade https://zuul.opendev.org/t/openstack/build/74b36ee10145413da2e6adb6d2154b4c : SUCCESS in 53m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dba1017888d34d9692c3b546315be8fc : SUCCESS in 50m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85fe37ba6ae343b18df498d360f82345 : SUCCESS in 1h 19m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1df9b6035954c1e8f95a0d80611b61f : SUCCESS in 1h 07m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b39f5b8de345d2a9bf966189d89cef : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1562b97ef08e4aaf99d05bcc3adc1575 : SUCCESS in 6m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/13b5346d0cce4026aaa7af4776093485 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02c3e8d6f69f4b0bbd61673af6626489 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1e256c9327d24f65b36ec320ab7cd97f : SUCCESS in 11m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c4f11e2b39dd44c18f4064f4ca1e98b8 : FAILURE in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/547b742187db463e97c3d1727b23e75a : FAILURE in 9m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/20ec579c6da24efd94d5139e6152d85d : FAILURE in 8m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6b4cfb8799334132846e139407a75c2d : FAILURE in 12m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/361c77399f8b43f7aff3dc9c53e28248 : SUCCESS in 1h 49m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cc8134db9d014d25983f29cf72807625 : SUCCESS in 1h 16m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/76147f4ecac74b519bf2fad9e01dfb71 : SUCCESS in 1h 41m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6aa9f276a0a54f9282d0f85c44aabf15 : SUCCESS in 1h 51m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5869add34a1f4464bc9e1e076eeb0a97 : SUCCESS in 1h 58m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d1fe38d20d5472d9e77f4c68710f363 : SUCCESS in 1h 00m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3da64d1f18b415dbcca8c01b57c3c2e : SUCCESS in 1h 54m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/56755db780ba40a0971ac5f06a23ca0d : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":37},{"id":"340327f7ef4fe543ae317412f47a7a9e1d0c1728","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 07:41:38.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"dc0781789b123384c2961af3592b71a95902dcb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 07:50:36.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f09ad42502304cb2b6d74ee9cc1499bf\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f7a18e736ed4de788ac0f6bdf9ac570 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/848d9450581e4f4dab8f41f7c551db01 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/534c54958a57472daaf046db574ad899 : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"bbefbef5fbd5ff6b663089498c3b2b3e6dbaeb3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 09:52:39.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/939af3819c8944a5a264897a70b662ac\n\n- grenade https://zuul.opendev.org/t/openstack/build/60c9cb2b722342f58889676818bfc3c2 : SUCCESS in 55m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1239d65c076a4465914ad4287ef4377f : SUCCESS in 52m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ffd94eb91f4c4c31b5538f9ab2a3dfef : SUCCESS in 1h 26m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b94282185564b18add58ac657fc468d : SUCCESS in 59m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40ded90ff8d6444195c6db7fe6374d86 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a15da53cfb2844b788b151f8667718e7 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f9ba1d7c6cb44f1af73c703b2abecdd : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2f3050a92f48f7b1b6ccc737cc2e06 : SUCCESS in 5m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/504be69c65de407998114cf3ad8aad96 : SUCCESS in 13m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/291090e5067d498cbc340aae6ee1339e : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/11008b8bb1f14b5599e21a3cc9605fd7 : SUCCESS in 8m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/648bd8040f26412a97823afaa22d4c0a : FAILURE in 7m 45s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/243d29fc61044341a67bfa7cd0157ad1 : SUCCESS in 11m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7be09fe3c2204134b48246f94ada2233 : SUCCESS in 1h 44m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a13c304d2d464c5a8110f2d0e6f1ac1c : SUCCESS in 1h 49m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e145f9836f6644f889a75a7a57d404ad : SUCCESS in 1h 36m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6e10c25803c2433ca9d99d96f87183df : SUCCESS in 1h 51m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4ae50d9d6a7c44d18c6e887ae6f20cde : SUCCESS in 1h 58m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a2b9775aa394f63836bd3a21fc3e013 : SUCCESS in 1h 02m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8b38a3fd71a4e6eaabe769b45d37179 : FAILURE in 2h 03m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/37633657fab54ea389f22aef0d57d5d1 : SUCCESS in 25m 19s","accounts_in_message":[],"_revision_number":38},{"id":"78edc889e3272910f9c34b85503d420e27a558d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-22 11:00:32.000000000","message":"Patch Set 39: Patch Set 38 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"25feb12477703d9a591fb1cbd9b21ab6ce65e7ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 11:33:04.000000000","message":"Patch Set 39:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5665c109fc241d082b6bfa02b22494e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2084416e494b4733b51ad12b00fadf36 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c744187c606549afacd0d82b721d3a0d : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/60a45b796b354262946a35c839cb9fc1 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":39},{"id":"1ddbb838372c762ea79b8777d17349284179e4ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-22 13:40:20.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6012ac21b3644ccfb3e0176539dd90a0\n\n- grenade https://zuul.opendev.org/t/openstack/build/e313dfef009e4bfa8d75682bef725999 : SUCCESS in 1h 00m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e5d923b646414029af3b52ead03706f0 : SUCCESS in 54m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7fcb28e1b1d401eb3c34db20aa106da : FAILURE in 1h 09m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2cd2308622094c73a6fc0fabe3c502b3 : SUCCESS in 1h 03m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09d3b4cd64224040a926eea6075c90ce : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80defbf0ba6c48858e875da8b063ad9c : SUCCESS in 6m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d22aa1bf076e41a6a2ad1cb634e039db : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d65a761da9c45f68face9d4a6cfd53a : SUCCESS in 5m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/37f30f5920fa46d2814340645ffd3551 : SUCCESS in 13m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7f232de2d6824d61b082a20f77b56a71 : FAILURE in 10m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18e3029c0a604de48e560b56c295e4d8 : SUCCESS in 8m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4c04204304d44c49b0a1d028a8a843a4 : SUCCESS in 8m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cb9bb9decc0a438e851f254c0c559fe3 : SUCCESS in 13m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cb1b7fde27b441e29b0b726eee0922e1 : TIMED_OUT in 2h 03m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ef2ab1f3ff90427f95fb2bf6991afeb5 : SUCCESS in 1h 46m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d182623a602540b6bae25a301be22e5f : SUCCESS in 1h 49m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/125aad77601146c591abf289906a9c15 : SUCCESS in 1h 47m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/11bbad424b0f4fb0bbbc9bb090408456 : SUCCESS in 1h 57m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b86f65753a954a2186238945213cdbc2 : SUCCESS in 52m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2dece6eb9a2348138eaca859462f1b95 : FAILURE in 1h 49m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f79ae3ffd6134090826ee7d04608eab1 : SUCCESS in 33m 43s","accounts_in_message":[],"_revision_number":39},{"id":"dec4e8b56eac75df010cf748162fb207e21480eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-26 09:16:40.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"70cb1f6db61b6ca8005c9f1efbf6c45ce5076cdf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 09:43:02.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0a451159404a8cb06f10c0bd94efe3\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15f09a90a22747e6b85c64fef3a12ab1 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/29a63d60091145f89c519c2c7e6581dc : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/50d4c29932cb41cfafa84b5e3645a034 : SUCCESS in 8m 46s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"202f248cb494584c6428f4bd07d8f97b11080748","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 11:15:02.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bab4cb9ae11a49509e1d32efaa62e612\n\n- grenade https://zuul.opendev.org/t/openstack/build/b0d9ecf93a0d41b2856ed0a5ce778bb6 : SUCCESS in 1h 12m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/854fb03a29514df998de23033049f5e7 : SUCCESS in 52m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b795fc1e965b4976840df52a899950d7 : SUCCESS in 1h 36m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7d2647c177214a57a79119b1cbdfd3b7 : SUCCESS in 48m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc4b24ec90d74e4e9fa6b407ef8b9e4b : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/261c7386755942d49a0fca345bb2abab : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f18b97e8e3b47dcaea6d01d45dc1bb4 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a08d89feabf4e66b1cd3e3c8ed7d782 : SUCCESS in 5m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3bb15cfc28fe4ad0abbb89a86ffd36ca : SUCCESS in 12m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a716f1d258ba45ffbd787dc4a6e1ea36 : FAILURE in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc769dc4cf9c4a2ea5fde7597db18499 : FAILURE in 8m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3c588d0ab8f14042910bf5ac48fc6a5d : FAILURE in 8m 50s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/896225ad52f048ccb84a124cd3fab0e9 : SUCCESS in 11m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fea24bbebcc0428289f8c20bfac08e74 : SUCCESS in 1h 42m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ff232609f5584a1791c6562a5a57d722 : SUCCESS in 1h 11m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8bbd636f6064f33a0b5b4359e85267e : SUCCESS in 1h 50m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bef658defecb4c819a60ecf8e89ef7c0 : SUCCESS in 1h 26m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fe3461f52c0e47cf9c0fcd847fd2ff95 : SUCCESS in 1h 48m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c39ec737c8da46558b330fd88a3be138 : SUCCESS in 48m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0677ee5246fc4e819e7d0e782285e032 : SUCCESS in 1h 42m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/11ac3e138bfc481ab25aef4f55e25f7a : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":40},{"id":"057205efaef997e26c90e33c38226027dae7ac93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-27 07:12:48.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"ba8300d1fac3c02575bf49432ff602a41494edbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 07:24:45.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9655d9f5a29a428dbe988902831f1c3c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/812ddcc294284eedbf017c5deecd06d3 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ade0c41c7764788904b5ae3c0f0a4bd : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db1027afe17d48779a0d437372de059b : SUCCESS in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"7d20f849dbb4425c65c46e6c1a39781d7dae3e26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 09:44:16.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/056f09ae7c3a43889e6627eab22d5b5c\n\n- grenade https://zuul.opendev.org/t/openstack/build/fffc02f3803d4277a53124b18123a5cd : SUCCESS in 58m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f20667a39c4644278d55eb00ebae895c : SUCCESS in 1h 04m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fcd9addc8cec454cb5a8aa8354a05ae0 : SUCCESS in 1h 24m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bec77f089604670bae93bf167ed6645 : SUCCESS in 1h 03m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3660fd7a94543e38f004258392a7781 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e94074aeb0b94f0c93e424ba4a4a89e4 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/798f6e16446b473c877f7b29b3accee8 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bee3aae489a44e0a428090d273c608f : SUCCESS in 6m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/43c821c49e9541ac9e3647c03471a657 : SUCCESS in 13m 34s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3aaf30ebd154c0ba4195357d2eb0f9f : SUCCESS in 9m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/736b4c11dcb244d5a65adb56e8a34038 : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c4bce98c0a4e4961a9a1aa34113c4dcc : SUCCESS in 8m 25s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e0ffff385d174f3a9525c8e89df93042 : SUCCESS in 12m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2d8233d11b234a978ca61d67d7037d77 : SUCCESS in 1h 40m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b8cac6a478f249a6a328831a8d4c5bf1 : SUCCESS in 1h 38m 53s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/338da1a00cc44f469df5160ac0dcc4d4 : SUCCESS in 1h 53m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d17f9ecc2b747e9b1206dc767f1ce7f : SUCCESS in 1h 53m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d1bb596074174cdea90b3bb6979d64c8 : SUCCESS in 1h 54m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e394891afbf4be6b419be2bfaa6d612 : SUCCESS in 58m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4cb03af156c84e0f8db63f10a0edd41a : SUCCESS in 2h 03m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d3f25891f3c6467baf9116aa4ef795aa : SUCCESS in 25m 45s","accounts_in_message":[],"_revision_number":41},{"id":"63f07930de5f629ebe6fd477d542e1a31426026c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-27 13:25:30.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"d33aa035cf5084d9f8587c96d68189c96f598170","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 13:35:48.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6ce5dcf99fb4e37b19453ddc6bee5e1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d27cae5847d452da06e2d632cc30a52 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/556018ee301e46d3aac586243feed230 : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/16b150851f2e418d9cfeee7dc9ed5004 : SUCCESS in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"58cfb31c76e18a95e330ce0cf57f130cf1f7948e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-27 15:26:26.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":43},{"id":"d3c33398c1a61d3bb09e5761cfa08be4ea96bc46","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 15:39:22.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dabda2788e44978ae2b0c946c2e3e59\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/124591c0ace14a8e9e241d3187076ee4 : SUCCESS in 10m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/46090387a24549058e719a0e78cef279 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dc7d7703dd6141e9be4ced68a8d509c2 : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"f7d4e1b4664e691bce441da48b7cf351b3ffe5a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:40:30.000000000","message":"Patch Set 43: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b9ffbef35014705ae8d16c561eb4917\n\n- grenade https://zuul.opendev.org/t/openstack/build/4086fda22fbd44ae8bef733e7c740df4 : SUCCESS in 1h 14m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c39783e0b4f445489f2f57a568e0658b : SUCCESS in 50m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eac052d129cc419bb73aeec3250d8295 : SUCCESS in 1h 22m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fbbfbba45134850a4d2727facf8c981 : SUCCESS in 1h 05m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8815ce01e2d469ea5e54f7275fb65a2 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a71ee4570e8642ab9bf8344da3d11c9e : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73c2c16c8e404c64bd32b5db7b776503 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e081d1ccf84fa890530239300dca1e : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f0086012f456410ca72afebd9be290f8 : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/187ffc2c5fa44a8da7ef806eea27d2bf : SUCCESS in 11m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9aba5547d9c2432082f9842f33f9e411 : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a8d0d0f4fcf64f09b81caa65c3b58ff1 : SUCCESS in 8m 22s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/291750b8c92642c7af20d300d5463e0e : SUCCESS in 12m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10170cd1d42b4798b99cc49a127e960c : SUCCESS in 1h 50m 47s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/add7b6eae8a747df932734c58dfc3725 : SUCCESS in 1h 27m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a731fcabbd2245d799bd4901ffa125a4 : FAILURE in 1h 42m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/25cdf034421b41ea933ae5d7285cfe8a : SUCCESS in 1h 39m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e20be4679fe14c389f27113efa7148b1 : SUCCESS in 1h 36m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8fba55ab0d145d4bc82755ae2c9f735 : SUCCESS in 45m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/026405fef4b0448799ea7e20ee22cb86 : SUCCESS in 1h 59m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4507b590309b4b0f98af6fc60f0f9dc9 : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":43},{"id":"84153a195eb35e00207a4dad0b7ac97c35f601fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 03:51:38.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":44},{"id":"80f2fb41c584a00ffae53b3d0b6a3285fe75f7a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 04:03:14.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3f51158f0a4b1ca124cff08bbdbb25\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3facfd7ab6eb4480a9050f97c6143316 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/819c0a7ed5ba40c39884251a54e3a7dc : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a7d65e87e7441f8ba7519ddec8004a7 : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"6ea5580df3691bbb2cbeeccd4f20e04153b1c731","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 06:05:18.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331f2022371e4eb19ffef16427b703e0\n\n- grenade https://zuul.opendev.org/t/openstack/build/1f860fce968547ee8acc6efdc3dc8f39 : SUCCESS in 1h 04m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d883200f6bf44bc78b2f44f56f128b34 : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/574d489658554d4aa2851b39df7fea82 : SUCCESS in 1h 27m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3afb912f1d8342c1abde1dee4699fac9 : SUCCESS in 55m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae76ef2806641d98dbd5b4f40eb49cc : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1705841c3e484ed09506046166be4339 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73ac1d437f29413980eabd55d57fe055 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d684cceb8b3547eead3022e96d808232 : SUCCESS in 5m 39s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/25ac7101bb8b4af19af4fa39ca326cd2 : SUCCESS in 13m 09s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3353b04bb5bc4b21943a9d0e7c922a1a : SUCCESS in 10m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d56352c7f8f24671964d924cf8ac9695 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b3d4283bc5b94195983256161fa18bf1 : SUCCESS in 10m 20s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/07a8c56f61814f9096cfc179ac350c69 : SUCCESS in 10m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0d05ba380464456bb15e39db12678d5f : SUCCESS in 1h 40m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/60db5c79f52f49668b5b29ac3f073417 : SUCCESS in 2h 01m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3338c135247741a0b94c8e4f82b7dc14 : SUCCESS in 1h 33m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4785f25c87f7400590b8d00c918a594c : SUCCESS in 1h 23m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5018389e18c14fe3a5b7aed44bc37e55 : SUCCESS in 1h 38m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e613930062d4d3fa6d0dd0a873fd648 : SUCCESS in 58m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b0703a3235a249c0863f364cc6661efc : SUCCESS in 1h 38m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/87ee271feeb24327b6828880d8c6c3e9 : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":44},{"id":"523fa4d013cbb09c9424cbd649e0ca82a1d5dfb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 09:10:56.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"1acec0340260e8033bef59b5f5deaac432de842c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 09:38:28.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7338c21338ab4509b73d66ee318a31d5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b7df09c000b44bf7b68029269275a3b2 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6607cd3f15b41fb911a8a8168ca94f7 : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cf68d7212095408cab3427109b82def7 : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"90c084b53157a0581a3d006be3323eeb1ab42b81","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-28 10:30:50.000000000","message":"Patch Set 45: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"64b3f9d26c2f0128f2fc099701e4d5de78b8f588","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 11:10:35.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":46},{"id":"a606772872e67c34a5043a06dc656e0d65f7f00a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 11:22:31.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc2e7819f3774e46863f4a8bcf873f65\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d6355346edf450685607f1287f442b6 : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/597311eee39349018daf983650350ffe : SUCCESS in 9m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4d0a5a15498d46548ad2ea34a79c1ed0 : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"824edda865b231efaed204cf789c5ef6d6e1b2a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:23:10.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12621483e89b463fb4f746dd5469f25f\n\n- grenade https://zuul.opendev.org/t/openstack/build/83626827aa9b47c2a7d6eca3662e5fbc : SUCCESS in 59m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7ef220c56a914875a922671b9152ee40 : SUCCESS in 1h 08m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2871a3926d184bab88aabb02fd85e936 : SUCCESS in 1h 23m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dde7642ad7614e96bbc03088b820c69f : SUCCESS in 1h 00m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94804681abdf4c3fb47d0efcd6b7eac9 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3cfd4f17676458c83dfa86d84ed405c : SUCCESS in 6m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b10161e1c714e7f99dee48299e7d791 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90bd52c5a6014f90aeea88074d8091b5 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/eafaff03dad84c7a9dc5fab3660c5257 : SUCCESS in 13m 25s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b9194eb213384ec685fc6104d17c9a5f : SUCCESS in 10m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/53445d717ac3443bb130be0bb33af16d : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/505a6eccc71c4ea586efcfb8719b475f : SUCCESS in 9m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d73e9c49869244f18d54d72cf1cbb1cb : SUCCESS in 13m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/89a3b9597d344c8480c6e976c54cfe33 : TIMED_OUT in 2h 03m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b4a9545a494841c0938733ed7bb27422 : SUCCESS in 1h 44m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ed1b5eb4cf5e4433ad93ba7b826a93c6 : SUCCESS in 1h 52m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2024d25f64954dca823301a2d0e59cb5 : SUCCESS in 1h 34m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2e9d5b5904014002b045cf8dbab50264 : SUCCESS in 1h 42m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f968e1849a04b0d95dda299afd62011 : SUCCESS in 1h 07m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/367adfd293cb4b4995bd15738c93875f : SUCCESS in 1h 33m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/35c6a10e49fe4f7b9b32bd9f20304536 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":46},{"id":"d599da23b4d5188d213009a4c43b3677631e3734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 16:14:46.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":47},{"id":"bf715a64906076ba26995e29be9298e7f548dac2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 16:26:14.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d5fa6f9558b444b839115b9980f56f6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ef61091d0c440db91e7f70e6fa59242 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/22ad744c22e14e47afdae87bc03ddd11 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/39e5c286756f4408b8e27ade3e28a738 : SUCCESS in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"e44a6aa8a07b6e886feead266a093ca7695c8725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 18:28:23.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cad0252746d4ce7aac56cc2b8b6a825\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c17682fb8cb46939103aae962b0bcd3 : SUCCESS in 56m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8ae948f896c1412d9fce616afe62e008 : SUCCESS in 1h 31m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67eb71394bd54f019e7465715a15d5a2 : SUCCESS in 1h 16m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b483989b98564cf5b31417e164043230 : SUCCESS in 56m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09724a2e950849a0a96b3063a475d5a4 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe3a4966aa0340f8b7c0f2c4cd5f63d8 : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/08d2796add82492989cf808d8ac6b6e2 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a192ee07dbc4a85a175d89849792434 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/861a95a7f00b44018a7ecc8896ba2ef2 : SUCCESS in 13m 41s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bd31d13ce8f42868fbe6e73874f692a : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/43ef4a0f2ee549dcb657f03ffdf72aad : SUCCESS in 9m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/287e3d53d4464e80be835d7971ff9d78 : SUCCESS in 8m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f63ebbaf41f0442b97eeb9f7882709e1 : SUCCESS in 14m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6e9b91e1d02f4e70be04d4fe5aa31334 : TIMED_OUT in 2h 04m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b931044c7a424b359b9e62ae535172ad : SUCCESS in 1h 35m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0ce59de6cd254f578b7a70878d586abd : SUCCESS in 1h 31m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/31bcb46129a44fe792999f6c7a609cf5 : SUCCESS in 1h 50m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3fa7a5488ed14352af5fc698ec0d7419 : FAILURE in 1h 53m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f02027ddd544083b752442037bceec7 : SUCCESS in 49m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ec524eb41994828be82287d7436ec1c : SUCCESS in 1h 49m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/280cd3389caf48da97349b73730315b1 : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":47},{"id":"8a44fff95d0b9d8cd6abf4c3703bb89a18b0b950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 19:01:34.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":48},{"id":"4600df240ba7cdd116f616475238844be83c13f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:12:41.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59fd2538b28544ed843cca8d7355840b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f463ae6260d04385b4ae81cc109db7c2 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a9d6192b5fd5468b9d08f94f1c6296b9 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1a786bba1ffd45c2b6e7362c027d8229 : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"0854a9a2a5eb0653cd6d9f2a41f2450d843b8cbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-28 19:21:23.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":49},{"id":"42f3e7faea1b0999fe12dbc19d0048e7cc9f3732","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 19:31:48.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9de3f731c37f45dda4f9aa53cff48128\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/69834904cdc84d5d8f166a618217fab4 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f857108471c4d4aaa023ccac65c0533 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dfd689a726554b958be3b2b61e7ee95b : SUCCESS in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":49},{"id":"d82945afc077ba7935cc29d6b8249ecc5a9a8344","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 21:38:49.000000000","message":"Patch Set 49: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c35895c088c47ee806d7431e161968f\n\n- grenade https://zuul.opendev.org/t/openstack/build/f0a93d6851d04319a66c9cf90f8d386f : SUCCESS in 1h 08m 04s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d86b404a34ee4c0eb67618f3d5241863 : SUCCESS in 1h 02m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3378939687e445fb9a4fc7fb44ce384a : SUCCESS in 1h 51m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4efb39f8f458439b8fdb7cceefa87960 : SUCCESS in 55m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5ce8c5b26ba4be096aa6fa105a71713 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70eeb3cf24084c4b91f13c84b6dbf982 : SUCCESS in 7m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5252a5eade544f008e6fc6e021ba9be1 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b5e54926cfd4597afaaf01903fda9e5 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/73cb719622074dcca7271e2510ef6d7d : SUCCESS in 12m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e68e704dbff0467faf5117a39205d00b : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f54fd76d896407e907928965c6f2297 : SUCCESS in 10m 59s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1ebfaed36c6b4e64bcbfa8dcbd58eeaa : SUCCESS in 8m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3e68cf4ee93843c49336eb5eb39d9b33 : SUCCESS in 11m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a0793193a5b74b42930d9b99db65f504 : SUCCESS in 1h 53m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cebdf5a1aeed4781a4980a89afefef01 : SUCCESS in 1h 30m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/57425eb09cfd4c9c89cf94e163124faa : SUCCESS in 1h 38m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/305726bb7d7448e68e515c60dd7b10e9 : SUCCESS in 1h 57m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1f76dc07796b40ecb075113d8f95312e : FAILURE in 2h 08m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c88fe78edf274af2bef1f9af123998f4 : SUCCESS in 1h 04m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1111ae15f5c4ced8e710c1c8cfb0094 : SUCCESS in 1h 40m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/626a984d27204586926ebae0ad1a02f5 : SUCCESS in 27m 39s","accounts_in_message":[],"_revision_number":49},{"id":"f7147df1fd7739dc00a613bddab94fed27fbfb7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-29 10:17:30.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":50},{"id":"eb9a6e237ff6f75676d08074bacf3cbcb312b62b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 10:29:29.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/442f47f273ec470e8e51c14868b5ea91\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7361b4e6d7e84eb4a6d69ec2d2f24d89 : SUCCESS in 8m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7e8bf37d27764a949d903bb679530d3b : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2468edb47b654c649cbd836c4aa915e1 : SUCCESS in 9m 19s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"fc354da47d93fef60b78c3df49f5d7924d06e9da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-29 11:03:29.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":51},{"id":"13ca35d860a0bcd8116c1dd4e5679fb9602cfbfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 11:16:59.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69a8ca7824eb4093b7e45802d8dfcd72\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5c35a22bdae4fd6a017274ddea593db : SUCCESS in 11m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e8aa8112fbf446839d5d24852dc8999f : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9df9594552c84621aafa073ce8df6169 : SUCCESS in 7m 48s (non-voting)","accounts_in_message":[],"_revision_number":51},{"id":"1741b4bb600224a3af9e665766f2a53efc76bcd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-29 13:19:57.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5d328d6f1b442b9b21ac774f0dc5bb6\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6a6535bfaaa4823a52461ab692247be : SUCCESS in 1h 06m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6b6c4b60665042cca0b754a563a82f92 : SUCCESS in 1h 01m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0d58dc9a06b944b4bbc2bbac0cc6bc95 : SUCCESS in 1h 18m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79b20b8e80894da7bc0ad7bc3d1f304c : SUCCESS in 1h 00m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c5386f92a8e4a96a824499cd468a912 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/633bd8aa3bbb4656bceb76a249f0a732 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/07e91c44276348b19ffb7c2f043ef414 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7544298a5584a20bfb84b77b4c6d459 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bd9d27f6dc174b278238f727bc028494 : SUCCESS in 13m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/796e1013b1ba481ca0be0c54d8312e57 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eecede14842541fdb2a667b5ba82ec14 : SUCCESS in 8m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/78cc475bb1b2492aaf5e5588fc812160 : SUCCESS in 8m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2f0b2a66c0ab4a698889247646b37e73 : SUCCESS in 10m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/025bd7c9b1884976acd7843f89b48806 : TIMED_OUT in 2h 04m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3f83d90a18c94620b398a8669ac98df8 : SUCCESS in 1h 54m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e7b82a354e0644e19b98d707e016325d : SUCCESS in 1h 42m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2fcb51e63cb1420cb4f980652f3efb87 : SUCCESS in 1h 57m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/784d0d80673c46bbba2571e5bd2473a2 : SUCCESS in 1h 51m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ba1e05438764bb388117363f35e2727 : SUCCESS in 57m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3b1a918187f41f59de8fded4af95a6b : SUCCESS in 1h 44m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/be133719ac3b4360b82c20aa4599d24f : SUCCESS in 33m 44s","accounts_in_message":[],"_revision_number":51},{"id":"a07731d02f13599d233f03c7fdd95617f79e939c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-04 07:00:07.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":52},{"id":"8ca3f0afb6ab0d13113f6c18dc29789926587745","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 07:13:15.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/990c869dbc60405bb31b6e4f9adefe87\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca5f8696629b491eaa07da96acdff7de : SUCCESS in 10m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e896b06bb5a044abb0e567a5b9f7e319 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"4496e66fa9c2ee508c4ae2f1f2a2cb6c55713a31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-04 09:24:46.000000000","message":"Patch Set 52: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f6104cc857d4b77af30a617eeb0708f\n\n- grenade https://zuul.opendev.org/t/openstack/build/d456688a4f4b40c084953fa3375eb904 : SUCCESS in 53m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a9d38ef64cb04e24b6069b5c7b17c0f4 : SUCCESS in 1h 36m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/facd15f8ad8f40beafa96c816efe5c62 : SUCCESS in 1h 03m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8f39d10269d450a8b13f3af8ab28643 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7d36e996bf649dca1d012bf08e5fe64 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bc6400a037234f66be1d553e33c312d1 : SUCCESS in 6m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8924cc9bf2054b54b82bfe0ce290be71 : FAILURE in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69912346707b47b79d24625547cff1c6 : SUCCESS in 5m 34s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c80670d56bbc42c6abf907c6e88f4bf7 : SUCCESS in 16m 14s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f70f7caee6a1493a82c3913598afe167 : FAILURE in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b55f67ccaf649689e0ca6cf20d3f5b5 : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9f4403b1524446e08e6b3a5fc63f4992 : SUCCESS in 10m 34s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/af06989e661e40ef97e3e91a3277b329 : SUCCESS in 10m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/be606a584eb74f2596d6f634aa34b7c5 : SUCCESS in 1h 52m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0b3c046eee404f2aba5d2a1f4dd0b57a : SUCCESS in 1h 30m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9e1f8f764b714d7a803b96a012a16ac1 : FAILURE in 1h 54m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e130992d23304726b992690524a59bb9 : SUCCESS in 1h 14m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b518c714bbe24334ac76f941c1dda850 : SUCCESS in 2h 08m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbb607905451479b84618468bdcbb335 : SUCCESS in 1h 01m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/380f6df171c84d88a369e6651d64324f : SUCCESS in 2h 12m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e3403e78a0d34e2d9b0085c5ef528667 : SUCCESS in 1h 47m 09s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f6246e3603124fed93ca43a60bb27595 : SUCCESS in 1h 30m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/519db7d4a58746c796ecf8bc7b0aa9d3 : SUCCESS in 37m 11s","accounts_in_message":[],"_revision_number":52},{"id":"e2c55b71459a38b27bc148b7969774a57d1c3122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-08 06:31:24.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":53},{"id":"9f429bc3c3e62858677f7b99de35ef32c861f57c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 06:40:02.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95280bf472d241cb89b978f3b8cff0ca\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1c44a4bc721748f5aff9bbea7c289024 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6331dd9c0c564054b844ce354e8530e0 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"20c2747bd7a985373b8b09f4a2507bf7fa876af8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-08 08:37:33.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f173265b7b5c49eab9269d1a28fb35ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/1bb6fd10307b4be68fe2d41d75e855b6 : SUCCESS in 53m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e15026fdacff413e859bed198be44679 : SUCCESS in 1h 14m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/619ae63367f54fbab414924264a8e21e : SUCCESS in 1h 01m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae010c49d15e49fb99b7987898885c3a : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cf11e158ae2410f97be95ccf8629807 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/280b7f0cf126407a851910d11d721679 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c9f6b9ceb044d539cad69dcfa2e2840 : FAILURE in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b120b2333c5447b7b8895e3d19c8e94a : SUCCESS in 5m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/bc625104643842e4993f3e67f6ff4b33 : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dfd48d9bf88746f9a772dfd97687ed64 : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4639a53812d44aada061c9306222ad94 : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b632b76692364552a26b193bbf412186 : SUCCESS in 8m 23s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/fb026da92b664ee0b6463839618a03ad : SUCCESS in 10m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1c57df961a2a42098a3ef1cd3813c931 : SUCCESS in 1h 30m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2754bddefafd4af7aaa0ff2865beb053 : SUCCESS in 1h 43m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e93ce80e2fbe417aa7811ed79e59667e : SUCCESS in 1h 34m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/19d24209bdb44381af23eb6e54af1712 : SUCCESS in 1h 52m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/443e38f0537c4c3294715913baf14e8b : SUCCESS in 1h 39m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f7b212fe03642aeaac7f124fbf97b3f : SUCCESS in 49m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e34b9db25d24c5ab0df80f95e055eb3 : SUCCESS in 1h 55m 21s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2ac64e9cb4e04a7dbce7fff46967602f : SUCCESS in 1h 41m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2c3744d4656c4157907aed9297a71cc5 : SUCCESS in 54m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/955880c6d7ee40318d27ceb601c047af : SUCCESS in 36m 09s","accounts_in_message":[],"_revision_number":53},{"id":"e32a324db45fc87d4f36129a7256a6f23a627f94","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-18 12:06:48.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"ecf82f4f6c3ed071ae855d12819d6bd388afef1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-29 08:04:05.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":54},{"id":"7a933f0b470a2103521dde72694682afdc3ea36a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 08:20:37.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d45498d330643ce8cb3be2cfc16a699\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/807a64edc7d34497a77ba5a8a5fc642f : SUCCESS in 11m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c567d63c7ad74529ad0510201ebd50c4 : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"7d08a6acfc5f3c3f154ac04434c2df8d30bc1190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:35:45.000000000","message":"Patch Set 54: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3333ca127dbc45d5ae7ec5f0bbf9a47a\n\n- grenade https://zuul.opendev.org/t/openstack/build/af4aff941c26442aaae2f5c6cf793ae0 : SUCCESS in 1h 06m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dd8c3b8778e948f1856c73416f5638b5 : TIMED_OUT in 2h 04m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/353e3db00b9948599a3e651495775039 : SUCCESS in 53m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31cea0cf531b4df59373762f3f994638 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cf85dff4d3d4de3ac7498a3eaa5dcbd : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4838ca9222cf4ac29ce195bbd85a6568 : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9793885b4f34839bd50503ba9215fcf : FAILURE in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1ad938701e84537baca0443bb1609be : SUCCESS in 5m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/36e5b748c6ff45928e9ae440d93e9066 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/315eef3123cf43f2a73555645d68a415 : SUCCESS in 9m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9df99f311121425a8930383e31f721d9 : SUCCESS in 10m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c99c6fe11a47477a9e17afd497352016 : SUCCESS in 8m 57s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8211bdcd08eb4d3a832f7d0845913015 : SUCCESS in 11m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/94b2ec506d0a418a9476dd0f5c51ddec : SUCCESS in 1h 54m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b837461be89e408e8db0ca4c7c03b934 : SUCCESS in 1h 46m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a34dbb4d93e44d7c95e3f7bd3238ea01 : SUCCESS in 1h 45m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/afa725c5b6db4dc4a38ff03a3880c7cd : SUCCESS in 1h 49m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/db26647f9f2547dba966515a9cac389c : SUCCESS in 1h 48m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9444df703c3d4ee18bfe3119f47b37c0 : SUCCESS in 1h 01m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3261da2b384e403bbe4d2c64fc8eb6ad : SUCCESS in 1h 48m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/418957b2e9bc443c8882e779ee382cb9 : SUCCESS in 1h 34m 10s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/30cdf4446cc44ddcaac0995ac70d5c22 : SUCCESS in 1h 11m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/675f3323498741d98dca37e151cb1616 : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":54},{"id":"b4025aec0f473272f469f701138bb41fae2b4cf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-04-30 09:42:34.000000000","message":"Uploaded patch set 55: Patch Set 54 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":55},{"id":"752bc4db74c49c103866f157f6d61a94a8f21ee6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:53:24.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3160bb2e074b2c82f253c57977d5c7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43dcb409cda04fe99a32ffb6cfd557b3 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3739fe7b5029460abde6b28869cc6ea7 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"9cea8ccc52f6932221086c231f050fec128e52f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:55:48.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e0d9d4ad1df448e8cad8f9ca6e51d38\n\n- grenade https://zuul.opendev.org/t/openstack/build/68952c398dbd42188ea561324d7ab171 : SUCCESS in 1h 01m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8ad3b5788af4d1f8307290342d6d4fb : SUCCESS in 1h 38m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c8673fda7b844ec8a4c6187db8c860af : SUCCESS in 47m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aba84566a0146298f9ed36c53fc65b0 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1adc78ecd78349ddad1e94ef8068adbe : SUCCESS in 6m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6da4c488cc94d57850def760543afd5 : SUCCESS in 5m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52a6703528d34ebcbeb19bae0aa1c9de : FAILURE in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/649532c42a664125a64381a688fc8993 : SUCCESS in 6m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5bcf4fb8272e4fee9cb3b09e066ee431 : SUCCESS in 13m 21s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1bf688286be34656bf39d79e0d9da5fd : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69a03b9c1ab64ddab6aac9f8577ad376 : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d439487cd3f34930845a7d6cc45ccc4a : SUCCESS in 9m 28s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0ca398ddb7f54de1a6ad73dde35021c2 : SUCCESS in 11m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a4cfd6c6d36747eabcf66a7c27ed27df : TIMED_OUT in 2h 03m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b0637c7ce054110b6bffa8a1622805a : SUCCESS in 1h 35m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b95ed7cdd653481ebde59371a7dbeec7 : TIMED_OUT in 2h 04m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4845519cccf24332a43f3e9eb70ffe21 : SUCCESS in 1h 51m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/88e3824bfbcb402ebb17470603b30975 : SUCCESS in 2h 05m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86824c35e54949849680cc1925b24f3f : SUCCESS in 58m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/025f39bd73f146fe8caf6c4d79dff528 : SUCCESS in 1h 44m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/04bd0b52f6b74d5090bfd576f4763f7d : SUCCESS in 1h 51m 21s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4d49b7b284f64d1596dca9f6ff626120 : SUCCESS in 1h 11m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b1802f7f3ab742a1a965d4abac1680e5 : FAILURE in 37m 59s","accounts_in_message":[],"_revision_number":55},{"id":"3028d0b431c2a93c9729eb2a1c52b05646542b87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 06:22:18.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":56},{"id":"dfc8524adc3b6d09e5cc5e1accb35198dcc04cd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 06:33:35.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/800987c7bf4247848db006beb118320b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d94ec6c969094c23a08f1327ee2a861e : SUCCESS in 8m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6f3a90aa3cfd4382976236012eca89d1 : SUCCESS in 5m 58s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"0ca9c3e32ae75bcb8ab5c4f9ca98811a8019b2fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 08:43:54.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/153cade1eb3a4dcda4bbcf7c7efe45a0\n\n- grenade https://zuul.opendev.org/t/openstack/build/639442d6a8bc4a3db63f9eb7c39330bc : SUCCESS in 57m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aff6cc81ba864d4eb255e9e42de1885a : SUCCESS in 1h 06m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22ff8e7bef8942a88858219ca78354e1 : SUCCESS in 1h 02m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72d7b69dcb4a44ed9ac42dcb685d5422 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bed2bd33a404b5192ecc9dfbad366de : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/669c1c80e63d423abcf530217bdf1ccb : SUCCESS in 5m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcf5642d7c814e658977d1fc6a93930c : FAILURE in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ba436e57fb4da3a29931a095e4815a : SUCCESS in 5m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f4847fd68caf4388b8b0045684d49063 : SUCCESS in 13m 03s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42ec371cafa9471196d5fd51d3a0616f : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3083013ab5904ca0bd7efd51a2b93099 : SUCCESS in 8m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e7927ded72ae4cb48f043c74cb93c169 : SUCCESS in 8m 50s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/89a576be0bab4146a6c62a6f1bb760ec : SUCCESS in 10m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b37631b12c07452190c752601871544c : TIMED_OUT in 2h 03m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b0d6d61ab69e4a0381fd08ec20269c6c : SUCCESS in 1h 50m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4deb485ce3e94cf580805003f9627ecd : SUCCESS in 1h 40m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/efbb595ccf334367993180aafe31b787 : SUCCESS in 1h 39m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d7d0ebaca9a48bb87c4d708a834c0d4 : FAILURE in 2h 01m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c53360800a18448cb57c77a936f11892 : SUCCESS in 55m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7718e96255f0412282afcc4feb315a06 : SUCCESS in 1h 26m 42s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/064588d7ef2648488fdc76a452fb8dd6 : SUCCESS in 1h 45m 36s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/84a0cfd2e564463389386e45e786c3b7 : SUCCESS in 1h 09m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7cd66a43a9084521b4a5c2fea0a80ea5 : SUCCESS in 30m 19s","accounts_in_message":[],"_revision_number":56},{"id":"08224ebcc903922878e83b45e9171bca6200bf0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-07 13:08:53.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":57},{"id":"a4fb819c79f60a15525ca053770038c25d10cac6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 13:20:43.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd7bb0d56424ba7a27803b053b58fd2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3133cd23891458a879ade38aaee109e : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/015fcc76d6f445ab8332a3c43763ba28 : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"39daca1e40758f314b5bbd809d2638ca5fb4b24f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 15:27:58.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7608ab4a18d14d50a3c784a0e34f8f4f\n\n- grenade https://zuul.opendev.org/t/openstack/build/00dd2347bed94674b8e08d1479842dda : SUCCESS in 58m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0def34a6564d4c28bec2174f2d32d396 : SUCCESS in 1h 42m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58c0f7332ef44edba0cbd1779496138c : SUCCESS in 1h 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/363b14126a2248d0aff814cd20f75bd0 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e2388ee1b01471c8095c949d900fd60 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a0e3584d439d4f569c49441dab4491b5 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f54f6821304c43189a69af8459f6f361 : FAILURE in 11m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e84cdf752f764732b88b7c2f6ef5881b : SUCCESS in 5m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/539056370c1b4062a25c3671eeb09df2 : SUCCESS in 13m 01s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c8159bdd86a74b3c911aeaa35a7fa24d : SUCCESS in 10m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85eef3b7960d4f609073000b4bfb47f7 : SUCCESS in 8m 40s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5cb5b8907cb449078ba8bb7cee4dc96b : SUCCESS in 9m 23s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2d02ed38484a4ec79df81738c6c40d5b : SUCCESS in 11m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/67882c960c80422088761bf86863e640 : POST_FAILURE in 37m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0a7f31371cec4467bd16e9ac8f6eb385 : SUCCESS in 1h 53m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1438e78b24c1411eb5facd8c29b2ffe3 : SUCCESS in 1h 44m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bab918ad892c4a05903e849b91e82454 : SUCCESS in 1h 39m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9ab3b120de874a059efccfb18a01b863 : SUCCESS in 2h 09m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ad50e6aa184462d854d56aebd8fb453 : SUCCESS in 1h 06m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/08e91db991034d33b3bee8465f826e97 : SUCCESS in 2h 00m 36s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b13091419ae74e3a8d536d266290950c : SUCCESS in 1h 40m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1c27f5952bc34ff5936c710ba878cf7c : SUCCESS in 1h 15m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/89dac79bcbe74288af77015dd21baaaf : SUCCESS in 39m 46s","accounts_in_message":[],"_revision_number":57},{"id":"3efb4acc02435c96b9cf13b172e63fefea2883af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-05-08 09:01:11.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":58},{"id":"ba7ca7c42b6c42421af2d3ec67c52669404ac8b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 09:15:04.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5be814ab0a4b239865c44b960ec653\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/012249fce1594dcb9e08d7faee408bd9 : SUCCESS in 11m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4f2f4a594d1b48daaa621eb32754598d : SUCCESS in 5m 44s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"365b8f815ace9f0d46b4a04a1454f04a41bc1391","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 11:22:20.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ce61e4e6c184b32846d2f3d20c9e08c\n\n- grenade https://zuul.opendev.org/t/openstack/build/b088f5825e224dc1a05196f093194240 : SUCCESS in 1h 03m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1a55b25373b44e70a7938c6ed2f9f1a2 : SUCCESS in 1h 19m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cc7f9f0e302443da87d9a3350c741d4 : SUCCESS in 57m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75445733bf4145a3aad240b57176228e : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7261b85b78d743a4a96f5180b2eef38b : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18888871029b4eea98301e3e2e0bd1bc : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3851071e6bb4d6e8a7687395efa6ef6 : FAILURE in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/624ced4b087b47a3af00ece08a6bbb21 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ff612ab0c43941d79ca0f36e365ea2c0 : SUCCESS in 14m 33s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d491e0e0f3804b7f96b9e408d740384f : SUCCESS in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c9f18a5893d4f71a6de9fd702f557c1 : SUCCESS in 9m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1c48e49f4b4949b3ba83933f799547e6 : SUCCESS in 9m 52s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8e31234945be442b9dff85f216633f29 : SUCCESS in 10m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9de44126f9bf4fe0ab97bed194f956c0 : TIMED_OUT in 2h 03m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d548ff96ce394f7892410a4bf314a28e : SUCCESS in 1h 41m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/24a3d144ffec4cbdb58079ee637d92cb : SUCCESS in 1h 58m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b4945c2dd861497ab060140422950e77 : SUCCESS in 1h 44m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/10acac54968044b9959bbed951957b91 : SUCCESS in 2h 04m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68067f812a3a4631865c36f3ff28234a : SUCCESS in 45m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa21cd38ef0b46d896cedf3b59739c59 : SUCCESS in 2h 08m 27s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6b46c3dd2c93422cb2c3fa3790bcd7a0 : SUCCESS in 1h 51m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2bfe6b8f7da1435689948b1df60619a2 : SUCCESS in 1h 03m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2e150880a4774ce8bc1be454d3ae6752 : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":58},{"id":"7c69833df6fda0a2d23bd3806c2e17265f4dcf86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-06 06:46:59.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":59},{"id":"bc9ad91455ce4589de5eda370bf1cef5ffd3f055","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 06:54:59.000000000","message":"Patch Set 59:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee70256e67584e30be715b8b441959d5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c395f3603f5847289730b419418ba403 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b3d39c443a745b4926f575e88937df4 : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"61508462c6311124be6428f1b278ca9169399c9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 09:09:48.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5644572308d141c383a23d16a25f9fc5\n\n- grenade https://zuul.opendev.org/t/openstack/build/4478c8a0bf294125bd4a0b19986d7a15 : SUCCESS in 1h 05m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c564e7901d1440fae828229925b7e56 : SUCCESS in 1h 36m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/10729175cc004646bd029079a8106e94 : SUCCESS in 1h 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e352d4bf014c73a413441cbd3b49fb : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02bcfbf908de47f4a38ff275c8df6430 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/125c642c097744a089d4cf35049ff5df : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cb16608b4be4e6cb5db5e4a4820b664 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aa933574cbc46afbe59e88507e40264 : SUCCESS in 6m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9e4c276eb4934f5d82e619c32f10751a : RETRY_LIMIT in 1m 46s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/876cd4c07252400daf2edb38ac69c3a9 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c0f08b6fbe34e99b0cea48629555674 : SUCCESS in 7m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c16fb5d5283c42be98fa107fcaf4d892 : SUCCESS in 14m 11s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f758278230f94efc927278ef5cc7179f : SUCCESS in 8m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d00019b56672429c84149cfb9b95c90a : TIMED_OUT in 2h 03m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e4e09d8bd80c463bac8a44c4e618557c : SUCCESS in 2h 01m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/24401158770e4a9a9df36a8bc205e4d0 : SUCCESS in 1h 48m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83ffbb7352b64d009d6095a4690bdc2e : SUCCESS in 2h 00m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d869c58109614668991139591f3da0fd : SUCCESS in 1h 57m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bbc1eab96f6f4275ae2a614e9b862cc0 : SUCCESS in 1h 03m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fc591a8235304c478e2e9691390f1785 : SUCCESS in 2h 07m 27s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/784909b8bb8a447fa54f163718f771bf : TIMED_OUT in 2h 04m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/baa40bfee5994c9baec83754c6f6c459 : SUCCESS in 1h 12m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3081e391022c4f4cbaaa9b4b26cae394 : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":59},{"id":"0f3efe8439f3034ae1375affb2962bba59085689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-14 10:35:46.000000000","message":"Uploaded patch set 60: Patch Set 59 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":60},{"id":"d28f0805f6ffd06195bc17e8fa1e55349bb56108","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 10:47:53.000000000","message":"Patch Set 60:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f13197ba3945ae8b76bc0ae64691c8\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5509565ca9ba4dacb107e145577ca63d : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/920a02ce80f94b87a851a110d8c78994 : SUCCESS in 9m 34s (non-voting)","accounts_in_message":[],"_revision_number":60},{"id":"f3a77cf0b63d1398269814b82ebfc87e6c39f11f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 12:53:17.000000000","message":"Patch Set 60: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/043c834d66854aebbcbcf91f6253d973\n\n- grenade https://zuul.opendev.org/t/openstack/build/2051b2211aef46ef8129dafc00f6603f : SUCCESS in 1h 05m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b217ef507fae48859e3a02ce9eced91c : SUCCESS in 1h 29m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13abab5fb76a4b228a410daa3316740f : SUCCESS in 1h 04m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/421a8c8e2a734607afa795a786eb337f : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1f43453832146c3a9c37114a794925d : SUCCESS in 5m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a6960390686b4f238e89971a988efce7 : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b84bbdc7f94243b3856ef0c64dbcd5d7 : SUCCESS in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09884b83e11b46abaa49ff0643355e08 : SUCCESS in 5m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c71568ffdae24d4bb8bc1a363cb34665 : RETRY_LIMIT in 2m 04s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/15415b6bbdb24575bedfc609e7fd6e35 : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/214102a282f94672be3faae92ae0cc45 : SUCCESS in 9m 20s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/13717bfd096f48aca3fb6025861d1ac2 : SUCCESS in 12m 13s (non-voting)\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/68e24c770d6c46ceb3a829dbadac0002 : SUCCESS in 9m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1a5093768295408dba83c4830cef3579 : TIMED_OUT in 2h 03m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a60d683779404c7db24c547950d3f489 : SUCCESS in 1h 49m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d426a8683a5b42feb7e79dc6d2f905b1 : SUCCESS in 1h 59m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/41db221a3a174b7c92aed59a2e7f8ba6 : SUCCESS in 1h 54m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0e7a73fe41cd4e1faffdf72db6ffe507 : SUCCESS in 2h 09m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00ece14f260840fa84b20e0cc45ae652 : SUCCESS in 1h 02m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a0e5a3d92bb4155be04426460c848ee : SUCCESS in 2h 01m 17s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/bf1cd1cfb60e4a8381a7f751ada36677 : SUCCESS in 1h 57m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9b388e61b8834921bdb3e84e2b7bbe67 : SUCCESS in 1h 06m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b3bbd862b464c3aaf77633d345ae909 : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":60},{"id":"266a2d3a83c0880845826b27f3121ab78e414d11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-06-28 12:46:31.000000000","message":"Uploaded patch set 61: Patch Set 60 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":61},{"id":"dab9ead0e1b8ab9c51faca668881a5e495751178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 13:07:43.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad9be9cfc444267abf693a5824a4362\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ee99cecf482488abb3d2f518141a62c : SUCCESS in 14m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44eba09acf7d4397aebbb1e27a37a4cf : SUCCESS in 10m 40s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"1c1fd17274dc9a77cfb0d89fe064d545b534c97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 14:52:09.000000000","message":"Patch Set 61: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/162b3e3c6a1b4eb3a03a5a68a2ad3885\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c250326977b4a6c9d3cd8aadd935869 : SUCCESS in 1h 01m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec55582a241648589e0eca740b07cc51 : SUCCESS in 1h 09m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae2c27de68ee429cbbf446048e098004 : SUCCESS in 58m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ceecad365064ebbb08f3d8fde2936a3 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff7e6c80438b4e67966ce5a05c58baa6 : SUCCESS in 7m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f68e4d92011847f4897d08d837e4d0b5 : SUCCESS in 5m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dac0cd3c69da47aca7ee72e31304cf4c : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a1ea97e5404237a0011d3f2ff1cd34 : SUCCESS in 5m 51s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/16e0ba8f07e942a8b42f59b8efe082c5 : RETRY_LIMIT in 1m 49s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f48e57483ea4d919f9c89a4134743be : SUCCESS in 10m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/be04c8b6b6d0490f9d2ac9e2ecc5aebe : SUCCESS in 8m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/88774d4a8d0f40e380fac3199c5507fd : SUCCESS in 12m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7b00c65d60d942b58aab988e64ee5801 : SUCCESS in 8m 36s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5959527244774170b8e89180e1386090 : SUCCESS in 1h 54m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5c93d1c1670942fd858659d2845d8bc5 : SUCCESS in 1h 49m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/452528ac852b48a89a215f58c5ba0b9d : SUCCESS in 1h 59m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6c54e4a0028041a1a11d49dc11a58216 : SUCCESS in 1h 56m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d9bda47ac7b546ab9c3eef87ccef37c6 : SUCCESS in 1h 59m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc14351610de4239b23db8bcc481127a : SUCCESS in 1h 02m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf7be82844f14be68d5e85f74a0ea4ef : SUCCESS in 1h 52m 00s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ca19624a795e4353b4c2c6949ac99b11 : SUCCESS in 1h 31m 10s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cfb2256a0a844796a84e082d1a581a2b : SUCCESS in 1h 12m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ab98ee3b590b4288b89486d89c1fd08f : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":61},{"id":"1e08e693b72b14f8a52d1e99f33b2c629a18a6e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-16 10:22:56.000000000","message":"Uploaded patch set 62: Patch Set 61 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":62},{"id":"b014777214d03e9e5d40037ec702cd3b9f85882a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 10:33:34.000000000","message":"Patch Set 62:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da75f8fc289046de941b7e86677a2f99\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e327d98ae95c41439f0581acca552bf7 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/94e4e3e370c34ed3a8729b83296f65b2 : SUCCESS in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":62},{"id":"ed5cf60349f4c394239858ee9dd9467a6bbdca69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-16 10:54:31.000000000","message":"Uploaded patch set 63: Patch Set 62 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":63},{"id":"9bf6d2f9321fa39204835d087aa27e5ed13ad668","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 11:06:22.000000000","message":"Patch Set 63:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e433ded06f644b08b8380f7f5074a7c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c1cf05728ee34fd695dc20cb6245346c : SUCCESS in 9m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9a3e8b62cc044b028a5b887f08bb8676 : SUCCESS in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"d5b121464e760b6496ce5545ca29ba42a2fefe4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:58:29.000000000","message":"Patch Set 63: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ab39dc7674f47d9b7274f74007e9b2c\n\n- grenade https://zuul.opendev.org/t/openstack/build/87bb742d1364482dab27ef9971323e13 : SUCCESS in 54m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/617bd6ff194a41b68a0c1bebd086853a : SUCCESS in 1h 15m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4c4030c8ec548c699c1df2ffb26887c : SUCCESS in 59m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec087ee39328456c969121049c23bf1e : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64aff4dc68d746a9aa209f11744c030e : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e0285c43aa534b188a877cf1318869df : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a364885532d240208583ee920ce1f939 : SUCCESS in 10m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ddb4b58c61b49a4b34d695b15b65bc3 : SUCCESS in 5m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/366fe3b1881e4eb3855682b2d0f6b5ff : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c91c2420d5243d886b77af5fa919147 : SUCCESS in 8m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6f6f97d5dc32483e844afee4bc2f1a53 : SUCCESS in 11m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c55bf38acf0b44268e9b76c5b0d93918 : SUCCESS in 8m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c5f2d396dd2844f087771257633258ba : SUCCESS in 1h 31m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dab5d830814747448a6b8c4b6e104b7c : SUCCESS in 1h 45m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f0fb51bc38ee48449c0de12b37c8949f : SUCCESS in 1h 52m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/512fde0c6cf646848194e6e4cb52d2ab : SUCCESS in 1h 44m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e44bc3eef8584b08ad0ce18b8268659f : SUCCESS in 1h 57m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d76520b903ab4cdaaf0ba03d3ce8e8b1 : SUCCESS in 57m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5881a1571b040c89232eaf5da007b54 : SUCCESS in 1h 32m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/18c7feb760c5462a8cb83b5769f35a30 : SUCCESS in 1h 39m 30s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c18ff10b3b57422a9406eb2320d490c2 : SUCCESS in 59m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f7b50734502a4449b6f12e998c7b5a5c : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":63},{"id":"788333e4656c927f0975804b597c653f4df48eac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-22 07:22:18.000000000","message":"Uploaded patch set 64: Patch Set 63 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":64},{"id":"d7286a77da6d901404f53d74850f5995b2ad8c43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 07:32:27.000000000","message":"Patch Set 64:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2dae34086b4496fbb274ab9551119fe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e8370ce725e4133a1083936c41f00b1 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/44898f34f4a0443ba100751d760e06cb : SUCCESS in 4m 48s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"fec98525d193f2d28ea98e278302799b861b8964","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-22 07:44:37.000000000","message":"Uploaded patch set 65: Patch Set 64 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":65},{"id":"88a169cb44394efce90cfce767799e8f59442644","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 07:54:32.000000000","message":"Patch Set 65:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54384d90e0784c27a7272cc9e4fb242c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/840cb60d9fd04e7ea3dae6c2dd48158e : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a5ade7653078420396e1a918f9c20c6a : SUCCESS in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"e6b7bcc5c4fa0d41653019cfefa83b2a359359d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 10:15:11.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73b784c4691436eb5aae449ae3440a6\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c2c636565b845b4843e61378d57f74f : SUCCESS in 57m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b68c28d145e4599bb3cfdf722c18d16 : SUCCESS in 1h 42m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb7ef71551c04566a9d9aebdfb7e236b : SUCCESS in 1h 04m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed87d762a9a34b42aa89736cb171d1f6 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3c60ab9578f4931833c9a65b0550669 : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5530a59871440aaa43e8b0e55b04df0 : SUCCESS in 6m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63b3c9edfbcd4380858f024c3e890bc2 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02134a3a74c4071b887f295becf67b3 : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f8fd11f905c4cbea806e1719cd15a9c : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bad757bc53ce434f8dc51329e614f51a : SUCCESS in 8m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8024216cee8348dab0299724959420f0 : SUCCESS in 13m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5d96825fc86f47719d8b6126e614ddb6 : SUCCESS in 8m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9ff754b3570f49e09fa66a232467a0cf : SUCCESS in 1h 42m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e67593daa684e5984519f8150166e03 : SUCCESS in 1h 46m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/38a9721c77e7423e8385c510779333a8 : SUCCESS in 1h 50m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8220dc0c12924c09b77a926b375295e9 : SUCCESS in 1h 42m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/edd5ebc33dc94b8b9ec7a38d1a08a14e : SUCCESS in 2h 17m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb045604d68d4cbabcc23c6ed0f663a0 : SUCCESS in 53m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/959e1f61579d49139fafc12d60712147 : SUCCESS in 1h 41m 50s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/000c0c5895f0401cac3015c84044929c : SUCCESS in 1h 52m 15s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/49bde0a50eec4527a36310104c996bb9 : SUCCESS in 1h 02m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/aae7fd125bff4dc1af78390d709879d5 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":65},{"id":"15c957026db1b93f2bffb9044f8c628b678117f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-25 06:53:52.000000000","message":"Uploaded patch set 66: Patch Set 65 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":66},{"id":"540db091fb8bb0634bb1a60784fb1769ea29b8d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 07:05:23.000000000","message":"Patch Set 66:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a93e36a92d4241009f88fe267822272b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3abb9f6cbd184c4586d2c603950805b8 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b9123177a82d4f2ba0927d4ccf9e136d : SUCCESS in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"5668e49004ee04f97b5f0180d4402982d364ac1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 09:07:34.000000000","message":"Patch Set 66: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6bf55f562b62431eace2532928bdfdc5\n\n- grenade https://zuul.opendev.org/t/openstack/build/182a60a1aedb4a4aac61472edcecc0f6 : SUCCESS in 1h 02m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df67640e8317414da518c16255279b16 : SUCCESS in 1h 17m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46757b790ab443c18e2260f7e44b7030 : SUCCESS in 55m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841a1b1eed174fb9851874666c70599e : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc60567c13ac44b083fd77cbf9d64d5d : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb21688e8a3d4478a94f567d58cf0913 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee648d624fee48089957e3b3b6c521e5 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edabc687f1349219b83cf4d6dbac395 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94e5c1f65edb45a4aa1b250569d48a06 : FAILURE in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d547bbbbab144bb883a69e45d3c3b66c : SUCCESS in 9m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2c6e0367af9b48dc911ebf3cd67ca65a : SUCCESS in 10m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c7855a9a700447c0bac4d553a09f8994 : FAILURE in 9m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/44497a94dc4c4a0d9f187bf75c2e7802 : SUCCESS in 2h 04m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1cbb3d7a744a4328868e3713831015ea : SUCCESS in 1h 54m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/176e63fdac4f45aaa760d26209bd10a3 : SUCCESS in 1h 45m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d34730b469374fef91e7427886b418cc : SUCCESS in 1h 37m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/44a76fe92b124d8cb5aaf6dc2a7e601c : SUCCESS in 2h 12m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6a4d5ba15334b62b4c69ce44d6ab28a : SUCCESS in 58m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/56d4bae8e15b4360a8112d892f2ed9e7 : SUCCESS in 1h 20m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/c0705289d30a4ab586f6c33fcdca1a2c : SUCCESS in 1h 43m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ae59dbf5df114c52ab1c92c3cb04aa0c : SUCCESS in 55m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d7007360f9e64aea83bc6456ab9b1a00 : SUCCESS in 24m 02s","accounts_in_message":[],"_revision_number":66},{"id":"ec76e8129a11594d24dafd8d511cc559a2f5f15c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-25 10:32:00.000000000","message":"Uploaded patch set 67: Patch Set 66 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":67},{"id":"8b6ab5b0ba665c59c611ab0eb17a750bc0315863","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 10:43:25.000000000","message":"Patch Set 67:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f79232a6fdde4406b91cea4586433bbb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b894f50a8ff49e196cc4d8d8cbf0234 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d1f2969b655464fb02957b4bd7de09b : SUCCESS in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"68e2be467b9c55de2cf0d3757b3cf339b68330d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 12:47:03.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098881e6ebb046839bb6c0a0b47d50ca\n\n- grenade https://zuul.opendev.org/t/openstack/build/2aa7b64dd6854e9f9754d6db943751e0 : SUCCESS in 1h 06m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/042c809823bc4a05a5d31932c2d698de : SUCCESS in 1h 42m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/701fe04f6cee49ab825370431daba261 : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99591953d9a4a52b2227734a58f77a2 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5931ef5e57b4c59abef32f12d7a2a18 : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d32fbb91fd1404298759fbcbd1779e9 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/404effa4c53f4f5180efef4b336353b8 : SUCCESS in 8m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/644ff7e1e900465389796f230196c5e4 : SUCCESS in 5m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1033601f8aec4af5ad4e921b448a54d6 : SUCCESS in 8m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2fee71dc629041bcb669c214d574f292 : SUCCESS in 9m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/42edb6a0e59c479cb2e8d4c18867689d : SUCCESS in 13m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7dee7c0f06874fac97991da43158deb0 : SUCCESS in 11m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/45c0e2e036b0468e8ae3134815ea696f : TIMED_OUT in 2h 03m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/115876ed87a448ac84c185496254c607 : SUCCESS in 1h 50m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/421d09a8e3864d30ac458a4e6575b10e : SUCCESS in 1h 38m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c5862228e9554dd9aa6c17f5d2177805 : SUCCESS in 1h 42m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f82a45e6294c43b18a97c25c67401e69 : SUCCESS in 1h 55m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d20399ec7c4a49c099d7f996b9cf03eb : SUCCESS in 45m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8b3c33959244857a20e337db1ba6891 : SUCCESS in 1h 41m 04s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2ac33183596a4730ab2257d26e590294 : SUCCESS in 1h 54m 27s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/88a86496397748868ed456dcfc1a92db : SUCCESS in 1h 07m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/487aa593d04443499327f050aee63ebb : SUCCESS in 23m 45s","accounts_in_message":[],"_revision_number":67},{"id":"52af4302b3bcf5061f798806b6028e9319a95086","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-26 06:43:52.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":68},{"id":"111c2036318ff50d3185c4aebd16af67a13ef802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 06:50:08.000000000","message":"Patch Set 68:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf19460ddc6541ccb7b23994caf6a502\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3d6a608edc044bf5ad1419aa5de33661 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e6a72908562a4e70ae3103b8a468f64b : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"aac649f74f310075e194b9fcfdad342113b9e6ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 11:10:38.000000000","message":"Patch Set 68: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8270e7fcfd2d431fb71c5487c0358aca\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8d20e793859488fab971c31e7e2ff2a : SUCCESS in 43m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6c2c5bca71da41028c8bc64aa990a039 : SUCCESS in 1h 24m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89d7ebe78ab94fceb12999a5b527c235 : SUCCESS in 1h 09m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5521b36a32443b98bcc5fb3591d1219 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4be08be04eb43ff853b66e59744ad36 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7237f806daa470abdc55d6e5e4c8b4d : SUCCESS in 6m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46abc552bc674ec3aad112814a613d8a : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1435b8e51404429887000140354e7deb : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ed2f1d4a1d343d9a1412d92c13a87b0 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e5d0c8908d348cba687afa57d220999 : SUCCESS in 9m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b8af83c36ac648939e676dd9608b5efe : SUCCESS in 11m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1601db6a1a244fbba037ed3a05a471b3 : SUCCESS in 8m 40s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ac8773173e364a5cbdb8a5a8cb63a6be : SUCCESS in 1h 33m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5a9a0647fd1a4ad4a7ec301409139047 : SUCCESS in 1h 46m 27s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a65b12d1b4d9442298441a339984d60f : SUCCESS in 1h 42m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/71dcdf13c3634960914e082e2a7ca462 : SUCCESS in 1h 47m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1a61367c752b40479f6a03c4cdfebff0 : SUCCESS in 2h 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5654101f05a4e2b819a929b20ccf56f : SUCCESS in 59m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00afb2ec32104d5caac403977424b352 : SUCCESS in 2h 00m 33s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8b977a10bac047338563e2e966284f7a : SUCCESS in 1h 19m 11s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9081ce142e1443d8b244351fd901627a : SUCCESS in 1h 12m 46s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0981b419babe4a75a73c4a3a67d0301d : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":68},{"id":"1a923c6626a33253e7bf2da8a0e7e3ecc1ae4e1d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-31 05:20:15.000000000","message":"Patch Set 68: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":68},{"id":"147eb3605d52867d9089e69bf818a87f91f93942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 05:20:58.000000000","message":"Patch Set 68: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":68},{"id":"9df7532173176b8b354991db2708b5ebc485de50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 07:29:40.000000000","message":"Patch Set 68: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49f48adea4914664b85ef2d9f511a1bf\n\n- grenade https://zuul.opendev.org/t/openstack/build/1843ba03414149aba9421871680b709a : SUCCESS in 47m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42bdbe1c0aa440649ec0e5d572ddc809 : SUCCESS in 1h 06m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/643526c7369343468cd1c60ed80d62a7 : SUCCESS in 45m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee59b7725f3341b8ab858ca63feafd39 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47e5818c8822492d8e3368a742d4b35a : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91ccf731b8df401b86fcea1bb20f755b : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5102d7ba7ccb4823b3e31b1f674eacab : SUCCESS in 4m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9f85002f6c44be39e2e554a5bd552b5 : SUCCESS in 8m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90fabb556ed04f8ea43c83fd7bd84f4b : SUCCESS in 8m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/57c197ab71f04280b8160073e2e1b491 : SUCCESS in 14m 08s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ed2569750aa741fbab80039b322aee11 : SUCCESS in 1h 57m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/969b82c76a2e42bbb350b898710d96a5 : SUCCESS in 2h 02m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/100b47ceee734df69468b9bafe1bc0a6 : SUCCESS in 1h 35m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7428df94a1d544509d4b852d95623a91 : SUCCESS in 39m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2c4a7c0bec1e4d9ab9c0081626dda60a : SUCCESS in 1h 53m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/deec86b2a21c43189bf0d0b9160aba80 : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":68},{"id":"d5f2a13b9129f407bd97a0fd870fcb47fb5a384d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 07:29:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":68},{"id":"3911d5c73b08b9a4ed8a49e78f91233abc1d1536","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 07:30:52.000000000","message":"Patch Set 68:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36de5977528748fba048a8838afd5e5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d80867de2b74dcfb59e8486aee3341d : SUCCESS in 53s","accounts_in_message":[],"_revision_number":68}],"current_revision_number":68,"current_revision":"6f4d9c434679c7440b026e1ffef38bc7c73cc5f6","revisions":{"b4511efcaf2ef90c24a38c60f248734d407fa41e":{"kind":"REWORK","_number":1,"created":"2023-08-04 07:04:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/1"}}},"commit":{"parents":[{"commit":"333dfad8845b5a5f9a605d20db2ed1a039d60368","subject":"Add api-ref for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/333dfad8845b5a5f9a605d20db2ed1a039d60368"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4511efcaf2ef90c24a38c60f248734d407fa41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4511efcaf2ef90c24a38c60f248734d407fa41e"}]},"branch":"refs/heads/master"},"3d3801ef9c49fa789e2bdbbc2c197888f0f159c4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-08-09 11:13:25.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/2"}}},"commit":{"parents":[{"commit":"ecc8c8173caebb1aeb65de98cd8b3a69edace718","subject":"Add api-ref for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ecc8c8173caebb1aeb65de98cd8b3a69edace718"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-09 10:50:33.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d3801ef9c49fa789e2bdbbc2c197888f0f159c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d3801ef9c49fa789e2bdbbc2c197888f0f159c4"}]},"branch":"refs/heads/master"},"31af0bcebbbb342eb11df0bf0fa7f595522d2d37":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-09 11:32:57.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/3"}}},"commit":{"parents":[{"commit":"ac2baf9b288de76401fafc49df5a6792dec469dc","subject":"Add api-ref for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac2baf9b288de76401fafc49df5a6792dec469dc"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-09 11:32:57.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31af0bcebbbb342eb11df0bf0fa7f595522d2d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31af0bcebbbb342eb11df0bf0fa7f595522d2d37"}]},"branch":"refs/heads/master","description":"Rebase"},"74be722b5bb4e6025d55cb3654fe77028bbac2ce":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-08-14 11:50:37.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/4"}}},"commit":{"parents":[{"commit":"84ee860ae96b7319bf518fe3de452585f925e434","subject":"Add api-ref for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84ee860ae96b7319bf518fe3de452585f925e434"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-14 10:51:37.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74be722b5bb4e6025d55cb3654fe77028bbac2ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74be722b5bb4e6025d55cb3654fe77028bbac2ce"}]},"branch":"refs/heads/master"},"5e5bc7f2c3e2e028b28b8c8b87dd7b158dd87d0d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-08-16 09:29:45.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/5"}}},"commit":{"parents":[{"commit":"b7cfbe4d651f52aa25916fc7a887ea0beac95d93","subject":"Add api-ref for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7cfbe4d651f52aa25916fc7a887ea0beac95d93"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-16 08:58:23.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e5bc7f2c3e2e028b28b8c8b87dd7b158dd87d0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e5bc7f2c3e2e028b28b8c8b87dd7b158dd87d0d"}]},"branch":"refs/heads/master"},"7739ef46b5b44983ae63949ea57b4b0944bc2d9e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-08-20 12:19:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/6"}}},"commit":{"parents":[{"commit":"268683e5845fb9a28c91c2179376cb19dfd13256","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/268683e5845fb9a28c91c2179376cb19dfd13256"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-20 11:56:15.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7739ef46b5b44983ae63949ea57b4b0944bc2d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7739ef46b5b44983ae63949ea57b4b0944bc2d9e"}]},"branch":"refs/heads/master"},"0c129ed4163a2eef1a87503743293b2bde2ee226":{"kind":"REWORK","_number":7,"created":"2023-08-24 11:40:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/7"}}},"commit":{"parents":[{"commit":"529b62dbfa2071f96683cf4ad4d3f1954d95e8dd","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/529b62dbfa2071f96683cf4ad4d3f1954d95e8dd"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-24 11:34:55.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c129ed4163a2eef1a87503743293b2bde2ee226"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c129ed4163a2eef1a87503743293b2bde2ee226"}]},"branch":"refs/heads/master"},"0bfc91808a427b214b3b70ba81711d81651919f5":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-08-24 12:00:55.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/8"}}},"commit":{"parents":[{"commit":"c6a3644d1bcb316f95875ee3d071fcb72d3916f8","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c6a3644d1bcb316f95875ee3d071fcb72d3916f8"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-24 11:58:49.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bfc91808a427b214b3b70ba81711d81651919f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bfc91808a427b214b3b70ba81711d81651919f5"}]},"branch":"refs/heads/master"},"336d45d73d5d51c3edc9bf9e37ffee01b51207d5":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2023-08-24 15:02:21.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/9"}}},"commit":{"parents":[{"commit":"5530ba9af056c865b51f3e5ea2064a05fc6a3b14","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5530ba9af056c865b51f3e5ea2064a05fc6a3b14"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-24 14:57:35.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/336d45d73d5d51c3edc9bf9e37ffee01b51207d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/336d45d73d5d51c3edc9bf9e37ffee01b51207d5"}]},"branch":"refs/heads/master"},"da00f1b6e6709b17496e074abdf721a837550f19":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-08-25 17:41:42.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/10"}}},"commit":{"parents":[{"commit":"3777ec966a22a4260d226c4da3b64c26fce690fc","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3777ec966a22a4260d226c4da3b64c26fce690fc"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-25 17:20:26.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da00f1b6e6709b17496e074abdf721a837550f19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/da00f1b6e6709b17496e074abdf721a837550f19"}]},"branch":"refs/heads/master"},"d1b1902119e3b30072b423822737be76dad57414":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2023-08-28 14:31:13.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/11"}}},"commit":{"parents":[{"commit":"623ba9177499b2be3dba90e4abf581d4562fcad9","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/623ba9177499b2be3dba90e4abf581d4562fcad9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-28 14:30:03.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1b1902119e3b30072b423822737be76dad57414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d1b1902119e3b30072b423822737be76dad57414"}]},"branch":"refs/heads/master"},"c51d4854f66de6db7c2cb31295da3eaac7c9517f":{"kind":"REWORK","_number":12,"created":"2023-08-28 15:25:36.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/12"}}},"commit":{"parents":[{"commit":"623ba9177499b2be3dba90e4abf581d4562fcad9","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/623ba9177499b2be3dba90e4abf581d4562fcad9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-28 15:24:39.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c51d4854f66de6db7c2cb31295da3eaac7c9517f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c51d4854f66de6db7c2cb31295da3eaac7c9517f"}]},"branch":"refs/heads/master"},"6e21d665db0b3dcd762159cb77a4bb6d927383b0":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-08-29 08:58:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/13"}}},"commit":{"parents":[{"commit":"9c85e3ded1b91aea71cef3bd7e5d252da0d93ebe","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c85e3ded1b91aea71cef3bd7e5d252da0d93ebe"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-29 08:08:27.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e21d665db0b3dcd762159cb77a4bb6d927383b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e21d665db0b3dcd762159cb77a4bb6d927383b0"}]},"branch":"refs/heads/master"},"a464a46382bfd2f8ce62531cb7332e243686d47e":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-08-29 14:21:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/14"}}},"commit":{"parents":[{"commit":"9ddb98f111e9c7f6a3a32f512b8b797dfb970e13","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9ddb98f111e9c7f6a3a32f512b8b797dfb970e13"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-29 14:17:15.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a464a46382bfd2f8ce62531cb7332e243686d47e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a464a46382bfd2f8ce62531cb7332e243686d47e"}]},"branch":"refs/heads/master"},"a8de1ac36095859014e5deaa164d9232f26d4172":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2023-08-29 18:40:41.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/15"}}},"commit":{"parents":[{"commit":"235bae2696abacd3e13eeed3e61bfdfa1f864b81","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/235bae2696abacd3e13eeed3e61bfdfa1f864b81"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-29 18:35:04.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8de1ac36095859014e5deaa164d9232f26d4172"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a8de1ac36095859014e5deaa164d9232f26d4172"}]},"branch":"refs/heads/master"},"6e260ab23434b4b7533aa683af8df3627b8ff260":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2023-08-30 08:44:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/16"}}},"commit":{"parents":[{"commit":"65e7e7e9785d5245146459899c6302ab856f3aee","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/65e7e7e9785d5245146459899c6302ab856f3aee"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-30 08:42:46.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e260ab23434b4b7533aa683af8df3627b8ff260"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e260ab23434b4b7533aa683af8df3627b8ff260"}]},"branch":"refs/heads/master"},"0962825fc327cf17f78d07fbf922e257ed4386b7":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2023-08-30 10:40:22.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/17"}}},"commit":{"parents":[{"commit":"83901a1f68ed0717431ed8fe763fdde8be26fc91","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83901a1f68ed0717431ed8fe763fdde8be26fc91"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-30 10:29:47.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0962825fc327cf17f78d07fbf922e257ed4386b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0962825fc327cf17f78d07fbf922e257ed4386b7"}]},"branch":"refs/heads/master"},"e4d0a1d4da6c86e487618f87f84f00f28e6fc771":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-08-30 13:57:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/18"}}},"commit":{"parents":[{"commit":"21a295ded1d529d4ed810250d1a19b8da1c987bb","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21a295ded1d529d4ed810250d1a19b8da1c987bb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-30 13:53:18.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e4d0a1d4da6c86e487618f87f84f00f28e6fc771"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e4d0a1d4da6c86e487618f87f84f00f28e6fc771"}]},"branch":"refs/heads/master"},"2a79fd654cf6b71b09c0d7c7a39b1454c76db743":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2023-08-31 07:57:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/19"}}},"commit":{"parents":[{"commit":"63fe00bf8d89f1fb169e6971528fa9ba75d9fc6a","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63fe00bf8d89f1fb169e6971528fa9ba75d9fc6a"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-31 07:49:15.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a79fd654cf6b71b09c0d7c7a39b1454c76db743"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a79fd654cf6b71b09c0d7c7a39b1454c76db743"}]},"branch":"refs/heads/master"},"de0f07be33ea2b4f85564b67c7530b83abd403be":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2023-08-31 10:27:37.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/20"}}},"commit":{"parents":[{"commit":"139164559db9118b8b8f9a18bbc2e0d4dcf60d35","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/139164559db9118b8b8f9a18bbc2e0d4dcf60d35"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-08-31 10:21:24.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de0f07be33ea2b4f85564b67c7530b83abd403be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de0f07be33ea2b4f85564b67c7530b83abd403be"}]},"branch":"refs/heads/master"},"e1afc6c083fde527669ddc71d6c21f33084d1649":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2023-09-05 07:25:17.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/21"}}},"commit":{"parents":[{"commit":"fa8e5358620560d59ce9c5642907a4d1e7dd34d4","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa8e5358620560d59ce9c5642907a4d1e7dd34d4"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-09-05 07:20:03.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1afc6c083fde527669ddc71d6c21f33084d1649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1afc6c083fde527669ddc71d6c21f33084d1649"}]},"branch":"refs/heads/master"},"57300909858ce30079b625e066823c4b445e42a9":{"kind":"REWORK","_number":22,"created":"2023-10-19 07:06:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/22"}}},"commit":{"parents":[{"commit":"186262596aa71eb381bb54d92ed86838a8ddde54","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/186262596aa71eb381bb54d92ed86838a8ddde54"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-10-19 06:49:23.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57300909858ce30079b625e066823c4b445e42a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57300909858ce30079b625e066823c4b445e42a9"}]},"branch":"refs/heads/master"},"d3036822d07d333e311928b791bb678c3fcf8c86":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2023-10-31 09:27:15.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/23"}}},"commit":{"parents":[{"commit":"6ec808b0dbc9cca4e81cfabd0ac3699ce06641e9","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ec808b0dbc9cca4e81cfabd0ac3699ce06641e9"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-10-31 09:18:22.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3036822d07d333e311928b791bb678c3fcf8c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3036822d07d333e311928b791bb678c3fcf8c86"}]},"branch":"refs/heads/master"},"ae4281046dbe9e04c90a360f887f2c59b9b09986":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2023-10-31 12:49:39.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/24"}}},"commit":{"parents":[{"commit":"de4dad1d35cbfcd5f9bdf8528a0ea36f4008aeec","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de4dad1d35cbfcd5f9bdf8528a0ea36f4008aeec"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-10-31 12:49:39.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae4281046dbe9e04c90a360f887f2c59b9b09986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae4281046dbe9e04c90a360f887f2c59b9b09986"}]},"branch":"refs/heads/master","description":"Rebase"},"6eac4427cfaa2bb35156de887c6341d254470e83":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-11-02 09:35:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/25"}}},"commit":{"parents":[{"commit":"be71e1a7a5b6036aca19b7ba68dfd55dd30f2e7e","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/be71e1a7a5b6036aca19b7ba68dfd55dd30f2e7e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-02 08:54:11.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6eac4427cfaa2bb35156de887c6341d254470e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6eac4427cfaa2bb35156de887c6341d254470e83"}]},"branch":"refs/heads/master"},"306c795b44bf750592e200583e338338dd8db178":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2023-11-08 11:23:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/26"}}},"commit":{"parents":[{"commit":"7460b5952a78c4f33467db66e096f00457668383","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7460b5952a78c4f33467db66e096f00457668383"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-08 11:17:25.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/306c795b44bf750592e200583e338338dd8db178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/306c795b44bf750592e200583e338338dd8db178"}]},"branch":"refs/heads/master"},"293bfa776ca1beaad08e711a2001bf6235c5c145":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2023-12-05 14:08:25.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/27"}}},"commit":{"parents":[{"commit":"021b21dcf4e23c7c4533334233d352d5773b331f","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/021b21dcf4e23c7c4533334233d352d5773b331f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-12-05 13:54:02.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/293bfa776ca1beaad08e711a2001bf6235c5c145"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/293bfa776ca1beaad08e711a2001bf6235c5c145"}]},"branch":"refs/heads/master"},"5dc83301d56ab094e669e181c68a684103024a80":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2024-01-02 12:02:35.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/28"}}},"commit":{"parents":[{"commit":"cb2be91ae081ff74613233f97249b6968bd7ff64","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cb2be91ae081ff74613233f97249b6968bd7ff64"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-02 12:02:35.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dc83301d56ab094e669e181c68a684103024a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dc83301d56ab094e669e181c68a684103024a80"}]},"branch":"refs/heads/master","description":"Rebase"},"712c6109b4c3bb0a57b80b7fa52d1bc1058a9242":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2024-01-11 07:36:44.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/29"}}},"commit":{"parents":[{"commit":"439c6c688b5ba13c30e2b78676b15e61286fbefb","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/439c6c688b5ba13c30e2b78676b15e61286fbefb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-11 07:05:23.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/712c6109b4c3bb0a57b80b7fa52d1bc1058a9242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/712c6109b4c3bb0a57b80b7fa52d1bc1058a9242"}]},"branch":"refs/heads/master"},"656df3207b856209ac01c780fb886c30081a9e44":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2024-01-22 10:58:50.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/30"}}},"commit":{"parents":[{"commit":"6c0ad5bc3569548e813a14baf7bdd6312e5b985c","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6c0ad5bc3569548e813a14baf7bdd6312e5b985c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-22 10:49:40.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/656df3207b856209ac01c780fb886c30081a9e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/656df3207b856209ac01c780fb886c30081a9e44"}]},"branch":"refs/heads/master"},"0496374eaf505456c1d58288387f04333ba6ede7":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-01-23 11:15:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/31"}}},"commit":{"parents":[{"commit":"1bb964f6a1bc56d5955c4f4975f421c198fa5bed","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1bb964f6a1bc56d5955c4f4975f421c198fa5bed"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-23 11:10:23.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0496374eaf505456c1d58288387f04333ba6ede7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0496374eaf505456c1d58288387f04333ba6ede7"}]},"branch":"refs/heads/master"},"3815edb7b8e8934f585736ada41aaf867bd1788c":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2024-01-24 10:53:02.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/32"}}},"commit":{"parents":[{"commit":"23a7fa3f61c5bd9bcd6fd9198eede6d8ede1cd17","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23a7fa3f61c5bd9bcd6fd9198eede6d8ede1cd17"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-01-24 10:41:09.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3815edb7b8e8934f585736ada41aaf867bd1788c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3815edb7b8e8934f585736ada41aaf867bd1788c"}]},"branch":"refs/heads/master"},"f507005f79ddb3a14e937b23b6a0fdca66e4b095":{"kind":"TRIVIAL_REBASE","_number":33,"created":"2024-02-07 06:50:50.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/33"}}},"commit":{"parents":[{"commit":"7bd93f018d1b8c156e69af85ee876a9400e90521","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7bd93f018d1b8c156e69af85ee876a9400e90521"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-07 06:38:01.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f507005f79ddb3a14e937b23b6a0fdca66e4b095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f507005f79ddb3a14e937b23b6a0fdca66e4b095"}]},"branch":"refs/heads/master"},"0b7f8f8ad38ebad74d1deb0c04ddb696e67b7fb4":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2024-02-07 09:14:43.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/34"}}},"commit":{"parents":[{"commit":"c36fbd89522d56a78a94916d3a51865b53ce0501","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c36fbd89522d56a78a94916d3a51865b53ce0501"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-07 09:10:55.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b7f8f8ad38ebad74d1deb0c04ddb696e67b7fb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b7f8f8ad38ebad74d1deb0c04ddb696e67b7fb4"}]},"branch":"refs/heads/master"},"628cd11904821254b357d192c61af6c9da4a7c7a":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2024-02-15 10:31:13.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/35"}}},"commit":{"parents":[{"commit":"c086f5afd7325e3b70b4c651da4213b603ff00d6","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c086f5afd7325e3b70b4c651da4213b603ff00d6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-15 10:22:18.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/628cd11904821254b357d192c61af6c9da4a7c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/628cd11904821254b357d192c61af6c9da4a7c7a"}]},"branch":"refs/heads/master"},"4dc908872df7f517dfc085a22219b3dc744d3d05":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2024-02-20 10:23:20.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/36"}}},"commit":{"parents":[{"commit":"459736fde680565f0bc282fd53ff141567edf474","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/459736fde680565f0bc282fd53ff141567edf474"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-20 09:54:34.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dc908872df7f517dfc085a22219b3dc744d3d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4dc908872df7f517dfc085a22219b3dc744d3d05"}]},"branch":"refs/heads/master"},"d89374d752f5136d8ec73e807076b5e33fd4b5e5":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2024-02-21 12:59:22.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/37"}}},"commit":{"parents":[{"commit":"73351405bc233f9a639049a305154ec3fdc6caec","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73351405bc233f9a639049a305154ec3fdc6caec"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-21 12:44:13.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d89374d752f5136d8ec73e807076b5e33fd4b5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d89374d752f5136d8ec73e807076b5e33fd4b5e5"}]},"branch":"refs/heads/master"},"793519a8c43d3ed8bb682b57ee8aa77a62cfd8c9":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2024-02-22 07:41:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/38"}}},"commit":{"parents":[{"commit":"dc03c447ecac971cbde1989f5a61a823c1968391","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc03c447ecac971cbde1989f5a61a823c1968391"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-22 07:37:43.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/793519a8c43d3ed8bb682b57ee8aa77a62cfd8c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/793519a8c43d3ed8bb682b57ee8aa77a62cfd8c9"}]},"branch":"refs/heads/master"},"94da23772953d0876d59a7bd6daeda413b5816db":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2024-02-22 11:00:32.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/39"}}},"commit":{"parents":[{"commit":"35bf8825c2698a416c9fb8d6c30b4ad25bd188bb","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35bf8825c2698a416c9fb8d6c30b4ad25bd188bb"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-22 11:00:32.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94da23772953d0876d59a7bd6daeda413b5816db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94da23772953d0876d59a7bd6daeda413b5816db"}]},"branch":"refs/heads/master","description":"Rebase"},"19a3847ceecf52d3d40b81b5c3a9640e9f2cdc07":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2024-02-26 09:16:40.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/40"}}},"commit":{"parents":[{"commit":"b3d62a5ae15079de04dab60d7dac65cbec0335c1","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b3d62a5ae15079de04dab60d7dac65cbec0335c1"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-26 09:12:47.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19a3847ceecf52d3d40b81b5c3a9640e9f2cdc07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19a3847ceecf52d3d40b81b5c3a9640e9f2cdc07"}]},"branch":"refs/heads/master"},"44d8cfa784eacf758381f586b3da10973ec692cd":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2024-02-27 07:12:48.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/41"}}},"commit":{"parents":[{"commit":"7df4ab5cf5a815316ced642cd574082198878fb6","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7df4ab5cf5a815316ced642cd574082198878fb6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-27 07:09:08.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44d8cfa784eacf758381f586b3da10973ec692cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44d8cfa784eacf758381f586b3da10973ec692cd"}]},"branch":"refs/heads/master"},"000c1d6dd0a7d96708c186cac93ca5fdf4bba703":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2024-02-27 13:25:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/42"}}},"commit":{"parents":[{"commit":"432766e27560075f3526507da7655eac28f8863f","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/432766e27560075f3526507da7655eac28f8863f"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-27 13:18:02.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/000c1d6dd0a7d96708c186cac93ca5fdf4bba703"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/000c1d6dd0a7d96708c186cac93ca5fdf4bba703"}]},"branch":"refs/heads/master"},"d3dcd9e4e658824d91c9b739235982b94037b89e":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2024-02-27 15:26:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/43"}}},"commit":{"parents":[{"commit":"6549eb0077b74e28fbdc50dd86bb182810332333","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6549eb0077b74e28fbdc50dd86bb182810332333"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-27 15:22:07.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3dcd9e4e658824d91c9b739235982b94037b89e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3dcd9e4e658824d91c9b739235982b94037b89e"}]},"branch":"refs/heads/master"},"9224b5e0ac848bfb7c04b9617b079728f450f1f7":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2024-02-28 03:51:38.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/44"}}},"commit":{"parents":[{"commit":"8a404a9e55a12bf6f988dab8614aee25ff0e694b","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8a404a9e55a12bf6f988dab8614aee25ff0e694b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 03:28:40.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9224b5e0ac848bfb7c04b9617b079728f450f1f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9224b5e0ac848bfb7c04b9617b079728f450f1f7"}]},"branch":"refs/heads/master"},"01fb6278c95fcabc95201f7fce5f2fd3bfedf873":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2024-02-28 09:10:56.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/45"}}},"commit":{"parents":[{"commit":"e3bff117b080f68ab534919251d2098002dcfbba","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3bff117b080f68ab534919251d2098002dcfbba"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 09:06:21.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01fb6278c95fcabc95201f7fce5f2fd3bfedf873"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/01fb6278c95fcabc95201f7fce5f2fd3bfedf873"}]},"branch":"refs/heads/master"},"071076607762981ccdc81dc25fd51dffda6c1025":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2024-02-28 11:10:35.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/46"}}},"commit":{"parents":[{"commit":"6d9fec30fd4efdcfdd0a9bba2035f9611c5227d3","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6d9fec30fd4efdcfdd0a9bba2035f9611c5227d3"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 11:07:08.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/071076607762981ccdc81dc25fd51dffda6c1025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/071076607762981ccdc81dc25fd51dffda6c1025"}]},"branch":"refs/heads/master"},"201248b48ebfa36f400afe1d02aea4708986747b":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2024-02-28 16:14:46.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/47"}}},"commit":{"parents":[{"commit":"b334dc9e8fc002e24fb0256ce3515e8399cb1a72","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b334dc9e8fc002e24fb0256ce3515e8399cb1a72"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 16:09:51.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/201248b48ebfa36f400afe1d02aea4708986747b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/201248b48ebfa36f400afe1d02aea4708986747b"}]},"branch":"refs/heads/master"},"070462f3b8f5a9c8a7df3c2693b715e4855ac4f3":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2024-02-28 19:01:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/48"}}},"commit":{"parents":[{"commit":"784c3ffa384474f7a2f2c161ed9ffecf0b0bd756","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/784c3ffa384474f7a2f2c161ed9ffecf0b0bd756"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 18:56:54.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/070462f3b8f5a9c8a7df3c2693b715e4855ac4f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/070462f3b8f5a9c8a7df3c2693b715e4855ac4f3"}]},"branch":"refs/heads/master"},"f249d374c901d1df512a1e53ab18585e5532faef":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2024-02-28 19:21:23.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/49"}}},"commit":{"parents":[{"commit":"0bcf477a16197f7a2adf51d2533ad16af7b1bfd6","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bcf477a16197f7a2adf51d2533ad16af7b1bfd6"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-28 19:17:58.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f249d374c901d1df512a1e53ab18585e5532faef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f249d374c901d1df512a1e53ab18585e5532faef"}]},"branch":"refs/heads/master"},"7a1fa000dacdd611954f896ea640607d891a83b3":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2024-02-29 10:17:30.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/50"}}},"commit":{"parents":[{"commit":"2b41597fe5e2564613f79411480125ec831e1672","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b41597fe5e2564613f79411480125ec831e1672"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-29 10:13:33.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a1fa000dacdd611954f896ea640607d891a83b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a1fa000dacdd611954f896ea640607d891a83b3"}]},"branch":"refs/heads/master"},"a98cf01fe373a5f1b1c03f8f77b9825562082675":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2024-02-29 11:03:29.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/51"}}},"commit":{"parents":[{"commit":"55459f1e6d64a61579c8e975cd76994181c5ea44","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/55459f1e6d64a61579c8e975cd76994181c5ea44"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-02-29 10:57:34.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a98cf01fe373a5f1b1c03f8f77b9825562082675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a98cf01fe373a5f1b1c03f8f77b9825562082675"}]},"branch":"refs/heads/master"},"6cd232ff91929f153a9844ca5d05f3f1a513437a":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2024-04-04 07:00:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/52"}}},"commit":{"parents":[{"commit":"d4dca485db6657e78a2678548da089a3e8a7bfa8","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4dca485db6657e78a2678548da089a3e8a7bfa8"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-04 06:36:19.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6cd232ff91929f153a9844ca5d05f3f1a513437a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6cd232ff91929f153a9844ca5d05f3f1a513437a"}]},"branch":"refs/heads/master"},"c94ba8030f41fc26cf5a64adbff47e2b17f6c1d1":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2024-04-08 06:31:24.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/53"}}},"commit":{"parents":[{"commit":"6a204740c665d3d982d6dac0cbf6f4fa526716be","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a204740c665d3d982d6dac0cbf6f4fa526716be"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-08 06:26:14.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c94ba8030f41fc26cf5a64adbff47e2b17f6c1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c94ba8030f41fc26cf5a64adbff47e2b17f6c1d1"}]},"branch":"refs/heads/master"},"381c69811878876a76f65be7a2141f8369d9f5f4":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2024-04-29 08:04:05.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/54"}}},"commit":{"parents":[{"commit":"3ba4e8aa5926298c220594b2b8daa65b09147d47","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ba4e8aa5926298c220594b2b8daa65b09147d47"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-29 07:58:15.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/381c69811878876a76f65be7a2141f8369d9f5f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/381c69811878876a76f65be7a2141f8369d9f5f4"}]},"branch":"refs/heads/master"},"805d573864a3f73ed048c71767161984b3ecfdd4":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2024-04-30 09:42:34.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/55"}}},"commit":{"parents":[{"commit":"ed070aa0cf21f247c36ecf3b755bb58056989b8e","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed070aa0cf21f247c36ecf3b755bb58056989b8e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-04-30 09:37:31.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/805d573864a3f73ed048c71767161984b3ecfdd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/805d573864a3f73ed048c71767161984b3ecfdd4"}]},"branch":"refs/heads/master"},"0a270480ff4f17963257ee5ce6cbc29a4d0853db":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2024-05-07 06:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/56"}}},"commit":{"parents":[{"commit":"eb3e9fcbd15a5ac7575e07658a4b8487fa641372","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3e9fcbd15a5ac7575e07658a4b8487fa641372"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-06 13:01:57.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a270480ff4f17963257ee5ce6cbc29a4d0853db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a270480ff4f17963257ee5ce6cbc29a4d0853db"}]},"branch":"refs/heads/master"},"7904532f457ba551adc200075e4aa8df7f75ff52":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2024-05-07 13:08:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/57"}}},"commit":{"parents":[{"commit":"f256cae25d41d6f36d66cce7cf7cab8430362741","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f256cae25d41d6f36d66cce7cf7cab8430362741"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-07 13:03:25.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7904532f457ba551adc200075e4aa8df7f75ff52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7904532f457ba551adc200075e4aa8df7f75ff52"}]},"branch":"refs/heads/master"},"8448968f7b367663a280aad7556e7d059ac0617e":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2024-05-08 09:01:11.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/58"}}},"commit":{"parents":[{"commit":"9a4b3ba7bf416ddfb953c664ce034650098d10fa","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a4b3ba7bf416ddfb953c664ce034650098d10fa"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-05-08 08:55:06.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8448968f7b367663a280aad7556e7d059ac0617e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8448968f7b367663a280aad7556e7d059ac0617e"}]},"branch":"refs/heads/master"},"a7bf9e0fcc869af62cf3165560d0390cefe4525c":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2024-06-06 06:46:59.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/59"}}},"commit":{"parents":[{"commit":"fde42755d553fd5f9109cbd041a0d6f1d78cb65e","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fde42755d553fd5f9109cbd041a0d6f1d78cb65e"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-06 06:38:16.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7bf9e0fcc869af62cf3165560d0390cefe4525c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7bf9e0fcc869af62cf3165560d0390cefe4525c"}]},"branch":"refs/heads/master"},"313a2be0da5d270bc279673c0017066e5e657556":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2024-06-14 10:35:46.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/60"}}},"commit":{"parents":[{"commit":"71bbf52784079bb1c1d1bcd4e6e8c48b372d4029","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/71bbf52784079bb1c1d1bcd4e6e8c48b372d4029"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-14 10:19:06.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/313a2be0da5d270bc279673c0017066e5e657556"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/313a2be0da5d270bc279673c0017066e5e657556"}]},"branch":"refs/heads/master"},"5cc1c826fea1252fb3076db241e4785a3ee90132":{"kind":"TRIVIAL_REBASE","_number":61,"created":"2024-06-28 12:46:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/61"}}},"commit":{"parents":[{"commit":"7d1673ece6b91a143a0a2e02a0f1c6458b87cad4","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d1673ece6b91a143a0a2e02a0f1c6458b87cad4"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-06-28 12:41:00.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cc1c826fea1252fb3076db241e4785a3ee90132"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cc1c826fea1252fb3076db241e4785a3ee90132"}]},"branch":"refs/heads/master"},"7f366526e0cddde0266e96cc4bf144d81552a5be":{"kind":"TRIVIAL_REBASE","_number":62,"created":"2024-07-16 10:22:56.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/62"}}},"commit":{"parents":[{"commit":"1e6b60f4f3308c8ec7ee8460678a509c41600697","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e6b60f4f3308c8ec7ee8460678a509c41600697"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-16 10:15:57.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f366526e0cddde0266e96cc4bf144d81552a5be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7f366526e0cddde0266e96cc4bf144d81552a5be"}]},"branch":"refs/heads/master"},"775a4dc57dad9f27f4104b4e97f5125f18016e5a":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2024-07-16 10:54:31.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/63"}}},"commit":{"parents":[{"commit":"b1a54de2f9ec3661b31edd957eead4addc215965","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1a54de2f9ec3661b31edd957eead4addc215965"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-16 10:50:02.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/775a4dc57dad9f27f4104b4e97f5125f18016e5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/775a4dc57dad9f27f4104b4e97f5125f18016e5a"}]},"branch":"refs/heads/master"},"2f3d63c58fe1d8142c61f0362bb8ff072914c157":{"kind":"TRIVIAL_REBASE","_number":64,"created":"2024-07-22 07:22:18.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/64"}}},"commit":{"parents":[{"commit":"fd34fba0fe377fe810a433895ae03f94e1818cfe","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fd34fba0fe377fe810a433895ae03f94e1818cfe"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-22 06:59:50.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f3d63c58fe1d8142c61f0362bb8ff072914c157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f3d63c58fe1d8142c61f0362bb8ff072914c157"}]},"branch":"refs/heads/master"},"b7985cd51280d80b60bf3708ce1f508abb26ba4c":{"kind":"TRIVIAL_REBASE","_number":65,"created":"2024-07-22 07:44:37.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/65"}}},"commit":{"parents":[{"commit":"c598adcbfdeaa0c70c21f5f3fafffa51dbf1a018","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c598adcbfdeaa0c70c21f5f3fafffa51dbf1a018"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-22 07:39:35.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7985cd51280d80b60bf3708ce1f508abb26ba4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7985cd51280d80b60bf3708ce1f508abb26ba4c"}]},"branch":"refs/heads/master"},"7901b0f2df86f7fc1ffb507909b55391a3a40ea1":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2024-07-25 06:53:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/66"}}},"commit":{"parents":[{"commit":"6ec56f2cc709bc8efa7a1e1740ecf65adc157f7b","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6ec56f2cc709bc8efa7a1e1740ecf65adc157f7b"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-25 06:48:03.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7901b0f2df86f7fc1ffb507909b55391a3a40ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7901b0f2df86f7fc1ffb507909b55391a3a40ea1"}]},"branch":"refs/heads/master"},"c678749c0b688fd8b01f42229fda6a50f6a7b4a0":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2024-07-25 10:32:00.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/67"}}},"commit":{"parents":[{"commit":"ceaa46dde84a92843298dbbacee169943deff1ef","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ceaa46dde84a92843298dbbacee169943deff1ef"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-25 10:27:29.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c678749c0b688fd8b01f42229fda6a50f6a7b4a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c678749c0b688fd8b01f42229fda6a50f6a7b4a0"}]},"branch":"refs/heads/master"},"6f4d9c434679c7440b026e1ffef38bc7c73cc5f6":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2024-07-26 06:43:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/09/890509/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/09/890509/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/09/890509/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/09/890509/68"}}},"commit":{"parents":[{"commit":"cedf499e86573640f3ec6c8519a66afff6b50c93","subject":"Add api-ref and documentation for new location apis","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cedf499e86573640f3ec6c8519a66afff6b50c93"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-08-04 06:43:09.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-26 06:34:52.000000000","tz":0},"subject":"Bump Images API version to 2.17","message":"Bump Images API version to 2.17\n\nRelated blueprint new-location-apis\n\nChange-Id: I6e7bdc2d092c32c060d2f3c72fc83cecc73ee7f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f4d9c434679c7440b026e1ffef38bc7c73cc5f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f4d9c434679c7440b026e1ffef38bc7c73cc5f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
