)]}'
{"id":"openstack%2Fglance~896223","triplet_id":"openstack%2Fglance~master~Iffa8ae6102c547184fab4ed0f6860a5a73da150c","project":"openstack/glance","branch":"master","topic":"sqlalchemy-20","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2023-09-22 17:34:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iffa8ae6102c547184fab4ed0f6860a5a73da150c","subject":"zuul: Replace use of deprecated regex syntax","status":"MERGED","created":"2023-09-22 11:38:13.000000000","updated":"2023-09-22 17:36:48.000000000","submitted":"2023-09-22 17:34:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896223-sqlalchemy-20","meta_rev_id":"222b14313ea7a493699f668601df4e3b90ca965f","_number":896223,"virtual_id_number":896223,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-22 17:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2023-09-22 12:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2023-09-22 12:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-22 11:58:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-09-22 12:07:16.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2023-09-22 13:58:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7ed182afd0261bb7b51bf63d5bf89dcb3e07058a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-09-22 11:38:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"688d59c0cbe7608599ee252a82d4e720041b21b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 11:58:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dfb154281fe140df80059ad36449acc5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/167702c8f2ec462ea542506b2210d020 : SUCCESS in 12m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/683ed8b4468544519b5a1656a4a9123d : SUCCESS in 13m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3d8b5e6c1d2b49368a9a7e7ce7c12a8e : SUCCESS in 14m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eac4846a1ae445f98b224ee9958b6969 : SUCCESS in 11m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4a0643da59a5bf05ea872a097aa465bd2cc419f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-22 12:07:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4d8ffc3bfe4e0908bcea8ef8b58fc1596b2142ca","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2023-09-22 12:20:59.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04f62bedf0a45537d737eb5751b56bdb6a7d1535","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 13:58:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84f56aeb1fb4bd890b4a9f7b92673ac\n\n- grenade https://zuul.opendev.org/t/openstack/build/30831fe940984e8c9601322fea1773ac : SUCCESS in 50m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7783479c3ff438195a1a6e2615c6630 : SUCCESS in 1h 23m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77ccf73eb77d4c1dbc4ccc9f97e0f425 : SUCCESS in 49m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddfc74d9747b48b0af8be412830a97e3 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33d405aa15254c1a9399ebf0a38760ed : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a78812fa7f94891aa88db803b189c39 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a95f011781a426f9cf7d0b216cefd32 : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd11224adf554702ace2472b3f0e77e1 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/038fff2fa8ee46f290d3e66699a7b858 : SUCCESS in 7m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5ef561d1f1b9498da66dd636ef6c1f93 : SUCCESS in 13m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f083a378db04d40a0baddde433a1d5b : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d891d91eeab4389a3760bd1c39af757 : SUCCESS in 8m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/41cc12dd29cc4b95bb5ad368e6412690 : SUCCESS in 8m 45s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3f730da485ff41829ef2a431eea70f06 : SUCCESS in 10m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/68681da56a804abcb77cea488ec73435 : TIMED_OUT in 2h 03m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/79e88bca3f214c13a1ac39871a43bc59 : SUCCESS in 1h 21m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/01376b955e4444569035261122d69da9 : SUCCESS in 2h 02m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/04ca05b6b3a94cb1b51417989cab7299 : SUCCESS in 2h 05m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2704677601724338a96cbaef082ac53c : SUCCESS in 1h 48m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75f11657db664e889785aa7480fb2068 : SUCCESS in 1h 01m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb0d5201c15848e38f10b5dbebfb3128 : SUCCESS in 2h 18m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0cbe18ad167480cb9667f958c9662a0 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"08a668fbc94f074173c15306e20d880aaafb9fc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 13:59:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57cf742a3229c2dfdd7928379438b4e46cf74d27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:34:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f352ebaf6248eca6d12b13c5195e02\n\n- grenade https://zuul.opendev.org/t/openstack/build/cf1d9da0101c44fda3b6c172de6626b1 : SUCCESS in 47m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f2f71089cf4489ab4fb2618356d6060 : SUCCESS in 1h 14m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91f9eddf1ab7477ba820faa4bee17c1b : SUCCESS in 53m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee9ff6d8a1e741358dc32be416d44216 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3c085cd12954d2f8ad6cb331ad162c1 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce112dd84fc14780a4e0abc923ca7e0b : SUCCESS in 7m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55de48c9fe28442588f4449e0ed2184e : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/912c7dd0c0524e3788db9cb5ffa21d0f : SUCCESS in 7m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1db56c944ec64a3f83d204cfaa3c28fe : SUCCESS in 9m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/24e291a95ae8435ebf594bc40f86b60e : SUCCESS in 9m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4caaadc2188c4335afc1dec37b0485ec : SUCCESS in 2h 00m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a65ef655f9cb4e67a5a5e8f099869a65 : SUCCESS in 1h 54m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c823760455d542249c41b4a39da0e7ce : SUCCESS in 1h 48m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dfcfd33704b7486c9a30b08d6750f2f9 : SUCCESS in 47m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a00219e2717749c6a041322a10d54054 : SUCCESS in 1h 51m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6c13bbc021a24eefaf93754212e6530e : SUCCESS in 36m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d664e990ccbd50d211962a9cb63ffcbf16acd12e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:34:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"222b14313ea7a493699f668601df4e3b90ca965f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-22 17:36:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f457fa8cee34225b62654f48da6c0e4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d6d298088d4aceab0eeb6f70b2be55 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b29dc9bebf4fa0393a170fcb1fa009f691b728d","revisions":{"0b29dc9bebf4fa0393a170fcb1fa009f691b728d":{"kind":"REWORK","_number":1,"created":"2023-09-22 11:38:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/23/896223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/23/896223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/896223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/896223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/23/896223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/23/896223/1"}}},"commit":{"parents":[{"commit":"78d0c8545d4c0580275aa31bc6231f4f32e95799","subject":"Merge \"db: Trivial parameter swap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78d0c8545d4c0580275aa31bc6231f4f32e95799"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-09-22 11:30:02.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2023-09-22 11:30:02.000000000","tz":60},"subject":"zuul: Replace use of deprecated regex syntax","message":"zuul: Replace use of deprecated regex syntax\n\nzuul now says:\n\n  All regular expressions must conform to RE2 syntax, but an expression\n  using the deprecated Perl-style syntax has been detected. Adjust the\n  configuration to conform to RE2 syntax.\n\nDo as we\u0027re told.\n\nChange-Id: Iffa8ae6102c547184fab4ed0f6860a5a73da150c\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b29dc9bebf4fa0393a170fcb1fa009f691b728d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b29dc9bebf4fa0393a170fcb1fa009f691b728d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
