)]}'
{"id":"openstack%2Fglance~899812","triplet_id":"openstack%2Fglance~master~Ie63d74d69b86149e4acb5af4b96c69f634cdeaf6","project":"openstack/glance","branch":"master","topic":"New-Location-Apis","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-07-07 13:24:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie63d74d69b86149e4acb5af4b96c69f634cdeaf6","subject":"DNM: Test Image status to \u0027importing\u0027 after hash calculation","status":"ABANDONED","created":"2023-11-01 07:12:04.000000000","updated":"2026-07-07 13:24:36.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b0a639b3825e373f6664ab555888cb5386804a0e","_number":899812,"virtual_id_number":899812,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-01 09:14:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-01 07:18:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-01 09:14:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-25 14:01:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"ed9215d67beab042de65fa43da15dd2a135aaf35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-01 07:12:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fba55a4e77778cba2e0b4792b9c6001a372f5303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-01 07:12:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0cd57a04f53789dacacb9a00b666ae1dda4985c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 07:18:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e1e508b831e45da8c85c3d85a5c9e75\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f988c7c9d5784aaa90264b1f8ca69b55 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0add2b79b1ab42b79172304c8645e647 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79fc4f5120174a87b76ada30e84f5e5c : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"54a6e3cb3063358da6b1b0ceb32b2e02693f2ff7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 09:14:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a9ce8b3a4b4e9d9b24731768f08b2b\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f021b9a69d94430a9ac0c660c6fa700 : SUCCESS in 38m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d0c6221b925b4225a5095f211e978200 : SUCCESS in 56m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f196a254bdaa493b8238a428a883527e : SUCCESS in 1h 28m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7cfcd78093594eeda3a8c8c12e2f7064 : SUCCESS in 1h 00m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8c1474a7fc04dc6b020c3e653f9f3cc : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ab88b70ade644e498af49d4178b71f1 : SUCCESS in 5m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/995756a6b2cb4679b5425e1f718c3efb : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00256a6b669d45169364b836a800b485 : SUCCESS in 6m 59s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a707e240538044fb867818c110c04f4a : SUCCESS in 14m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a34d3ce63934207b678a339582b63ae : SUCCESS in 9m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5ab9e87ec6ae4392b81bccaeb0462686 : SUCCESS in 7m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/42b75844c0014ef0920220e87e6db2e4 : SUCCESS in 8m 57s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1aabfcb125b14aa992fa72670493f586 : SUCCESS in 9m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/39e79d6a6ade4a79bfe5c17b2e47e59a : SUCCESS in 1h 48m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/74bc6b3cd1ed4f23882e092c83737561 : SUCCESS in 1h 24m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0370821c5fcf4facb2f09e798c3bb7b7 : SUCCESS in 1h 55m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/de05327b351f4b598ea9b73f704c3dbe : SUCCESS in 1h 21m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3675c48da2a24c7c9785624a2e1749ae : SUCCESS in 1h 44m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fda0ab2cb6ca45d0ae45b4b97bfd827b : SUCCESS in 1h 02m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c255c978001646df83dfd713c7737ebe : SUCCESS in 1h 17m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04201b8d017447b4a7506d035a16567f : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1eabfb78136c87e0705d37bde01d9c85c2561ba0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-25 14:01:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0a639b3825e373f6664ab555888cb5386804a0e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-07-07 13:24:36.000000000","message":"Abandoned\n\nNot needed now","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"14bf53ab0f868faf8e1347953008347f25bbf068","revisions":{"2d334809906a155c268be6a47658468e27959ba9":{"kind":"REWORK","_number":1,"created":"2023-11-01 07:12:04.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/12/899812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/899812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/899812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/899812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/899812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/899812/1"}}},"commit":{"parents":[{"commit":"d3036822d07d333e311928b791bb678c3fcf8c86","subject":"Bump Images API version to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3036822d07d333e311928b791bb678c3fcf8c86"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-01 07:07:54.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-01 07:09:07.000000000","tz":0},"subject":"DNM: Test Image status to \u0027importing\u0027 after hash calculation","message":"DNM: Test Image status to \u0027importing\u0027 after hash calculation\n\nChange-Id: Ie63d74d69b86149e4acb5af4b96c69f634cdeaf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d334809906a155c268be6a47658468e27959ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2d334809906a155c268be6a47658468e27959ba9"}]},"branch":"refs/heads/master"},"14bf53ab0f868faf8e1347953008347f25bbf068":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-01 07:12:42.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/12/899812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/899812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/899812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/899812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/899812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/899812/2"}}},"commit":{"parents":[{"commit":"ae4281046dbe9e04c90a360f887f2c59b9b09986","subject":"Bump Images API version to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae4281046dbe9e04c90a360f887f2c59b9b09986"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-01 07:07:54.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2023-11-01 07:12:42.000000000","tz":0},"subject":"DNM: Test Image status to \u0027importing\u0027 after hash calculation","message":"DNM: Test Image status to \u0027importing\u0027 after hash calculation\n\nChange-Id: Ie63d74d69b86149e4acb5af4b96c69f634cdeaf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14bf53ab0f868faf8e1347953008347f25bbf068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14bf53ab0f868faf8e1347953008347f25bbf068"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
