)]}'
{"id":"openstack%2Fglance~899868","triplet_id":"openstack%2Fglance~master~I0c1affff05a2a4c2838cfad2b637e7660b056eef","project":"openstack/glance","branch":"master","topic":"centralized-cache-db","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-02-14 10:16:06.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-14 04:34:15.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-02-08 15:24:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}},"hashtags":[],"change_id":"I0c1affff05a2a4c2838cfad2b637e7660b056eef","subject":"Deprecate sqlite cache driver","status":"MERGED","created":"2023-11-01 14:58:19.000000000","updated":"2024-02-14 10:19:21.000000000","submitted":"2024-02-14 10:16:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899868-centralized-cache-db","meta_rev_id":"9b3d1b6cff72bec026defea1801cb2fb50235e6f","_number":899868,"virtual_id_number":899868,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-14 10:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-02-01 14:27:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-02-08 15:24:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-02-08 15:24:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-01 15:06:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-01 17:08:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 10:35:26.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2023-11-14 19:36:23.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2023-11-22 06:37:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2023-11-28 15:02:41.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-02-01 14:27:49.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-02-08 15:24:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"2809719777746c4c71e9beb308aacb538c21721d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-01 14:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb7cff7f8c082ad1efaa0dda9f3e53ebd8bd4f59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 15:06:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d770dd0c18492e8d0ab9b33c13d26b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/222d6ddbd4144ecc9e9482145b4a54e8 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b0390fb54f96463586cb091ab11e5582 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a7c59e75646495595f50a6080f792a5 : SUCCESS in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e5d05def33cb899c8fb6ed8fd07c0467a7e438e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:08:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/590cc2c5ab5d41478686a269d7ee5297\n\n- grenade https://zuul.opendev.org/t/openstack/build/78246abf14514195b0aae04eb1d7d38a : SUCCESS in 1h 01m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a4edcc773d1249d39111ba130537cc12 : SUCCESS in 46m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63ef1bb8386d45ecaf42e020eb23e463 : SUCCESS in 1h 26m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe535b25b14440c9aef36c29ef9413ad : SUCCESS in 1h 01m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9449bd968b5e474496283fd249b53f0c : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02c3749431404aaea2d84fcd8a6ac6a6 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e2de277f6864a5a98ea79ecf864c182 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8077ae3901b943c5b5fbaf13c321bc49 : SUCCESS in 6m 48s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d93b2d37b9dd416384b5b385b0db0d6e : SUCCESS in 14m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f9976b354544de6a08c3d8c63b3cd5c : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57517a7b75a946068c61f87a759a86c4 : SUCCESS in 9m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bac17fda04644d0d889c161b3292427d : SUCCESS in 9m 02s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/eca3b99e06af485a8240be9ba0f8b028 : SUCCESS in 10m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cf778f3a94004a609f8af70adf13ceec : TIMED_OUT in 2h 03m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/20de956f34274e4a91881968a018848a : SUCCESS in 1h 38m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8a0fc185f5ce4dc68e3f0e814b863cbf : SUCCESS in 1h 49m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/00a4d98b98c0445c8c9ed0699a8b5250 : SUCCESS in 1h 47m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/417509a6a5674931b0933a7ea4bdb49c : TIMED_OUT in 2h 04m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/285afdcd6e224a31afc4f6a4417eb01b : SUCCESS in 48m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f27bc94eb60847e19a48ee92d3550a3f : SUCCESS in 2h 03m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b42b38a3e98a45738e0dc0637962f3ca : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":1},{"id":"73c9425edda73815d6cbd499e37311903231e24e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-02 07:29:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a236fda435e0a4d2598677e22922c50752df36fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 07:39:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a555df6578c34a609c74eb46e919a73e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/96dc8b3b71864fdba6a0cf563e651396 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c2c1eb5db7847528360977f476e4ce0 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/877be8f412af4bfa9a314b1d56e02dd0 : SUCCESS in 8m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b96877ca9715a35b3a094df1619d004f8e271376","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 09:40:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bdded4c58213457ebf6942e14396367c\n\n- grenade https://zuul.opendev.org/t/openstack/build/ede7c8a04ca94291b56d043a71e25a1d : SUCCESS in 1h 08m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d802c5c6778e43c9b02e2f3f4809665e : SUCCESS in 1h 05m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2d49bc9366c45fc8d0f4e4e68b085a8 : SUCCESS in 1h 27m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f72561a173a4eb79bca1b8de5382196 : SUCCESS in 1h 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2dd668a804d4101a021655b66cc086e : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6dc9b9a04f9d4db281fc48851a57265d : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ad416c0aa3443cf98aec89b3e5c37df : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7dee7166ffb48cd91a0a955af9b697b : SUCCESS in 7m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d6688fe554854ce2b51beebf026ad790 : SUCCESS in 12m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/83d7237634aa43cd8cbb8d97a18b8ef8 : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/59ab0237711e48ce9cd75b29cf2c5d6a : SUCCESS in 10m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4eb38408cc0342da93ec9974b485b8df : SUCCESS in 10m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/35fb1b0bb27349fe91f0283dfc933601 : SUCCESS in 10m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/17e71173cd354cc5aff97188a0403e30 : SUCCESS in 1h 38m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0c7d0bb446604d09b660b96701b164ba : SUCCESS in 1h 07m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7deb29261daa4d6eb05700d56f3f8fa1 : SUCCESS in 1h 28m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5c7ea857228541ca9cb22685c2569b4a : SUCCESS in 1h 36m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0a7cd3ebe7404bce9a8f7e3aae992e2b : TIMED_OUT in 2h 04m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dec9332f4bf84bfa9428eea78547cc77 : SUCCESS in 1h 10m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62db8895eaf94f9f97c8470bc4e7d55a : SUCCESS in 2h 02m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/79781f9634e6452ca34c92f36efee685 : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c83fc6ad9a9180c00ea63ba3eebc5388e9fbf308","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-03 09:48:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bcebcdfc54e2713421c7e183c12f7e804347f003","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 10:02:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e23b080e0884b67831f829ca66b41ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42fbb5a80de44c538d0f43b9c4a494fe : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/484ac5c8278f470c9f0b0def4c724320 : SUCCESS in 11m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7eaf4bfe681e4969a27eb0cf91e12b6b : SUCCESS in 9m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e61fe6652697357ef2aa75172764982fa1ce1067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 11:48:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9db70edca76b4848bfe5efda6371044f\n\n- grenade https://zuul.opendev.org/t/openstack/build/846114b443904f34b14562bfdcf90b44 : SUCCESS in 1h 05m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/599fd94f7c1e4961984f54fee857e39e : SUCCESS in 48m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7de0b0708127400b8f46c1c880209e36 : SUCCESS in 1h 21m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/643c78921d434a2ebc29c03094020fac : SUCCESS in 1h 06m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c749935115f4a77a932270b8b429ae6 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aab08be36ec64de79f14bb0e3226b846 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/222bff202b8d4d9f92b99e4c1c94b832 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4c26246f9c47dcbf792a2525e2f033 : SUCCESS in 6m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4a2261a456654c7984bd96c7f6d83011 : SUCCESS in 13m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4329e5c9c5524c6e840b7bbc447577b2 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69ebfa7dbcca4e8985d73ea9624fc510 : SUCCESS in 8m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d27501ae042e4a638afbd85d3450b60f : SUCCESS in 7m 23s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/63c4ff01cc4a4aad94191b1bc585cbe6 : SUCCESS in 10m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b600618ca7ee42bd806a209aa749a99a : SUCCESS in 1h 47m 47s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/adadfb07f5b34419bfe3d9915a9e6756 : SUCCESS in 1h 51m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c72aba2e2bc4d96864cd593085f6d11 : SUCCESS in 1h 40m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1e7f43ae71fe479290bb716054ba52c9 : SUCCESS in 1h 46m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9669bf3ecc01427e9d0a0c8cc4d74015 : SUCCESS in 1h 42m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9b279f5551f4a929ec3e08a7c3817cb : SUCCESS in 1h 02m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e7a94208d48c496c9c5a975f753afc31 : FAILURE in 1h 43m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6fb3187c2f204a27bdd38adc78115903 : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c5849447eaa17f3dd20fd2760f99c251a8340ac4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-07 05:01:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b4f15ec1d82a9820c48fbf33ba42e5e7d125221e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 05:14:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00ae1cc407694c91aba3e34b46e12adc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/462aa12dddb14a25ae31ad9d204baf55 : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e609dd0f4e664481aef5f1db13ef39ad : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd33637411484490aab5e9ab20f4365d : SUCCESS in 10m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a8fd7e27ebc352c77a71b152366f3b58e7fd16f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 07:29:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff4b01ab0bf746969f2076df67e32eb6\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f2befb4783a4bfa9759123729100f2b : SUCCESS in 1h 06m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/55111f87c28a4b02971f310751083384 : SUCCESS in 42m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c2570bf70d3463e9753e0bc2d00ad56 : SUCCESS in 1h 22m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a9ebbe054c5a4ed0a1b3c39d84d8df99 : SUCCESS in 56m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d47fcd7f7f2a4ebf93642551831bd6fd : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5447e4f07c147adad6f8cff08c3d727 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a909a45d921746c69068c792a9a3ea38 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c02ea2f96ae8479f8c53d836ae9b6947 : SUCCESS in 6m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1f9bd24c0ee446eebb0ce4c3f51991bf : SUCCESS in 11m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/00dfa52432614661a205e05fc1bf5a7d : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/896ce2add4e34b61b20552a9955b1202 : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6a42d41fd13c4d62b724e0363e048188 : SUCCESS in 9m 13s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e450835c765f46dd95ae60a513c49468 : FAILURE in 5m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/318d46043ee74ee99d2f2042a74acae8 : TIMED_OUT in 2h 03m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/706a944dc1a843b98ccf7d16a8cf5b40 : SUCCESS in 1h 14m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f7f44b5bd3c44199a9b93b1a843376c5 : TIMED_OUT in 2h 04m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ccc5200c26854c6cbbb5e41fb0dfce91 : SUCCESS in 1h 18m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31bbabf9e6e24618bfaa3ff0539c8bc4 : SUCCESS in 1h 34m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ef61370257c4189882503e05ba95735 : SUCCESS in 1h 04m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/555ebf3085474809b6c505476fe2fe4c : SUCCESS in 2h 20m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/748f30f183c64b1ca7737e798cc52940 : SUCCESS in 25m 05s","accounts_in_message":[],"_revision_number":1},{"id":"081c737d8747de44669acd1cdc322e347b09d390","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2023-11-08 10:35:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4fcd3963b3d532448326a594841bd190103b2737","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-09 15:21:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb91b19f564b9c7948c3834bcaa408f663a0d5b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:34:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65164cb82ef44753869a9e0136f00af8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa6cc4ab6667477784f532651ed25719 : SUCCESS in 9m 21s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba8233a3f3134cc7af349879c3cf8cb6 : SUCCESS in 9m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/79b10700f28f4e90a1ce9117b567f16b : SUCCESS in 10m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"964584d3512843e66dab0740633e94fa11dabfde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:28:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/725619296f6040a2a718536f945b6b0d\n\n- grenade https://zuul.opendev.org/t/openstack/build/d6f0639851b44025bef7cc1bdedc9c16 : SUCCESS in 1h 06m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4d41608e7cf543f7a3cb315bbe640eee : SUCCESS in 44m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bafea4bac3b4ddda6155c429c2bfd58 : FAILURE in 1h 14m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50b94bd3f472439d95d152a71259b61d : SUCCESS in 49m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5246c3dce774f60a1f66513043a0359 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a7bb522312745dcb0c9d88931de719f : SUCCESS in 5m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb3c1b46223c452ca331904e5a59d952 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933a2032b38e4b3895fc60fa586f431a : SUCCESS in 6m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c917b80698904e91ae9c8c4622c22540 : SUCCESS in 15m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7674757995c04d469a1ce5c1fdfd51fa : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f26528bc571141c88ce18bb8b47574e2 : SUCCESS in 9m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2e0405315a8f42d7834cf9c2c9dd810d : SUCCESS in 8m 58s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c9bac1cf50f54391bcc4bd8fa619a314 : FAILURE in 7m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/caf244aa7b314092978823e649e7fbbc : SUCCESS in 1h 57m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8a2df29a8bee4b1093d420262ea2ba30 : SUCCESS in 1h 47m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3cf318d6bce947c186be59d64942f6d5 : SUCCESS in 1h 59m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e6905e922dc34d558c1880d741ba026f : SUCCESS in 1h 32m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d680eb23700f44fb950c46c33278b3c1 : FAILURE in 20m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f75f24538b8a4530ae186f711ceb5790 : SUCCESS in 49m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0967f6ed712445258a1825bc0801ae89 : SUCCESS in 1h 57m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/87eac30f8dac495aaf0f72b39cd11bfc : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3e8dc1fe6c58c6ff12fcb3e88a1df537cde4e723","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-10 05:03:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dd0d5924cb196e14ff662aa52e3b502fbae8d25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 05:09:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6ef189f9f144bfe8a5a4c6cbd9b327c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee4dac873d7840f9933a662648d6fcfd : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/21e1a8085d684476a6ec80a6acf0ae61 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a4a007ce36240a98eeeb9493e957c3c : SUCCESS in 5m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8f13b12b040144fe4a6d732ca862c1169fe7066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 07:14:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5cfb5e00bd245bba0783ab98437ecf1\n\n- grenade https://zuul.opendev.org/t/openstack/build/a76b2efa1a064bc19188653d565a067f : SUCCESS in 39m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b919de9c3a8646ce9fbc8583b25fe3b2 : SUCCESS in 38m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0fc4b29ef12641b08e9f36c4dae3b157 : SUCCESS in 1h 01m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/003908dc7c314fa2b361a4d265835561 : SUCCESS in 42m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79894981ad1840988b1481700bc674aa : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ba77b0ffcc84a2388704a2fb10752a7 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98025c8f32b74e8ba8ad6b6186fae2e8 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e0ac3745d074e0087a0ea51840cac78 : SUCCESS in 5m 40s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/461de6f2bd294756bbd2e1421936566d : SUCCESS in 10m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea08e9401c8e4b6987d60b87abeb473c : SUCCESS in 13m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1493c10010a4929899c3c85132bfacd : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/eacbad987b324b6aaf6ccabd23985654 : SUCCESS in 8m 00s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0f4f15ea9ea046f1a223ca55fbb3d134 : FAILURE in 5m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a818e449640e449183bff21f1717b276 : SUCCESS in 1h 58m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2a30061ed4b3404c8014858ac0570391 : SUCCESS in 1h 46m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d0ab3c1a2c6a47fabcef42e05ffa4917 : SUCCESS in 1h 42m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7ef22dac50fa4a1cb918b754b0d938ab : SUCCESS in 1h 46m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9a3c6b5fe92d429fb4eaaf158da3607a : TIMED_OUT in 2h 03m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37459912685e4a85a4fa6c45b1ddd4b2 : SUCCESS in 50m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ef83ea592284833bc3b9d7884c0bdec : SUCCESS in 1h 57m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ed9db65580ae42429d43b0f7252060e6 : SUCCESS in 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7fd228917a64d2f6e448b632c234fff4d5dbd8c5","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2023-11-14 19:36:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6a298f0081139b503abc910f2481138cc5e6b42","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-16 05:26:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e177ccfd920d1b69239814c114973eec6eefb4b6","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2023-11-22 06:37:15.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"465c4dfaf0835885e2a93a6374344a5686e06854","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-22 07:13:35.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a906ac7f4ff7bb2dc1413d2bc339e2aaf0355907","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-28 14:21:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99c140ce2b0faa4279440641d372461504a57fd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 14:52:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad25a5666bf04345be5b6a514b502fe4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0c0840ee1e04305bcb1a9226ddcad2d : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b34866587fae476aa2dc8c90edac94b4 : SUCCESS in 14m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/25ff6d4a7dc54c2db00f197acc6e7764 : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fd2b069861dfcb742099df2706c9e9e4d9a8b262","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-11-28 15:02:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56a12390529e8587fba2435057662367c37d0f2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 16:34:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5a4f3c051454813aec1cad57304bc98\n\n- grenade https://zuul.opendev.org/t/openstack/build/337a06cdd5414d289e8b9af027d8ba09 : SUCCESS in 52m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4e9c3df9d4514894bd43b6ec496a58b9 : SUCCESS in 1h 00m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ffbc838eb7be4aea80270039077812af : FAILURE in 1h 16m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39178e973c2d4ef4af646993d13d6ae6 : SUCCESS in 1h 08m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d93d5445488e4e699bbca66a4a79f59e : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb9da154ca5946c9959cebc12e10e1e5 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc05a984ec2042409b96186017d45613 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4041cad4fe5b441f9a523962aecf9257 : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/c508a2c5d37749c4afbde7336dd0fe45 : SUCCESS in 12m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb29d8be128242dcba2679da8272eb9c : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/581b1b3161f24eff9472ca567ca2aa7b : SUCCESS in 9m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4af14dfed355498c9c6a990d6d40c4f1 : SUCCESS in 8m 32s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/836982e8277d444da4b57e13c91d3af4 : SUCCESS in 10m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/21f38f95a4d543a38346084c49128626 : TIMED_OUT in 2h 03m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0cdc7c8fbfa143a2bc3f520397603741 : TIMED_OUT in 2h 03m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d55774211cbc4e60b0c03f66dc45bb0c : SUCCESS in 1h 39m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c71490849a2d42d2ac7af3466f22a3b7 : SUCCESS in 1h 45m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e3f2e41b751a47ed8c8adae6ca4947d2 : SUCCESS in 1h 56m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbc5fbbe630541acab3680e99730434c : SUCCESS in 1h 10m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7250c38628334b72acb51f3f4ec209a8 : SUCCESS in 2h 05m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/26ae72ab0e6b478990dea33bb7268bc7 : SUCCESS in 27m 12s","accounts_in_message":[],"_revision_number":1},{"id":"13a021eeadfd1cd77b3d8a938812ddbd6ea5d537","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-29 07:55:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"361f92fa2ff2e2beb9c55426b320d64d7d022193","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-01 04:35:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfe2d16bfee78feaae25075e09fbaa1e4fc4e248","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 04:50:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a04fcc78b6a4b9c97ecfd2ee4680237\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2c9ce48f94674c43ad6bcd91963e7d7e : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d8e8780b43304d2d92e9cd3b3f1f2e7d : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eccc1b28030a475eb6398cdfcde8fc3c : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d401cd55db45b7d59db69527c882576dbdb5ac7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 06:36:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af9a29c0fc1d470fb191315585fe390e\n\n- grenade https://zuul.opendev.org/t/openstack/build/361989ecb3c04c84bd3e34fcf736ac1b : SUCCESS in 1h 05m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6c7c918a93ac48d18f9ab996afbf40d0 : SUCCESS in 47m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e63575ebdb944081b48dc1c212967ec2 : SUCCESS in 57m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/687b6dfa40bc447d841321f019c41cae : SUCCESS in 1h 04m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56999492a0dd4c87b9e518bdaa7c3913 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90d57a3a7d1944a6a9236fa60583d63c : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8da64e195cb48c680cf8d257f75f516 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58047b5851a14945a25c8e928dff86ea : SUCCESS in 5m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b9e96dc8109241dbbd3734d1996b42e5 : SUCCESS in 12m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e3dc1c5708d048bf872315b414c05bff : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3d568898b5a54dd2b16461cbea2dd4ce : SUCCESS in 8m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/80c9a61760674f36a53ec42d06919d7e : SUCCESS in 8m 19s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/04a67e1dde00499ab21057dfff5095c7 : SUCCESS in 10m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/376392dfea4e45eabb770570240c0c59 : SUCCESS in 1h 24m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c9481199c1ab42e39465ecfbfa63977a : SUCCESS in 1h 47m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/545cff6422134ad195e2a342f29bb7a8 : SUCCESS in 1h 54m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/142149c2b81c4758be9bedfae8f3c614 : SUCCESS in 1h 11m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b67c83f1dcfa43b6a35d48c0af8a7569 : SUCCESS in 1h 19m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d4933a4b4ad49ab8f84ebf7fc7e3fc1 : SUCCESS in 47m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/97fe4400251048a481f9398d05ae1463 : SUCCESS in 1h 23m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5de0afa6054c49f2b8f957bfbd476a55 : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":1},{"id":"72286e1b1fbaba6f5da5b293fd79d691f2eaf4af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-07 05:23:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7934b7a720f3c82c3b8ac1753eab8214dad004d8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-07 05:38:38.000000000","message":"Topic centralized-cache-db-1 removed","accounts_in_message":[],"_revision_number":2},{"id":"c8809909c6f38750b845a4c4eb4192330f886cce","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-07 05:38:52.000000000","message":"Topic set to centralized-cache-db","accounts_in_message":[],"_revision_number":2},{"id":"4bf5f4217b112ddd47bb3f4369f45537dabf3443","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-07 05:39:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9deec3d868f42499497d8bcb61c5b551bc49b1a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 05:39:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a167df8ed4f942a88b57fcfa3856db5c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/90535456249e4f9eac15d98930d1b999 : SUCCESS in 12m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aaf24c15a1e44b8fa9fae13ad814f6f9 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6cb378d42d7844a3b7085679875cff34 : SUCCESS in 12m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"19120d8e6abf8ae7a6aa68bf9c440e5d68eaa38d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 07:44:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed9ff309b5474a01905914c73a8ea50c\n\n- grenade https://zuul.opendev.org/t/openstack/build/1531e4eb916a4788b956931cb4a3a65f : SUCCESS in 1h 05m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3c334d0058144664b7b08d9e82c72d14 : SUCCESS in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f85babbc9fa44191bba0a54e6cb49a8e : SUCCESS in 1h 28m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc965c1b88944d65aab37ee6ce5d9761 : SUCCESS in 1h 07m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bcb5d9f05bb4d69b7a83580f470a20b : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b462481680423f939f920302405f4f : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b22a70b3eb24d759f173780dd14c43d : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5627f104674943be98e6a1effd43e488 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73fc90f8b22c4e94acb2521cf20fede0 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8632e43cbf7f41fcac4fbc6e4cf95907 : SUCCESS in 14m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/816a9f5dcf9d4d679555cd4993aee7b3 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8d3e0d09243468da1b04fcec4fd3a30 : SUCCESS in 10m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1daae056d700497e8c6f100d542f8aed : SUCCESS in 8m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6f9cdcfffb31403b87980455f8b02287 : SUCCESS in 11m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aaaa1785f7d44b6eb5359f9400dbd0a5 : SUCCESS in 1h 28m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e7b79b1dfcc748c294dc790aa0a3de28 : SUCCESS in 1h 57m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bddd29231f1e44a785e818e9e288fe3b : SUCCESS in 1h 31m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a6fbb25c22614b7e91ae20f2a5b216c9 : SUCCESS in 1h 22m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53a324db8b3542a6a3bfc5a0dabaa449 : FAILURE in 2h 11m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df321b93bb674f1d9deb5f5e8a9022f9 : SUCCESS in 1h 03m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c85cfeb86714e399d0864b853e7fc63 : SUCCESS in 1h 34m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2c7e0b4a72d544d4b7c9a92cae9b7fd2 : SUCCESS in 30m 21s","accounts_in_message":[],"_revision_number":2},{"id":"942f723a6973cd89c843d55940c5b6f83794d462","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-18 06:44:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cd7609206582823f1c57d984a5cfc26428876e39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 06:58:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afb9475efc4547a48c96eafc565aa97a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cb98299a16d4e8183fd46d552b4bb47 : SUCCESS in 12m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/149c80f2663943d787eff86501296a18 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cecd540414174463bcf9cadfc22fc124 : SUCCESS in 11m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"caf153c38b369e66c5f0cc744c53a54b07895802","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-18 09:06:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0216ddffb52644a08d255b6dfd08edcd\n\n- grenade https://zuul.opendev.org/t/openstack/build/716d467d314b48b183ba4d2253a37a94 : SUCCESS in 1h 10m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/182bd1921ce5484383958c0a59402563 : SUCCESS in 58m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60a5d7efd3f9459ebe40ccbc3a20678c : SUCCESS in 1h 08m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b3352e5e6ae4f77924c05be6fc1cab1 : SUCCESS in 1h 09m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5851cea7b3fe487996f48f0eae9298d2 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73c3109c60774f34aec2707799deb53d : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/694c73934efa4f1684c7c3edecaacf46 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/571df7eef61f48ce8935ced0e7e4261b : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2c4ac0f506a47e7852b125536db7d54 : SUCCESS in 7m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f505b28b8dfa4f9599c63c1d01eac090 : SUCCESS in 15m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a3ff9ad043cb4ddfb403d7a704951964 : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d044e637919645a1ac2a86a4107b898b : SUCCESS in 8m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bb5d652dfd9644649e23a859f5a3b64a : SUCCESS in 10m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f16eb93d79f741ecaa1b7fd4e685e178 : SUCCESS in 10m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7097e0ca7751462ea91dac7ed466a793 : SUCCESS in 1h 51m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7fc5a45457794965bd44cc0c73362f13 : SUCCESS in 1h 37m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/afeac62f4b2344079e716ce47d3f88a9 : SUCCESS in 1h 57m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ff821638322448c8ba575f461185030f : SUCCESS in 2h 01m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0cd882f04293413d9f00148cb25faf3e : SUCCESS in 2h 03m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfec8e90eacf41abb0f95d05acac4b85 : SUCCESS in 1h 06m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ade5f6bbf26a48a2b08236c1c5746bbb : SUCCESS in 2h 16m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ad5786163d44685bfd7af374a360721 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":2},{"id":"4c2708897abfb2a5400a9d17e6ce8cf0372da3ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-16 04:12:07.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b6ac888f683a35ad16534bcfff3447cedb93a264","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 04:20:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94712c4972c740fdb738912a48f02b3f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9623e792d4d14bedaad87358a2bdbc37 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18f485ca4da043c38bb8d7bfc2f403fa : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/07b30c3bf8ff475c9092d8f313b92649 : SUCCESS in 6m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0ef6f2b578ee0aca890c1db6a0a3fef5401668d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 06:24:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b5ebf17c95b4d4395b99ce124c09f27\n\n- grenade https://zuul.opendev.org/t/openstack/build/019a5c99ca3f4c3eb26223e8500b7237 : SUCCESS in 49m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/64fe76548f904569b34f2e7a7c0fe841 : SUCCESS in 1h 03m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6f54f7d9b2e44b2d94810f070d008578 : SUCCESS in 1h 28m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9a08de85b1a4371866fe73334c0c806 : SUCCESS in 30m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6315e649723d4aad9fe9eb78d75291eb : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/079055b0aa904e979644d78cacdeb7ea : SUCCESS in 7m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae7a5d0327a04f5a8a2cfc103b2a0609 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dd40fb33e544242a8e0ba6f9bd8e826 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51063bba39dc4f4ebdbb4b991795b243 : SUCCESS in 7m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/6226b0cf127746f09f78d59a23ece064 : SUCCESS in 14m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/132a9d7c05cf4e2f98cbc5af1439281d : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f70a52d9fe44390ad4454e94f12fd3d : SUCCESS in 9m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7c131096185e4dd18f87c008a925bb64 : SUCCESS in 8m 45s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cefc9cf5e89346ccb73f0290de843ecb : SUCCESS in 10m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aaa37b43c6e34dcf85d1f78bbee207f4 : TIMED_OUT in 2h 03m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d5de08b170404601aebbea96a6d0b116 : SUCCESS in 1h 46m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a5ef954d4fcb4c6db43a0167afaff8c9 : SUCCESS in 1h 56m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a86f04d8d7ff4db086c6f394fe405614 : SUCCESS in 1h 10m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/08fba25d57a7430480424f98c61b947b : SUCCESS in 1h 50m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17c25dda627f4b319e2aa1579a639d7c : SUCCESS in 1h 02m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5d38f1575f9044b48edab1e193907491 : SUCCESS in 1h 48m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb60edb606184fa78593c78b5d1c709a : SUCCESS in 39m 33s","accounts_in_message":[],"_revision_number":3},{"id":"4d4068efb96651f98973ff2bdf143dd24442fd5f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-25 05:18:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d51d582ad37909d7fd563b3ab465a73c6fb83a0c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-02-01 14:27:49.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4e124904e7b25eb88875e75fb60d2f1e182045fa","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-08 15:24:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"83ecd7ec416951a3d3eac57a2cc31efd6767589b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 15:25:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7ec3c7eb4fd0bec86911a8a793a2de960e66d957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 18:42:25.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8e9ffdf03814efbae2057465b1a53f3\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff99a421723949259a8827798e89828e : SUCCESS in 1h 03m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c558b09eb1344378b8d40e036c033a88 : SUCCESS in 46m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92bf02a7a39d4684a78c2d0849539efc : FAILURE in 1h 15m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b075dc602bf4fc9a309b64b0f6f9127 : SUCCESS in 36m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af8ad8a0d13b475e8416f7c9a0217c85 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/648c4007d784455395048a1cdfa03ca8 : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6602ffb42f140fcb1825fc5e8a539b5 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a058d451be114e14841c47000fa6e15e : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c17f4c68e7c4d53b302cebd5f31c29e : SUCCESS in 5m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e3e557bbdb141108b7a8a20ef8441ec : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/bbf4d9084d3b42c2a15e499c36f48be1 : SUCCESS in 10m 09s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1fadcba61195445ab908836c1099969f : SUCCESS in 1h 35m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/48063a9b4fa9472d84d79995952b3022 : SUCCESS in 1h 37m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/95ba48c64d714d06a1bdee5fb44f47ef : SUCCESS in 1h 44m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aae0ee57ef0b41789ced7b6e39f62e73 : SUCCESS in 1h 06m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ebfbafdd86b74563bf05c4e4b5579438 : SUCCESS in 2h 09m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/37d8686305b54a25aa0c8bb13849ef19 : SUCCESS in 38m 21s","accounts_in_message":[],"_revision_number":3},{"id":"256f043fe7bd8afac972e01cbeb3738d213949e8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-09 14:45:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f693fd73db5d6f9a52122f48ef630e9a82fea657","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 15:00:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a1d411002db4d449662e71e589ffc33\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9dd7d20940934e228cdb14eb0d2b3e1f : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c186eaa83e2434696ed9b23dc89b330 : SUCCESS in 8m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af36c4e92bd84e2f8a50cc10f33c3399 : SUCCESS in 7m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"57b7d4f81d23149006b54a9b2ad30ea5dd11f45f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 17:11:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e71152b0f0f2441facb43279bd19a6bb\n\n- grenade https://zuul.opendev.org/t/openstack/build/7dbb42103592459baa723370d792f450 : SUCCESS in 58m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5bed96fe6d364ed6908bd73d80e0380f : SUCCESS in 58m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/be54f38227cb4f49b6538b2a2a3d2e36 : SUCCESS in 1h 38m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/729c021e0a85442bae77be996825114c : SUCCESS in 34m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff60eeeed9ac4a8ea3163d982c548dd6 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bdb2df1ccab4f16bfdf37daec5ed534 : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3367efa262b04d20b2c8dccb109d7521 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517eec659600401b91868cbcca032cd9 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21731f6639df4cb8990acd0051592106 : SUCCESS in 5m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/063fb8c377d74f28879028bedfc4e7f2 : SUCCESS in 11m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ae7729190374fc5838ef4c435275efb : SUCCESS in 10m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2bdecc9b2ee6473195b5fa786de7eb0e : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/61426ac169624a75ac43cc34fc806db5 : SUCCESS in 8m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/493646a091544101947408f46316aa98 : SUCCESS in 13m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/eb42082d3f2345b4a4624620c3efb41c : FAILURE in 1h 56m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/219b2260320345b8a47bc3d8bce7f18e : SUCCESS in 2h 00m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/98029b47d008455098dafc25566f7b31 : SUCCESS in 1h 46m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/801e7e1b14954d6191cecf633f889cb2 : FAILURE in 1h 20m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/489c32fe52da4efe85c0e6e242f839af : SUCCESS in 2h 04m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e91bfe1bd6084f38b5b699fe3f6c5548 : SUCCESS in 1h 07m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cabce8cc759843bebdeafb818a5772f6 : SUCCESS in 2h 17m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52653649bad94bc38e52284aead1eaf0 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":3},{"id":"1490a7fb85c0c65e9d46021548ef3117eccb8ad7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-12 14:50:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2f16132ae2938cc416d25c5a2fd96938b5e86ec6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 15:04:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab73835704d448aa2306de80f6d8217\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1e83ce596ebf4616899172311f4940b4 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8201a28584c94b8881a6c8dcf49c2195 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1df07c0871104bd481652701153bdf58 : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"969fb4704d1a84bc7a969015d16655680d51cbec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 17:03:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d980d60a4944615a9a86dd2e3e27e55\n\n- grenade https://zuul.opendev.org/t/openstack/build/6eb19f724a5f42be812980af530b6af1 : SUCCESS in 1h 04m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/511272c2e4dd493b932d4d1ba77ff662 : SUCCESS in 1h 03m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c4c9d1ce4b0d4f38bb5ea5c86bf32853 : SUCCESS in 1h 42m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7fee41dffe6b4845910b696bf0a15714 : SUCCESS in 35m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e9ee14e5e9b42d2a3e9e2cd24359c19 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b551c37925b640d2bc828bb1b501bd31 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b43cd6a6b1d249d7a9a01790f392158d : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab175eea38f41b4b84af726082e176b : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/834b0015735c4b8caa447d47bdd61ffb : SUCCESS in 5m 35s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/337f9555515b4151b01df052bb971bc1 : SUCCESS in 11m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d5188c61c7e6407c8afadcf3039198b8 : SUCCESS in 11m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5804d61efa2645c5aac70d39a1d0df37 : SUCCESS in 10m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/60ad8da1b89c4e8a957360bdd0f3d60f : SUCCESS in 9m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/26b43fd439a741a994dd4829983dbf17 : SUCCESS in 12m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/96a3efd47e294335b6fda03ca20908cd : SUCCESS in 2h 04m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cfc5a2885b3b40ebbdf8c48db5429859 : SUCCESS in 1h 38m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/37861bdf4aed48e3a8232ea2eca03988 : SUCCESS in 1h 41m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ad8fdc862b354289b5220b8c95ea1566 : SUCCESS in 1h 45m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6469f75bd57146fc90eb556d11ef370a : SUCCESS in 1h 54m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00436d2ea0ed47d88f1496a11d70feac : SUCCESS in 51m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c27e89888364e26a82fe2f4744f7f9e : SUCCESS in 2h 01m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0459e4b11f944271b663d6c8d6495031 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":3},{"id":"c7a95fc344031ac1cfee4fec68e1d7ed795a608f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 17:03:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8747221449328ba75a60950982f92a361b3a2040","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-12 19:25:58.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74c0a89743464232b097b41a41b253bc\n\n- grenade https://zuul.opendev.org/t/openstack/build/57c7e42dca474dc38c925f67c8ee5997 : SUCCESS in 1h 09m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/554f283da2074ec2bb1560a38bb95707 : SUCCESS in 53m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ffdd1c095723423ea51234fd1ec2ad5d : FAILURE in 1h 45m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ce7139fa036e4cd083cbf931264feac0 : SUCCESS in 26m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d4c2f9c66b4f5a95223e7fff252746 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d32e359d245d4c13bb72b0c82b9f3378 : SUCCESS in 5m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d49c5b2378ec42f398e95dec20a2a206 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79f48d3818b44b28556689b2bf5d438 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/813c8c0391a549d884b13dd6136bb96a : SUCCESS in 9m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a599ecb4ba7340db8b7750749b36a965 : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc56dc06aae1465a9ccb7b092f4f6bcb : SUCCESS in 10m 48s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f45629626dd043709252fac5a81a5e3d : SUCCESS in 1h 39m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97860cf27b3a473699a334c32191a243 : SUCCESS in 1h 36m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7d56c66341424ae997873cca75ea711c : SUCCESS in 1h 42m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4e6247c68194298adb7dc598ae64812 : SUCCESS in 1h 07m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3c5e2c6a2c2344ea8ba9344d741c55ca : SUCCESS in 1h 51m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6b5ea42640f9457ab86f08d4783436dc : SUCCESS in 24m 53s","accounts_in_message":[],"_revision_number":3},{"id":"8906c891a6689a8c09aef9a0b59ad9e2f1d91b60","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 04:34:15.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7cca68576c389e21ca2f72c11f224dbbbea860a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 04:45:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28bf18d0ee614f9db4aa1f356ad8dede\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65ddceaf8a984812ad572e4f6950a9df : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/72512e3e30b24680b02e97337aa7ea12 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5dae984bbc4d4118ac76279a7baf57c6 : SUCCESS in 9m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"be88494e0e4546647d9f1cac095b657d23818ec9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 06:50:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/858228fec3bc4eb98bf1d534284544a3\n\n- grenade https://zuul.opendev.org/t/openstack/build/54c5c55c8dd745b4ad88f47d1ca203dd : SUCCESS in 59m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c01e2c96c8464800bd52553c4fbc0dc9 : SUCCESS in 57m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/aac9dcf658244d0bb52cec27ed8a7ef0 : SUCCESS in 1h 17m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a63932ae2b334e50aa865f3091bdaad8 : SUCCESS in 24m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22905d518df14f0e981615da207510fb : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94befe71236848d3a1745020a3263ac2 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f10d219e53ea43d085fad70cd8df3b47 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68e5f65e69a44346a39750ba880139d8 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d01c74ba7cd44a68f691386454ac1de : SUCCESS in 5m 59s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0b89b9930bac48f0a8b7e6d5e752f70c : SUCCESS in 12m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/add972be65eb4dff95e0b8c4efc3ea2a : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/54de8db245a64284b82329d26cf43480 : SUCCESS in 9m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/59c9e259fc7a4fbf82551dfe8b177086 : SUCCESS in 8m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/aa10d311747340ce91e961c2d12da834 : SUCCESS in 11m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2513615857d341e883b95e89d27a39c9 : TIMED_OUT in 2h 04m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8e1ef8e8ebf3465da234bf7f631cf8c6 : SUCCESS in 1h 28m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/26abd396003f4cb784f9ce9a58fa30ad : SUCCESS in 1h 31m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4d2e6d2782ec40d3969984f2977b9ef7 : SUCCESS in 1h 44m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/221aea87f3a74d5493557361661b6793 : SUCCESS in 2h 09m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aace7bb46efb4a8c8dcc371b96a59477 : SUCCESS in 51m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a42e422337df49c297fd24cee5ce7430 : SUCCESS in 1h 47m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f833b44bca8b42c0a51c816674595152 : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":3},{"id":"5f5bcd636b7fde43b7732e2f55e2b8e6cd4954d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 06:50:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d1dfa98bdde68456a242d25adf17375ff27649d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 10:16:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e3841a6f1194c849901cf4c6f3e9fec\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b5a4386e85d47b08f39dc5474101039 : SUCCESS in 1h 01m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/219e656997a94cf7963d5eedc70e8c44 : SUCCESS in 50m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c10fe0adc8f44c4a611cd42592058e0 : SUCCESS in 1h 22m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9523221ce0624666a22972badddc3e14 : SUCCESS in 31m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2ea04f896174658bf53e48dd05e3b38 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7753f1a293e34dabb10fbdc3175738cf : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfaf67b9feea4f6d978c9778d9a97c9f : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40dcad3a1e540e5841d8cb4b4d08aef : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca8bf2cd41804d1faf4b4e26621a7b15 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd89ed25e32e4fb190d50b5c8cd83fd0 : SUCCESS in 9m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/baabd66901ed42678a3eda13bb891083 : SUCCESS in 9m 43s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/679f46e1221746e7b66d1b35531d0164 : SUCCESS in 1h 58m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f39e39d36cc6460ca9012d9de9c42ab9 : SUCCESS in 1h 46m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9d19d8b6c2124b2eb42ac29a78bb5872 : SUCCESS in 1h 47m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7a86048c26d495cb7cda32eab8f2386 : SUCCESS in 1h 04m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73cfe1d045fc459090466bbf66639f98 : SUCCESS in 2h 14m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ae12a491938c437bb8cb5fe56621cd9d : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b54148abb8cfc1e6d51d7fe27cf83f6285828e40","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 10:16:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9b3d1b6cff72bec026defea1801cb2fb50235e6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 10:19:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0a8134d12b84cd69f4ab586e8c31849\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02a9cc47622141ef85e5ce0c5e4be250 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b65d6f0fff04412991597415c3158774 : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6708fa60e970b1305c87478fb433a83d0be8bace","revisions":{"f410368958e13c3e041512dda50367101e94c855":{"kind":"REWORK","_number":1,"created":"2023-11-01 14:58:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/68/899868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/899868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/899868/1"}}},"commit":{"parents":[{"commit":"03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963","subject":"Merge \"Unset VENV python_interpreter in fips job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 14:37:35.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 14:37:35.000000000","tz":0},"subject":"Deprecate sqlite cache driver","message":"Deprecate sqlite cache driver\n\nsqlite cache driver and related configuration options will be\ndeprecated from this cycle.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I0c1affff05a2a4c2838cfad2b637e7660b056eef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f410368958e13c3e041512dda50367101e94c855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f410368958e13c3e041512dda50367101e94c855"}]},"branch":"refs/heads/master"},"e3b8ed90b6ec7b1946e06ff0d24e9555f04f3c39":{"kind":"REWORK","_number":2,"created":"2023-12-07 05:23:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/68/899868/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/899868/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/899868/2"}}},"commit":{"parents":[{"commit":"64e6da6fe1c810befab1e5a59230ce8711dce709","subject":"Increase timeout for glance cinder multistore job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/64e6da6fe1c810befab1e5a59230ce8711dce709"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 14:37:35.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-12-07 05:04:33.000000000","tz":0},"subject":"Deprecate sqlite cache driver","message":"Deprecate sqlite cache driver\n\nsqlite cache driver and related configuration options will be\ndeprecated from this cycle.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I0c1affff05a2a4c2838cfad2b637e7660b056eef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3b8ed90b6ec7b1946e06ff0d24e9555f04f3c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3b8ed90b6ec7b1946e06ff0d24e9555f04f3c39"}]},"branch":"refs/heads/master"},"6708fa60e970b1305c87478fb433a83d0be8bace":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-01-16 04:12:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/68/899868/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/899868/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/899868/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/899868/3"}}},"commit":{"parents":[{"commit":"41fa8223b7dda10e2ff310097bb86700f254015e","subject":"Merge \"Add Packed Virtqueue extra spec and image properties\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41fa8223b7dda10e2ff310097bb86700f254015e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 14:37:35.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-16 03:51:20.000000000","tz":0},"subject":"Deprecate sqlite cache driver","message":"Deprecate sqlite cache driver\n\nsqlite cache driver and related configuration options will be\ndeprecated from this cycle.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I0c1affff05a2a4c2838cfad2b637e7660b056eef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6708fa60e970b1305c87478fb433a83d0be8bace"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6708fa60e970b1305c87478fb433a83d0be8bace"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
