)]}'
{"id":"openstack%2Fglance~899869","triplet_id":"openstack%2Fglance~master~I9991a4e945bd4d231e942096445fd5374ff7ff2a","project":"openstack/glance","branch":"master","topic":"centralized-cache-db","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-19 14:26:25.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-02-09 14:35:59.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}},"hashtags":[],"change_id":"I9991a4e945bd4d231e942096445fd5374ff7ff2a","subject":"Add new tables for cache operations","status":"MERGED","created":"2023-11-01 14:58:19.000000000","updated":"2024-02-19 20:36:11.000000000","submitted":"2024-02-19 20:35:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"899869-centralized-cache-db","meta_rev_id":"48690a0f20cd7d3f26aaf86ec659083d097af1dd","_number":899869,"virtual_id_number":899869,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-19 20:35:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2024-02-13 18:52:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-02-15 01:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-02-15 01:43:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-01 15:05:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-01 17:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-08 15:43:28.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-02-13 18:52:23.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-02-15 01:43:45.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"7f161a809f8045cd5262153736237c9220544dad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-01 14:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"343f012eda1c4ec7d3ac112b4cb94d18140f8df2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 15:05:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/132c0663e6d94b2fb9c5e6100caf794e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d3ba800432243409e5ce8c7b57ea754 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/45788230e71d4e81b3b3e664c4073247 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/390f646d591747c29c401f196bd5b1dd : SUCCESS in 5m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66cbce1112deeb4c24218db154295bb7c3622ca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:13:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b484a74b01df4ae6af135f365ea5e7ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/40ca4113b67445d4abda4a8e0b39d3aa : SUCCESS in 1h 05m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1e2e014a24594128aed900f1115b1f75 : SUCCESS in 56m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/40df58b46dab4be9bb578432c7a3b1fe : SUCCESS in 1h 23m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cdbed160fa34a9fb607ba13b30c756a : SUCCESS in 48m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aef4e6b33f644e058e588a9c6508276a : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/434addd3844c4ac3aa6c428314d45fa3 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb395576da1445d3956702c4faf66fa3 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d2c6580d0d41b383ccfd085468792d : SUCCESS in 6m 27s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/55a973b255b340568ca333cf5bfee117 : SUCCESS in 12m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b25bf6cb2a034939bdd63550a020db18 : FAILURE in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5a525511ce74423a866feea2ce9070c7 : FAILURE in 8m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/85107b8366e84ff2a8126ce586f187a7 : FAILURE in 8m 18s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/988e79aa047b4091a86c37b8aa739e38 : FAILURE in 11m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c981eb8d6947415ba0836f11ccf31c58 : SUCCESS in 1h 40m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ed03013b201f42fa95f5b6db6d0e446b : SUCCESS in 1h 27m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/58e105fdfc0448ba994b79b638b6b4af : FAILURE in 1h 46m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5f29094eebf34f05be9e0d868fac0438 : SUCCESS in 1h 20m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3de180b249ad4867ade71296e97cc50f : TIMED_OUT in 2h 05m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8613ca98c752466b9a61b7ff35f22c79 : SUCCESS in 55m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20d2ab5de0e6483f8931141f45a36264 : SUCCESS in 2h 06m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b59676e32804d9d9006767cbe6e80c7 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ab1c4272b8200e02594ec8b51b334b300d22f32c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-02 07:27:29.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2fc4b5876bb4f10e6a800f4fec42ecbda8c7a4d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 07:34:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa79f0c241644db806df779b5a1cd44\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/399fdaa1f00d4b8281ce0f8e7ab99d89 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03839d048a714ed893b595a6ea686b36 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b0c6be8b459c40cdae7ed03083129cbc : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"539eb8e8c352337661f573063116360590b52771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 09:40:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/08c0bb6232aa491296d8d48c6772e51f\n\n- grenade https://zuul.opendev.org/t/openstack/build/766016e213bb49b18a5caf5ae22ec77a : SUCCESS in 49m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/23afaf50402d4789b8e717c3b21a8990 : SUCCESS in 43m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d28d9e9aac864661a92a94883e0a625e : SUCCESS in 1h 21m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6903d9cfb32a46b89b6435b4cfb52b2c : SUCCESS in 50m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7c7eda407b4fbfa827c15b746bcebc : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03b8b04e0cb14adf803e4aee839b9ce4 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30c633a9168b42d1acf05dc8cfad026c : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a191912f36d74a2d91097694bdcf22e0 : SUCCESS in 6m 26s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e837d6d68cfa48de914e1cc4e9160be6 : SUCCESS in 14m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ecd5616a70b742b0999f50d2ddcab12a : SUCCESS in 11m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8d33891d7635421b89494b31c462590b : SUCCESS in 8m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/698e0db463404badaea8742a16b75814 : SUCCESS in 9m 14s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/958bce0b162e4f6fba54efb661b39a6f : SUCCESS in 10m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4ae8a3cbbc3d432cb8fbcdf469a200fd : SUCCESS in 2h 05m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7891cf315c794d91b15f2494b7691d26 : SUCCESS in 1h 53m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/31bdca08f1a34e17b44aa9fbe28ec98a : SUCCESS in 1h 25m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c22b6f1b62944da868273e89db61b43 : SUCCESS in 1h 51m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4a9021858769494198ecc89c62b56b15 : TIMED_OUT in 2h 04m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ec48cb1736244569655ea9a60852d8e : SUCCESS in 46m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/773c1ee429cf44dc9e7f7d38302e79db : SUCCESS in 2h 07m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/826531fe149a4319b44ca820df375d02 : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":2},{"id":"b8bd219f4881ddc82442643056e6d655289bc4ca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-03 09:48:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"569b9f989ca6e5ecb86d759e9a482bf594c5a160","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 10:04:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/713b6c1a0b3b485db1cfb49512897a4a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93a2d2998f0f4cab86fc27835a86ad31 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66081cdef7934d22b8ef6d25c2205a8a : SUCCESS in 13m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c185d6995d7940ccb08ce2460aca2f5f : SUCCESS in 12m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b8bc06f7f402d70e43a1e4b1db58046aeaa8100c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 11:45:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a64daa50809e4c45a79a38afc7583d6c\n\n- grenade https://zuul.opendev.org/t/openstack/build/7925b2ebe1874419934194903f6d5c2b : SUCCESS in 1h 03m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/63cadfb345d9450d80a70d2abd27aa59 : SUCCESS in 1h 01m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/51233b671f1f4412a5b9f9384fbf5bde : SUCCESS in 1h 16m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d10fd67b74dc4e50a542584848bb91a1 : SUCCESS in 1h 08m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd713f97ed5a4f3985acf3e589f4d7d6 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c02a58f7e2a411c9a13b503d2ca0295 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/258be4432dff42e68a9daa693aace012 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4083228858eb4875b3427996bd387088 : SUCCESS in 6m 24s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ffb8a2bb39a34b73878e8432f8d0a592 : SUCCESS in 13m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5000e46ab553402485555098a05ce98b : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/19e9ca856960436d848370ef29296425 : SUCCESS in 7m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3f6fb5ae2204a6b9ec4aa386eaf5af8 : SUCCESS in 8m 26s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5132343be2914a76aa26a0186fe44ba2 : SUCCESS in 10m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/53a6571d29de4a4b893954ca7e3788e3 : SUCCESS in 1h 27m 47s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4d83e12c06a144c69cc227bfb25951be : SUCCESS in 1h 43m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/14810cc829514a82a311312f4a0186f1 : SUCCESS in 1h 51m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8c8450f552154598ac512a91943882dd : SUCCESS in 1h 29m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e4f299367bae4d599a5598455b5cce41 : SUCCESS in 1h 44m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2eacf1a698dc4b4c80ef0a8b9f27e94f : SUCCESS in 1h 00m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e051f28fbb8c481dac29ae5f2cf75541 : SUCCESS in 1h 39m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dcfcea2692dd4109b03ace5a36ac6313 : SUCCESS in 21m 50s","accounts_in_message":[],"_revision_number":2},{"id":"0215138839541d59d7ebe0ca36b10a1d6c15c272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-21 04:49:46.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"10c640ad75dcf8e686ccde2061fbca63c29c086b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 04:56:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00d5bf7e73754f939041239bcedc4d1a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22c2bf8ddc024711929f4d25b5d01167 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4806bde026e046f59cc8cc0ed4c813e1 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4ab17a5a21024cd19de86345e78ee96f : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"316b6486e384fc62869d24ae8f19a4fd87e6941c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 06:59:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9679179b3e1941a2be6b3fef04c3bc47\n\n- grenade https://zuul.opendev.org/t/openstack/build/5c6fffa534d24b7aae0f4bffc3956216 : SUCCESS in 1h 04m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ddd49cac0236489d874ba564cc6e9c1d : SUCCESS in 56m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17078e5861c4442b81da80513cb6b790 : SUCCESS in 1h 24m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acb29b0bc3ea470f877a5ea03befb26d : SUCCESS in 1h 03m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5771649e86bd46b49980bbcb3d03018f : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7510313b0fe432a9dcc44a0f1b116a8 : SUCCESS in 5m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6be7dccd331748a1be27e0a3aa2b7904 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9477f98ffdb24332bbc017b324486caf : SUCCESS in 6m 58s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/08f17a9b36864bc0b32ae8e18ab29d17 : SUCCESS in 14m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/535bc2d0af2e418a807f4efce8213f34 : SUCCESS in 9m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/75dec46bc978486c9e46c646056cb920 : SUCCESS in 9m 36s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e55dfcd779894e02acb47ea572e2da1a : SUCCESS in 9m 16s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/15ad9b54676f4d90871b87813bde941a : SUCCESS in 11m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d5ecb760a6ad44aeab37eca862b387ae : SUCCESS in 1h 49m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a0c6e436408041a89786a08f3df4a0b6 : SUCCESS in 1h 44m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/082f74acd33a47d790d396393d773195 : SUCCESS in 1h 27m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9be40c78de9c4c3c842d14442c8633c1 : SUCCESS in 1h 40m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b9f1c2992dee4c909df803649a083a52 : SUCCESS in 1h 53m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de67a48eb58d47c2991631f179899a6a : SUCCESS in 1h 00m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aee770502db940ed91e80749356e4c5b : SUCCESS in 1h 46m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b8b057f04bc4bd9869dc461a7ba6a04 : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":3},{"id":"644ac5077cb3917022558f1fb59b29a36b54dff1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-16 04:12:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b958a5e6817956af4a406cdde0eab95cfeb32432","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 04:20:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b24b6063089a46a3b724220a626729e2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd6bab0d4e1d4957b3b64c56d8c36c62 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/76f0ba812d6449d58ed748f1a8552694 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e0f30e0955984368b9621b6eb14ddc18 : SUCCESS in 5m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"eaaa59f293479a86c3ee6b9a60d34b08d4047412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 06:22:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2034e7f0ba464530a1fd43c03c3454c9\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f00b989325947f09307a637d11324ca : SUCCESS in 58m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ab8bec260a7542c78eab6603bc635daa : SUCCESS in 45m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fea6946f23a7481ba509071cbd68d4f9 : SUCCESS in 1h 27m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a16408bc25c14f5ca5f297fcf427877e : SUCCESS in 37m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed49b764aa94ef0b132864143dac32d : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9032b5dad84469e913f48e01d7f923e : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9dab318381a24780a4b555b3c6318b52 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff229ca8551743e2b7de063b52ec3cc8 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ab374d34edf140aba719e00eaf9a1f76 : SUCCESS in 13m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39d8e095befd4cb4b8ebe653c29e269e : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2547de7668e3411f9036a6d6501aeaa0 : SUCCESS in 7m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/08e28e4cd9cd4fe2b1d5250dce950562 : SUCCESS in 9m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/273f0f069c9445d0b6d9dd336195e4f1 : SUCCESS in 10m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cf9ee21fee2540c584bc0348175d1444 : TIMED_OUT in 2h 03m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d34a56e6ce5d478cbf6c5d4cb1af3ca0 : SUCCESS in 1h 47m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ccc251ea18f4d6697dfa31dfae7c67d : SUCCESS in 1h 52m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec87881bf73a468785ccd7e32e305ff8 : SUCCESS in 1h 45m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7de442ca0f5d45fd9dd229e713c28325 : SUCCESS in 1h 44m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88a9c27ef17b4e609f4efa46059f85fb : SUCCESS in 57m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b77b7f6f9abc4db1bd27c89f3f26f1ab : SUCCESS in 1h 54m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/73ef3a40935b4e7e94f85aca9c14d45a : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":4},{"id":"170c46ffc63c8054d60d6798f8461e96a417832c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 18:28:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"531e1e9c0636dde25827aa08252a3c984c78a69d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:38:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d23d9b2c18d4566bcc555fa380d67c5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14c34277075043188dcf59f0e450f518 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/23ded1d4e5cb41b0856ec96a79e2cc2a : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/525775a55e1145cba77cb1d4507fed87 : SUCCESS in 7m 21s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"2e8cd4404ace4419017c082293268b35648e11fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 21:05:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a49e3bb1460548b38b758088f87556de\n\n- grenade https://zuul.opendev.org/t/openstack/build/5ab503850947405d823ca1a2a19328b2 : SUCCESS in 1h 02m 37s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/34db5bc75d554fd291c3277419740deb : SUCCESS in 57m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39ba97f83cbc47bd8dcbaeb519587a28 : SUCCESS in 1h 35m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8de161c25294ce2bd2749469646b419 : SUCCESS in 24m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d0dfe01b9d486aa0261b9b4a21ab61 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f122ef39b4244cd981daf8d2e343f28 : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/373d25851d1e485c9b6501cc377d062b : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/992623233ea94b7f8581e09e6d38f79f : SUCCESS in 7m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/2f3f913ee59a47daa030e402f182fbc8 : SUCCESS in 14m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d97a0ef255f04ec089ce92621c45edfa : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0b14f864f5484c6e97dca617a619ae19 : SUCCESS in 8m 32s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e3894531bfad4d9ea6b409b3bd3883c5 : SUCCESS in 9m 56s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0d03638fa03847e197b5e8d3e5608d68 : FAILURE in 11m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f824248208dc465a9116f57afd146e92 : TIMED_OUT in 2h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/24550903596f494fa97198b3b0cebf17 : SUCCESS in 1h 45m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3eb0cffe06f64ed59cc831ad251f7e83 : SUCCESS in 1h 47m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec1649462d5c49e99cb3599a0e60192b : SUCCESS in 1h 56m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/10eacd9b29ea4e54a53a7b2718473ddc : SUCCESS in 2h 14m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/617996a233a941f88e1f5f7751010e04 : SUCCESS in 1h 04m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4801e83ae71b4889bdb074814e79873b : SUCCESS in 1h 47m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4da202a26aa04a2d8cd0bd28e3aa7dba : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":5},{"id":"ae7e786f2cc05f776d1e6d713d9c0085f53dbf83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-18 05:34:58.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"beda22f8ddde090588ebc34c33cdf66024295748","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 05:48:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3e194fa15f41658d56b8bd648e7af8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3cc397c371824c08b9bee2649ded1019 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7bb1c7db005744fd9167d5aea58cc39c : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ce0d29bc23bb40fa9b7bafaef798bae2 : SUCCESS in 11m 37s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"207ed45574c27f73a8b51a3d5fbe050de9f77753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 07:52:43.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/120e6724e3f545b9a598ad86ed587b28\n\n- grenade https://zuul.opendev.org/t/openstack/build/8a82134e70284219aa0520a9dc2991da : SUCCESS in 1h 07m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b474a2a83e3949b0be8d0c5886f034f7 : SUCCESS in 1h 01m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6e50b696a074e348c8bba331decb1f5 : FAILURE in 1h 37m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d64d36023a114f8f88435a9299d3c3c9 : SUCCESS in 35m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c3d52c48da34c2d98a371cc3924a639 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4d84723b1a34858a7ae39d569d428e9 : SUCCESS in 5m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5bf2477cd6824ac486f63dd8c8924d18 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877c96b607c64866a454e7781db6a276 : SUCCESS in 6m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ff3f15deb4d54c119c57364343e934a0 : SUCCESS in 11m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c1e4e1cf18143ca9672eb3772d4b903 : SUCCESS in 10m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d1acde3c77cf4afab2adf4b948306fbb : FAILURE in 8m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e927e922c47d4486aec383b662197bf0 : SUCCESS in 8m 57s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b64ad3be01dc4c2bad2658c6e65ff397 : SUCCESS in 10m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7172c295065e4d538e666e9d76fcee8f : SUCCESS in 1h 41m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3e73f489da5f4e769c425ecdce0f5b59 : SUCCESS in 1h 55m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/68619e3a1fba4d28910ec78622a45c7e : SUCCESS in 1h 52m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb9bfec5fde647e99bfcbecb2927c644 : SUCCESS in 1h 30m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20770fae5fac4b97981527e24c818558 : SUCCESS in 1h 57m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4202ad9493bf4eaaa09ff9957435a80e : SUCCESS in 1h 23m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c926dc098d94bb8868b6bdd46361d64 : SUCCESS in 2h 04m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af830fbd183945c098f4a3274a033e21 : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":6},{"id":"eea0d5655aeb9c3c3feb8183b97bebe95b53dbce","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-25 05:20:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"76120fed1b065e94bf488771611309fe1daf62af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 05:26:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/976d460d56844eb88d775240af019168\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65d8d9dab7934b1aacce78d9e4b0b75e : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/016d373e545c4f6f8a1471e322c042e8 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ba2b34319959456986e4d9b028172d4a : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"479798437516885f410947085e7bf15ef7125833","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:34:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09aefaceb3b140df8e07cd27a954fded\n\n- grenade https://zuul.opendev.org/t/openstack/build/f266e8b5388d4a4eb8fa6a08aea512e9 : SUCCESS in 1h 08m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/a95c50a861c74a05a86be70be1a5b31d : SUCCESS in 1h 07m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/865f2377b4dc47edbecdfd2f576482ff : SUCCESS in 1h 23m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f9fbcc1a7304a0a953a983c4af5734d : SUCCESS in 38m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cb437e2f3b4268bd8ea737c600a976 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af16029c144c4f298a7515c39dc5a055 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6ad180aa65447adbffca8edff01efb0 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fb71a16fdcf47adaf8066ff5341f419 : SUCCESS in 7m 18s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a5e412746ef14c048178dd088ce5ea28 : SUCCESS in 13m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fadc49e67e554487908eb1809517ba3d : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5831a32c0f6d49b79d7ab119405f9cef : SUCCESS in 9m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/5360fe4a35b64961bc4fe3db5a2b2e74 : SUCCESS in 9m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f2ccbd9e439c470ba534263207a90597 : SUCCESS in 9m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca917ab18089418390c634a516d5971f : SUCCESS in 2h 03m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dd4c4277f2764cddbcfaa29dc3830e56 : SUCCESS in 1h 47m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aa893322cd2b44faaaa89c00ad60f858 : SUCCESS in 1h 54m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0e1f977ece6f4f06b03478e61b0e3a92 : SUCCESS in 1h 11m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/06e17315ada64378868e5ca3a3532350 : SUCCESS in 1h 58m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/173b647f468e407985876df1b42cc3aa : SUCCESS in 37m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a25603d3ba904881ab2aa31ed4c1d818 : SUCCESS in 2h 07m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5197bd11b70f4906b07bf349734e3e7a : SUCCESS in 40m 46s","accounts_in_message":[],"_revision_number":6},{"id":"d1a59f084445a77903148eb6234b77b19b4f97a1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-08 15:43:28.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ff78d6dd5c18f3e042a8385fcc5abc1fc5b22b9b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-08 17:44:50.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"6d5bb0b26764b2d4811667a3f18012905c0d6782","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 07:28:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"00f69ed08c6588aeff787202b5c83544d19289e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 10:05:42.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"0b7e5fb1a0b03ccec33acfaed41d62f8b2b5b1ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 10:18:05.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/832ce147bd8c441492f899c3b1a031da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5daa854424042af8b53d7abb769be45 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/69b33996db6c4586aed6ed807e317058 : SUCCESS in 10m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d14fa4af3a544f4fa7174b92fd6ac394 : SUCCESS in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4a45a7ddcb9c2a822c54df4ba013662da2f13d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 12:10:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0775374883054e10a7869cacc93491ee\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0c4f78b03d347aca954c6e94c9db811 : SUCCESS in 56m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eee9c563bd4840efad12fc919f83ccc2 : SUCCESS in 53m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4649b069091842d4bd5904496c6767c1 : SUCCESS in 1h 12m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1265e6672f3347d9aa52bd9cfb22506f : SUCCESS in 29m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c53123e89244feae0f0baa2b420855 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2af9829889244dd48bc89149960b2bb0 : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c253ecaade7d4b21b6c2bc401b6351ec : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bf7ed3c0842414dbf811b96ccf9ff52 : SUCCESS in 5m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/560122ec459845ebbce61bd634145b0c : SUCCESS in 12m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6886b88d4e8a49ff89eca438b80bb36d : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8b83954d4791430d9e05a2b002f65c27 : SUCCESS in 11m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/90b74a5b478b43c69bf46a1b5238ec15 : SUCCESS in 10m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b82ca0c0ff4541a885f362b97b3889f0 : SUCCESS in 13m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe106dd94e1a43bf91dbbff55d96feb3 : TIMED_OUT in 2h 03m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8f515fd76ed84e84bbb82d98339b92f8 : SUCCESS in 1h 35m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c07d074aa164b1488320463fb750440 : SUCCESS in 1h 37m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ffad0a7fd84a4a2c8697f1e03ea882e8 : SUCCESS in 1h 50m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/42c99d9296424d578c86ef4d418fee36 : FAILURE in 1h 56m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a136de50f9c0408492fc698dd6dd2063 : SUCCESS in 53m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3ff616dc994a40dba968d7fd9c0beb80 : SUCCESS in 1h 54m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4610ceca30ee407b87ad795537abf1cf : SUCCESS in 29m 27s","accounts_in_message":[],"_revision_number":7},{"id":"955b36210e88925d2bb2d196e745e60f9412ab1e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-09 14:35:59.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fb9db82ce3e05d2085e7d9dbc0c0dd559f3a9b77","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-09 14:47:36.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c10c49053e6cce74db336d42cfed971caeb77365","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 15:00:41.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529fbd56aab6479c9780b3e466439784\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/38366d69b54a4dab99a0353a23c17d97 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/037af33ccc27454aaf1fdcccb93eab30 : SUCCESS in 10m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0d0df2ec41b24c42b4a6379e4741b00b : SUCCESS in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"accefd7c44e7b7c926bf2ebbabed94b25844b154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 16:48:39.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7725d3409c94686b92ee4f57074cac3\n\n- grenade https://zuul.opendev.org/t/openstack/build/d290cace593241e8a995fc99ec4917d2 : SUCCESS in 57m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1425f12e77f04f679198ded6b64b4537 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/036d1da417d0420a8b896bab18ad9fdd : SUCCESS in 1h 21m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92e7627fd97a4a48a48238d0aa906ffa : SUCCESS in 38m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6ec873b34a64715a95d071ba14cc38d : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72a962ece042486ba9917f0a77bee41a : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e2c20f0e3574eb5bf4e4ab17a6cb036 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad83914c398a4c0cbbc63e362ea2e026 : SUCCESS in 5m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fc70c30722b14c3e8f5ffa11f9ebfb7e : SUCCESS in 13m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a3f146a77e7420ebe42417c2d9ffecb : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/96b3184c5be044a1b5361f0a6d35af6a : SUCCESS in 8m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ae3ddce0522b4e969f7c2ce3409d236d : SUCCESS in 8m 43s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/994ecbf10d1d4e3aad822ff1ff5b16a6 : SUCCESS in 10m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/104dc6303bb8422291ca1b8fc3bc7d68 : FAILURE in 26m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6314f56d882d4d4c8aa28500c2c2aab1 : SUCCESS in 1h 25m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/abdca3e9f01843d886b8ce322cc92795 : SUCCESS in 1h 54m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/727c75a47b2f49d19c1dfd5222c0310a : SUCCESS in 1h 50m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5fb847f27d2741abb1b2b54cd6b6c78e : SUCCESS in 1h 57m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b4de236a8a38462ebed28d17458496c0 : SUCCESS in 1h 00m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5cec9a2dd14a47f0861722abd333458b : SUCCESS in 1h 45m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ec84771fcb74d0c94356d6637fed4b1 : SUCCESS in 58m 29s","accounts_in_message":[],"_revision_number":7},{"id":"acc6319c61c6eb13f9d19d7636c544aace8d04fe","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-13 18:52:23.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"71aa9b7ccc1548eb4bbec234aa9da00f2ab3578a","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-15 01:43:45.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8770ddea9283213940de5db64d107742b1299a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 01:44:30.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"2c70671056b8a4b93a4c8b5eb0b9a1f89d49d6f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 03:54:11.000000000","message":"Patch Set 7: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/36bcbe97e6524867a2e2694a2a04d645\n\n- grenade https://zuul.opendev.org/t/openstack/build/6d15f0647884440d8df9dd31bd7105d1 : SUCCESS in 48m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/518e2f6bbec54fb39587e1121acd7657 : SUCCESS in 44m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/585b1900ca9940eab93d517b92423a1b : SUCCESS in 1h 24m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5106162785ab47cdb93fb2ec0a8643f4 : SUCCESS in 22m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/031bdc6eeb844e03a09402d50ad3b36a : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de27c4eb81c842a698a038296838d88a : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/691e867c06544a238e5947e8158215f0 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df1d16d3e6c14894a7665cf52ae9ed6a : SUCCESS in 6m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/609e2224dbec49cc87c1ef886bc148d6 : SUCCESS in 11m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4f28c062cc647db83982d798772d4e6 : SUCCESS in 7m 45s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a28110ccd7d14d58b4a3aa6cd744e16a : SUCCESS in 1h 39m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f49ba12648fd4e018bbe389dd693805c : SUCCESS in 1h 47m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dbcd831800954c29a71bfd63dbca5c6d : TIMED_OUT in 2h 03m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4144d68d33764ad784a343767beeba75 : SUCCESS in 1h 05m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/11cdfac5a95846f28f8797d78a782c66 : SUCCESS in 1h 59m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c30419e61784bb8ac675b5ec133ac88 : SUCCESS in 41m 13s","accounts_in_message":[],"_revision_number":7},{"id":"cf54ee689bf0dce511232ec30749ec7dfa64492e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-19 05:29:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"27f36bfc5e4cac04b911ffc3a4d4b8e8eeee9af2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 05:41:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d6c5c0c9fc410484fe5ae41be486b7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/636630423be746f3afe94a50fd54aa39 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d9d647b844c248e8ba54249e1fd63835 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7a1db44de6a040ccb725ffb12533a2cf : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"797d38fae6c698b38b5cb2fd25f1fb54711ab778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:49:59.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b56c60028663405f9a048fb78f4bb644\n\n- grenade https://zuul.opendev.org/t/openstack/build/f8e8867560084d06b418f3ce5d1171d5 : SUCCESS in 47m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/401e5fe4340244be9ac49d3c7888f70b : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/06817d5cc34e47088bcf77f199e31b2a : SUCCESS in 1h 04m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9f7900c0b094fd7b00d8424aeebfd63 : SUCCESS in 1h 05m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41da77ed0dc14796b6099c0f028eab08 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ebe77bc3b8641688f6792facd2e76c0 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4436fd8066e45eab7a521ea05f16fce : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e95535ff3c4879a22aa526f20e1de2 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/02bbc6098a294f8b8a009dbb57610cdb : SUCCESS in 10m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ad82623dba947268f469dca7f779c25 : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d8d3aa024dd84db891f4f7882a863ff7 : SUCCESS in 7m 48s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ebe03d80357249d69a854dbb9c1c330c : SUCCESS in 7m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6c69c8b235c842c78d7eb443f44da69e : SUCCESS in 11m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a42a4b9eeb042caa4cca3bbe689f8af : SUCCESS in 1h 59m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c1d6f3cdea014c01a0ff3a0dacda46be : SUCCESS in 1h 45m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8e4cae0d3df34584aa5684a817863821 : SUCCESS in 1h 43m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c0d7cbb7be9c4cf0a0af7d2e4438f0a8 : FAILURE in 1h 21m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1492c58ab5354468a12cd2c71a9c8811 : SUCCESS in 2h 14m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea7d68f3c3fb42fea2ff68f54aa992ae : SUCCESS in 1h 01m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/95409e06813346a88b5ebd2f32e69a1a : SUCCESS in 1h 48m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a620845ad7354babb49c80e341970fc7 : SUCCESS in 25m 12s","accounts_in_message":[],"_revision_number":7},{"id":"4a7cd8b13340ae3d1d57feb010b67e277bbd226a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-19 14:26:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c8b562f7b425b96b163e15683a0fe956db42d90d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 14:38:29.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7482a77c4b64ea690e0076d645ff3c6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a25424e138c84f628f97828ce085ae37 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a47bb820209a4564867561f089cebf4e : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1d58232351b0491eabd6aca7d4765d95 : SUCCESS in 10m 13s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"af3d3d43bd9f8e0634294ccc6f296894a0f272c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 16:31:37.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b8b13929c86488daf92a9a02c1638d9\n\n- grenade https://zuul.opendev.org/t/openstack/build/a57e82f2b16342f7a96956bd473c0636 : SUCCESS in 58m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/19c61243422448ef9de0f8fab57b4b16 : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9db06f153907497cb1ec422a694baa2d : SUCCESS in 1h 20m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e1c0c3d9381468489dee6b54c734463 : SUCCESS in 57m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea9ee1a3324e4b97a0986ff1066f67aa : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c420d648a67e494b82011909f91250a5 : SUCCESS in 6m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4babeb375b3f44eb8ce1d8edbc3efe20 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6687998e844b450eaae016ed301a3591 : SUCCESS in 5m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ba519e2565ab4a73a422bcc34c9bf88a : SUCCESS in 12m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f435ac8b5149432586d981f7d64717ec : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9391bd1eb40c4624b056a1b73dfca89d : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dd6bee02ec5b438d9b5286c45036d1a6 : SUCCESS in 8m 20s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ce0712df373d468688cbfcb2a3038e05 : SUCCESS in 11m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bdbd48c9c9fb4d13a68747af600e7f28 : SUCCESS in 1h 48m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2d727f3ad5494f7598baea777c0d1215 : SUCCESS in 1h 58m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8ec36441eb824e01be922a7a48e95b10 : SUCCESS in 1h 57m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/927ef2317a5a4a3cbba29eb8026e5b22 : SUCCESS in 1h 31m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9386c56ca4d442fcabfe897fdca9dabc : SUCCESS in 1h 59m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/020800ceedf74601af1b7f8795501635 : SUCCESS in 50m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/63d7e0f07f464df4af957f5b77d22159 : SUCCESS in 1h 48m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f8ab48bf54d94f5dae152ff782d033e6 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":7},{"id":"fbba51faece4a261325f4856a2f3183c2601056b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 16:32:20.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"728091ad50a6b2b8bf47cd634c7141c63966ac44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 20:35:00.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/015bce9fb37b405d827716c56d0b05ac\n\n- grenade https://zuul.opendev.org/t/openstack/build/912c1a93be574e25af27445e2974c716 : SUCCESS in 1h 06m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ed9711bad52c428a969d209ac804d43b : SUCCESS in 48m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97dd067ecd5146acad0560ba7288f484 : SUCCESS in 1h 16m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b63d818c2b93460d9ce5ed224747df27 : SUCCESS in 51m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db137f4618a54398b76fefa6691b8cd7 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dcffe81e427742a7b5743c3acd83838e : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c30f0c49395b46768a4a24d6de607730 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22ac2ca57c44dd3811ca23cfd3d9274 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/94d769eaa2d84e72bd5f01c7b87329ff : SUCCESS in 9m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fc0f604750354cd4a651944fff092175 : SUCCESS in 10m 08s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2f821992ec2e4e2a99c21808c7d084a4 : SUCCESS in 1h 38m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a5b70f301d4d4e3f9f8b0ef0698604d0 : SUCCESS in 1h 44m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e1d0d595c35a405faa3496cf7b6d8acb : SUCCESS in 1h 39m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d929c421653a4924989781792996cdd8 : SUCCESS in 59m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e978a0b2e8c042e1978aa911a774d966 : SUCCESS in 1h 38m 13s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b6b3c12646b4a5f9a66682e0fb92c9f : SUCCESS in 36m 27s","accounts_in_message":[],"_revision_number":7},{"id":"eb07346e98a59cf05d0f7b57332c2061f32aaf97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 20:35:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"48690a0f20cd7d3f26aaf86ec659083d097af1dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 20:36:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd7af706a8eb486e9ebf6f6241c6c593\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2332919e98c84e629968b238df549f5d : SUCCESS in 54s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"b6bd03d4bc5cd3eef8b73b72b3998da66b95cbb3","revisions":{"27c449d1a0b94fea35c90a3f5e9b4ba57c28288b":{"kind":"REWORK","_number":1,"created":"2023-11-01 14:58:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/69/899869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/899869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/899869/1"}}},"commit":{"parents":[{"commit":"f410368958e13c3e041512dda50367101e94c855","subject":"Deprecate sqlite cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f410368958e13c3e041512dda50367101e94c855"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 18:11:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 12:22:39.000000000","tz":0},"subject":"Add new tables for cache operations","message":"Add new tables for cache operations\n\nAdded expand script and model to define `cached_images` and\n`node_reference` database tables.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I9991a4e945bd4d231e942096445fd5374ff7ff2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27c449d1a0b94fea35c90a3f5e9b4ba57c28288b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27c449d1a0b94fea35c90a3f5e9b4ba57c28288b"}]},"branch":"refs/heads/master"},"9b0177193ba6bbfd531114c289f4fa66f3641d2c":{"kind":"REWORK","_number":2,"created":"2023-11-02 07:27:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/69/899869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/899869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/899869/2"}}},"commit":{"parents":[{"commit":"f410368958e13c3e041512dda50367101e94c855","subject":"Deprecate sqlite cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f410368958e13c3e041512dda50367101e94c855"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 18:11:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 06:16:36.000000000","tz":0},"subject":"Add new tables for cache operations","message":"Add new tables for cache operations\n\nAdded expand script and model to define `cached_images` and\n`node_reference` database tables.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I9991a4e945bd4d231e942096445fd5374ff7ff2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b0177193ba6bbfd531114c289f4fa66f3641d2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b0177193ba6bbfd531114c289f4fa66f3641d2c"}]},"branch":"refs/heads/master"},"de528dd3037adce08d10a4928897d0582c6f9542":{"kind":"REWORK","_number":3,"created":"2023-12-21 04:49:46.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/69/899869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/899869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/899869/3"}}},"commit":{"parents":[{"commit":"e3b8ed90b6ec7b1946e06ff0d24e9555f04f3c39","subject":"Deprecate sqlite cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e3b8ed90b6ec7b1946e06ff0d24e9555f04f3c39"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 18:11:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-12-11 06:01:06.000000000","tz":0},"subject":"Add new tables for cache operations","message":"Add new tables for cache operations\n\nAdded expand script and model to define `cached_images` and\n`node_reference` database tables.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I9991a4e945bd4d231e942096445fd5374ff7ff2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de528dd3037adce08d10a4928897d0582c6f9542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de528dd3037adce08d10a4928897d0582c6f9542"}]},"branch":"refs/heads/master"},"a2000fad156dc879e8591183f0d31ae0632a97ea":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-01-16 04:12:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/69/899869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/899869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/899869/4"}}},"commit":{"parents":[{"commit":"6708fa60e970b1305c87478fb433a83d0be8bace","subject":"Deprecate sqlite cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6708fa60e970b1305c87478fb433a83d0be8bace"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 18:11:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-16 03:51:34.000000000","tz":0},"subject":"Add new tables for cache operations","message":"Add new tables for cache operations\n\nAdded expand script and model to define `cached_images` and\n`node_reference` database tables.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I9991a4e945bd4d231e942096445fd5374ff7ff2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2000fad156dc879e8591183f0d31ae0632a97ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2000fad156dc879e8591183f0d31ae0632a97ea"}]},"branch":"refs/heads/master"},"e7ec379ba37ac4ab3265b7814b0a2c8d20f4c8ce":{"kind":"REWORK","_number":5,"created":"2024-01-17 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/69/899869/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/899869/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/899869/5"}}},"commit":{"parents":[{"commit":"6708fa60e970b1305c87478fb433a83d0be8bace","subject":"Deprecate sqlite cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6708fa60e970b1305c87478fb433a83d0be8bace"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 18:11:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 16:42:54.000000000","tz":0},"subject":"Add new tables for cache operations","message":"Add new tables for cache operations\n\nAdded expand script and model to define `cached_images` and\n`node_reference` database tables.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I9991a4e945bd4d231e942096445fd5374ff7ff2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7ec379ba37ac4ab3265b7814b0a2c8d20f4c8ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e7ec379ba37ac4ab3265b7814b0a2c8d20f4c8ce"}]},"branch":"refs/heads/master"},"3c11ef2047dddcd864e212b68cf103b2b4577d05":{"kind":"REWORK","_number":6,"created":"2024-01-18 05:34:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/69/899869/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/899869/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/899869/6"}}},"commit":{"parents":[{"commit":"6708fa60e970b1305c87478fb433a83d0be8bace","subject":"Deprecate sqlite cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6708fa60e970b1305c87478fb433a83d0be8bace"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 18:11:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-18 05:17:00.000000000","tz":0},"subject":"Add new tables for cache operations","message":"Add new tables for cache operations\n\nAdded expand script and model to define `cached_images` and\n`node_reference` database tables.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I9991a4e945bd4d231e942096445fd5374ff7ff2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c11ef2047dddcd864e212b68cf103b2b4577d05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c11ef2047dddcd864e212b68cf103b2b4577d05"}]},"branch":"refs/heads/master"},"b6bd03d4bc5cd3eef8b73b72b3998da66b95cbb3":{"kind":"REWORK","_number":7,"created":"2024-02-09 10:05:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/69/899869/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/899869/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/899869/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/899869/7"}}},"commit":{"parents":[{"commit":"6708fa60e970b1305c87478fb433a83d0be8bace","subject":"Deprecate sqlite cache driver","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6708fa60e970b1305c87478fb433a83d0be8bace"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-10-31 18:11:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 08:40:46.000000000","tz":0},"subject":"Add new tables for cache operations","message":"Add new tables for cache operations\n\nAdded expand script and model to define `cached_images` and\n`node_reference` database tables.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I9991a4e945bd4d231e942096445fd5374ff7ff2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6bd03d4bc5cd3eef8b73b72b3998da66b95cbb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b6bd03d4bc5cd3eef8b73b72b3998da66b95cbb3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
