)]}'
{"id":"openstack%2Fglance~899872","triplet_id":"openstack%2Fglance~master~I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6","project":"openstack/glance","branch":"master","topic":"centralized-cache-db","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-02-27 21:57:36.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-27 21:57:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6","subject":"Read global config file for cache utilities","status":"MERGED","created":"2023-11-01 14:58:19.000000000","updated":"2024-02-27 21:59:40.000000000","submitted":"2024-02-27 21:57:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"899872-centralized-cache-db","meta_rev_id":"0088376995409def28ab771e2c57aa392b4ca71e","_number":899872,"virtual_id_number":899872,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 21:57:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-02-27 13:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-02-27 14:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-02-27 15:00:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-02-27 15:00:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-01 15:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-01 16:58:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-28 14:45:17.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2023-12-19 06:23:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-02-26 18:27:50.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-02-27 13:22:57.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-02-27 14:58:49.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-02-27 15:00:49.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"b23a38ef7694d49dca1f411e195a3a4195e5b73c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-01 14:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d01b13efcb10de56e2603103a9dfacf9ade6104c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 15:12:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afc47d36320c4a448214f56356f0a6c1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/32070a13ebfb43fbae250a48fd91d29e : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2fde5c511ef045b29dabc6d9d1df84f8 : SUCCESS in 11m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/94888fc13fff431e855b0ca86a942c18 : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1391869f6167fd4ffefe614aab89a6ce156bafca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 16:58:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1683818748014eefb90ca5bb520c5202\n\n- grenade https://zuul.opendev.org/t/openstack/build/e0971dc2a4af4a44854bc4ba7dfbac76 : SUCCESS in 1h 04m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c7b5341666154a90aee9f41171b5bb0f : SUCCESS in 47m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e5e6cb231f14c269b748ce0ffb4d8ed : SUCCESS in 1h 18m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af23da23f69d40008c3fb7593ff022a4 : SUCCESS in 51m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2463d218f9b447ebab9036d204970775 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69965a42ac34583b5d979ddcd9d0c8a : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/187fdafb9c3d4ec298746bd4a760f1ff : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be63c9d829424706b9df678c168b990b : SUCCESS in 6m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/54332042ea684b73b941d02602a131b1 : SUCCESS in 12m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/13013150adcd4dfe8be2409f57de4a78 : FAILURE in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/487844c7146d40dbbf7c7ca3000aedd2 : FAILURE in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/328ed8dd07a64f85885fb6b411b14836 : FAILURE in 9m 15s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8ba3946989694d2eab8395a1c93f58f8 : FAILURE in 10m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1c22596abe614df588cb0cdfe9dc6446 : SUCCESS in 1h 47m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/79075c80df2345b0bfd9aef7819a3346 : SUCCESS in 1h 52m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/921e0c3b440e4695a0d1ebea10d4352b : SUCCESS in 1h 42m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/820a7515d969432eb1bbd41f40ade04b : SUCCESS in 1h 53m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/154ec4306cb4442aba3dcde9fe919dd2 : SUCCESS in 1h 46m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2995c622cc6844f59a6d638f3fe60c82 : SUCCESS in 1h 03m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/de0622cbd58b4cd293d404ae4809de60 : SUCCESS in 1h 43m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d9cd9711501f40cf89c9fa0a5cf9a5d6 : SUCCESS in 30m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2376c6a9bca44cfdd9108521e7100a96917572a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-02 07:27:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"efd3741bcbaafcfea457a16b72f218ae2992ce9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 07:39:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c08250b50c142878001b6b2f50a2fa4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1a2a345a53a44bf803aa5226ee7eb93 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ae5281fdcfd243d2aed8249c5cc87278 : SUCCESS in 9m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a5f61af34b324bed999caae15fde6756 : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fa953393a7d6a02e97582d84559982f649b2d309","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 09:41:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff3f091a0274465ab136f53fb8a6d118\n\n- grenade https://zuul.opendev.org/t/openstack/build/23cdeffd5caa47aab5313dc14a843084 : SUCCESS in 50m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3330fe9e700b490c8d293d985f44c8c4 : SUCCESS in 54m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/02f3e1e8c1174e9d806f9ec74e6ec21f : FAILURE in 1h 04m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1521f77e56374bf28a7d2f5088a17eaf : SUCCESS in 1h 02m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d6e2975cee4ebfa8a764c5029963b4 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58a98965a9eb455bb534a27f67ce2564 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73f0bd308448434c91d230cba55e816d : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ef7eab1bc44400db0e1f12da74affb4 : SUCCESS in 5m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/48b1bc952d6741d1bca07201169662c9 : SUCCESS in 12m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f35242a3fddd45c28d03549202dfb355 : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5ad16c198dc7489e9f3f64323d9e38d4 : SUCCESS in 8m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/36be82e43438493bb673de412e04a684 : SUCCESS in 7m 55s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ab0df7d5efdb46ca9ddbde013878ea75 : SUCCESS in 11m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/078898863db442bbad5308902bd8d1e5 : SUCCESS in 1h 37m 05s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bc459c717be5450ca8002071d91038f6 : SUCCESS in 1h 29m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c11361280cff4951adec61480b5c1eb1 : SUCCESS in 1h 59m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/319c19b8d77a41579956c06d5ee14417 : SUCCESS in 1h 44m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5cb0ce4d776f41149d5ecb258ca48bec : TIMED_OUT in 2h 05m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0b19a00fbf449cc967c944b84f1fe86 : SUCCESS in 1h 02m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/12c694642e0b4625ac759130b198eb85 : SUCCESS in 1h 42m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c5dd8f4138a44e7d8ebd0892c61eeefd : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":2},{"id":"05abf4962349741ccb9962d70b093b1871e51d44","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-03 09:47:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"81ff2f21c9ada0c5c8f9b217936c28e3f47fa672","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-07 05:00:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5b66adb31746362aeaa80d7a564d1078535fa134","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 05:13:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acc87f136fd74a5189c7b187210cd758\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/efeb430fa0b247b5a0f4ab7f62464b91 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5c5e8a801755442bae62c0027ecb53bb : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9dee7c09614f499b84e1234f907b8846 : SUCCESS in 11m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b868c2603f22d27d233b7687b6461d850968d27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 07:09:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/981d7637337545499133e089eaf4ed98\n\n- grenade https://zuul.opendev.org/t/openstack/build/8491a1c1f6574cf0b4f279abbd2a2ac5 : SUCCESS in 1h 02m 42s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/687016adcf684430a6061fec7c9d551f : SUCCESS in 1h 04m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22597637c13c4652892ed1afaa894b10 : FAILURE in 57m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ad43ab6b2aac4ebfb7ccf83a352b5ddf : SUCCESS in 43m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c3fdd64b7f4b32a5260582000489ae : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32b177abbc5b41609dc239ece46dbccd : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f45550bac661485aa263278786d9edf9 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c7328224414118869f3504f2102cd0 : SUCCESS in 6m 08s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8dbe82fb2a1f485ea2bd550f75835ddb : SUCCESS in 10m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f79f6c1a4384fa7b874ccabbcb8bf7d : SUCCESS in 8m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/deb2556cc0344f3484a1ab2fb47222bd : SUCCESS in 10m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/531a2d6fb6eb40f4bdfad922e7b5e81f : SUCCESS in 8m 28s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/51372b58cefe482db0b8afb9923d6e33 : FAILURE in 8m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/db5de2b2c848402d86bae18b48d4574d : SUCCESS in 1h 58m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0635fef90d2d4ffbbc2c7ab11a8c93cc : SUCCESS in 1h 25m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fd5b292bd9e9494b8600742d3997d5cf : SUCCESS in 1h 14m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7914e496acf04580a9762f6804ea5604 : SUCCESS in 1h 45m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6af2d936b0654d4aa4039298cc9df6fd : FAILURE in 2h 01m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bafb71986e841d6a1011aeee3b851ef : SUCCESS in 46m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7816249c94c141f4ae80e17dda982b62 : SUCCESS in 1h 48m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7d78f194c4a24a50bb097442bd87a910 : SUCCESS in 21m 17s","accounts_in_message":[],"_revision_number":2},{"id":"aadae51c2ff0634e78b16ed50d93a3fc911056bf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-07 07:17:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d19f12497a6edbc8600a3770d92f7c79c5df2a1c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 07:24:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7c5928f82cd464fb8a9a0e5d22f22ae\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ae1c947577154eada0007b3691642b39 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b9ef5e2c3cd8471f9e3b8805880d46d5 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bd050e68d103408d86afdda8ecde5e51 : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47528f99049a0831f9e9240a1508ef3735027b52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 09:12:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/214b6946d7d84b5d899081bff2ba56c8\n\n- grenade https://zuul.opendev.org/t/openstack/build/2fe509b4d5734c63a28559458187e1fa : SUCCESS in 1h 03m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/90b6b85246724c69a4c6b1b94a79952b : SUCCESS in 58m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de0020aa15784478b6be6f1619fb1858 : SUCCESS in 1h 24m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/78a0a3ec93fa4e2f86ae854c6dbcc9c8 : SUCCESS in 59m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4283922e4574942bcb50aed1b684af0 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d9ce80e8dea467ea84d0d45dcc50119 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cae1e348d72f4fcf8dc9a1052e761103 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/332d98ce83ad4fa580516cfbb51d8d02 : SUCCESS in 6m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/dd2041a4625c404ab825021bf1d1de75 : SUCCESS in 12m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/48a157b1e2a7473683ad56ad2cf40d1d : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/87016937f3ff47e68195a84dc5a80b58 : SUCCESS in 7m 30s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2c23911f44ef4b3a8341fa84e0561e6f : SUCCESS in 8m 01s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/94f99d11d36e459494fc9e7866bcbeb8 : FAILURE in 6m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/89e59756704145dea9799b9518ffbcd0 : SUCCESS in 1h 23m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/837f4236af064e9ba3ec0e9b74e2ea7a : SUCCESS in 1h 09m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cbd0b35285444a17a3949f073f9e61ab : SUCCESS in 1h 10m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/115a0d8cff6c472cbd37052f644fb3f7 : SUCCESS in 1h 11m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/00ed9bc65aa14c26833a47891324f206 : SUCCESS in 1h 35m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e91b0954ce949a1a03c368c2eb6b07b : SUCCESS in 39m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fc121e9dbc2f487fad1f0f497379944a : SUCCESS in 1h 47m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/45b2833900a943bf8353c3488a8e1220 : SUCCESS in 38m 04s","accounts_in_message":[],"_revision_number":2},{"id":"41844f924fd4737efa9dc33fbc6f991ac3e446cf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-07 09:23:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba30ec61398f71b1d1523acdec0462c93ac1ce09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 09:39:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fde107e48c54f64bc3840b69df52e98\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/11752dd0b53d47699da604bf9e14b331 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5900c022392a45f3bdf9b0139d806f2f : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2bb15956b9fb4480a1e140f2d7123053 : SUCCESS in 9m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8cd93dbbe6c0d47fc693c733ad03439d6f02f2fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 11:37:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6b1fd0dbd9b14ac18d80323d82c6be5e\n\n- grenade https://zuul.opendev.org/t/openstack/build/bba3f023b4c44dbbabf13e20a173db27 : SUCCESS in 1h 07m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/21d5b5b976fc49b4b628becc6d31aa18 : SUCCESS in 42m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea924a32e1044bef81f0e056ba76572b : SUCCESS in 1h 27m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/398baef39b1a4f3fa07a9c66fba44ec9 : SUCCESS in 1h 07m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1539f10c48f4c76a62a4484bbebe60e : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d42ac93a253a43caa15d29711ea1a774 : SUCCESS in 5m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84e9f337b1cd4600b2f8d5dd8c90fb88 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e22e7e171ae547b2add57ee8357513fe : SUCCESS in 6m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d99ed191e007482d9d78193712d7afd1 : SUCCESS in 14m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ca797847edff416590d2a6acb0f24b65 : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/071c445eaf724937845b6939e9b08eaf : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4128d04b78ba444eaf9b35eb2b971428 : SUCCESS in 8m 10s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6f7f06f27ea04a3bb6282b6581253cf0 : FAILURE in 6m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d50884ce43ee40ff8b1d83e6fb79bd45 : FAILURE in 1h 52m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/28226989e0184919958930bcebed4382 : SUCCESS in 2h 01m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c3fede1d91ca49a0888fa02b7117dae3 : SUCCESS in 1h 30m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e0fe66cb57414f4b9f96bc90656570a6 : SUCCESS in 1h 49m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6e45ab20043b4757ae96ee105cab949e : TIMED_OUT in 2h 04m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b585ab2ad7147199d0bdc12b9b1693b : SUCCESS in 45m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0b2d3044d9624d6e90f7ca269be6767e : SUCCESS in 2h 07m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/047891df8dcd41c6b256dc1b341c566c : SUCCESS in 38m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b9ae00b72f6a565c09626a868ae61962dc299c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-08 14:37:54.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b3106824fd6ff1faefa871935a9413738c060f3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:52:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98ff6afd40df4bd1b42b3bcce1306255\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2299d7676cd4481fbb2f2610870a1745 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c5c33ec848f42e68716ad18ee6b1942 : SUCCESS in 12m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a4b64918e6934d678e10fd37f1bf8859 : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c5a8267c35c45afe36678a453d4cbf52b032de61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 16:45:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c72b49782d534e07bff1b5bdbd2556a9\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc34d06de7a744ea891a76ba8d4cc085 : SUCCESS in 1h 06m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e8db63a9eeb841b6a844c652a5595477 : SUCCESS in 59m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eb570c5e2d9f4cbdbe5b587ee481fab3 : SUCCESS in 1h 32m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24fede101fc2440bb4cc2c31d05b5950 : SUCCESS in 1h 05m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cd52b67b93247b0a3ce55282aad484e : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f8016ee1a3447a898205d96db78f389 : SUCCESS in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa3a4622ec164ef782ba4c12a3c7f716 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8a55ebcb4e4c03909bf5dd1b1401dc : SUCCESS in 6m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/3a9a3070af2046949eb57ef0f07ac606 : SUCCESS in 14m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/93b0bc07118a49068771998d60abb1bc : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d9c3d8d5b7d1404f9221a6bb41fabb16 : SUCCESS in 8m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0c9921479e924e55808a7ef41aa87ec5 : SUCCESS in 9m 28s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f94ae4cc26924f26b48335fd3687d95d : FAILURE in 7m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cb903ff0470246889613098e4d0bfab3 : SUCCESS in 2h 00m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c265db38d7104292ac6249d20c61fdb3 : SUCCESS in 1h 35m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90149a92ee824aaba226fc84af669465 : SUCCESS in 1h 25m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f7469c605e1c4b9bb7ce4e811144b6ba : SUCCESS in 1h 47m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ef75dc754cca4ccf87ad91eab22d471e : FAILURE in 1h 42m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9380a56c950c4dcb924832382b79f7c6 : SUCCESS in 45m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f5e48211f0824220a908b148350cb8e5 : SUCCESS in 2h 01m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/66c830fbd2394582929e1078a42bc2f2 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":3},{"id":"9d743e42cf3aa6d0c100cea73bf52ef3e02fd6fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-09 15:22:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4963652d1338038d355f242507f8b08b83774a9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 15:34:18.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5213b5abc8264da689056ac33f5d5f35\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/446a30068c604eb7a0bac3fb2e2c4689 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1323358417194515a6870639d9b336b9 : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ac07fdd1813e46b1a331014b1626a58e : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0824e86266827b0bd84fd04e178148077793bf51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 17:26:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/013c2c682cfd442dabb288989736b9f1\n\n- grenade https://zuul.opendev.org/t/openstack/build/93420c5a9d014d098e6e2cd80c360768 : SUCCESS in 58m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b9d072387afc4bcc83d5b212adb9e1a9 : SUCCESS in 53m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5c4a8c50b884fea9958abbb000e04b5 : FAILURE in 1h 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b8377c6b92b4a6a8ecd18c4db051caf : SUCCESS in 1h 00m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c52c61a32ef442cb9c2d9a830ef72f4 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a2ab9a2ad143279e0a852d0ef6a241 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8a060d23c3564091853623355bd0aa90 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/957fdaeaab6547ba8a5c0da2b419ff4c : SUCCESS in 7m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ce6280f73a034b5eb4f521d098290ce0 : SUCCESS in 13m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f6bdd3d9c605425c9ce7be0d288cd59d : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dca5e43ab36c43929c269e027fe857c9 : SUCCESS in 9m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/68f8939e3ad54fd7acfa1f4d4d497a18 : SUCCESS in 7m 35s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/88fb2d40248044dab93565e59188263b : FAILURE in 7m 26s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7703e4420799458ea9d7135b8b4de18d : SUCCESS in 1h 50m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1de9b4956d44424cb274de6d7991f1d7 : SUCCESS in 1h 42m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9d9e45c6d7ad4c5e9a1e662165d4b05f : SUCCESS in 1h 43m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9cf33da87eba4341a5d0300796297bf8 : SUCCESS in 1h 33m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5b88c251c2ad4f5fa16a54fbad7faec5 : SUCCESS in 1h 40m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/868a7a9c72d7406eaf0c0975cfedf8e2 : SUCCESS in 59m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/28bd7a233c7345d2bfb3793117ef034a : SUCCESS in 1h 58m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a5f92a273ad04578bdf0d1a6e7ee721a : SUCCESS in 37m 45s","accounts_in_message":[],"_revision_number":3},{"id":"dff41f0e83057b5c2eb7ca32480bb5ffabbeb381","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-10 05:05:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2e5d77731c8d8c5eadd89a8b5242c6aeec59ed97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 05:11:34.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fbaaed9e9d14cb6a2d9b060cf5399ab\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6499f8a70da94dc9ba96ec265bca3690 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d55c0f90077d4514b72a81adb42bc7c9 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/afa891be2f3749bcbaa6785a4ff253f3 : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d010b6be115ff4f02b3d1ed7a7a6ff3e29161da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 07:14:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0bb535ae4bb4090925af08f57fbe245\n\n- grenade https://zuul.opendev.org/t/openstack/build/8fb6aa7d15f7452e90e5628ee84634c5 : SUCCESS in 59m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8821f2273ee644e59cf864ce3c8a768c : SUCCESS in 40m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/81f8a56b1fc04875ac065e6ce27c20d1 : SUCCESS in 1h 34m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/05eefd84e057451aa79f1db104ccb062 : SUCCESS in 1h 00m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d07385cccaf14218a10ee657ab276d33 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d01ea8125323470e960728a762948466 : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2181d94a9d345cca2470ba8ec487a37 : SUCCESS in 8m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/652320353566428f841e96c66ec7f19e : SUCCESS in 6m 29s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/fb2989104e264c52abdb7d8e3446528f : SUCCESS in 14m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9cffe138459a48c5b82429c4379cc567 : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fbab7895c10948e8bd04d550be9f5b13 : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4e597473b830451bbc74a6dd8d376a34 : SUCCESS in 11m 58s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5296716595254329ac15565123d55d15 : FAILURE in 5m 54s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/76e38838d32649bfa388c23d99f68673 : SUCCESS in 1h 51m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93a4e69481fd433e84a708b1e5329f5f : SUCCESS in 1h 13m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f43a7325e80b43b982f56b2096b4ee01 : SUCCESS in 1h 15m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cc432698d00f4ccb8d4e7dde79a6559e : SUCCESS in 1h 25m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7197b17511b84af79e872f0e55484899 : SUCCESS in 1h 42m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/585fdbc33f124f0989e9f60776246e7d : SUCCESS in 58m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e2fe0871c6440b8b8cb322d5ea8d189 : SUCCESS in 1h 50m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e71190f9ca1483095b233ee51cc2c59 : SUCCESS in 23m 40s","accounts_in_message":[],"_revision_number":3},{"id":"8c08f143993aaebea67d8a3b636d7fbb49779487","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-11-28 14:45:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"85cd31e5a61093f8a4cae98895e2372473d7e8c8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-19 06:23:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac745e8be721f656bae87e394bf86dec4e6fcb84","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-19 06:36:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15f8dbadadcc0edd97c3cbd128794192d4c7a448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-12-21 04:49:46.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"49dbba7030dece22b8d95cc70824e1efcb651c9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 05:02:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a44b01962a8b46f2a134dc6c42dcd1f4\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fd4b8ba2bc3482aa09cdaf91dcfd265 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eda780632ed24a2db90fd8a97dbd80f3 : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/34b7a0c3c2f9466390f21c932c8335b3 : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d47620a696779b84713918899bcb0ad0ed6d21d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-21 07:03:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1787533ec4564b2d8a6332a706f0afcf\n\n- grenade https://zuul.opendev.org/t/openstack/build/a374deef9b5a483c871ddf3553be3c58 : SUCCESS in 1h 11m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d2e528aa1bd741c58b00d1218037c9c2 : SUCCESS in 48m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d1aced715da04f16836b360de3621c68 : SUCCESS in 1h 29m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c139c4b02db4ee9afc2368ad86dd74a : SUCCESS in 49m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15216a647a0a4522bf479033bfd7ffd8 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5719c25ef5b4ef1a5200b268a6e826d : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c079b6eed61441382b0d04974b957da : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/909ad12bd47b4e6f962f089be45c5ed0 : SUCCESS in 7m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/249666d3419c493a81e1cf557abaec45 : SUCCESS in 13m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/441db647b4f34db0bc8ef7b1a3dddfdb : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4fdced479e824abeb0b1866bf2ba5ae3 : SUCCESS in 10m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d3da38dcb0c84b3ab580dcec1b405369 : SUCCESS in 9m 03s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a5eebbd336364500b0dd7d11d9a6b2d1 : SUCCESS in 13m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/064cfb0cb9164bb0a165273bea6d8c40 : TIMED_OUT in 2h 03m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87e14bb9b8834ded8a478e853c4856e2 : SUCCESS in 1h 25m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d0faa45dd2640769512d41345c7b027 : SUCCESS in 1h 24m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/28c1f18f36e44bc6b5c780b495fd3090 : SUCCESS in 1h 24m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f885d4d0e825479ba724da22bae9aaf7 : SUCCESS in 1h 36m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ffa9b64bdba44e76ac14fa6d0d20204c : SUCCESS in 47m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7fecf27de49e47e9ab0c61fdd8b4a25e : SUCCESS in 1h 46m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cd4fb3f7437042f9bb8d0f490a2d03d9 : SUCCESS in 24m 56s","accounts_in_message":[],"_revision_number":4},{"id":"42a81695a1498980023a3256ee9c02fcfc808654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-16 04:12:07.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"478062b8ff2240d37c400fd90d6625cac677e8b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 04:23:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8324da96e2c467f9cee5ac0b79106b8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5f5db1d69f3411e9b967d51a17a19ae : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ad8859c1a0564f5a97f3eb51d80bcce1 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb0915ff05564d26993430cb3a658367 : SUCCESS in 8m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f09d45de9f52e096a97bbd1dca02f4a6a79ab144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 06:34:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64b8bf2919f94191997055d23ca300fc\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba274fb7185a4afdbbc6a77472494bef : SUCCESS in 59m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b244d1ac54374752bbf3acf41a38b916 : SUCCESS in 56m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d84c1b01ce9c4e19b780d5fa0f7e6fcf : SUCCESS in 1h 25m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8832e418725943afbea7afcfe0b00177 : SUCCESS in 26m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96658db1cdc64d0490f686ef192fb24e : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2412c24a963f48169f08e9a4cb8814c2 : SUCCESS in 5m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc6cced76dae4b5d8b3c902e0cfdd830 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f72db55000b487fa3ab412d5a28e8a8 : SUCCESS in 6m 49s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cf0df949db974b379453febcf33b0570 : SUCCESS in 14m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/538b39fb150945f5b9da5a5eae3ff05b : SUCCESS in 8m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6361bcc124cf4ad4b2cec202bf574c7c : SUCCESS in 9m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/12f70a22ec3d43b29e4a3b5b9e6d6be0 : SUCCESS in 8m 41s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0a5a547a583f44ffbdd28d6d85e07e53 : SUCCESS in 11m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/958167392bd94afabb070101a85ceb6b : TIMED_OUT in 2h 03m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/49ec1867173148da9c71495a3fd676c4 : SUCCESS in 1h 23m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/73e618ca716041048be4e90fbe2c6ebb : SUCCESS in 1h 47m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/47cfcb883dac47d4b49f5b72453e94bd : SUCCESS in 1h 59m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/04b668ff65d147ba9a83fb974cbc6579 : SUCCESS in 2h 03m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6192057234f643bda82e4342253fc7f0 : SUCCESS in 57m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5ba657b2d6e43b79626316f3b46d4ac : SUCCESS in 1h 50m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d3b9b02078b44e35bd462fc6f2be3dad : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":5},{"id":"8380c860fecf66abec45e6186e6f71cbd2159137","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 09:26:42.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1f3b2555da49d1133961338408b0d29eb3b44947","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 09:36:06.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f8db6ca49604491a64bcdd5424631a6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/476e882496394453aa89abde73bf5c12 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a5b2d0ad75df432a8d52991bfdd51e22 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f26463f6e054fc490e01a5222ef92c5 : SUCCESS in 7m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"854e1959d5cdd0ed7651daab9a3e248b58237f4f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-17 09:37:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"430d9dd77652a99b2475d0f67ef07504546b3233","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 11:57:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c9f2c6302d454476ab60eaca30223624\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c854c3a3f1b42d4874292568b767c5a : SUCCESS in 47m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e9e098ed385a4406869071ee8b44a0c6 : SUCCESS in 48m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0785b777182f4fd09f0a9f28b5f56e90 : SUCCESS in 1h 13m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/956de372ab8b4c959a4a073a3af72f56 : FAILURE in 53m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c485ea95f94a6d987cde3268c35f59 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf27223a2e7a4f948e21cc4e9f5d7c8e : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/66afb582e12144b5ada42639783f2228 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0d8b9c127ff4da79f04c735d19259a7 : SUCCESS in 6m 31s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0d67d3813da04404914752fa47bf4393 : FAILURE in 13m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c281839dab84a0bbfd64e6c29da16c8 : FAILURE in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a153f362ab68447aa40cb47e6b24b4e4 : FAILURE in 9m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b5156922bba34530936b43c1fffbdb22 : FAILURE in 7m 45s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/379cf9e99d0c47f5a8a80e5a0635f562 : FAILURE in 11m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/80a008bfc2b84234829be7e7f054e4fd : SUCCESS in 1h 57m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8e4e409968ba488d9c1b9d3dcbdd0554 : SUCCESS in 1h 43m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/016eb781f95849cfb6878e39dbee46ef : SUCCESS in 1h 46m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d85c70077f9f4e7086507861ce5d8f92 : SUCCESS in 1h 44m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e71424b999494086bc8ac91cdb77be41 : SUCCESS in 1h 37m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c84f172d6494114b3ccfbcd7506efd4 : SUCCESS in 53m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0d48226748c84e0699938f8f34542d07 : SUCCESS in 1h 53m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0f8651513e8243cf922b765836de6c88 : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":6},{"id":"c9782bffbb27398b38e8666bda4a700acb45411b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 18:28:54.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5b36d448700cb1a6db78085eb1155eb9f32b0924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-17 18:37:18.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"b18475070c9b092b4e678f4955369eaec9f330a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 18:50:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca1753ce873429f83e6e0a4140249fd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a335ec77561e40d7b32de8d239f00719 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3336ddeff63e4846a9cf852c736e8f57 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9019061ff487474bbcdab447cc7fa648 : SUCCESS in 10m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2ba5a0ba11582300f33ed2cc07e8cf73db6b44b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-17 20:50:39.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e1ea662293d46dcb8630cfacb41f1d6\n\n- grenade https://zuul.opendev.org/t/openstack/build/83f053a4311e4e4698b136d3a35f2867 : SUCCESS in 56m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/222a6743a2e54ffdab3c84d7ab93a1fb : SUCCESS in 1h 04m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c0f77b8334948038f8b20348640ea8b : SUCCESS in 1h 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b92398e57f64babafa967df315036e2 : SUCCESS in 33m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c193a2fa04ea4fe38c048311fe61a6ad : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77dfad08876f4e539d8bf410b5780996 : SUCCESS in 6m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a1384ea2de94cfcb3e6200bff6ee50b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82220febcefa4ac7b92239a020550eaf : SUCCESS in 6m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/51d294adea1e4940abd6596763e6fa75 : SUCCESS in 14m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ecbcbf0803d44389bf552bcb494a079 : SUCCESS in 9m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/19c4c7348a6f47388758e6ee33bd5733 : SUCCESS in 8m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e640ad9c61ca46d38a873ef2a0efa6ab : SUCCESS in 7m 46s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f17a4e25bb0f41d4806035e88a3e5ca0 : FAILURE in 11m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d8196c22e27d468d9ed3694183b4ad91 : TIMED_OUT in 2h 03m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ab70a6307220402aaf6bf06c8c6e76e8 : SUCCESS in 1h 31m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/44e8f6f6acdd40bc8294ac75ef7ea80e : SUCCESS in 1h 32m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a8c6afe64ee6498f9b4e2a7cdb6bdef6 : SUCCESS in 1h 26m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a4d904a3a94d490ab2301e595a4b0b93 : SUCCESS in 2h 06m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ce846ffbf744a2880707bfff246d921 : SUCCESS in 1h 03m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e148c84e8d7843ca860e6787fb3ef156 : SUCCESS in 1h 40m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/287de57fc0fb44f1af1e112873438b48 : SUCCESS in 49m 04s","accounts_in_message":[],"_revision_number":8},{"id":"dc6095f021df80f16e88a01e5e27f60d50bf612e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-18 05:59:22.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"3980e0e476a5ca1e3dce07c4ca62e458e426c5b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 06:13:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/772812c4a7be4b63baba9c182b2877ef\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc7927ce96b548b8a38d4c42e5745a17 : SUCCESS in 11m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d69f7310545f4796b3a9f8cbe48d7df2 : SUCCESS in 10m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5c9e2abaf68a4bf98bce839806ac4372 : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"30f31b73571c8b0da37eaed4dbfd89d091b902a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-18 07:59:05.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecfd7e121f5b4a2d8b3f1e91a6b648b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/6393995260c14e88a153ef4f83de4cb8 : SUCCESS in 54m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8375c34fb28e428ab402b58468a6c5ac : FAILURE in 52m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bacd7c57d42f4bbe9f3185f9467aad49 : SUCCESS in 1h 43m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ba0e9a4d50644b9ad92c43354e0bf1d : SUCCESS in 42m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f302cade89324575a27766aa8139578d : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4688c38b33f4c44968e5566d95e73c2 : SUCCESS in 5m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60e3e054655e4152bfc327221059f361 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3de79127670a4267a1a272e322200983 : SUCCESS in 9m 00s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d2822e7a72cd4740b7f2c2339557b4f2 : SUCCESS in 12m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22fde697c9cf4f7fb67c1af54e41c1f4 : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/05af2b1eded042d895d6ea6496ab4ea3 : SUCCESS in 8m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6ed1df2382b84fdb8ef27891f951c65e : SUCCESS in 8m 49s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b760ff53f5604690b5ece0e74dc41601 : SUCCESS in 12m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e39f17621ada44389b85b021080ac320 : SUCCESS in 1h 37m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eb18a3cc5f044984a7d9944c05658cb4 : SUCCESS in 1h 20m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/147eee95d4dd4bbfbbaa166a3a5e13e7 : SUCCESS in 1h 39m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2ff1ab23691241e8be0c8a62098c94bd : SUCCESS in 1h 35m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8a74c7235cf44380b65c61ebdb1df8e9 : SUCCESS in 1h 31m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f179ba38ec194f409b0428d571169f64 : SUCCESS in 44m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1883f3a61a17419f8a84cae2518bfe32 : SUCCESS in 1h 50m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b79f8188697c454b9e691791ce3e24c5 : SUCCESS in 44m 56s","accounts_in_message":[],"_revision_number":9},{"id":"ab434bc54ed7148fe07ef8500cc48b821f1bb9c1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 10:00:29.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"802e6004db435c95d85edeac57faa9ee8c4054c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-19 10:17:56.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"b21e92ab8d03b926e691c58342981c62876e8f37","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 10:50:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b61943d0672408697c64d503e07fe58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88866f3e61564bcaab69f5ffc74b3527 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30a375b8b762483696a17669c4fb3c24 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dddcf1f7ce6e4ede83917e40a1abceff : SUCCESS in 7m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"81b7c0005741ea72fcd06501310354437c7a3fcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 12:40:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7d831b30ff7464c9db25dd724bfbbd8\n\n- grenade https://zuul.opendev.org/t/openstack/build/340b0264e7b346ba9482a3cc1996122c : SUCCESS in 1h 11m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f60a758fba5e4ac9b7704615cf442d2a : SUCCESS in 1h 02m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13506751e1004b09939e5ebe69711bbc : SUCCESS in 1h 28m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/694aaaddb9ab43a79db9edfa456a7ade : SUCCESS in 35m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e0c67b054c14f939ff3afa0e9f4fd3a : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f16e0a0a400f407a91ecccf8b9ec300a : SUCCESS in 5m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f0c2a1b2062440c9e36cf5aac7dabee : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/547eb73c55d9428e9f13e18742775524 : SUCCESS in 8m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a24d86b68b324c9fb4fa8d7735127dd4 : SUCCESS in 14m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63b8e959c4ad4b54aeb687542e2d46b8 : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7bc2c1e78ae94c9eb3bc21991be94dc2 : SUCCESS in 8m 38s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d22d7f0ee69541f9851d7fcd094ada95 : SUCCESS in 9m 51s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/269d4b5706b745c08c56abede58a7464 : SUCCESS in 12m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/929721bfbd0d4b498562fb3a7afb1df4 : SUCCESS in 1h 48m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d084894820be47719f7545c79b819924 : SUCCESS in 1h 41m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fc561a39e07d4c83b0b882b14c77e840 : SUCCESS in 1h 46m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/75c1b6ad463e4be38f5b83bd7e91e0d8 : SUCCESS in 1h 54m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ac776d39d30c49cbb5ac269e68a1d2ec : SUCCESS in 1h 46m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca5c82a449734cd89df2454caa58ef68 : SUCCESS in 1h 04m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/43bfe6ac7bd14ff79c7ea3ddcdb811fe : SUCCESS in 2h 03m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9136ae0dfd634510b66f9349bad7c8be : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":10},{"id":"028caa102fb9ed9daf2fc04ba5f6ff3646fd04a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-24 14:41:03.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"7baecc3d40c1247e2d536d5cd7013bdca12ff1f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 14:56:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/335987ae74554d8f924b60def9acb56d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4399f87919f34a70b387f85e9ed27e95 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7509716951742189067956344647f58 : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd9785cf7289429fa292a6ee9c073489 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"54ce77a8a72bb011f965ab6f0364f72afda91b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 16:54:39.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a1204a7c3184ff19912654490229227\n\n- grenade https://zuul.opendev.org/t/openstack/build/ac52a2670fc44b698b333f99328504a3 : SUCCESS in 55m 34s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fa5ac6a7330246ceb74c88bae1dba3b3 : SUCCESS in 54m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7621d4bf8084fddbaf1adefad528190 : FAILURE in 1h 05m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43ab935410d14d33a8fbf21ae94efba7 : SUCCESS in 33m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0bba5c2897e451db8d5687086d9c76b : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/248c7ac55e5d4671b39e6c191894133b : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5c6cf4a864ad49a589612bdf0e7b29d8 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2bc84aeca7a4e4fb65000815f4c2cb4 : SUCCESS in 7m 07s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/76ed4f3eb3c64fb29e95ab0b01799227 : FAILURE in 18m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c7fbe7d6bb9e4f4594b8e5645e7131ce : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1bd0df19efab4dba897c728f0b89fbed : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d636be10fb6d4bc5b8994a3d535e57de : SUCCESS in 13m 31s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/296381c64082469c899863ff954eb0cb : SUCCESS in 11m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/421a5eff1f4646fab593e0944aa16c24 : TIMED_OUT in 2h 04m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a3773b5040f04671907cae5492683d5f : SUCCESS in 1h 39m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/954c4cd601964ac7818550b12815f8be : SUCCESS in 1h 57m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/67091cfba7f145d2ac20e36bea571bd9 : TIMED_OUT in 2h 04m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6b45f2906344ac0a3cd54886d732d01 : SUCCESS in 2h 09m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f7e55a35f9a4f1dac508f8a96164195 : SUCCESS in 59m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2c76d667ba9947a2bce0184223c7b062 : SUCCESS in 2h 00m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/720c0006061841fbb9f437d7614eac1b : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":11},{"id":"8652549985ca49cf82192e1691508ef7a3f271b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-01-25 05:22:33.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"1a042cfcefeeb28a3c8e21cb7e7ba567800b21ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 05:29:33.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a712b22e75e49459c40418626ce36b2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dcc1ee9cc7084afc950eee1f7038b6ef : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/18b936b71e9f40c09d2affc787cf284c : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5b5ba995aaa042b6946f93a804cc3bfe : SUCCESS in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ea43ce65b3552b342c20c1c5515944917fb5b080","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 07:52:44.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed41bb0d9c1447fab332f1b9caab2a0e\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2e6290a5c194eee8b80f4f0ac79fb98 : SUCCESS in 1h 02m 15s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e7c65faa574940758dac75359e34f762 : FAILURE in 54m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4a2b5338afeb46d294c45dca3753dc42 : SUCCESS in 1h 35m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e14cc2d11d343faa418952962ca1585 : SUCCESS in 39m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de56228825ab4405ad2a0813b0450608 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/526f27ef9d4f45b9bd31261e344958cd : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/97bef0a11c7a49eabef8b83afb563f33 : SUCCESS in 7m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78496264604f466da17e832d6295fa94 : SUCCESS in 6m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/93454d6473554690810b856ee89a2827 : SUCCESS in 13m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc0925aae56b4fac9e0dc113001f47d5 : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eb3519f9f0bb42389284db146fe62c71 : SUCCESS in 9m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6fa64fc7859948d19eba41c20fea62b1 : SUCCESS in 8m 19s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/803ff3930a1943e4a14d1ae4677f8847 : SUCCESS in 13m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/87235bd5227041f590ce536a678c7df3 : TIMED_OUT in 2h 04m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8e727dd0f32f415cb8aaa6d86c73b8e2 : SUCCESS in 1h 55m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dfc2951fb5704a98a126b8a88fc93d58 : SUCCESS in 1h 52m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1230f5f3c47f439188a42c561626ed2b : SUCCESS in 1h 48m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/449f2a0024864607954e3b0a393549d6 : SUCCESS in 2h 23m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5c4f4e7f32545248602ec633c05de84 : SUCCESS in 1h 11m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c065ac8a04b4022b8a595b6e1833e70 : SUCCESS in 1h 55m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d84ef7ffdb7142beba5c1ddac685e984 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":11},{"id":"360f7b871f38154c62d06f028b3fa003a66bd7f5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-01 06:06:46.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"4f5199baebdffaa6fe8cb9e5accfdabd642813be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 06:20:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d9e085ca447485db7db3e7b1582b10d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3da30206799942109bec99cc0bb223b2 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2173fdbec3bb44d8b0c6dbe2355397d9 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b58d9da526254e04a0ab15bf282845ae : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a510347a8fb7581cdd2de581f09886a7c6b3f184","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 08:16:48.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e14ae8382c3b475c856b694e74426e0b\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1da803f6a9f4ac78a79631202c81dc4 : SUCCESS in 56m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6b965e9ca50947dd92bf2cb7c0dea00e : SUCCESS in 57m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3afe912a19e04bc4be8b1b85d1c9729a : SUCCESS in 1h 14m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b4f8b44cbb04c1eac4db8c0f38b38ec : SUCCESS in 30m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bd33bd18f14903bca877ad676f4b81 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33a40ab8ba91457ba553366d087a49c5 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bc6b968c15a4dc8aa2ecd5037acb064 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756a8ec994354efa87c83248eb53b2ee : SUCCESS in 6m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cbf00384d61046029cdff5e3287e5cbf : SUCCESS in 13m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/54da4ee831fd46568a0d0b3e3012aacc : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d6672bb48ff47d998e1fb38ad16c580 : SUCCESS in 8m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d953288f37b04ec19cd5e1bae57189d9 : SUCCESS in 8m 30s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1d9e887450e44f9eb09fa79458dc4156 : SUCCESS in 9m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/367f8dbe2ec047a3be885c4e5bb4bb98 : SUCCESS in 2h 04m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/64e79f66b67c416f8581aca7cd5bebae : SUCCESS in 1h 40m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a527fa8a915349b7a059fe887220c64f : SUCCESS in 1h 38m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/558b920fa8d7467cb8bbbfb15859a47d : SUCCESS in 1h 30m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/135a13040fb14555aec158de4c6e9a36 : SUCCESS in 2h 00m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/726798fd69124c798d244d32c25c9d1b : SUCCESS in 51m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b07b0701e954abfbe4b6996c1695461 : SUCCESS in 1h 46m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/94a96a3280b7405e894b4bb9d5e5dd36 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":11},{"id":"0e53e978bdb95c4c344ac1391079fe5675b8afab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 10:05:42.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"e2f1cea467425dcef86cbd6e107f2d3f759a6c6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 10:15:42.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd8f779f8ae7411b84ef23729cc5708c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6b83649588f429c86008d1d23c92000 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/66c02c9294e04897a829f4f3993a024f : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8ae268c2cf8243c2be48b953585d2f4b : SUCCESS in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"dd221c99ec104d2649393f68bca2dc224ab21c9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 12:10:37.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a53a5f1c83d84c78818eaf534d4c9d77\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ca4dd0b291b468a8f6d3036914b6e0c : SUCCESS in 56m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/31c46a2dbb2b4b2a8088486c28199bd8 : SUCCESS in 58m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e6c1fa06310468fa72fba19d8e101d2 : SUCCESS in 1h 24m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7959ec4242044f70a539d784c276ee61 : SUCCESS in 31m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3844fcb143a44611980663bf4ad5f22b : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dddb40a12f124253ac8e322127bd844f : SUCCESS in 6m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2f6e99b76c1c41d28e0ba039377c7e4f : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96644df604d4a77b61d06453bb8927b : SUCCESS in 7m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/a2b7dbb77a834d3ab254019738b5cc68 : SUCCESS in 12m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47da7c0a1cce4df58c81f0f823df0469 : SUCCESS in 9m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4a81409bef6e40899231ddd2e92e429c : SUCCESS in 7m 41s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f7744ca5cc1b4db69f1653d487e523da : SUCCESS in 9m 29s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/abb1cf691750433f90b3e637245639a2 : SUCCESS in 12m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/86b6996dd0334ba49dd639f260753426 : SUCCESS in 1h 44m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4bffebcd05b64b45a4771ba446d20f2f : SUCCESS in 1h 46m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ce7f15b9e7f54166881df9bf0f197b4c : FAILURE in 1h 30m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e7dbb40aa6ff48809d69ece5fb967bcc : SUCCESS in 1h 23m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/940a5ca1295049a1992513a09a51d2ec : SUCCESS in 1h 49m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/135314fe5e2f45059702ede75dc87cc2 : SUCCESS in 55m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3f30014500d64a5aa445b16487c97d84 : SUCCESS in 1h 44m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/75bbfe8654e543d5b151d30a66c1fe1f : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":12},{"id":"d16e590019430f91872ad5fb742680f63601dc47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-09 19:54:58.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"eccd5938f4b4977e7004f0847bd37aa80ecb0283","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 20:12:03.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3577bc6b9024d4c896873b9386dd3a2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4877dedd62d348e993143dd50b13e244 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fd805bbfa3bd4da5a34853577c0e6e82 : SUCCESS in 12m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/055376b79ddd456a8c76cd2feca334c8 : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"4515fd1f8ad4f09ee38b2fcd146ebab5b2593c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-09 22:11:22.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/602933a9aa0048309f05b87e47aaf5d6\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ead1032e96749518960c90aed7c8ef6 : SUCCESS in 58m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/eb7ccfe8aec949a9946c86221e208895 : SUCCESS in 55m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8eaad935161e41ebbbb6a37c79250cbb : SUCCESS in 1h 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b847d19dee024228acf8f88673752fe4 : SUCCESS in 31m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3645dfee7d1a4a81844cab059fd735a9 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1d549e3fadb44b1b8e9e41029968f9e : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a5c47032b3c4423a2ee6d9ceb9b9d86 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aaad3336abe4d9ba2411d41dc0d6960 : SUCCESS in 5m 52s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b8a30d249bbe430e96985e792c71ed4a : SUCCESS in 12m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5535275b281e4342bed1267e6f7bb6df : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d11c911ad3b840ad8b32c616868f8260 : SUCCESS in 10m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f9b75c3ec8b646bea5b61ad60e7ba6b6 : SUCCESS in 9m 06s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8f5661cf2af8473cbe69a13549809103 : FAILURE in 10m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/74f50db8a1344e5da2e2bd96bd87d836 : SUCCESS in 1h 52m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/31ff82f4a6ad4a7d9b2523cb3dc58ce5 : SUCCESS in 1h 44m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1cccfdcdabb44d3ca1c1e6e4fea0c7af : SUCCESS in 1h 36m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a9dd7c7709f04c96822c6443cab0d1e6 : TIMED_OUT in 2h 04m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/01b1d85b1cce45b6b19cb66c4d21709c : SUCCESS in 1h 53m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21a8ed6a650f49f7951e18d660d9abf3 : SUCCESS in 55m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/032f0e25bb5243a5b59f269267de8e5c : SUCCESS in 1h 47m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9c2f5e410e424feeb606ca70a423fa18 : SUCCESS in 41m 47s","accounts_in_message":[],"_revision_number":13},{"id":"2509ceccbdda6335aa5760677bc49adee8f633aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-13 18:28:54.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"457856b61eb238a585c3690b441b8974f912509d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 18:40:37.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2804d2dd1d3e4ffea48c285fa1eb4c2b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f3f53f22d22470da0fae6b787c9654c : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3904547413484b9aa6745b6bff414265 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/65ef30d9902f4e028cf398758a8dd3f3 : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"483cc31d4fcde882a48c1ffe3495bdf6587ab69f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-13 20:44:57.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fdf8b58834c482387f06b82562d3a7b\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f28df793e864d548a91788cfb6dc5b5 : SUCCESS in 1h 01m 41s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b5e7f5c809ff4056a5ac3bf5f3f7ce10 : SUCCESS in 54m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ec8ee1302d8436b9a0659b59fe2f6b3 : SUCCESS in 1h 18m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ec05363367744419edfe4ebdf295f97 : SUCCESS in 26m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e87b55d3c68a416ea988eaf131afe171 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/774bfa5997904c319895a9edebeaf499 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88b0a5b300f44ee9baf0d5caef2f0fc4 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed5568b443949ec9f0ad4ce70682e3b : SUCCESS in 5m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/cc61f0d9ae914865baf63f75022022f4 : SUCCESS in 11m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5173ac995c8d4402a37c96abe4d23827 : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e946b2db8ff54b0db6dcc2fdbf775342 : SUCCESS in 9m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2dfb3125d47949d3ad102b9f353fd58e : SUCCESS in 9m 44s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cd224622100c4cdd9310e91423aa080a : SUCCESS in 11m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/72ca2e75a5da413f8773800adb948cea : TIMED_OUT in 2h 03m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d65ff0997f324fe490659b82fe2bcd56 : TIMED_OUT in 2h 02m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/55093d3ffd704a92a640d801c71b5314 : SUCCESS in 1h 37m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/98c76313b64048c286bf61c06d3a1943 : SUCCESS in 1h 38m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2c4c933a5e894e1193e6bac40be75763 : SUCCESS in 1h 32m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e17aebf659614ef1bf570a1f756324a6 : SUCCESS in 1h 19m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20632edc20674340ba21418733f20fd9 : SUCCESS in 2h 04m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ca68b4b61d3c4068aeae1e6124fc48c7 : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":14},{"id":"33fedd054b2e47d740664ae67364109abd4f0186","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 04:35:13.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7d0a40b86958d9c736736e2a0cbbaf7c4c6f82d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 04:48:01.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ab252ea2604461ab6790659e2a8b251\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/025388c3c23d4c0fa5cb1dbdfa42411e : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5a8722e224194337a2a82641b89e2f61 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21063123b9674774bc1a8bc1da5b86b8 : SUCCESS in 5m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0b15ead3cb5bb1cc622964a733395b885ede2525","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 05:35:50.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"7e435b8b4cf5e91d51628c4a90b363e94672261b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 05:48:43.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbbd45235b4d4dceb0b7f62080293ba2\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a7066f27a634c6e819659dcbeafd508 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d6b1136b028a4819aff3b12c396e8370 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/05b1e18c91804583a200f4c00205521a : SUCCESS in 5m 19s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"c6c8de86db0128b74ec82e6b58913a3e92737f35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 07:40:59.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70fab34f33554dd1aeee1d9e0d12512e\n\n- grenade https://zuul.opendev.org/t/openstack/build/d843eacefb8a4a6dabac04982676093c : SUCCESS in 52m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/8b3951722f074804bc735faeee3d52fe : SUCCESS in 55m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/923873b546a84b2e9f9624ae05e2646e : FAILURE in 1h 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85a1efd215d444caa73ba232fdca611d : SUCCESS in 28m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c635237cede46cea98476198bd56c75 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b570e01f0ab94bc699bd62f7c5452b6a : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/976e198e02cd4e31ac56dd1ff0372ede : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daba4471f5274c5b87b93e8a0e1e4a5b : SUCCESS in 8m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/09d49dba84494141b0fd595e900f417a : SUCCESS in 13m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a4a635a9c8e44a14a5cca140ca881262 : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6b9752664f40461e86be2167b8b12f02 : SUCCESS in 8m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a73b38f659e441c589a82802e48aa818 : SUCCESS in 8m 01s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/326e7cecc7044fa797d371b4de7bc8b6 : SUCCESS in 11m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/86cd902585d44443b772a97e973e1628 : SUCCESS in 1h 35m 48s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c1666ce9582340bb83d1cbc6ce8d0c73 : SUCCESS in 1h 53m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8b5cb11b7e24558b4dca4f09b81b90f : SUCCESS in 1h 38m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5f7663aa197743e4aeccb8211a3fa2fe : SUCCESS in 1h 31m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8b9d9932d2be4a439d6b662672783d4e : FAILURE in 1h 52m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6db990005caf4b2f9e71d587370e8658 : SUCCESS in 57m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce236f0e403c4f4a907d97eda3353b1d : SUCCESS in 1h 53m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5fd5b11e7f01440fbe9e66800264c47b : SUCCESS in 25m 02s","accounts_in_message":[],"_revision_number":15},{"id":"f2578caa437703059ca630a95f8c9ddbfd95a339","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-14 10:01:39.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"343867fe76e5497f4b21b62e707849c475984759","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 10:11:24.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b52cfcc2959b485baccf3aa804ef2a73\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c1b2320aedde42aab42f296ad64ac28e : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7dab657e46a34b5fa7d7d20e2b9ab46b : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d0292e2dae34474b8c0bffad86faea68 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"b7304ab2a304c2bac78556374caee74f2de102d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-14 12:23:32.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f79f8aebe5d94d65b9cfe35d4e4637c4\n\n- grenade https://zuul.opendev.org/t/openstack/build/35636c51381249dc89e113cba2f55ffb : SUCCESS in 1h 08m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1a24d985ec7e47c5bc8fc09e2ae331b5 : SUCCESS in 56m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/50060b5ec088490ca32e435298cd4f2e : SUCCESS in 1h 21m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ae7a882a41904ae3a13a67f55de15a37 : SUCCESS in 36m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3c62eafa26e4b87acccaaf78e4f81da : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e839baa6eb3447096418a8274f4541f : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5b76175a40204c87b82444d1fdab8258 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8fc5da80f1944be8a737319353686c3 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/9a86cf4f6fe54afa8f743f8ace3e304e : SUCCESS in 14m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/985c6413e9f24314a3f329e955b2fb45 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/445e1a727b7147aa9a9d7f5377b51416 : SUCCESS in 11m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/397801ad967741f3a531f088733022f7 : SUCCESS in 8m 18s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/24136b542afb44fd93ed45f6040e722a : SUCCESS in 13m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f80db0c4f088453e925052d4a540ba65 : TIMED_OUT in 2h 03m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/232f44b397994d42858fb948b6116d03 : SUCCESS in 1h 33m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0b894936916c4a03926d668990037d9d : SUCCESS in 1h 31m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f52e51cd6f2b41a5abb427eafb7a3113 : SUCCESS in 1h 37m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc52d97bb402440cbdc519a7f2c59066 : SUCCESS in 1h 48m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91b3d01d86c64ef7aa1c3a87f3a3b527 : SUCCESS in 42m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0ef5d38ba6f048918e159fb49f27b3e6 : SUCCESS in 1h 55m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f0291e8e5e4d4d82a7f6337cdf9b30ce : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":15},{"id":"cf6251f473148eb59b40893ecd6cfc6b8cdc13d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-15 17:46:06.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"87297d1226ec2791047cf30e3580ff3e112ed1ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 17:58:56.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b2f69fac6be4166a0fbcb214131c292\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cda109d80c1a49a28b4972844a9cce53 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8eb7a789b3c47288dce6bac276d002d : SUCCESS in 9m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/13d3f155573e44558d5a13bd1b89c73b : SUCCESS in 6m 43s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"ba0bd342a40906784da85b68657cb578b727d518","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-15 19:55:02.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0fddd754362470291782ebcf3ebd9af\n\n- grenade https://zuul.opendev.org/t/openstack/build/8386e61204f14e40aa5a06e1f181de46 : SUCCESS in 1h 00m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ac447f146d4f4ab58228ac7c7133032a : SUCCESS in 59m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de0aaf8dcbbb4a3c9ff576c88a82e901 : SUCCESS in 1h 29m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9fe47151620844038be98f7b4456ab8f : SUCCESS in 38m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c07cf07542549f89d4b84b4676a3da7 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dec240cb45364b1aa2486e8c630c2222 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30900eb588c642a0b76c970145b549a6 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/798484ed3bed4305932c95631096b9f3 : SUCCESS in 7m 14s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e7f3486f4b7b4b8ab883ec0a31256452 : SUCCESS in 14m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf8eb5f31d654be196a7041825c37ae6 : SUCCESS in 8m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/21eaaa6a56fa43e2967b321e5ce7bc9b : SUCCESS in 8m 51s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d06933c99cef4d2b8dc7d6bf2466f119 : SUCCESS in 8m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/798d4b2cf21e44f2be7d8b2e718bab96 : SUCCESS in 11m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/203a14ac731043329ce3dac6a00610ce : SUCCESS in 2h 02m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f48a3c214204698be120ed4c4081865 : SUCCESS in 1h 35m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70ade1bde2804ac49546eb2a151d8d00 : SUCCESS in 1h 27m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ac262e81fb1429d8de75945ffaf14eb : SUCCESS in 1h 29m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/87650b3b8b324db5a65b5757f7253a27 : SUCCESS in 1h 43m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/729f5a0930b74c97a00faa4e8807ea72 : SUCCESS in 1h 01m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca4e8e29cd214dd1a9bd215556304f76 : SUCCESS in 1h 45m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/88e848321fc5493ab0737ee0ea636196 : SUCCESS in 26m 44s","accounts_in_message":[],"_revision_number":16},{"id":"ecf171434fc9c6da816ad453d026a4566dcbf539","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-16 17:12:14.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"7e3484c87463943c84510474d1a7cb50f99f6758","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 17:27:42.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bca22fe6cad342bc8affdd2ea2c32080\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a99ae0e925c643c095e9bbb315b11da6 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/caeaec4c982544629ac94dce187d4b58 : SUCCESS in 11m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ed2856102174408d8d2a99c8bb9d7104 : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"6ab490cc666a342f7614056930dd2bc58146ce16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:32:27.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a76e9ac2ecec425eb55a763abc0f6d0b\n\n- grenade https://zuul.opendev.org/t/openstack/build/98667174d14c456eae4f8a9ed5453dc6 : SUCCESS in 55m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/825699ae01ab4b2f8467493e5cb09e88 : SUCCESS in 57m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0532cf48db79414e8b1978dc1d3e135f : SUCCESS in 1h 35m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5293cd818a74edf8b15471af7df9815 : SUCCESS in 1h 02m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2367e43cacef4261a3705158ac0a5543 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcc3e8215a354a6fa5297f865116c7d3 : SUCCESS in 6m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a2abf045008447db1be3230e48356b4 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9b7699438744c79420bcbb418f1d60 : SUCCESS in 6m 36s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/69a467d446d04a44b6a0c6a32ebd6f27 : SUCCESS in 13m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db91a8f95e6f49ffa06f498a19b0ba83 : SUCCESS in 9m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1baad8e6dffa433e9894ef53c883cc7d : SUCCESS in 10m 03s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3dbb2d82ffad49c48063e80090cbfa27 : SUCCESS in 9m 11s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9a019538bbac4f50a89bbe503e3c507f : SUCCESS in 10m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9857ffc4489d4bdb82df93e90cfa50f0 : TIMED_OUT in 2h 03m 55s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/759cb6d7cf8f42c6ace1000adabe41ec : SUCCESS in 1h 46m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fa64d05f9c0140d2b02363c7c937581b : SUCCESS in 1h 49m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/de56ec47beac4f1a8afa3d7860d6b865 : TIMED_OUT in 2h 04m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ae8bfb625bfc4bf1b2a180455bf6af61 : SUCCESS in 2h 06m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a10966256d2647b99954c2f99997331c : SUCCESS in 1h 04m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d7a85e6e4c27475591221570bba4c93b : SUCCESS in 2h 13m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ae63fb033e214bacaae5b6457814895d : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":17},{"id":"c8ffc5d5288963fab2b44dfe01dd00c97d081dff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-19 05:33:41.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d15fc2ecc8e4972daabe85637e2cdb9c197ce452","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 05:44:30.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d905535e979b4f0dbf98bd47b9920ead\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3996883ba8234f3cb6b86fa91a1aeb14 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9372374575aa49868ef01e0e0f12841c : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d33270491f5b4d1fbdffdc8269c231b9 : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"d87dd99462e645933f293d6650aee2f487e8abb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 07:55:30.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9af2d3c677004a37bf8fd1f9e6cdac0e\n\n- grenade https://zuul.opendev.org/t/openstack/build/be9c9bc137cc446bbb05d5531592aeb4 : SUCCESS in 1h 01m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6aff0e3474ba48a8b031d549aea1da5c : SUCCESS in 58m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7599ddcec641417096db8a674ea2a506 : SUCCESS in 1h 13m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/211fd95b179c40cf8f63b4bb6d3a1697 : SUCCESS in 48m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e586712e88e425da17371f55cc140d6 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e15d477ca05a4c2f85ca8878c8ef0eb2 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e7dfafed9ac4bf9937bfa273ce2bad2 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c6c7b86e7394e0ea981e8274b379b73 : SUCCESS in 7m 57s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0971dbb51f5e4058aaf77214baa5d68a : SUCCESS in 13m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0b69ec4bb9684bac94b7332799669dfc : SUCCESS in 9m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/74b28f1d27f44d65a690e11df16dda06 : SUCCESS in 9m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1cd9dfde19cc4854a2fb48f01dc4aaf4 : SUCCESS in 9m 21s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/00814ae3b6de4f85b9d2ebdd59ba7b99 : SUCCESS in 11m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4b0bbe4d78514b9188a39c7d710f75df : SUCCESS in 1h 59m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b3925c4f458a40c4a60696c733a7a567 : SUCCESS in 1h 51m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7861227252664e14bf6c5ebb84203027 : SUCCESS in 1h 29m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/94eb723eb650487d94cf1496d26e9946 : SUCCESS in 1h 41m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7276c69a63b84afe90eb259d2c2e92d3 : FAILURE in 2h 09m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1ca6c06291d4aa8a44d94ac916b451b : SUCCESS in 57m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4d5f74ce4e1d48e583816289bf446e78 : SUCCESS in 1h 45m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1796580d2a2440cc86f9df3eeb61d0a2 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":17},{"id":"8ff6293641ae622a070133a334c5a634ce7c7f25","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-19 14:25:25.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"9a75df94a624fc21b23eda288e0dd2941bfb4204","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 14:33:09.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4feadd1c7e114f0dab4f15cfd045fe82\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/62b07fb018354c3ab2c67d53747915db : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/27a98325793643e7a4c77b33c03e9aa4 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/156f3e8ab6a648d79a1fe688ad35b276 : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e55f2287dfbf2f89f400eb431cb102c70fbdd280","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 16:57:30.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37f4c59390454c0ebf290afd71f097f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/0f099f208c11483ca3ea372a939cd463 : SUCCESS in 59m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c0985206f715472592e28928bb6d5b7c : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67f28645b742488db1f2bf69bcd740be : FAILURE in 1h 16m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d958589010d47d0a1356b4ea2d790a3 : SUCCESS in 1h 04m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9007f0edb0c04481a1f89dddcbba7a63 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c36b86d054e4d43b976ea6bde33ef4f : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fe7766db3624949bb74e39ecd8e0264 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5cb61c657ff4c1e8d9832cfc53e374f : SUCCESS in 6m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/999b1b6d5a4f4f6a884066e75f043238 : SUCCESS in 13m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2881ecda1284ef0be8e0c145df06c81 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ef4dec9c19fc42cd8cf0bf796e4851b1 : SUCCESS in 9m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/89738ba198484abfb913a33aff95c00f : SUCCESS in 9m 53s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/77f943fc9f44412ebf1be70055de398e : SUCCESS in 11m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d2119d8c94b84894a553a2c71f19034d : SUCCESS in 1h 51m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21e1874ba2b44fd0b837c46b0529d2b9 : SUCCESS in 1h 40m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b769532681234ec98acab3f7a0a92771 : SUCCESS in 1h 43m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f9f404b326404b349ee3f842d7c12b29 : SUCCESS in 1h 47m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8adf7a98946d453db5baf181f6d7fa9b : FAILURE in 2h 24m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d49011d4d3864b8a89ed74cb01b7b6ef : SUCCESS in 50m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/742d6f16caba445fb2442e926f6bb07b : SUCCESS in 1h 51m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a3293f950374cff8128b94c85dd65e0 : SUCCESS in 39m 46s","accounts_in_message":[],"_revision_number":17},{"id":"e17e916b4c63f63dbb89605552991ce1af5ad597","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-19 16:59:56.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"36b07309662d946de05a6e1195b75b4253bab4a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 17:09:37.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3394e57c9db4b3cb9f376a10afd9c39\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/94e6d44f94ea4341b131c88ee4537c1e : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a47d1b543a3f4214b838a07c62822cdc : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2e6a6acf08b9429e86d2ca5759de69a4 : SUCCESS in 5m 22s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a918248517e7f32e273c8664a2519871be193a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 19:18:47.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5d26224a71440a3b4d17159c592c1db\n\n- grenade https://zuul.opendev.org/t/openstack/build/932cac989a9a49679e1d3a2152600d68 : SUCCESS in 1h 01m 14s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d9a7c1bba79340948ea1e51eca498b4b : SUCCESS in 59m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae5d84c37b22463ca236c183f2b81f09 : SUCCESS in 1h 23m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7a2affbf517e4aa0826bcdbb71ccb6a1 : SUCCESS in 58m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e333e44c21644afa936c67e4fbca82a : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cbcb6d88fb448bf946eb5e116277d1e : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f534c16440d4d7ea256a4d135b03a21 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9157dbd1927b4f52bd0fa0c7dc588d2b : SUCCESS in 6m 05s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d03abca274fb4801ae64009bd917ed6e : SUCCESS in 11m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e5b97a6887cb4fa9aef3502ff56ae6c7 : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/594234db46bf43c99c755ed35c1bca01 : SUCCESS in 9m 05s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3c792b77400443c99a555e487b581e16 : SUCCESS in 8m 04s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8f8231b2875044e9bf78699f902aad97 : SUCCESS in 11m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c4604c80e5e04be980620875af15575e : TIMED_OUT in 2h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b53376cb478448689c51c57e3e7daa10 : SUCCESS in 1h 51m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4b06bb108e54981a3001fb2fb2b73fa : SUCCESS in 1h 38m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/45626a2a2d6b4a22b903486fca0964e0 : SUCCESS in 1h 26m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1c5ef00e5f1f4fdbae2880965a27d1f6 : SUCCESS in 2h 06m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/04fd6232a6f04d988b1166419b6e0031 : SUCCESS in 55m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9e33224725364127a43df3f9aafb5c9e : SUCCESS in 2h 11m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a3c8624a297944a783dd71a3087a78e7 : SUCCESS in 31m 53s","accounts_in_message":[],"_revision_number":17},{"id":"84968fe91f5815499eced594fa8ecee96a61d246","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-20 15:02:07.000000000","message":"Patch Set 17:\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"d7a325d57f3f62131f8ab026e258b1dc9c7bd044","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 17:46:18.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1f8b3a159a80ac6d30abe4e1a55bf5e79b0b2629","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-20 20:30:08.000000000","message":"Uploaded patch set 18: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"9e69d4a9495c7de7f977fbf2b025b1ff5114f19c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 20:40:27.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d503365b5fd142f1b243c5a7c2c06d68\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e46a2640591c49d9b7945b1201a079f4 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6316a6debb5a48ee92d831b39297e847 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/99a7dafeb3c14f0db58f482e29bdac10 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3a5da7399ece9debc2e27a734dc0ca84ef12348b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 22:36:44.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/28e1cef1b00c43538a99ccedc892c0d6\n\n- grenade https://zuul.opendev.org/t/openstack/build/f856bb0ec02847c99a0d4ba72353c10c : SUCCESS in 1h 09m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bca89925f64a4a6f9c3e5b6763c0370c : SUCCESS in 48m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/70dcab3bb37f44f4a4955f42a2dfabb6 : SUCCESS in 1h 12m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7d900d7c6254918b0e760b3904466bb : FAILURE in 48m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1be8bc5daa7456ca11d387d24a74727 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0335fc74988f4503b88983cb68a5852d : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7f23b79aa9534407882027ed3631e37a : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0871257778f34d7b98d3ad20ef11f2ba : SUCCESS in 5m 56s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/016dca5572f44497bb655c4df13ed863 : SUCCESS in 13m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/562ff5bf087845f0917669f7a9885298 : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cfc519b9308448758712999ecb66e22d : SUCCESS in 8m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/05685d38041f43e5a18a53b5184212a1 : SUCCESS in 9m 33s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a210fb59471246c09a135239866db4aa : SUCCESS in 11m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c70ed914ec2e461583ce6265f6824577 : SUCCESS in 2h 03m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/924206b141734891a794e8c432840396 : SUCCESS in 1h 45m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8ee97c350e84b41a6160e11b5f643b9 : SUCCESS in 1h 57m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a269e599583949df8189fa0fb05d25b1 : SUCCESS in 2h 00m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/edb58846a5224c3daf688030c20321ea : FAILURE in 2h 04m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49db1f6bdf84442db7f2e81dc15030fa : SUCCESS in 55m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ba0e2b2de2441649c4d370d1b890fb2 : SUCCESS in 1h 43m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c1933ae2df6b4fcca7cfd58433453dae : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":18},{"id":"baf0b1e4da8a5b2ceec28be45edc842a26d6e323","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-02-21 06:56:00.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"28d533b6b7edfa1ac54f2c6cc20eaf34a835d40d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 07:02:09.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed3fa1725f9c41c588b953807296e11c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91d802c82867478fa0aed4cd1bca7ec4 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/373ac4f7b3bc4fe78f753b107919ed1e : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/94b250c4bf464b7ebca8ed14f48f9d50 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"5efcc71826f9fca33b7c87096a5eb30729f1417b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-21 09:01:11.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4d231c416b247a0b683d0e37d1a60dd\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c42c735caf14620a1284261cd422e64 : FAILURE in 57m 28s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/14bd4dd3179d4c59b8eddceb577514d9 : SUCCESS in 55m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/af879adf164e45ba966b4a23fa12802e : SUCCESS in 1h 11m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e7e8250957d146adbe9492f4ef3982d0 : SUCCESS in 53m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/162ba378dfce4858be7a6a411dd9f3f9 : SUCCESS in 6m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a062f259339e43fda7fab6f2e116bff7 : SUCCESS in 7m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/903eb90b55b84c91a1d38794abdd823c : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2042a4cb0a0d428bbba94b6bde93c6c1 : SUCCESS in 5m 13s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0d53f5f0ec6945ce843d342f77e19223 : SUCCESS in 12m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5a4e387ed264f71b4c7a31ce96313cc : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f385e977df8c4d02b26a5ada6e47a3d8 : SUCCESS in 8m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/679d603d4d74453c80d1b963936af399 : SUCCESS in 7m 27s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/75e81405136a4137a93fbe2b6ea3216d : SUCCESS in 9m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/18d001590db942ad8f4638c247922210 : SUCCESS in 1h 54m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e6785347582e41cfae6b146e084895f2 : SUCCESS in 1h 47m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/398e3afea6524676a21c7ff5b538e225 : SUCCESS in 1h 53m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/950b5521f4e84629bae0f378e2c0abc1 : SUCCESS in 1h 41m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1fc09ec2a90d4520a69bccc52b69d996 : SUCCESS in 1h 47m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69e6bd076c684b91ba137731d36ad67e : SUCCESS in 44m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fbe9a6c5789848c9a6a2439734ee9cce : SUCCESS in 1h 39m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5d8e2f6c84604aff8a440196a7451a6e : SUCCESS in 24m 14s","accounts_in_message":[],"_revision_number":18},{"id":"edde9a872331043ff0d2fd0f14ea5960a8cd5a9d","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-02-26 18:27:50.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a897157890cdcbff0fca680acb66383317003680","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 18:39:53.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cb8737887ed4b07a7e2f903687d7ad6\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2cf39c84c5254350b9ade1cd00b9f0d8 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a56db33f9ac64c2f93722d099b18a7f3 : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d56b0dd7b2eb47fa9dac1a9204504229 : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"8987e8244f21bd9b5cfd847d970c23a4474050fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 20:31:30.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/815438daa942412b98f730aa9322f489\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a642f4746a5433191378db2d26b0d16 : SUCCESS in 1h 08m 46s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bf88bc7a2b2f4d5584aeaa804648728e : SUCCESS in 48m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0dbdb079de7340fa98403c3e9ab83f74 : SUCCESS in 1h 34m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ba60a3e05d9243caa8699936b2dd0ba4 : SUCCESS in 1h 03m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa25efc8c0154fada06d71324e3031ee : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9705d821f88e424bba521936f904925a : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e2862990a17462f94adfd85037e6c40 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0234535596d4e55a6d24458a17a18b5 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/7d7de3e6e37b485197503f3e6ba310de : FAILURE in 13m 54s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4aefd3a01a8a429285e1e45eabc31c47 : SUCCESS in 10m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0751b02f11bb4dd48168aac5f9192c31 : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7774abbb7fb9423792dc32423b83ebe4 : SUCCESS in 8m 08s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9d9a3c1385134575a19adb86b132d0ba : SUCCESS in 13m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/48f87f75e02c4e51a13693f18ea473fb : SUCCESS in 1h 43m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/802026f605324a22bf0ccc5176fc4e40 : SUCCESS in 1h 07m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1ef9c5745a214a1395074a658e1d8697 : SUCCESS in 1h 27m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/12429f0a783046598bd315a1c8ecec4d : SUCCESS in 1h 53m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3ed8a29701d84ca1aa805f37addac157 : SUCCESS in 2h 02m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c1f93c7887d490da80ea800d321755d : SUCCESS in 41m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1dd322f8a0054bd69f8802f06ed8478d : SUCCESS in 1h 44m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0c0ee275f14144768341083084b279df : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":18},{"id":"e604e3f36a5efa8d336fa8dd6e06dedcb132f73b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-02-27 13:22:57.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a7138bcf0b013fcf68265362a030b07e2a2c7d6c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-02-27 14:58:49.000000000","message":"Patch Set 18: Code-Review+2","accounts_in_message":[],"_revision_number":18},{"id":"ea761da0812c8788887f644e720e1a4f54f634d8","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-02-27 15:00:49.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"610024b8b2220dc72e10172852335d676e0f4622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 17:55:28.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"ff2634d57d259ff33a937846a87e4bf8dc786d61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 21:57:36.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f94298ab2f1247c49f5758bd7ad91846\n\n- grenade https://zuul.opendev.org/t/openstack/build/e3e2be2ff0324bfeb074da685c3d82cf : SUCCESS in 46m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4b7a68cadd504cf8838a0da0fc62debc : SUCCESS in 55m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/47f947ebed8944528d3997e40a5fd46d : SUCCESS in 1h 05m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/de8e52298b694d67a27eadb0bee88652 : SUCCESS in 47m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27389a268e3e45d3ae43f44bb223b82a : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebcafd8d9e38410e88c78301c85ad57a : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d34a82b0229e4a14a98b693ab43ed669 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec7aa7d6e8da452883461d6e32d630a3 : SUCCESS in 7m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fbd6d3a9b5144b25a9361791e7f86e9e : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/addff3e0047847b88d12e4bef526877a : SUCCESS in 8m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7754bed0f67f4787a598dd8cca9b1b83 : SUCCESS in 1h 43m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5983e5d74cf14ce6ada961c634620228 : SUCCESS in 1h 43m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fc8b9df46c3048c280aac077eb29c42c : SUCCESS in 1h 15m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/718557c73322405e8d1151eb3344730a : SUCCESS in 1h 02m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd619639101349669b14e9d78b11452e : SUCCESS in 1h 57m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a0c14803616a480e9d273d99088ba85f : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":18},{"id":"73fe6e4ecd7dc078b32e249524ef9fd43df6b6e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 21:57:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"0088376995409def28ab771e2c57aa392b4ca71e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 21:59:40.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b24f5f7f0f5140feb4096f93538360ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b1e293506424e06b36922b5aa7bbb89 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"9c7820740a5ceaf31336da39624226b9e398e84d","revisions":{"a28e8154849351d09d9e56f0d06f7a71ae2f127d":{"kind":"REWORK","_number":1,"created":"2023-11-01 14:58:19.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/1"}}},"commit":{"parents":[{"commit":"995c10c36f3aebb53eb4aa3f24549967100624cd","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/995c10c36f3aebb53eb4aa3f24549967100624cd"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 14:37:07.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 14:37:07.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a28e8154849351d09d9e56f0d06f7a71ae2f127d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a28e8154849351d09d9e56f0d06f7a71ae2f127d"}]},"branch":"refs/heads/master"},"ad45433cc6d0907cac4f668523004ffa1790e485":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-11-02 07:27:29.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/2"}}},"commit":{"parents":[{"commit":"495cb6f541ce2379bfbf125da5ae43644cc0b97e","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/495cb6f541ce2379bfbf125da5ae43644cc0b97e"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-02 06:59:38.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad45433cc6d0907cac4f668523004ffa1790e485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad45433cc6d0907cac4f668523004ffa1790e485"}]},"branch":"refs/heads/master"},"dc138e8b33903a939500ac8d91e5677f71b9ed72":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-11-08 14:37:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/3"}}},"commit":{"parents":[{"commit":"7a64552637451706db6395f7f28aef2d1091e846","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a64552637451706db6395f7f28aef2d1091e846"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-08 10:55:56.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc138e8b33903a939500ac8d91e5677f71b9ed72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc138e8b33903a939500ac8d91e5677f71b9ed72"}]},"branch":"refs/heads/master"},"e821788e1983719cce9469be767e8d4f4c5c4200":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-12-21 04:49:46.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/4"}}},"commit":{"parents":[{"commit":"b02063b85709aaabade55aab3749c099a90c0e21","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b02063b85709aaabade55aab3749c099a90c0e21"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-12-11 06:04:26.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e821788e1983719cce9469be767e8d4f4c5c4200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e821788e1983719cce9469be767e8d4f4c5c4200"}]},"branch":"refs/heads/master"},"36c7e57e5a360c03beccacab6386c3ad9889f159":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-01-16 04:12:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/5"}}},"commit":{"parents":[{"commit":"954b5af748311d33abc35f7a27b5c074c27347ca","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/954b5af748311d33abc35f7a27b5c074c27347ca"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-16 03:52:27.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36c7e57e5a360c03beccacab6386c3ad9889f159"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36c7e57e5a360c03beccacab6386c3ad9889f159"}]},"branch":"refs/heads/master"},"e2e9c0d5241492f4395e066f73253bd9f94c569e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-01-17 09:26:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/6"}}},"commit":{"parents":[{"commit":"73a1ba764faec73b59ba85dad56a3891adbbe0a0","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/73a1ba764faec73b59ba85dad56a3891adbbe0a0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 09:05:03.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e9c0d5241492f4395e066f73253bd9f94c569e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e2e9c0d5241492f4395e066f73253bd9f94c569e"}]},"branch":"refs/heads/master"},"b8bb8f350258d2320de1f5d77cd73743670ad7fc":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-01-17 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/7"}}},"commit":{"parents":[{"commit":"1ad4681047b42735881f0e23ca58cf659d339822","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ad4681047b42735881f0e23ca58cf659d339822"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 18:05:09.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8bb8f350258d2320de1f5d77cd73743670ad7fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b8bb8f350258d2320de1f5d77cd73743670ad7fc"}]},"branch":"refs/heads/master"},"eb3f78b91b2b2450431adf34cf1a0aabe7d86817":{"kind":"NO_CHANGE","_number":8,"created":"2024-01-17 18:37:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/8"}}},"commit":{"parents":[{"commit":"d7a89d308ff84d9d96715c9cdfd3e74e4f969dfe","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d7a89d308ff84d9d96715c9cdfd3e74e4f969dfe"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-17 18:17:43.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3f78b91b2b2450431adf34cf1a0aabe7d86817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3f78b91b2b2450431adf34cf1a0aabe7d86817"}]},"branch":"refs/heads/master"},"1681e3406fd867ead99c44ce73ef68a236cec230":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-01-18 05:59:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/9"}}},"commit":{"parents":[{"commit":"0cea286cc0dd536d4a46ab87e8e48eb7752a6aca","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0cea286cc0dd536d4a46ab87e8e48eb7752a6aca"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-18 05:37:24.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1681e3406fd867ead99c44ce73ef68a236cec230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1681e3406fd867ead99c44ce73ef68a236cec230"}]},"branch":"refs/heads/master"},"e350172f7069b162186cd1b760464765b4ad9b5b":{"kind":"REWORK","_number":10,"created":"2024-01-19 10:17:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/10"}}},"commit":{"parents":[{"commit":"ae5d182ff53d14ab88400ae9cf49cb16103be1f1","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae5d182ff53d14ab88400ae9cf49cb16103be1f1"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-19 09:53:13.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e350172f7069b162186cd1b760464765b4ad9b5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e350172f7069b162186cd1b760464765b4ad9b5b"}]},"branch":"refs/heads/master"},"a0640e18fecb0e5d8293ca3e6af26323416186a0":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-01-24 14:41:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/11"}}},"commit":{"parents":[{"commit":"5ce687f2d1c643bde6ca438a15f3eb9bbe9b9d5b","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ce687f2d1c643bde6ca438a15f3eb9bbe9b9d5b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-01-24 07:32:50.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0640e18fecb0e5d8293ca3e6af26323416186a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a0640e18fecb0e5d8293ca3e6af26323416186a0"}]},"branch":"refs/heads/master"},"2554a17323de076416284d09c10a9a85d3b3c296":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2024-02-09 10:05:42.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/12"}}},"commit":{"parents":[{"commit":"45cc8471e81abb26e0228cbc2a359fecf512e15d","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45cc8471e81abb26e0228cbc2a359fecf512e15d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 09:27:49.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554a17323de076416284d09c10a9a85d3b3c296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2554a17323de076416284d09c10a9a85d3b3c296"}]},"branch":"refs/heads/master"},"8508e43088ea4cc1782920b147774bc5022f1c0e":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-02-09 19:54:58.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/13"}}},"commit":{"parents":[{"commit":"6601279bb7895cbc85bc7381db86a3820c4217a0","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6601279bb7895cbc85bc7381db86a3820c4217a0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-09 19:05:31.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8508e43088ea4cc1782920b147774bc5022f1c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8508e43088ea4cc1782920b147774bc5022f1c0e"}]},"branch":"refs/heads/master"},"43ba1b7011391d9550801dd5c3620b721c938835":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-02-13 18:28:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/14"}}},"commit":{"parents":[{"commit":"ee6cc44bfc196b1f599a29baa965bbfd7b1c173b","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee6cc44bfc196b1f599a29baa965bbfd7b1c173b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-13 17:28:23.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ba1b7011391d9550801dd5c3620b721c938835"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43ba1b7011391d9550801dd5c3620b721c938835"}]},"branch":"refs/heads/master"},"1fa50957a063b9e20485f93db0388f580a689a53":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-02-14 05:35:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/15"}}},"commit":{"parents":[{"commit":"91e8fd90934985389f45a165ad3012bef28b4016","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91e8fd90934985389f45a165ad3012bef28b4016"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-14 04:55:14.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fa50957a063b9e20485f93db0388f580a689a53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fa50957a063b9e20485f93db0388f580a689a53"}]},"branch":"refs/heads/master"},"0ac629b32cdd341e88bb480ecf66f7a4870f07f9":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-02-15 17:46:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/16"}}},"commit":{"parents":[{"commit":"50cf2b0b00e0320b1bccd6759e94e721914a1727","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/50cf2b0b00e0320b1bccd6759e94e721914a1727"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-15 16:56:45.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ac629b32cdd341e88bb480ecf66f7a4870f07f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ac629b32cdd341e88bb480ecf66f7a4870f07f9"}]},"branch":"refs/heads/master"},"322dc1e92d299e4ff4b95a0838ffc026aa1d7446":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-02-16 17:12:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/17"}}},"commit":{"parents":[{"commit":"0bbaec2eeb489b1f362e066749c3f71d05e02602","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bbaec2eeb489b1f362e066749c3f71d05e02602"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-16 16:40:54.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with cenralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/322dc1e92d299e4ff4b95a0838ffc026aa1d7446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/322dc1e92d299e4ff4b95a0838ffc026aa1d7446"}]},"branch":"refs/heads/master"},"9c7820740a5ceaf31336da39624226b9e398e84d":{"kind":"NO_CODE_CHANGE","_number":18,"created":"2024-02-20 20:30:08.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/72/899872/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/899872/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/899872/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/899872/18"}}},"commit":{"parents":[{"commit":"0bbaec2eeb489b1f362e066749c3f71d05e02602","subject":"Introduce centralized database driver for image cache","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bbaec2eeb489b1f362e066749c3f71d05e02602"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2023-11-01 20:55:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-02-20 19:51:52.000000000","tz":0},"subject":"Read global config file for cache utilities","message":"Read global config file for cache utilities\n\nMade provision to read gloabl glance-api.conf file by\ncache-pruner, cache-cleaner command line utility tools\nto make those compatible with centralized_db cache driver.\n\nRelated blueprint centralized-cache-db\n\nChange-Id: I88d28dd086b6b57bedd9eda73534fa3e05dc8cc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c7820740a5ceaf31336da39624226b9e398e84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c7820740a5ceaf31336da39624226b9e398e84d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
