)]}'
{"id":"openstack%2Fglance~900026","triplet_id":"openstack%2Fglance~master~I299c3c8274aaf7183a1b257432b06b4f5e1b823c","project":"openstack/glance","branch":"master","topic":"db-migration-version","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-03-01 06:02:39.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2023-11-07 04:59:26.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-03-01 06:02:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I299c3c8274aaf7183a1b257432b06b4f5e1b823c","subject":"Change DB migration constant to 2024_1","status":"ABANDONED","created":"2023-11-03 09:32:02.000000000","updated":"2024-03-01 06:02:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6e14ae1a214cef6ea0cd141bdf7ba12936bd0893","_number":900026,"virtual_id_number":900026,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-03 11:38:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2023-11-07 04:59:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-03 09:32:50.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2023-11-03 09:32:50.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2023-11-03 09:45:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-11-03 11:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"686aa2045fe3036aea46b5c351781149feb1d989","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2023-11-03 09:32:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"462633de8988196cd7ad57783cb72a35a97efb02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 09:45:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4ffa0f6794e4841bb35283e83f5e1ca\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/421c968740864b0b83ae2b10aaa169f8 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49c18c597c1a40868d39f5985bd95b6f : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/473090f8c502473dad9d89d532291212 : SUCCESS in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"753e41ef87a3aa7c5046adcb8b8e4f85cb730dc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-03 11:38:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50ac15058a6e497f94ec04072b8095ef\n\n- grenade https://zuul.opendev.org/t/openstack/build/0c26c9622c9243f394818e4cc5d1dfc1 : SUCCESS in 44m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/830d641f7ed74953ae559a4f465ada76 : SUCCESS in 45m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f191becf9f54df1ac51e36b23296dec : SUCCESS in 1h 02m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40ac5d51663543efab8bf37a1b06a413 : SUCCESS in 43m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c48cc7715fa4ef8bee20b48ff93c54d : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b4401d42d3b419db7dc239af384b08f : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ae6e11acda047e69e8b7d7f641c2ae3 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2d47b71709464cb4f87cb2f2486f54 : SUCCESS in 5m 02s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/5191c37eccbf4b97a962c782cb885815 : SUCCESS in 14m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bb891b9d9a3a42e39a6f494fefda637e : SUCCESS in 7m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99d2187d68bc4e178ee9c9f5936e5ad7 : SUCCESS in 8m 34s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b22e399d8bd44007b39ceaf43d80dff9 : SUCCESS in 8m 32s (non-voting)\n- glance-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/18e3a47e676c4bdda418608e48ef59bb : SUCCESS in 9m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/93d18b6950224cf7ab23aad90baa4c01 : SUCCESS in 1h 58m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/959c93cd82a147ef9c32358d5b7c48d8 : SUCCESS in 1h 42m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bac69b647d3a49dcb8d077e34584b9d8 : SUCCESS in 1h 20m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3fb24bd37cf34476a752a96dc93cdc93 : SUCCESS in 1h 23m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/da26faec9e044130b6c7b93007b3ed1e : SUCCESS in 1h 51m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/329e7e6f39d64cedbf6db1e1e062aa7a : SUCCESS in 53m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0eef11b69a624b22878a4f352b5c65a7 : SUCCESS in 1h 52m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8df5d9614fc143bfb1dd6bb696909e5b : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":1},{"id":"619006aa2e0193b5a46efc8490c5b17a70c00958","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2023-11-07 04:59:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e14ae1a214cef6ea0cd141bdf7ba12936bd0893","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-03-01 06:02:39.000000000","message":"Abandoned\n\nThis is not required now, already updated constant in one of the centralized_db patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"321ac6653738cfe6ddfaf76e259a27b437c437f5","revisions":{"321ac6653738cfe6ddfaf76e259a27b437c437f5":{"kind":"REWORK","_number":1,"created":"2023-11-03 09:32:02.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/26/900026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/900026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/900026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/900026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/900026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/900026/1"}}},"commit":{"parents":[{"commit":"03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963","subject":"Merge \"Unset VENV python_interpreter in fips job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a14a9b9c3fbbe106d1f2cc1e33fb594a5be963"}]}],"author":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-03 09:28:51.000000000","tz":0},"committer":{"name":"PranaliD","email":"pdeore@redhat.com","date":"2023-11-03 09:28:51.000000000","tz":0},"subject":"Change DB migration constant to 2024_1","message":"Change DB migration constant to 2024_1\n\nChange-Id: I299c3c8274aaf7183a1b257432b06b4f5e1b823c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/321ac6653738cfe6ddfaf76e259a27b437c437f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/321ac6653738cfe6ddfaf76e259a27b437c437f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
